)]}'
{"id":"openstack%2Fglance~900172","triplet_id":"openstack%2Fglance~master~I9e09fa6396a959d97e690a08dc318d3aa0a78cfd","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-07 05:49:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9e09fa6396a959d97e690a08dc318d3aa0a78cfd","subject":"Drop unused pyOpenSSL","status":"MERGED","created":"2023-11-06 13:07:01.000000000","updated":"2023-11-07 05:51:06.000000000","submitted":"2023-11-07 05:49:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900172","meta_rev_id":"47386b7486e34d0553519f7e7568906d6dda8893","_number":900172,"virtual_id_number":900172,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-07 05:49:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-11-07 05:02:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2023-11-07 05:02:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-06 13:13:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-06 13:54:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-07 05:02:33.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"df5fa7b6372c491b4207f59315aeed868ba062fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-06 13:07:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"686858a0a82438c6e06a07810bf83b30b34d5ac5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 13:13:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63ff58c3722a4b3d80624f9fe7d8d5a8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/676b739f804941bba6f3af1802142fee : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e41629a36d974a1fae5c55956fdd8185 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/379cb1806d6e4e408b5f92515cf58999 : SUCCESS in 5m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f86c47a699d6c7b0a6439061f34477256c1f97c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 13:54:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/078dc5b2d4804a3bb9fdc2116ac88a7f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/98cfc16b272b4fb29c805d42c25dac56 : SUCCESS in 2m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4527edf7a9bb467f8d1ad9845fb61a6f : SUCCESS in 44m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c90bb021068a47fbbaed84bca88a4548 : SUCCESS in 45m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7db6ee4a362742a485a51ad1d82cbdaf : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14d0eac7ce794fa9b8537a9afbd6ee97 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f4bbc5f78384b6899bf70c6ab94f636 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53e1ad97b48f46008f51b1d7a2e78fa8 : SUCCESS in 6m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/91b0ffa9893845fa901aa228d22337b8 : SUCCESS in 11m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ef7ef50b4d15471eb7685df517facf3f : SUCCESS in 10m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/699a1affdbe045d798509c9897a27707 : SUCCESS in 8m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6bc14d6ad5604a64b67bd3472d2ef05a : SUCCESS in 9m 21s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/20dab85703ac426fb94586f0e4c6c0c0 : SUCCESS in 12m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/de3b39ef32e34f30967e41d96ce8bc6e : SUCCESS in 28m 17s","accounts_in_message":[],"_revision_number":1},{"id":"048f8fe9344faeba410c12c1556f9dc69da6af31","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-07 05:02:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3df3134d2b3facb01177d79877133375c21a3a06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 05:03:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"576a4419d1726956171c93d2a582ede6939207d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 05:49:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/222e3caec95f410f9e6d2e2e3fee0cb6\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/348065ac155a4f33b8099f35357e9cec : SUCCESS in 2m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1322fb85eeff488990c1651d8c3a293b : SUCCESS in 45m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8533067e7e48447dafc3b28874ad912e : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f32c11d876f0420c9071252c46439b06 : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/201e5deac3fe4e0a94025fe95e669676 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/840c08fe06c941db88a78d9da6dd1eb6 : SUCCESS in 5m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7c8aa095b6944acabf29fb70f980a3af : SUCCESS in 8m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/65555bddced94894a5f2322196febbe6 : SUCCESS in 7m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/86613ef04a0f421f9852047666f6961a : SUCCESS in 41m 48s","accounts_in_message":[],"_revision_number":1},{"id":"0a25b7a3e49d6e8b5435b826da6c0f8c4954f9f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 05:49:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"47386b7486e34d0553519f7e7568906d6dda8893","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 05:51:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a88143a700dc461499728ee83e320b0d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2b35144f69449e5a5a960a2e0af9512 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08eedc8c455c0faeda680c54c9994afa4be050a0","revisions":{"08eedc8c455c0faeda680c54c9994afa4be050a0":{"kind":"REWORK","_number":1,"created":"2023-11-06 13:07:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/900172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/900172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/900172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/900172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/900172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/900172/1"}}},"commit":{"parents":[{"commit":"03a14a9b9c3fbbe106d1f2cc1e33fb594a5be963","subject":"Merge \"Unset VENV python_interpreter in fips job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03a14a9b9c3fbbe106d1f2cc1e33fb594a5be963"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 13:06:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 13:06:39.000000000","tz":540},"subject":"Drop unused pyOpenSSL","message":"Drop unused pyOpenSSL\n\nglance no longer imports the library directly.\n\nChange-Id: I9e09fa6396a959d97e690a08dc318d3aa0a78cfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/08eedc8c455c0faeda680c54c9994afa4be050a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/08eedc8c455c0faeda680c54c9994afa4be050a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
