)]}'
{"id":"openstack%2Fglance~900276","triplet_id":"openstack%2Fglance~master~Idd36655f1e495fe40138b9245c8e4f1f81c89d19","project":"openstack/glance","branch":"master","topic":"move-tips-job","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2023-11-09 15:16:56.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-11-09 15:16:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idd36655f1e495fe40138b9245c8e4f1f81c89d19","subject":"Make glance-tox-functional-py39-sqlalchemy-tips job non-voting","status":"MERGED","created":"2023-11-07 09:45:59.000000000","updated":"2023-11-10 11:00:27.000000000","submitted":"2023-11-09 15:16:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900276-move-tips-job","meta_rev_id":"e8d40c44c1d6b6a83b34044038193d68eaf7e206","_number":900276,"virtual_id_number":900276,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-09 15:16:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-11-09 13:24:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2023-11-09 13:24:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-07 10:03:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-07 12:04:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-07 18:42:18.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2023-11-08 17:52:58.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2023-11-09 09:32:26.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"d7922e7085aa6f6cc03c3e857255bb84b97292e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-11-07 09:45:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa1926d40517de559d4db9b559f2b7a2af068322","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 10:03:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47052b1e9ad1410388107cf0c7a4def0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c274e6204fba482e8a08047aac26a4bf : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d3035cc4b98d4d3d93315fde783f1cc3 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/365f88f4fa0942e18dff8d2b3ec8927d : SUCCESS in 6m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc4264eab9bf89713801bba5a194645d402507ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 12:04:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb94a25f44604e8699fe48cd7c3be804\n\n- grenade https://zuul.opendev.org/t/openstack/build/f9498c749cff4ffa9416e833b3cd07c6 : SUCCESS in 58m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d1d76d47016742159c74e9f7e52e8e41 : SUCCESS in 1h 00m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/98fea5c610cc49a6aef5807a9b909b02 : SUCCESS in 1h 41m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d84e58e321847d3bd126a519be3dffc : SUCCESS in 1h 05m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b65b7b66273a4a159b19e8c56f7fd4ea : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9a6d5b8858141c9aa382816b0b05281 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5cd15e8707b84aca8d6c05791fd0cc37 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eda195f385094b20bdce6e9a4f674cf1 : SUCCESS in 7m 02s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2a8a3927366a46a885f9528d59e26dbc : SUCCESS in 14m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b5797bd49a7e49a89f9acbbbf475461f : SUCCESS in 9m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/87f58a5c576344d89a79301674479e5a : SUCCESS in 8m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/78a6ff721dc044278eb295ae428b4122 : SUCCESS in 9m 42s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dc7929c8b73f4b0f9d170e2958562b1b : SUCCESS in 1h 50m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6bfa9bea6ff84c90a0f72901442bd7e8 : FAILURE in 1h 42m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/58ea01df9c2643fdb46eaaa88958d4b9 : SUCCESS in 1h 38m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e4140ba2c1c54d3f88a3a9ba3831068b : FAILURE in 32m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9df6b7c9f6634c3e8a04468947a6f2fd : SUCCESS in 1h 45m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9594804f8f5c44308229750990c144ee : SUCCESS in 1h 02m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9ac4759119c741f988066a5e23fcc0a6 : SUCCESS in 2h 09m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8ae0b7851acf41ddba3962e6897b279a : SUCCESS in 38m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c7a47755f068cdc48970aaef2fc6aa30a8f7e314","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-07 18:42:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1d1378803a485ba7b59ed30044c7c9910af5cc2e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-08 17:52:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed4a6c2b69b71d7b5633cd1160c0275e8086d561","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-08 17:56:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"05906394761acd790b020156875f486770e348b2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-08 17:58:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4559249f8eea7563896f9af23fc9aa878fec044","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 18:11:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91b116cbcbd54006afbd2a53212b9ec3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/107fc1a7be734b509468bb7c14b9645a : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b70f5329097b4a979e99fc53dd92343a : SUCCESS in 11m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4d0590abd2c14510aa373b9543a62e90 : SUCCESS in 6m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"669e95b1a372159ead2ca52f7223f5fff0108985","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 20:09:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/94c25a99f08f42ec91951766d68812dd\n\n- grenade https://zuul.opendev.org/t/openstack/build/14192ec842574f11baa7d1577f00308a : SUCCESS in 49m 45s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/389acc760df4447198e4e819aeb9caf3 : SUCCESS in 46m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e7f2b6a8c359477eaea21afe3183f64a : SUCCESS in 1h 08m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1db072c1d7445838b710c408056d65f : SUCCESS in 53m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1083a2bed42405482bdfe5a193e6a87 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7642882f4ea44cbc81fd5ebc6b85e5b6 : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1c9ebfd9144427cb653a8b3c870a82d : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa5ca8b7c2494e79bd45438f66283d0e : SUCCESS in 6m 53s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a9460c5403d049d483062f68811f33a0 : SUCCESS in 12m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8bde688510d347208dbe888c5c785c9a : SUCCESS in 8m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/53f11817a82d41e3b5853cf55f5dbcd8 : SUCCESS in 9m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f92582da2ff84654872f81b620c5ca3f : SUCCESS in 10m 34s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8be4fd282afe44b48532cef0ff61c30e : SUCCESS in 1h 26m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ef6d2bc05d1b418488757643c5de4002 : SUCCESS in 1h 18m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0e1ce300fc394324bc7c860aedd41992 : TIMED_OUT in 2h 04m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b475a145b5ae4456bf9d0fd5cffc0be1 : SUCCESS in 1h 37m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/110de71e5e2645af87b24b4620e0f6dc : SUCCESS in 1h 51m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bd39e9a6c9e40f8a28a1094bd6fb889 : SUCCESS in 46m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/375a300874614038b8b23cc9986cf9c9 : SUCCESS in 1h 50m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eb846270d1f24ac6a318b4444bb5cca0 : SUCCESS in 38m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3b5be0d1aa48d35ee794667961adea4b69b28665","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-09 09:32:26.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a26ce4e4e9e35dd15da0e9573b5dae6a88d6853","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-09 10:43:45.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9848d83d916331aba6a97ab9bad345edb9ac32f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-11-09 11:12:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2b0b32bec40ea28f13d4e65be27b1ab435815a82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 11:23:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6afef9ac4ca447d989310df400f521d7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79792ae627c8491db3f9cfd00458b750 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a2ef9a8ff07a43aeb53e13a457be05ed : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/66b8d291ecf74d6d9f6f1f181436efc3 : SUCCESS in 8m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"21f3be43582fbc4dff3b7e96577308c8eea1bddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 13:17:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bc2619cd5b1447f9b8bba8aa50441ae\n\n- grenade https://zuul.opendev.org/t/openstack/build/e894fe085c6b4b719aa494eeb48a07ad : SUCCESS in 55m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c68459307ca54f1594aed2f21f00439a : SUCCESS in 50m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/47ff8d5bdc284285af1bce44b1b30246 : SUCCESS in 1h 05m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5faa79ad044401587e91c8358853b23 : SUCCESS in 54m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cad2c773e6a741d8b07606a6e6664f83 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6622b12a2c1140dabc01057699ec2b5e : SUCCESS in 7m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c0718d374194a30b388cebb11eee8f9 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d08f202b43ba4d7d950974e0bae9606a : SUCCESS in 7m 30s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8fa84d3596eb46dbaaa9b1650dfc8ba0 : SUCCESS in 12m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e2312f24cb4a487dbe8627379b6e02c8 : SUCCESS in 9m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7b16af102c7c494e8e5f0149fa8ac75a : SUCCESS in 7m 33s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7cb55c25c1fe4880a6f438485e6a8cb7 : SUCCESS in 8m 13s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/4e05c06c55ff4e7b8147b38179c2b7fb : FAILURE in 6m 08s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/302cd55ce30040cea75b9c403cdf1c8e : SUCCESS in 1h 59m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9f0e4ebb78804291a4304476d007c46a : SUCCESS in 1h 23m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/72bbc914aef44aeca282e235f9d494d9 : SUCCESS in 1h 29m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1f5e9c49faff4160ba44452265da026f : SUCCESS in 1h 45m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/87e0c0bf705d4a99bc012a0ce594e5b1 : SUCCESS in 1h 23m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/585d25e2b5a0480c9d77cb0a37a903ef : SUCCESS in 46m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cda57789d7674e399e306797c8e997a9 : SUCCESS in 1h 41m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8a56b4cceadf4f4ca5ca9409879ae618 : SUCCESS in 27m 38s","accounts_in_message":[],"_revision_number":2},{"id":"93eaf40e12e2345b3d4a5ef9cd4c18492c7dcf61","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-09 13:24:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9d55042003bddf8187da7c7b240dbe15fc22f2d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 13:25:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f2232c843eea2c71af6ef26e1486c3ee6d1915ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 15:16:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/298c99a66be54829acff8f5c038af20f\n\n- grenade https://zuul.opendev.org/t/openstack/build/bd49781fbbd1462e93a76651237311b8 : SUCCESS in 46m 51s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ae29349cf11446e48b5611bad455ac94 : SUCCESS in 1h 00m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/49ecaa5ad00543c89595ab4f4e27549e : SUCCESS in 1h 20m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/598a92437c994eda8c585aad6c45c7cc : SUCCESS in 1h 04m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/363aaa6d0fa94c22838c7cc4ee5e5e6d : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09ca302cf3a3402fa76e806a81ff594c : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b6691a48f054aaf9f2266956687aae6 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e045d70bae9d48a08afb0e7577e6ab46 : SUCCESS in 6m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c772627e3d3644d3b33ac23e54bdef53 : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0dd2d7b59c3349eaa9a13f6288194683 : SUCCESS in 9m 44s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ce39cfac69cd4135a35eca952794a3ba : SUCCESS in 1h 33m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fbc3df149e8e4e0dafcb1226e4b3fa6f : SUCCESS in 1h 43m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4401cf61288147aa9d5004da061f368d : SUCCESS in 1h 42m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8970241ac25461fbd922117a90608aa : SUCCESS in 49m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4aec47c856f94dbc8bcd35f7da3f663e : SUCCESS in 1h 50m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f99c5628ddda44058825245b4d7205bb : SUCCESS in 38m 45s","accounts_in_message":[],"_revision_number":2},{"id":"313437ee7db21773fbd48af85e802d75b80645cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 15:16:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"83998cd824ae9b7d51fb89c9d30b9b486d6c5cab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 15:21:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c98bcd675ff046468221add0171a2c9a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bc294cabefb4618ac56b08fd4d113ed : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2},{"id":"e8d40c44c1d6b6a83b34044038193d68eaf7e206","tag":"autogenerated:gerrit:revert","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-11-10 11:00:27.000000000","message":"Created a revert of this change as I21e3911aaea363dfd2d46974431bf34efd72282d","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"caa6e5d965c1467d7df51ceaa29660fe5140f1c3","revisions":{"b67b81b5092eacef45090f9d572300fc41e07c8e":{"kind":"REWORK","_number":1,"created":"2023-11-07 09:45:59.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/76/900276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/900276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/900276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/900276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/900276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/900276/1"}}},"commit":{"parents":[{"commit":"08eedc8c455c0faeda680c54c9994afa4be050a0","subject":"Drop unused pyOpenSSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/08eedc8c455c0faeda680c54c9994afa4be050a0"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-11-07 09:38:56.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-11-07 09:38:56.000000000","tz":0},"subject":"Move glance-tox-functional-py39-sqlalchemy-tips job from check","message":"Move glance-tox-functional-py39-sqlalchemy-tips job from check\n\nMoving glance-tox-functional-py39-sqlalchemy-tips job from check\npipeline to periodic where remaining tips jobs run\n\nChange-Id: Idd36655f1e495fe40138b9245c8e4f1f81c89d19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b67b81b5092eacef45090f9d572300fc41e07c8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b67b81b5092eacef45090f9d572300fc41e07c8e"}]},"branch":"refs/heads/master"},"caa6e5d965c1467d7df51ceaa29660fe5140f1c3":{"kind":"REWORK","_number":2,"created":"2023-11-09 11:12:26.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/76/900276/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/900276/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/900276/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/900276/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/900276/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/900276/2"}}},"commit":{"parents":[{"commit":"08eedc8c455c0faeda680c54c9994afa4be050a0","subject":"Drop unused pyOpenSSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/08eedc8c455c0faeda680c54c9994afa4be050a0"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-11-07 09:38:56.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-11-09 11:09:11.000000000","tz":0},"subject":"Make glance-tox-functional-py39-sqlalchemy-tips job non-voting","message":"Make glance-tox-functional-py39-sqlalchemy-tips job non-voting\n\nSqlalchemy issue[1] was breaking this job which is currently fixed\nbut in order to unblock the gate moving this job to non voting\nuntil the new release for oslo.utils[2]\n\n[1]: https://bugs.launchpad.net/oslo.utils/+bug/2042886\n[2]: https://review.opendev.org/c/openstack/releases/+/900382\n\nChange-Id: Idd36655f1e495fe40138b9245c8e4f1f81c89d19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/caa6e5d965c1467d7df51ceaa29660fe5140f1c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/caa6e5d965c1467d7df51ceaa29660fe5140f1c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
