)]}'
{"id":"openstack%2Fglance~900418","triplet_id":"openstack%2Fglance~master~Id34f9ae7a2639023c8b6a7937487e7af5c2dccad","project":"openstack/glance","branch":"master","topic":"centralized-cache-db","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-02-28 02:25:41.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"Id34f9ae7a2639023c8b6a7937487e7af5c2dccad","subject":"Move sqlite code in common module","status":"MERGED","created":"2023-11-08 14:37:54.000000000","updated":"2024-02-28 07:00:13.000000000","submitted":"2024-02-28 06:59:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900418-centralized-cache-db","meta_rev_id":"b1d62000536d0b833301c5feff4289b0aab9b83a","_number":900418,"virtual_id_number":900418,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-28 06:59:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2024-02-20 20:30:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2024-02-27 15:05:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2024-02-27 15:05:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-08 14:52:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-08 16:49:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-20 15:02:36.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2024-02-27 15:05:27.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"f70735f3f80470ab9873935c7bf520c0ab50a8d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-08 14:37:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2fe408b6bac5532c3150162f636fdcf921894d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 14:52:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85de885906f440e2bda4cb8ac1a9d35b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/22e4cbf8d88f45bd8598124c875e042c : SUCCESS in 10m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/187d503177424af2b424cdc2cfdf2e07 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2c1cf5a650c6458581a9e83be2f41c95 : SUCCESS in 12m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4702dd3f5363d0ced9f8def4bae6dbb7a611fe54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 16:49:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1613a850b3a4fe38cd277167f0331a5\n\n- grenade https://zuul.opendev.org/t/openstack/build/f1b751ac58f64b3c882c5ec72ca17deb : SUCCESS in 52m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e65bcd95385b40e5a323d8a8e7aa5bd7 : SUCCESS in 58m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/19bc99924ea84c428286dd10cbffa581 : SUCCESS in 1h 32m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5d29b707d4b4b13bd068270b588dc45 : SUCCESS in 1h 00m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/600c9af40593472a8e74afb3eec02558 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9100d71c01284467ab9a8304ee9b1d2e : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/781d1e002f084e5183e69556b0b8cc72 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33297b5cf0e64829bab0ed9b82951479 : SUCCESS in 6m 32s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/92f0f68fff584843af34623b83bdeb3b : SUCCESS in 13m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2c972c5f376343f087fb0b0d7b4d5975 : SUCCESS in 8m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5854e33505bf4a07b6e24c4f381736ad : SUCCESS in 8m 05s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2871965bf5884cc0ba4ccbaa0c7455c4 : SUCCESS in 7m 52s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/e38ee5c420d343d1829b023a074dc467 : FAILURE in 6m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c9bb4534bd0b4cf98545f9cfe8424982 : SUCCESS in 1h 56m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/93fdcabed43e455487b680838865cd2b : SUCCESS in 2h 02m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/eada5fb8da9c4bc4af6a3f588ba2a4a5 : SUCCESS in 2h 01m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7c5aab4114a74668a2b35f474d78d138 : SUCCESS in 1h 41m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9db253933e0b40eaa620594f2923b7a3 : SUCCESS in 2h 00m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35dfd2257ead49418b4fbda747dddfb4 : SUCCESS in 52m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/738eb9a89b6b4b1bb2a582c2f908039b : SUCCESS in 1h 46m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0c1d3faad24947fa88ab0f276af15e21 : SUCCESS in 44m 50s","accounts_in_message":[],"_revision_number":1},{"id":"982c0dc25b735aaf604902e50330b6d662b03c9c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-09 15:22:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e04f5ea7431e7723fa6b0c63f52a527395a30edf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 15:35:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff21cfc38dd143bb9c3346dcd7216b73\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf75e7eb223844c491bdbda69f0f3e0c : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b2cb52afed014104a642bbcc190dccb3 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/544f59ca029b4296ac9e320872154970 : SUCCESS in 10m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"893da8224d85ca11848ed49567ff37ceb44d5827","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:50:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07ad325bdebc4562ab1c8919eb7f2d24\n\n- grenade https://zuul.opendev.org/t/openstack/build/38b7b778197c4ef9be577fae75bfdfd0 : SUCCESS in 55m 42s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e30d3a1d97944a22bcc22aae3871f04f : SUCCESS in 1h 02m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9d882ed9940e4221be52dad3fb25475f : SUCCESS in 1h 04m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3756a040cde44d4984d572770b87ab8 : SUCCESS in 57m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07d4cc85182444d58148f0a16ba60254 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bfb80bac53542839fa7a3e1aa5fbc07 : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17ca1b8a2c034d0191a4f590d87557fa : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bed97b8d1f0a47098f253795812d330d : SUCCESS in 6m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b92ec02ba1474f5b82488345865b888e : SUCCESS in 13m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f33e592ff8bf4e3e87427a94d15c9e58 : SUCCESS in 8m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0fef38553f73479dbe25b010daba6d21 : SUCCESS in 8m 13s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1f17dc894aae4a0ab7fbf80b3075ed5c : SUCCESS in 10m 38s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/3479490e631d4cc0a23f792a958eab47 : FAILURE in 7m 30s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cf8d274e9550486692ed48e48edae055 : TIMED_OUT in 2h 03m 19s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1da28ea72faa434e876abd55348cbbcc : SUCCESS in 1h 52m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/56c02a38e7674e638e04869dae81fff0 : SUCCESS in 1h 27m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/23337ffa0f6541358cb8fb75bc7686b9 : SUCCESS in 1h 45m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cbc1da6a03104597a0e4dd798f90fa0d : SUCCESS in 1h 52m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db9d9f1b1e8943559706daf5d3aeb691 : SUCCESS in 1h 01m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3db4f6b26c36447ca689e0c340e72f63 : SUCCESS in 1h 51m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f2f92eab5a2f44059759f54013728968 : SUCCESS in 45m 03s","accounts_in_message":[],"_revision_number":1},{"id":"2c91881d1ed399a5db9fd16e3e60c1e08d674917","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-12-21 04:49:46.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4412e6e3d0601cb876d4ad31b0799b4cdbae5aa2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 05:03:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75d66a25e6514965923a37e66bf3d656\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5bfbbfe10f5f4b56988ad7f0d13c3454 : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a28d2352ae294e369702a870bc17378e : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7c19d6d8621c4111a433f687fb20e5b0 : SUCCESS in 11m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ca0e711c1a005656563be4891aaf746d043f9f0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 07:06:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/886b29385ee6433d87bd2ca882fb0ee9\n\n- grenade https://zuul.opendev.org/t/openstack/build/6e2d772625f64db89a2c9f876a2151ff : SUCCESS in 48m 45s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/76285c93d34a4518a680944b1a3ebb7b : SUCCESS in 49m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a33a91e7eac447dd915e07519d5180af : SUCCESS in 1h 09m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c62cbc786df4b6ca712861871642d57 : SUCCESS in 48m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff080e09686e4c28af20913fa96d6453 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19467f8072e8473ca60c987f853a0018 : SUCCESS in 6m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/26d208a3976a4f388c63754ffb18f2ff : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f11d458f95547abaea5d54e86f77a9c : SUCCESS in 8m 04s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a0d5fee03ea14198be962966f830a1e2 : SUCCESS in 15m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/15e49f21f5d6450c805e702a731f4a4a : SUCCESS in 8m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1cd3ad822f0340c489680df4d46481aa : SUCCESS in 10m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ad686592238a4801ba3b75a6ac9a4ee5 : SUCCESS in 8m 46s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/463378ed4787473e994f447b664620f4 : SUCCESS in 11m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b207862f2c744fabbb67ee7d94beaf5c : SUCCESS in 1h 57m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7d1d438eade24b57b1692b21c172c00b : SUCCESS in 1h 54m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d298aee5fec948bb92ceab78a587b4c8 : SUCCESS in 1h 39m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/aedf042e985641b7b502f22db3c4ea96 : SUCCESS in 1h 57m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/461995fb7aea4d19bf242b16da7f4de5 : SUCCESS in 1h 46m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72d3ab7458734116b2d46af57d4d8549 : SUCCESS in 56m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1c60dab367194284b23566e88b251f4c : SUCCESS in 2h 05m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2b725517b5264e71872e6544cb00411d : SUCCESS in 39m 10s","accounts_in_message":[],"_revision_number":2},{"id":"098486f498deec7a8d884adc12907dc645bca1fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-16 04:12:07.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"2913e36d4c73326dac16d508cf02db0920bb7e76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 04:24:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab4850f0d47746298525f608ed9a51f3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ac6992cd230409cbf5ba309fe2ceb0e : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/209f45f676ff4cff8199feae85e95ac1 : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9254cacacbe745eea4bbb00affa48aa8 : SUCCESS in 6m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2371d44d80f48393ff6101ae07cb47ab434950d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 06:35:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/347b202622de4bb7981538b24b7b0c2a\n\n- grenade https://zuul.opendev.org/t/openstack/build/442f4a2e022746a6b0288998fc10b59f : SUCCESS in 53m 05s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7a89fab6b1214cca8fe95ac9636f8ffc : SUCCESS in 1h 24m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f31ab9d3341c4688b592bc96229bafaf : FAILURE in 1h 30m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cd302de2e37475ba7741176418b4213 : SUCCESS in 24m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dffa02d049c54f0ebd1ad73dbe49c20f : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33f5ee1bcfcb40f88e62939d220c277c : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/148f3a53d7e24f7eb1ecdaaeea362cc1 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04a22fbfe3234bddb9e943f40a66ada7 : SUCCESS in 6m 16s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/aabe271c610b445ebbf0ff20eb540cb6 : SUCCESS in 14m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bfb8b10deeee4adba74d3281bdcf80c5 : SUCCESS in 8m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/28e3e270315540ee89c8fd6de4f89667 : SUCCESS in 8m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1c2822a5a7c04b9da28be6f53a23ca80 : SUCCESS in 8m 21s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/14bb6f56d6f64354928ca1d61fcfb1bf : SUCCESS in 11m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/612a314f619b4cafaa19419b1783f14d : FAILURE in 2h 02m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/aa7ca0f9339145c79f71dba7d783256a : SUCCESS in 1h 41m 32s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e142dfdcddf64638a78efe56be9d0792 : SUCCESS in 1h 51m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/751ac9f271ec4284bc4f4e6271f206f2 : SUCCESS in 1h 26m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d44a2e009bf9446ab2bfb8a93926ccc7 : SUCCESS in 2h 01m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5306379973048fbb84c0f50f8163648 : SUCCESS in 1h 02m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8c28ff2723f34da986e2fc93e474ea09 : SUCCESS in 1h 41m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1b6798a948954806bd1d7a71874fb4e9 : SUCCESS in 32m 02s","accounts_in_message":[],"_revision_number":3},{"id":"15093e0bfa5e901145bdb09f302c68e2cea4cc24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-17 09:26:42.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1ffc65934f5d27d312ddb9c7167153fb43377f3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 09:38:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee62dcc7b11e41088fa62872b3d907be\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b145887675e649429a3172a3dfd4ca5c : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/97dc4d4ab4984353b8fac8aeb893c1d3 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/efe1365555054eba992f0e2e80a08b85 : SUCCESS in 7m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6148cb917ee28f9a945431b214d194c3a70b94ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 11:50:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9205dc4ee66747d7a6f5379840ba1c36\n\n- grenade https://zuul.opendev.org/t/openstack/build/e31a500b0c85471a9284080449139b58 : SUCCESS in 54m 00s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/25ef3952932647f189d2b34f786bc6d3 : SUCCESS in 50m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d6a49d416af1442489c083440814872e : SUCCESS in 1h 33m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/965fccacbb0d441db5c833961d26052f : SUCCESS in 41m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cb2f6d39bce41ae96fabb1d420b0f8a : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47a343a055ee4dbbaadffd872ea272ff : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ba86fbba0dcc460d905887f76c1999ef : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cf263c4b17a4371871797ceb59db151 : SUCCESS in 6m 00s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/fc3c87e48b4b4cbe9e56872bdfe15b12 : FAILURE in 11m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5fe6d6ad62a14857a853f36ead0db8c2 : FAILURE in 9m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6c87c0e2c88d4cdf9f3eed32c3875311 : FAILURE in 8m 49s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b1e39fdd63e1459e8a9762ae2d1c60a2 : FAILURE in 8m 26s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/9a0fd7fc6833459694170f1ac560e07d : FAILURE in 10m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/12da85bd6ae3455ca0944ce02e4ef162 : SUCCESS in 1h 48m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c2ccbb2d69ce47d6930842fe4cb55134 : SUCCESS in 1h 51m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/31056c1359d84836898c91e1f3c890ca : SUCCESS in 1h 37m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c496244d2e6743b8a7b9f8a6e132b2ee : SUCCESS in 1h 44m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7eb310f5cf2043a9b9ee7c85a810a135 : SUCCESS in 1h 34m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c0310b99984460581afe0b2c3aa487c : SUCCESS in 1h 04m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bb674705eaf440588c429dc1973efb75 : SUCCESS in 2h 17m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e49acc51d2144a49afc2662a96525019 : SUCCESS in 42m 49s","accounts_in_message":[],"_revision_number":4},{"id":"de066bed4c506d3fd0d5c21d8ce85232dc4e7337","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-17 18:28:54.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2aa269b887dc07d71d5503c852fdd46825d09317","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-17 18:37:18.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent tree, and commit message as Patch Set 5.","accounts_in_message":[],"_revision_number":6},{"id":"54fc15dc2f084f2dfd10bc31199c637727ac7f67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 18:50:24.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ce3a6d0f3414f98b08929365469be21\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e99f93beb2504492addc2b18ccadfdef : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e3a515b197504cd4b6cd17bba59b8b42 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a2015b01eea54c6198eefcb4f1f7c804 : SUCCESS in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b7e8e81a0d9654ed6fd3646fa0fc2924e6da8605","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 20:50:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b2a5f6a1a964a9ab1350cc4dca5dbcd\n\n- grenade https://zuul.opendev.org/t/openstack/build/cef2411479e14a7b86b18179e886b18e : SUCCESS in 1h 00m 54s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/54bd31034a2545778e075fd5609de862 : SUCCESS in 49m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6bc145a828cd4352bb57678e27f34190 : SUCCESS in 1h 33m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fb6dadb8a7e4a35adea5057390e7a60 : SUCCESS in 25m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/765f9ee568ab40acb5e3101a664b6147 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c5c615f416e4fb4b28154780dc33c82 : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/676d6094ab014a5a90a57a77226d5ced : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f432da66bb8245b596e4b20e09928df4 : SUCCESS in 8m 01s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/068535ee4dac4b3aa791b431e143060a : SUCCESS in 12m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b1e489d0487640d6bdacb68f9f7bb33f : SUCCESS in 8m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a3d12ff57d114cca9e454e3271d67c06 : SUCCESS in 7m 41s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/34f78384842d4dc596623ca5bf4d6438 : SUCCESS in 8m 54s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/270326c02f8d4c6e9278ec7d746d94d8 : FAILURE in 10m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7ee0475d5c024d428d6c4751d30b3028 : SUCCESS in 1h 56m 19s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/212fe5ad644045dd88185ed21607ed60 : SUCCESS in 2h 03m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/933aab757de543a79220b7817ac1ce33 : SUCCESS in 1h 58m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0859d609554244ae9d5886f2b4514452 : SUCCESS in 1h 47m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dc71a8fab47c44f292d18e22388bf4e6 : SUCCESS in 1h 58m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77fe0b054615420ea81800caad8bb4d7 : SUCCESS in 51m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/375e2cb401574fb98c6e592ddb9747e1 : SUCCESS in 1h 59m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/18cf9e90ef1343279388231839fed70f : SUCCESS in 40m 19s","accounts_in_message":[],"_revision_number":6},{"id":"0e8e704433cf93e9db04554da6b04c52468b6a85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-18 05:59:22.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"1c0cb3d9164b1d31eec367ab17edde2e359b8679","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 06:22:24.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c45e4fb085374f45bd2c5998bf519716\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/45657782e8c14d61b5b9415def2893a3 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8f0be77bcac84c72b0d68eef34fac5dc : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bf20c74c30624d3b9c48ba48da9f5009 : SUCCESS in 6m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"68255421be9907f9c948424c3c292cb6122c14b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 08:13:11.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cde364c981464ee2ac09c3b86d0c962e\n\n- grenade https://zuul.opendev.org/t/openstack/build/957c9224cf634f279105f737833a3eff : SUCCESS in 53m 14s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/95ded78f42684a64ab5bd59cf2c13988 : SUCCESS in 57m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/68c613da2d8f4a9a95839f7313b7e9f0 : FAILURE in 57m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9243fe22750d405a94796eb4a6a542d5 : SUCCESS in 24m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd598f4fb5d6465abb3f01c4dc84bb24 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fd6e8036342481eb46635396946fc9f : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/223dada84ee14d4490b3acd854a2ce56 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aa719f57ed344ff9d2ff74fdee9008f : SUCCESS in 6m 37s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/94f1bc688a52423ea4d03e05bac96cbe : SUCCESS in 13m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5a2d90ed0c154a95b3e144638c017067 : SUCCESS in 9m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5962ff3a77ec43d690a813d82eca536c : SUCCESS in 7m 46s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e092d8b8eb934e24a9b748bb91d3fce8 : SUCCESS in 9m 14s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a5d9b4fb907b46f29efb00e0b131db29 : SUCCESS in 11m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8356917dc9504375a99b465cba9a97a8 : SUCCESS in 1h 58m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9506bc14034b4f879fd3ee0ecb54ca31 : SUCCESS in 1h 46m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2ce9949e5d37400cbf5d7a8b1752e452 : SUCCESS in 2h 01m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6272e5588b924c6f86e45536340c5395 : SUCCESS in 1h 58m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3fd6e0c904824a689d7400f340e29ef1 : SUCCESS in 1h 57m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3f23117773a4a43a898497e99b9da3d : SUCCESS in 1h 10m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a76625bc7338401783300e8f3ae846fa : SUCCESS in 1h 49m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/164fca59d0c14d0eb629f16effb28b47 : SUCCESS in 26m 46s","accounts_in_message":[],"_revision_number":7},{"id":"fb0a0393108513da977001373e3b711dfef369f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-19 10:17:56.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"16e61c4ac5a904722f11fab13445a56e72d2330b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 10:57:29.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/098b2c2428854c9fbbd1802a3841c1f5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3b5b9e10a1f481a8cf6747653828407 : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bb2d5e603dc04cb4945632ea0758e1e7 : SUCCESS in 13m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2474313398e74ce9957c80e95fb27f83 : SUCCESS in 11m 40s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a516b5e5e26ecdd872efaaab96209e8e365ba028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 12:41:22.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ad5136fbbab4a1b9153be74790dc0b2\n\n- grenade https://zuul.opendev.org/t/openstack/build/daab523fb5074d5c9d28713373563114 : SUCCESS in 1h 03m 44s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/84873307049e4b5f92103f10de686154 : SUCCESS in 1h 02m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/796a9c9220ef477d9e8dfda112d1be07 : SUCCESS in 1h 47m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96cc246532604929ba0b15a04f8227d7 : SUCCESS in 36m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f80edafb0bf2477497de495bea25703a : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8097caf9dd1a4983ada6939bf6e51ac7 : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a5035591fa25437c9fbf422706e82d2c : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6c495d2e45543068be6e4228824b40f : SUCCESS in 7m 14s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/62758e601a8643618791dd15211a0e8d : SUCCESS in 16m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7f65e62ba19349de8c85db47407dceb5 : SUCCESS in 7m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5869b3d1719348ad9abb4211b168e7ca : SUCCESS in 7m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/edd874f454f440e0a2963480c11b9f07 : SUCCESS in 10m 34s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/d6ee6904d966400eba7ee4e941600a31 : SUCCESS in 11m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fa9aa674c4064232a1cc153ad709c311 : SUCCESS in 1h 39m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6bd20ecb61624aca8c86d2e18d779064 : SUCCESS in 1h 33m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ec3627ee80f94c64ab84d4ddbc25d0fd : SUCCESS in 1h 37m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/300ea3f4fc83479981777aa7a65ba8b0 : SUCCESS in 1h 28m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2d945d82afe94ab5bcd0776ad8d06d57 : SUCCESS in 1h 26m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8865b8534a984bb2b094031c86122aea : SUCCESS in 59m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f29d6c9dd7f2466ab47f7215b7010c40 : SUCCESS in 1h 54m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/69e948406a754447b2af2aa3f6482f0a : SUCCESS in 35m 17s","accounts_in_message":[],"_revision_number":8},{"id":"4fb932380df1c2d109b9f141fdaf06e57602a5dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-24 14:41:03.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"8652df2b07cb5856e373acf827afb9c1e69ba26a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 14:58:49.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f28450ba79d45a88b2e41effaf6dafe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/34bdff767a5b4a7d899a5335a5a2741c : SUCCESS in 12m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/45f01e3a3ea84289a9196b34e0625f54 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d9ce07da79c744e4b08e2199c7955b07 : SUCCESS in 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4a0502226bb9a1efc803e3bc506b07d4ca2ff6fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 16:56:25.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5d8b68c92f6402b9f241347dcfe6d75\n\n- grenade https://zuul.opendev.org/t/openstack/build/1f0c85a0be974e0593c0a7ac87fcb77e : SUCCESS in 1h 10m 13s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/761fd0ac5f4149deb1277d1357489962 : SUCCESS in 1h 15m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4ebdb5867fab40cd8565cdc688940d4a : FAILURE in 1h 25m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f95535aa45b49d8b83748c33804125c : SUCCESS in 32m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc41961c582c4df3b94137819ede269c : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a61c66ea83f7435db62053c50e6593c1 : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/91460aeb89364b048a00d0552df3ef58 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb4c4319a364f71861d20a4e2434bc5 : SUCCESS in 7m 41s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d2f0477c012641198e1f17e7a5850f7c : SUCCESS in 12m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e9fcf0a100ca4b56825288bf0100f3b2 : SUCCESS in 8m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d72dc38b8de049abbb270721c4116724 : SUCCESS in 10m 16s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c52b7580731745b4934f22f6c245e470 : SUCCESS in 9m 05s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/213c11901ce14a47bfa07ca2d5cc865c : SUCCESS in 13m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/adc286699c3040248b8344733be8a90d : SUCCESS in 1h 42m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b52c2436e7904d3c829928d59fc50411 : SUCCESS in 1h 45m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f221a35ed2ee4d748917f138a2ca08e7 : TIMED_OUT in 2h 05m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3887ed7b96fd43339a8390adbf6fa2aa : SUCCESS in 2h 04m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0032068b83c64f4c9c47136aa0ee4c4d : SUCCESS in 1h 39m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55b74bc2883940dfa6e9999a7a070ded : SUCCESS in 53m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/996dc245f27640b7807e143412f0096f : SUCCESS in 1h 59m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/06248540fc9241df88095d9a360b5dd3 : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":9},{"id":"4f20e62200f0212ec8e3f21eabd1cab458a21419","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-25 05:23:06.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9c2c410127dc3e4197a85c09959e698c711e10d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 05:30:11.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5e1abe593d643048693bc748fa71103\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0bd7b56f0e1646c6b9781abb28fb9a39 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c95f8b4e38af4943a03c43731eb47550 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2a967b8839cb4f7cbbc40455726942b0 : SUCCESS in 5m 26s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c5da7a4c59a75c1192bb620f5589b62e8508dd70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 07:47:07.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f04d5a558f1426f8fb0ccc7de1dfe94\n\n- grenade https://zuul.opendev.org/t/openstack/build/aa2f0980952d400aa52115653e2c3028 : SUCCESS in 1h 00m 00s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c8a2c723486f4648bd2dd571af957eed : SUCCESS in 55m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6c3573d4375e4c4dbf39a507b1ff4356 : SUCCESS in 1h 27m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e453751627347aeb027aed07909a337 : SUCCESS in 31m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd89d9d2c29e4f62989c9bb7bfe42a04 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d629f9324ac1418fb09038533bb1e471 : SUCCESS in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f7c6f7a655ee4b5291381ba227b11300 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdb49cc3723a4c8786183cf914c97935 : SUCCESS in 7m 00s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a8ddae24768c441b91c9b25f575b293f : SUCCESS in 13m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3d033ebe5135406292c15fc7cf0d43f0 : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f1eda50ff7014908a76b61ff685d3527 : SUCCESS in 8m 29s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6702fad0229c468e8090238035c6627d : SUCCESS in 10m 36s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/36fc0c9c0a814a1db7e3ffabbfa9d229 : SUCCESS in 11m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/084889223185416ca452358c2d5b7981 : SUCCESS in 2h 01m 19s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0cb01a27dbc34852aa49193fb9757b3b : FAILURE in 1h 34m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8375fd8d0e5b4a998cede5676aac3d35 : FAILURE in 22m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/97aa9a7852bf40e7bfc8c06d980db2d2 : TIMED_OUT in 2h 05m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/764ee970dbc94b678f68ce7b3647ab99 : SUCCESS in 1h 53m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afbd64d02ccd44ad993bb97f049ab262 : SUCCESS in 53m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b59a773fc4364865bc2e5034e071c24f : SUCCESS in 2h 18m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/786db1645b0949eab5b47686ba508d2e : SUCCESS in 40m 49s","accounts_in_message":[],"_revision_number":9},{"id":"4cf1065c81d015e55d847451ac4eaedf3196b16b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-01 06:06:13.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a62a5bc95543787c7e88b7bafec152cb5216110c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-09 10:05:42.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"bb05c7a002fe1613eb5e550a1357dbc164539087","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 10:15:59.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d3bebf8c66345e0b4aff981f17c1188\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/27f0f4d1667c46158a18f88c46a7d0b3 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5ae01ff0a5fc45b2bc2ae79043eb2363 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4765d696694346eebaebb91f1087f704 : SUCCESS in 7m 56s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7024142234ab53804d1f45bba4197f4e5b74a608","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 12:26:34.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03049414994c4d66a31d05ee2e558d4a\n\n- grenade https://zuul.opendev.org/t/openstack/build/6cc3553759a144f4a8eccb3cf6dd4437 : SUCCESS in 55m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/59ced3dcde254b46b2668f9dc287ef21 : SUCCESS in 52m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c83242a289e4fbdb9c1ef14199b111c : SUCCESS in 1h 23m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9305b5439a14c4ea09130a97539de9d : SUCCESS in 39m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dc7592602ad42e19ba906bc458333fc : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17d780bdac4748be8891bb5553ecfbb8 : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17a84134714040d1aecbe3850360cd3f : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a6ca706b6c14674b11c67c191f1ef7c : SUCCESS in 5m 58s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/6dab33402bbd40a9998c08642f20d464 : SUCCESS in 14m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fb6f4dbab5cb4e618cfaae18bd74a65e : SUCCESS in 10m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1bb639c655154808bf85cc44167e6b3e : SUCCESS in 8m 27s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3594983ae6ce41d9b9f954efb9350302 : SUCCESS in 8m 53s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ead11891a380434fbccbc325b30fcc98 : SUCCESS in 11m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f65df49fe328453ab734fb6bd0cc5f90 : SUCCESS in 1h 53m 32s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/72359fd130de4eb18360eb05d344ab9c : SUCCESS in 1h 51m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/12851d33bea94e739c1840ca3fdf239d : SUCCESS in 1h 48m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/abbad891922547a6a31120beb8310532 : SUCCESS in 1h 52m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7b3caabba0a14272bd4b417967823a7b : SUCCESS in 1h 39m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd4ca6ca5c98416a88241d7e09d43a19 : SUCCESS in 56m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3f75baa684934cb89509875591fa3c32 : SUCCESS in 2h 07m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5732f55a4e414379b0a608dc06fcafbb : SUCCESS in 26m 12s","accounts_in_message":[],"_revision_number":10},{"id":"befec2bbe39c2cbe328e2998a34d64719f1a0980","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-09 19:54:58.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"9fbaef92f84ff3ac59e95710fcc245e3e87ba02e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 20:11:00.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71fd29aa6a984374866cdd663f74526d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9de6083ece1e4fee8c1a35ef315afcc9 : SUCCESS in 11m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/38599af92c2f435a8eda59d2f8650249 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2ef940b78ed94190867e901b8a524ec3 : SUCCESS in 11m 13s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"1c9c769b976ff816edf0a46213d0079d957115eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 22:20:54.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb63f2d1f5dc4434accbbafeb29d24df\n\n- grenade https://zuul.opendev.org/t/openstack/build/294b70f32ef14b0fac632ed0a9109701 : SUCCESS in 1h 08m 22s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/78c776d7758f4e39a8fe9da880f27696 : SUCCESS in 51m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/01d2068a73314272b6fe58b3197b8d4d : SUCCESS in 1h 31m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/040b9fa10e3144219e09a636808e3a25 : SUCCESS in 27m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ea96845eb74126a5bbdd5f9a4ba93f : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5ef17fa6a844da88768def6ca0000ca : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0017ccb5eb7f4c6d8ddf6d234931655c : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50c87407cdfc4355893a9a21b38c3057 : SUCCESS in 6m 19s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b5addfaeeae04210bb89d6e93a4620ee : SUCCESS in 12m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d17a08e0f17343b58da063fd0d5f6a21 : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8990c2e5a35944fc925b93d76a49d29c : SUCCESS in 8m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d50fe29bedfb44429e86f095081552b9 : SUCCESS in 8m 19s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/39933ea3978a42909ffae93e172f66b2 : FAILURE in 11m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/338597d04acf4dcfb7ace49c3ea79b4e : TIMED_OUT in 2h 03m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/316c609dec7b4103a360b88527b48f97 : SUCCESS in 1h 42m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b1e4f94c7c6d4a9b8a4d02abc247d621 : SUCCESS in 1h 44m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8e8cd331bed44e2ead7c342988b443cd : SUCCESS in 1h 45m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ccc95b5e50bd4cfeac05213b5b96db2a : FAILURE in 2h 05m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4e228b865964230a9296a7c777c1a04 : SUCCESS in 47m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1f1deac0557e4a6ca1d4632761081282 : SUCCESS in 1h 55m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0b3b03b72bbc42d7942a84cf157b7854 : SUCCESS in 39m 15s","accounts_in_message":[],"_revision_number":11},{"id":"c7c1450d9247c2b676467f9e81a391caf348bf43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-13 18:28:54.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"33e879a1922108ea427a3008a1e26b82fa0bddd3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 18:41:58.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63178422e7534eeeb1c8e3cc630a1651\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/764924870bc844e4a867784c4d1978c6 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ccaf77044ab64f1bb4a9f28b1e21bdc3 : SUCCESS in 10m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/32050dd5db224c7283f42258ca18e3b1 : SUCCESS in 9m 21s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"672e32b279d709dd9fcd83e51c282d3d5ec71bb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 20:43:38.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/49876718c7e345dcb573568295e722fb\n\n- grenade https://zuul.opendev.org/t/openstack/build/46503ea233864b6a8d436a2c2c020cd7 : SUCCESS in 1h 06m 57s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/858a239e69bd43e9af707edd0db8db87 : SUCCESS in 54m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d471055739ab40da9560d2282ac3b3e3 : SUCCESS in 1h 01m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02b04a1eaa5f4ca9bdd3a78ec9890836 : SUCCESS in 37m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f93d31149244bdbb77639a5f55acc0e : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe69af9b5cdc45a1b6afac573ab579b3 : SUCCESS in 6m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c5c9570aec74c61a40ef41c1d06e2df : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bdd74fe110c4c09a49e3dac2825c6f9 : SUCCESS in 6m 46s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/73c2b2ef2a04400987d0b02a9a970679 : SUCCESS in 13m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/be0200f5660248278ccc8fa44b2ea1b4 : SUCCESS in 8m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3858ea458a564c11b713a00f22398013 : SUCCESS in 8m 23s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b6d29d46aee4474d8d36839d12f39ce9 : SUCCESS in 7m 58s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/641df7ff366947299662500510ee5939 : SUCCESS in 11m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f7f371d478bd42179def8614e482806b : TIMED_OUT in 2h 04m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/980b9cd5a0dd435e9a6f8ec9c9960f6a : SUCCESS in 1h 28m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/06c1b34bd60b420f9013b350ad259232 : TIMED_OUT in 2h 05m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/90719dd620784df28419aa97ffa7034f : SUCCESS in 1h 31m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c40c33ee80c54733af24f7a9e57834c7 : SUCCESS in 1h 49m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/948ce8e96710478999b5d95d1fe54516 : SUCCESS in 1h 03m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5ec1697ea25f4039bea6f8b693f109f9 : SUCCESS in 1h 53m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e0c39ff63632495aae1ea44df8511a3a : SUCCESS in 27m 30s","accounts_in_message":[],"_revision_number":12},{"id":"329055b9d7e53f1f88b96941d6b1b1a8f33aa074","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-14 04:35:43.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"961ec29d5c20196c46e9672545ee3affb940ab7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 04:42:47.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3db6c578f8d54c62880d39648b9492de\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5a0a08b1e584f3d9944094facf2159c : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/88c9871f13334ea6b1d64bd6ddb6cf62 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dded81e41a4240008846b1efa1bb1ca2 : SUCCESS in 5m 44s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"163ff62ae80b3ddb5132910beee96d0ed2c64384","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-14 05:35:50.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"f7cea1f824296083f2921202a4ee68b4f4bbf913","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 05:51:10.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a90d0e3f95814bebbe14ef6ccb609447\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a06f802081d4188afa0160a1bd59115 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bfce000afbc744fd9cd048dfcfc6a26a : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c3780123721b4aacabc8dc7a45b8ebd9 : SUCCESS in 6m 56s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"39c112a07ccf58598bd82d009c34d04f470d840e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 07:48:18.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42755d04a37145d7a34e499e19c90cab\n\n- grenade https://zuul.opendev.org/t/openstack/build/78738500ea0e4c728e85fb07ad8bbbb7 : SUCCESS in 58m 16s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/06786675dcd64e728ff4bb4dd7598920 : SUCCESS in 1h 06m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e9829fc8e4364f318fd4476f1b88ceb8 : SUCCESS in 1h 06m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14837997fbe2402ca0831386f42d60f3 : SUCCESS in 27m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac371c7046644e9f991de6bfa9d642a5 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5828067a65a54c54aa41bcfc8ac96bd8 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1fe0e00152f544b9ba8b4245650b0e5e : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acb07c25f11149268717b61eb5832711 : SUCCESS in 6m 14s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ae7f8a4a01614a26aab40c47e97ff4a6 : SUCCESS in 13m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/674feeddc2e847f8b65a93a633c30a38 : SUCCESS in 9m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1ea68d75088f4feb8d2a4ace80bbe12c : SUCCESS in 10m 24s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1e888ab6bc1a4c02abebd0e9f08a3871 : SUCCESS in 9m 07s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/96fa792db1b641468ca98af9e9afddec : SUCCESS in 12m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a2e4ce32cc1745cc860559e3271b74a3 : TIMED_OUT in 2h 03m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/36fa1c9bbc3c4e96aa3309eaa391f5d1 : SUCCESS in 1h 45m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/341e047bb2de4b59ad6227166fbe948a : SUCCESS in 1h 37m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5dd643e9ca6a45ecb5f1d030c361c79a : SUCCESS in 1h 47m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/44ec0f69da0448ea8fa529357edc6c43 : SUCCESS in 1h 27m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19b1f64189754afeb4d00d10458c2fa1 : SUCCESS in 1h 01m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/70e0cb42faf748369afd95f15481c2c8 : SUCCESS in 1h 52m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/320e79e9510f433582c606922ef33a60 : SUCCESS in 38m 37s","accounts_in_message":[],"_revision_number":13},{"id":"06682d940dcd0c0cbd55ab209edb6c8602c63525","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-15 17:46:06.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"afe03d637325b6df71beae248ac879883b587910","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 17:58:44.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c8cae930dc64981bbcd7e5f7bbf7803\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9c5947b8c7a4680a92c799fcbbcbd82 : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/50dc05ba5f204a3d98eae084d71ccd1f : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8dbb5799d0b44381bd387be64e6f47ea : SUCCESS in 9m 45s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"0d227bea4bc3360169aa355705d4ed8a1a3d9d50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 20:25:13.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a15098a73411483bb0b34ca56797ffb9\n\n- grenade https://zuul.opendev.org/t/openstack/build/e6e0afd2d50b4d9e977d5e53111d3652 : SUCCESS in 51m 05s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c9ffa77574cb450685dbc56b5de34e19 : SUCCESS in 53m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6d64444c658d4c0fbb7481c55970ecde : SUCCESS in 1h 24m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b41435e6be54b0186f2918857ecfdde : SUCCESS in 24m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76a81db85eac45149bc33d6d3cf30e49 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebbc8c9d86f74d47803ea1946a52189d : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70ca0a41cefd429da885c0b8e6302812 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aacfec1517fc42b3bd9cfc6921a5c1fa : SUCCESS in 6m 51s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/02a297bc79e446ebac2596b4e24f944e : SUCCESS in 13m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0854303dd04e41ceb1bd6bd2b4178a7f : SUCCESS in 7m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/66aafcb35c6047f3aa8baa938090bcdf : SUCCESS in 9m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/bc919d7b079a485fbcf3cfb1296c8f16 : SUCCESS in 9m 19s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/51439d1a3fd1412995bbed584ddcc2ea : SUCCESS in 12m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7303fbe808844f988c13a827bb53615a : SUCCESS in 1h 58m 48s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9ab53d0cbe4b48d3a27054eb37cb7151 : SUCCESS in 1h 46m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b85fd02cd10e4ae7a9322c7cf6c86644 : SUCCESS in 1h 56m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2a2d5eaa76a6454e8ff90be7beacb4f8 : SUCCESS in 1h 44m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1a4fa1103302406bae262bd7333d9c65 : FAILURE in 2h 06m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75cff7d817744c21b8eac07f0ab4805f : SUCCESS in 58m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4cb48ba7c2764212ae06dea60f941cd7 : SUCCESS in 2h 23m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/95868383ce424b5ea8a4e06f3e395ebd : SUCCESS in 39m 11s","accounts_in_message":[],"_revision_number":14},{"id":"8f5daee25351dc306f5f5f97509dda90197ead67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-16 17:12:14.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"778a9c1703d4eed37fefa9d963bc44ca14a18be2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 17:24:46.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/773843876bcb475c8dfb92603d9a7379\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4dcd79417c364246a1753701581d0c8d : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/abb0c3ad9ad143e195d93269f01241c3 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d9942d145eac4afc941846fad1abeadd : SUCCESS in 8m 13s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"0695de9ae929fb2b5a0f3a3ea64f480100cc55d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 19:20:52.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9182ca7c1dc64c3eb3eda1de0d9d6c8f\n\n- grenade https://zuul.opendev.org/t/openstack/build/4c95e49384e34162bb57cdfd3774e406 : SUCCESS in 1h 05m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/36498d0ed94d408693989fa4c292ff90 : SUCCESS in 1h 07m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/68cfdaa00bf74540b4a2cd002151a8cd : SUCCESS in 1h 30m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d936bf363bf84acdb6800766f54e3a3f : SUCCESS in 1h 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/419dc7327a9045c398691cf02953f47d : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3605652201824f708907e1d379cca932 : SUCCESS in 7m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5bb77e38ceff49d7b077a3975d853643 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/001377a103f4459ebdc08720d9b3b607 : SUCCESS in 6m 12s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3fb1f6f3a63e4259957f27aaf6ab55f1 : SUCCESS in 14m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d9d251d04eb44809b263dcdd79dc0319 : SUCCESS in 11m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a5f5c4b2532b45a2868585aa1ea8ec3e : SUCCESS in 9m 27s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5fcb195cd4cd45718c31f7d69fc3a2b9 : SUCCESS in 8m 47s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/da78080664a7430fbcbf23328b71d019 : SUCCESS in 11m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4574b5fc82db478bb618a4db7390e76d : SUCCESS in 2h 00m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3969eef2ef054154abfc6fffcdaae969 : SUCCESS in 1h 57m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b720f7beae6445e4b785e9e4b583e41b : SUCCESS in 1h 23m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/601811cd9ec748fda422d73c9ae6e9ae : SUCCESS in 1h 52m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a6a15becde264945b63399e7e5c1c1fd : SUCCESS in 1h 54m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af32f5affc084947b2046e035dec9bb2 : SUCCESS in 1h 00m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/18a846034cba4756bd9c7f98e219590f : SUCCESS in 1h 46m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9ce5d4afe69a499f95b79f61e9a8746f : SUCCESS in 38m 38s","accounts_in_message":[],"_revision_number":15},{"id":"bbdc1437b917d47cf7175c858664245c9cda32c9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-20 15:02:36.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"99bdad659592022ef8f57158efe058da68ab61de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-20 20:30:08.000000000","message":"Uploaded patch set 16: New patch set was added with same tree, parent tree, and commit message as Patch Set 15.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"56d85d94fd78ecb11af1ed0d43619c3a1c8475b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 20:42:45.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae5b4c9f023b4fff89a64eec0b566d8a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f56235106f94bf390e49cf3ef02f990 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/940fca64c5024e8d8637806c2db17ffc : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/19172a358784479eafcce84e987dcd98 : SUCCESS in 10m 11s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"5479fcd05e0906a56913668345cd26f11b96cb31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 22:43:48.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2a612b7961f48f3a572ff3d6fab6f33\n\n- grenade https://zuul.opendev.org/t/openstack/build/636fcd8bc4044183af450f0f8ef34612 : SUCCESS in 46m 55s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0b9c78ba11964d4888eea301f7647ef1 : SUCCESS in 46m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4f6d446938a4412ac05bad08802133c : SUCCESS in 1h 39m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8bda3eaded4413eb554750f1508dff5 : SUCCESS in 49m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/525c39a7c4e2475aa6e676cc1e1cc312 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf785429c32f4f009abcb89e0000abe5 : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/335994eebdc842c59aec0689c18075b5 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8d57536dd1c48489deb0d4ae6cb36fc : SUCCESS in 5m 48s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/34e65f23732845cbb3ce92a224c0346e : SUCCESS in 11m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1421346b86eb41f28fc3dc42efee5cff : SUCCESS in 8m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/41d3ec2ad6ce4ae8a01a9900235d9e4d : SUCCESS in 7m 02s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9b5afc1019b24e75a03e63f120a314e1 : SUCCESS in 8m 38s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5da16b9169874e91bcc6bd5066cf3486 : SUCCESS in 10m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/37956f1574e1421db6094583c3c101b1 : SUCCESS in 1h 33m 37s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/21930f0ae9fb4504aace1caad0ccc963 : SUCCESS in 1h 57m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1e378f1f823845bfb42428a96b1c3773 : SUCCESS in 1h 53m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b78064691237482f92c502aec082fe14 : SUCCESS in 1h 35m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/105748c2cb9e44b2af3d88fcdc2f12f0 : SUCCESS in 2h 11m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ef41949132b4edabac8e033497ccf66 : SUCCESS in 58m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3a87fec193b5410aa0278132d6270af5 : SUCCESS in 2h 07m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/aceffd0f83e94456b79f5f9fd4bc32b7 : SUCCESS in 41m 56s","accounts_in_message":[],"_revision_number":16},{"id":"9bc12f09b82a3c84d32d310897d61527926dedea","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-27 15:05:27.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"7a96172cae0f1888cf3f6c69573e5cd77acfb2e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 17:55:32.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"1c980092b99f80794449ac4031a499336821c8e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 22:06:20.000000000","message":"Patch Set 16: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ace284314fd4faf8143eb17f545e993\n\n- grenade https://zuul.opendev.org/t/openstack/build/c08a6efd84684fdb89e494c11d9eae56 : SUCCESS in 58m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9d4bfc8feebf4b5eb0fa126b7d506704 : SUCCESS in 59m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d78f919f818f424c8049c636124ff0d7 : SUCCESS in 1h 20m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d68723599a847db80b422465f319eef : SUCCESS in 1h 00m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc564e2471a84d2794f41344fe0b98f5 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22d8c8deb4f54499802b77b7585cb71e : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bed70b09d41b459e98571a1bc3f46a84 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9770e56afec945a29d2eb774223a356e : SUCCESS in 6m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9710ce8f57414c2a9c934bf2c8669514 : SUCCESS in 11m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/40e258b561c84794bbfdf40427abd6d8 : SUCCESS in 9m 20s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a35fe25701ac4aec938ec973f4de9af2 : TIMED_OUT in 2h 03m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/243d5e82e6ce46278d29062fa9514229 : SUCCESS in 2h 01m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/da5d28fce58c496f9891331f428cae23 : SUCCESS in 1h 33m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3de024844742483f8f0181d37b401743 : SUCCESS in 42m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/343d672e18434489b27adc7b692a6638 : SUCCESS in 2h 03m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6ab5c97950764155b0399db2bde777ea : SUCCESS in 30m 22s","accounts_in_message":[],"_revision_number":16},{"id":"7438e51320ae8e86fe8abf0ad0d073c262acffec","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-28 02:25:41.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"b8fb4bd2bacf91676f2bf084537b85e1b8f03789","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 02:35:50.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9947ed8984ce49cba198e556a595f50a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd17e141e6154b50a9ed2a92c687150d : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b3c074e053534bcc8ccfd9934ca7f5a9 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a1d167673e2d4913a9007cd49186395e : SUCCESS in 6m 15s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"1f2479bb2681830019b9f356b43d46d8a6c1593c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 04:56:40.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2b1fcd709dc45b7854ff89a7a3d5396\n\n- grenade https://zuul.opendev.org/t/openstack/build/206eaaeebaff40738febe99d85b7ddc6 : SUCCESS in 1h 04m 03s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/774bb43c53494290805b0670cfb60bef : SUCCESS in 59m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b12cee029fbb43f0b45182169656c234 : SUCCESS in 1h 21m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5da00e6931b0455c80f3fecc567d97fa : SUCCESS in 1h 01m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aec30f4eb1dc4c1fab13a77d66537065 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a241c4cfe97f4c5d94eb6da92ca71756 : SUCCESS in 6m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d60982e0d96b4643a2772e0946090bca : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/720f62d1ef764011b5672e8b05f0e42b : SUCCESS in 6m 31s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2bd2af71a7f5474da194530e6e1c918d : FAILURE in 15m 32s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/af27542cb43d4b64a60414f80e0cc28a : SUCCESS in 8m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/825a12ade1884f05855d2974ed7ba2e4 : SUCCESS in 8m 22s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d5cb7a177945455fb13fd2f18b8c07a1 : SUCCESS in 9m 09s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/397e8633fafa4d33a5e9b8dc33041ca9 : SUCCESS in 10m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0b1b4382d62049c0a415a1ddf4280be4 : TIMED_OUT in 2h 03m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bc2d4302c0c74e1b9f6a9de6a5399089 : SUCCESS in 1h 59m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/365ac072f0a04ba7aef473b88174a105 : SUCCESS in 1h 28m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/86e464e3d82c4374890b9394b8810db3 : SUCCESS in 1h 52m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/73301716437549be90545df72b0f0bfc : SUCCESS in 1h 48m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ea911b5bf01460384c1d85a93ded49a : SUCCESS in 1h 01m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b744b9794bc64d08bdeb118845d8ae53 : SUCCESS in 1h 59m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dd662b055f124a3f98df9f94be9801ca : SUCCESS in 36m 02s","accounts_in_message":[],"_revision_number":16},{"id":"a6b71e3e5d1951c39ae3f5a1a5e8c64ceaa3982e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 04:57:35.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"f94f731af5227972f620f29a6cf32e88a8c6079d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 06:59:06.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3de080c28f55496a9d801307e9f93d24\n\n- grenade https://zuul.opendev.org/t/openstack/build/66ec7b5e81464781bf601c26b30bfe8b : SUCCESS in 1h 07m 05s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/981362d092f946469d4ff36233b2d4cc : SUCCESS in 48m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6ddcabc59472428985871684c218c247 : SUCCESS in 1h 04m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d03fcf2fb05c462f9e29b2e018fc5b61 : SUCCESS in 1h 00m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/164dd3b9c4d844cf87a242d8576e743f : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67ee6c6a7e924d1b8a014b0ed54f7d74 : SUCCESS in 5m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c07827f6dca4a64a76c47a2078ab16a : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09e4115023984552967e9b24cc2f5f4d : SUCCESS in 7m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/247da567e3c448e69e91625085657706 : SUCCESS in 9m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/38062b373af24e74912cf7619772efc9 : SUCCESS in 11m 19s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/855ee2dc5bfc48d283474133478d5c79 : SUCCESS in 1h 19m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6651654aebfe40f5a9d8ee5d7f59a4b7 : SUCCESS in 1h 35m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/161b28d79eaf4539a34fb4106dcf53b9 : SUCCESS in 1h 50m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7df951b32c0841278bfd53b19a2f6e40 : SUCCESS in 1h 05m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5783dbf2493f49539ab947d6aa028016 : SUCCESS in 1h 55m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ebfedd21b90f4066af21638112835e94 : SUCCESS in 43m 00s","accounts_in_message":[],"_revision_number":16},{"id":"96e0461bb2d7eae3b335d0a877cae33d88e5bd29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 06:59:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"b1d62000536d0b833301c5feff4289b0aab9b83a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 07:00:13.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70cdca553e824e4ba1652e0a983b82bc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e43f4b960710409b88186d9e8fb82b0d : SUCCESS in 48s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"21333c1ac85e52e31a29db2529abf3dfdd04e9a0","revisions":{"436cb274b0f78441d065788118810eef77648d5e":{"kind":"REWORK","_number":1,"created":"2023-11-08 14:37:54.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/900418/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/900418/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/900418/1"}}},"commit":{"parents":[{"commit":"dc138e8b33903a939500ac8d91e5677f71b9ed72","subject":"Read global config file for cache utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc138e8b33903a939500ac8d91e5677f71b9ed72"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-02 08:07:05.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 12:28:31.000000000","tz":0},"subject":"Move sqlite code in common module","message":"Move sqlite code in common module\n\nMoved sqlite connection code in new module so that we can use\nthe same in migration phase. Once sqlite driver is removed then\nwe can remove this common module as well.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: Id34f9ae7a2639023c8b6a7937487e7af5c2dccad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/436cb274b0f78441d065788118810eef77648d5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/436cb274b0f78441d065788118810eef77648d5e"}]},"branch":"refs/heads/master"},"65caf39e9e8e70e4aa9c96c486d9b30ac0152758":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-12-21 04:49:46.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/900418/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/900418/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/900418/2"}}},"commit":{"parents":[{"commit":"e821788e1983719cce9469be767e8d4f4c5c4200","subject":"Read global config file for cache utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e821788e1983719cce9469be767e8d4f4c5c4200"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-02 08:07:05.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-12-11 06:04:26.000000000","tz":0},"subject":"Move sqlite code in common module","message":"Move sqlite code in common module\n\nMoved sqlite connection code in new module so that we can use\nthe same in migration phase. Once sqlite driver is removed then\nwe can remove this common module as well.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: Id34f9ae7a2639023c8b6a7937487e7af5c2dccad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65caf39e9e8e70e4aa9c96c486d9b30ac0152758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65caf39e9e8e70e4aa9c96c486d9b30ac0152758"}]},"branch":"refs/heads/master"},"5a4d46d25aaafb913a8e4430db9ff7c897e8ac92":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-01-16 04:12:07.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/900418/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/900418/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/900418/3"}}},"commit":{"parents":[{"commit":"36c7e57e5a360c03beccacab6386c3ad9889f159","subject":"Read global config file for cache utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/36c7e57e5a360c03beccacab6386c3ad9889f159"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-02 08:07:05.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-16 03:52:43.000000000","tz":0},"subject":"Move sqlite code in common module","message":"Move sqlite code in common module\n\nMoved sqlite connection code in new module so that we can use\nthe same in migration phase. Once sqlite driver is removed then\nwe can remove this common module as well.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: Id34f9ae7a2639023c8b6a7937487e7af5c2dccad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5a4d46d25aaafb913a8e4430db9ff7c897e8ac92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5a4d46d25aaafb913a8e4430db9ff7c897e8ac92"}]},"branch":"refs/heads/master"},"4c3271e21407be403da50cc379c46329a54ef999":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-01-17 09:26:42.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/900418/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/900418/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/900418/4"}}},"commit":{"parents":[{"commit":"e2e9c0d5241492f4395e066f73253bd9f94c569e","subject":"Read global config file for cache utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2e9c0d5241492f4395e066f73253bd9f94c569e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-02 08:07:05.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-17 09:05:56.000000000","tz":0},"subject":"Move sqlite code in common module","message":"Move sqlite code in common module\n\nMoved sqlite connection code in new module so that we can use\nthe same in migration phase. Once sqlite driver is removed then\nwe can remove this common module as well.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: Id34f9ae7a2639023c8b6a7937487e7af5c2dccad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4c3271e21407be403da50cc379c46329a54ef999"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4c3271e21407be403da50cc379c46329a54ef999"}]},"branch":"refs/heads/master"},"d1de77b7168ba8ec5a2fa0cdb4b80915e907fe64":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-01-17 18:28:54.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/900418/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/900418/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/900418/5"}}},"commit":{"parents":[{"commit":"b8bb8f350258d2320de1f5d77cd73743670ad7fc","subject":"Read global config file for cache utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8bb8f350258d2320de1f5d77cd73743670ad7fc"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-02 08:07:05.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-17 18:06:41.000000000","tz":0},"subject":"Move sqlite code in common module","message":"Move sqlite code in common module\n\nMoved sqlite connection code in new module so that we can use\nthe same in migration phase. Once sqlite driver is removed then\nwe can remove this common module as well.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: Id34f9ae7a2639023c8b6a7937487e7af5c2dccad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1de77b7168ba8ec5a2fa0cdb4b80915e907fe64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1de77b7168ba8ec5a2fa0cdb4b80915e907fe64"}]},"branch":"refs/heads/master"},"4542d95045c83f7cfff5c505b5c313c1fe773619":{"kind":"NO_CHANGE","_number":6,"created":"2024-01-17 18:37:18.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/900418/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/900418/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/900418/6"}}},"commit":{"parents":[{"commit":"eb3f78b91b2b2450431adf34cf1a0aabe7d86817","subject":"Read global config file for cache utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb3f78b91b2b2450431adf34cf1a0aabe7d86817"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-02 08:07:05.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-17 18:18:10.000000000","tz":0},"subject":"Move sqlite code in common module","message":"Move sqlite code in common module\n\nMoved sqlite connection code in new module so that we can use\nthe same in migration phase. Once sqlite driver is removed then\nwe can remove this common module as well.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: Id34f9ae7a2639023c8b6a7937487e7af5c2dccad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4542d95045c83f7cfff5c505b5c313c1fe773619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4542d95045c83f7cfff5c505b5c313c1fe773619"}]},"branch":"refs/heads/master"},"e42c532998da66b87a52813daf7e3589d6fd5df4":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-01-18 05:59:22.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/900418/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/900418/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/900418/7"}}},"commit":{"parents":[{"commit":"1681e3406fd867ead99c44ce73ef68a236cec230","subject":"Read global config file for cache utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1681e3406fd867ead99c44ce73ef68a236cec230"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-02 08:07:05.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-18 05:38:38.000000000","tz":0},"subject":"Move sqlite code in common module","message":"Move sqlite code in common module\n\nMoved sqlite connection code in new module so that we can use\nthe same in migration phase. Once sqlite driver is removed then\nwe can remove this common module as well.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: Id34f9ae7a2639023c8b6a7937487e7af5c2dccad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e42c532998da66b87a52813daf7e3589d6fd5df4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e42c532998da66b87a52813daf7e3589d6fd5df4"}]},"branch":"refs/heads/master"},"de93e8b3c444e6692d612492b6ac80fbb3442631":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-01-19 10:17:56.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/900418/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/900418/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/900418/8"}}},"commit":{"parents":[{"commit":"e350172f7069b162186cd1b760464765b4ad9b5b","subject":"Read global config file for cache utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e350172f7069b162186cd1b760464765b4ad9b5b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-02 08:07:05.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-19 09:54:07.000000000","tz":0},"subject":"Move sqlite code in common module","message":"Move sqlite code in common module\n\nMoved sqlite connection code in new module so that we can use\nthe same in migration phase. Once sqlite driver is removed then\nwe can remove this common module as well.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: Id34f9ae7a2639023c8b6a7937487e7af5c2dccad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de93e8b3c444e6692d612492b6ac80fbb3442631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de93e8b3c444e6692d612492b6ac80fbb3442631"}]},"branch":"refs/heads/master"},"199859d9647ca9748deb54d6b8c7cf2004408096":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-01-24 14:41:03.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/900418/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/900418/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/900418/9"}}},"commit":{"parents":[{"commit":"a0640e18fecb0e5d8293ca3e6af26323416186a0","subject":"Read global config file for cache utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a0640e18fecb0e5d8293ca3e6af26323416186a0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-02 08:07:05.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-24 07:33:29.000000000","tz":0},"subject":"Move sqlite code in common module","message":"Move sqlite code in common module\n\nMoved sqlite connection code in new module so that we can use\nthe same in migration phase. Once sqlite driver is removed then\nwe can remove this common module as well.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: Id34f9ae7a2639023c8b6a7937487e7af5c2dccad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/199859d9647ca9748deb54d6b8c7cf2004408096"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/199859d9647ca9748deb54d6b8c7cf2004408096"}]},"branch":"refs/heads/master"},"8f5f9dcee4ee0458d60cdb0b35875cc274ea71bf":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-02-09 10:05:42.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/900418/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/900418/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/900418/10"}}},"commit":{"parents":[{"commit":"2554a17323de076416284d09c10a9a85d3b3c296","subject":"Read global config file for cache utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2554a17323de076416284d09c10a9a85d3b3c296"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-02 08:07:05.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-09 09:31:00.000000000","tz":0},"subject":"Move sqlite code in common module","message":"Move sqlite code in common module\n\nMoved sqlite connection code in new module so that we can use\nthe same in migration phase. Once sqlite driver is removed then\nwe can remove this common module as well.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: Id34f9ae7a2639023c8b6a7937487e7af5c2dccad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8f5f9dcee4ee0458d60cdb0b35875cc274ea71bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8f5f9dcee4ee0458d60cdb0b35875cc274ea71bf"}]},"branch":"refs/heads/master"},"b84991d048e0872335200f297f53349f820a31bf":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2024-02-09 19:54:58.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/900418/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/900418/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/900418/11"}}},"commit":{"parents":[{"commit":"8508e43088ea4cc1782920b147774bc5022f1c0e","subject":"Read global config file for cache utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8508e43088ea4cc1782920b147774bc5022f1c0e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-02 08:07:05.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-09 19:06:52.000000000","tz":0},"subject":"Move sqlite code in common module","message":"Move sqlite code in common module\n\nMoved sqlite connection code in new module so that we can use\nthe same in migration phase. Once sqlite driver is removed then\nwe can remove this common module as well.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: Id34f9ae7a2639023c8b6a7937487e7af5c2dccad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b84991d048e0872335200f297f53349f820a31bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b84991d048e0872335200f297f53349f820a31bf"}]},"branch":"refs/heads/master"},"d6f33f20d63661264a6b2b2a3608b37fc64b1caa":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2024-02-13 18:28:54.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/900418/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/900418/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/900418/12"}}},"commit":{"parents":[{"commit":"43ba1b7011391d9550801dd5c3620b721c938835","subject":"Read global config file for cache utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43ba1b7011391d9550801dd5c3620b721c938835"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-02 08:07:05.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-13 17:29:29.000000000","tz":0},"subject":"Move sqlite code in common module","message":"Move sqlite code in common module\n\nMoved sqlite connection code in new module so that we can use\nthe same in migration phase. Once sqlite driver is removed then\nwe can remove this common module as well.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: Id34f9ae7a2639023c8b6a7937487e7af5c2dccad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6f33f20d63661264a6b2b2a3608b37fc64b1caa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6f33f20d63661264a6b2b2a3608b37fc64b1caa"}]},"branch":"refs/heads/master"},"5293fb207729c89b48e1d175adab7a3f08208074":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2024-02-14 05:35:50.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/900418/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/900418/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/900418/13"}}},"commit":{"parents":[{"commit":"1fa50957a063b9e20485f93db0388f580a689a53","subject":"Read global config file for cache utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1fa50957a063b9e20485f93db0388f580a689a53"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-02 08:07:05.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-14 04:56:38.000000000","tz":0},"subject":"Move sqlite code in common module","message":"Move sqlite code in common module\n\nMoved sqlite connection code in new module so that we can use\nthe same in migration phase. Once sqlite driver is removed then\nwe can remove this common module as well.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: Id34f9ae7a2639023c8b6a7937487e7af5c2dccad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5293fb207729c89b48e1d175adab7a3f08208074"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5293fb207729c89b48e1d175adab7a3f08208074"}]},"branch":"refs/heads/master"},"7a6a289d3f0b2d1d5592899bc1b6daf884afd9ff":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2024-02-15 17:46:06.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/900418/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/900418/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/900418/14"}}},"commit":{"parents":[{"commit":"0ac629b32cdd341e88bb480ecf66f7a4870f07f9","subject":"Read global config file for cache utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0ac629b32cdd341e88bb480ecf66f7a4870f07f9"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-02 08:07:05.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-15 16:58:37.000000000","tz":0},"subject":"Move sqlite code in common module","message":"Move sqlite code in common module\n\nMoved sqlite connection code in new module so that we can use\nthe same in migration phase. Once sqlite driver is removed then\nwe can remove this common module as well.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: Id34f9ae7a2639023c8b6a7937487e7af5c2dccad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a6a289d3f0b2d1d5592899bc1b6daf884afd9ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a6a289d3f0b2d1d5592899bc1b6daf884afd9ff"}]},"branch":"refs/heads/master"},"0168b1a0c0ff51470b0d9071e18f384896b9437f":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2024-02-16 17:12:14.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/900418/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/900418/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/900418/15"}}},"commit":{"parents":[{"commit":"322dc1e92d299e4ff4b95a0838ffc026aa1d7446","subject":"Read global config file for cache utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/322dc1e92d299e4ff4b95a0838ffc026aa1d7446"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-02 08:07:05.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-16 16:41:57.000000000","tz":0},"subject":"Move sqlite code in common module","message":"Move sqlite code in common module\n\nMoved sqlite connection code in new module so that we can use\nthe same in migration phase. Once sqlite driver is removed then\nwe can remove this common module as well.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: Id34f9ae7a2639023c8b6a7937487e7af5c2dccad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0168b1a0c0ff51470b0d9071e18f384896b9437f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0168b1a0c0ff51470b0d9071e18f384896b9437f"}]},"branch":"refs/heads/master"},"21333c1ac85e52e31a29db2529abf3dfdd04e9a0":{"kind":"NO_CHANGE","_number":16,"created":"2024-02-20 20:30:08.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/900418/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/900418/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/900418/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/900418/16"}}},"commit":{"parents":[{"commit":"9c7820740a5ceaf31336da39624226b9e398e84d","subject":"Read global config file for cache utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c7820740a5ceaf31336da39624226b9e398e84d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-02 08:07:05.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-20 19:52:39.000000000","tz":0},"subject":"Move sqlite code in common module","message":"Move sqlite code in common module\n\nMoved sqlite connection code in new module so that we can use\nthe same in migration phase. Once sqlite driver is removed then\nwe can remove this common module as well.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: Id34f9ae7a2639023c8b6a7937487e7af5c2dccad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/21333c1ac85e52e31a29db2529abf3dfdd04e9a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/21333c1ac85e52e31a29db2529abf3dfdd04e9a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
