)]}'
{"id":"openstack%2Fglance~900419","triplet_id":"openstack%2Fglance~master~I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd","project":"openstack/glance","branch":"master","topic":"centralized-cache-db","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-02-28 16:33:32.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-02-27 18:53:41.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-02-28 14:22:03.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd","subject":"Migrate from SQLite to Centralized db","status":"MERGED","created":"2023-11-08 14:37:54.000000000","updated":"2024-02-28 16:35:11.000000000","submitted":"2024-02-28 16:33:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":44,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900419-centralized-cache-db","meta_rev_id":"6c784061ad4e5a09465f50359b3a025bb1a60cbf","_number":900419,"virtual_id_number":900419,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-28 16:33:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2024-02-28 07:11:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2024-02-27 18:53:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2024-02-28 14:22:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2024-02-28 14:22:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-08 14:50:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-08 17:01:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-20 15:13:22.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2024-02-27 15:22:03.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-02-28 07:11:04.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"}],"messages":[{"id":"23f21d5540240289fa6749988f2d50f78ac0e574","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-08 14:37:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2324c0ae296a68a3c2e8e0362099c1e4645813aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 14:50:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/105763cd51f1499696237375bc5805d5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f6ab6a2431e4dc3abc79a91a14c619e : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/26da0650f339457b8ddedb4bc1d45403 : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/98b7b361df4243179f6c03e9795a0c60 : SUCCESS in 5m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a0ba2fdb0c2fdab5efacb8af9e7ef7b3a52e87a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 17:01:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dfc5478a3ac24d8bb1b77bc351716f6d\n\n- grenade https://zuul.opendev.org/t/openstack/build/0c0a3d01564744938deb1728c6a15121 : SUCCESS in 1h 16m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/88ec673ba87a4123badd86ce0c15f5a9 : SUCCESS in 1h 02m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/57b07f84b2984451a0a101924d759515 : SUCCESS in 1h 32m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dcaf35668d7e423cb52b534a22101d9e : SUCCESS in 54m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55a98e596d3943b296a37f098d495f92 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6364510f3e3b46c484e29c5b62df0c8b : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd92f31feffb4b868d89b73b4949483d : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e6a3d490ead409ba2e7654128ea19f0 : SUCCESS in 7m 01s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/951da78353ea4fa6ad9e51468c9a94ef : SUCCESS in 14m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cc3ece0499ca4e4687492bdbc5d04ff0 : SUCCESS in 8m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bd126561b3554fe199081b156666c268 : SUCCESS in 8m 02s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c3cb8fce68b940ab987589552f9c693a : SUCCESS in 8m 04s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/80c9f3214f9146a3abf5e963948592dd : FAILURE in 7m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/46b915f889d045bab4b75a35b85d4459 : TIMED_OUT in 2h 03m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b55d7d1ca6544fef95d59571eaee44ee : SUCCESS in 1h 47m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/db8f0a95db184037a970c2cdf0c798e9 : SUCCESS in 1h 40m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f9327ccea63f4a8cb3062f9072c1a045 : SUCCESS in 1h 30m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dd5931fa4d4e4387b8d0ebef923a8e61 : TIMED_OUT in 2h 04m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d1e1d4684a243f6a1382ecdbb9e324d : SUCCESS in 1h 01m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d3649eea83ab403087ebfffba41ac51a : SUCCESS in 1h 50m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8f836a7e1ccd45619baeca174655af14 : SUCCESS in 32m 30s","accounts_in_message":[],"_revision_number":1},{"id":"30a1fe8326a9623293090208ee238aad89f4d620","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-09 15:22:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d61a2677f3fc0b9ca1f198933260448016ad57f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 15:43:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23f24a9279aa460f9ea0e474488c4595\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4459e485be59464c860aab1e5284903a : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eb6abf37877c4835830c9f38772f9c31 : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4fbd3c2f3f7d45cfa6f063c80b09c040 : SUCCESS in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4a71a46f20150723f5221ed09d40bade4bd9808","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:43:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f3d09280394489a89369b45410d1169\n\n- grenade https://zuul.opendev.org/t/openstack/build/a5cf6337afdd451eaece0ff429ba811f : SUCCESS in 1h 04m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3eea72bf78164b6db18ef077e42ef591 : SUCCESS in 52m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a7641a218f5d4c26a24d5b4de97afa7f : SUCCESS in 1h 35m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dcc969a0093c44749d832acbf896a616 : SUCCESS in 58m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69d02c3f79c24442b4d6fad71d74a92f : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/605834137c0d4cf19add8a22c3ef02bc : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c842f00928d147a0af128306c5e45653 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09f7f16055ea40d7aa51ce4c14de25d3 : SUCCESS in 5m 44s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/825d0adc7f1245efac38ccab6dcdf591 : SUCCESS in 13m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ad128a38b78a4689ba3f7a42e0d3d2d7 : SUCCESS in 8m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/911776c2219f47e1b323230a3894912b : SUCCESS in 8m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4cd739345d0644349b23b547955b7a76 : SUCCESS in 7m 59s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/1c525d281be2498d8bcee3ddbccd543e : FAILURE in 7m 21s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4915cfdf4f314c109df93148d54996e7 : SUCCESS in 1h 44m 47s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/67971863869c4802ad3de14be26c883d : SUCCESS in 1h 29m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f2778907d3834d3a8181e271df06fde3 : SUCCESS in 1h 45m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3a9e192acc85406d8c59fcce09c45525 : SUCCESS in 1h 50m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d16700f539e846bbab37416dead66a62 : SUCCESS in 2h 03m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88769b2a27c9493a8bedb0a78a0703de : SUCCESS in 43m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2d1018b7aea04ddb8fab76e6d8fdedbb : SUCCESS in 1h 54m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/24ec25bbb02741888015b8bb4a00500e : SUCCESS in 38m 40s","accounts_in_message":[],"_revision_number":1},{"id":"b352cffbfceb4641809b8463e713e3ab37f3cf4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-20 07:51:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8bc3ac52ac274f28149986c03f10fe1709abcee3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 08:03:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68ade87a387643a0ac5bea7f03183f29\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7a650a24bf3498591d57f4d87a54102 : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/843f616202e9464081a991fe5e2578a7 : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/999b26a79a4c42519ba74aad506fd1c5 : SUCCESS in 8m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"97133891f23efafbee0ecede6ffe73aa0c59bad0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 10:03:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/396adabdbede45d0acc88325293c3161\n\n- grenade https://zuul.opendev.org/t/openstack/build/140ec18484a248b1a81a7c5de1fa400a : SUCCESS in 51m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f9a2569438bd421cbe17c6e21aadf99a : SUCCESS in 52m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2ec399f5cb474a6b96694eb51cad0040 : FAILURE in 1h 24m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a319c20a54424feb9f1a0d2664760a75 : SUCCESS in 55m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8772a80ef7164500b1cc5f1bc4be2ea7 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06b9d8a34af7416e9db88208b0cd67c2 : SUCCESS in 5m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/12681d8d22f14ee095b4b26507ad1c21 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b387cbfa05c44b8a564189a770a9740 : SUCCESS in 6m 30s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b00ee8ebcf084d10949a9e6736bb21d6 : SUCCESS in 15m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/137251b0f510457d934751294c07caca : SUCCESS in 9m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/14aa024dfbe648d8902504c14c84011e : SUCCESS in 8m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/be58af0376134a0eb86d85effe04179e : SUCCESS in 8m 30s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/3091d19422a6471791d2bea6a05d8910 : SUCCESS in 10m 18s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f34ac2edb4d644a1b74375d5ecca5f49 : SUCCESS in 1h 55m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2db8c634223d47bfb088e3f3d75c2f45 : SUCCESS in 1h 22m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/01847973ce5a4e2587b7ce0dd2eeeeb9 : SUCCESS in 1h 34m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/51beb71d9e794d0891984a175b79aef4 : SUCCESS in 1h 42m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2262bddfca694896a32fb9d86bc02d98 : TIMED_OUT in 2h 03m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4d951f7d7c64de381af9c0a3eaf4adb : SUCCESS in 43m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fb5f106403e94ad9b345c2fe4d5331fa : SUCCESS in 1h 36m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/84dd62fc5d264f9b8c996001e1767e39 : SUCCESS in 24m 40s","accounts_in_message":[],"_revision_number":2},{"id":"de64cb6d28b82a622f3a2b177d911a61f4688e69","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-21 07:55:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ff1de0d2ade6d55ed1f065ce4499a417ac22b21b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 08:08:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ede24b55e3214e16bb8cd8307091b5f1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c2533bfb7714d93a9a6d8c3c458dd1b : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fbe2a184055443cd8f65d2917a240f9e : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cbeb75632d434482b41c5a94bcc3034b : SUCCESS in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f5b3d7c9497309108a4ccaf161ddf3c05d9b34d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 10:08:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f98b996722c24dc28c8aaf28bf9943f3\n\n- grenade https://zuul.opendev.org/t/openstack/build/c5f2c304dd0548f293213103c1aa46ea : SUCCESS in 1h 02m 55s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e920112a13824eadbbd1eb7f592f2c0f : SUCCESS in 1h 01m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0c0749016ba74e7eb07e29cade038b30 : SUCCESS in 1h 26m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e08af91dfa8406f92998fd47dd5881a : SUCCESS in 1h 03m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7239560d60d34843a883b03d51cdf1f7 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d8f0c93a3404cf09b746cf9ee90ab7e : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4ca46b62f310443bb85ff2ff4fc729be : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c308a197b4c4c038784ee558db1fd30 : SUCCESS in 6m 27s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/277d224b60f04c2a960ab4bede1a176e : SUCCESS in 13m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fbef89792bcf4251823e1d48ce40fad9 : SUCCESS in 10m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6c3dcefb3d3246cb8fd127d56dc32ea5 : SUCCESS in 9m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ed03634a516542a5aab43f704f8d2e0c : SUCCESS in 8m 15s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/8ba9d98553f94202b22f2c951e13fbc7 : SUCCESS in 11m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a570e7ca257b4dc39a20d17e315ec26f : SUCCESS in 1h 46m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3ee68d5f91fe4a1eb5cb1aacdc23fe4e : SUCCESS in 1h 57m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6a7c219c2e9e49bb9fda0156946528ce : SUCCESS in 1h 35m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3cc047e16a134b78b05dfd057cb15b29 : SUCCESS in 1h 55m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1e55c14b9424490dbebfda00a22b9d36 : TIMED_OUT in 2h 04m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43a04300caf645639136005b59662124 : SUCCESS in 53m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4617d86983254c1c806b5f6ba0bf0b2c : SUCCESS in 1h 41m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/882e538314e347bfa46cb1c46314d8eb : SUCCESS in 30m 00s","accounts_in_message":[],"_revision_number":2},{"id":"81836a1eb31d9a880db6ab3cd8bc10b9e83243d7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-22 12:38:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"37930a41f4345f583dd1f19760927b36c42f79d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 12:44:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bd295dfb6b04c68914d86b80440cece\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/596b9de0275c41de8437748c1cf164e3 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e53ea77ea74c48699c8c5ad4f455b9fd : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9004bd07394e4348828af72b83e8248d : SUCCESS in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"690fcd6a25fefb160dec2cccfd5a0df538bd0f95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 14:42:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2df32982b00a40b0a17cf9c1d1c4940d\n\n- grenade https://zuul.opendev.org/t/openstack/build/ee3b8357a89b4f3f9d2090f249ea0054 : SUCCESS in 56m 22s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/edadc05df9bd42da8a7908aa4e48e13d : SUCCESS in 47m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/42940e925bef400d82ada8083a804643 : SUCCESS in 1h 31m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01fbb077768546608644b632c8d632cf : SUCCESS in 51m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e68f702207fd4ffe908086345f9c0627 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfcebd0eed184afc84a40f6d24a6543e : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30cc9cee83a2475a81190c5883924fe9 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/914495c967694a4fb4727c92ca2afa20 : SUCCESS in 5m 44s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/bd28f1ad05d34e2ab5e568d8893597d5 : SUCCESS in 14m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/255deb4204f542cc99587506baa7147f : SUCCESS in 9m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1b666ea68eb0447f851a048250b1202b : SUCCESS in 8m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8a4c583fbb344bba924b328e514dbf50 : SUCCESS in 8m 55s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/2090d5a90db24a16a8ec9425e69e739c : SUCCESS in 10m 23s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/272641d502aa4e3eb52356e05809678c : SUCCESS in 1h 54m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b50ae8dfe54b4fa78057a3a054d49672 : SUCCESS in 1h 35m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/eabf4abef2a54f0e9c5b09fa24c7a373 : SUCCESS in 1h 41m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/adda38cd77614678bd33e40d2789efa1 : SUCCESS in 1h 19m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/79bfe6cbb6aa472b9441c2a1da91df2e : SUCCESS in 1h 43m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8f9f74616ec4894bf4c849a60fd90ae : SUCCESS in 1h 01m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8568e74b17ab4d22acd9a039e974c04a : SUCCESS in 2h 01m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/47ea23139d9c472f9dbcc7ea67c98d07 : SUCCESS in 41m 01s","accounts_in_message":[],"_revision_number":2},{"id":"f1b96eb9478c838ca367e6850bc4af11af119033","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-12-21 04:49:46.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9984395558e0df7ef0b788117670bf3c0999c295","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 05:02:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6791d16ae2b94b7aa6c6f830c3cc10b4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e3638d83ec741ed81d04f212ad2800d : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9c9df80134044fd081f23974cda59f77 : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a19c76f661194ca4a113bba009ea3f4f : SUCCESS in 6m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f10ef17fecae63322f05bf0c885d2a9738bc6f6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 06:59:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ebd3d44a5e44461ac4ab3edbe28c221\n\n- grenade https://zuul.opendev.org/t/openstack/build/cf84e35c9e9e4d5e8c0e87f135273be1 : SUCCESS in 1h 00m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/79a0557e7c5045a3a6e45633492550c2 : SUCCESS in 1h 13m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c8e430c3b00f4a23b95fb10e00466534 : SUCCESS in 1h 26m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5c379441cc84473a83deca940045680 : SUCCESS in 1h 12m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a893b768fdc4069a48392e9d74b5b47 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/540b995d8ec748f0a8b346c3084e2b54 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f19af7338d84f08aa347b8773e00352 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22084926597a4b09a3287ebcda39d652 : SUCCESS in 7m 02s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e0d134e9b1d44d6f8a33f52379eede27 : SUCCESS in 13m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ec5a36464fc44991bbd8bbf6d32afe35 : SUCCESS in 10m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b3fcd5049f1d403597b20ca0c495fc3c : SUCCESS in 8m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1b2dd3ce27bc4a78abf0774e568fbc24 : FAILURE in 10m 19s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f4939a1d779e4a11bec08e2d523652ab : SUCCESS in 11m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ba2326ddca484d25a2ab4c6c8863ad0e : SUCCESS in 1h 49m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d9b91d3e82644463b26bab4cd50f7b0e : SUCCESS in 1h 39m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/70b8433a5894467c95f4f17b228cb563 : SUCCESS in 1h 25m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a8cc7441cb7143a1b9f6aecdedbd83a9 : SUCCESS in 1h 43m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/69d08ca127914c73a7da53396c7b0644 : SUCCESS in 1h 52m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3bf027225cf446a816f22cd96542ff6 : SUCCESS in 52m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2bd03c3f4fc84120995a380a3ea927c2 : SUCCESS in 1h 47m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5c8cdcd2cd3c425587a82d513b540aaa : SUCCESS in 41m 28s","accounts_in_message":[],"_revision_number":3},{"id":"78efb2c2ab29ecc864a545520b6e431c2862163b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-16 04:12:07.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"8e7ce007c4919248a8000aae322351829affc5ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 04:23:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1785582ab4c741dc8b1afaca4d2f24f8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c3bdbad7ce44742b03376a910d3369a : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/155863e923d5460aadd6d478a3fb7d03 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b6a9a7344d84424dbb606b2744aeae07 : SUCCESS in 8m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"edba4be68a2a62f50a8eac3722415fc4b8732463","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 06:33:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2cb62cab77e34e1cae6de7ac9995f281\n\n- grenade https://zuul.opendev.org/t/openstack/build/53bb1cd171ff480c9a1e0211c884aa07 : SUCCESS in 1h 04m 24s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/969ce02574844bb0a8936a7d6b3c7374 : SUCCESS in 1h 07m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/643163bc920d4922a4c9ef71e0da189e : SUCCESS in 56m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82d0c19bce25426b9a0ae9814bc62028 : SUCCESS in 29m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3304a703d354404190c766107cc66463 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5aecf23d40f490eaf9f753c40019f1a : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a005d17f3404b998c88018d37e9e2c2 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0146a14da0784dbca2cd959aa8afa72a : SUCCESS in 6m 32s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9be1ebbabb9940268696cee8fe36ac85 : SUCCESS in 12m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b42052f8b318489ab9a24d8432f19861 : SUCCESS in 9m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b9247fbdada34ed1bd7182bfa5e26eb4 : SUCCESS in 7m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/827889f30aef4975ae081431e2e3be20 : SUCCESS in 9m 18s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/73ab2eea94f543199cb3597eb9d18c5c : SUCCESS in 11m 23s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3d0d4f42986a4677967385a37b9a9107 : SUCCESS in 1h 51m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/55e62b5fadc54bd59ad91aaa78080987 : SUCCESS in 1h 09m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/898f614769ba4a7aa1b6819a7b5801f8 : SUCCESS in 1h 36m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/53c8debe9f754d0aa4cf98a1131300ee : SUCCESS in 1h 32m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f0c39fa6ccfa437e8a9eb26d496039ff : FAILURE in 2h 11m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/719592e139ba47e2b853452bafc3e4fb : SUCCESS in 1h 07m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/46eb763b0ea44013b54052edab451b72 : FAILURE in 1h 16m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/55b27e1f40744d8ca6b8a949e2e2abb2 : SUCCESS in 27m 13s","accounts_in_message":[],"_revision_number":4},{"id":"7c98545773701b4277592281a3d3ef9863e8ca07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-17 09:26:42.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2ed54c838e7bb8903591e344b30cedf22174bda3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 09:41:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2060ea594114f539395ecf4c88a159c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0bfcabf768a44735beb3709e0a413cf0 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b7e6f58297114cafb977b78da0c4b260 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ee960067f2044ad8adc89a82ff14b91b : SUCCESS in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e5074da8ce55bbf2b986eb11245367ced2cd3ed2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 11:52:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a8b9ae281d948ed80f3bd7325021e30\n\n- grenade https://zuul.opendev.org/t/openstack/build/5fc5239d2091483ea6f71b4256b630d5 : SUCCESS in 1h 04m 24s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ddcdd177b2bd41b3989226ab2ab85aea : SUCCESS in 58m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fed5a1ecda4a4b9aac5c188f7eeff37e : SUCCESS in 1h 25m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8f722a0208a4b4e8d6eaeb825b206eb : SUCCESS in 29m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ade5bc396af48bb99d74a44dd066ac5 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c84c5addd4a4244af90e19baa64f8d6 : SUCCESS in 5m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1943b87a8e5e4a2c9602bc71f30cb7a4 : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bf0fdfbd3af4e1b94ff2329ce84c534 : SUCCESS in 6m 28s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0a1d849a1b244167922d97cc2bf91f03 : FAILURE in 13m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9da8cd77163944a6956c5af9fa49ba2d : FAILURE in 8m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cbc0cc01fb5e4f148cb9969f246c45a7 : FAILURE in 8m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b5b6651c664a4654a4e0e2dae4736b11 : FAILURE in 8m 07s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0225b5ac2d684ad3add8e8194b621b14 : FAILURE in 11m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c681434c5dd04a41b8ad7e41214e8f19 : TIMED_OUT in 2h 04m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/268619e0cacb4cf78906e65db5676185 : SUCCESS in 1h 54m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ce4b8a75da4c45ccb1d512ad675ea4a7 : SUCCESS in 1h 32m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d24a9f7e1132458d9c31a7a14fae70c0 : SUCCESS in 1h 48m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b6fd2a6635b0481a9f5c17fd67b0be24 : SUCCESS in 1h 50m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da92aba27c514744a18fb6f62292bd72 : SUCCESS in 1h 03m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8836f15ee96d4c229475e570d9cc0690 : SUCCESS in 2h 08m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d84dab3642ba421b81c250014c3bd2fa : SUCCESS in 24m 24s","accounts_in_message":[],"_revision_number":5},{"id":"0ed49b7424c2191987413f2534d1aad6eb87abd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-17 18:28:54.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"864bb6c8f1deb16bf2e7e96c223bd4627207342f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-17 18:37:18.000000000","message":"Uploaded patch set 7: New patch set was added with same tree, parent tree, and commit message as Patch Set 6.","accounts_in_message":[],"_revision_number":7},{"id":"74fb01dbd706d2475d9843dcd575f9f171138dd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 18:50:34.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4ee8acb825646daab093c494f7c4cde\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1af1226a1a354930bdf9d9984f6741cf : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2c7534de37b740738ac30f5b94d3ee40 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2d53224fc7c44ed8b635adc3ff83fc8f : SUCCESS in 6m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"df06552f51dcd88bcffcb5868f0d892fcd615ff1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 20:48:36.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/02f91e1e24614695ae2d67eb5085e3ed\n\n- grenade https://zuul.opendev.org/t/openstack/build/5f2b1bab1d4144d7afc3222c229e7d3d : SUCCESS in 50m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a3659ca03d4e4fd7974fc3ccda551f2d : SUCCESS in 48m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b31d10fb2bfb4517882247efdfdaf77c : SUCCESS in 1h 24m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f9bbd9b10da489db7c422da54d07d8d : SUCCESS in 28m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acc4453748f44cb9be6c89250f1cdbc5 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80133e60cad54eebb10062eea7d6c40c : SUCCESS in 5m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/201dc8ab9a6343dd8b2b677a6d024616 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36c759e91c5d49d4b0f0a148a28542c6 : SUCCESS in 6m 21s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f160d2ae4aaf4930aec3c8f9b6017c64 : SUCCESS in 13m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/692db23afae94b7a9502b47ecc7d8bf8 : SUCCESS in 8m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5daf4eed068048c6b056b69f76296826 : SUCCESS in 10m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ddac42550e1348f9b7146753b00cee5d : SUCCESS in 7m 41s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6128714cbf5c4a828ebd0fdc862b595a : FAILURE in 11m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/08ea5e364b4842fbb87304894a8d90ad : SUCCESS in 2h 02m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e8d8969b751540abadd9320c1d820ef5 : SUCCESS in 1h 47m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/87bc04373ed04102818c3cb224a6389c : FAILURE in 1h 29m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/738ab8279d184500a9e10fca6d934043 : SUCCESS in 1h 47m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a31da84882254cc8a686b570947c2bb1 : FAILURE in 2h 00m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0dca003ee50c479192d45375609ea8ea : SUCCESS in 56m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9946df2aab5743f3a086dbd1c0c21708 : SUCCESS in 1h 51m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b88daf07b27a4ad4ad3d174f4ee5dcb7 : SUCCESS in 29m 23s","accounts_in_message":[],"_revision_number":7},{"id":"d86152087c70bee3253375f63e0a65d0e2a6cd95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-18 05:59:22.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"5d7c3a4b7183235b7fcfe00187d32eae439e4cbc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 06:09:09.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/179948f5e9e54187b4c2cee85ab79a5c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14cca8c55198454888b45bfedb6f7a10 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7b5f301302b244b0ad1e60f1534f5740 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f5a936737b6f451c80dc773117c4454e : SUCCESS in 6m 58s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3d9d20fd3d99fc7f63bd49c6a449b8fed76d8cc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 08:19:40.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c2b8e5341b04c5eaaf46d4d9465379a\n\n- grenade https://zuul.opendev.org/t/openstack/build/b037c75865bb48fb8ca4ac62eccc40c8 : SUCCESS in 1h 06m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2351f72409f843e897fe337be1e297ab : SUCCESS in 1h 21m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e6698efb03b84ea5872a891b65378321 : SUCCESS in 1h 21m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/353f19fbfdc84bfd8961a371c87c2825 : SUCCESS in 23m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b86552e40a0d4428b89ee660afd646f0 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7db337f6982413891134e8995d3a087 : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f911c7bc6fea4c6b907d233ceb1907c1 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0578acfe2fea45d292332bc5084d1ff6 : SUCCESS in 6m 32s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b46a9476c2e6450a8f7c3a6d78302744 : SUCCESS in 13m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4a5d595876764898942bf5b1969acbac : SUCCESS in 8m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/43c3baf123394a5cbbef2ce340dae528 : SUCCESS in 8m 11s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1508a9aeeb7f4c4ab3b1a8f1e22fabfb : SUCCESS in 9m 16s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b9238dd8aeb945b394a9ecbdc0756154 : SUCCESS in 10m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/062eb2fa43624be89188f13343edefc3 : SUCCESS in 1h 25m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cc4ee441bd54463baaf5998a22afa738 : SUCCESS in 1h 50m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/749191478cb74f66a07c622c36e6fd03 : SUCCESS in 1h 54m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f68579bfe49045f8963c11c16bf0f197 : SUCCESS in 1h 43m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/acd223b49d7a42ac9ca263517aa9e438 : SUCCESS in 1h 29m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d05184a720034d03be83585a8bced291 : SUCCESS in 38m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/38cec2a8772742fb9bb29e89baf73e76 : SUCCESS in 1h 49m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/beac8135c8684e58aa86714222570cdf : SUCCESS in 46m 01s","accounts_in_message":[],"_revision_number":8},{"id":"c829a3583c96e03c728cba153fb047593068990b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-19 10:17:56.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"7f258ed18e11e9b3506df5be8bbc5046fab2667f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 10:55:30.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e555ccb45b34f7eb758ce6f76ee97be\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ac5480c416c4d40b68504fc9f185889 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ec61b5128290437585eecd69b5a531ef : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f0b45e4db1b54c4b87a9b67c1e8a5c9c : SUCCESS in 8m 58s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1897b44e526f9c3d3833b1765d96d440d7975285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 13:20:15.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d91e8009ead4328b789214fc9567fa2\n\n- grenade https://zuul.opendev.org/t/openstack/build/2a47c8974313461ca91993c02c193b4e : SUCCESS in 1h 05m 50s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/66d5b68f6fa746b68bf33fa7ca9aad55 : SUCCESS in 1h 00m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9a2e751347704df3956d34cb2176471b : SUCCESS in 1h 35m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb47700e71e94bc497478108390135c9 : SUCCESS in 38m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f7ffb0295454feb970142033d27787b : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5313a7da1c9b470db72e562a6f6cb60b : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae7639988b5540688b264b5846fc96a0 : SUCCESS in 8m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd1ea4283c66494dbef5434f5553bc30 : SUCCESS in 7m 14s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2d836ddd13804155b73b31c767bdaae0 : FAILURE in 14m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ddbb85a6f6674eae9749ef428036640f : SUCCESS in 9m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c2a527786278413e8347585a252ac0d4 : SUCCESS in 11m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5d10dacaa2eb4e5eab9f27917684faef : SUCCESS in 11m 06s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f618d6823ca24c7f9294ce807e6002ad : SUCCESS in 10m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/04f8b4043c184fcd9e09737633cf018d : SUCCESS in 1h 50m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6dae9bd9a8c64a1b9714bbfb0a6d484e : SUCCESS in 1h 37m 43s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/11f29223b63f4fb28276cb02d74163c0 : SUCCESS in 1h 37m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d80f33b17d924470a93931c981f95d12 : SUCCESS in 1h 39m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c576b56fc85e415b9200ad72b9014c84 : FAILURE in 2h 32m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/817e64c65f384f5cab0c1d3c413dd4cf : SUCCESS in 1h 07m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fad086cae8964105bc6deb466258ac73 : SUCCESS in 2h 13m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8ea5a13123d945ab8603ff8592c0bffe : SUCCESS in 30m 21s","accounts_in_message":[],"_revision_number":9},{"id":"72c85cf5d949460e1d50c9e23489b5dcbcad5976","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-19 17:54:23.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9f8cf66b84cc2f362cba7626983e045ec0cef9ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 18:05:15.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/817b24a5cdb7411c8dd1ab4bbe52e044\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/47f92bb54e854a01810b0448f0be1814 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/69a7a2c2dd0e47c4943f94f289606544 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/13d735c012944a158ec68291df44a484 : SUCCESS in 8m 49s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d499c26c11bd2795c34fb655a536f1d8e709022c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 20:03:21.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcf4942955724d9e8e683029b4cdb7a0\n\n- grenade https://zuul.opendev.org/t/openstack/build/21c0542048ac4a58bba63583a11c0e58 : SUCCESS in 49m 33s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e864c37a5e594068b784c70adb954e1e : SUCCESS in 55m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dceec4d605b14daba03a9885eb0a9f3b : FAILURE in 57m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95587fadb5734dd5a40d8c329b2e7104 : SUCCESS in 29m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e3ac8fc87044978bc8be4b6e714ee2b : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99a89d1ab9364d86a616092c9712dc0b : SUCCESS in 6m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dced79ab41e74b1e8c2684840f76ed05 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46b8e57b723340a9a4de4d3a250dd58b : SUCCESS in 7m 49s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ea63252fabd04e9caf63d4794f464664 : SUCCESS in 13m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/001675fbf00a454abb47d3bc7700f344 : SUCCESS in 9m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3f29db3fb70447bd948331567e8cf483 : SUCCESS in 7m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fb80ac7b983e44dc9f704d8608f3c0ac : SUCCESS in 8m 00s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/3fe56df02cc4493a9d2804b3e1621146 : SUCCESS in 10m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/413a335c24a341ccbdceda992754a9da : TIMED_OUT in 2h 03m 00s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/36d286b51a2648038182440c8f198191 : SUCCESS in 1h 46m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c28b7efc104f4a75826c0512fdbbf44f : SUCCESS in 1h 47m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2018c5cba8ae4a94979870a222715f6f : SUCCESS in 1h 53m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9acc5cc95a7b4718abe716b93087638c : SUCCESS in 1h 52m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1afc3ba36d348d3b0d6a126fa9763f4 : SUCCESS in 1h 04m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/671ac4977b724a1299d8a56afaf9b49a : SUCCESS in 1h 50m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5ee17d79a4e2442d98f31715fb29c942 : SUCCESS in 39m 01s","accounts_in_message":[],"_revision_number":9},{"id":"ced93b282bdc7afe356fcccda5dd61ce3a8e8144","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-24 14:41:03.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"4f9d6f5e93aae2c358676aff0581af979efd7ee2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-24 14:43:48.000000000","message":"Uploaded patch set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"9e8ac6dc96bcc7ddec3a3def01cd0ef5fcc0f5e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 14:50:32.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/613d97d67f6b4880a2351f5194e46345\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/62a39e35193641e8b41d335cccdf07f5 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7577d19d53514a79a2f50640ddd510fe : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cd7f8cb313ca4492980add417e79a721 : FAILURE in 5m 21s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a0b0c178098655965ec743a3c3b6b6f3ed648852","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 17:06:31.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/761cef351d6440048b7382da7c1d41b7\n\n- grenade https://zuul.opendev.org/t/openstack/build/14196cf3f49f471580f8011f663329f4 : SUCCESS in 1h 10m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/26540d359a524ee482166a801aef606f : SUCCESS in 57m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d488858982144cb1988a6f5fc0a9d7b0 : FAILURE in 1h 37m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5983a4c404224666b77b89ef86768868 : SUCCESS in 32m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a7925f075df4eb0b6cda73fced782dd : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c57a6fb97ae248259d6cdc1d3cd2d448 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e46f95b7638472982a924c949c049a2 : SUCCESS in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2943a84d3be411bb4319be4aaa8bbbf : SUCCESS in 6m 21s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/fee42e3242bd4153b822fd5149a8e8dd : SUCCESS in 14m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0ddd0222e5a542acbbce42dcf0bcf0b1 : SUCCESS in 8m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2a845c1213d644deb48d66d19f7a9f20 : SUCCESS in 9m 17s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2114b26a919c4047a5724dd9d75157a8 : SUCCESS in 8m 41s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0b8314da32ba4ce4b100b403357b003e : SUCCESS in 11m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9447b609a06b480482e0c68930e2a779 : TIMED_OUT in 2h 03m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/575fc1b4d7524fa082f75a55b032d99f : SUCCESS in 1h 53m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/adb9d5da741246a088bc9cca38e561f0 : SUCCESS in 1h 46m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6139df9170fe41a4baa223f139b8018b : SUCCESS in 1h 49m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9a89b4bd82f349dcb3f24db950bc60b1 : SUCCESS in 2h 19m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9f59bd6081d4c049a2b621b74af2f3b : SUCCESS in 1h 07m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ad4c5832ac85448cb2468ce8bcf23341 : SUCCESS in 2h 03m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d38960973df64d13b644b33213b2d3c6 : SUCCESS in 39m 15s","accounts_in_message":[],"_revision_number":11},{"id":"b9f3d1806adf0584fae25f294526b382ee796eef","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-25 05:24:10.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5499d8118388a9430225f2aa1fb83e8c4a26cbea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 05:32:28.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28e3e1f7396a46f8adab8d19b3b58170\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e6635dc6b8c41dea8ed7fce04b87b9c : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1326cc0e29374d5ea72ae862acd7bc5d : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7ee1e03febd74fd1b122058eec571d6e : SUCCESS in 4m 49s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a243a4160ae36ee59fe3f326705e76e4874a7c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 07:36:46.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a41579b4e0b4f12bb536802dd70ff4e\n\n- grenade https://zuul.opendev.org/t/openstack/build/4d37170e12a1497e8bb94d1bb6231d0d : SUCCESS in 1h 02m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/26fbb38b33414ab2b2034431407d2f96 : SUCCESS in 46m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/36045ace046540f98d19bfb07fad78b2 : SUCCESS in 1h 32m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9fb0ae03f217449787c5b18317026dfa : SUCCESS in 25m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdeff6bc3df240cf938ad3755392f5b9 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52aa23c1b7554a87b255114de7f9c7bc : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d23a2a2cbd2e4e1aa8023cef36e15a78 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e2b837aa337479abf853047f8312eaf : SUCCESS in 7m 15s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1e79191d5fbf4f55bad73d1489ad348c : SUCCESS in 14m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d4f5156fddb74330866a0edf50ab6fe9 : SUCCESS in 10m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f8c412ffe9b341719039dfc08b8a6224 : SUCCESS in 9m 41s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d0e30a5cbfa54478a0df3508558a453e : SUCCESS in 10m 11s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/2e3a1e03eb27404b88164096293b7763 : SUCCESS in 13m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/28278f36800d45e580046ad3a97e8bcf : SUCCESS in 2h 03m 03s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/84197fe40b094b839ad3ad0c8d566d95 : SUCCESS in 1h 51m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7f22798efa554011a2ae0e0c5d6592ee : SUCCESS in 1h 52m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/80a695740a47438c83d80004e01a14fe : SUCCESS in 1h 48m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c8441589548c465fa70e2f67e5c6ca09 : SUCCESS in 2h 07m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3af3eb7f8594411f8f2267669443cc41 : SUCCESS in 42m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/750dcc64a7a64624ba3846893ccc3ce0 : SUCCESS in 1h 51m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2b8769a4fa2a4cb983e8d3a796d10000 : SUCCESS in 29m 01s","accounts_in_message":[],"_revision_number":11},{"id":"dc0158079ef5528f6f61cd98691e7dfe9fb61644","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-31 21:23:20.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"6b4766062e946c2f1d8dc9cf53184945a2cebec9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 21:35:43.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82475f1d6e6a474dbfa86d1d1b8a9216\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/97189b9a88b64433af3092630991db16 : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/427fb3bd9f4e4bfbafc2fbe883fdd03f : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8f4218526a7a4631a2e35e53a5bb5204 : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"42bf5f4e2cfa3708f608b4805dd1f2cba18cb363","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 23:41:24.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/439b72d61fd542f4b443515f1ec9d319\n\n- grenade https://zuul.opendev.org/t/openstack/build/c1374234705242c582376caebdbe3d52 : SUCCESS in 1h 08m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f09f3d65a1884967b9a4e4115cf6f9d0 : SUCCESS in 36m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f7e80f212907431a81585faf4ea84afa : SUCCESS in 1h 17m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a4add7f31dd4aedb1c55b2471f1726c : SUCCESS in 26m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9812d10989ab4f769449d8befd9a0250 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1427189e8f72464d8ceaee396d51f1e8 : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb3c4e23ee624ce681155feef7f8ca59 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8f83040f89f425a803e30b973567564 : SUCCESS in 5m 08s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e26cd6ecac1647648e871be4d32e1b84 : SUCCESS in 14m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cc084e86c3734d20a63009d9bcd009ef : SUCCESS in 9m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ccc4ee9445424872a4c42d47d442830a : SUCCESS in 8m 05s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8ce2e1681ec247c096f4079845a23ade : SUCCESS in 8m 53s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/9f52d5267c0643e6a58ead7caf7a8cf3 : SUCCESS in 11m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7555c841b9874c4aacb91a5969a6849a : SUCCESS in 1h 53m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9a03ba9c7a244b38acd59b27dcc4b8fa : SUCCESS in 1h 52m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/411110141a5648c5b551c33356b24d30 : SUCCESS in 1h 33m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8e8e788bb1794481ad6cd456883dd938 : SUCCESS in 1h 42m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ea63b990d73a4d60b9414939aea7db47 : FAILURE in 2h 07m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c99e7327d51b4870943186671cfd9429 : SUCCESS in 1h 03m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c582e6f5c9a14b9da94dfd7fa76ea67d : SUCCESS in 1h 32m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c401c0e8a6854d2ebefca301711c5c12 : SUCCESS in 40m 30s","accounts_in_message":[],"_revision_number":12},{"id":"fbc01c3d7882b9f93a1190bb9caa147bb38a251c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-01 06:04:59.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"5e27bf889ce2e2e4311da9137b53e05d9682187f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 06:17:36.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aab40c46e4334838ac1e1ebc12884259\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d3fc3f9ae55c4a1181f83968a74590a4 : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/223939f5c3d14810908556715a135322 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/38a65d63cd994dacb1222f85e494c319 : SUCCESS in 10m 35s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d9142fb97bf2aa6bff95c2ed05e0efe936eab7fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 08:31:48.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bff65175856b495dac9bb4febd4d75ba\n\n- grenade https://zuul.opendev.org/t/openstack/build/4c9eb4403e9a4579828b3b392d241ac0 : SUCCESS in 47m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1acedd2942a8445592cace48a6ff96cd : SUCCESS in 1h 02m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b6df45bc24414fa2992d025a72d5dfa5 : SUCCESS in 1h 45m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/434eac366e6f4ec19aa71586eb30ca80 : SUCCESS in 37m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9b6f54c4bf8451b80b84ecb330b4fb1 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0918dd36d1ea4d5e84a8b6cca9cb175b : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f4bc83fb26941ccbe3ad98e3c38b5a9 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3393bcb511eb41c0941a569a57a642fb : SUCCESS in 5m 53s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/98c6656882264a2092318f457c45d909 : SUCCESS in 13m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4d24054839fd490eb9d78d4b2c46a208 : SUCCESS in 8m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3f5a4433d6e64bb8b08e2ebb073bd4cc : SUCCESS in 8m 32s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/79b9c14494544c2b8fc9f275f5ee471f : SUCCESS in 8m 53s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/30b3c35792d34f5284c1f99ab33aa5c3 : SUCCESS in 12m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dd5db577c77e462b817deed82d6729e8 : SUCCESS in 1h 52m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/573446f44cfc4b28a3f2d81ebfa60707 : SUCCESS in 1h 38m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/267b660411954c939c5538e2fe925821 : SUCCESS in 1h 25m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/edd42352c99941bd9b0ce37de0ca8080 : SUCCESS in 1h 57m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ba31c2429eb344e7ba8c615aa4d5626e : SUCCESS in 2h 06m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9496a41b0ea94607bdadaf8de57acd39 : SUCCESS in 1h 06m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f42e1cb564b447adb7865b0ae6746947 : SUCCESS in 2h 16m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8382c029d10f4ac7badbb4d8b595180d : SUCCESS in 31m 31s","accounts_in_message":[],"_revision_number":12},{"id":"1312e623443c3246af8f6a46e27c31c12d6a430e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-09 10:05:42.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"6b3711628b6b9504b8600d39344859772059b314","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 10:17:30.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b291ba02678347668bf8bb2258c16554\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bec95bb97e8e41a4951ed32e8d25427b : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9b177d7bf6074ea286283c7d0816d940 : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/01079c4c4347401dbad76c811694cf16 : SUCCESS in 6m 31s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b3c73a2c6ad1f7fa3686d3b15d9382c09c95857f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 12:08:47.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37eea407fbb943daa4487a646f16ce67\n\n- grenade https://zuul.opendev.org/t/openstack/build/a800fcc6368341ccb5d12252dca62446 : SUCCESS in 1h 00m 55s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ad200e53bea8476ca39a91e264a974b0 : SUCCESS in 59m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a240307dee64456b728b53c055f845d : SUCCESS in 1h 22m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/118dc68a952645a39637c3a0ede8794e : SUCCESS in 36m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abdcf539c22342b993aeec565c545a69 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dbca2c527f54c6fade7750523ce159e : SUCCESS in 6m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/623a0c6383d94a29a6725c50dd4b066c : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5d6408cdfc94b21b60e0a763c65e150 : SUCCESS in 6m 25s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/77d5b28b26794e67b3722d2b687d999b : SUCCESS in 14m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d6bb2f43b1ce4132a7cd0c5b83db1db4 : SUCCESS in 9m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ece2d8dcb0694df591ffb9b590596f1e : SUCCESS in 8m 27s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/17972f5e738949bda81d81550e3a17b0 : SUCCESS in 7m 56s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/e5e8ebe6613d4a4ca34f4e7b18c55aaf : SUCCESS in 10m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/219c4e2c5ce04c9ba6c3d10d8ec5e8ba : SUCCESS in 1h 49m 28s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/651b58ca291c4d819c02401a54aa2aee : SUCCESS in 1h 38m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c5d509cecfa14a30adc20d6d25567442 : FAILURE in 1h 49m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c9149dae9bc247fd98bc4c53c3740547 : SUCCESS in 1h 50m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9638dec851804748bb9005428c19a170 : SUCCESS in 1h 55m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b6aa134f7c24df19e26cb2e89e40fab : SUCCESS in 44m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/109d42a57a59409084eaa0f12f62b5a2 : SUCCESS in 1h 44m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/265c4989953a4a04b64b87bed9ea1356 : SUCCESS in 36m 38s","accounts_in_message":[],"_revision_number":13},{"id":"b329f010422eea827ab99478e85056f7f6eccef5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-09 19:54:58.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"2d8a1ad7688e7b3879b3df968cfa4cbe4a85bb32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 20:11:56.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54bffaf0ad70494580c9ec00a0931aac\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ef627a82e594cd4b27d1f2fc091b7df : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/665afac8c62645c1bf1ff0f46bfad63c : SUCCESS in 12m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/82929fb954e54cf5a21bfc11eca7afa8 : SUCCESS in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"342c3996ac5e31bd1ea6d3e591bed3d4b112070f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 22:12:53.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3b499950e55453f8d47850e0e5a566c\n\n- grenade https://zuul.opendev.org/t/openstack/build/53ee8124461f4232a1fc066a564717a2 : SUCCESS in 48m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/615e7946c73549ebac0943269b4845d9 : SUCCESS in 1h 02m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/35556c643be14fcfa4cafd275a5b93bc : FAILURE in 1h 26m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82009d8817bd49679b0a006ecaf5b778 : SUCCESS in 35m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b1e3e71db0e4c59bc64beaee6868c0d : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bebb66b400614ef2bc194a009af9c714 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8642dd75a5fb450781f2083073593317 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d6d7a79c3e94d2fa363e4e7e66186c1 : SUCCESS in 6m 08s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/03c9788e76aa4f9a93b98b9f6c7038f0 : SUCCESS in 11m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/caf8b7c262e24ceda9d8aba8434b0ba2 : SUCCESS in 8m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ff5e47b70e84488b9de8a4e5b6109461 : SUCCESS in 8m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/96be1ad9fd1f4d9ebc8291f6b4cc03b2 : SUCCESS in 9m 24s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/34e4e00b9b914d70b9b32537bf6cd052 : FAILURE in 11m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fda77aee2dca416b92c5696040a2145f : TIMED_OUT in 2h 03m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/aeecb148cb9242b4b84e0d3c147d2159 : TIMED_OUT in 2h 04m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cb414a18e1f04efa972c969060a45f4f : SUCCESS in 1h 53m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/27f98bd666b44296b91a5c4b24a3d24e : SUCCESS in 1h 44m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/414f36f60689446194df5e8d84e7bfc2 : FAILURE in 1h 54m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3745adb012941b29d0a320b6e91e371 : SUCCESS in 1h 09m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7cbb46351275449a96864ca8b5b4a177 : SUCCESS in 1h 47m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f893a71a7bbc4966b306e718e93db7bf : SUCCESS in 44m 27s","accounts_in_message":[],"_revision_number":14},{"id":"55973d8ecf9433d6ac28a801df585445c0c4cff2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-13 18:28:54.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"b984fc0f838d729761e6d052f15c2d0c6f19b166","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 18:40:42.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/621d9a8d45804b4b933a562bcaeb0d0c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a2811ba20a341d6935a91cc018ed18a : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d1b216fc8c254e1dbcc19f578c2162d6 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e8e8cb01388d456b96698991658966d4 : SUCCESS in 9m 17s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"6bed6a2b9960e96fd3a9d69d50e393f76503fd98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 20:38:51.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41043c5bbdf44108ac4de1133a26fa7e\n\n- grenade https://zuul.opendev.org/t/openstack/build/e26e856a20514f049092ebfc276e450b : SUCCESS in 55m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e4b77dd3c3674e4d9281032636222ffd : SUCCESS in 59m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ff748aab09e44d728a16c8bc0c8a1cab : SUCCESS in 1h 56m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b83d7792f84e4253bfd44f094006503c : SUCCESS in 22m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/044297faaea84d779c1312c22550a4f7 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d10cb7e5615f4bfe92a59dc0df8e3d21 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ee097c83db04eecb453fb1bdd7673d3 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ca0ce5ff83145aaa6a219463fd64c3f : SUCCESS in 6m 01s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d5d282dc47ed455a8093aa46df182fb3 : SUCCESS in 13m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a6797398032a41a884bc6f9fba73b450 : SUCCESS in 8m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/030591ea19654bf4a32b38607694ef3e : SUCCESS in 9m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/42b7ac59c2b54e01acd548f202859524 : SUCCESS in 8m 30s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a61a9e0de0e54aa79d8c9e71c38b1be9 : SUCCESS in 13m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/11d015e311a34c4e883588b3770e53e2 : SUCCESS in 1h 58m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/51374268bf14469ba0a55baf55933c23 : SUCCESS in 1h 44m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b74761d826f34b669468c04177df5956 : SUCCESS in 1h 30m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b2ff1f6451ed496ca457e4b69a77ae22 : SUCCESS in 1h 44m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/43748e0e3dc046e6a66499e8c341e65c : SUCCESS in 1h 42m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/973bf993d26e47ec82c82f97bf09f259 : SUCCESS in 59m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e97155c8eb6148498ac52362ae3c42b1 : SUCCESS in 1h 43m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/36f43a640ac34d3185a8b5308fb910de : SUCCESS in 27m 25s","accounts_in_message":[],"_revision_number":15},{"id":"41d281728db1d811b6150fdb6471b7dafea4346d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-14 05:35:50.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"67d6f50c716589a2f3b4bc4e82259407a88257f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 05:52:23.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3a8ea620ed24069b2353fe8452db0e8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/60c2ea0eec1943af936b24cb97ad29c8 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6d271da3c0ba4b8c896cf51aab5b9720 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8e03f8813746458084ecec2fa44e9167 : SUCCESS in 6m 49s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"716ebdcdd140219013964f7c6f9cdfde4035401a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 07:42:56.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec7cc6e874c1436586bc5b534e01b4ae\n\n- grenade https://zuul.opendev.org/t/openstack/build/52c1b96459cb4702ac1469536c5aebc8 : SUCCESS in 52m 22s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/74664c6534254dd1a45bd99556c68455 : SUCCESS in 1h 08m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d94b56f40b5d4efca814fbbd0d83c812 : SUCCESS in 1h 17m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4559dab30b9943f0a7287037779b2782 : SUCCESS in 23m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8043772595ac4c5ead628ecb7e54cf43 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35916148ec784414846c38984b439f57 : SUCCESS in 5m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f2a300cc6dc4810ad8f3787f61ea781 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/180b91051dcd449394f907b89166b0ba : SUCCESS in 5m 47s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0d1bfbbb5bde4b55a6f6a5ade79b9207 : SUCCESS in 13m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2b1be6eefbb3486190ee59b0e51f6203 : SUCCESS in 9m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e036e1982b1f434c9eae263905a3c2dc : SUCCESS in 8m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4dab5094f07b44f495415572c8a3c109 : SUCCESS in 8m 30s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/43aac710238842cbb4157279f34c8be9 : SUCCESS in 11m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/717f95f0edb04a3a9421d3be689bb208 : SUCCESS in 1h 38m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0338fe3140584ae286d67d1eb3cd5e7d : SUCCESS in 1h 53m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2ffdaeef9c714b08bf4c98a29ea28dc1 : SUCCESS in 1h 39m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f90f7345e88b43898d6f1d3b508a7e10 : SUCCESS in 1h 31m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1ca7253227a04db29c3bbcc765b66ed1 : SUCCESS in 1h 57m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/763c33475c6b4d37928eacf8a58bb9ff : SUCCESS in 52m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e86a3999f0b94fbd87b51b995f134f59 : SUCCESS in 1h 43m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/71c46876e6424f8dad15dc071dfcd7df : SUCCESS in 29m 38s","accounts_in_message":[],"_revision_number":16},{"id":"04c0e459e5e91a82291b6f513bff74b5b2823208","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-15 17:46:06.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"0ef5623fe6cc6fc41acf2f003b3960b520883213","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 17:57:16.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e61c9037f184d02b3edbdb313be9c12\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cdd4aeacf304439e8aa6c9c30524ba93 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c44a74614ff44980a748b8b90e37ccd3 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d12d1f61cce246a2a284180d057a9aca : SUCCESS in 7m 09s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"9f3d64e07068e186e5ddca4fa2869bdaeee650e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 19:59:57.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/35174ef8f0ba4859aba7e1a9c0b4aa69\n\n- grenade https://zuul.opendev.org/t/openstack/build/7f7831a2d6a84d628d870bb71aeb3bb1 : SUCCESS in 54m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ad2eb9fe34304ec5b4c62bcc4e776028 : SUCCESS in 57m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ff6e8fd892dc46d39e717cf0916ddf5a : SUCCESS in 1h 21m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdac7bf8c3f24dbf9430e294c8a8dd43 : SUCCESS in 28m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddeb7547d7f44564820fd335bbd2927d : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44a1e04256c740829ccacfbe357cafa9 : SUCCESS in 6m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25aacaf09eb14abbb4ac4b988db08444 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4a8a3440ac74f018a170a2749dc7b87 : SUCCESS in 4m 58s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3fbb82cc3b2e4474a38e7577a31e8975 : SUCCESS in 12m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/819c9462421948ebaa7eeba9762ebfb0 : SUCCESS in 9m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3555fc2c19c54e14b72690fe522f39f2 : SUCCESS in 9m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e5894e08804741cf9772d735b0752baf : SUCCESS in 9m 07s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6f6b3c030495446b9ebe8f2a190315a0 : SUCCESS in 10m 58s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/86a80f1ef8cb4051b40e8528a8f69ab2 : SUCCESS in 1h 43m 56s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/35bae465e61f45119be0da6cb2b219e9 : SUCCESS in 1h 25m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/85d8d35a614a44f9ad1e213c63c14f30 : TIMED_OUT in 2h 03m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/08907b9eda454c19a0da0741cd6193ce : SUCCESS in 1h 37m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1d9c593a42cc44289355fd3b19432900 : SUCCESS in 1h 42m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc6e9821233a4c5fab2f4f7d304d82e9 : SUCCESS in 1h 00m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bf395fe783c64a94b34c4238f6b358b0 : SUCCESS in 2h 03m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/28a547579b3f417280fbb68f5f1b640a : SUCCESS in 35m 45s","accounts_in_message":[],"_revision_number":17},{"id":"a27512a45030cb15c11da48aaec8fb37394d460c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-16 17:12:14.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"a87a1ec8c0d346830d786e5b92ee54d8f9a68e1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 17:27:42.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a080b43065b45f3a622a7e1a83ef7b3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c6d31b40b6f41718fe3a6cc018fbe3e : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fc947ac694e94560b97403b87df20c87 : SUCCESS in 11m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/de47f1a6ee8641589c7f7cf2f6cc9061 : SUCCESS in 6m 23s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"e13418f2825ad06baa9cc541c96976009bf95837","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 19:16:50.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43c34f437bd74c019a8988683e2a428d\n\n- grenade https://zuul.opendev.org/t/openstack/build/dac9c72a176542f3b5ec01894c451532 : SUCCESS in 1h 08m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7efd21c0104b49f2bbae49b77766d633 : SUCCESS in 1h 02m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/99be178d4d094155b1e4d33eff96173f : SUCCESS in 1h 28m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/785c2d964a3b4ecdb806390f7204c744 : SUCCESS in 1h 06m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24f0c9c69c844ecab03526e9de1a7b1b : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e5aa18b0e724169899a12cfdb12c3d0 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/578264d20d9d41eaab67dcd1df6e1317 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/522922116311453eb00d5fa7cb41319f : SUCCESS in 6m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/241eaeac8e8b446d99ea50702f3d6b85 : SUCCESS in 12m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/92b93200f6314767a58f18808ed35765 : SUCCESS in 9m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f703e5f170ea49549f232fc75e252f81 : SUCCESS in 10m 16s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7baedfae7f05464396796b7d468962da : SUCCESS in 8m 23s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/8fa3c45404d64accbbe85cbd55012c66 : SUCCESS in 11m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/590db593ed134620813ffaeb2ad85db4 : SUCCESS in 1h 51m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/62d6c67149584d94ae4b052ae7062a71 : SUCCESS in 1h 23m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bceb3ce3117b43c6b505e1aa43c8e460 : SUCCESS in 1h 39m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c1305a8338c2445fae2f048086e38779 : SUCCESS in 1h 38m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b71a628e6f914e148c0b4f6857bd18ef : SUCCESS in 1h 56m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7ac4064048b4a3e92a1ab5a6c7dc085 : SUCCESS in 50m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7610fd1b82ed4f6eadfbe3aab4f854aa : SUCCESS in 1h 50m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8dfe6b719a53449e8255a53f76f4e36b : SUCCESS in 34m 23s","accounts_in_message":[],"_revision_number":18},{"id":"a65873ae9bf19f99ca49a05a267447a11c77ff26","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-20 15:13:22.000000000","message":"Patch Set 18: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":18},{"id":"fe07ef9cc7b7e6e38263faeb7f3e953fcd246ee6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-20 17:43:32.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"d3ade3df405d83cdd5837272f24c253752075499","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-20 20:07:20.000000000","message":"Patch Set 18:\n\n(4 comments)","accounts_in_message":[],"_revision_number":18},{"id":"46540653d6804ca61c7a2516200b97902d94000d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-20 20:30:08.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"fe10b861c9fdfe0eeea3651d07946f8613ac80fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 20:42:37.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c07e14130dd47b6afe23faca8ffcb71\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/535e060527dd4a1a9776e10ce9721ebb : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0dbb4265e9b84c038d84a8b9232b9811 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e2ed72dc3408447ca9fa29babb79aca8 : SUCCESS in 10m 02s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"005da1d669da1c3e7198e120b6f69f40638b8f80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 22:42:30.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fea6f01b99ba4de58333b7ff6d8e9196\n\n- grenade https://zuul.opendev.org/t/openstack/build/c5b99c5279a248dd8b8046193450fe7d : SUCCESS in 58m 34s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/707d390dc80b467381608f329bfb6a99 : SUCCESS in 1h 01m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fd9938802cc34c52ae1dc0c50026f424 : SUCCESS in 1h 17m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc08b32ba39b401597712485364e0e9b : SUCCESS in 55m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/796d9fb43ca94ca4831bcea3c08a5fdf : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ff1653b4b894e54bcf0cc296fff444d : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0bfed11f357a499a9d4f79494bc82b34 : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17ba160725414998ad0a60f7f439fe12 : SUCCESS in 5m 26s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9153e42adf9441f3845ff9383929aa1f : SUCCESS in 10m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/db1b25938f894f529e3f0edadab3e764 : SUCCESS in 8m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1e4af2b5c6d04fd6a04b73ece8d2a1ab : SUCCESS in 7m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5a0ed2fe1f92493986c48e3b4d638dd8 : SUCCESS in 8m 26s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b55b2d9140d74a5eac04806030bc86c4 : SUCCESS in 11m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/474c5605bf104f7a9b213e6989a0327f : TIMED_OUT in 2h 04m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8c127e409bcb4cea8b46504dd200ab12 : SUCCESS in 1h 47m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/171da05ffb4643838a1a95cbf4a7832c : SUCCESS in 1h 41m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/59e60e1535524f6ea6477e8455ead366 : FAILURE in 1h 45m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9bdd34749e784005814348b9f37b82e4 : SUCCESS in 1h 47m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c69cdbf756364776ab16c582ab91ed2e : SUCCESS in 1h 12m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a5843846df4943fca7106169bb69964d : SUCCESS in 1h 47m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/09a5f09392e6451a8ba31b2dae6673af : SUCCESS in 39m 50s","accounts_in_message":[],"_revision_number":19},{"id":"9dfcccce3979dc9fd23c0466449f75f8946f7376","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-21 05:26:08.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"030ebe2c3ab22376584a3426cb5156b46458b90f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 05:35:39.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/452a7b8d85ed40d4a868ed53f84a6d62\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/59c909411aac4339b8ee769d80e504a1 : FAILURE in 4m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b5b4ba8a56924827b1ec9baf698a7dc4 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5059e310e7c34c71ac0f64e9ac3394ff : SUCCESS in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"7513e0d07c2a8d239b962ca23ffc7e600c8bbee4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 07:39:19.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f579a95987dc4e55b9019190e2ae596b\n\n- grenade https://zuul.opendev.org/t/openstack/build/b995b043bc6a44b5b778a82509d26ed4 : SUCCESS in 1h 06m 13s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/18fdab65b20a4e50945f2f41d741f38b : SUCCESS in 1h 05m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e5f349eb9fce40d882f9ff9cefcdcaf1 : SUCCESS in 1h 59m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3963557766640778e158f0317f7e870 : SUCCESS in 1h 08m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeb802d730ee46da8bc136aa2b8a0e38 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92138a5bee594820bdfe559f4213cba4 : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c9fa408e8c1466dba79da4892fe8320 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16ba1576032644729672120862958015 : SUCCESS in 6m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/30b4fb240d594c3a8efc0ced73269132 : SUCCESS in 12m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bd3d2ec9b4f548959557b673b7d732fc : SUCCESS in 7m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/64f35d751d4f45ddb353cd47ce705ed6 : SUCCESS in 9m 38s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/011a8369c0f2450691404e6c3dc9b77f : SUCCESS in 8m 39s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/28f06d180ed940a9bee1561afc33ba15 : SUCCESS in 10m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/09961cba87cf45acadc9bba2dea665bb : SUCCESS in 1h 50m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7393ab6c4e684aa690aad7f564df9d6e : SUCCESS in 1h 47m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5d8ceb9533d0476087b753b15b533877 : SUCCESS in 1h 52m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c1bbfc5aed4747878f16cdb6885a9eef : SUCCESS in 1h 55m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3b93d21864604dceae36611f7d69b92f : SUCCESS in 2h 07m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98ebd72c99614d8597bae212c33f98e5 : SUCCESS in 1h 00m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b3057a2a54d84e5a8d7095563ac1b93d : SUCCESS in 2h 02m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f50a6db523c346febd68845443236b8a : SUCCESS in 41m 43s","accounts_in_message":[],"_revision_number":19},{"id":"54c51bacee96ebd5856950f4a8d5d958c1f2e5d7","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-27 15:22:03.000000000","message":"Patch Set 19: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":19},{"id":"8cba7ec609654269c62a1941605202f5839c7ead","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-27 17:24:36.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"5bae28400cb583a9c81bca30c882d7654d49517a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-27 17:31:44.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"e26333653f4ae0f2a2e15e43b2aa73654f5ba6f2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-27 17:38:38.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"b27e86a3b2d17e9e4a384189b04fc7dee793311a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-27 17:55:54.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"febcd1dacfeb7599c76c2476ab192062bb217fe7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-27 17:57:41.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"10ca5f1bd390db8d337aa569b1d9623314b70a91","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-27 18:02:50.000000000","message":"Patch Set 19:\n\n(7 comments)","accounts_in_message":[],"_revision_number":19},{"id":"27895071e1ad6ee1e523c498d2f2945350aeb701","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-27 18:03:28.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"5367ffb4e2d2f5a97b824bfe3130382fa7410698","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-27 18:12:20.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"18f1388936e5a01f8d3ab8bcddd474e086b7d51a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-27 18:13:15.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"2bbdc2d345239b76157b9974d0d6c0aee38b3fa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-27 18:45:05.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"7b218cb984bddf0c7237ebb8b4c87a57ae503d80","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-27 18:45:38.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"b6fb79ff09de76951721214ba7f00bfd6ddbc0ba","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-27 18:53:41.000000000","message":"Patch Set 20: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"3a52f0c911ae85d72effac48eb302508a8c97c8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 18:58:47.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e84d774b27d466aa83cd26f5759af75\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93c57ca2548442c2ac3e54448c21ee9e : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8ae335fc1fa9469c9139d4b81e719554 : SUCCESS in 11m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/75e72407e5734bc1b8585a0ab22ebc1b : SUCCESS in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"65bcd18a196e8638c444cfdff9f07fe14436c0ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 20:52:00.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e797a2d208941c3ab3df64e697a0b07\n\n- grenade https://zuul.opendev.org/t/openstack/build/f5ca091309c7441a9da4868ddcab14fe : SUCCESS in 1h 00m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fc90e9e553884c3faf9ba9a407c0f6a9 : SUCCESS in 1h 01m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3f3aa35915784e49a461b3a4a2253ce9 : SUCCESS in 1h 26m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e863485b3c142f992b16eec61ede1c4 : SUCCESS in 50m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cccbd86e4484d31a342264b1b0e1ee6 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b18bba5cb06450492c51600d83f6052 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23068fe2b49848cba8dce634e4670475 : FAILURE in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3521927cc0964c079f4adab1ea0c4cbf : SUCCESS in 6m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/31322f37d76f4803b98f49878d96c77c : SUCCESS in 13m 31s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/43a0e7cfd48642f6a1f0e970f88ad659 : SUCCESS in 9m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/af210abfe9104714b2bc5fcba87c48f8 : SUCCESS in 12m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7ea3ba14890f491a8cab0fcf27cfa36b : SUCCESS in 8m 36s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6e4875eb5c5641f78c4ad16ea3e364de : SUCCESS in 13m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d4503fb1ddc142d0b2d020672dbab5bb : SUCCESS in 1h 58m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/92c63e4d586d448a8cbc578842848583 : SUCCESS in 1h 38m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/30c6846c94f048e88297488ac9a73643 : SUCCESS in 1h 42m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/17f438e6095a49af9f0a7e4a5d5fa241 : SUCCESS in 1h 44m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/67d740ce67a94e799978abb8302f7db8 : SUCCESS in 1h 59m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/283b5769301d44b7a3595c1208eeaf6e : SUCCESS in 1h 04m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/80f8df471f7b42f99526f5bcb14f6a62 : SUCCESS in 1h 56m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f68b34558690446281e56acad839c3bb : SUCCESS in 38m 22s","accounts_in_message":[],"_revision_number":20},{"id":"f361bd9690e37c317e1c19b078483bc89547918b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-28 03:58:46.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"9bb588729386af408611a5c463b2afdab736d5ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 04:08:22.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9d0e620be814ca2a98000f81878bf51\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d84cd15484e14007a240aa2ccc95cf7e : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b485728ec35e48fbb607942147954298 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b7fc7c7584bd4a7fba3138511716182c : SUCCESS in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"bd8726f8b06b042b813c9f26c3565ff26bd75a88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 06:08:56.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4398e30ad2343748abe2fdba33eb48a\n\n- grenade https://zuul.opendev.org/t/openstack/build/0e083f94b222449fa9a6672909417e90 : SUCCESS in 58m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/42e0fd60da8f414bbee627936f6a08b0 : SUCCESS in 57m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/66f7507895ee45c1a78df4d4c824384a : SUCCESS in 1h 24m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfb0f0cade6c4c14a8a9f921cd376d5c : SUCCESS in 58m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe617529fda94de79f6f1e8fbf205bef : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c253eee2f1d4444f8900161ad522ca59 : SUCCESS in 6m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84cf1fd45e2d4635a08e781428f45a1c : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87bd83e39a6642f1aa790591261b7a7b : SUCCESS in 7m 37s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b2c18ed940ce43ae88d146b09ffd51be : SUCCESS in 14m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a5b725f943b74ea4a50d8dc3ca4b55fc : SUCCESS in 8m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4a69261736d540149839a8218ad23427 : SUCCESS in 9m 12s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4ac37f609dcc4c559b43863da1a4c94c : SUCCESS in 9m 27s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a73272f53c254537860e6907c551bf6f : SUCCESS in 11m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/390ff52ed9cd455187b954f437604acd : TIMED_OUT in 2h 03m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/55e825eb19a246f187f06e0cbbeaa833 : SUCCESS in 1h 33m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/43345ddd24dd4d14a79d24518f7eb41e : SUCCESS in 1h 55m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ecb33e28c06e4ff5ae77f1024eb78891 : SUCCESS in 1h 54m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f248c98019944fa885242f7b6b75a53f : SUCCESS in 1h 55m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72533cdce6fb4f9996a6620ab473f8df : SUCCESS in 1h 08m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f90b94608aa445e8ac4f3bb1b1a028bf : SUCCESS in 2h 02m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6a38f32bc1c847609a7944da42125462 : SUCCESS in 38m 20s","accounts_in_message":[],"_revision_number":20},{"id":"b3f9fd9da30567db6a5fa42b62b2d0f8c66072df","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 07:11:04.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"669b53223b5a2c2998f5caa9c107798f43fed3a5","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-28 14:22:03.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":20},{"id":"ab59681dec94c8fb3f7d4e41c0a56985e10050c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 14:22:42.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"48a148e02351de415e76ba64f11c557548d57c7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 16:33:32.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a119a1bf9989412cb84e8af4871d2f2f\n\n- grenade https://zuul.opendev.org/t/openstack/build/4a9b47965cc14ae8b9be9a9627bb648c : SUCCESS in 50m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2fc389f3fabf4447ab13b820c80437c7 : SUCCESS in 49m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/42ae31c22c6d430b8c0977f0f77fce12 : SUCCESS in 1h 14m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d32843829234e87870c731990e426b2 : SUCCESS in 1h 08m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/820f8920ed16417caf7a312aa314a86c : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/106b1a94b48942efbf2dc242fe1958fa : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/02922c0c65b34f9b9d41b7a1ee250b58 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d834ea09660344c9a0ef2af90d37b2d3 : SUCCESS in 5m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6382233d736044af9f85c864eb355a4f : SUCCESS in 11m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8354882424604e828b644ef278371c74 : SUCCESS in 11m 41s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7d21647cbb7448319627a24ae439f48a : SUCCESS in 1h 33m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/518e6c9d72fa49ae8d2afcc59198dcc0 : SUCCESS in 1h 34m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/30dff25fb2c0452ebe35be92b8063431 : SUCCESS in 1h 55m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/755da93d215c4987a7da6bff80337847 : SUCCESS in 49m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b0928f0dfaed46ecbfc39e2025b7bd9a : SUCCESS in 2h 05m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8b11a13487a34441aa6f371cf22188fb : SUCCESS in 43m 17s","accounts_in_message":[],"_revision_number":20},{"id":"85934508ba852e0a05b7c755eddd7a3eab02776c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 16:33:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":20},{"id":"6c784061ad4e5a09465f50359b3a025bb1a60cbf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 16:35:11.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9e398255a1a48ee97c79ed23fa5c591\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f806adda5474ff7b55b5bd4915319fc : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"781ce285604844643cfc30232406a402fa32c30e","revisions":{"b99f7c0ed0bf1f3d124e656d68fcd7dc2f65f75a":{"kind":"REWORK","_number":1,"created":"2023-11-08 14:37:54.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/1"}}},"commit":{"parents":[{"commit":"436cb274b0f78441d065788118810eef77648d5e","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/436cb274b0f78441d065788118810eef77648d5e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 14:20:12.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nToDo:\nUnit and functional testing\nRefactoring and manual testing\nMultinode migration testing\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b99f7c0ed0bf1f3d124e656d68fcd7dc2f65f75a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b99f7c0ed0bf1f3d124e656d68fcd7dc2f65f75a"}]},"branch":"refs/heads/master"},"fb207b55602ce771ca06b2b3ac843ca88986ed3d":{"kind":"REWORK","_number":2,"created":"2023-11-20 07:51:06.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/2"}}},"commit":{"parents":[{"commit":"436cb274b0f78441d065788118810eef77648d5e","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/436cb274b0f78441d065788118810eef77648d5e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-20 07:34:44.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nToDo:\nUnit and functional testing\nRefactoring and manual testing\nMultinode migration testing\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fb207b55602ce771ca06b2b3ac843ca88986ed3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fb207b55602ce771ca06b2b3ac843ca88986ed3d"}]},"branch":"refs/heads/master"},"d317fc928f5650f991685954a43ed5b0d7714375":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-12-21 04:49:46.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/3"}}},"commit":{"parents":[{"commit":"65caf39e9e8e70e4aa9c96c486d9b30ac0152758","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65caf39e9e8e70e4aa9c96c486d9b30ac0152758"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-12-11 06:04:26.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nToDo:\nUnit and functional testing\nRefactoring and manual testing\nMultinode migration testing\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d317fc928f5650f991685954a43ed5b0d7714375"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d317fc928f5650f991685954a43ed5b0d7714375"}]},"branch":"refs/heads/master"},"166c4f89ec1cf607c82849b25a65553f345c54f1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-01-16 04:12:07.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/4"}}},"commit":{"parents":[{"commit":"5a4d46d25aaafb913a8e4430db9ff7c897e8ac92","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5a4d46d25aaafb913a8e4430db9ff7c897e8ac92"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-16 03:52:57.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nToDo:\nUnit and functional testing\nRefactoring and manual testing\nMultinode migration testing\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/166c4f89ec1cf607c82849b25a65553f345c54f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/166c4f89ec1cf607c82849b25a65553f345c54f1"}]},"branch":"refs/heads/master"},"d03387c77f0b48c01eec8d5a6a7318fe4618ee71":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-01-17 09:26:42.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/5"}}},"commit":{"parents":[{"commit":"4c3271e21407be403da50cc379c46329a54ef999","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4c3271e21407be403da50cc379c46329a54ef999"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-17 09:06:45.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nToDo:\nUnit and functional testing\nRefactoring and manual testing\nMultinode migration testing\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d03387c77f0b48c01eec8d5a6a7318fe4618ee71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d03387c77f0b48c01eec8d5a6a7318fe4618ee71"}]},"branch":"refs/heads/master"},"9c16075e9c29cefa39b0332ae853c7b7f3f9d5c2":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-01-17 18:28:54.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/6"}}},"commit":{"parents":[{"commit":"d1de77b7168ba8ec5a2fa0cdb4b80915e907fe64","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1de77b7168ba8ec5a2fa0cdb4b80915e907fe64"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-17 18:07:51.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nToDo:\nUnit and functional testing\nRefactoring and manual testing\nMultinode migration testing\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c16075e9c29cefa39b0332ae853c7b7f3f9d5c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c16075e9c29cefa39b0332ae853c7b7f3f9d5c2"}]},"branch":"refs/heads/master"},"4130784cecda33a9e8e5035f35f70fcc57697ea3":{"kind":"NO_CHANGE","_number":7,"created":"2024-01-17 18:37:18.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/7"}}},"commit":{"parents":[{"commit":"4542d95045c83f7cfff5c505b5c313c1fe773619","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4542d95045c83f7cfff5c505b5c313c1fe773619"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-17 18:18:32.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nToDo:\nUnit and functional testing\nRefactoring and manual testing\nMultinode migration testing\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4130784cecda33a9e8e5035f35f70fcc57697ea3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4130784cecda33a9e8e5035f35f70fcc57697ea3"}]},"branch":"refs/heads/master"},"10eaff28420bc0fd8a64125d36cc93d6678829c3":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-01-18 05:59:22.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/8"}}},"commit":{"parents":[{"commit":"e42c532998da66b87a52813daf7e3589d6fd5df4","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e42c532998da66b87a52813daf7e3589d6fd5df4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-18 05:39:40.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nToDo:\nUnit and functional testing\nRefactoring and manual testing\nMultinode migration testing\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/10eaff28420bc0fd8a64125d36cc93d6678829c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/10eaff28420bc0fd8a64125d36cc93d6678829c3"}]},"branch":"refs/heads/master"},"ef3d4032322426491ddfafdf5634a85c74655717":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-01-19 10:17:56.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/9"}}},"commit":{"parents":[{"commit":"de93e8b3c444e6692d612492b6ac80fbb3442631","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de93e8b3c444e6692d612492b6ac80fbb3442631"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-19 09:54:48.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nToDo:\nUnit and functional testing\nRefactoring and manual testing\nMultinode migration testing\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ef3d4032322426491ddfafdf5634a85c74655717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ef3d4032322426491ddfafdf5634a85c74655717"}]},"branch":"refs/heads/master"},"c24511e1ed800d5f6c58044ff7e67db5cde6a148":{"kind":"REWORK","_number":10,"created":"2024-01-24 14:41:03.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/10"}}},"commit":{"parents":[{"commit":"199859d9647ca9748deb54d6b8c7cf2004408096","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/199859d9647ca9748deb54d6b8c7cf2004408096"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-24 09:42:35.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nToDo:\nUnit and functional testing\nRefactoring and manual testing\nMultinode migration testing\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c24511e1ed800d5f6c58044ff7e67db5cde6a148"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c24511e1ed800d5f6c58044ff7e67db5cde6a148"}]},"branch":"refs/heads/master"},"df5f3af13560ae7b5c7a821945f609064d2febb3":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2024-01-24 14:43:48.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/11"}}},"commit":{"parents":[{"commit":"199859d9647ca9748deb54d6b8c7cf2004408096","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/199859d9647ca9748deb54d6b8c7cf2004408096"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-24 14:24:35.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/df5f3af13560ae7b5c7a821945f609064d2febb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/df5f3af13560ae7b5c7a821945f609064d2febb3"}]},"branch":"refs/heads/master"},"9ae4ca6c265b0e5d4f90ea2e6aef9ee72b602955":{"kind":"REWORK","_number":12,"created":"2024-01-31 21:23:20.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/12"}}},"commit":{"parents":[{"commit":"199859d9647ca9748deb54d6b8c7cf2004408096","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/199859d9647ca9748deb54d6b8c7cf2004408096"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-31 20:52:06.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9ae4ca6c265b0e5d4f90ea2e6aef9ee72b602955"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9ae4ca6c265b0e5d4f90ea2e6aef9ee72b602955"}]},"branch":"refs/heads/master"},"d9701f95cf600d497965b45fce3a63fef51e4c53":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2024-02-09 10:05:42.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/13"}}},"commit":{"parents":[{"commit":"8f5f9dcee4ee0458d60cdb0b35875cc274ea71bf","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8f5f9dcee4ee0458d60cdb0b35875cc274ea71bf"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-09 09:32:12.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9701f95cf600d497965b45fce3a63fef51e4c53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9701f95cf600d497965b45fce3a63fef51e4c53"}]},"branch":"refs/heads/master"},"12d5df76f0a0e2cdf6d1005cec74c40ac37e093f":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2024-02-09 19:54:58.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/14"}}},"commit":{"parents":[{"commit":"b84991d048e0872335200f297f53349f820a31bf","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b84991d048e0872335200f297f53349f820a31bf"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-09 19:08:02.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/12d5df76f0a0e2cdf6d1005cec74c40ac37e093f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/12d5df76f0a0e2cdf6d1005cec74c40ac37e093f"}]},"branch":"refs/heads/master"},"66cdb6092a0ea7ddb7288ae458adc056c13cb89c":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2024-02-13 18:28:54.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/15"}}},"commit":{"parents":[{"commit":"d6f33f20d63661264a6b2b2a3608b37fc64b1caa","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6f33f20d63661264a6b2b2a3608b37fc64b1caa"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-13 17:30:57.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66cdb6092a0ea7ddb7288ae458adc056c13cb89c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66cdb6092a0ea7ddb7288ae458adc056c13cb89c"}]},"branch":"refs/heads/master"},"a2bedba22e439bd9bdd275933e248cc809a2da3c":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2024-02-14 05:35:50.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/16"}}},"commit":{"parents":[{"commit":"5293fb207729c89b48e1d175adab7a3f08208074","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5293fb207729c89b48e1d175adab7a3f08208074"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-14 04:57:49.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2bedba22e439bd9bdd275933e248cc809a2da3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2bedba22e439bd9bdd275933e248cc809a2da3c"}]},"branch":"refs/heads/master"},"e9489f6b535e0c20461bd33c5a69b6f319343d21":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2024-02-15 17:46:06.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/17"}}},"commit":{"parents":[{"commit":"7a6a289d3f0b2d1d5592899bc1b6daf884afd9ff","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a6a289d3f0b2d1d5592899bc1b6daf884afd9ff"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-15 16:59:49.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e9489f6b535e0c20461bd33c5a69b6f319343d21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e9489f6b535e0c20461bd33c5a69b6f319343d21"}]},"branch":"refs/heads/master"},"7e638b099a44e74240a53f12de04630af8861b27":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2024-02-16 17:12:14.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/18"}}},"commit":{"parents":[{"commit":"0168b1a0c0ff51470b0d9071e18f384896b9437f","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0168b1a0c0ff51470b0d9071e18f384896b9437f"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-16 16:44:08.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e638b099a44e74240a53f12de04630af8861b27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e638b099a44e74240a53f12de04630af8861b27"}]},"branch":"refs/heads/master"},"0f2a3644e92bec030ac07f9562215a2b6ba2eff0":{"kind":"REWORK","_number":19,"created":"2024-02-20 20:30:08.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/19"}}},"commit":{"parents":[{"commit":"21333c1ac85e52e31a29db2529abf3dfdd04e9a0","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/21333c1ac85e52e31a29db2529abf3dfdd04e9a0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-20 19:53:04.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f2a3644e92bec030ac07f9562215a2b6ba2eff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f2a3644e92bec030ac07f9562215a2b6ba2eff0"}]},"branch":"refs/heads/master"},"781ce285604844643cfc30232406a402fa32c30e":{"kind":"REWORK","_number":20,"created":"2024-02-27 18:45:05.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/19/900419/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/900419/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/900419/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/900419/20"}}},"commit":{"parents":[{"commit":"21333c1ac85e52e31a29db2529abf3dfdd04e9a0","subject":"Move sqlite code in common module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/21333c1ac85e52e31a29db2529abf3dfdd04e9a0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 08:51:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-27 18:16:04.000000000","tz":0},"subject":"Migrate from SQLite to Centralized db","message":"Migrate from SQLite to Centralized db\n\nMade provision to migrate existing cache records from\nsqlite to centralized db and delete those from sqlite\nafter successful migration.\n\nImplements blueprint centralized-cache-db\n\nChange-Id: I2a840915bb6d9d72d6379cae09ebec0e51a4a6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/781ce285604844643cfc30232406a402fa32c30e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/781ce285604844643cfc30232406a402fa32c30e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
