)]}'
{"id":"openstack%2Fglance~902260","triplet_id":"openstack%2Fglance~master~I2d21d8b098355a6cc4ceb1de8f9d7fefa19be965","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-11-30 16:14:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2d21d8b098355a6cc4ceb1de8f9d7fefa19be965","subject":"Increase timeout for glance cinder multistore job","status":"MERGED","created":"2023-11-30 12:10:31.000000000","updated":"2023-11-30 16:20:14.000000000","submitted":"2023-11-30 16:14:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902260","meta_rev_id":"af22d4b29412a48472c3b93e4f9e7a1bd2517ecb","_number":902260,"virtual_id_number":902260,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-30 16:14:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-11-30 12:15:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2023-11-30 12:15:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-30 12:15:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-11-30 12:26:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-30 14:19:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59e9eee434ff0f1e9e2c7e542727aea5f6b01f81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-11-30 12:10:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56fef9b73dd2cba434d05df3e9a1a3b287221803","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-30 12:15:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d2e41505fc8ec0f491706fdedf37aab62a997f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 12:26:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89c0c8583f084c6bb8c24b31ec57072b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad74470bb91348ea8d8a4e62005c0bf2 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e4097fc683b640ba9f087b2159a86e29 : SUCCESS in 12m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1ced7629714d40a4b052d34b94e09596 : SUCCESS in 5m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"315542f231789694046666998080f23311a06d2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 14:19:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b95fbc40d7ac489b84732eef727a7bb0\n\n- grenade https://zuul.opendev.org/t/openstack/build/216d391e393e47d5952efa5477ec5583 : SUCCESS in 59m 48s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c2edec30dfa9405a9c488c9051e90877 : SUCCESS in 1h 05m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/00e21bb0a68548b19ce9d0d3700a196d : SUCCESS in 1h 19m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/486f7d49c0884cadb4ee3caa7e79e1b3 : SUCCESS in 1h 32m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2231029c4ea4483480ae5f4b2a727c13 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42b2d6cfefb04009a76104b41f2f3df8 : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe6b593889cd48639713cee13542d480 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/492d1a8565de4525a8960dea6b1870ef : SUCCESS in 7m 13s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/586d73290d6e4e4580cd74b20d20249e : SUCCESS in 15m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d042467ff57b4434baaef8ddbbe0bddb : SUCCESS in 8m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2c8dd723ee1247d7b4d5bb811e2aaf31 : SUCCESS in 9m 27s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6ad9eadde99e493eb94adb6914399e8c : SUCCESS in 8m 14s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/d025ce2f23544d2ebe27f600061f6c40 : SUCCESS in 9m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a7ee33ce391245a6b65899288e0676f2 : TIMED_OUT in 2h 05m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ff3ab674adce41bd97673eff59b2e38a : SUCCESS in 1h 50m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ef3df19350c14cebbd46e036c1d8aa56 : SUCCESS in 1h 42m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/785aa19c59ed41a490d63b3941bdb3c3 : SUCCESS in 1h 39m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/29720fe6d3074127a04cbd8717d57b6f : SUCCESS in 1h 30m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/632caf4f248a4b5cac7bd277b3881ba5 : SUCCESS in 1h 01m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/43860a3f6f164a279c0cb9fe5ce6b1cb : SUCCESS in 1h 55m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/338a6a586f2b4034a98084a08fce365c : SUCCESS in 30m 54s","accounts_in_message":[],"_revision_number":1},{"id":"49fe0bc28c181ba9f70c81fb796752f76a3d0c8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 14:20:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"92866789bfc8c79403dd2f2febf7cc917b47a220","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 16:14:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de98d72a90204842b7f0918cfec5a9f2\n\n- grenade https://zuul.opendev.org/t/openstack/build/aa80944cf46c44579f88b260788a8f49 : SUCCESS in 1h 03m 44s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e135748614e34ed3be2dbf289bc4efe7 : SUCCESS in 43m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a05d93d2f7db40ce91d1c9e14ff91db9 : SUCCESS in 1h 41m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef7d4f3b980d4aab90ecefdece03f027 : SUCCESS in 1h 08m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d74d0ff7bcf14785a4903c4d2867f9a5 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fea6185104a44fabe653b70ae29475d : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/82eac8d1b9994205817731380141a4dc : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a781fed890dd444f95bc63ddd7acf06f : SUCCESS in 6m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/853e2972a3794b0bae7a210a33d424b8 : SUCCESS in 8m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f23c57a075734e5ea3822dc067a81610 : SUCCESS in 9m 52s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/072fb8ce688f48bfb14598cdabfb377e : SUCCESS in 1h 35m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e94966c435fc49b3a131373c6b5b4634 : SUCCESS in 1h 48m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/73296b3e66e84a21b73f73c40b118afd : SUCCESS in 1h 41m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ef08a5e22104daebbe370f87ed48f91 : SUCCESS in 42m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e54a7c99df5f418e83f6196c4cb79f95 : SUCCESS in 1h 52m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/15d55368f904401f9a5431fdaa4e66a8 : SUCCESS in 30m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e9f532849c320e45d35d6bc1fb7dc70eda6270ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 16:14:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"af22d4b29412a48472c3b93e4f9e7a1bd2517ecb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 16:20:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/282c6bd20b8a4b57b327f901a828c2be\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fdbccf10ee84ac38ec73d65ca0701b4 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64e6da6fe1c810befab1e5a59230ce8711dce709","revisions":{"64e6da6fe1c810befab1e5a59230ce8711dce709":{"kind":"REWORK","_number":1,"created":"2023-11-30 12:10:31.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/60/902260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/60/902260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/902260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/902260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/902260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/60/902260/1"}}},"commit":{"parents":[{"commit":"db41ce67283b02e3b610b81a3131923d15b3e7a8","subject":"Revert \"Make glance-tox-functional-py39-sqlalchemy-tips job non-voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/db41ce67283b02e3b610b81a3131923d15b3e7a8"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2023-11-30 12:04:37.000000000","tz":0},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2023-11-30 12:10:25.000000000","tz":0},"subject":"Increase timeout for glance cinder multistore job","message":"Increase timeout for glance cinder multistore job\n\nCurrently the glance cinder multistore job times out from time to\ntime in 2 hours (and some minutes)[1].\nThe issue is not specific and generally points to a slow devstack\ndeployment or a slow tempest run[2] which exceeds the allocated time\nprovided for the job to complete.\nUnless the job is checked thoroughly to determine the areas of\nimprovement, a short term fix would be to increase the timeout\nof job (which might also be the ideal solution if the job is\nlanding on a system with inferior hardware).\nThis patch increases the timeout to 10800 seconds i.e. 3 hours.\n\n[1] https://zuul.openstack.org/builds?job_name\u003dglance-multistore-cinder-import\u0026project\u003dopenstack%2Fglance\u0026skip\u003d0\n[2] https://review.opendev.org/c/openstack/glance/+/899868/1#message-081c737d8747de44669acd1cdc322e347b09d390\n\nChange-Id: I2d21d8b098355a6cc4ceb1de8f9d7fefa19be965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/64e6da6fe1c810befab1e5a59230ce8711dce709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/64e6da6fe1c810befab1e5a59230ce8711dce709"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
