)]}'
{"id":"openstack%2Fglance~902648","triplet_id":"openstack%2Fglance~master~Ia31be8ca412fcc4a28f91cb63b5daa48e3d49f98","project":"openstack/glance","branch":"master","topic":"image-encryption","attention_set":{},"removed_from_attention_set":{"27665":{"account":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"last_update":"2024-07-31 15:56:55.000000000","reason":"Change was abandoned"},"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-07-31 15:56:55.000000000","reason":"Change was abandoned"},"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2024-07-31 15:56:55.000000000","reason":"Change was abandoned"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-31 15:56:55.000000000","reason":"Change was abandoned"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-07-31 15:56:55.000000000","reason":"Change was abandoned"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-07-31 15:56:55.000000000","reason":"Change was abandoned"},"28271":{"account":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"last_update":"2024-02-12 10:17:10.000000000","reason":"\u003cGERRIT_ACCOUNT_28271\u003e replied on the change","reason_account":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"}}},"hashtags":[],"change_id":"Ia31be8ca412fcc4a28f91cb63b5daa48e3d49f98","subject":"Add support for image encryption","status":"ABANDONED","created":"2023-12-05 12:57:11.000000000","updated":"2024-07-31 15:56:55.000000000","total_comment_count":9,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"69e7ec1ba37798456f8c9250bdd1d864377784e2","_number":902648,"virtual_id_number":902648,"owner":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-12-06 20:07:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":-1,"date":"2024-01-10 05:35:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":1,"date":"2024-01-10 05:35:03.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-05 13:04:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-05 14:19:09.000000000","updated_by":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"reviewer":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"state":"REVIEWER"},{"updated":"2023-12-05 14:19:32.000000000","updated_by":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"reviewer":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"state":"CC"},{"updated":"2023-12-05 15:22:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-10 05:35:03.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-01-10 05:35:30.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"},{"updated":"2024-01-10 05:35:30.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-01-10 05:35:30.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2024-01-10 05:35:30.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"dd398ee04adf57c27af807352a2bad09f53fc2cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2023-12-05 12:57:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f69b821cd39a7848ae9ff4b2f9c6e50d0fb8f3f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 13:04:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14b7a6c521c44472b530d4038af3d513\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5d9323d67494006990364855fbafd41 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d781420d6c3341b6a96f25d1de4945da : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/677050e6542a4ff1bba78cbcb8740d77 : SUCCESS in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"21cdc93a90e9e9c09dde6db31ac91b89e4bfe73a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 15:22:07.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ed5f1b0277c4621b247aad290aea299\n\n- grenade https://zuul.opendev.org/t/openstack/build/f08d108b15e04961b694cd5327fdcf49 : SUCCESS in 1h 16m 37s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d0247a12491541d09a8694c1e8d3faaa : SUCCESS in 1h 03m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1af225c8e9774e5c9f6510b273e6c6c8 : SUCCESS in 1h 13m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17f5c3005972431597d82592c6f908bb : SUCCESS in 1h 16m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f862c88983d40babe21e5da9cacd174 : FAILURE in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/367b7a2f696e4d2487a498e8fca3ef36 : SUCCESS in 6m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0cc843c169f64681b1e5b65f70a8667e : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/897096eb1ba5406bb3317e930865fa00 : SUCCESS in 6m 52s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7faa49717de3421c9838c8560fc1bbc3 : SUCCESS in 13m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/07ca823873a94951a4bffa61dbbdc58f : SUCCESS in 7m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e87ac1e13962404093c2c3f0aebcb331 : SUCCESS in 9m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/50aaa3d9ffb24414bde68aac7cb66f1b : SUCCESS in 9m 05s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a9e0dbc2fa774dd8a047337bb6a6a39a : SUCCESS in 11m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a95d83e4ce364f7fbe79689c9cea8e44 : SUCCESS in 1h 53m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6b17c81da22241d4b2b2a03e38f8fdf4 : SUCCESS in 1h 41m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/25b33522cdae45d089374f4ad897c2bb : SUCCESS in 1h 35m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/df7ea183ffb1444088f5796f1354b482 : SUCCESS in 1h 59m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b16b29119e0c4ed9891edf44e313e924 : FAILURE in 2h 15m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47fe62b1216e48eeb5e598843f47d476 : SUCCESS in 1h 05m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6499340b3a3048e495f89e4baf207be3 : SUCCESS in 2h 00m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2d0f3494b3034869b5a020536ca8da89 : SUCCESS in 32m 39s","accounts_in_message":[],"_revision_number":1},{"id":"626020dd57a4e492cc9f929b44b67f8ef89e4f50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2023-12-06 17:44:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"91408c7088d6f3e8029895aab2a583770c5e1eda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 17:51:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7213a71045af468e91691d81dea75156\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/224e0a25a210485aa8db4196310a6bf2 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5eb8757809c045d88f8002c3ae22b417 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e532788e10884119a8c1236f6c27836c : SUCCESS in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4ba6dcc6fd403709226b5a18a3cb8621f2d39a6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 20:07:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b45122905d714b7abd49a156d5981dfc\n\n- grenade https://zuul.opendev.org/t/openstack/build/413fdd52ae344f718215db28ff154f3c : SUCCESS in 1h 02m 58s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6f9d7143a9c74875b593f9ae47089428 : SUCCESS in 58m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e9933e8a184249f39c868aba6740e5c7 : SUCCESS in 1h 30m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b2b60a5f0a749ac88f79f6106cc0841 : SUCCESS in 45m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4908e400b7474d7ab1edafec2e2867f0 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/488d869609ab43fc9e2588be981f9a3b : SUCCESS in 5m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f0429c2c51a940ef8707998d175af5c9 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2b2d7c3f99f4557b92d8e119da42f52 : SUCCESS in 6m 57s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8bd959dfa3f7499eb5ee633c00cdf199 : SUCCESS in 12m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c291060564384ad089328ac1d03272e5 : SUCCESS in 8m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c065e1199b624b4daf21b6892cee1623 : SUCCESS in 10m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f069460ee1cc49dfb610d40bbb17d4ee : SUCCESS in 9m 14s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/7e3597441eea4530aa2eb361ce307f38 : SUCCESS in 10m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2390503caeca429693388ab57f5f40bc : SUCCESS in 1h 43m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c49fc48a6c6c4418996a8d48242f8e3b : SUCCESS in 1h 32m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/36977993d52349f78ca6dd9fde209bc9 : SUCCESS in 1h 26m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ef523b04c31a42998c82c35ec4858c45 : SUCCESS in 1h 22m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/df37fc829d714af9984203c33f572fdd : SUCCESS in 1h 32m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efacf2c812264259abb73f840a4988c5 : SUCCESS in 50m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/894a8d1581eb4ad1bdcb2eb54e4bf1b1 : SUCCESS in 2h 16m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bd114c98aef34cd7aa2ca7fd2ca6bc28 : SUCCESS in 39m 33s","accounts_in_message":[],"_revision_number":2},{"id":"5500c7e4ed19611ca97ebd5f9387e8bb8f32ebbc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-10 05:35:03.000000000","message":"Patch Set 2: Code-Review-1 Review-Priority+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"961a9c5658cb72b98e2ac780d0331904ab458dd6","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-02-12 10:17:10.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"69e7ec1ba37798456f8c9250bdd1d864377784e2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-07-31 15:56:55.000000000","message":"Abandoned\n\nWill be superseded by an alternative implementation as per https://review.opendev.org/c/openstack/glance-specs/+/915726","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"963692aad98ef67e57ed71a3920df994695ad715","revisions":{"d7511e1a5f4dc908e6df45174e83128eac34f2db":{"kind":"REWORK","_number":1,"created":"2023-12-05 12:57:11.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/48/902648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/902648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/902648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/902648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/902648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/902648/1"}}},"commit":{"parents":[{"commit":"03a14a9b9c3fbbe106d1f2cc1e33fb594a5be963","subject":"Merge \"Unset VENV python_interpreter in fips job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03a14a9b9c3fbbe106d1f2cc1e33fb594a5be963"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@secustack.com","date":"2023-12-04 15:21:25.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@secustack.com","date":"2023-12-05 12:56:49.000000000","tz":60},"subject":"Add support for image encryption","message":"Add support for image encryption\n\nFor the image encryption a new container_format \u0027encrypted\u0027 is added.\nA validation for additionally required parameters is added for the\nimage creation. Registering and unregistering as a secret consumer\nfor the key of the encrypted image is added to the creation and\nremoval process for encrypted images.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@secustack.com\u003e\nChange-Id: Ia31be8ca412fcc4a28f91cb63b5daa48e3d49f98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d7511e1a5f4dc908e6df45174e83128eac34f2db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d7511e1a5f4dc908e6df45174e83128eac34f2db"}]},"branch":"refs/heads/master"},"963692aad98ef67e57ed71a3920df994695ad715":{"kind":"REWORK","_number":2,"created":"2023-12-06 17:44:33.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/48/902648/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/902648/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/902648/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/902648/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/902648/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/902648/2"}}},"commit":{"parents":[{"commit":"03a14a9b9c3fbbe106d1f2cc1e33fb594a5be963","subject":"Merge \"Unset VENV python_interpreter in fips job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03a14a9b9c3fbbe106d1f2cc1e33fb594a5be963"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@secustack.com","date":"2023-12-04 15:21:25.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@secustack.com","date":"2023-12-06 17:44:14.000000000","tz":60},"subject":"Add support for image encryption","message":"Add support for image encryption\n\nFor the image encryption a new container_format \u0027encrypted\u0027 is added.\nA validation for additionally required parameters is added for the\nimage creation. Registering and unregistering as a secret consumer\nfor the key of the encrypted image is added to the creation and\nremoval process for encrypted images.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@secustack.com\u003e\nChange-Id: Ia31be8ca412fcc4a28f91cb63b5daa48e3d49f98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/963692aad98ef67e57ed71a3920df994695ad715"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/963692aad98ef67e57ed71a3920df994695ad715"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
