)]}'
{"id":"openstack%2Fglance~904422","triplet_id":"openstack%2Fglance~master~Ic41aca29142a8143d308642912c3ff7a9536a04e","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"36644":{"account":{"_account_id":36644,"name":"yuguizhou","display_name":"yuguizhou","email":"yuguizhou_yewu@cmss.chinamobile.com","username":"yuguizhou"},"last_update":"2026-07-07 19:09:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic41aca29142a8143d308642912c3ff7a9536a04e","subject":"Log displays information about image updates","status":"ABANDONED","created":"2023-12-29 01:36:47.000000000","updated":"2026-07-07 19:09:14.000000000","total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"f39f9b17d9c47459211d2b5ab1e50f1deba4b2eb","_number":904422,"virtual_id_number":904422,"owner":{"_account_id":36644,"name":"yuguizhou","display_name":"yuguizhou","email":"yuguizhou_yewu@cmss.chinamobile.com","username":"yuguizhou"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-12-29 03:53:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":-1,"date":"2024-04-30 11:47:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-29 01:42:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-29 03:53:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-30 11:47:42.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-04-30 11:50:34.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"}],"messages":[{"id":"c3befdbc0d2c90647893f39cfdc9e5c9393c24f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36644,"name":"yuguizhou","display_name":"yuguizhou","email":"yuguizhou_yewu@cmss.chinamobile.com","username":"yuguizhou"},"date":"2023-12-29 01:36:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb53122d29b02faad006d4cbd4aa6cbc509feba8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-29 01:42:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87634de8f371475ca612b93fb54efb12\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b887ce4a04cc44ed80ab3bb6e2a86ef9 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8790ee85caa7407c97285012fa1dc144 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f57e5e821d794d788193d1b3a8bc7380 : SUCCESS in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c8fbc2e20e41462aab331258c69231ca8c8c5028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-29 03:53:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0d5ab7e3fb44f9cb249593597e9d46c\n\n- grenade https://zuul.opendev.org/t/openstack/build/e00db540d87242e4a22b778b6dfe1dbd : SUCCESS in 1h 04m 07s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/269b4d2cb6724e1da32e3da7f312c043 : SUCCESS in 1h 12m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c9c61820fe3a42b3824be5027d98cd5d : SUCCESS in 1h 39m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45c8191f8f6d490d824a958db65efa6a : SUCCESS in 1h 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62567dc870804edb819625ac368413f2 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/290d8d9073c34b129f72dc1ddfd3c73a : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/39dd77400a6248c6bfea38b34cdc5520 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/979964f90d9d42639ae7251a6c8a009b : SUCCESS in 5m 44s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d9ba9f19d9c74350be10f8b4b6921501 : SUCCESS in 10m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0f491884a61d4fc5bb977d48cca35a3f : SUCCESS in 11m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b528c879d1824c4a9fbe0ef89612fcc4 : SUCCESS in 8m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/37290308c3134648a03bf03380a447ce : SUCCESS in 8m 28s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/042e10d9e26c4b06be8d4136a9b3e501 : SUCCESS in 11m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bebc430f8c4e47e7867e6bd02140e58b : TIMED_OUT in 2h 04m 27s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fcad8198f2894788ab8e9d59eb8187ab : SUCCESS in 1h 37m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4255a8295865434a9bb71766a0169ca1 : SUCCESS in 1h 29m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ef2935f818db4dd9a050c7ed409a24ee : SUCCESS in 1h 36m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/df6ec83e14e74b508b34aef7d49f5d98 : SUCCESS in 1h 46m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f757c35d00c4d2d854efc08dcd957b6 : SUCCESS in 42m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1bde2352d8204e74a60b87f24b75fcc0 : SUCCESS in 2h 10m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/388f03c5c12046078a80eec0ff40bfec : SUCCESS in 47m 17s","accounts_in_message":[],"_revision_number":1},{"id":"276e58b9bbdf15615251a6d1f0cc02df23851c5d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-04-30 11:47:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ea9bfa90e427ceaa38b0bbbdab27ba712b6acbf9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-04-30 11:50:34.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f39f9b17d9c47459211d2b5ab1e50f1deba4b2eb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-07 19:09:14.000000000","message":"Abandoned\n\nNo answer in 2 years, I\u0027ll abandon that. Feel free to restore if you\u0027d like to see this change merged.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e0025d3e53d37f813cdd62876d2d805c9ce8f07","revisions":{"4e0025d3e53d37f813cdd62876d2d805c9ce8f07":{"kind":"REWORK","_number":1,"created":"2023-12-29 01:36:47.000000000","uploader":{"_account_id":36644,"name":"yuguizhou","display_name":"yuguizhou","email":"yuguizhou_yewu@cmss.chinamobile.com","username":"yuguizhou"},"ref":"refs/changes/22/904422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/904422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/904422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/904422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/904422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/904422/1"}}},"commit":{"parents":[{"commit":"41fa8223b7dda10e2ff310097bb86700f254015e","subject":"Merge \"Add Packed Virtqueue extra spec and image properties\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41fa8223b7dda10e2ff310097bb86700f254015e"}]}],"author":{"name":"yuguizhou","email":"yuguizhou_yewu@cmss.chinamobile.com","date":"2023-12-29 01:34:08.000000000","tz":480},"committer":{"name":"yuguizhou","email":"yuguizhou_yewu@cmss.chinamobile.com","date":"2023-12-29 01:34:08.000000000","tz":480},"subject":"Log displays information about image updates","message":"Log displays information about image updates\n\nWhen an image is updated, the updated content\nis recorded in the log\n\nChange-Id: Ic41aca29142a8143d308642912c3ff7a9536a04e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4e0025d3e53d37f813cdd62876d2d805c9ce8f07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4e0025d3e53d37f813cdd62876d2d805c9ce8f07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
