)]}'
{"id":"openstack%2Fglance~905051","triplet_id":"openstack%2Fglance~master~I703ff8cdebd2e929ab775f9b601a8542fb413153","project":"openstack/glance","branch":"master","topic":"py3.11","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-01-17 20:18:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I703ff8cdebd2e929ab775f9b601a8542fb413153","subject":"Update python classifier in setup.cfg","status":"MERGED","created":"2024-01-09 04:21:25.000000000","updated":"2024-01-17 20:19:43.000000000","submitted":"2024-01-17 20:18:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905051-py3.11","meta_rev_id":"4524b7fb330fd8967f1b010f8d85d5af875ef414","_number":905051,"virtual_id_number":905051,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-17 20:18:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-01-10 01:57:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-01-09 05:04:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-01-10 00:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-15 05:24:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2024-01-15 06:01:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-01-10 01:57:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-09 04:36:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-09 05:04:04.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-01-09 06:35:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-09 22:47:51.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2024-01-10 00:57:38.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-01-10 01:57:51.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-01-15 05:24:38.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2024-01-15 06:01:31.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"19c29146246749d55e5ff9f045b8df327d295069","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-01-09 04:21:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7f05bf67844617cb87d581c41e34d4233a36ad5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 04:36:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe7b46974453485593345a7bcc1a4d8e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2e47953464d4c4691bc8558ac5b250b : SUCCESS in 12m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/82d6c91bd26f40dbad69d743002bd03b : SUCCESS in 11m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/388916be9843426aa9c0d94a7c6c988c : SUCCESS in 10m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a6d53bea68b24c6e11b7273c0332d8b920a40ddf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-09 05:04:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"38bce16d74caef02a58d9d85e810bd92257f4b57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 06:35:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d6f3c246c284ac9a114966261905718\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f3d08e4502074e2f854db39798d09c1b : SUCCESS in 1m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/f8b2ba039dd84ea89b9d3b4c896f4bda : SUCCESS in 51m 07s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0df32831b95d4d14ae3ef64db0a6568a : SUCCESS in 54m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b2524e5c30354e8ebe871178cae47165 : SUCCESS in 1h 35m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69189a6711e24fe793b02aad4310bb43 : SUCCESS in 40m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/850397abc7be43e8a90dd73a15277f52 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b076827f74a44e4b80345d9d502f6df5 : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97f4943cd2d742508d6c694110e633f5 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fda2709c3ff404b8a9d9d6ac17f88e8 : SUCCESS in 6m 04s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e14e8697ca0e4fc4a25a0cea42c92476 : SUCCESS in 14m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4fbe2f9e7a5a4639b42bdd33df4d2a9b : SUCCESS in 10m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d929615e0e6b4cf29391492d96f105bd : SUCCESS in 8m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2e12dc53782846fe83b9d1d4b814bc6b : SUCCESS in 8m 36s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/d94e64f7d42e4595b0292ec311ad4a80 : SUCCESS in 10m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/880c87e94dbe4c859cb8f9307e2c8c11 : TIMED_OUT in 2h 03m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/aa4ef7db762648afa3e0902e99af5da2 : SUCCESS in 1h 40m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/68e7172883d042a6bd5e6965025dfe9b : SUCCESS in 1h 17m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/29bb07c775a44ae5b404a7af9dea810d : SUCCESS in 1h 22m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c840aaba0efc417fb15c1bf34b0f3ed0 : FAILURE in 1h 53m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d74a7302444d4acc9699c6a616194ee5 : SUCCESS in 1h 02m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5a8b5914574c4f21acb5e410e0b92f02 : SUCCESS in 2h 07m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5019e020e90743be9d7fa3e90bb94109 : SUCCESS in 42m 25s","accounts_in_message":[],"_revision_number":1},{"id":"9f50ac046c4e5bfe2ed24367ffb6a39ea197eb1d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-01-09 22:47:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e9a29824662a16523e47d587075b92d2ecf3d9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 23:01:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee50efa4cdd94a7298c4e8a2574f24ae\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2dfda6a25194d4a9a636c6609d802cb : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2b63c928032549cdb5f9884c1b360f50 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1aaf270dba464d6c9d414b454a126f76 : SUCCESS in 5m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c67a1c2fe3296dcb1c4bed74d66a21649a6e8a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 00:57:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/07a31de77fbb48cfacfe52527620e2c6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d9cb29e2034c4dfe8e16116961ec51e5 : SUCCESS in 2m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/f93ec4c7ecdb41698663235cc1365294 : SUCCESS in 40m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d93018500e0b494687a151805dbc2290 : SUCCESS in 1h 07m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6d50489e43174562bdfc5bd560b2cf1f : FAILURE in 1h 15m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6710159b64242349b6e080ca1c7b2f8 : SUCCESS in 22m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/902b3ac260c946b6ba7a173016f6e9f5 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a1e033befa64f5da2cb7aeae3082d21 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5a12c5c3f1a44bd5a35237cd33a64d7e : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a98d51cd8074dc69ff6be44a4bbbc9a : SUCCESS in 6m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f0fce9416adb450887f08880ee4ea53b : SUCCESS in 14m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4fab65f0c0994cc2a1e7e7a2e38a8a30 : SUCCESS in 8m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c0c35ca0cb614712af82ea38347013b5 : SUCCESS in 7m 44s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4949653e273c48ea88a6e0494e33fff4 : SUCCESS in 8m 45s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/e6eb78823f6c44e7bd5f1794921646de : SUCCESS in 10m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/16c4d3ad5f33498bbdd1acc34857822c : TIMED_OUT in 2h 03m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ca63d692dfe141aea295e4abec9822b1 : SUCCESS in 1h 40m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/32109dfd019847ab89ac58c73dfb5ea8 : SUCCESS in 1h 44m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f91513000a1042c6aa5cc20a78c0c448 : SUCCESS in 1h 55m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9491e8a992724345b96eaa10237b530f : SUCCESS in 1h 55m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/888dd3fe0b9547ab9311ef4396013ce9 : SUCCESS in 48m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/69d7397770ff4659b245e6d0cbb1d9cc : SUCCESS in 1h 32m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6e85e12422474ffc9e075501e693f397 : SUCCESS in 23m 22s","accounts_in_message":[],"_revision_number":1},{"id":"fd60568081f2ee48b01000f7c31eba8879ae8d37","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-01-10 00:57:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"228beb9d60fbae4efe90dc2590171d836fd15395","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-01-10 01:56:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b41661fb5c530549a810e7d238d6e11d042efaf4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-01-10 01:57:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"65a51e6608e7eb0855ac2dae379765d06b32d9c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 02:03:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e039e34c7764a17a832fd30fc2b4b17\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5df7d6edcac84ee6914da7457001b605 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/321398c9d13a4ab6b3b8ad494166e04d : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2f0289ccf85b47cabe223671c0fb256b : SUCCESS in 5m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06970a830785d3295f0eddf4b47bdf891adc7c25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 04:28:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1dda981924a421cb6e64eea322220a0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7fa5be3d53004314b47bd0b7a636413d : SUCCESS in 2m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/92de8155f55f4230b28cee2b8dcf0eef : SUCCESS in 1h 05m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a6d2619aedcb43fea8309b6ed343233c : SUCCESS in 1h 01m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b8886ddaa14465da17c781c2568609b : SUCCESS in 1h 27m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5a0dbd741cf490aab227a3e4bd80730 : SUCCESS in 38m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35bb7dc64a7d4b2d941950d349aa724f : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7981f24d1ab947e39da4b7969617f9d4 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/208c5544b28a4beb9896e05cc67ac2f1 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cedf8556a607420086f02d7760b6de2b : SUCCESS in 6m 52s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/882581b644244eb0a088c3d3155992b9 : SUCCESS in 14m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/35633ae558b54efbaf0931b2e897d072 : SUCCESS in 8m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bc358ca813114b3fbe1d93882da1939a : SUCCESS in 8m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9c1bc72ebe154021a0ba5453f26e6620 : SUCCESS in 9m 08s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/020950d2034c402aa8876db740a02f2d : SUCCESS in 11m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/92b4f26ac75e4d8fae4e44f7ed8323d5 : FAILURE in 1h 21m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a74354da3af5490d9c0e09e53e150719 : SUCCESS in 1h 51m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/67b0fdfa163b4ccdb5dfba934dbc5aaf : SUCCESS in 1h 50m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b76049b22fc8471485c3505074ba352b : SUCCESS in 1h 50m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b65fbaf582e14035af529118f1924c8d : SUCCESS in 2h 03m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e97128de2cd44a6979f7ad4affa8590 : SUCCESS in 1h 10m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8be8985a7c234189a136e91a3d8a7843 : SUCCESS in 2h 02m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5a24c3707b44434284f584539aced65a : SUCCESS in 52m 41s","accounts_in_message":[],"_revision_number":1},{"id":"c8f290a98c92abd45e76b284472300f58127085b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 04:28:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"980026a2a4a5701cbe6bac448ea9b1c275a15312","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 06:32:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3285facc65d4a858d26cd96413aa33e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6144de4d36b84686bc5c1c6e46d415da : SUCCESS in 2m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/01b83864f74a41baa5583c4aa55c6fae : SUCCESS in 49m 44s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7d582416f55344aba5bce05d467ad7c0 : SUCCESS in 59m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/deec30bfffcb466b86cc862fdb678c07 : FAILURE in 1h 09m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5244f9bb582a4e7685acbf552c2ee4db : SUCCESS in 38m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6ee1a7677bd403e9fa41cda65c51660 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c831cfc23eb45648d0fc4a59b8940c3 : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b305212dacf04fbeb2e84bbe8f687b9c : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7c7f3d6f7e44e7a990bf55bff427f76 : SUCCESS in 6m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c5759b45c2b74f7080f443fee5ee6a1e : SUCCESS in 8m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/03e47da0517c4ce093aa78d66f294365 : SUCCESS in 8m 15s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/806976dd61f14bf6ade2957313968088 : SUCCESS in 1h 58m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f3a006135c27479090fa153f2c817fea : SUCCESS in 1h 47m 24s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bb7e6133c8804a5e80589427bcabadae : SUCCESS in 1h 24m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e3b4d67bd4146cba01efe99da5fa93b : SUCCESS in 1h 03m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ab668a00617a426aa78f6ccd00844764 : SUCCESS in 1h 55m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1edb597d120641b2aaf08eeaa94de8a1 : SUCCESS in 40m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f490d65500e5aceba62ea966823dfe067a5a0282","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-01-10 12:47:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bdb7b83a4148670955946d0293268fd8199a21b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 13:01:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/026906809c53417dbc352196ad3bd296\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/075e8a9458f64b3a945224f67c41a56a : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a2dcdab9a35f48a8be951fa37a0dfa0e : SUCCESS in 10m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ea4e99e852414fddbbe27f28432009dd : SUCCESS in 9m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8669bac1fbeca6770c5ca35cb2d45598c79f5e17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 14:55:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a04efb2d043429597d06f301bb61400\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/42df84e58831441b9092887f30bc3845 : SUCCESS in 1m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/9e217882be8d4bfb8365968b0d8ccbad : SUCCESS in 56m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/36dab661d8fc4dbb9066aed3923aeb61 : SUCCESS in 59m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/822d1eee8e4146f7b5729db54850b675 : SUCCESS in 1h 24m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54323cf484094a6097c22b400eaba3ad : SUCCESS in 37m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1b5d8b9a687400f95d47eb131e1111d : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54c8003bc77a4776b0da351da60d2d77 : SUCCESS in 7m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ed9e9fb485c47b0985a7ad307e8aa12 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f072162faa944fb698b025e159256f68 : SUCCESS in 6m 28s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d3ef21eb61874646993483701945057f : SUCCESS in 12m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bf983749f56143839a4e0f5b88b7ed8c : SUCCESS in 8m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d8267f7a571a48eaab0ad428a77d1a1d : SUCCESS in 9m 59s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a8d2027308754a9d9259a6972a8ef07d : SUCCESS in 9m 26s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c95b412f2d9c46659196ae0908367e76 : SUCCESS in 11m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2e9524c2ede34272b7d849bae1802690 : SUCCESS in 1h 58m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b435c46297aa43f1a266a0892b5b27a5 : SUCCESS in 1h 38m 43s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ddd0f5383fe743399136639fccf5ec2f : SUCCESS in 2h 02m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/67c41321e2ae4e2bbe565a777295841a : SUCCESS in 1h 47m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4c8df805c37a4424ae20ef3f5b9468e6 : SUCCESS in 1h 54m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee0c48b59b8b4dda882e8105e4704660 : SUCCESS in 1h 01m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/562d42f35b6d4f76a0d8619cafd7026a : SUCCESS in 1h 59m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/917abf70218c45f29a7f905d7615f027 : SUCCESS in 40m 14s","accounts_in_message":[],"_revision_number":1},{"id":"52061c3de603af26feb0e93cf182a84c1a573c5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 14:56:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f0ccf0f54c18b9b5423c69be7614277e16135ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 17:14:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/646d07e414a64eaea551bfdd0ddc973f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f51d0c963f984b5d9bdd4e802dd38ec2 : SUCCESS in 2m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/44724a3970e34501a035477aef9f4210 : SUCCESS in 1h 10m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cf6ecdd1befd439791805baae87399fe : SUCCESS in 50m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c5e6349a38f4b39a47c82e5cbdda1de : SUCCESS in 1h 27m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0e6f965dc294e2eb38bce482ca1b0ba : SUCCESS in 41m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7244e3d80ae44fc2a3e7f2f1494bb015 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5323a99513934660810fab2ca22104e8 : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9877607f15cc49428a630acd272e8cb2 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe75f2f9da6144eba785f2c4d47bad33 : SUCCESS in 7m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/29b464f5dd6843cababc52a4f00c9d7e : SUCCESS in 11m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/27ff78292a3949c7b1c6bcde95cbd538 : SUCCESS in 8m 05s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/701597a79cf84f90aa236a459924d969 : TIMED_OUT in 2h 04m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d1ef6df60cad43e985c01ac244df916b : SUCCESS in 1h 51m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/223ead18f2d44d31923541b02bc8546b : SUCCESS in 1h 43m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6c252710aa14530ac048fa441491927 : SUCCESS in 1h 00m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/92c73c2544514c508c2ca1bcbeb36dfe : SUCCESS in 1h 55m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2e753855719d4711b24dba3e3b240227 : SUCCESS in 28m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0bb4070bfd8036285c1faf2b315ac5712d247e17","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-01-11 16:52:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd54c89728c9131749075a18564fad6f8510b5ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 17:00:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fa55b243e6646999f5fc251ce5bff48\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f5d66aa9e6d4a7691e9a2d395d44df2 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/419dd08defe244b7871482692e3a681c : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/da289b407f7646288390f0e770b9045e : SUCCESS in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23d4f41d715d720093e2a90294f1a879b2b0b3b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 19:20:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/27ba20eff0304510b9d25b41aa3a021a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5c2b39bdcef1463cb55ce90675007d24 : SUCCESS in 1m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/d1c4a30a11c4424f8f58634b91d5e33a : SUCCESS in 1h 10m 34s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6581f98ee5e94b09948ceb2d0424fa39 : SUCCESS in 57m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b6fd7cd2ee344a949bea99648794ea47 : SUCCESS in 1h 30m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/495d1861d3e64bb48dd4a3300c0e90cd : SUCCESS in 32m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e71bfce1d57437d84e7ea0b901d4d35 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5824135ec6104787a5594d59b9fa5ad1 : SUCCESS in 6m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/424d9bff64ce4b42baf1ae4199066ee1 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c192dbf64b14c178ea9303a351a7393 : SUCCESS in 6m 34s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/07a5c9a044bc412d9fadc0d33e9de61b : SUCCESS in 13m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e3cc84a3c4054682bd19ebfe6954d208 : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cb9ece6debd7481ca5b73f47cc80019d : SUCCESS in 8m 30s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/07b7bc926adb47dd8e6c505b00d0a773 : SUCCESS in 8m 41s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/466bdd8fdc054473b5e038e2d22ad817 : SUCCESS in 11m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4889ed1fc4e8400280ca982de280d270 : SUCCESS in 1h 40m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/efc86141f60a4a0c98a0cdaa0fd08323 : SUCCESS in 1h 47m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b121658bf6a8433090443c0eb497d7ac : TIMED_OUT in 2h 03m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/59f4daa745ca4a38b6c0daa7881ac8d1 : SUCCESS in 1h 40m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/60f766a806a449e1b432c235bc6223e7 : SUCCESS in 1h 48m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7d107276c3e487cb3bf5f7e3660f2ee : SUCCESS in 1h 05m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/18dd6b1a8cbc4067b6854cfc9f82ca15 : SUCCESS in 2h 21m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3bd7faca2d6040fd9db95218fe7e5908 : SUCCESS in 30m 22s","accounts_in_message":[],"_revision_number":1},{"id":"944739a0447d924ace5aaf3fdfbd7f9ee0befe64","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-01-12 01:55:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cff27b7e80f198c80d98e33e21b09f5fb4d109fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 02:02:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e617f884d49a4fa58c2fcb8dff1e2b2e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa6ba820d2fc49c5b0919ecc01be27f4 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2900c794528445cd87e5722414f63147 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1413576eac93461085ac7ccf5dd976d6 : SUCCESS in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf240b93ef9ec832fdc5e63fd5e1da2b459d2814","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 04:09:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c10e71a58ec479393135bfce29513f4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3804c5c1c1a84666a509908a20f717a4 : SUCCESS in 2m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/f80ccd811b284678985546e311625bc3 : SUCCESS in 1h 06m 56s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f4ae729a0903413293a2b0539fde68c4 : SUCCESS in 1h 10m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4db23e2e98754a23bcb10ee42b723560 : SUCCESS in 1h 28m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66a4075ef4c645bba58caae9be51c6be : FAILURE in 37m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/585a4c8b085d44f4b908998d06dc0911 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/043bcfc95085480693f7665b7a20e1bd : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eef2da416af0446f9cd2ce6355730895 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2060bdfccbf4441cab4071fcd774c60f : SUCCESS in 6m 35s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a063b7260eca4584b32be391cd7f9406 : SUCCESS in 13m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ba5aa736faa84ff487282005d0ac3246 : SUCCESS in 9m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d6379d787973401ba6c750f522c06aff : SUCCESS in 8m 24s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e53e58f6cad2435c94297c728efb6fa6 : SUCCESS in 9m 06s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/485abf9179c64c03a60d6ae73916f9d5 : SUCCESS in 11m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b236afcd3884482ea530e233f1eb4fec : SUCCESS in 1h 54m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1fb055fe6c5041338ad833edba1da231 : SUCCESS in 1h 53m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/eb2799bf2f724ed2adbaa53b930afbf2 : SUCCESS in 1h 46m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/58c4a6f18d834f27af3e93d22ce5f204 : SUCCESS in 1h 52m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5162dd88d63a4746b6530914a92a7ce9 : SUCCESS in 2h 06m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90a676af0fdc48cb8a1f325154ae1ff1 : SUCCESS in 1h 06m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6d69c0ee6ed94075b118ff4d5d6c93fc : SUCCESS in 1h 58m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d2932ea0202b43b5a090071b4fec6872 : SUCCESS in 45m 09s","accounts_in_message":[],"_revision_number":1},{"id":"780b7ea09ed12aa66f998ab65749b06d217d1e80","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-01-12 13:22:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbb74a7f18c5d9686f3db80a2f9f607ec42d80c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 13:40:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b25c208bc3a40358af9c631efc02e52\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/279b7c13526043ee92d38c4a97ab9aac : SUCCESS in 11m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5136f4b5c1d3478d87cbb9e729fcd8e3 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/baaf7233c5cf499c9562b49a9388b0af : SUCCESS in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"218e03dbe262d3328dbcc1a862bbc72788730b86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 15:34:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d25cac9f202c4c3785894ea250d227de\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/742626fbe46b4e9788cbcb7817ea39f9 : SUCCESS in 2m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/cc52242ee5d94fa3b89d61241e3afc60 : SUCCESS in 1h 10m 15s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/26a627f110a34e1d94e9c75077a21449 : SUCCESS in 1h 01m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d8927ee8e1654a4ea62816640d9460d2 : SUCCESS in 1h 39m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a0e0487f0c34609a5bb7d97a1f33ad2 : SUCCESS in 38m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dec517644bb447699f0a4f7ca270cbc9 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f63c78070b8c4ffd8493bad761a63f09 : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de1fe54c8ad34d5291fa6633686d3465 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af677e1495644ad8aff7ced6f2e4cbba : SUCCESS in 5m 57s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5c9a31c62b5e43cf9aa84d9b458f9ab5 : SUCCESS in 13m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a08b2ba843554fa68d8b7761a0605b50 : SUCCESS in 8m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/15ee6d05c0844d939d56f35e1b86db40 : SUCCESS in 8m 16s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ae62169d9e62452981270f12cd5b34c9 : SUCCESS in 8m 16s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/2788d0e20aab49998c3667c61a027148 : SUCCESS in 12m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ab081fde0d29401bb6af98b1af245b95 : SUCCESS in 2h 03m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8b490864f2ef4a348a28cdd2de2acf91 : SUCCESS in 1h 53m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/03f99d470fa44e0680796021be9cafc9 : SUCCESS in 1h 45m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bbd7932a82354cd29235e5cb55a4a4dc : SUCCESS in 2h 02m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/594036fbf0c941f2a393313b9a92fa40 : SUCCESS in 1h 50m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9851ad0d19d44269977c376b278b7504 : SUCCESS in 46m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/347ccc5fa6264e36a1202b2aa70f90a9 : SUCCESS in 2h 01m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c57d42fcacc5462cbb7332809cf7180b : SUCCESS in 29m 23s","accounts_in_message":[],"_revision_number":1},{"id":"18cefa12a2155c0d3509c2697f7db85968a8efd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 15:34:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fb106a233340b533b98bedca1582952de59f99b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 17:48:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/521bb1195d42403790e91194e437a77d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/572a93065ee34d798148442dc68a6fca : SUCCESS in 1m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/2656aac22f924a049fd986b91e07d446 : SUCCESS in 1h 11m 56s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e14df50313204484a4b89a1fb25d036d : SUCCESS in 1h 01m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1a4d925649094b4082bd9cf2921f4f1f : SUCCESS in 1h 27m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68e58122eaa142dc89bf1d334ced8979 : SUCCESS in 32m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a8e3a656cf74b61bb8bb477b267d9ec : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0eb3bcaa9d249829bf7befb42277ad2 : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/48048b99fdf345bbb1afa2c3b404bfc0 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee6feb679f5f4ca0af1ea472b92f3f8a : SUCCESS in 6m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/58ca90238c414d50af6ea1b247ae26dd : SUCCESS in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f8a34e0dc137453db5c88df83defaf0d : SUCCESS in 8m 46s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/89551d20f1014008baf3853b8d93e03b : TIMED_OUT in 2h 04m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/912dee1b85804695a2dbae82108828a6 : SUCCESS in 1h 54m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c181229796824edb85dc4c03844d1409 : SUCCESS in 1h 52m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/054753fcfb8b40f996c192400e63bdf4 : SUCCESS in 55m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c79efd4a8e7d46b5975642edc6b028eb : SUCCESS in 1h 53m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/57ef3a50464f4e7c95c218681bd3d47b : SUCCESS in 39m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ac110cfac398d302231500eee3cf7e397914f911","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-15 02:02:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be79edcd49d52abebd14eb4eb427b9f0801503c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 02:35:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3281b24bbd9f4395a43cbd6e57c30173\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/86f75c5088bb4cb7af40d1f1e3362827 : SUCCESS in 12m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/068bad22754b40f5a626b72ad94f8a1b : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/634bb343e72340d99d0cc00b71fefc6d : SUCCESS in 10m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"74ba7aea7354e600f67139192f55deb139a81ae6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 05:04:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d99a02ef13843948dc99d9e073d9b50\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/26a1d2ede1fb4319a2d88b23a24ef223 : SUCCESS in 2m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/929140f5411940e791d5a80bab57fbf8 : SUCCESS in 1h 10m 42s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a7ba4774f5e140d2875c22f324840749 : SUCCESS in 1h 10m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f73dbdebd1044c048d3f01e7adb3548f : SUCCESS in 1h 08m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b4aa97d638e4543bccbb873b4c6cb39 : SUCCESS in 22m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ab773b592664dc59fca80f29a54a262 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f86c80bd5cc146438032d5f0ee0e4e4f : SUCCESS in 5m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5179a11008d44c189e554760a22bfdf1 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0fb836e1a6c493db1b65623aeca8bef : SUCCESS in 8m 00s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5057001621ec42198d5f7f77dd2af747 : SUCCESS in 13m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a0d66b45039545fa94360bebaffcdb1b : SUCCESS in 9m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/988449d925c944cca1f36cf98df8b760 : SUCCESS in 9m 34s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/eb9e2c5b324b43b9b1583ea53113ff8b : SUCCESS in 9m 12s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/22e29f75d8b54886b5f45986d67a0ec0 : SUCCESS in 11m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d2e6907ff28d4fe68e1cff74b9748dde : SUCCESS in 2h 03m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2d5214796de148fe9375bd72127bd235 : TIMED_OUT in 2h 04m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e8f675487db74d3798a933c4a95f7e1d : SUCCESS in 1h 51m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1ee106d0ee52421d8eab1bb711e480f3 : SUCCESS in 1h 56m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7f527d2f61084b028a3523100d501785 : FAILURE in 1h 49m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/301dbef03640460abbbdc5a417c9f0ed : SUCCESS in 1h 03m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/605dddd256e74889a22d737dcf500900 : SUCCESS in 2h 23m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f9a3a162751c4acb9b6a1467ad2c3bb0 : SUCCESS in 45m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b35663b8b4e5f7a0942dc0e9163ccf266ff607fc","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2024-01-15 05:24:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b371f3c3745aa353f7218ebfd96aea8d0782e0ad","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2024-01-15 06:01:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a77147ac66c16bdcac93cfb8862148d19b7e4a15","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-01-16 18:23:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa616c8e53f322961cfba8aabcc72f996a51e320","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 18:35:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a875618197c04f3383b1d8fbc523b3c9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0471ebaded049a69843da60e452aa48 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d1a725757da34e128aad1b8aff245915 : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/704cbd5459cf4c9aabdbd971ffd16c3d : SUCCESS in 10m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5629b7d6e8c35ed5501c976dd3d601846221e683","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 20:47:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f9e44ced8454a928a85f991697bd731\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c50ceaed2e094ef599d49385a582ab57 : SUCCESS in 2m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/1ffe4b4b07304ae59d083061e0d10c2a : SUCCESS in 54m 33s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f32abc89bcd14859957935d048312d04 : SUCCESS in 53m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/677d155f571e423b8afa17c90e2b8715 : SUCCESS in 1h 26m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e340f7103e9c44bda72ba7b1e0bcf237 : SUCCESS in 27m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/229a10790fa24e2cbad5d459c6879e76 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a41b7614a21c40efbac90db1776aea0b : SUCCESS in 6m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d6bce633f4545999e8443889489afef : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25af4e04cc734711bfbf3bb7f86a8fe4 : SUCCESS in 7m 31s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e2aef9a53d3542e08b6faf8725ed478b : SUCCESS in 14m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8a3ee7a2ee4548b6920c3aeca6d4fee4 : SUCCESS in 9m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a3378b9a6853448598a91a1b550034ad : SUCCESS in 8m 34s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9157ffc1120d41ef9678bd2ece6b0a9e : SUCCESS in 8m 43s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/92b59e30f5b844bfb4639e403a61f7db : SUCCESS in 10m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1d3a1d0ca50c47419e8a8f8258bd473c : SUCCESS in 2h 04m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9308bb8817cf4cd2825d7b3866e69795 : SUCCESS in 1h 51m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/55c3d786ba9848a1a30322214c510e11 : SUCCESS in 1h 51m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/47e9991a92a946b78f4d82b7398c750c : SUCCESS in 1h 47m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1279a32d85dc48f98a28381147acf814 : SUCCESS in 2h 02m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11f46cfcb635422a8eaa135d53c18969 : SUCCESS in 47m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7a142bea050246e1ae4b0466e838eda5 : SUCCESS in 1h 52m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/298a89e016464d52a61ae69092821f06 : SUCCESS in 32m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f67559c4aae9cd0427cf82f719f7d7a22b8bdaa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 20:47:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa0f64477c9a0f69ab6c0ab9f37a3bc773d02836","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 22:55:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/acec5172231e4c68919c0be71d8c1184\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b25fb7af0f6d4bc19a53f86aa306a91d : SUCCESS in 1m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/299cb902889e4773967f10f34fd456d4 : SUCCESS in 1h 03m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8c8a4c82e92048eea34c9f90346340a1 : SUCCESS in 1h 04m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef5405f3419b4847b9a413bff76123f7 : SUCCESS in 1h 19m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0802e7877854ff0b081c25eff577154 : SUCCESS in 30m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebbdedaf94934c9db81f9f84674d548f : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d53028a82329418e9720375822e86a34 : SUCCESS in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ccb404d813740459033d574c6c5986b : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ee4f651961b4ef19df5e24bc258bec3 : SUCCESS in 5m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/621fd6ea9b564154b39210a573edd755 : SUCCESS in 8m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d12052b816664a56bed254d12192c013 : SUCCESS in 8m 50s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1273f630cdb5491b83138a62fb0bb015 : SUCCESS in 2h 01m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6fc17f4e8a1e4fb7be7cd982f2c94e67 : FAILURE in 1h 46m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fafaf80a225a4818be5c511bf9909f24 : SUCCESS in 1h 40m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/256be0cc100349bf9da76c019068c956 : SUCCESS in 1h 06m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fcd51449559747dfbe319730b8d5d74b : SUCCESS in 1h 41m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/77e2c312e3414fffbaef3b2bf4821e52 : SUCCESS in 33m 11s","accounts_in_message":[],"_revision_number":1},{"id":"20c22073a72fde8e972adf2827f6ad6366781a39","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-17 14:22:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"779a155e4066dee8c37eb272fb6103540955e12f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 14:48:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e5a072d7e5e40f0b7a1655ebec71559\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3963993ecbf4481c8954234bd42ac74e : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7438f791636645d0825afb45ec90bafc : SUCCESS in 10m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7bf4843beba046c990784ca804a1ad1f : SUCCESS in 7m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f30f34e734d94d71b2a5567356a6b906f7fa647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 16:55:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7a84b66644c4e45bc8f78caec79ae09\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f935bd41382d4e279d9e0c8c48031c49 : SUCCESS in 2m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/4430b5ba8c884befbe6fca85839ab9d4 : SUCCESS in 59m 21s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4a341b75d311414491eb1407cabbdc55 : SUCCESS in 1h 19m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/04ca0cd0ef1349e5aee631a9e9e8dae1 : SUCCESS in 1h 28m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1b2764530584403a8d4d1bae16c6ac9 : SUCCESS in 25m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f667abdfeeb14f7ba03bf7ca619b8f80 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b47d1f06fe894706a414a7971ddc5be6 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/551e4116c5724b84bea6357a5ac896fc : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65ad10e1a554fb5ba3f7caea0efe0be : SUCCESS in 6m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5d1f20072ea746fe852be28282d0fdea : SUCCESS in 14m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d4aa26b3ce48448c85811482839d4086 : SUCCESS in 9m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e9350e19ba4848e280d117a45aa447d7 : SUCCESS in 7m 52s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/01cc1542d8d54a4cb7f1578e2a225d95 : SUCCESS in 10m 29s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/698fb964998e4c30922c015719c609e5 : SUCCESS in 10m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/697b1789b4dd483c869cee0977ce1c49 : TIMED_OUT in 2h 03m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5e006273f8a74f568aeadfdb8a2c089b : SUCCESS in 1h 36m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8787e0bdec554be484562d7d1a3f9f5a : SUCCESS in 2h 04m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/527a05dde1b74f5e8d823cfba5b7c710 : SUCCESS in 1h 46m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2467ff457970461d8c200fa4347239cd : SUCCESS in 2h 25m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d54e9ba80834e959c745aa7d6c7c979 : SUCCESS in 1h 04m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/730198cb5ee34d4381f2ccc75c247f8c : SUCCESS in 2h 00m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/26c0fadf54ac4960882bd7f396b3b8b2 : SUCCESS in 40m 56s","accounts_in_message":[],"_revision_number":1},{"id":"7c66f0ed2aa7a38c402d0b5f9137b58ff6aea87a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 16:55:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3880a332341efd1cd97ad2216cb326c935dcdfdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 20:18:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c6c8425cce7411ca3dfc768342349ab\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ead3282fd4754fd89d43080073e6ec80 : SUCCESS in 3m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/d64d9631951b43e8b30752d6bf575603 : SUCCESS in 1h 06m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ef18987cb38c428a9045544a1ff5149b : SUCCESS in 56m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9593a073ab4b4f818baf01b4ffc848f2 : SUCCESS in 1h 26m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bde79f0b521745f7a478bd73939de6d2 : SUCCESS in 28m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3720c99c66c948a0acb48e13148da23e : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b49db196ea44728a46db391d60eae97 : SUCCESS in 6m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/024eafe9f9d44beb98c73ff01cd34a91 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e3e2d2084814dec9fb6f9c75027bafb : SUCCESS in 6m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3a5c12e3ec6645c3bd0adedec0ef8515 : SUCCESS in 9m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8aece26752d149b5a152d3e4fa124921 : SUCCESS in 8m 15s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e43c625202044c01a38703d94b3fe1c2 : SUCCESS in 1h 45m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/743ffceeda614a3faa735f6285befacd : SUCCESS in 1h 58m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a90504f68bd64f9ca1f1f49aab8b14b3 : SUCCESS in 1h 43m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae5d3309b4c345f09a826542f91db2b0 : SUCCESS in 1h 06m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9a9b03e3b3894aef9d9fb3879dfc8604 : SUCCESS in 1h 35m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/47ef2ee63eb148d29b863eee2a52520e : SUCCESS in 27m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0dacf1a828c4f81bc8d3794f8fff44b8321d5580","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 20:18:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4524b7fb330fd8967f1b010f8d85d5af875ef414","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 20:19:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e9e7dfa33a84b2a9971649f90fbd580\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfb574ed03d54ccb82e29a6f325f2c7a : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d460333737e1d630f7d80ece3b248b91ac93440e","revisions":{"d460333737e1d630f7d80ece3b248b91ac93440e":{"kind":"REWORK","_number":1,"created":"2024-01-09 04:21:25.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/51/905051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/51/905051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/905051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/905051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/905051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/51/905051/1"}}},"commit":{"parents":[{"commit":"41fa8223b7dda10e2ff310097bb86700f254015e","subject":"Merge \"Add Packed Virtqueue extra spec and image properties\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41fa8223b7dda10e2ff310097bb86700f254015e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-01-09 04:17:34.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-01-09 04:21:21.000000000","tz":-480},"subject":"Update python classifier in setup.cfg","message":"Update python classifier in setup.cfg\n\nAs per the current release tested runtime, we test\ntill python 3.11 so updating the same in python\nclassifier in setup.cfg\n\nChange-Id: I703ff8cdebd2e929ab775f9b601a8542fb413153\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d460333737e1d630f7d80ece3b248b91ac93440e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d460333737e1d630f7d80ece3b248b91ac93440e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
