)]}'
{"id":"openstack%2Fglance~905315","triplet_id":"openstack%2Fglance~master~I704393892799ee0d5edf45fabe07afe01473e38d","project":"openstack/glance","branch":"master","topic":"bug/2049064","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-17 15:34:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I704393892799ee0d5edf45fabe07afe01473e38d","subject":"Fix test failures with oslo.limit 2.3.0","status":"MERGED","created":"2024-01-11 15:23:48.000000000","updated":"2024-01-17 15:36:12.000000000","submitted":"2024-01-17 15:34:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905315-bug/2049064","meta_rev_id":"c4cab1cfbe272385d1889fd3dcc12bc83f25e383","_number":905315,"virtual_id_number":905315,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-17 15:34:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-01-15 08:55:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-01-17 13:46:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-01-17 13:46:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-11 15:38:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-11 16:23:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-15 08:55:33.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-01-17 13:46:36.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"af2e37c22f653ec382add28ab11e16863557ff4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-11 15:23:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0db3d33ae0049707a4228590db43b7ce685a7ea3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 15:38:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59c45a49ff1b4ad3a216bae24887bf35\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd34563b4a5047f28eb7cc225fcc4001 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/adbb81d132fe41f6ab079612c262f281 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2db21afbf3c549c080ac4ded4c93fa2f : SUCCESS in 11m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62a13fc6cba4d8b678b2435b76941b53af67df3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 16:23:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/04b174e65a2d49a8894056362b4ad610\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9d5028c167d04365891f5495ddc3e6f3 : SUCCESS in 57m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ab6dbbe66e74e4a8ce0b222f9f5ee2d : FAILURE in 42m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fdfdad8eee943ef98d92609c9b70190 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8e68c42a8eb49979c63bcf99407f92e : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c247bd1eccd4c77bf7407ea5488b472 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e1be406cea74417b744d6e214b002c7 : SUCCESS in 7m 30s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e8f47cb84fb74b1fb050ff8c7807219e : SUCCESS in 14m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7fbe8f5d16904c40b8fcd324171ef011 : SUCCESS in 9m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bdb13d410aa842ff9b572c983d382bc5 : SUCCESS in 8m 48s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2830fe6344f14c47b983dd322b4d0212 : SUCCESS in 9m 17s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5b251a7f64d44316bf52f7fe1a81e977 : SUCCESS in 12m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d5dc578f13ac4ae3863932ead93f9984 : SUCCESS in 41m 01s","accounts_in_message":[],"_revision_number":1},{"id":"08e209a13c781ac75cd8fbf6855dc33c00e15c7a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-11 16:27:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"175e64c23e96c89cd5b934b8f0e29152b5fc37de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 16:41:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d81b04d1dc9b476683aae3b964aa4221\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9882efca1fad43028e72daaa80acd261 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f66b4ec847924fb1927beb088f9a68f5 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1c977fc3e03f4387a16540a7d4c216fa : SUCCESS in 8m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b67e5ec88099bb136e4919b74da365f8d262f529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 17:17:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6cb0e820a874daabc746cd049817a02\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7831456209974e2abb631829cc48f7e3 : SUCCESS in 44m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b1be93b83604219a6720959854ef12a : SUCCESS in 24m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b14aed425b854c45a050700f8e37d97e : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3aad362317e4a9ea6fab27816181fcb : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2aee86b530554a1d80340c81e47e4a60 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7500cade4b1b4996af5edf1fb73d8688 : SUCCESS in 5m 55s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/13a7f827c48449acbae3fe8ce5d74b19 : SUCCESS in 14m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/78994f45d2694fc8b7a80f30efbbf6c9 : SUCCESS in 8m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d1ebb594797a434cb3b394f601026f6d : SUCCESS in 9m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/979fc376ff6742178514656f1bd82689 : SUCCESS in 8m 43s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/725207c9c8564729a07825fcd60997ff : SUCCESS in 11m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2ed9e6073e314543a4e6e73e97f85f65 : SUCCESS in 40m 45s","accounts_in_message":[],"_revision_number":1},{"id":"1c0f1fcee1bf1f9100f06d3fceb57d318bd6f9aa","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-15 08:55:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d03fc180119399530534d9d135a1506527d01b46","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-01-17 13:46:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6e1c459342ce35ee3f4ae3ba855684e747c24c8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 13:47:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d81a2df75d3894a8dfbc8064c15f9d169e1f2a9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 15:34:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d44e3d9cdc9412a82ab90708b633901\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de95a56bab2d476185772097ddd9ebac : SUCCESS in 40m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/913db3d5d4e64c5a80ecd68394f4511a : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9f615cbca134b15804e0b7550ddad02 : SUCCESS in 6m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/832e5ce296ed411083c6b98e85799fcc : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b741bfb5af7c4e05ae22233807c2b9dd : SUCCESS in 6m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/53681450a0ce48d890ebdb8dfa5c25b4 : SUCCESS in 9m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0d6fba9711fb48189142bdbb125aad8e : SUCCESS in 8m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d35119cbfe724290ad23e1f92755f3a0 : SUCCESS in 43m 00s","accounts_in_message":[],"_revision_number":1},{"id":"73ef613fad416bf9ca7f05be7d26c4bed804c2d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 15:34:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c4cab1cfbe272385d1889fd3dcc12bc83f25e383","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 15:36:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffe70321e59f49f1adb2269e50ec74f8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a18a583170b842e59107775c1a43b497 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5378396b84755dedbba2a34d2e9c93add8c27176","revisions":{"5378396b84755dedbba2a34d2e9c93add8c27176":{"kind":"REWORK","_number":1,"created":"2024-01-11 15:23:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/905315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/905315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/905315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/905315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/905315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/905315/1"}}},"commit":{"parents":[{"commit":"41fa8223b7dda10e2ff310097bb86700f254015e","subject":"Merge \"Add Packed Virtqueue extra spec and image properties\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41fa8223b7dda10e2ff310097bb86700f254015e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-11 15:20:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-11 15:20:42.000000000","tz":540},"subject":"Fix test failures with oslo.limit 2.3.0","message":"Fix test failures with oslo.limit 2.3.0\n\nThe oslo.limit 2.3.0 release contains the validation to ensure\nthe [oslo_limit] endpoint_id option is set[1]. This fixes the test\nfailures caused by that validation by overriding the option in all\ntest cases where unified quota implementation is used.\n\n[1] https://review.opendev.org/c/openstack/oslo.limit/+/897509\n\nCloses-Bug: #2049064\nChange-Id: I704393892799ee0d5edf45fabe07afe01473e38d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5378396b84755dedbba2a34d2e9c93add8c27176"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5378396b84755dedbba2a34d2e9c93add8c27176"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
