)]}'
{"id":"openstack%2Fglance~906317","triplet_id":"openstack%2Fglance~master~Ied8c4366eb1cc2ebf83a96bd824631a4d8af3a00","project":"openstack/glance","branch":"master","topic":"deprecate-location_strategy","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-02-15 16:20:27.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-02-15 17:48:22.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-15 20:32:04.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-02-15 20:32:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ied8c4366eb1cc2ebf83a96bd824631a4d8af3a00","subject":"Deprecate the \"location_strategy\" option","status":"MERGED","created":"2024-01-22 20:34:13.000000000","updated":"2024-02-15 20:34:02.000000000","submitted":"2024-02-15 20:32:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906317-deprecate-location_strategy","meta_rev_id":"90eec60ece0b6bd10e32bb2fa20088ea538e419f","_number":906317,"virtual_id_number":906317,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-15 20:32:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-02-15 16:20:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-02-15 17:48:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-02-15 17:48:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-02-15 17:48:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-22 20:40:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-22 22:47:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-23 06:22:32.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-01-24 02:24:27.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-01-24 05:16:27.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2024-02-06 15:11:22.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-02-15 13:59:13.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"3cece3604f8e68b7d15b56cc70ab16bf14f1468f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-01-22 20:34:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e629a86bb08de34a18b84ca571ac3ac5a6169fd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 20:40:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3fff943f2374c0f8f46b5fc1cc2c356\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/47013a5761df487fbfd12dcb977a40ab : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a258ed0fbce64aef84e7c28c9b857edb : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3367eff7236f4276a4679a531569285e : SUCCESS in 5m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"20b2e84662b732faf081d6825a8e9ee2fab0637b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 22:47:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a282870b0557404bb1dd80e5cf3b1559\n\n- grenade https://zuul.opendev.org/t/openstack/build/670ce05cbc434f54ba6d324abb925660 : SUCCESS in 39m 57s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9036aab175194a1ebb4bf992c541bf4b : SUCCESS in 50m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/65423e17b15f4ec88f4f2235c670de5d : SUCCESS in 58m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db72076aea054c1bb41efbfcf02bf8e9 : SUCCESS in 26m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe9636b7777b4c75a3ecf93db9fbfde7 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6cfe867e8b84d1b8abbf0f43d6bc886 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/813f94c1a8784665bed349adddfba772 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27e356bf0b12428bb32ea0e008a5e275 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78a69b4850ea4cd1a6b0c1d4c2b18b93 : SUCCESS in 6m 47s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2db7f24413084ac19daf424c7856bdbf : SUCCESS in 13m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8da1771ff30145e4b958f65bb7109c30 : SUCCESS in 7m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8f4c96df818644f896fbbc274b736d22 : SUCCESS in 8m 45s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d4170b47b7fb4b8a9f99956ac612f0a1 : SUCCESS in 8m 29s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6880e16571cb4068b3b040962e632777 : SUCCESS in 11m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/21a8219270724d30a6e5c568386d5c0b : SUCCESS in 1h 54m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d6cabca9b5964a65b943be1c157a7ea1 : TIMED_OUT in 2h 04m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/61abf158736d4fba8031d8f595569403 : SUCCESS in 1h 52m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4a13b712d50e4179a75315f5d06f57d1 : SUCCESS in 2h 01m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3c691205c2fb4ceeac1c1bb29f935e4a : SUCCESS in 1h 58m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf802b9437b440dfa89785c3951d6a14 : SUCCESS in 1h 00m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ed9de44f329648b6bbc5c0ac918b54d3 : FAILURE in 1h 42m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/89977f5ec4754ff0bddfe173cbab8787 : SUCCESS in 23m 22s","accounts_in_message":[],"_revision_number":1},{"id":"225e552c19bd9ba663b0b337ceaf6465eca05f74","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-23 06:22:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a0f850d19cd19fadc832d6afc2b9ba1646c25f70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 06:36:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c47e8ee4a304198848d1e84a70708b4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2fd74b2d6aba46cd8f73e53716d07591 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b6440bb2e2804f14a50c5dbf72b64353 : SUCCESS in 11m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a7d10c80757947c09438fae4fb6b08c8 : SUCCESS in 5m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"642f0cc2439ca95640530bb7fa6bdf95bd70eaeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 08:27:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d89d3fc65f6445788759ace80739f72b\n\n- grenade https://zuul.opendev.org/t/openstack/build/3429913e394949f4baed31b0dd7dfc3a : SUCCESS in 55m 40s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2c9f6fa4ad18415fa5fd7cc721a97d1a : SUCCESS in 1h 05m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d05a6d76a9e24a40b813c65595065556 : SUCCESS in 1h 34m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfc2dc467172469689d099d9d852910f : SUCCESS in 36m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e29ab8212e540e3bfdced5e482785de : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d07ee9b75bfa4a8ea44d7713374d227a : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c051899126584e5581346946916243e2 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6699e95bc6f245bba5dddc3d34178696 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0aff23f8e5674d6d964a365e9e481c5c : SUCCESS in 6m 28s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a7bc8c8c8ab14264a2ce0b4f0dd06e3f : SUCCESS in 13m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/19558b5de9144575a667a5eb12b6ea2d : SUCCESS in 8m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cfd555b1c4654a8bb11d29df64f5157c : SUCCESS in 7m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ed32e02c7f864d858ac9a4ddc0885b00 : SUCCESS in 9m 40s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/cb654fc23b7345b992a36cffab97b17d : SUCCESS in 10m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5e409c85ea4540e4a84658bbad8026cc : SUCCESS in 1h 58m 50s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/40db99ae9e3d45dc8515b8b3d293ed75 : SUCCESS in 1h 22m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/495562fe9a6641afada2a9c67fd46684 : SUCCESS in 1h 29m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a876aaab84214eb3bf79b16bd253a76b : SUCCESS in 1h 25m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/14bf0ac47ab84cf3946a4f9cd0095ab5 : SUCCESS in 1h 39m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a703846cd7a24ce4833438dbb0e1e403 : SUCCESS in 45m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/318943e3468d4defa1b7a86a860f5108 : SUCCESS in 1h 49m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e1cc099add1c4482b688efdebd6e86b7 : SUCCESS in 26m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b597240ebb1ca2da1a36b3f24a3e3dba4d11c5a5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-24 02:24:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fe7063bbd71491ec1cf343fd81073520ca19f0c2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-24 05:16:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4eb2bfbeb176bfd3b977fe14f50245891286dc9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-06 14:48:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"80ed4f1c10420d0374cab4667d8a36bdb1638d3b","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-06 14:48:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"05231053549351fc6f40c54a73d97c18982b7f62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 15:06:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18673fb077544108821242b384f9123b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/21f5e49c58ea42b58266802681481fb4 : SUCCESS in 14m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0371346b793c41cdbbdeddfcf181b12a : SUCCESS in 16m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6c89dd01b89a46b083dfc5de18e3e3b6 : SUCCESS in 4m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ecbebb15e5f97e8f27ab799b6323f1e705b5f3f2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-06 15:11:22.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c64c441753ce3c531a7aab0ef4892c93e06e207c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 17:08:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f91e36fb80ba44a183b337047e661b62\n\n- grenade https://zuul.opendev.org/t/openstack/build/543c58bd13fb4292a6229f204ee9a93e : SUCCESS in 49m 44s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0b8eb6f4ebe8406caff58f215168481d : SUCCESS in 1h 04m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/12e7f22f14cc4e4ca5e8f49428ba43f7 : SUCCESS in 1h 32m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cda19c7da734272b8c7a1710a303202 : SUCCESS in 23m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e32b4dc49b29461c9b82cfb6ae20354c : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4333c75055e54d7d889266025d376b4e : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/253379363d484f37852d257f151887ff : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7440649817a24bf7bed0552cfe2c89c3 : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb64805eda8445c4b022ba4d9cd6f297 : FAILURE in 6m 00s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/885203702ace43079f0584501490844f : SUCCESS in 13m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7a4f038e143c49bba1a1001362271471 : SUCCESS in 10m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/85f80a7d7ad941a88fdbe0d3971bbf78 : SUCCESS in 9m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d875ed8febdb4873b06675ff12846d27 : SUCCESS in 8m 55s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/e2c84ff341024a79af0e730c6453ce56 : SUCCESS in 12m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b7780841667e46fd9f95328e34a49026 : SUCCESS in 1h 58m 45s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/55b7323764f9423ebda1171aa42c3d09 : SUCCESS in 1h 52m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d0c40e37ce114c28978d2e419f813e34 : SUCCESS in 1h 44m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/21bbc518c6ea44d8b7a9fb9ba6f61d6b : SUCCESS in 1h 44m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/28be49abd175402ab47eb037c87dba6d : FAILURE in 2h 09m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fe3628b517a4d06859b2aabaab75b74 : SUCCESS in 56m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9d78b97ffe864de0979c8072545ec07d : SUCCESS in 1h 47m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d7cc8bf628554df48a971aa31f13cf64 : SUCCESS in 38m 00s","accounts_in_message":[],"_revision_number":2},{"id":"be9ed6d06d8b8e14fd74aaa42f981cfb1d5d62ae","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-15 13:59:13.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"584a3483ea2519c27474c53b620fb9eb7f1ab5ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-15 16:17:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c64924e331e0862c1dd9e97fb54937002e191241","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-15 16:18:15.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f23ae9265761e3bfc03df5ef8c564aef85f183c8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-15 16:20:27.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"99ba53dd3280c1e97da78108ea2a43cb66e80d91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 16:25:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d136793a2de44b3c8848736ca4bad2ae\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/63804323d56646b9a9fef0658ad1aaa3 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3404457086854d23b48865355c6ab7d1 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eede6b9b50c34c869dc2686ffecc9cbe : SUCCESS in 6m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"29151565186aba967da895a9fd6c1237125b6a10","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-15 17:48:13.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fb182bf4bd994bdf8472e16fe5724486cdad8a6c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-15 17:48:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"03e04ce7955b1a26d14768e7b2283bc3265b8446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 18:25:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4da8fd88ccd49eeb3b0da4abb4f232d\n\n- grenade https://zuul.opendev.org/t/openstack/build/d61438a60d4847efa146fdecd6df1fe0 : SUCCESS in 1h 00m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/09de438465e142549ea92a44d07eb001 : SUCCESS in 59m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fbb508f8d7fa4f608ab2488658324447 : SUCCESS in 1h 22m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62937ec55db74f00969e62415c648457 : SUCCESS in 33m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a6840cc0e9c46139f7eb4838281832b : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0aa0e6cd4eb64a8e942f92a01d6a40fb : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1157a97fef654a39822b5919602c4fbc : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e9777ccf52448f39a13c309a38b8fee : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d35ceafbe37843b78bdcbf28c96c0504 : SUCCESS in 6m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/402dbc8ea0d0427b957572ed2441d073 : SUCCESS in 11m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/adb8f8c9fcde48fead43bde215e93190 : SUCCESS in 10m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6d5c5662fc594e85889f838f9894a120 : SUCCESS in 11m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d21ca000c45f4eae87bf573bd6aec46a : SUCCESS in 8m 31s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/33a581c323464094a72f67e5a7c7fb38 : SUCCESS in 12m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/99d1b411bbd643e8a9557c1719acc6ca : FAILURE in 1h 53m 50s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1a5ec1e04a7f46cab963cdb1ba5f4910 : SUCCESS in 1h 35m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3a245d21f9bf4f49af3e6fb27cb258c7 : SUCCESS in 1h 40m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e2303f4351624e589f898b9e761910f9 : SUCCESS in 1h 40m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0264804462c4451b9e13c7023b1c8c14 : SUCCESS in 1h 43m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c82ea38b1c14131ad446764eb19afae : SUCCESS in 58m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ac0e4cb1f6aa40fd8a1ea4276c6c646d : SUCCESS in 2h 01m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/877b66fdcc8b4cd3957215d4cd865882 : SUCCESS in 29m 30s","accounts_in_message":[],"_revision_number":3},{"id":"4cc01887eccbdf40ecc90afec3df7e57cd124b33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 18:26:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f5d54c25431af09ae631691933ec71dd76054b87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 20:32:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9636522072934c319c8e6554efd41651\n\n- grenade https://zuul.opendev.org/t/openstack/build/2bc0dca5b4b140849025d9a42498fe20 : SUCCESS in 58m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a2fe939f31204af7976331e03916aa57 : SUCCESS in 57m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f85584309ed24b0aa7910e040f8ed36a : SUCCESS in 1h 27m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b309baa10e514615bbb505e01c5eb01d : SUCCESS in 20m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33eaf31fd6cf4b8f91881730726006da : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ca2f3fd115a4af8813f24311db0e84e : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/127c21ef8dd44670a68a118f439bacdf : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01c44e021d4f4337be9f868b1fb5dd10 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/293edca455bb407cb9c4f89244a1ad00 : SUCCESS in 6m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/23dbacc190cb402c8a1d0f8a58c6b381 : SUCCESS in 8m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/420174c768434e1e941103bedd99fb46 : SUCCESS in 7m 58s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9805c944bbdd4a369cae55e5c71c5a74 : SUCCESS in 1h 59m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1d4e25ce307743f2b1dc7d82335e2efc : SUCCESS in 1h 51m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/01c081a8850f46a7852539ad60d0970b : SUCCESS in 1h 56m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4c9e65413cb45c4bec270bec187ac6f : SUCCESS in 42m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/991505cc164f43b8b54b56928df990d0 : SUCCESS in 1h 52m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b7f73898295e4a959a768dd6b768e73e : SUCCESS in 38m 30s","accounts_in_message":[],"_revision_number":3},{"id":"5bcb83a12254c5a34bda0bfbf41c95d849201909","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 20:32:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"90eec60ece0b6bd10e32bb2fa20088ea538e419f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 20:34:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b2cb15e8e384beaae0d05484185b5be\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdc2077b8f5340d3a31abe64bd3ba48a : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d643d137a344faaa6c2470ff6bf83ec : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bd70adf049f38ffda4b0d79f06930180cfa9e7c6","revisions":{"1813ae9656c3f0ecfc0675fd61872e196138d88c":{"kind":"REWORK","_number":1,"created":"2024-01-22 20:34:13.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/17/906317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/17/906317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/906317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/906317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/906317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/17/906317/1"}}},"commit":{"parents":[{"commit":"f429b53e0474553992f43c7045498156ee36d5cb","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f429b53e0474553992f43c7045498156ee36d5cb"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-01-22 20:33:29.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-01-22 20:33:29.000000000","tz":60},"subject":"Deprecate the \"location_strategy\" option","message":"Deprecate the \"location_strategy\" option\n\nChange-Id: Ied8c4366eb1cc2ebf83a96bd824631a4d8af3a00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1813ae9656c3f0ecfc0675fd61872e196138d88c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1813ae9656c3f0ecfc0675fd61872e196138d88c"}]},"branch":"refs/heads/master"},"f78288f902af3a2e75226487101f6042a64d0596":{"kind":"REWORK","_number":2,"created":"2024-02-06 14:48:01.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/17/906317/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/17/906317/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/906317/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/906317/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/906317/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/17/906317/2"}}},"commit":{"parents":[{"commit":"9f896ab03d3516d05d410fc545a6faf7fb7554e2","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f896ab03d3516d05d410fc545a6faf7fb7554e2"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-01-22 20:33:29.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-02-06 14:47:41.000000000","tz":60},"subject":"Deprecate the \"location_strategy\" option","message":"Deprecate the \"location_strategy\" option\n\nChange-Id: Ied8c4366eb1cc2ebf83a96bd824631a4d8af3a00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f78288f902af3a2e75226487101f6042a64d0596"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f78288f902af3a2e75226487101f6042a64d0596"}]},"branch":"refs/heads/master"},"bd70adf049f38ffda4b0d79f06930180cfa9e7c6":{"kind":"REWORK","_number":3,"created":"2024-02-15 16:17:59.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/17/906317/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/17/906317/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/906317/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/906317/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/906317/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/17/906317/3"}}},"commit":{"parents":[{"commit":"9f896ab03d3516d05d410fc545a6faf7fb7554e2","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f896ab03d3516d05d410fc545a6faf7fb7554e2"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-01-22 20:33:29.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-02-15 16:17:39.000000000","tz":60},"subject":"Deprecate the \"location_strategy\" option","message":"Deprecate the \"location_strategy\" option\n\nChange-Id: Ied8c4366eb1cc2ebf83a96bd824631a4d8af3a00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd70adf049f38ffda4b0d79f06930180cfa9e7c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd70adf049f38ffda4b0d79f06930180cfa9e7c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
