)]}'
{"id":"openstack%2Fglance~906359","triplet_id":"openstack%2Fglance~master~I5f791c58ae857f6c553276dd9808799c3db3aa4f","project":"openstack/glance","branch":"master","topic":"New-Location-Apis","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-07-02 11:48:14.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-06-24 05:02:29.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-07-02 11:48:14.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-07-02 11:48:14.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-06-23 16:06:07.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I5f791c58ae857f6c553276dd9808799c3db3aa4f","subject":"Make location URL compatible with cinder backend","status":"MERGED","created":"2024-01-23 11:15:20.000000000","updated":"2024-07-02 11:49:20.000000000","submitted":"2024-07-02 11:48:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":71,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"906359-New-Location-Apis","meta_rev_id":"91d7d5a60e91b7342591c8d384451de250062ecd","_number":906359,"virtual_id_number":906359,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-02 11:48:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-02 07:03:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-06-28 12:46:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-02 09:13:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-23 11:28:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-23 13:36:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-23 14:46:11.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-01-24 05:21:10.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2024-01-30 06:35:40.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-02-12 21:40:43.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2024-02-20 14:48:15.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-02-21 11:06:38.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"3baab8e2354780928b44d600f3d53553b1b3decd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-01-23 11:15:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f85f3e7e1940c5b2428d5a36ab3ac30775062e75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 11:28:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cf260617acd43a8861577bf820b3b00\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a80f6bfc054f41f9b9b74ab66db63fcb : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a462d94460e94c029a7aea29a64404b0 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3fee42f2c03b4a399658e94ed5a7a1d0 : SUCCESS in 10m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f663e665b04cf26481d586cdbd3834a4fc3d569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 13:36:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b9dacf167514fbcaf9ba9b00d8ba49f\n\n- grenade https://zuul.opendev.org/t/openstack/build/4ec924d7ca4145fa9c3c9681d037b05d : SUCCESS in 1h 04m 08s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6a34577e6e42414c83cf78cda9088203 : SUCCESS in 1h 03m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e87a43a3083141f9af9b2a0e4cafd9fc : FAILURE in 1h 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2264678261984f7bbf73933aaf605224 : SUCCESS in 26m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ff734ad032e4550a5a52a3bda76a216 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ecaf8fd8e134189b31f2644d727e2d8 : SUCCESS in 5m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e790feb4a8e04e2183bf9b40473a74a2 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf394f38223744ee9dc6c269651b97de : SUCCESS in 6m 35s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8a34009e2cd3461e876f548189ca1a61 : SUCCESS in 11m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b6ba40c80fe84afa85965ba29c519783 : SUCCESS in 9m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/62e93b64190c4e07814dcf7b7cdc2790 : SUCCESS in 7m 44s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4c9c0dcfbc814f779673c0f12b2ef7ba : SUCCESS in 8m 57s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b0eb14a3acb049c7aec65ee5425a24ce : SUCCESS in 10m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c2cbd34c22de44bbb2908fb7e577ee51 : TIMED_OUT in 2h 03m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6d0bb44bc83c4dd894517cd473c333d7 : SUCCESS in 1h 47m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/83a1e55a3ff6458382eff5e2e48c139f : SUCCESS in 1h 47m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/162757d489234cb9a0aedaafa7ac66f8 : TIMED_OUT in 2h 05m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/846242e921904e5990fc3fedbb1627ef : SUCCESS in 1h 50m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70fc8e85bec84bc98fbdce5c8b7ba971 : SUCCESS in 1h 12m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fc7629c935254dcfb473677ebbb15705 : SUCCESS in 1h 49m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b0775b382f4649b39607ec153ecd5c72 : SUCCESS in 37m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6e5df914c40483880d6782ddebd2c8820fbf778e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-01-23 14:46:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6bdffa1ef8a00f3cb454a22cbc352a492c455e1e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-01-23 14:50:58.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5fe7e662cbc7207a472c7a893eb80ecc058cfcf7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 14:56:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b39b237700984b70860f3e22400da5f5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/719896f657ae41fc855d6f5c78e7334e : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a762467f88a44fe0ae299b3d09e460fc : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b88fb06649de4d09b7aa843b603266f0 : SUCCESS in 8m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a13cb3399493dca39a64b3c79f79410dd8b0792","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 16:52:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d30bb2067ae44094bcddfbc5b9b8d9de\n\n- grenade https://zuul.opendev.org/t/openstack/build/3d00d512c75448f08cab7f3e244e2861 : SUCCESS in 53m 22s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e186aa6ce85f4128b01c5cbecb4c989f : SUCCESS in 55m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/91f279a58bd94a9cb095c4c938630005 : SUCCESS in 1h 41m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92752732b07c4827813fca6200bc2967 : SUCCESS in 28m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e29b8d0af6c849d7a9a87de5b6a39755 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0618b8689af14112a1cdc5f23a5cbefb : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cfe2a9c6f4b74c70a6677e82910fbe8a : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c8dbe427c2b4e52bbe04ce7d5f49da0 : SUCCESS in 6m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4a7ba4960e1c42dcbe71ecdf0a2589b2 : SUCCESS in 15m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4fa5dd6db69d44719fca9b858360986f : SUCCESS in 9m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8c1988561c3243a4a94227b52ab6a594 : SUCCESS in 8m 27s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a301fea4ec1c4015ae773c2fc711a225 : SUCCESS in 8m 59s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/346e91b89bbc47778211d38668156210 : SUCCESS in 11m 02s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b0207809c7c14ae585ec316e57a5a23d : SUCCESS in 1h 57m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e706aa312dfd4c269f88070a846bae14 : SUCCESS in 1h 34m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cc7e317beffe4021a06b8b62641af1e2 : SUCCESS in 1h 49m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c9a67ab10ddd48f7aece77b51853c768 : SUCCESS in 1h 38m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4e9515cd055347669f80d6c72392d6a9 : SUCCESS in 1h 38m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf572063b7f44541ae13b8b43efc9903 : SUCCESS in 1h 04m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9f3a6158f854480e8a622362e4f45991 : SUCCESS in 1h 42m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/41f38deec7314763b31952188162edd8 : SUCCESS in 40m 30s","accounts_in_message":[],"_revision_number":1},{"id":"61d1c70c2a9dda9a97c6cce123ffe010b90e134f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-24 05:21:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80bad9e69588652d8dccaf7497d78d482264ffe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-01-24 10:53:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ce99ae2e2cf8b64563d57ae9d5d49ae7bb7299b7","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-01-24 10:54:45.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ce5a6150ba1e7a0c06f3ed0b4dc9ef2410d79d00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 11:06:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b811f11be6314236ae9f631ef61f7c39\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03d3992964174aada6c9c3647c702d97 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/70b6881e67fc493591b709ac5159db52 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0a7a4e454cba4d17bf0912440f610bcb : SUCCESS in 11m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"158f9770986be71930eada40d41785493411b5b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 13:09:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a07f7a376f264cf6b8650419e3ce3597\n\n- grenade https://zuul.opendev.org/t/openstack/build/e1f4c855cc91468fba91339f426e50d3 : SUCCESS in 48m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5c05d3de067b4ae2abb321459747b8d8 : SUCCESS in 1h 09m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4985ed8f99724ff0b05f48a739b59cfd : FAILURE in 1h 24m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b3c65ba9d164856acd20d954ff1c35c : SUCCESS in 30m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/365fb09f5d5b487ca803e4605591cd1d : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21f5c8fd6b6d44b8b25c38d714a04f74 : SUCCESS in 8m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/345d1d67b9894a50bac373f537c0bd7e : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df9281dd76694f54925c906b2b3608da : SUCCESS in 6m 45s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ab73f270c3c94741ab3f4feadfbfcd7c : SUCCESS in 14m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4ad2b8d81cb2420da5dce50a2cc0365d : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cde9ca3116e3490f982228aca3361920 : SUCCESS in 8m 29s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d89486a69f7e419eb8dd1f4e9c2eb084 : SUCCESS in 8m 27s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b78536c1b672401e8fe57adb3554260a : SUCCESS in 12m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f294d15724f6485d827f0d8da9830d44 : TIMED_OUT in 2h 04m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d070a68fd2b84b7ba48f3a9e62b94560 : SUCCESS in 1h 22m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/857c05d705704727a78fa2abbe598ea8 : SUCCESS in 1h 38m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2bbe3299bc5341f5b6450f18ead848c0 : SUCCESS in 1h 42m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5df89f8f09934b45adc9f44c7bcf19c2 : FAILURE in 2h 07m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef54e9447f4c42578d693b776966d453 : SUCCESS in 50m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d9e92e2753644e85bae15b4e5c1849d5 : SUCCESS in 1h 53m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7bf063316611481dafa93067be57ec4d : SUCCESS in 38m 01s","accounts_in_message":[],"_revision_number":2},{"id":"c0338e40939f1a2445f204950bbe08c9137ae54e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-30 06:35:40.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7f90c5ad61cf053f136aebd16a07e6901d0d83d6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-01-30 13:25:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9e130aa32959a445d7eb983184170af25e616ad2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-30 13:30:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8eafffa9d9bade20d99be1a826341f4a2a569101","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-01 10:33:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2bf4b059def8e9969fe5d88d2e1b8b1e7d3cbfd7","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-01 10:35:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3c7fe4438b83ea8d1156b66c590ad3d961cb3b36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 10:45:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac83aa8cdd674d17a33f96a968970ea5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a547f01a7f644109c51453041ceaa72 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cd8c117cbfbb4c1d9ca9c7cf5828b2af : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0a270f0becf64836ae0321fa72455940 : SUCCESS in 9m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"beae2efe85c317de1935e9d10d93748dc1a3e762","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 12:43:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60101393de354de5bc78bc8a8f008eb8\n\n- grenade https://zuul.opendev.org/t/openstack/build/747a3b826b074c1791ae73d205b67c38 : SUCCESS in 1h 05m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ce9ecaf196634c039ac231506ed42d96 : SUCCESS in 1h 04m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7a85629a00344a0fadefc914e2f3d4b7 : SUCCESS in 1h 19m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c2bda6872514811a6b37a140e188339 : SUCCESS in 33m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40a8321a45214085a3b81e7e0424557e : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34985853c6ce49fda549d442923be13b : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0426a5b2a1b44d6abb6cd46ad0c9be32 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e63efbca08949f9bfa7f3b9d3171d4c : SUCCESS in 8m 02s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b66650a9cd3f4594a202a8cda9ec376f : SUCCESS in 13m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/99303219e7754489b8c420b846c2b440 : SUCCESS in 8m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/88c2e305415340839460409f53cdbe1a : SUCCESS in 8m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f6e5702050f946d484e27d493db39065 : SUCCESS in 9m 33s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f14d4ff999b4455ca170217ff926abab : SUCCESS in 10m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8a9a129ad96a422d928968308c0e6fa6 : TIMED_OUT in 2h 03m 22s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1c7147ed65f744a8a63b6b0076594af9 : SUCCESS in 1h 45m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/50b05a103d8444db8f041f4890f24f30 : SUCCESS in 1h 55m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2b778ffef93b462eac5405e4239bcb1f : SUCCESS in 1h 54m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/de73409e582f4dea80da62f0dcd2fbba : SUCCESS in 1h 25m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da1d1ece006f4dddb907d13bc2c4b8c9 : SUCCESS in 1h 08m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fb7e48b90c9646e0994a5b04af0159d7 : SUCCESS in 1h 37m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8fd208c757754220ae08dce0c813859a : SUCCESS in 37m 20s","accounts_in_message":[],"_revision_number":3},{"id":"6ee0b9514ae2aaf8974f0062d495f018e62b930c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-07 06:50:50.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ca1624488700a55cd16ef0b66bb699ab2f44103d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 06:57:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee772c5165b34d4a835da768db43ab1a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cfe726c2fe014ab4a926b55677b64ca1 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b4f37b10d42143959b7dd6e3a6191d56 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ea5d6fce2cf0492ca9d2d302e3181662 : SUCCESS in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"28057904e0b4b7386a0e344ec2c78166d5a37568","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 09:06:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e31bafa3083c47b38d2ea2afe13ecc6b\n\n- grenade https://zuul.opendev.org/t/openstack/build/584be6d2fdc2422d98a6838a54416222 : SUCCESS in 1h 06m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6cd66b81309747dd8c129900838b555f : SUCCESS in 55m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bf1399c0fe05441c9cf420907ed556a3 : SUCCESS in 1h 25m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e6c15488e124515a97d3658939bd278 : SUCCESS in 22m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fafc5fbd144942cb8fbf92c9e2ab0be7 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ca6500d1a2d4599b2efdc3863534760 : SUCCESS in 6m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1eeeba5edb00467cadcffe17d3135b86 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccf3a8b0db2d45ceb2c9fcb89948de85 : SUCCESS in 7m 04s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/23112e6745014be5bee7bb630c9cb61d : SUCCESS in 12m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d863dad4b7384cef83be0da5c52e67e9 : SUCCESS in 11m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/587dc9afa9924678ae6b048a4d77b38d : SUCCESS in 8m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b28aef10da99420796417cbb93acfb07 : SUCCESS in 8m 27s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c22644d8fb0040b2b340ccc8a9d1ecac : SUCCESS in 13m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ad6fd1226b05489da2516c957682ef88 : TIMED_OUT in 2h 03m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7f6c07df17ff429291e250fd9bef53b5 : SUCCESS in 1h 29m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1484a0cc337a409a9fbb725c0416a6ee : SUCCESS in 1h 44m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f233768821354c69b8c21dbee1cdceb4 : SUCCESS in 2h 01m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7c540f1322bc4f5e84996d6cd1a7d2d7 : SUCCESS in 1h 53m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2129d284df0c454d97fa4e7691c7ef28 : SUCCESS in 45m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1e40b0c8d41e416baebaf5e94e814e04 : SUCCESS in 1h 47m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3422b719809c4a39893cfaba4243739b : SUCCESS in 27m 22s","accounts_in_message":[],"_revision_number":4},{"id":"bef7c69c0d187153c24e19e728a1b8d3f2926ab6","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-12 21:40:43.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"1087037d96ee0b7be0ffb5bf9c0bf942588d29ca","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-15 09:09:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4a459b592e9daf34db8325f17e790c0f4dc434f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-15 10:31:13.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"b14d758bae9fdf11556acdc2b3a079dac9856774","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 10:44:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9a85a2c8767400e95578500666d8883\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6cfbb6059f144c75bb691b5156fd185c : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/495d37e1ada5460bb1db32f5d3bbe6bb : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2a03f101706c4e54b0ba3836bd1c9c2a : SUCCESS in 11m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cd52ff41fee747e97dd7ac9bf0cc399f0b1eb083","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 12:47:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f2bfcb5153f42a5aef9a2ec046b5589\n\n- grenade https://zuul.opendev.org/t/openstack/build/1f735b4f0fd74eed9ef2b0f49587872d : SUCCESS in 48m 54s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/46d6da761c6d4dc6bf8e9e7c71facf4b : SUCCESS in 1h 03m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d3a14bfa3270466d8f2c052c3e107416 : SUCCESS in 1h 25m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b64413c6b09f4ae2b464d111e2b356e8 : SUCCESS in 34m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04bd28fe8040452090b835295a0d9394 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c0fd7e3fe034eb5925f57f5b144955e : SUCCESS in 6m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5021673546c84fbd8d23fcae2b02b104 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/208a5f8f7a1e48b08d06bf4560904a7c : SUCCESS in 6m 37s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3bbce06956364f5f8a3e9ec8c6d01064 : SUCCESS in 13m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/714f8c85e11846c08dd236739c5f72ff : SUCCESS in 10m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c429c369ae7f455680cfd72a9d681ef0 : SUCCESS in 8m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a529f93310a24d2690d12502b08f2093 : SUCCESS in 11m 10s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/1cc328d2ac8545f48a053b84ed7c4e4c : SUCCESS in 11m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f43b9ed487644e01985726f684a8556a : SUCCESS in 1h 54m 44s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/caa6accad91546d0a75a027251de87b9 : SUCCESS in 2h 02m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cd620e292f494f62a3849517f9c2570d : SUCCESS in 1h 47m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/880c8b02ad23412f8f62be0b4a6e39a2 : SUCCESS in 1h 26m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ada823621c934713b4fb38943366ebcb : SUCCESS in 2h 14m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/653bb06d848443f3892f0e2a2b6d07fe : SUCCESS in 1h 04m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d8a0e4162be9453282fdc15043750ac0 : SUCCESS in 2h 01m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/db251ef43f1042888161ac2d6816753f : SUCCESS in 23m 01s","accounts_in_message":[],"_revision_number":5},{"id":"7ff6be56767a140a6c75e812656ccda96d61b669","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-20 10:23:20.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"1a8e5530d018499d2b1023df6745d138b42bb6ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 10:34:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a1dc8830ff44f27aff28095d8e7b7e7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3927ab05179b4b58ab4b3ef80957cf8f : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/347a5c2b54214f2fae8553bba932a776 : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ee35186f8fed4fceb8179ade49530709 : SUCCESS in 8m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"478d3a12518a9af94fad87db9cafcaf1b4ba492d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 12:29:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5dfe6d3e2214bdfaf8a7bf35e2c4655\n\n- grenade https://zuul.opendev.org/t/openstack/build/425d80c89d994741adbd28e520688a2c : SUCCESS in 56m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f8ce3599e6e940ddafa99a941f05281e : SUCCESS in 50m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5cf9a76ae0824cd9aa9147f50edeec19 : SUCCESS in 1h 18m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ac3991fd85f4746812cf034b6dbd5df : SUCCESS in 45m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c63391cc01f541d291035724b51ec7ba : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d3521a582724920aacadea7c5074d84 : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be96190c53b54ff1ae617c57919d1a11 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68707cd60f594cd9bb9fe6a83a58e17d : SUCCESS in 6m 06s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/66b31c4631a241908ba63f6e9c84a165 : SUCCESS in 11m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/823c06e333de4548bfadfd97bf0a5661 : SUCCESS in 9m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/277ae6c9754f450eb1a33df086701376 : SUCCESS in 7m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/70edb41e4cf44d5e91c609945d03231b : SUCCESS in 8m 12s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/9ed8ab4afd0d4806a5954f90098719d9 : SUCCESS in 14m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e5c102fb8358498ab184b33331df71a4 : SUCCESS in 1h 51m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c3897682611e448885dae257b6592f42 : SUCCESS in 1h 24m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/daaafc97a8c04adcabfa1089817709dd : SUCCESS in 1h 43m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4ea7397955fb4f79979ebade3ad5fa51 : SUCCESS in 1h 49m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/199bd69cac334984ab0d420772fa23cf : SUCCESS in 1h 59m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12f5e37e7c994227b25ef7242a37d9c3 : SUCCESS in 59m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/075e6f3e27f04d5da5da5732ce163023 : SUCCESS in 1h 46m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/af61a4d2384f4aa9a6a2d39930073a8d : SUCCESS in 30m 37s","accounts_in_message":[],"_revision_number":6},{"id":"ed6a1ad6216bf548f550b3122d69cf8841d2cc5a","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-20 14:48:15.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ffcacd1c3b0a0069d7415ccc04c922926f72107a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-21 11:06:38.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8ec82f187dd86939dd411320ae7e41647de26728","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-21 11:09:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"22ee0a7de398282b6b5d00d63a7f3ab119b26153","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-21 13:32:37.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2ea8ce7257768fd3a9b71e8fe7022d794f0b9a03","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-21 13:47:54.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e06de51206443f183ce7923d29691669f55f1e6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-22 07:41:38.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"2d19cc6a76e4b8d3da2353c399e23672b1b4cc68","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-22 07:42:46.000000000","message":"Patch Set 7:\n\n(8 comments)","accounts_in_message":[],"_revision_number":7},{"id":"f97ba79252de84eb904d5079e9cb162f34b959eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 07:49:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1258c2f2faca4a0b9f1a356daaa8fc52\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/99f15aa9de454eb2921099b562cd67b8 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fc8c4d2d9fad4069838a1ba24d949107 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/898caa9fc4504302b7eea9e45e4b0ac7 : SUCCESS in 5m 16s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7930d4bb69f8abb8097c7ca99112a5f29a7ca8f4","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-22 09:23:01.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e5cd4702072632e64fbd4c5c100d5770f5e86c1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 09:55:48.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff47c18b748a44a3b9524c441ce8e692\n\n- grenade https://zuul.opendev.org/t/openstack/build/9508f2db95fd46afab84b20ae486a5fd : SUCCESS in 54m 13s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0e4b4c09f6744c42992b4be3e0f3cca9 : SUCCESS in 1h 02m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ba2c0851f43d438d83126ff4b59a1531 : SUCCESS in 1h 27m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18710e71bcd24145bcd80f34aee0dacf : SUCCESS in 1h 02m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8830723cd10494e95a79ca30b41751f : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b801eda8aa1e440fbe55df739d352d53 : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4ec9bcb421d4b749dfeae407487d0c5 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/849ee2e873564c0ab85a9fcc1b1d971e : SUCCESS in 5m 35s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/72d00d4604f344d8a98b76d4f2633059 : SUCCESS in 10m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a241eea2dbe74f5eb5a5873ef78fadee : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d7788ece664446efaeb0ddc862555f31 : SUCCESS in 7m 24s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a6c91026aeb94f1f84ae1b6d1403e0fe : SUCCESS in 9m 14s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/1261bc86b28e4318be95a3a09fca1c47 : SUCCESS in 10m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2579eb805a1b4343bd92fd1ca9b47fb7 : SUCCESS in 1h 49m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8933973cfcd84e78ae01471f3d4fb5a3 : SUCCESS in 2h 02m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e1a8e58265074f29a599bee628461858 : SUCCESS in 1h 46m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4bf821fbe35a424ebc5041524faec9ad : SUCCESS in 1h 49m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a1be179ea70c4006929c1b2f7e2ef62d : SUCCESS in 1h 58m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3cdb696906b04b7f842edddd024daf12 : SUCCESS in 1h 03m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4bfd57cdcdfa44079c9de7b96a4c6bc7 : FAILURE in 1h 59m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6175025c79ed46d5a990436d5dd28e24 : SUCCESS in 30m 12s","accounts_in_message":[],"_revision_number":7},{"id":"77a9ccb69accc2a0374642ae0ebf11b1f50a4603","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-22 10:25:55.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"f3814a0592ff6aeb6d3bbb4e7c8fd77e696d79d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-22 10:58:39.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"2ff37c706cc057b5518ed974c4f5064e61a2d03b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 11:15:59.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75ac560ed5274b7094708342e3833b2a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1cd88863afff4590a7608f7bc9b83832 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cb324eb3ce3e4cb9bd97edb3f216dd22 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aa6dc15550d04df7b6a8b9d2a1db0b38 : SUCCESS in 6m 58s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2b338b565c3fb65b25d2dbb61eb778810c302e7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 12:57:10.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/310d0877a90d4e448427313eab64375e\n\n- grenade https://zuul.opendev.org/t/openstack/build/54937805cef942478960a68f7878405d : SUCCESS in 1h 06m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/349e37af0e974df5a7754a2e3a99225d : SUCCESS in 1h 01m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bf7370b9ebf645938f109d868b5bcc3e : SUCCESS in 1h 48m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d745cba3d7b74709843bbaec0590ec43 : SUCCESS in 1h 05m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3312d2b0028c4ca98a337a0c03d0efa1 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad06536676d14fbe94bb0f1a52045a10 : SUCCESS in 6m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/90afb352b645434a903d61a58e2f5f2c : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf7654af6e284c22ad014e62503c4c2f : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88c4c326908e4582a82124b5423e72d6 : SUCCESS in 7m 00s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/962c4dee8164456cbd30009e704eb6b2 : SUCCESS in 15m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e0482924a18843c4b22c875f77c5667c : SUCCESS in 7m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/40857c2b396647b2aa74c51b109f9703 : SUCCESS in 10m 05s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/dcfca0a0e8024090b6843a7facd345fa : SUCCESS in 8m 18s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5e0eae970127418994cbe367b675367c : SUCCESS in 11m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8e1519b723fb412b94384f32fad76b78 : SUCCESS in 1h 55m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/404fd7917c954ad792e6902937abd228 : SUCCESS in 1h 31m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/57164337a93f43259b3fe56279923408 : SUCCESS in 1h 48m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9b67308cc134448698fccd0e9f2f5d44 : SUCCESS in 1h 32m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e924e15b14c04eae8b9cbbe2fc1fb89b : FAILURE in 1h 26m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4aff68e9997d4d1f9986d26cb37b5cd6 : SUCCESS in 53m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5e0c9422a7e649cb8d9e4c0718e13fa5 : FAILURE in 1h 45m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e4191fd3f9d243479e229b7cb46b58db : SUCCESS in 41m 11s","accounts_in_message":[],"_revision_number":8},{"id":"3ab5e52bc55f7fb08fe8fcc3fb138963b73dc813","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-22 14:24:36.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"882d3499fe8d478b58746b50def4225fa6693638","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-22 16:10:40.000000000","message":"Patch Set 8: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"9baa241321209454a9340cdb2328c1368a36e892","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-22 16:11:41.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2eecb3670857f860218bac733703d6c4136b0c3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-26 08:34:52.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"0c68da5c49203a377c6c959b07528a536c471c9a","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-26 08:35:39.000000000","message":"Patch Set 9:\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"0064e443617ce908f5250582fdec2f1a9d390cb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 08:45:06.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b20d9dfd279c42b3bbbd62755a8ad620\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d57a81a9987c413e8f552d74df7bfd20 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9eda1654fb474dd4a19b0116fe3f61b1 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bb30831f9f3a41e0a613f3e0acea8b42 : SUCCESS in 8m 07s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1e80795bf61eda7deb002f7f67e5eab877192c21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 10:45:42.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dfd91c8efa1846f0964e9a4a57588cdc\n\n- grenade https://zuul.opendev.org/t/openstack/build/7fd9912aef444f01b0ab8facc77ea6bf : SUCCESS in 52m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a9ecfb19741240da8c1b2c65227503fd : SUCCESS in 59m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0e75afa4fc224a49a2ec9d03091dcaa9 : SUCCESS in 1h 33m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f7bab8537d9481590fd6c92065484ea : SUCCESS in 51m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33695f1b47d2407f87a698a05e92c7d3 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8858ecf3b952416988070328195181cf : SUCCESS in 5m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/22ea590cac494aab8d5358bcfe5a27a3 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bcf2e76639547f096fc1f9cfc46b27b : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63296882f0be42c1845cfa9cf6cea843 : SUCCESS in 6m 32s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/893213af4d064b8d87086115a247d304 : FAILURE in 12m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e99dccc7e65c4a45b155d0b461687ae6 : SUCCESS in 8m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ede1850af8a34cdb95064d03ccbc7fbb : SUCCESS in 8m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/81273dce7d754e188f36ea74fb67ae40 : SUCCESS in 8m 08s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/42cd9e78bc7a41be83810b3e01a84a87 : SUCCESS in 11m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e37c7c31c1204b51b89f178cf0e26f5a : TIMED_OUT in 2h 03m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1d60aada780f48b8bb9515bdcb3e8678 : SUCCESS in 1h 55m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1ec4671803e5489caf9e83c199627767 : SUCCESS in 1h 33m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e835458d8dbd45bd89296bec43e7ec01 : SUCCESS in 1h 59m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/11d6b255350b466eb84479ef713dc8d3 : SUCCESS in 1h 58m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c654dbf6300414a90dcc58850040b90 : SUCCESS in 1h 03m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/db215d8049524e0a8d2187a80b998f81 : SUCCESS in 2h 09m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/166d29509ee34f89ac94dba9191f7a05 : SUCCESS in 30m 21s","accounts_in_message":[],"_revision_number":9},{"id":"a1a60b5ec37c3d9dd4748b4a96a5d0212d7a7777","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-26 11:07:29.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5b53175ca0bc8df79af5b5de0731154895fb79e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 19:01:34.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"8a4ca31e83497ef0a36a01a5e1e3295cfe96b176","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 19:09:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd7602eb2af94807a505aef58c8efe6e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4729b46f07f84f3e961166123e25ef61 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5d8370ccdde94382a74708367226e497 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d49ac80108b54ece8051363fe0c0a82b : FAILURE in 4m 45s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"689ee1a7d2d06cab8d4ac72d85eef640da10eceb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 19:21:23.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"adc8bde39636a26efbaf3be6c137a72983c8450b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 19:32:33.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cc82971a23a4871baa913ad79ca94e7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab8179abad9c498983888294cad56af6 : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b4a20945eb5a4cce92929729f6e72a16 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6bd91b0f32c8440da87363f93d2a50ed : SUCCESS in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"961b30326e01cf90a274819705ccb5dc822473bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 21:19:38.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0529e5356cff44ef89e4f47a61249d3b\n\n- grenade https://zuul.opendev.org/t/openstack/build/dfb580cefbf74213a72af851ce9cfb06 : SUCCESS in 51m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b8489cee0e8e4e31be91ec8b8c342474 : SUCCESS in 48m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f0740b3c0a0e4fd0a642f1165e771ded : SUCCESS in 1h 30m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf6712c7b34042779c59855dee4aa2bd : SUCCESS in 47m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59e05430b2b04646af9645a1489006f5 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c06e57201b8478cba37b41eba1c85d1 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d49f03f92414bd58073fc949fbeea10 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09817d927207481682e66d94418a2089 : SUCCESS in 7m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7011a1e3382a4b7bab8e5fc3efcb11d5 : FAILURE in 12m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8847b371e91544fdb5c0d1c396718574 : SUCCESS in 11m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dba6129772a544a19dee586bfefc11ab : SUCCESS in 9m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1ba46b02588a4acdade83615d1ef7976 : SUCCESS in 11m 09s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/945e3a6c2c92448c8f4192340e5cf0b7 : SUCCESS in 10m 11s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/76951e66e51846b683ceda234a2b7ae0 : SUCCESS in 1h 51m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5fc4dd9d2f794ffa84186d1f83e6fe41 : SUCCESS in 1h 27m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f09ac32328014c55b6151fe89adb2555 : SUCCESS in 1h 35m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/be5d6de9ed294d42985c2eec7646768f : SUCCESS in 1h 41m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ca928715007546bf956e9ea3807f22cb : SUCCESS in 1h 51m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac7f3da3e97c439e8b619c0fccf74e74 : SUCCESS in 57m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e9068477d3b74223b42a0f32766bf0f2 : SUCCESS in 1h 46m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/57192750d3e84565bec1facf7a737af9 : SUCCESS in 27m 31s","accounts_in_message":[],"_revision_number":11},{"id":"63305fe1b40f765af05346b959534aa43bcd149b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-28 22:06:14.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"75569a9fff7d942fae91eb44700ceabc23b096b4","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-29 05:08:16.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"6532255362be3a83309fb67b97cfbb4aca350302","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-04-04 07:00:07.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"0d62db0b6df94f21985b1052ffd727ef12cfad7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 07:06:07.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a07334e36704d358cd9947b80bda1e7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1958307466c04de185207df88d7944ae : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f8156ca8ac364efaab1ebf61cd9dc5d6 : SUCCESS in 4m 30s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"fce105a9ee9b7735affa817cc9fa52bdc175c6b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 09:09:45.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d08aebb301a40d7a8f49791427385de\n\n- grenade https://zuul.opendev.org/t/openstack/build/4cf0d3febbd44c37bcd48e2cdcf2b8f6 : SUCCESS in 1h 10m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/19a3754ba1d5402f8b50c758ceaedcc4 : SUCCESS in 1h 15m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06612c06b36b475da741dc321c261172 : SUCCESS in 1h 04m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f5b06fdc4f14ce1872dbe8af0f1d84c : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0affdf1ab0df4e08b9a008351ee65bfc : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34ab39bc542344599f00c7a8e04aab65 : SUCCESS in 5m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2b8f76e10d164b8a853e689807ff44b5 : FAILURE in 9m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e863e04a87641f8824a9309e30a286c : SUCCESS in 5m 54s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5c8b59af0b26412e8c8d7b9a405aa9e9 : SUCCESS in 14m 04s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bee0a5fcbd5a446282ba11ee383c99bc : SUCCESS in 8m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4d638f1bec46463fbb27bba58403f28e : SUCCESS in 8m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9b0c7aab1d3641a1b92d2e2283d64145 : SUCCESS in 8m 31s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/d2d8802cbf7e4240bdc9820efe8492a0 : SUCCESS in 14m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2f12e54baabf473a8b8d17e8215ae2e9 : SUCCESS in 1h 43m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7dde4c8d77164433b2e04f6ccc76d81c : SUCCESS in 1h 21m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5a62789ac04f410d9f1e162015929563 : SUCCESS in 1h 20m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/84468c67891d4c399c61428341b25d7b : SUCCESS in 1h 41m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dcd41db16f31445f9f96ed9f4568731f : SUCCESS in 1h 52m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b958b443d3544e780db0f452a77d18c : SUCCESS in 1h 01m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3ba2a09d7f3c4410bd330ee06b395e17 : SUCCESS in 1h 49m 24s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/4690b4d6b7f34a6db467a4aa08d6b2b4 : SUCCESS in 1h 40m 34s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1107519f37df49a89027e2f1d2db2449 : SUCCESS in 1h 13m 16s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9357ac62926744619df18fb8d70cc755 : SUCCESS in 35m 42s","accounts_in_message":[],"_revision_number":12},{"id":"132c9aa00804650a39db540cbcad1ff2e77adce1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-04-30 09:42:34.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"217db0cbd0ef2ba10e616dad09c2a637ec7d6402","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 09:51:08.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82705de809cd482091316ec9f8a9b8cd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/97ac5bacda2447b48a9f620b386181be : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dffa0ebcefb246219408400d8493224e : SUCCESS in 6m 00s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b16895cca0ce740023097d9886b1f89866319596","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 11:55:20.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/355bb550d6ae43c39423f79362ab7374\n\n- grenade https://zuul.opendev.org/t/openstack/build/a426142ba5a640728ac94cdf739e62d6 : SUCCESS in 51m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9da0d18366ed459ba3aad07f412d6a6d : SUCCESS in 1h 20m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fa292b8177a4164bd270cad1a46b54c : SUCCESS in 1h 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/920b5e082d954165aeac9f1dc3669ec8 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b39f61852d44325adebbd041e493890 : SUCCESS in 5m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ca711e7d95146d9a42adaa397ecb0e3 : SUCCESS in 5m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45454815fe984fa88bd5a5c451372c87 : FAILURE in 9m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cef31f05c4864bbba4aeacf98a6be60d : SUCCESS in 5m 51s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/304a0bb35eff49e9aa64a626e1988741 : SUCCESS in 15m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/49d5afc591274bd7807765c05f2066ca : SUCCESS in 8m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a63e68d14c864124a31939ae4a7345db : SUCCESS in 9m 48s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9ff68b2529c14e24b98a95f13fa00f5c : SUCCESS in 8m 32s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/13bbcc492e074c3890acec097fbdc2c3 : SUCCESS in 10m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7c72239bc1604ef8a3154c37a08eafe1 : SUCCESS in 1h 34m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8c9e9b75888644cc93a043fe6dd647a4 : SUCCESS in 1h 46m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a2fc43ffe7b746eabc6a65e3fa648494 : SUCCESS in 1h 36m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/09560863eb3445e0af4b434924b9554f : SUCCESS in 2h 01m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/86a5572159c64ac7a5dab827c11cfc81 : SUCCESS in 1h 55m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d757b2c6170a405b9d107029a40f47f7 : SUCCESS in 53m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3869fa9af30943eaac2bec9f682eb363 : SUCCESS in 1h 45m 46s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/4dd5311ed14440d69bcbfc786615e0ec : SUCCESS in 1h 44m 51s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/332dcc4a04ac43bb9dea8df683722a5d : SUCCESS in 1h 15m 43s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/23827065959b4b9096f4f6d63709f44c : SUCCESS in 36m 31s","accounts_in_message":[],"_revision_number":13},{"id":"058e170651aeab569fe366e9c062ca6ff1f5d30a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-05-07 06:22:18.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"b284a8dd64cd29c290dbd582ba6461bff83acc1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 06:29:49.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8be1dcecee534c8aa7acef453ec452af\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d8313ff710134ac6ba4aa3ff56309fca : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8cc29a7cdea043f6b391d96a1ab251e0 : SUCCESS in 5m 38s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"0bf62de882290ffd544e2e054a3e552625f74920","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 08:47:00.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/61f967f108b7451ba011952a9bac6cd7\n\n- grenade https://zuul.opendev.org/t/openstack/build/ab52144937dc4f51a7a2a0706a4235b7 : SUCCESS in 48m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/46e233d009c44940b1648984bbfe420d : FAILURE in 1h 23m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb2b9de557724715bd1d607e5ecd751f : SUCCESS in 1h 02m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd94b29b50ab4db7b74c750aa937f999 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fb639f94c4548b1a806ae273400be79 : SUCCESS in 6m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b60304f69ac24f1a9ff5b1b3c2c78e67 : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69c705f18794470b9f1975892c8874c8 : FAILURE in 12m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/099b9f3bcd274a95addbc08c9ffde8e1 : SUCCESS in 7m 02s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7f3d8fca1a48487d9477e9761147a8da : SUCCESS in 16m 45s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5d8bd07e3e0e4765864f00d43c2f030f : SUCCESS in 8m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c71cf3a2846841e5866de87f59944b5e : SUCCESS in 8m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7b728e0b0fde40e79d899a4eeb4bea10 : SUCCESS in 9m 34s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/078ee850622e42ecb1d66f988cb9f819 : SUCCESS in 10m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/17413bf2cc2141fcaf136d2ac3136c3f : TIMED_OUT in 2h 03m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e7a6278c9be54e8aa1c161eb4a70e5cc : POST_FAILURE in 46m 31s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4f45fe91f02c4b8d87990e7206f3a791 : FAILURE in 1h 34m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/11cc80dcdecd45e89227eb93372d0dce : TIMED_OUT in 2h 03m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5e958352136d4b3faa791df0fc8bf1d6 : SUCCESS in 2h 05m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2858ed2b46784df6bb204645c8af84f8 : SUCCESS in 1h 01m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f99c5354f7e2476c8af5b835a6598966 : SUCCESS in 1h 47m 13s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/39f383aea40d47f484348a443ef560d7 : SUCCESS in 1h 46m 06s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/04d176ed1e0448f88476cb2703fa2f69 : SUCCESS in 1h 11m 26s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/91999678f059494ab19dad0d2793c45d : SUCCESS in 43m 11s","accounts_in_message":[],"_revision_number":14},{"id":"775353765cba88eb714128ef30b45cd277c0ca71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-06 06:46:59.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"1a364b55b8636846ff5788577a5e9c8be77f593b","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-06 06:49:31.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"86378bcea62bfc35c0b3d280e09f101323fd8817","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 06:59:59.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c6fd7d742584f4b953985f5313e2248\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cbb3d7f907a545ffaf319b28b28daada : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/62fe4eb6cb874e4c9c5cac14e0e95a61 : SUCCESS in 8m 23s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"1da394a09aa75576d14ad868873009ebae1791d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 09:01:47.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/486ba16b9a9044018da8778dff14537c\n\n- grenade https://zuul.opendev.org/t/openstack/build/9282dea0371c4b0e986dbdfc4e0360d0 : SUCCESS in 59m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9fc86e22abb740b28e4b3db8a133a2b1 : SUCCESS in 1h 30m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63f3368725214f9eb50da7c36c29347c : SUCCESS in 59m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac6f786868d54ceaa3f36f1f5d69afd2 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78473f454bcd4ec790a8360489eb514b : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/57c0dc7089b94f59baf5639f20d26563 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/33a0eb14b013457b973d273f4baf8d4a : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d00f826b14c14c1d87841358707bafb9 : SUCCESS in 5m 28s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/98f3a26cbe7c4e76b38203aaaf4ff3ea : RETRY_LIMIT in 1m 55s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7631ecfe65b54ea3b5fae1102840ea20 : SUCCESS in 9m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7cd8f44dfc514d04a6d8fa6fc47ea6ff : SUCCESS in 7m 48s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2aca2fde2ecf40c5a7e01de1fd787030 : SUCCESS in 11m 26s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c045f3102f4340c89734a7a2efa8a97d : SUCCESS in 9m 03s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f0da40b5c00342c688c32e98ad90d2f1 : TIMED_OUT in 2h 03m 44s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0a1ca3f1740d46199fff41307b40c87d : SUCCESS in 1h 58m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/36cef0cfe13f421183381bec5e0105a0 : SUCCESS in 1h 45m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e0e4f373a98c478181ca109104342879 : SUCCESS in 1h 54m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2494568bd91344e6ba49c7a7a4eb3ccc : SUCCESS in 2h 09m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60119a57f35e4c00874ea96c49825964 : SUCCESS in 1h 01m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0682f572ed764d7db2cda4faea1f7aa8 : SUCCESS in 2h 04m 43s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/47c0847637634ca799428022528db07c : SUCCESS in 1h 51m 54s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a60247f7c0ff4534924d4469074c858b : SUCCESS in 1h 01m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/04c9bc89636944beb35eef57deda7af9 : SUCCESS in 36m 36s","accounts_in_message":[],"_revision_number":15},{"id":"b5ae9b1beb34561a75ac1d723c370fd53c09f620","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-06-14 05:32:30.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"1e21769e2339fc32ce8febb02c28ad9fd4893273","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-14 07:53:49.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"91541a49e75292e77c8e81d320053982235f843e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-06-14 08:15:56.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"d9de431402519eca5953c4862f5fb76b2192a430","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-14 10:35:46.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"c62bf939cd8da1abed2ace821c7c746b9d05ee72","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-14 10:36:36.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"5880cda5c62ecced5ab362136a4ec4908dcd403a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 10:46:28.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14c39b434aa14bc488c4e992ead42299\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8395679acd694952bd5894100480e388 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/71ca76058d0d469197b8a54fb860d6af : SUCCESS in 6m 34s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"01ee15dac37852cb6d7a7bdad3c1586d3105a921","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 12:46:40.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b942b130ceb644c589a3e4032151b165\n\n- grenade https://zuul.opendev.org/t/openstack/build/42d5ee07a3cd41318a7dc6d1d17a54ab : SUCCESS in 1h 04m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/516d605dc3644719a90e80d54c994e32 : SUCCESS in 1h 31m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34fdd14b952e4fe4b7e4668964218b03 : SUCCESS in 1h 04m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01e0315985f54c909aa2a203cbf520f4 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa79adaab34c4500a702233846dde488 : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d8e0638929747c8a9218b80294a505e : SUCCESS in 7m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8fc4cf2d483b4c2194119585a4edab84 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a373da0c27c54736b4ff546c8eea255b : SUCCESS in 5m 43s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9dc191d67fbf4663b3792bc46faacca1 : RETRY_LIMIT in 2m 36s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1962c884f8144e35b87fff1b24fca02a : SUCCESS in 8m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/970c105c2c1445038277f156534318dd : SUCCESS in 9m 30s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1cc0e93d20194ad4919181cad63a43a5 : SUCCESS in 11m 19s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/077f3e044c924f0ca90f36a4a1a04c69 : SUCCESS in 7m 48s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3ebee3fa4b334845bbe3b8c9061d50e6 : SUCCESS in 1h 59m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3b3542e43af74eec942e66462b9011d6 : SUCCESS in 1h 53m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/64ab553431eb44bcb18aff229ef1c156 : SUCCESS in 1h 58m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/951723a136034b8e971f74fd39a6e9de : SUCCESS in 1h 55m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/068fc74e128841639e4229ca3afb2917 : SUCCESS in 2h 01m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e83e24dd82c44af0bcd3a594421e958c : SUCCESS in 58m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/383b8e0edd0a47008173638af0b97ddb : SUCCESS in 2h 06m 27s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b0816251f736442f8f438ada227c4612 : SUCCESS in 1h 55m 32s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3aaaa9ef05d342c68286fe53690953ec : SUCCESS in 1h 08m 08s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/956a5eb7a41445f1bba48ca36a84dba3 : SUCCESS in 33m 41s","accounts_in_message":[],"_revision_number":16},{"id":"0765c3144f4e0b07e5f431647f0b58548c410246","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-06-17 18:52:53.000000000","message":"Patch Set 16: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"f10c3fa6b80fe22c7db48df7301501add4a74188","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-20 10:58:02.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"afdee52b647c0b15664a77c0df7ead3879d1368e","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-20 10:59:01.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"b89a314126bc7698f93d9552c27e3bc62ccfb54e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 11:05:13.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c2ee9c202ed46ddaf491864bb5c0328\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab91694fa3fa42c390b132db36b7338e : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5934aefb83e1401e93ef57377d85b8b5 : SUCCESS in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"2439ed7d07d8eb7f285d1a4cb5f7756e396376ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 13:35:06.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/235a01d15d5a4327970c9873a4d58177\n\n- grenade https://zuul.opendev.org/t/openstack/build/0b9ba6c7bb9e4a53bf467f3758f27905 : SUCCESS in 56m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f2d4062971f24a01acccc06c9794caf1 : FAILURE in 1h 34m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66f25e70bd714d529f1a87cb837eff00 : SUCCESS in 1h 07m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6422ff6b7f3046a0b032681f0a399c02 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98619e41849b4191bb92c961b05f4993 : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d90e50f0b2104c959a59af795cb102b0 : SUCCESS in 4m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a745160254b344c88fbd3ad8af4a1bd5 : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98837ada575b4893901d67b7e82f7621 : SUCCESS in 5m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3cca61246bf1420a82cffc7864c98306 : RETRY_LIMIT in 2m 31s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a0daee9082d04da8bb5006815b6520c2 : SUCCESS in 9m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b34f5747ec674d6cbfe7bf75fdffbc18 : SUCCESS in 7m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/03b560f266ef44feb3a105848493e7f1 : SUCCESS in 12m 45s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6465fff3f73b48749567514ef25b4f37 : SUCCESS in 8m 29s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/63576689701d442fa7e194bab6938354 : TIMED_OUT in 2h 04m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/220e04a29aa549e482dda17cd59e80b1 : SUCCESS in 1h 39m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/891b6d08c3f346df9b887504a680035d : SUCCESS in 2h 00m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/59985113ee1e424883da5c8a0413ab0d : SUCCESS in 1h 41m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/066b4affeb10439182226190e0877af9 : SUCCESS in 2h 19m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d870c820f9dd48ac803d03dcbcf9ed78 : SUCCESS in 58m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cea24fb299704eb8a989e4d25fd30220 : FAILURE in 2h 33m 08s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/ad10f0997b004aacaaeb181219d971a5 : SUCCESS in 2h 04m 29s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/12a691237bb745c48ee71525b3ef13f6 : SUCCESS in 1h 01m 51s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e46fd46b20624fe4bd564f2e2d3499f7 : SUCCESS in 29m 04s","accounts_in_message":[],"_revision_number":17},{"id":"55839dedd55c2f224409e49f0d9dca427443b9e8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-06-23 16:06:07.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"48dba812d5eb48754e178f0d45f4b3967d44b71f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-23 16:13:25.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/441c240a91ef4c188dce0116702ec334\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9f60e58516e42c6938abc2627739290 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/58c43c43b32f4f3b8acc695a5873eb00 : SUCCESS in 5m 25s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"16121ff6584c75e91dc377aa2792181edac43b7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-23 18:17:47.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c27eb0fd06dd4ee7bacf7abe8d0dba2c\n\n- grenade https://zuul.opendev.org/t/openstack/build/511ce2926d814f76aaadcd62f961f258 : SUCCESS in 42m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/df02c1d24c304c7a8e134aef82f2ca93 : SUCCESS in 1h 07m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1af5544abaa41dfaec63828a3c61996 : SUCCESS in 45m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd8e9a39d2fa491dac1f97791d304518 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0278ffc76d6e4f6d941c0549bb9333af : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d395108562c3493ab767e4c0a5442c41 : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b146d8ce5f04f44ab4e3b3fa7e90b26 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6a78474f0174294be26f9cf496aefdb : SUCCESS in 5m 16s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/91b159ccef6d488298263e7d7b9c94d0 : RETRY_LIMIT in 1m 28s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c96315f30b6a4f59915414548728a6ae : SUCCESS in 8m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1aa640141f824f55aa0a20f02710f4c9 : SUCCESS in 8m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f7f635ac38134ac38835d6352dd4d093 : SUCCESS in 12m 06s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2b95c8b884924e4683f9a8e79ebe8acc : SUCCESS in 8m 25s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6395fc56e4284c9e804e9b218a47dbf7 : SUCCESS in 1h 57m 12s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/20df701c69f24b55b717bfbc264af50b : SUCCESS in 2h 01m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6d2533d4508449e8bd0356a1a017f1ad : SUCCESS in 1h 56m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d64cd8d6bc88426eb3b602e87df4ca0a : SUCCESS in 1h 58m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a8f9fa5bf1b64eb3be2d2a0f9260f391 : SUCCESS in 2h 10m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2700939b1eb54e89bbdd95d031b0c35a : SUCCESS in 1h 00m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6d2bdb3682644f50a07a18031b5023fc : SUCCESS in 1h 58m 14s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/88bdfe63fe4c450b93b783b42ef7f06c : SUCCESS in 1h 59m 52s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/de8d696b5e14407c83eb70d900eb297e : SUCCESS in 1h 07m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5444f1b8a25a4c19adc07fe798be80a7 : SUCCESS in 36m 18s","accounts_in_message":[],"_revision_number":17},{"id":"126ba9154c3ea7da144ee59af4434adf338cf7fc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-06-24 05:02:29.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"4e8e85640354146ed3adbf49aa72d84c41250ea8","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-06-25 13:14:03.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"dbb4e62ac3fb1e1b10398837563e6f5b7be2a348","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 13:14:46.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"0c977d5273c0e078edbca240755012080c52e93e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 15:29:47.000000000","message":"Patch Set 17: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70ec9a5793c6415f8415c9ba2a1e75f7\n\n- grenade https://zuul.opendev.org/t/openstack/build/5920b9b4f97149449aad244b9e397c8c : SUCCESS in 1h 00m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8455ddaf00d04290bdff13a8bce1006e : SUCCESS in 1h 31m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b853eff23a545fe8f3f2ac07b4759f5 : SUCCESS in 54m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4e40c08e84c425290998d98f8d67ae9 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/960c5366ca7a43cab801760fcc46b712 : SUCCESS in 6m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8cb1dd96885d449f82fa3c6d0468975f : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f60fabc008546b8a986ffd1eba4896d : SUCCESS in 5m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7bfe16f1b38c42ff94d3ad6aa60900b7 : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/21f52be3e9b6433788612315df2eb00b : SUCCESS in 8m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d1a985f8f9ee4c40887846195fed88cf : SUCCESS in 16m 20s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9179c571a1e84c4dae46298431dbe3dc : SUCCESS in 1h 53m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b0532e0bb6d34ec0be6f6f9e705c4235 : TIMED_OUT in 2h 04m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0556cc749e19431eada015dc5f719960 : SUCCESS in 1h 57m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbaf7e789c8242d28445f6e63cc0fd42 : SUCCESS in 48m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/58c3e80f07304fe09ce3bb986a8c3080 : SUCCESS in 2h 07m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9f12fb1f97f446ecae656bfa25388e69 : SUCCESS in 40m 03s","accounts_in_message":[],"_revision_number":17},{"id":"31a55a90d4d7419d273c3743e5b53d33b194d764","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-06-25 19:20:14.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"a5fb6f081eb7ca6d011be9e24d54a270bddddc9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 19:31:54.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5e39898b071452888f8803755e82714\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b6004e8407045e98513912e227cf67b : SUCCESS in 9m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b01bfb0d9a4a422abec00422d6de4610 : SUCCESS in 5m 28s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"928c8a07d4e3b13c6e4a1bfc758b615c6f7a761b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 21:38:14.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f6ac60e94f74866a7d33e0421e51cda\n\n- grenade https://zuul.opendev.org/t/openstack/build/3ce366a16f2741c084913ce4c4b230ac : SUCCESS in 47m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/75bdf99f8f6f4af08ae96606b6180c0c : SUCCESS in 1h 15m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a116f1e2720c4ba2b05cb28a47e6f9db : SUCCESS in 58m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c55ee446afbf48158b10a5657d70c1b4 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ad9740c126d416e9d4a0dfc6262c7d1 : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/415bdef6c6d948e8b1c7f99bb7b84895 : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ef8215370f68406e879629515300b329 : SUCCESS in 11m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3a421214af46cc8daad8d5c05837fa : SUCCESS in 5m 35s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1d7cd2c9b83242489dad476aaefb6353 : RETRY_LIMIT in 1m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2627a73ffe16477ca88f0391eddbe91d : SUCCESS in 8m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aadbd00fa0f94a21974fa3a65c983857 : SUCCESS in 8m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/60ef00e5560d471e9f81e7ae3d58f347 : SUCCESS in 12m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6c3e5012d5524493b0649169ce2e2741 : SUCCESS in 7m 35s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3524098fda5744b1b936c1b85751d460 : SUCCESS in 1h 47m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6c8ef23a89b045a085954cb7beb8e3ca : SUCCESS in 1h 50m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/79e7caaf35a44dc28a82fe8433c833d4 : SUCCESS in 1h 52m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/52444bbde9994ed6b8be1c0c2cb90705 : SUCCESS in 1h 44m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/999a66acfb3e46fd923da8b441f35745 : SUCCESS in 2h 12m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40c7c73f034e4f06898323a9954ce4fa : SUCCESS in 43m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b58fe77c23194691b51893ab04644792 : SUCCESS in 1h 56m 45s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/806021c676a74738a7c460a0931e6a2d : SUCCESS in 1h 52m 35s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f46b383d5be148b6b21e8564fb147de6 : SUCCESS in 1h 05m 45s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/84e1202d18584b68a0d1599fa4778768 : SUCCESS in 32m 57s","accounts_in_message":[],"_revision_number":17},{"id":"89e13a68cede9286b07dbb2fa20f7cc0c8b6df09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 21:38:49.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"9e5cc61224b0e905e4c5601b2c1c7366d6ce0b58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 23:49:19.000000000","message":"Patch Set 17: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c914cdb9b9d84656b1379ede0d0049ba\n\n- grenade https://zuul.opendev.org/t/openstack/build/95a92005d8954a789be88e429490ca96 : SUCCESS in 46m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b2d01d86953c403fbb015c526b86ddd2 : SUCCESS in 1h 36m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07416078e38f4678b64a0ee5f315baec : SUCCESS in 1h 04m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bc565cfb3054c8d93e9ffa334421a17 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/baeb9b0953ed43bf93a0b8997b234325 : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cacd5793b5da43508d36fc85103ff995 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a38cff226614d0b870454cf9fab1f63 : SUCCESS in 6m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cf5f3d3b2f744ce3a19880027ddd6a18 : SUCCESS in 7m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0be1959af49c451ca1a1b650594dd903 : SUCCESS in 7m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ac7add897d894f2ab77ac0af610f48dc : SUCCESS in 14m 20s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/dc3f56eb022b40c7b6d273455419fb93 : SUCCESS in 2h 04m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/278843a6e9ba409a9b16a3a6ef1b7a1e : SUCCESS in 1h 33m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/76317a020c8c4de7aee4ce2121aafdbd : TIMED_OUT in 2h 04m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec0e2a0937334ad092b1f20ef4571862 : SUCCESS in 54m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3fefda5aa9364f35ae9fb46167dc192a : SUCCESS in 1h 47m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/be94784d46e249d7beab698376da528d : SUCCESS in 30m 52s","accounts_in_message":[],"_revision_number":17},{"id":"f2335553e146a25fb48b2e78a1ede8848a14ed6e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-06-26 06:14:48.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"feba3211ac4228033121215a643aaa1df804e7e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 06:20:24.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d34d21e6b7c4b21ac4e6aa76b0c315e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/164815f345204f69ad6a321f63d8dc47 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a18588f150ff49908f213d2491f2bf54 : SUCCESS in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"050a940efc87d09b67ba06f7c9b4e0399046865f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 09:19:45.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c07144ebe7d463a8bc476e728c02beb\n\n- grenade https://zuul.opendev.org/t/openstack/build/5550a951ca48446aa39c6ef7fd4da4b0 : SUCCESS in 1h 08m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bd1c2d16b1694180a483f1e1c7adb210 : SUCCESS in 1h 31m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1774ac4e0d942febe67e66a9019f970 : SUCCESS in 1h 03m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/added2e9c3ea49e9b3c6e1d62502e6a8 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba0cf1d97838404695a68ccabeefb6b2 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2fccac9efbf84a3db8c36cac7c4ddd0f : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ecd2279fccf84f798d2ffbc1ae6d17d2 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec27cf6d8a194746abf3db5f35812907 : SUCCESS in 5m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/cd3626e3e0df4a14be54011df4601c5f : RETRY_LIMIT in 1m 34s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/95cab4092e5942068a62cc5d0a18ea24 : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/13029d1c861640729e1b516d2efc8785 : SUCCESS in 7m 42s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cff3ddac030b458b9f7b97f935e95e86 : SUCCESS in 11m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b6a1af02a36e4147b6fd09d0a1c0dc15 : SUCCESS in 8m 29s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a47b66fb8cd04d07b42f954035cb6218 : SUCCESS in 2h 02m 03s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4b2e1bef64e04e40b61092a7b2e4164d : TIMED_OUT in 2h 03m 32s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c71a3a19f1dd4c2ebc168408b6b5eef9 : TIMED_OUT in 2h 05m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b01b768fb15f45e3a33fe414b170220f : SUCCESS in 2h 02m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d7c1bec6eb9b424a8751c7a97d6c2751 : SUCCESS in 2h 11m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c02f722934c545bba5c2c404082abbf5 : SUCCESS in 1h 04m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cc9f72b3b4914c0e8bf059b3ef0fd604 : SUCCESS in 2h 30m 03s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/331e1153739448eeb13bfb1410317b98 : SUCCESS in 1h 59m 08s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ca2a17c97b7544a5b97457046cdc1c62 : SUCCESS in 1h 07m 09s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/38555a8393704c7287141896c0d2cbc0 : SUCCESS in 32m 33s","accounts_in_message":[],"_revision_number":17},{"id":"bf2aacb9bf97c4ee47f340a070973a8505a56fa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-28 12:46:31.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":18},{"id":"7055de1c05a2cde791227c353fd4c809f7f323e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 13:07:25.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6eebeb83998f41e296c41fe4cbdaab36\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/54fb87a5092b41cca8e5db74919a930d : SUCCESS in 13m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/52eb0d4745084373972c08d93f1480b4 : FAILURE in 2m 58s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"e7b9847df047cf60513ed3bf2762a0cb0caa7624","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 15:03:52.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/38da4c3471184005be03fb0a0935ff84\n\n- grenade https://zuul.opendev.org/t/openstack/build/56c6ce9616464fb99726b8cabd85a4db : SUCCESS in 1h 15m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/da7960f9aa9249498a189f1e4dac1b9d : SUCCESS in 1h 39m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a446eccaf7d4f1e9ca7ba58d06b3297 : SUCCESS in 1h 00m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b091de06c944fd19a44cf9bd64db506 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b17297f4b654ab5a23ba05b04625185 : SUCCESS in 6m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec494cf83fa04a369a2c7738de28819d : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fb8673613c734693a79d85566ed4713a : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a91efb9d51414daf959efd36c202bb61 : SUCCESS in 7m 21s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/03b002bf23104e7e9a1e8efd31993666 : RETRY_LIMIT in 1m 48s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/70a7dbf66a524c6187af8e9024d036df : SUCCESS in 10m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f37376d9096447b7894fd1eae8f48d24 : FAILURE in 10m 24s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d99c4ff0513f4c47a5dbffce05141a93 : SUCCESS in 12m 13s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d2ea746b7ca8460c98db066d5c91dcef : SUCCESS in 9m 55s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cdbf10db95e84d5690d1f92acab9953f : SUCCESS in 1h 57m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2b8d16de341f474b9eead985db237102 : SUCCESS in 1h 52m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/444bb01950aa417aaba73809683598af : SUCCESS in 1h 43m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/676bc93b21854a18903bfcd54e9a0744 : SUCCESS in 1h 51m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8b7eddf09b1e4a0797a34ff8fd0164ff : SUCCESS in 2h 10m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/822141ef43da4b61972b19d93b403735 : SUCCESS in 57m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6dd86980a593474abe5be3fe7db02425 : SUCCESS in 1h 25m 24s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/f941321fa20b4e4a912d17dcefc8df22 : SUCCESS in 1h 36m 28s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e7e61d8f03f74e18b075b27fdd3f931f : SUCCESS in 56m 51s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ceabfafbbdbc4227a8cb51fc8aa349f2 : SUCCESS in 27m 38s","accounts_in_message":[],"_revision_number":18},{"id":"bc94bad82dd32a02f4c97ff41a03034f43105b8c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-01 18:15:21.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"85b9099a43bb5fe04c8b0b48749e1527c70d386c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-01 18:15:26.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"c015d4e348cf8d31e1486a5d4196fdadb189d150","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 18:20:22.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c128bcb37b4489eb98af01bc389108f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6fe0406e7bf34830b1af44b639de974e : FAILURE in 4m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ae55b7a94a3f48d3b2a8c466cd16b2a2 : FAILURE in 2m 54s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"61f5bdad6f04b58fb74df028da18a452b2ee37c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 20:36:12.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1c826a2b1ba4a5187e1d493a6099fec\n\n- grenade https://zuul.opendev.org/t/openstack/build/2aff8a7fcd094de69634ad4b95821fd3 : SUCCESS in 1h 12m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/adce961bf4fa4b528e718b37778a5faa : SUCCESS in 1h 26m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1088ac7038f4e3ea1de7c8716244184 : FAILURE in 59m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5fe5ca72a414fa3968fb86396ed35f9 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7e1bd532c6b4da5a49f624d0ea19ea8 : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b265d32a476947b08c20f6f8cb989ce2 : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/244a8f66c45145efae0c3cfdb3acfdb0 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/087b5be8b5a84fd5b8fbfbe0e97cdd82 : SUCCESS in 5m 57s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ab8f207cf39b4b36a62b05c61011eb28 : RETRY_LIMIT in 1m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b0ea26c956da45a79d9d5691baccc85f : SUCCESS in 8m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7623d136aba744dcb68b820412cef569 : SUCCESS in 7m 29s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8b9d08e8977e4be19dd27f72f14df889 : SUCCESS in 12m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/19424d50ebe1489fb9e449a5fb1ada41 : SUCCESS in 7m 41s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7e11caac7b4c491daaba89d76b96ef5b : FAILURE in 1h 38m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/24e29ccb921f4625a3852595d993361a : SUCCESS in 1h 57m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bd978c0017b3422da5c625cf043071b3 : SUCCESS in 2h 03m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/88df09aecd5e4f44a8a2230c3ecc7ede : TIMED_OUT in 2h 04m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/56763dec9a29449c888767eba67fa9eb : SUCCESS in 2h 14m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f34d9455a8f4100877be0f171824632 : SUCCESS in 1h 06m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9ce5aa4613dc40809869fc713924b1f7 : SUCCESS in 1h 49m 49s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/cac3ea883aa04ed9b37b652f6ffb68bd : SUCCESS in 1h 54m 16s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1ad18028e1574b9b8e1324da2e8c9b24 : SUCCESS in 1h 10m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/032fff1b3d0943cca8e16636508f096a : SUCCESS in 35m 07s","accounts_in_message":[],"_revision_number":18},{"id":"0009566769e3feb1e2fd8f4f0070b3b9671a5631","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 07:03:22.000000000","message":"Patch Set 18: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"560e7c7e6ba2df052f2e0b8b1d9212839cbc4c74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 07:08:21.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81c681e3bf744fbc9332594d5b038d50\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3ac9725f0237421293c53590c2c15cb8 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/276baccc80754af1b6eb6bbb6881b2d4 : SUCCESS in 4m 14s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"66c5dd4bd99fbb114e6bfc8f93fb96461f7171c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 09:08:32.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a256281bf674613a144d0edff70685c\n\n- grenade https://zuul.opendev.org/t/openstack/build/80bb400ee6fa4fe7bae900b2772553e8 : SUCCESS in 1h 05m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ea30dc40b564f0a903fb4ae8fe70541 : SUCCESS in 1h 30m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8c5ee0e03b841359953d93e308bd4ad : SUCCESS in 1h 05m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5e803f6b37a4130ac80d793c2b8ba91 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc635ce10e2b46a2ac60f9793f2025ec : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef65232d55f749c988fea25524946baa : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/357eb61a51a64172b942f827a128c206 : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6827df80794a4cec999d90f53ec75f63 : SUCCESS in 6m 08s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/55f213686a3c46289907ff4de33f6a4a : RETRY_LIMIT in 1m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8aec6a7700d84d8896d787d50c92beec : SUCCESS in 8m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/969b2d17f45e491faac3096f6626675f : SUCCESS in 8m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/76f6e181e7014e868eb22e2c33c13658 : SUCCESS in 12m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/de7505bf9f564db8ab89b38c9515e947 : SUCCESS in 9m 28s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9a961f67fd0447918ecdf1e4b5197aa3 : SUCCESS in 1h 52m 21s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5887894698804369a1c53c8c82ee26e8 : SUCCESS in 1h 57m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e802e9d8f50f4c38b7c5173f1db69c2a : SUCCESS in 1h 55m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d6a1713a16e445278c8793803b118e81 : SUCCESS in 1h 56m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0017182b9ca343e49e5fb3282b6599ca : SUCCESS in 1h 46m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0579c59a74e544c695bbc13236ef7306 : SUCCESS in 38m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/12c998bdb2154f869d14dbe58e66602f : SUCCESS in 1h 57m 23s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/fd72badfa8d04373a9a00a5bb75464b5 : SUCCESS in 1h 41m 31s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/93316b756221456aafc9b9ddfecf72c4 : SUCCESS in 1h 01m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5c813c5bedd84bc89f45bd1c2f6839ce : SUCCESS in 27m 32s","accounts_in_message":[],"_revision_number":18},{"id":"938fb43a6fd615caeca5ec23cbde8de93e292162","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 09:13:39.000000000","message":"Patch Set 18: Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"5886d9b5d4056e5f4397c2666506a7d6e2a0dc0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 09:14:33.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"2ae379d1ad5b2e6cb18effc442b7dc0224d3dffa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 11:48:14.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34b39ae6720c4be4be95b1ed0192fd76\n\n- grenade https://zuul.opendev.org/t/openstack/build/d8a8d955b4ff4e48b6ef5a983c35bdb6 : SUCCESS in 51m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d03885993b204646952273d8f5010580 : SUCCESS in 1h 33m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d686392c792440949ea819fba21a37f5 : SUCCESS in 1h 01m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94bfeaed6bdb4bca9d9c7a68abece8d3 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c44790baa0c24f1fb9f31e2c55221612 : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a8dfeb6b8bd6489a8b4b28633cb6e9b8 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd842a29402d417e83c6bb3cd6eafc31 : SUCCESS in 6m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c854d91bde4e49a987cea86ce90cb2f3 : SUCCESS in 7m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1887e1f1c028408eb8d2ffbd5d2f973c : SUCCESS in 10m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fba54464d13843daab25c8d969542dd5 : SUCCESS in 12m 16s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7b77a0baa7864a65ade1672521391b9a : SUCCESS in 1h 56m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/965c0fc4f03f48efacd8a60271ee5b93 : SUCCESS in 1h 47m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ee49eb1bfbe04792ad46b981ed17ddc7 : SUCCESS in 2h 01m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/222885940cd24be58c95bfee7b327da2 : SUCCESS in 56m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dedf99d69884447183da33861c414160 : SUCCESS in 1h 30m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/71491e7a7c3d4a7484b03776a8f510c0 : SUCCESS in 29m 38s","accounts_in_message":[],"_revision_number":18},{"id":"c1e9012a34b39119b87eb61fe9d281479a6e7c64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 11:48:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":18},{"id":"91d7d5a60e91b7342591c8d384451de250062ecd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 11:49:20.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf9e36de43c54a3c9c6165e80956e84c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc56434d827e4073b1f16e9f9ff6375e : SUCCESS in 46s","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"1c523cf94e89470c008163a31385a6cff3658c93","revisions":{"86eb10c4396b7b280f5ba7f400fecd36575e6c5a":{"kind":"REWORK","_number":1,"created":"2024-01-23 11:15:20.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/59/906359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/906359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/906359/1"}}},"commit":{"parents":[{"commit":"a0a3abf1f4394d5ffb1ee6434207f97d83301acb","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a0a3abf1f4394d5ffb1ee6434207f97d83301acb"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"subject":"Fix cinder location url compatible with multistore","message":"Fix cinder location url compatible with multistore\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/86eb10c4396b7b280f5ba7f400fecd36575e6c5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/86eb10c4396b7b280f5ba7f400fecd36575e6c5a"}]},"branch":"refs/heads/master"},"2f32d688cee03c122df5ba06880bc8d0d1a95f13":{"kind":"REWORK","_number":2,"created":"2024-01-24 10:53:02.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/59/906359/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/906359/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/906359/2"}}},"commit":{"parents":[{"commit":"f429b53e0474553992f43c7045498156ee36d5cb","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f429b53e0474553992f43c7045498156ee36d5cb"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-24 10:33:42.000000000","tz":0},"subject":"Make location URL compatible with cinder backend.","message":"Make location URL compatible with cinder backend.\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f32d688cee03c122df5ba06880bc8d0d1a95f13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f32d688cee03c122df5ba06880bc8d0d1a95f13"}]},"branch":"refs/heads/master"},"e51ac6a4df33ad98d1fea47ef8ef5002c613d533":{"kind":"REWORK","_number":3,"created":"2024-02-01 10:33:16.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/59/906359/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/906359/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/906359/3"}}},"commit":{"parents":[{"commit":"b6b9f043ffe664c643456912148648ecc0d6c9b4","subject":"Replace usage of deprecated [DATABASE] sql_connection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b6b9f043ffe664c643456912148648ecc0d6c9b4"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-01 10:17:18.000000000","tz":0},"subject":"Make location URL compatible with cinder backend.","message":"Make location URL compatible with cinder backend.\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e51ac6a4df33ad98d1fea47ef8ef5002c613d533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e51ac6a4df33ad98d1fea47ef8ef5002c613d533"}]},"branch":"refs/heads/master"},"431836b5650982c57c034aa370c11d302e23517f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-02-07 06:50:50.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/59/906359/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/906359/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/906359/4"}}},"commit":{"parents":[{"commit":"9f896ab03d3516d05d410fc545a6faf7fb7554e2","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f896ab03d3516d05d410fc545a6faf7fb7554e2"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-07 06:08:26.000000000","tz":0},"subject":"Make location URL compatible with cinder backend.","message":"Make location URL compatible with cinder backend.\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/431836b5650982c57c034aa370c11d302e23517f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/431836b5650982c57c034aa370c11d302e23517f"}]},"branch":"refs/heads/master"},"3c0966a0bc848a7de6cbabbe8be4e20509dbe394":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-02-15 10:31:13.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/59/906359/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/906359/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/906359/5"}}},"commit":{"parents":[{"commit":"7fd459fef29f586d681bf4e3e3639bd061adec9f","subject":"Merge \"Deprecate sqlite cache driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7fd459fef29f586d681bf4e3e3639bd061adec9f"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-14 11:52:36.000000000","tz":0},"subject":"Make location URL compatible with cinder backend.","message":"Make location URL compatible with cinder backend.\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c0966a0bc848a7de6cbabbe8be4e20509dbe394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c0966a0bc848a7de6cbabbe8be4e20509dbe394"}]},"branch":"refs/heads/master"},"4fa1381ee0e4125367498f2a853990b53d2717a3":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-02-20 10:23:20.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/59/906359/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/906359/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/906359/6"}}},"commit":{"parents":[{"commit":"02be49c998c578284052e73ec1a6bef7193afc8c","subject":"Merge \"Add new tables for cache operations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02be49c998c578284052e73ec1a6bef7193afc8c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-20 09:37:13.000000000","tz":0},"subject":"Make location URL compatible with cinder backend.","message":"Make location URL compatible with cinder backend.\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4fa1381ee0e4125367498f2a853990b53d2717a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4fa1381ee0e4125367498f2a853990b53d2717a3"}]},"branch":"refs/heads/master"},"2792c038eba6cabb35291fa2aa7c21e99074a1b5":{"kind":"REWORK","_number":7,"created":"2024-02-22 07:41:38.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/59/906359/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/906359/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/906359/7"}}},"commit":{"parents":[{"commit":"02be49c998c578284052e73ec1a6bef7193afc8c","subject":"Merge \"Add new tables for cache operations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02be49c998c578284052e73ec1a6bef7193afc8c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-22 05:39:57.000000000","tz":0},"subject":"Make location URL compatible with cinder backend","message":"Make location URL compatible with cinder backend\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nCloses-Bug: #2054575\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2792c038eba6cabb35291fa2aa7c21e99074a1b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2792c038eba6cabb35291fa2aa7c21e99074a1b5"}]},"branch":"refs/heads/master"},"ccfd8dab4bb3770714b3e141fb4dff2913a4b2c9":{"kind":"REWORK","_number":8,"created":"2024-02-22 10:58:39.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/59/906359/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/906359/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/906359/8"}}},"commit":{"parents":[{"commit":"02be49c998c578284052e73ec1a6bef7193afc8c","subject":"Merge \"Add new tables for cache operations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02be49c998c578284052e73ec1a6bef7193afc8c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-22 10:56:10.000000000","tz":0},"subject":"Make location URL compatible with cinder backend","message":"Make location URL compatible with cinder backend\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nCloses-Bug: #2054575\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ccfd8dab4bb3770714b3e141fb4dff2913a4b2c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ccfd8dab4bb3770714b3e141fb4dff2913a4b2c9"}]},"branch":"refs/heads/master"},"983bde9f1f963e31744d9682893a32adb57e3d98":{"kind":"REWORK","_number":9,"created":"2024-02-26 08:34:52.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/59/906359/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/906359/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/906359/9"}}},"commit":{"parents":[{"commit":"02be49c998c578284052e73ec1a6bef7193afc8c","subject":"Merge \"Add new tables for cache operations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02be49c998c578284052e73ec1a6bef7193afc8c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-26 08:32:50.000000000","tz":0},"subject":"Make location URL compatible with cinder backend","message":"Make location URL compatible with cinder backend\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nCloses-Bug: #2054575\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/983bde9f1f963e31744d9682893a32adb57e3d98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/983bde9f1f963e31744d9682893a32adb57e3d98"}]},"branch":"refs/heads/master"},"bc336fc7b082dc51be04a6bc6ad96c72b7c75ab5":{"kind":"REWORK","_number":10,"created":"2024-02-28 19:01:34.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/59/906359/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/906359/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/906359/10"}}},"commit":{"parents":[{"commit":"703cf5ffddc1dcc9a7086b8dff220cf5627d308b","subject":"Merge \"[docs] Configure centralized_db cache driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/703cf5ffddc1dcc9a7086b8dff220cf5627d308b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-28 18:48:13.000000000","tz":0},"subject":"Make location URL compatible with cinder backend","message":"Make location URL compatible with cinder backend\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nCloses-Bug: #2054575\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc336fc7b082dc51be04a6bc6ad96c72b7c75ab5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc336fc7b082dc51be04a6bc6ad96c72b7c75ab5"}]},"branch":"refs/heads/master"},"7e9bae465c327a352b50393d415cda2e8b70fec1":{"kind":"REWORK","_number":11,"created":"2024-02-28 19:21:23.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/59/906359/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/906359/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/906359/11"}}},"commit":{"parents":[{"commit":"703cf5ffddc1dcc9a7086b8dff220cf5627d308b","subject":"Merge \"[docs] Configure centralized_db cache driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/703cf5ffddc1dcc9a7086b8dff220cf5627d308b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-28 19:09:28.000000000","tz":0},"subject":"Make location URL compatible with cinder backend","message":"Make location URL compatible with cinder backend\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e9bae465c327a352b50393d415cda2e8b70fec1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e9bae465c327a352b50393d415cda2e8b70fec1"}]},"branch":"refs/heads/master"},"4b2ff484e24bb314255a29696792d40f944b753f":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2024-04-04 07:00:07.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/59/906359/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/906359/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/906359/12"}}},"commit":{"parents":[{"commit":"0bcd6cd71c09917c6734421374fd598d73e8d0cc","subject":"Merge \"Docs: Add info about the cinder store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bcd6cd71c09917c6734421374fd598d73e8d0cc"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-04-04 05:57:40.000000000","tz":0},"subject":"Make location URL compatible with cinder backend","message":"Make location URL compatible with cinder backend\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4b2ff484e24bb314255a29696792d40f944b753f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4b2ff484e24bb314255a29696792d40f944b753f"}]},"branch":"refs/heads/master"},"08420bd734975ed8c5c3bc335724fe2223ea5e29":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2024-04-30 09:42:34.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/59/906359/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/906359/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/906359/13"}}},"commit":{"parents":[{"commit":"79f128746ce3c5c2c3b0719dd7e377e0bda6bf76","subject":"Merge \"Cinder: Remove ambiguous warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/79f128746ce3c5c2c3b0719dd7e377e0bda6bf76"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-04-30 09:19:58.000000000","tz":0},"subject":"Make location URL compatible with cinder backend","message":"Make location URL compatible with cinder backend\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/08420bd734975ed8c5c3bc335724fe2223ea5e29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/08420bd734975ed8c5c3bc335724fe2223ea5e29"}]},"branch":"refs/heads/master"},"65ec7c3dac511ecd2c51f32442bf0f874f060c2b":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2024-05-07 06:22:18.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/59/906359/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/906359/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/906359/14"}}},"commit":{"parents":[{"commit":"95770a25e33cd531216104c788ac5dacda4e36cf","subject":"Merge \"refectory: fix comments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/95770a25e33cd531216104c788ac5dacda4e36cf"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-06 11:17:42.000000000","tz":0},"subject":"Make location URL compatible with cinder backend","message":"Make location URL compatible with cinder backend\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65ec7c3dac511ecd2c51f32442bf0f874f060c2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65ec7c3dac511ecd2c51f32442bf0f874f060c2b"}]},"branch":"refs/heads/master"},"45dc70a3d84c720f135e576560b0007899f3e491":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2024-06-06 06:46:59.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/59/906359/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/906359/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/906359/15"}}},"commit":{"parents":[{"commit":"dfbfdad77fdc91899f64d2ff186097cfb2065ab7","subject":"Move all periodic tips jobs from py38/39 to py311/312","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dfbfdad77fdc91899f64d2ff186097cfb2065ab7"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-04 06:14:35.000000000","tz":0},"subject":"Make location URL compatible with cinder backend","message":"Make location URL compatible with cinder backend\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/45dc70a3d84c720f135e576560b0007899f3e491"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/45dc70a3d84c720f135e576560b0007899f3e491"}]},"branch":"refs/heads/master"},"c0e96abf3fb195f666431123f1ef37e12b950834":{"kind":"REWORK","_number":16,"created":"2024-06-14 10:35:46.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/59/906359/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/906359/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/906359/16"}}},"commit":{"parents":[{"commit":"dfbfdad77fdc91899f64d2ff186097cfb2065ab7","subject":"Move all periodic tips jobs from py38/39 to py311/312","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dfbfdad77fdc91899f64d2ff186097cfb2065ab7"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-14 09:43:17.000000000","tz":0},"subject":"Make location URL compatible with cinder backend","message":"Make location URL compatible with cinder backend\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c0e96abf3fb195f666431123f1ef37e12b950834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c0e96abf3fb195f666431123f1ef37e12b950834"}]},"branch":"refs/heads/master"},"61a63881111ec0d0f7f0d2ccfa68df153878abac":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":17,"created":"2024-06-20 10:58:02.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/59/906359/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/906359/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/906359/17"}}},"commit":{"parents":[{"commit":"1b11e6e3a29185acc627b988ac5ae32856492aa7","subject":"docs: Correct grammar nit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1b11e6e3a29185acc627b988ac5ae32856492aa7"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-20 10:50:07.000000000","tz":0},"subject":"Make location URL compatible with cinder backend","message":"Make location URL compatible with cinder backend\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nRelated-Bug: #2054575\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61a63881111ec0d0f7f0d2ccfa68df153878abac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61a63881111ec0d0f7f0d2ccfa68df153878abac"}]},"branch":"refs/heads/master"},"1c523cf94e89470c008163a31385a6cff3658c93":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2024-06-28 12:46:31.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/59/906359/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/906359/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/906359/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/906359/18"}}},"commit":{"parents":[{"commit":"4dec531b0aea15e8d970fa775a6e90068ddd5b03","subject":"Merge \"Use constants for common rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4dec531b0aea15e8d970fa775a6e90068ddd5b03"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:00:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-28 12:13:47.000000000","tz":0},"subject":"Make location URL compatible with cinder backend","message":"Make location URL compatible with cinder backend\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nRelated-Bug: #2054575\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1c523cf94e89470c008163a31385a6cff3658c93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1c523cf94e89470c008163a31385a6cff3658c93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
