)]}'
{"id":"openstack%2Fglance~906977","triplet_id":"openstack%2Fglance~master~Ia739af13e0f88d0861e453a4d54ecc626e32b34b","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-15 15:17:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia739af13e0f88d0861e453a4d54ecc626e32b34b","subject":"Drop ineffective options from config file","status":"MERGED","created":"2024-01-27 16:20:48.000000000","updated":"2024-02-15 15:18:24.000000000","submitted":"2024-02-15 15:17:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906977","meta_rev_id":"d81a8dbafd8f23d4145fda3e4b729f711e030fb5","_number":906977,"virtual_id_number":906977,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-15 15:17:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2024-02-15 14:30:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2024-02-15 14:30:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-27 16:27:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-27 16:59:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-15 00:48:11.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2024-02-15 14:30:32.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"0de36cbe84ab7e480a8b8612dff3076545b00f49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-27 16:20:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"526cc1d95909f55f65a597db0bad1c3b7dcd194b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-27 16:27:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/042f4fc44fdd49ea8926d5aa229abc80\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0df4990fd22a4c9898b2d781dad21717 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4ef70872f3c14f86b01f9e82efbd1341 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9d29a3c4659b4a45a094410c4cc9d3be : SUCCESS in 5m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2cfa65a51d92a2b85547af2302bf729daa0e5ac8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-27 16:59:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac10dc3c1b2f4ed3a1c239a16a37876c\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/08cbe7e596f44907a882a43aac261e2a : SUCCESS in 37m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66ea8b74c4144ba5bc6c76bf3f9e3ea7 : SUCCESS in 18m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/656fa94f27394004b3ab7fe6cb7a7212 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/462f29e5da924963aa39d60be8af85bd : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dcdafbe3b1064c69886213548b3075ea : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b05179658cf4288b470733a480f9420 : SUCCESS in 5m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/54ec7e5ae83840c6a11c2e1265062222 : SUCCESS in 9m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/409467294f5649ff88d70486feb5e443 : SUCCESS in 8m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0bf24ed06aaa4f3c84c862acd464e153 : SUCCESS in 7m 23s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1f2d091b1a9149d9a4ef6f38f5fac162 : SUCCESS in 8m 31s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/98e24b7f39ed4bf791f0a803e0b2397d : SUCCESS in 10m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f12b1bbe751a4f8fbd83efd02271f799 : SUCCESS in 20m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d4d5bcba11add2688ba50f037cbe7dbd54aa74d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-31 01:16:12.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1d22b83f8ac16e7664966a5e44f95561453e9a10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 01:22:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2601ce1b5461407ebee171e6c318f10a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c34905e8559645098d400c163c65e733 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f7026b5b4249407988da1e9e638dadd7 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6636d1959444486da2e06ed16e371fd2 : SUCCESS in 5m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cc91d2822cf586c296c1d4367531bd3d23d380b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 02:00:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb5766e738dc444d9a4b0c1b3c20fd34\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/058287457c31443480c49b6c59c077a7 : SUCCESS in 43m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d6425c4b62d40da8d408b47c0997e69 : SUCCESS in 19m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/544d1538bb2c491c99cd8b47d5088e0a : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1753bb073d25429e8c49136eab6454b1 : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7fd4b850369d46f99110d3c10dbd9433 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c08a16c0fbb7416da13cf7afea105feb : SUCCESS in 4m 50s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b732e26722b7434eaa6d3757b72555a3 : FAILURE in 14m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dbeddfa50ea042be97f917e912e76143 : SUCCESS in 9m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/26c5008d1507409daa9f2031adb7b2a9 : SUCCESS in 8m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a001ddca397a4b8eb8a32d567d21f592 : SUCCESS in 8m 07s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/9f6a6fb744ab460baab8559b0bc67b08 : SUCCESS in 11m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/47e33e5fb40e4ed68f08b9275ad23e0d : SUCCESS in 24m 16s","accounts_in_message":[],"_revision_number":2},{"id":"fb57959204041d88e4fc63d3dc522b61887aa9c1","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-15 00:48:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cddd3683591954f66366f00d7fd9258c5b63ffec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 00:54:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4b3df371dd0439fb09aaddc3cc05f9d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/07b8d33dc39642778c148f1ec9f6390d : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d57f43816c514be794e51fd5b0dc82cb : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5bd42c0ce69f426f963ad74b19a03d0b : SUCCESS in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18fd6fbed381034a5be92e2297597ce99ff74dee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 01:57:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db72d28efa1a4f69a8fb13535f1f4682\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/575922c2d6674ea2a65e6fe6392eca31 : SUCCESS in 1h 08m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d334199ac824083bf7e41df08bdcf5a : SUCCESS in 25m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a6939080acc4c6fbb7f39a4e50ccfb5 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e78d7682e12c4f52a09531e6adaea80c : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d26a6f3ce0d447ad959199a39e45ceae : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2dca614bfe4458ab322e0368915924c : SUCCESS in 6m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2422f01b313442f68a7cde66f6dad0c0 : SUCCESS in 13m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4b70e77081ab4cd086029a9a0422e4e6 : SUCCESS in 8m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ba0c8dc6b942427e9981a3b03f1b6f14 : SUCCESS in 10m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a7cf7489ada643a197ae402ab257181a : SUCCESS in 9m 34s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ee6c5fd226a54d32b0040d1e39da1f81 : SUCCESS in 10m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e579132c32e547299c35f5bc07294760 : SUCCESS in 37m 23s","accounts_in_message":[],"_revision_number":2},{"id":"4bf4e5c73879bbc3341a24ba0c98b712fbb7f16e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-15 14:30:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3af090798f720c8b92d40888bb9ee87262625fae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 14:31:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e78a77b2269aa8219b87775c262fd9ea749b291a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 15:17:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff0926d8fdb241e6814f028513db7608\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f927abf37a6046d3ae94bea09700ead3 : SUCCESS in 37m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/542e68bb8ac24969a25af0e55a1dd425 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f31f022319b34a3dac4774a07ba10185 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e0ccb737edf4ff18bd2a12c63f34a12 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14bfba9efa9046329974f7bc381b5f89 : SUCCESS in 5m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/83cb77a4c5944409a0618f921a291f7b : SUCCESS in 8m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d18444af62c54f24a2dec16685b97e45 : SUCCESS in 8m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1ea30579a4364519accec5bd62ca0e4a : SUCCESS in 41m 09s","accounts_in_message":[],"_revision_number":2},{"id":"500bc30477ce3ece4adf59839d6cd0bffb61ce3f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 15:17:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d81a8dbafd8f23d4145fda3e4b729f711e030fb5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 15:18:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca6e6725983246fc974d0bab52880171\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dee874dee4f4d08b3a441dd28dcd37e : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1e99767d0e213d73b50c52241848bbc01971cd94","revisions":{"76f3ad8514adcb9f454c85e8778a1ba3bdfc827a":{"kind":"REWORK","_number":1,"created":"2024-01-27 16:20:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/906977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/906977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/906977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/906977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/906977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/906977/1"}}},"commit":{"parents":[{"commit":"b6b9f043ffe664c643456912148648ecc0d6c9b4","subject":"Replace usage of deprecated [DATABASE] sql_connection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b6b9f043ffe664c643456912148648ecc0d6c9b4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-27 16:16:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-27 16:16:25.000000000","tz":540},"subject":"Drop ineffective options from config file","message":"Drop ineffective options from config file\n\nThe following options do not exist.\n [DEFAULT] send_identity_credentials\n [DEFAULT] send_identity_headers\n\nChange-Id: Ia739af13e0f88d0861e453a4d54ecc626e32b34b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76f3ad8514adcb9f454c85e8778a1ba3bdfc827a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76f3ad8514adcb9f454c85e8778a1ba3bdfc827a"}]},"branch":"refs/heads/master"},"1e99767d0e213d73b50c52241848bbc01971cd94":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-01-31 01:16:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/906977/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/906977/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/906977/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/906977/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/906977/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/906977/2"}}},"commit":{"parents":[{"commit":"b6b9f043ffe664c643456912148648ecc0d6c9b4","subject":"Replace usage of deprecated [DATABASE] sql_connection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b6b9f043ffe664c643456912148648ecc0d6c9b4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-27 16:16:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-31 01:16:12.000000000","tz":0},"subject":"Drop ineffective options from config file","message":"Drop ineffective options from config file\n\nThe following options set in functional tests do not exist.\n [DEFAULT] send_identity_credentials\n [DEFAULT] send_identity_headers\n\nChange-Id: Ia739af13e0f88d0861e453a4d54ecc626e32b34b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1e99767d0e213d73b50c52241848bbc01971cd94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1e99767d0e213d73b50c52241848bbc01971cd94"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
