)]}'
{"id":"openstack%2Fglance~907111","triplet_id":"openstack%2Fglance~master~Id94e93e3ba3fc207b39c7dbff92495805aa0f6f9","project":"openstack/glance","branch":"master","topic":"centralized-cache-db","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-03-06 08:17:12.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-03-06 04:10:23.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-03-05 19:55:28.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-03-06 08:17:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id94e93e3ba3fc207b39c7dbff92495805aa0f6f9","subject":"Make `centralized_db` cache driver default","status":"MERGED","created":"2024-01-29 18:47:26.000000000","updated":"2024-03-06 08:18:35.000000000","submitted":"2024-03-06 08:17:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907111-centralized-cache-db","meta_rev_id":"3af5cc2a4f1209ecad2a15f9048363837bf8d5eb","_number":907111,"virtual_id_number":907111,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-06 08:17:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2024-03-05 19:56:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2024-03-05 19:56:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-29 19:00:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-29 20:58:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-28 14:47:15.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-03-05 10:15:11.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2024-03-05 19:30:52.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-03-05 19:56:30.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"f32eb9a6e68e0a1082a7027704725bd7804bf83d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-29 18:47:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b876ff4f2188296dbd4405e9bd934ca53e8cf327","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 19:00:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fb547a5ec6b48b4967201d66ccb5b0e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3183afc30ea64963bbf4bed2ba75d29e : FAILURE in 10m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4e4943694f7d463e91be795e85bb5e5e : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a1a5ce0b0e6248be9c493cb4a73810b4 : SUCCESS in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eabeb797fa122a24e0979c3fad5a9889442075af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 20:58:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44f9e4e08c634c8f9d762fea8eca0bba\n\n- grenade https://zuul.opendev.org/t/openstack/build/862a3dc28ffb4560ac594fc95633d09e : SUCCESS in 41m 40s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/84743ad73c8642ee91cce6a7e7e82795 : SUCCESS in 1h 10m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/936e486089c3402a9ba8c4eb4a836f65 : SUCCESS in 1h 17m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a313f6304b1427a82582f315c152414 : SUCCESS in 32m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef0b73494fc243d1b43adaeac81a3199 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6175e20b0802442bb3387e04d33934b1 : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/12944f8fde634fffb0955163cbb3e418 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740830faa4094cca88353197f385ca11 : SUCCESS in 7m 19s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/23c2bb64fa134db0ab010654a61850aa : SUCCESS in 13m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/624ad35264274cf9a0a982f05980a060 : SUCCESS in 8m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6a85fe132ce9413c90d9ad58d184928d : SUCCESS in 8m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6a6cca127423430c99ae7f71f4dfee18 : SUCCESS in 10m 17s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/89ced73416f6415e8584cce2750738a5 : SUCCESS in 14m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1c77209a0d814bf79919d6e02a43e2b6 : SUCCESS in 1h 57m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/680c4294fa6f411499cecdbba5cf92e7 : SUCCESS in 1h 22m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/53bce0f7272e4430b2e53162c48b0da4 : SUCCESS in 1h 39m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0f69154391d84a929afccbaaa9ff37b8 : SUCCESS in 1h 56m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dc9c4f305ed2439390fb77293a840e87 : SUCCESS in 1h 40m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/425006c9a2ec486fb3aca93bbd474c4f : SUCCESS in 50m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9fb922e2e99f4574adfc3996f3814825 : SUCCESS in 1h 50m 08s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/ebe7bdf7d5a34be6874f23972d605ff2 : SUCCESS in 1h 19m 49s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/be05b2a9dad941c196dbf4040144abf4 : SUCCESS in 53m 54s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dae481c3dc1341e8bd289460abad1504 : SUCCESS in 34m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c7e9cb006e60bf1fb8922422fe47c543b78be616","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-31 17:54:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f10481b3a41030a75807fbe1e5a17bed6874945","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 18:09:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e37eb942f794a21b21b7cdb4570ce1a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec9b4ecf53bf4fd3a21745774c224ff6 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aba149ee81024c86b63749f8207da97b : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3109b400e3e7411fa460f1c219186716 : SUCCESS in 6m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd930ba669064e0240f5849d5c2a75fb5c011fa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 20:03:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/607cbefff6104fffac4ded9784bcb927\n\n- grenade https://zuul.opendev.org/t/openstack/build/21a98f53034f4cc4967de4964d888d5e : SUCCESS in 1h 08m 41s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7cd7aff1a61240688e6c72a341b9b21a : SUCCESS in 1h 03m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d8bd21d012334869936829afe50408eb : SUCCESS in 1h 19m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b22741c39f9d4b969756b0fc9dfed64b : SUCCESS in 40m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16f5eab34232419b8fb84fd91576ad7e : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5200955b43d4c4fabc2b82b6e709316 : SUCCESS in 7m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84ee6439a5a642dda5cb8609a78bbf4a : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8daa031867bd4f3bb6eddf5a87eadc17 : SUCCESS in 6m 30s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4776b8b40a6a49cbaee2596e185b2332 : SUCCESS in 14m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/faa65f0aa5234b6eb452c00524e946d9 : SUCCESS in 10m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/63d24955c44a40f19703f33a1bd4f4c8 : SUCCESS in 8m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ee37935198a84324b594e301f1b50303 : SUCCESS in 10m 25s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/1a5fea9e43c942f28414d80adad514c5 : SUCCESS in 11m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8a7522fb88af4d4ea48fe53fbb12cfb2 : SUCCESS in 2h 02m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a1dbf2d5ac014b8ca85edf5ef8f79afd : SUCCESS in 1h 47m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/173d2ce8648b4242a4f8fa92032ab9f1 : SUCCESS in 1h 45m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d705a19e07bf47c7849fc53686f170a0 : SUCCESS in 1h 41m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7289894fb235422b89bf1cb30896fc1b : SUCCESS in 1h 49m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60410812d2cb44548cf3b0c0b0f879a6 : SUCCESS in 1h 28m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/564643b5267146389bf9df4538cbbdf4 : SUCCESS in 1h 58m 26s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/75177e63170d43ecbca3654c88eb3ea5 : SUCCESS in 1h 39m 35s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/283b03c3108449c8a25073f0f197c669 : SUCCESS in 1h 00m 48s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/39bfc2e5d18a40139be25fda11996232 : SUCCESS in 44m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4a5d7bbc1090240490a527e7642b7273c06e614f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-31 21:23:20.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"fbcb3d45a17cdb6aa34e2b24d1915f990f398a01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 21:49:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71b537eeb9ca401a8db37c31678ca4b5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/39cd26cedd36448c82551f0792a5ce1e : FAILURE in 9m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9fda3ec093e149e08a7422258daf1645 : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/defc66b9feea436fabeb35efae0826e7 : SUCCESS in 8m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7ffcef5da2fa3ad219eff2fe78fb13f14f8bb32b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 23:58:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/961ae05f0dc541919658bd6130440e64\n\n- grenade https://zuul.opendev.org/t/openstack/build/86986bcbfd7c4d578fd6b247c2c2629a : SUCCESS in 1h 09m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/27f5482b16a248839f97668987091e25 : SUCCESS in 1h 05m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fc214a519d124ea2bf8b201b1ccac7b1 : FAILURE in 1h 22m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67855cdd95134eab8cc7419750bbb57d : SUCCESS in 33m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54f04771d88c4840b56bf255f947df9a : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9c0146cfe1f4f6488c81a0f17602781 : SUCCESS in 5m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3fca49bebff94e6a9b1ae4da139ec1f2 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaf3c58fe4954efda53858881129ab77 : SUCCESS in 7m 24s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0f616c0d43ab4746becb0035817adb98 : SUCCESS in 14m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/77951619099546ffbb4c297f4424fecf : SUCCESS in 9m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/016a98e4c10e44c2a9f6db1c14ddb921 : SUCCESS in 8m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/21e3200adbbf4687910898d2c076e9a5 : SUCCESS in 9m 03s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/4237fc87025c4f57ad8038164045c21b : SUCCESS in 11m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c81d9034157a41318f44686e6275495c : SUCCESS in 1h 43m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0730cb9297ed4130b2a31d6f2c4ada1d : SUCCESS in 1h 57m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b2d986753dd8421e8dd0bc7facb6b40d : SUCCESS in 1h 54m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6b258c463a91423abc758fa918a7652b : SUCCESS in 1h 49m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/29b93bd937f643a7920ec0bb713a48da : SUCCESS in 2h 14m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/928ed932f8044a31b561be04ce346a63 : SUCCESS in 42m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f20136e0a10347bfbcf0500241aa9c2b : FAILURE in 2h 10m 19s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/385cdb37bdcc4061aaf690dd576fdcb9 : SUCCESS in 1h 38m 58s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/771260c25bb74d91b1c94c6a7ceda71c : SUCCESS in 1h 06m 34s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3c41644378f54797887ae9e5d188f291 : SUCCESS in 41m 14s","accounts_in_message":[],"_revision_number":2},{"id":"3fac0f622430a35be2234b99a5b43eb7e22f95ea","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-01 05:45:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cbd94927eadbd0a2a4824073164a726d260331af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 05:59:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c99aa46bf73f4a2da81a9eeda9224f5f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b97464d509fc414a868abfec9d3439d5 : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8a0c553e08ba4205b1d793d9283b6781 : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5c6ebb3194c14395ad2a862e3947488b : SUCCESS in 6m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2ace6d6a7e7904ab182d691a9f051dd44c802238","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 08:11:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9135d1bbba63444db1b2444e77d50e30\n\n- grenade https://zuul.opendev.org/t/openstack/build/329e02658b0c4afea7aac8f29febaf3f : SUCCESS in 41m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c3fd2b5a31e349b48cadad932498daa1 : SUCCESS in 39m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/395f959a9bcd41c9a01555cd012cbd52 : SUCCESS in 1h 01m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b8f2852b6164997b32f0a8af2e1dada : SUCCESS in 32m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/814048c8258f4a2b8d110a93ee2f2db9 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00a294a1d63b4857a89bf6e221a5af67 : SUCCESS in 6m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e26b95e109d048ae9a993f3c9f977a38 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89bbfa20cdd14e74884b92c01d1bbf99 : SUCCESS in 7m 55s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c8a387cb54a24dd6bc44995edce73bac : SUCCESS in 14m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/722894cbbe1646b98c0cb9bc0d41e4e4 : SUCCESS in 10m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4689c398ebbf4831838ddb9fe2861b5f : SUCCESS in 10m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a3d61bf277464f2182baf2fbf2a84121 : SUCCESS in 10m 03s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f6417abc521b47ee93ed15917da1baa4 : SUCCESS in 11m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c4acb8d160324d0bbeb6e2864e1ac3e6 : SUCCESS in 1h 49m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/74c65df7448d489f8a32ebf018ee5076 : SUCCESS in 1h 53m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6886cf48e6274c86b2cdb49845c5b87b : SUCCESS in 1h 49m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0c0bd678729d4f21a1043568c9fc9ff9 : SUCCESS in 1h 45m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4bced8f59d684a3e854086dd3b67ac6a : SUCCESS in 2h 06m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db348abeb84547c98f0e52b2c5075e4e : SUCCESS in 1h 01m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/94b8d12b65f14219bcde9bbabe5e124d : FAILURE in 2h 14m 22s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3d4fc5176adf4cfebb235b66d65e68b8 : TIMED_OUT in 2h 04m 49s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/76ee71706ead446b94f77b5cf8afac5e : SUCCESS in 1h 21m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8bfc2af05884433eadb556bae6193b52 : SUCCESS in 32m 56s","accounts_in_message":[],"_revision_number":2},{"id":"3a82e599d337998c6c37746eeea7194898c1d247","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-09 10:05:42.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"21d54dd13858fd96d4e687ce5a6fb2f2617f3c3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 10:26:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4078c75bc3bc46faa5f8dbaad7526754\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8fa732cf8a7431a93e6d6cdc97d3c48 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e43887f17c7942bfb6cea2a56106fe79 : SUCCESS in 10m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c49ac606ef6643a58947706400f306b9 : SUCCESS in 5m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c3719a1937bc5d8f9d5ba17326369ff990fbd16d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 12:37:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ba5c6408f2a491db113ef2adc6b8420\n\n- grenade https://zuul.opendev.org/t/openstack/build/b624837a0bf24a7a848f552cc92629a2 : SUCCESS in 1h 01m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a6a8d9f936284be6b0be7e6a91f8704d : SUCCESS in 1h 01m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/80047dc955d84fb4a7262eca76a2fd94 : SUCCESS in 1h 22m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c58f025b73b24b3eb0acd5c5202434cd : SUCCESS in 34m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ea6ba44453f4546b61b3657b979eb4f : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16828aa43a514cc187f88c6b0d88bff2 : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/348af11bf5e943da84f7132c41c76afb : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d658777b03964fcf97d7aa2effa663b1 : SUCCESS in 6m 12s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ef040f829b7b423196e85889a450b865 : SUCCESS in 14m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ad7736fbb17b412c8ae8d1ad75f6f244 : SUCCESS in 8m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0532aa9546bf44c6a5e55ed271dc8276 : SUCCESS in 9m 35s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8d4de3c5bb8b44f6a17ef81e150c0888 : SUCCESS in 9m 03s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/bf46bae6edb94825a2443df6ae985033 : SUCCESS in 12m 02s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/24e1e0a785834989a5aa0ce5012c3681 : TIMED_OUT in 2h 03m 28s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a2f73898e0ec4fe98ca464127bd1e518 : SUCCESS in 1h 53m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/448eb5b628bf41bcb03400c9af021349 : SUCCESS in 1h 50m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1d99e398d3fa4831865959b82baa02e1 : SUCCESS in 1h 43m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b93f4cc0001940c99604677120dbf13a : FAILURE in 2h 16m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f453e4edba0f47718b4772e49c53c38d : SUCCESS in 1h 12m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d26ddd0ca7f14b54ac9e5c3042f15656 : SUCCESS in 1h 56m 46s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/2dcac0f56161499a80ca2a2db7d0edc7 : SUCCESS in 1h 47m 56s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d796d42d4f6d4fe5a9587aba8c79bcd5 : SUCCESS in 1h 08m 06s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cc7b9a39f9224a13ab51f86a6a494db1 : SUCCESS in 41m 04s","accounts_in_message":[],"_revision_number":3},{"id":"26d1f7454eac31eaed8876f54fb3d4a28037e221","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-09 19:54:58.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"0678fff69cfa9b1d67e32dde01d96d6769521b28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 20:13:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9b73beae86a4f8483292e90bcce1e41\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/23f881ff9b4445c795115fb0557a886e : SUCCESS in 14m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0f6429ba6f224f65a9c9e27a17b98a6f : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1c9546cf157c4fb2b58ec7622ec6bde4 : FAILURE in 11m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f59563b8517b5e0625726d46da591b071202a17f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 22:22:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcf0e7b99fe14dfbab147a501b66c587\n\n- grenade https://zuul.opendev.org/t/openstack/build/1e7d34b5e8c146f68f9cac287ad8c428 : SUCCESS in 51m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c640541b8b854574b55d6dfea25ddfb5 : SUCCESS in 49m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a1f80c561a0e463c8d04708516396d50 : SUCCESS in 1h 18m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdd3a062c621463a977915e9a4fffb20 : SUCCESS in 32m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7da9bd6a09d249bea4f5fc180581d261 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ec7bec078a64e8ba6a05926c2ea02fb : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a002b5483a1b4a469cdbf367f2a9e176 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61f31939f6ba47cdb0581c4bf560d615 : SUCCESS in 6m 26s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ddefdb185f97440982fc6ef55964a495 : SUCCESS in 12m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c39f1e30c4194bd5b692bcb807499b6e : SUCCESS in 10m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c2a2c84b066d47cbac07dca41a42a394 : SUCCESS in 9m 38s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/682c2085430844c5b90eb36002950b0b : SUCCESS in 10m 09s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ba74b2001c934ed7a5c17af7384e812c : FAILURE in 11m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0aeb1e0041724948aa220df1bda9d6d8 : SUCCESS in 1h 45m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/99a63f9b9fd24cb2ba6f0a8901fe9abb : SUCCESS in 1h 36m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/814725bd9dee40b7a53c26794e4237a4 : TIMED_OUT in 2h 05m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/de14e23d8dd84b109f48024140992d66 : SUCCESS in 1h 35m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d44a2052ef0145e8b804b65742f7bbd7 : SUCCESS in 1h 47m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5ec4bdbb22148c99885dd3070621e3f : SUCCESS in 48m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0abab827adfa41a4ac112d8d90507d12 : SUCCESS in 2h 04m 43s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b46d72f12cb34aa780158abda61fd8ee : SUCCESS in 1h 50m 17s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e07912f9cbdd41c8a58127936d8a588e : SUCCESS in 1h 07m 44s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b99f63b0d9ad47529d320f37c93c82b4 : SUCCESS in 32m 10s","accounts_in_message":[],"_revision_number":4},{"id":"99465cd145aed708a0297f3aeec96587445b3d56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-13 18:28:54.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f348d3dc9258fe95f2e9e3efedcdfde80586a4c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 18:48:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e97de7456d7b4ca9b7a59532883ed854\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0c98420d538494fbc124440d060d15b : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9491c977901a4e15ae984736c33196a5 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/109aad0ff85c41869f81ff6778db5fc6 : SUCCESS in 5m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"75da84cec82ad4e62c3f05bba3f15d627ca40343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 21:43:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f37bc8501bc2494abc16d48dd245598a\n\n- grenade https://zuul.opendev.org/t/openstack/build/7cd07f5c90da4a19900e0181b4a96bd9 : SUCCESS in 1h 12m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/10b5711f6ac44121b1f52ae9055a5c5e : SUCCESS in 57m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b34990ea3eb0440a92241c9e027f93ed : SUCCESS in 1h 30m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b86fa197c16d41cd9c342a5cbf4bd4c3 : SUCCESS in 30m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baf0054c202e484c9b1fb8dcf014de28 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fde5714fe0ce43969e9a483969f24d51 : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a856a22a67f40888d22d37f01340c43 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/429bf80b3286435f8476aac8d1950dc2 : SUCCESS in 6m 12s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/784dfac07b9a4bf49aa8ee262aada7ae : SUCCESS in 11m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/27c643b6ff7e46bf9aa2567b78153ed9 : SUCCESS in 8m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d389373ebef5453ca0a0a407bba0a94b : SUCCESS in 9m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a4c06c5438b746d791d0e66e82656ed5 : SUCCESS in 8m 29s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/06b4b29d541b4dffbe8d11518f18c5f5 : SUCCESS in 9m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6592171e2c4d413695d2573d806f5f82 : SUCCESS in 1h 47m 27s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a8bb1f33f0814f1786adad57bffa03d7 : SUCCESS in 1h 35m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ca70105f3dfb4057896355b8413eb295 : SUCCESS in 1h 40m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dffb9d5b6c864455a041178b79a5efda : SUCCESS in 1h 33m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/900ede7bd42e418cbf698bf8082a08a5 : SUCCESS in 2h 04m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ea6e680de2747d2b295c4a1b02118e9 : SUCCESS in 57m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3cabf9ddb61f408187baa76f2336f841 : TIMED_OUT in 2h 34m 02s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/efa2759050bd4cc28c746604230d7137 : SUCCESS in 1h 47m 08s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/54046fc740df446a81e3f2712a59b913 : SUCCESS in 1h 19m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/723128ee18a44d60847468bf43210260 : SUCCESS in 40m 00s","accounts_in_message":[],"_revision_number":5},{"id":"4311ec6f7ad4547e2f5618ae0ce45ae1f19a6344","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-14 04:42:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fdca42cbe080010348290ca1aa3ba602c92c4e1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 04:49:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/904f76a3d7844cb38dfa387823cd4744\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/985c6746d7574183842448c32de03fe6 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bcad06819ca94da899a19a927b5c5a16 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1a115434f95a40ac95a76242899b8595 : SUCCESS in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"248d006910647c071008f36136867d9dfadedf5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-14 05:35:50.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"f38e63dac4df811375eae0fff0305494eb113a63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 05:57:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/241474ecada94ef1b7f0ae2c13a11d2c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d659997fa4a48bd93743fd91d0dc6cc : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/79437ab8e585429899316814c92f61dc : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/83a80cbda40444e6a413b7642b45aaa2 : SUCCESS in 5m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"05c7e94d68616b985bb1fde450274e20c86f01bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 07:41:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/754f5b3c5a21440086182e2d44e02f46\n\n- grenade https://zuul.opendev.org/t/openstack/build/6f4983f06ae5400481839e9857afe606 : SUCCESS in 54m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/889e30ef67d54f6f803c8415b2f19572 : SUCCESS in 1h 02m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1753894730924367a75fbee34751f001 : SUCCESS in 1h 23m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80b2d93347594f71b03731f69c6f0eb0 : SUCCESS in 36m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43b557c3bb8f4051a249995bcd84781b : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36d451a8976b4f0ca8131474cf1ee9eb : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/abcf6f577ee9455284f45ad4e3fd4bc0 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/042ed1c1193e4995b001b2630bd26cf6 : SUCCESS in 6m 46s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e0cccb81ef3c4ad89b1aa4981d1e42c0 : SUCCESS in 12m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/44a31187531a408f8d841ec9d67c8570 : SUCCESS in 9m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2c6d4633033a4994b73429986015318f : SUCCESS in 8m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3a16c28ceb434b528c4a8c234c0867a1 : SUCCESS in 9m 15s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/42ec2c8b55524eb1827865d850c09417 : SUCCESS in 11m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fd2ad8b2c83e4e2995d9d441d1cc62d9 : SUCCESS in 1h 53m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/eb68b49bdb9645adbbbe8354f07b1ba3 : SUCCESS in 1h 37m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1dbd9fbcbc6f404884db1f7790cff881 : SUCCESS in 1h 20m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6618dac440fd4833907bd3b6f3516bac : SUCCESS in 1h 30m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1fb92dd5e2c04ca5a3ab8b71fb3b2237 : SUCCESS in 1h 54m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0fcea778efb47d086b7e415fc08dbce : SUCCESS in 1h 00m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/870d8502af2c49f3b34525e0594d9cf7 : SUCCESS in 1h 56m 35s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/592915cc007b42b99ee8a85dba8aef3a : SUCCESS in 1h 24m 26s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e88a68fc8b9e4cbfb47d657e5cf91778 : SUCCESS in 59m 01s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/458af1737d62422f8ce2c5bfdff5ca61 : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":6},{"id":"658abb32897af1b9920ff3fc99e89ed14ff64c84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-15 17:46:06.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"7f0a1a2fa59ecd8ca479ef0d5d72ec3b29124229","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 18:06:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/affa7a21ca3243669f931ae99f582d4c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6722e40d4df34407816d1a06b78f2450 : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6d526f7c392b4211984c718844054ca1 : SUCCESS in 10m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ef2d3829dbfe41cfa82861dd14ea19e3 : SUCCESS in 7m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bcce71da7a85b69e798f9486caf2fb29dda35d2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 20:16:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56811bcdf7834959b3dd5f4768b8e02f\n\n- grenade https://zuul.opendev.org/t/openstack/build/a4c24f4844dc4c2cb5e23dcacec37248 : SUCCESS in 1h 05m 20s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/09cf5951c77b497f864c0eedcb81d9a7 : SUCCESS in 1h 03m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f366b3d084d04170b14ebda150608923 : SUCCESS in 1h 23m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9754de5bb3c2462ab1019f88fbc570c6 : SUCCESS in 36m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83f9795ce94d40c190daf038dbed6eb4 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5cf21f5b74043e1b46d38c0c0035fab : SUCCESS in 5m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/71cd83aefe4c47c78b9dd8037336006b : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/831196c01773473fa37345a227e73230 : SUCCESS in 7m 51s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0dcf888982c1433792cd99a5a1cbfffd : SUCCESS in 13m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b6eabcec0e234fef8b849ecf449749e2 : SUCCESS in 11m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0b15cba73ea04cf0b3ee841226ff7a30 : SUCCESS in 7m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/539ec16658a74d1cbddf556230b5ea3a : SUCCESS in 8m 45s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/e6b372061050448f92c407fa3af9c266 : SUCCESS in 11m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9fbfbf0605b0421daeda0341daf41c30 : SUCCESS in 1h 22m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d863fdc09bd241e48d8b962be5ad3711 : SUCCESS in 1h 40m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a62a40ade21643d1a16737a4aa8e0aa8 : SUCCESS in 1h 09m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/21959d187d3b44a19e0e3362248e71b6 : SUCCESS in 1h 36m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b194a2c0d5434519b3c13051a98a5423 : SUCCESS in 1h 38m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd280daa38824f70ba9c57745477a129 : SUCCESS in 51m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/29e1231a904f40e49ba0bd58850d89fb : SUCCESS in 2h 14m 45s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/60dd8481d8334b3888041c994b152b46 : SUCCESS in 1h 41m 10s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/6bced594b292432f87fa4be031e1a171 : SUCCESS in 1h 10m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8abc48ef088544c18c1a0c0f8efe31a2 : SUCCESS in 40m 44s","accounts_in_message":[],"_revision_number":7},{"id":"eaa83a0c43ba03c4ebea4ab7785a11a23284a2ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-16 17:12:14.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"5b95c26adb2ad1ca98bd857fe39b15e56d5a6534","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 17:22:42.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65649394f2fc499e95b03efb471caeb9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/92ce48e370aa4ad79cf35b4351b4660e : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9289a0a8f93d4e89b4e8934b64999c95 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9ea22357df9e4be396299cc8cda20110 : SUCCESS in 5m 43s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"63dbcc3962d1b822c8221918fbad8cf408107e24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 19:25:24.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea975fefe5594b1fb5827373c359689c\n\n- grenade https://zuul.opendev.org/t/openstack/build/fb9876326a914be1a4354e74dde175fc : SUCCESS in 1h 15m 07s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/73215806a13b4299a974a663db488d33 : SUCCESS in 1h 05m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/04704ed19cae45c5a90d9a8a2677a1fc : SUCCESS in 1h 24m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83ec3341d91445d6ac7e9f8f0ece7fc4 : SUCCESS in 1h 07m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0353ff40cc7f401da542d2fbbae82835 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8aff7bb6bac477381b2db6854dcbc3b : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1ba04ffd75d4a1380a3fe53ff989652 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6310c857c89d4fd7a02268b4d78f1093 : SUCCESS in 6m 38s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/77118578c854438a83b8e05242df7b66 : SUCCESS in 13m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4c4cef00244148759a254f1948353887 : SUCCESS in 10m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/06a5ffb576d04e08b4dde5fd01eaf38a : SUCCESS in 8m 35s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a69963006efe4005a9c4400fa3a5280e : SUCCESS in 8m 29s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/9f7d0ec227eb4f9b96350bcf938fbbea : SUCCESS in 11m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d399066de53a44e383f05da22bc39233 : SUCCESS in 1h 40m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0e9477a6d4ae427a8319fc0ad239a7d6 : SUCCESS in 1h 55m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6599f28e9e5f409aa18b64907c660e94 : SUCCESS in 1h 41m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/31e5f2381e0a4ea798bd2b4dc6afbd21 : SUCCESS in 1h 34m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bc9d58a430124fe9a961d07267a4ff02 : SUCCESS in 1h 46m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4efb8782da024968a40989266f477c7a : SUCCESS in 57m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/90f2d150fa4f457e8eb6c851d14fc30f : SUCCESS in 1h 43m 24s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/882092e11edf4247b8c686bb1798ce75 : SUCCESS in 1h 35m 44s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/77aecf4fdf264d6e983b5b3ea47ea99b : SUCCESS in 1h 01m 27s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/683b0a16ea5e418daef9df704830742c : SUCCESS in 41m 07s","accounts_in_message":[],"_revision_number":8},{"id":"ef34142939fbdf9d10fe63af93719f94c0597b9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-20 20:30:08.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"730acd2f2967b02864ec59dbd764c2b14a4bd9b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 20:44:12.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b85aa8bf14c746cabed4cfa66b06bbb2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c6a53123da8434ab6eaf9f0018303c8 : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6ae1c0de61564aa1b090a5f77958ad1b : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2c311b5c9e104a6ab4a311e6a8fb2caf : SUCCESS in 6m 56s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b6f815df5426f75dec1b1291be00135962cf1ccd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 22:59:00.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb8cea8a4b6146b2ba72ba1da58f8e6f\n\n- grenade https://zuul.opendev.org/t/openstack/build/3fc1478585574544a2affac023ee8293 : SUCCESS in 1h 02m 05s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ac316eb6ada9403bbde7af53ca5fe04b : SUCCESS in 1h 01m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ab415e41f9e642a9b9fdcf3bb4c546e9 : SUCCESS in 1h 27m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5781575a946c4d53adcb3e049e0d3a3a : SUCCESS in 1h 01m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f531ca9685e44c098a81f7e1c6fca72f : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f53a86a09f3f42388583495f53795daa : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/907413c3b941425fb52cdc84a854d54a : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95457bee2ca749a282a2fa128c0b6cff : SUCCESS in 6m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d106ec790f4e4a099f568c58891fec61 : SUCCESS in 14m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5ca65ed135734090b11503de3e480028 : SUCCESS in 9m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8cca36b251504c37941825294ec5eda1 : SUCCESS in 9m 17s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7e309b3d79fe482cac72d1a3c5d442e7 : SUCCESS in 10m 21s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/68b0ab74f696431db96033d09c626630 : SUCCESS in 12m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3de21afcd8894a60a53cdc06a5c31c29 : TIMED_OUT in 2h 03m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/88fec067fe1c477fb9a904cb0504620a : TIMED_OUT in 2h 04m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f8b901eaa07d4703a46ca6be6e675e79 : SUCCESS in 1h 50m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c3dffb14c82948749e4aa09e755b3578 : SUCCESS in 1h 48m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5c053ad371314f2f92f0506762ffe594 : FAILURE in 2h 04m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84e8913279154b5e96693b513c2e6325 : SUCCESS in 1h 00m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e70be408925047ee86ffaecd7b728cb3 : SUCCESS in 2h 05m 41s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/aa5c87310bdc428a90093f5ea97e6627 : SUCCESS in 1h 17m 06s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b481c6e21a2642a18b7d7493d67c55b1 : SUCCESS in 1h 07m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e014e86a4319455598b15de2cd132c52 : SUCCESS in 39m 12s","accounts_in_message":[],"_revision_number":9},{"id":"d70b4fc76ae72b807f3dc28c666dabe913b98acf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-21 05:23:52.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f1245c1374cc158b8094741fefbc3323471f683a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 05:37:18.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d71035bfbf214803ba33a5a932d06f71\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2fcb6c5b99d9460a9e4c2755a1b96d12 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9be3f96bdf0144e686c81adf33b135a7 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1d3e019dd5574def838aeb0bd599eff3 : SUCCESS in 10m 55s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"34730d45cddf13f0962f89f19cdb548db66deb6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 07:39:28.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c045a4efa7454d7f90e0ab7ff410819c\n\n- grenade https://zuul.opendev.org/t/openstack/build/3405a74034f44de7b46518c335196b0e : SUCCESS in 47m 15s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/66385ac9faf648fc80e0aca5a21e2ce6 : SUCCESS in 43m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b52a524f6b9d468382b55ac944fbc1e1 : SUCCESS in 1h 29m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df70659c19884a00aa18ee03dbf67810 : SUCCESS in 51m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dc94762e1a44d039fa501582be5b6f3 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a5f379d3eee4513996fc6485ce97c09 : SUCCESS in 5m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/503b08f5ee604eb8bc4f05f205e8055b : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c2e6f4be26d49a783a5abb4fc959dd8 : SUCCESS in 6m 26s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7f1029590d024ddc80e5ae30873638e9 : SUCCESS in 14m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f2228881732b4abd8017a8ff5cca347a : SUCCESS in 9m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/02f11d5a766046c1ac021a9bbb66769f : SUCCESS in 8m 49s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c87e6cc11bb84e9080ff4d27ff9ef71f : SUCCESS in 11m 10s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b06d27dfebe64666b5c13f76f3713f73 : SUCCESS in 12m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d7e7caece4af4de3b846d979bc10ac42 : SUCCESS in 1h 38m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/03e23a118659458c9ae07803763ab38a : SUCCESS in 1h 40m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f13a7a1e716e498590c3d427b44ce502 : SUCCESS in 1h 43m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d525c05099c94466afc80bb06569aa50 : SUCCESS in 2h 01m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/97d54f52ab8c46f493e3c1ffdb4f80a5 : SUCCESS in 1h 42m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b7b6b9be0e4478a8f7c5ab9742c4009 : SUCCESS in 40m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e3a9fdfb6c2247c688f1143dcaee272f : SUCCESS in 1h 24m 28s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/161029794b3a4aec96157b10d82a2ff8 : SUCCESS in 1h 39m 12s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/61d809111ab142d4ad796c9262702790 : SUCCESS in 1h 02m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c07aac9a5a1a44499f3b38b1996456d4 : SUCCESS in 42m 15s","accounts_in_message":[],"_revision_number":9},{"id":"17a65be497bdb4dbe4429e3ab9a27ac0f6b489d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-27 18:45:05.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"d9c16d83a44ef96404c050cee53475ff6d1de069","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 18:54:00.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fc7f878bc2d4e23964297863481a73e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13c32cc833034847863532e56d097619 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e8b17575f2a4185a4632184bdaf1671 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f10f33f2d8274fae99ed5da54b8c09c4 : SUCCESS in 6m 05s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"2385dc3ac9b111e7c4499c00d8ea1d0c98d8bcf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 20:53:14.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c5d4ece84104ee783e157fd1f43fef7\n\n- grenade https://zuul.opendev.org/t/openstack/build/d35bd45ab73343ab9922533ec7aaee97 : SUCCESS in 1h 10m 01s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8bd434d98f314805861bb72756019df6 : SUCCESS in 1h 01m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b41e6c193c2a442cb2111f8e927480e5 : SUCCESS in 1h 28m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/465dd7be0bb547ec857c77c888cd809a : SUCCESS in 1h 01m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20bcc7b1ed8245298e8d7eb0b1e20a79 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d0764f41a3f4bef92e41fa3c1616643 : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c9676207a3104688851ddd74c8844e72 : FAILURE in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0908d173f3f04a79b2e979e1ec334b0d : SUCCESS in 5m 48s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/6a4c869de5784bbb9b76e00988592dd5 : FAILURE in 13m 17s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0990815f3a04498b972d06db5329f85e : SUCCESS in 9m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9fee7e97f8ee4b41ad7d11460b1b4215 : SUCCESS in 9m 22s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ad74b4228a2f4fd289c0b67a7a63b201 : SUCCESS in 8m 32s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5120dcce8560486381a1de6cf822ce8a : SUCCESS in 10m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/41175da37f54442b9750cebd1ee56344 : SUCCESS in 1h 55m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/31dd52970ba84442bbe40e24929fcfe0 : SUCCESS in 1h 48m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8cd6d85dbfe0431eaef48e6a5d68d69e : SUCCESS in 1h 29m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/eff71e262ca84ecd9a32d8da0b0f7f18 : SUCCESS in 1h 48m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c59bc09914714e7f934fc9ae7669c2f2 : SUCCESS in 2h 00m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8ac10c2e2f24c88a1b835df5ce619a2 : SUCCESS in 1h 06m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0abef0d003a54fe6b8fbe9297291a574 : SUCCESS in 1h 47m 52s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/8b7a0bce9b8d4288818ca6b57fbf6bb2 : SUCCESS in 1h 42m 13s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f35140c7a5ab4b3794f61a951ce31298 : SUCCESS in 1h 05m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f038429df43946a3a3007c23cfd1a132 : SUCCESS in 28m 39s","accounts_in_message":[],"_revision_number":10},{"id":"4a6895b00934e0017a2edac687864c74c9d26581","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-28 04:13:45.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"197eb0bde6f93c678a31bd666506b9597c705731","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 04:24:43.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f275aca33774a4ea51784061ebcce60\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc87573ca1f1450cbe293da777f30b85 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/73fa5e983ff8436ea935100bf42e1401 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4b97b5765a544b5799d4131115a39a12 : SUCCESS in 8m 11s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8ffcffcda52cae0d45b65984de8804ec8ff4c666","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 06:18:57.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9533cad950d44845963521a62d99b124\n\n- grenade https://zuul.opendev.org/t/openstack/build/96c5f563c7f44451acf34f2b89245f51 : SUCCESS in 58m 03s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8430ee68f32c4b4e83102745c871a3a3 : SUCCESS in 54m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a7f39af2d41a44c289285426c9b59c15 : SUCCESS in 1h 23m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75b9fd585c2141b0b481dfc6b7e055e0 : SUCCESS in 59m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/206be9e441524585ba23b428dab574e5 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ed2416c3243414fa89111011220a246 : SUCCESS in 6m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73e00b6093ac44c481f8c00cb388f274 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6c1592a23b94f7bbe67dd7e95a21d8f : SUCCESS in 7m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c484518e84034b8c9320b39f0558aefa : SUCCESS in 13m 06s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/12c993d9948241589115d9d53e7cd09d : SUCCESS in 10m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d4f26908a06148beb414a12dbc164522 : SUCCESS in 8m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a9f27efa27054187ac4e6ebb3044768c : SUCCESS in 16m 38s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/48f813d5f8c84e79b4d83c729e30e46c : SUCCESS in 12m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4749acf642974e58b9cc481116f1347d : TIMED_OUT in 2h 03m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9bd62f721be943f1a034f6a87cc1050c : SUCCESS in 1h 34m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0f0653c8376c41889c754c6a1c58eee4 : SUCCESS in 1h 43m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/83f6c237b4b04104b4e538f4b2636307 : SUCCESS in 1h 37m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b127f7ddd63a4ebcaae76093ffc94f9e : FAILURE in 1h 36m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec5993de744742a8bc9e74d8af1872a8 : SUCCESS in 51m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0107b0ed6e704b93ba6b8a8674bffd30 : SUCCESS in 1h 48m 08s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/7b9f559b1fac4ecab27de9f5e6b7fc47 : SUCCESS in 1h 39m 25s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/61e30513925d4f368b23410901e4dd6c : SUCCESS in 1h 08m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3cc35bd4e9e445a58c04236cb1c7ca03 : SUCCESS in 37m 16s","accounts_in_message":[],"_revision_number":10},{"id":"6236f5aeb9fd6c676d577b38956a7861024fd6c1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-28 09:39:39.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"dbe3d644776c458c3c3bafcdc99e605f27162891","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 10:01:00.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/901649, https://review.opendev.org/901429 are needed.","accounts_in_message":[],"_revision_number":10},{"id":"8926398ab5f007353f69f517d4bab070ed2c70f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 10:01:09.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/901649, https://review.opendev.org/901429 are needed.","accounts_in_message":[],"_revision_number":10},{"id":"2c544be3adf7d0ee1d00d51a77cddcb9d350e43f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-28 10:04:42.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"14fcc74e41ebba94bd194226eed04563ad2d367e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 10:22:17.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f152e3c708044d0bc9ebe3ccf1abd2b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4de884d0a1c142c3a1d48919d5ee6ac2 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7550d110166441b68ed113873f1adf21 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/227c593a620f4b8cb00041f4adb74c74 : SUCCESS in 7m 47s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"827e792067debdbca2bb84c306352ea49ea4e715","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 13:56:38.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4bc2d99f1f247a187f64c4f5a46a470\n\n- grenade https://zuul.opendev.org/t/openstack/build/3c77be07dd19493da0a2b58906c9d89c : SUCCESS in 1h 20m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e2846d9f3705465fae46a886f013244a : SUCCESS in 1h 07m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9371bb827a21425d8a3a9d5d7c467f91 : SUCCESS in 1h 29m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2779444fde7d469f8e47b876ae256f93 : SUCCESS in 1h 01m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9acdb02d8b346b4a88b61060ef948ce : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eed3943fd19445fb83740f1d07c412b5 : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/da3dfe865b8d483c99718ea7c8cfb828 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04fe4b0cb70c45679ac689e4482501ab : SUCCESS in 5m 48s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/935efaa9a79a43329dd9a77715ecbbd1 : SUCCESS in 12m 13s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/526fb61e2c464025a3f3cd4a63ec1fc1 : SUCCESS in 10m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/18b20131528c4fce9d111cfdb9a1843e : SUCCESS in 8m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/22ffdf867ce24dc9979335e77562cf78 : SUCCESS in 9m 59s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/767360c50291402b9e3945a11e393e7d : SUCCESS in 11m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5512bd0eb3cd4d0d81e0ba029c1e0ca5 : SUCCESS in 2h 01m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/660db050926244a7861a4bf2898fa0a4 : SUCCESS in 1h 55m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/34a459b496834347a1e1c158a49395ec : SUCCESS in 1h 44m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f7105daf7de64e52b06f99bb56e3f993 : SUCCESS in 1h 50m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/36a2a317e8444a09bbd48b9f842db86b : SUCCESS in 1h 38m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/140f429581ef488990b054f42b76c6da : SUCCESS in 52m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a3ff4d30fe8f4671ac8333164265c8c8 : SUCCESS in 2h 11m 18s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/346fd107b9ac4453b8c60c2455f33ad7 : SUCCESS in 1h 55m 58s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/42cbb0b9e8c642f490e653d026c53f80 : SUCCESS in 1h 16m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5b04e52f8c804ffaa732e8efd20c93b0 : SUCCESS in 51m 55s","accounts_in_message":[],"_revision_number":10},{"id":"cca4296ebb861818a12c31fa8f66b00b2ecc69c5","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-28 14:47:15.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"405640004434c621e2aa573a36a3d828a5d0641d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-28 15:06:06.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"689f88784d7c19dbac4412b3e43d912b0486d128","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 15:30:44.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5015b47d17a4d5daad3696077c0149b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7811d341a794c588de644e813fde708 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7ea5c30a519143cea2e181e84de5c279 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/951f6554d0f14aac82785d23a148d4a6 : SUCCESS in 9m 03s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"fe8c9c4ae4410cc4fb0b244af612184e04542739","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 17:45:28.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/26f3f2449dd44414b7187022f5745937\n\n- grenade https://zuul.opendev.org/t/openstack/build/1ffd1de82cfe4e47ba84bdc3c8e229bb : SUCCESS in 1h 08m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3fd6ae88f0974caeb3051f07bfe88a8a : SUCCESS in 54m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c6e7c8d90ff2486f8f7f92b02f9a2ab9 : SUCCESS in 1h 23m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a83b67e3b25461091c0685a0323adc9 : SUCCESS in 1h 03m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e0049273bfc4cc68091e0c67b7e376b : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/693ecd2966c94e46842cb41898272cc4 : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/81c68bbbcb8348a1b7d4a2117b3fd813 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85aa51c44b4143ffaaaba51beebcc1ab : SUCCESS in 6m 13s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3aa272e261ba4825af9aa4481e466ad0 : FAILURE in 12m 58s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/748cd839f6a2424cb69695c513a95895 : SUCCESS in 11m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/525ad5a7d1b8466b9e7f0d7a53acdfeb : SUCCESS in 9m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/80356c3fe48248f19f192473022fdc64 : SUCCESS in 12m 39s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/e76557f93d0e46caa25e858763d9d5d6 : SUCCESS in 11m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ac28e97e7b704553bb6373389f78c206 : TIMED_OUT in 2h 04m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e435d6abbaae4c5695545ce718010628 : SUCCESS in 1h 51m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a0ddcf13de7a4aa28d73a0d83b8b301d : SUCCESS in 1h 48m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/db98b6f6a4f4404a9394145e61508fb3 : SUCCESS in 1h 51m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/45ac4755c6014194829e6adc2835445f : FAILURE in 1h 41m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22fe583302734189979e722e9441e2d4 : SUCCESS in 53m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f8c5ec1e011a4cfc9362b49d9abfdad1 : SUCCESS in 1h 59m 24s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/a8b2fe05f577421693ea3c188cdcb2b0 : SUCCESS in 1h 59m 30s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d27ae4538c8b475e9c0957d57eb6b398 : SUCCESS in 1h 20m 37s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/28dd5a2c56e04b6a91b58cf6934ec918 : SUCCESS in 39m 39s","accounts_in_message":[],"_revision_number":11},{"id":"28fe3875929b5267dacf7590bb165bde468eea60","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-28 17:53:34.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"1b19ab9c9cf8cc90e979654fa2cb28d16abd128d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 18:03:55.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9552ae0bcca43b6b8c880a27a5698db\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c1a590b818684b00873ef88ff6037605 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b64a255cc60d416cadbf440255f8fe10 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8ec0ab8a270c47e787035b481d397ae3 : SUCCESS in 6m 01s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e11eb3f3df9c9a611775a39aaffbd19d3bb66c39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 20:03:25.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f88c4ba3f874310b8bad7e1f6004066\n\n- grenade https://zuul.opendev.org/t/openstack/build/01d6b1acd3494b6d88fc29148c8705cb : SUCCESS in 1h 05m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/77891202ef78406db4a8699b3bbcbecd : SUCCESS in 1h 05m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a5cdea962dce40b382cea952c7edee9f : SUCCESS in 1h 32m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35e6c41fea1248bd9b3d13470688c094 : SUCCESS in 58m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ca44c95f98f4467986724363b2c6898 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5233e53dd7a4d7d8f84d8fd838c9f11 : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/197595d884d643b4a8fb606296a3c1ff : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/330a9a42781d43a8bb000d3cb8f323ed : SUCCESS in 6m 26s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9a7f844c6c1647bb9167bbbae94fab95 : FAILURE in 15m 06s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/973024b45ef34754a7783692255f74cc : SUCCESS in 8m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b9231bb7b9ae4860b174e1745d04b550 : SUCCESS in 8m 46s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3a49c7c660174d028883b86657bfb3a9 : SUCCESS in 8m 43s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/74fc113ad14649cb915c96bb687f49c9 : SUCCESS in 12m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d489e7a2437e428e9909a2b0b9e2172b : SUCCESS in 2h 04m 00s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9b91ac283d2149bf8bb8aa228135119a : SUCCESS in 1h 48m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cf4603a7b2ff4d959a20c4ee0d047a17 : SUCCESS in 1h 42m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/21ac517c7c2449c09713bf1be798efdf : SUCCESS in 1h 44m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/259ad853b05d4a53aa80d33fe902d9b9 : SUCCESS in 2h 04m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6a075d315444946b4aa225f0a379b81 : SUCCESS in 50m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dc898937636646fc90c5a23596a4ed64 : SUCCESS in 1h 46m 28s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/f33e8872937b4bfd805ae9d6287e37cd : SUCCESS in 1h 19m 59s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2f308548652c4d899bdb832778904f59 : SUCCESS in 1h 04m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/730f1059128d4cb3946dd937e3527f0b : SUCCESS in 37m 11s","accounts_in_message":[],"_revision_number":11},{"id":"5dc522f4e27d64ac81ffa3386a925e9232668d94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-03-01 05:45:41.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"957bb1c70da2546836edfd94f57b77cda0d8fa71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 05:59:54.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/021ca671b709470095b07df1f9a8396d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/353eef20f25542fcbcd1ad53d7b78423 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3639483c234647dbaa7ad1e04f8a2af4 : SUCCESS in 11m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b4f2c4c3256343c48fe6dd42a6b3d8da : SUCCESS in 4m 55s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"2220b2c1033aad8c444bd0f534ca56488f4717a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 08:00:24.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e66684a6a5a84759af2d37a4a1b53308\n\n- grenade https://zuul.opendev.org/t/openstack/build/46e8da967bee4afb8fe084ec7b0cbd76 : SUCCESS in 1h 00m 21s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/dab2af906df44ef5a82bf301d87c0745 : SUCCESS in 1h 10m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a5d303e4f9e74e6599fb20e70a295648 : SUCCESS in 1h 16m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80aa78fe15f54f1e8732bc9de25ff8db : SUCCESS in 51m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0b8af6372e5419ca8e58310b3692cff : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/935e741c004c4fdfb82e1df282983c77 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e816a20dd7da431fbda6f5d1ba91f1b5 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a3f742af3a543f4b866dcb251f26707 : SUCCESS in 7m 36s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b6b662328e534f3d85b2f2dbaa9516b4 : FAILURE in 14m 06s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6d7034e44abe44c89e04874f79d2ec1c : SUCCESS in 9m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/364fc57d6ad54d73a88ffa0f7ae899c6 : SUCCESS in 11m 06s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6d100658803549bdad8c88a2d5d7762a : SUCCESS in 9m 15s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/8daf332158ba4f708437982b9aaeb8ed : SUCCESS in 11m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0e503a036815448d8082091c87205a71 : TIMED_OUT in 2h 03m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/64b6617e53814828921f9c515e01f68a : SUCCESS in 1h 39m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7488fda9e3f847f3bfa1e4d6dd87d9d4 : SUCCESS in 1h 43m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f31efa0f590c45a49f22ff2ca3a6ba9e : SUCCESS in 1h 35m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ca1cd3b554e048899df5d217435d44ba : SUCCESS in 1h 53m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ff0df4cf4fc452091ee1fff3fc91990 : SUCCESS in 1h 05m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/17dbabfb61e84e7b86b34bd3dd10e8ee : SUCCESS in 2h 00m 35s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/8eda1f921d1549dfbb30a073371febf6 : SUCCESS in 1h 24m 03s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/60d14bbd33464ce9894d6f93532c9b3c : SUCCESS in 1h 20m 38s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/49e066323c4c45b5b409501264d8e95e : SUCCESS in 40m 30s","accounts_in_message":[],"_revision_number":12},{"id":"367574f87d27bf97079c5cd2f8ff7218258d8e67","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-03-05 10:15:11.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"6260469e518132d06eacd2062e9080fa9ea57a89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-03-05 18:39:34.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"f3039055a3b8db224e3994fb58e51ffd2e5e3509","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 19:06:13.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/743f4702f62c4d9a94c71613d758223b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6006b8cefe14341a802d480d0f59174 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3d51015ba8f849deb170f7d001a68061 : SUCCESS in 16m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/96712a83a1b949d5a95f1483b8db4d2b : SUCCESS in 17m 07s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"adbf413823a60839d7c686acd3a403ff783d0fca","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-03-05 19:30:52.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"fda969b1c8986e4ccd12248e229865fb55e3ac39","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-03-05 19:50:27.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"b942a66e11ae39765dc36352ab0458064be0814b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-03-05 19:55:28.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"6dc05312547839dead60f0537fb02f05989855ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-03-05 19:55:40.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":14},{"id":"b422c7dc2613d904dd61e282be45f81f3cf28971","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-03-05 19:56:30.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"dae07b521f2053210520199eb8acb7a6b4532e42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 20:00:56.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/731a634697b540379d7a23b7bc7bc63b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab6064cc162342b9917e9eb12acbc12b : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a90917983d5549dc8262a26bbbe7d2d9 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d744721b05c84538bfc04fe619676ad2 : SUCCESS in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"8bceef061190a7a6533f707f05bbd6dc20656e05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 21:49:23.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22b00425ad4e4dceb270841683fc124d\n\n- grenade https://zuul.opendev.org/t/openstack/build/691df326f03c4997bd271a4f5e021947 : SUCCESS in 59m 37s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f2fb7a77772844959fcc3f4e938bc4dd : SUCCESS in 1h 01m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1b3cf858d08142479e005b666cf56746 : SUCCESS in 1h 34m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05647f44df9241e2a49f6469c9343cc0 : SUCCESS in 1h 07m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd8d69b47d9a4bf0a8138276d42d3fa9 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a38fe494ea584f57a6915e8cce64fdd9 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b09900224a89452ebbee82bff3006ae5 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35cf01b39002439c97c847764e33ee25 : SUCCESS in 7m 01s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4b3e7f652535434d9690c88ba715c2d3 : FAILURE in 13m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d5cbff0dbcee49f086c92474bf374a4b : SUCCESS in 8m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5a34aff7a1c248baae354c5d91434c24 : SUCCESS in 9m 25s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/97fb9b98a80048c5ac12c4b4906309cf : SUCCESS in 7m 41s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b9bab6750fda4af2811c4ea19eb95ae3 : SUCCESS in 10m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3dbfae5f956746b5b4974dda6a28080a : SUCCESS in 1h 35m 03s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9cbb490ff6d844acb0dfd16b512d1fe8 : SUCCESS in 1h 31m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0c0c73dac6584a0c96be1da33f284ff4 : SUCCESS in 1h 40m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2a267a43f61d435b85ad1b9718305af5 : SUCCESS in 1h 24m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c16da4bd51cd449dbce495a595cabcb4 : SUCCESS in 1h 49m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c3699522be94a13a37404f27e296bd0 : SUCCESS in 1h 01m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c4240acf98f847c5be329b3a932efb44 : SUCCESS in 1h 39m 42s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/c4165d8a1c5b4324bd0fc02a1b05986b : SUCCESS in 1h 47m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d83f7b6c9fde4ff58ed50414291b07a0 : SUCCESS in 38m 02s","accounts_in_message":[],"_revision_number":14},{"id":"d03e4b513379718c8ca16c3f559598e95e012dd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 21:50:21.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"2b3736e16e4bb020aa6a221c32384162b083a873","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 23:50:47.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f07a269409da404e8c8d89c6fa92518a\n\n- grenade https://zuul.opendev.org/t/openstack/build/dd8ff6a1f18b4e3a9addb395b3e7a6b6 : SUCCESS in 53m 11s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2e2b631533c34818b9121db74529f272 : SUCCESS in 1h 00m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f5a7c1430d00430db9d8b7cd55da3e78 : SUCCESS in 1h 17m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/978960f815c84308bb2ed08ff74e33e7 : SUCCESS in 43m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dc31ea4eb94483d8b2f73cdbc3746e2 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c9053f721f0423f9e0c32062b675e01 : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bde3b05a7c1b4f7d9cb562a3cfa6aa19 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8662ff12357f4fe2afcb5c044c8600c0 : SUCCESS in 6m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/39f563d88bf94cc3acf4c5056919f10c : SUCCESS in 9m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c9e4bd6c773d4fef94dfbf576d32771d : FAILURE in 9m 06s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/84a8c4aa44554c82a7e83b316672906c : SUCCESS in 1h 08m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/35b81f1635fb4f23b02f9afbf8e08d79 : SUCCESS in 1h 19m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/02f6034023e3461b8f6b1e43a80a7e82 : SUCCESS in 1h 52m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9f1fbcfb6d34724b58b01d78e67e6b2 : SUCCESS in 58m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cdd62357f20f4e53b2f00ffed515820a : SUCCESS in 1h 52m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a6320e26bdbd42e5af227d0ed7de2c73 : SUCCESS in 33m 14s","accounts_in_message":[],"_revision_number":14},{"id":"99aa1e22ee643b7d40eea2af5251147a69be1bf4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-03-06 04:10:23.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"c5e4512afbfb0dc6d81fe85841a2432589a7c031","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 04:22:19.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d359dcf4bf424ea280d858f06176541a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/32e28346998e490a9ee75c6fe52f500b : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b4d46f7b76fe410ba30e66520e0b30f3 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0c4241587950478ebd784c5b4af71ffa : SUCCESS in 9m 52s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"0ef502ead743ffadb17b040a50fafed34e1312bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 06:18:04.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71cf60ac72974105b0cc33df1f6c3c0f\n\n- grenade https://zuul.opendev.org/t/openstack/build/3371e22c74f1498db7c0e88982d9c406 : SUCCESS in 41m 56s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d2364849fe224101ae0da8f277069356 : SUCCESS in 51m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/235e064523e54b5b8ef1c93958868ac9 : SUCCESS in 1h 32m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c54f9e0180e494d99d616709b35b59f : SUCCESS in 51m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/267c45d8815f4f4b929253c533ebb089 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e190bb964214ee2b348f769b7f62a24 : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ddd3fc20695d47c78b56a8812ab15e7d : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1895f94b764d4a2d862ce7106854ea32 : SUCCESS in 5m 09s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f82ecc2593154b45847a7e06f33df5cc : FAILURE in 11m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cd6dc579e6f443c69455169af4b7987b : SUCCESS in 8m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c38811547ccf4a51953192877b2fe425 : SUCCESS in 9m 16s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f916eba368a5439893c9717e41f2fcb6 : SUCCESS in 8m 32s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f2744e28951745da8a267903e15fa4ed : SUCCESS in 10m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dd15367f203f4fc2935750b4fab6b703 : SUCCESS in 1h 54m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/505e6483a0c047c78faa4947a059c2a6 : SUCCESS in 1h 19m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6a7b81dac5814ee4951c02a6ca9d60f7 : SUCCESS in 1h 27m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2a46082dd73a451290b08157f13a9931 : SUCCESS in 1h 58m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c910ff033f12414987b9b858415ad600 : SUCCESS in 1h 36m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d1a37345e474d58acc2b0817e5c3246 : SUCCESS in 36m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2e32e0ce62a849d5896bb8bea397d3ff : SUCCESS in 1h 31m 41s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/03c1d7f338e140389d9d2ca35d1d0747 : SUCCESS in 1h 37m 18s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/741d824cdbe34908897d9b686ad388c5 : SUCCESS in 25m 15s","accounts_in_message":[],"_revision_number":14},{"id":"8ba06973822fb3684848263590c3ad8af8c7dea7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 06:18:39.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"c3b2fa1a26080b957dfd79af8709549fbf66e993","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 08:17:12.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6526e4312b841c79fe756f20891a034\n\n- grenade https://zuul.opendev.org/t/openstack/build/27d2e7f2fec54a6480a95f7c5df24618 : SUCCESS in 53m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f98cf911409f4bdaa22374199ce71204 : SUCCESS in 1h 00m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b8062b789ad94e2aa8bea4f2cf7335d7 : SUCCESS in 1h 31m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/209dbece547e4f82958a8355ff15d2be : SUCCESS in 58m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53f3feb26c3945e58827a521ac98a578 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eca77240f51741189435603c4c662fe6 : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/888a9729cc3c40d3b001be62fc8e742b : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ab4cdd5068a4d2e999dd9269ecddb31 : SUCCESS in 6m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/48b243c988b54274951d831806b0ce43 : SUCCESS in 8m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ec055fcae2314ca69c10971d3aaa5882 : SUCCESS in 7m 43s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/06c0c602402f4216b42d5be3719b84f9 : SUCCESS in 1h 37m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4ec8579440294b0ea0d84a2f49e5759b : SUCCESS in 1h 30m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/490eb41facee4e089a7d0b2140b5e8b0 : SUCCESS in 1h 28m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff159c96530245129e108a400356f6ca : SUCCESS in 42m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e5d29919e70d49ffbf6d84d80d4f8c79 : SUCCESS in 1h 56m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bfbd1033010f462f871fad37f1d4c0ea : SUCCESS in 27m 54s","accounts_in_message":[],"_revision_number":14},{"id":"12675d7e0dd0c25b6fe193d0713f1bb53d7d3817","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 08:17:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"3af5cc2a4f1209ecad2a15f9048363837bf8d5eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 08:18:35.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d341c840c1254821949f38c0674d026e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/339c2eaba7844c1fbc9a80ee52145392 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"9b9111f81996a22527f9622d0707891e445cbd5f","revisions":{"4beb50c37251b4c673ee71ae173e9d5ebb4ab441":{"kind":"REWORK","_number":1,"created":"2024-01-29 18:47:26.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/907111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/907111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/907111/1"}}},"commit":{"parents":[{"commit":"637ba087a55962f8c5e024ba3adbde39c386906b","subject":"New grenade job to upgrding cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/637ba087a55962f8c5e024ba3adbde39c386906b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-29 18:26:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-29 18:26:10.000000000","tz":0},"subject":"Make `centralized_db` cache driver default","message":"Make `centralized_db` cache driver default\n\nMade `centralized_db` cache driver as default driver so that we\ncan test it using tempest jobs in gate.\n\nImplements blueprint centralized-cache-db\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/907110\n\nChange-Id: Id94e93e3ba3fc207b39c7dbff92495805aa0f6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4beb50c37251b4c673ee71ae173e9d5ebb4ab441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4beb50c37251b4c673ee71ae173e9d5ebb4ab441"}]},"branch":"refs/heads/master"},"8fc297210f1d37909923417f7f652cbdeec06016":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-01-31 21:23:20.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/907111/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/907111/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/907111/2"}}},"commit":{"parents":[{"commit":"6687244557f7f945e41d719f9705aab083d31a31","subject":"New grenade job to upgrding cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6687244557f7f945e41d719f9705aab083d31a31"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-29 18:26:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-31 21:02:49.000000000","tz":0},"subject":"Make `centralized_db` cache driver default","message":"Make `centralized_db` cache driver default\n\nMade `centralized_db` cache driver as default driver so that we\ncan test it using tempest jobs in gate.\n\nImplements blueprint centralized-cache-db\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/907110\n\nChange-Id: Id94e93e3ba3fc207b39c7dbff92495805aa0f6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fc297210f1d37909923417f7f652cbdeec06016"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fc297210f1d37909923417f7f652cbdeec06016"}]},"branch":"refs/heads/master"},"b812b8eabddd62abc475d6f859f7f9dc0b153b10":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-02-09 10:05:42.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/907111/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/907111/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/907111/3"}}},"commit":{"parents":[{"commit":"35c47dcd4e3f650810ecff19566514a904c82d23","subject":"New grenade job to upgrding cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35c47dcd4e3f650810ecff19566514a904c82d23"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-29 18:26:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-09 09:38:46.000000000","tz":0},"subject":"Make `centralized_db` cache driver default","message":"Make `centralized_db` cache driver default\n\nMade `centralized_db` cache driver as default driver so that we\ncan test it using tempest jobs in gate.\n\nImplements blueprint centralized-cache-db\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/907110\n\nChange-Id: Id94e93e3ba3fc207b39c7dbff92495805aa0f6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b812b8eabddd62abc475d6f859f7f9dc0b153b10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b812b8eabddd62abc475d6f859f7f9dc0b153b10"}]},"branch":"refs/heads/master"},"dac28647af962f82ac367e6444605a86a17545bb":{"kind":"REWORK","_number":4,"created":"2024-02-09 19:54:58.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/907111/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/907111/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/907111/4"}}},"commit":{"parents":[{"commit":"d9e3dbeb040f4465194c2c8bbeeae2511c195f3a","subject":"New grenade job to upgrding cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9e3dbeb040f4465194c2c8bbeeae2511c195f3a"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-29 18:26:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-09 19:32:09.000000000","tz":0},"subject":"Make `centralized_db` cache driver default","message":"Make `centralized_db` cache driver default\n\nMade `centralized_db` cache driver as default driver so that we\ncan test it using tempest jobs in gate.\n\nImplements blueprint centralized-cache-db\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/907110\n\nChange-Id: Id94e93e3ba3fc207b39c7dbff92495805aa0f6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dac28647af962f82ac367e6444605a86a17545bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dac28647af962f82ac367e6444605a86a17545bb"}]},"branch":"refs/heads/master"},"e38a117cc3b3dc912f9326364e1531a9f7e22d60":{"kind":"REWORK","_number":5,"created":"2024-02-13 18:28:54.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/907111/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/907111/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/907111/5"}}},"commit":{"parents":[{"commit":"358aec067c04d705b305a08b4d528287014b3620","subject":"New grenade job to upgrding cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/358aec067c04d705b305a08b4d528287014b3620"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-29 18:26:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-13 17:58:43.000000000","tz":0},"subject":"Make `centralized_db` cache driver default","message":"Make `centralized_db` cache driver default\n\nMade `centralized_db` cache driver as default driver so that we\ncan test it using tempest jobs in gate.\n\nImplements blueprint centralized-cache-db\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/907110\n\nChange-Id: Id94e93e3ba3fc207b39c7dbff92495805aa0f6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e38a117cc3b3dc912f9326364e1531a9f7e22d60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e38a117cc3b3dc912f9326364e1531a9f7e22d60"}]},"branch":"refs/heads/master"},"afd1e6439f5fadda983f42ddeedab50c7300a522":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-02-14 05:35:50.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/907111/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/907111/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/907111/6"}}},"commit":{"parents":[{"commit":"f3d85d6622767dabe7f031d37e125e2a8fe12668","subject":"New grenade job to upgrding cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f3d85d6622767dabe7f031d37e125e2a8fe12668"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-29 18:26:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-14 05:00:26.000000000","tz":0},"subject":"Make `centralized_db` cache driver default","message":"Make `centralized_db` cache driver default\n\nMade `centralized_db` cache driver as default driver so that we\ncan test it using tempest jobs in gate.\n\nImplements blueprint centralized-cache-db\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/907110\n\nChange-Id: Id94e93e3ba3fc207b39c7dbff92495805aa0f6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/afd1e6439f5fadda983f42ddeedab50c7300a522"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/afd1e6439f5fadda983f42ddeedab50c7300a522"}]},"branch":"refs/heads/master"},"4cf46e97002dbe39c9aee4ca4c90bff1f5eadf0b":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-02-15 17:46:06.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/907111/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/907111/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/907111/7"}}},"commit":{"parents":[{"commit":"69a8617ba5dc4fbd6cb9d9f6b628a0ba68604e7a","subject":"New grenade job to upgrding cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/69a8617ba5dc4fbd6cb9d9f6b628a0ba68604e7a"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-29 18:26:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-15 17:03:00.000000000","tz":0},"subject":"Make `centralized_db` cache driver default","message":"Make `centralized_db` cache driver default\n\nMade `centralized_db` cache driver as default driver so that we\ncan test it using tempest jobs in gate.\n\nImplements blueprint centralized-cache-db\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/907110\n\nChange-Id: Id94e93e3ba3fc207b39c7dbff92495805aa0f6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4cf46e97002dbe39c9aee4ca4c90bff1f5eadf0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4cf46e97002dbe39c9aee4ca4c90bff1f5eadf0b"}]},"branch":"refs/heads/master"},"1e9569266f0b0a444390c12b7dc2bc2cfd686688":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-02-16 17:12:14.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/907111/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/907111/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/907111/8"}}},"commit":{"parents":[{"commit":"db1db378b980876126da15f941e7b871755308a1","subject":"New grenade job to upgrding cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/db1db378b980876126da15f941e7b871755308a1"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-29 18:26:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-16 16:49:32.000000000","tz":0},"subject":"Make `centralized_db` cache driver default","message":"Make `centralized_db` cache driver default\n\nMade `centralized_db` cache driver as default driver so that we\ncan test it using tempest jobs in gate.\n\nImplements blueprint centralized-cache-db\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/907110\n\nChange-Id: Id94e93e3ba3fc207b39c7dbff92495805aa0f6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1e9569266f0b0a444390c12b7dc2bc2cfd686688"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1e9569266f0b0a444390c12b7dc2bc2cfd686688"}]},"branch":"refs/heads/master"},"4f973850591dfec379cda9b9cdf190e2e1e2705d":{"kind":"REWORK","_number":9,"created":"2024-02-20 20:30:08.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/907111/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/907111/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/907111/9"}}},"commit":{"parents":[{"commit":"555e8ac32ce6d8fa576c4640126fb10a3f27f18c","subject":"New grenade job to upgrding cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/555e8ac32ce6d8fa576c4640126fb10a3f27f18c"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-29 18:26:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-20 20:09:57.000000000","tz":0},"subject":"Make `centralized_db` cache driver default","message":"Make `centralized_db` cache driver default\n\nMade `centralized_db` cache driver as default driver so that we\ncan test it using tempest jobs in gate.\n\nImplements blueprint centralized-cache-db\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/907110\n\nChange-Id: Id94e93e3ba3fc207b39c7dbff92495805aa0f6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4f973850591dfec379cda9b9cdf190e2e1e2705d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4f973850591dfec379cda9b9cdf190e2e1e2705d"}]},"branch":"refs/heads/master"},"14aa05c1ca0ac512e3146e7ebcbf90adf2bca419":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-02-27 18:45:05.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/907111/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/907111/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/907111/10"}}},"commit":{"parents":[{"commit":"9821985f72e03a3b47405a248eef8af567f80afa","subject":"New grenade job to upgrding cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9821985f72e03a3b47405a248eef8af567f80afa"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-29 18:26:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-27 18:19:13.000000000","tz":0},"subject":"Make `centralized_db` cache driver default","message":"Make `centralized_db` cache driver default\n\nMade `centralized_db` cache driver as default driver so that we\ncan test it using tempest jobs in gate.\n\nImplements blueprint centralized-cache-db\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/907110\n\nChange-Id: Id94e93e3ba3fc207b39c7dbff92495805aa0f6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14aa05c1ca0ac512e3146e7ebcbf90adf2bca419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14aa05c1ca0ac512e3146e7ebcbf90adf2bca419"}]},"branch":"refs/heads/master"},"2f6405df5674adb553e020da5dba57d5869425b3":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2024-02-28 15:06:06.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/907111/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/907111/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/907111/11"}}},"commit":{"parents":[{"commit":"ae9ec496e386c200cfdeb88c7aff673068aba111","subject":"New grenade job to upgrding cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae9ec496e386c200cfdeb88c7aff673068aba111"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-29 18:26:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-28 14:45:44.000000000","tz":0},"subject":"Make `centralized_db` cache driver default","message":"Make `centralized_db` cache driver default\n\nMade `centralized_db` cache driver as default driver so that we\ncan test it using tempest jobs in gate.\n\nImplements blueprint centralized-cache-db\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/907110\n\nChange-Id: Id94e93e3ba3fc207b39c7dbff92495805aa0f6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f6405df5674adb553e020da5dba57d5869425b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f6405df5674adb553e020da5dba57d5869425b3"}]},"branch":"refs/heads/master"},"dcd9bd185733205e4c8677de9b8d3afe67ac90a7":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2024-03-01 05:45:41.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/907111/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/907111/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/907111/12"}}},"commit":{"parents":[{"commit":"64815cf76a24a77baa62ae55473919b5ff02d5ce","subject":"New grenade job to upgrding cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/64815cf76a24a77baa62ae55473919b5ff02d5ce"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-29 18:26:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-03-01 05:25:37.000000000","tz":0},"subject":"Make `centralized_db` cache driver default","message":"Make `centralized_db` cache driver default\n\nMade `centralized_db` cache driver as default driver so that we\ncan test it using tempest jobs in gate.\n\nImplements blueprint centralized-cache-db\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/907110\n\nChange-Id: Id94e93e3ba3fc207b39c7dbff92495805aa0f6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dcd9bd185733205e4c8677de9b8d3afe67ac90a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dcd9bd185733205e4c8677de9b8d3afe67ac90a7"}]},"branch":"refs/heads/master"},"cb9d34e01d0f6fe160b4bce84903ad56f8241844":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2024-03-05 18:39:34.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/907111/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/907111/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/907111/13"}}},"commit":{"parents":[{"commit":"bd955b1189da170c12fe06fbd42556d91313747c","subject":"Merge \"Use centralized_db cache driver in tempest jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd955b1189da170c12fe06fbd42556d91313747c"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-29 18:26:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-03-05 18:19:15.000000000","tz":0},"subject":"Make `centralized_db` cache driver default","message":"Make `centralized_db` cache driver default\n\nMade `centralized_db` cache driver as default driver so that we\ncan test it using tempest jobs in gate.\n\nImplements blueprint centralized-cache-db\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/907110\n\nChange-Id: Id94e93e3ba3fc207b39c7dbff92495805aa0f6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb9d34e01d0f6fe160b4bce84903ad56f8241844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb9d34e01d0f6fe160b4bce84903ad56f8241844"}]},"branch":"refs/heads/master"},"9b9111f81996a22527f9622d0707891e445cbd5f":{"kind":"REWORK","_number":14,"created":"2024-03-05 19:55:40.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/907111/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/907111/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/907111/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/907111/14"}}},"commit":{"parents":[{"commit":"bd955b1189da170c12fe06fbd42556d91313747c","subject":"Merge \"Use centralized_db cache driver in tempest jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd955b1189da170c12fe06fbd42556d91313747c"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-29 18:26:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-03-05 19:35:20.000000000","tz":0},"subject":"Make `centralized_db` cache driver default","message":"Make `centralized_db` cache driver default\n\nMade `centralized_db` cache driver as default driver so that we\ncan test it using tempest jobs in gate.\n\nImplements blueprint centralized-cache-db\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/907110\n\nChange-Id: Id94e93e3ba3fc207b39c7dbff92495805aa0f6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b9111f81996a22527f9622d0707891e445cbd5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b9111f81996a22527f9622d0707891e445cbd5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
