)]}'
{"id":"openstack%2Fglance~907526","triplet_id":"openstack%2Fglance~master~I7d63951ff080dc699b8d11babc0a5998d90621e4","project":"openstack/glance","branch":"master","topic":"stream-optimized-vmdk","attention_set":{},"removed_from_attention_set":{"24434":{"account":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"last_update":"2024-02-16 21:47:31.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-02-15 14:46:45.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-02-06 15:41:24.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"9642":{"account":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"last_update":"2024-02-16 21:47:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7d63951ff080dc699b8d11babc0a5998d90621e4","subject":"Support Stream Optimized VMDKs","status":"MERGED","created":"2024-02-02 11:02:54.000000000","updated":"2024-02-16 21:48:34.000000000","submitted":"2024-02-16 21:47:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907526-stream-optimized-vmdk","meta_rev_id":"c797eb0f9cfbcd4a0b90809f98733a16b5056243","_number":907526,"virtual_id_number":907526,"owner":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-16 21:47:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-02-15 14:46:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-02-06 15:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-02-15 14:46:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"}],"reviewers":{"CC":[{"_account_id":32624,"name":"Rajiv Mucheli","email":"rajiv.mucheli@gmail.com","username":"rajivmucheli"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-02 11:17:45.000000000","updated_by":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"reviewer":{"_account_id":32624,"name":"Rajiv Mucheli","email":"rajiv.mucheli@gmail.com","username":"rajivmucheli"},"state":"CC"},{"updated":"2024-02-02 11:17:45.000000000","updated_by":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2024-02-02 11:17:45.000000000","updated_by":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"reviewer":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"state":"REVIEWER"},{"updated":"2024-02-02 11:24:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-02 13:12:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-02 16:06:14.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"3583dd9f209f7065397fba38986a0ea630941b6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-02-02 11:02:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57b6d8b65c5725297f47c05e86c40f44eeeb6d7c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-02-02 11:03:16.000000000","message":"Topic master-m3 removed","accounts_in_message":[],"_revision_number":1},{"id":"082a972169019e0b2f98716a7c55b49592771097","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-02-02 11:03:20.000000000","message":"Topic set to master","accounts_in_message":[],"_revision_number":1},{"id":"4333ba2c9f12d6d4af1c86f3a52c8621a8eafae1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-02-02 11:03:35.000000000","message":"Topic master removed","accounts_in_message":[],"_revision_number":1},{"id":"7f7e5a35b553d14acf7ea10e8e22a10d9b718dc8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-02-02 11:03:54.000000000","message":"Topic set to stream-optimized-vmdk","accounts_in_message":[],"_revision_number":1},{"id":"8b982a94fceb94022bf89075efba26781d0fff93","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-02-02 11:10:28.000000000","message":"Abandoned\n\nMessed somehow up with changing the topics...","accounts_in_message":[],"_revision_number":1},{"id":"6029ae182cb58199992d87b821355234b38c7e5a","tag":"autogenerated:gerrit:restore","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-02-02 11:10:52.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"03c6a839075face743df75bb96b49b9f63711d70","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-02-02 11:11:12.000000000","message":"Topic stream-optimized-vmdk removed","accounts_in_message":[],"_revision_number":1},{"id":"be20467ec55dd3f4b4460cdab89afe4571030e81","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-02-02 11:11:29.000000000","message":"Topic set to master-m3","accounts_in_message":[],"_revision_number":1},{"id":"70b0310a942a31e0ba005006b5c25384f8beca5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-02-02 11:13:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c25b1f35a0a44492322122e30729d57a6d39bdf4","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-02-02 11:17:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"082d8f6a2facf45e663ac1e4a81cd47fe66bdf8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 11:24:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9aa0933f55944f19341d351b9461bf8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/96bc3186656444d3a2ace3b8cc8282ff : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dbe55842c7924420bfaf5b04f28291dc : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4b26ed3e2d694b4eaf040dc9d31a5758 : SUCCESS in 5m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6afb2e22d5e1701e85a3c8e60c7267928f98bb84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 13:12:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/564cd4de979449b3b29c93a404e95335\n\n- grenade https://zuul.opendev.org/t/openstack/build/4eca34f02e35420eac85eb35a3379a84 : SUCCESS in 58m 16s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2ec88fc0c0c343ac999cce64d4acfac5 : SUCCESS in 40m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f62d8a20fa6c4af3a03d6ccbb324fe78 : SUCCESS in 1h 03m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c750f8923b2e44e4a6dcefcc12a5bdb3 : SUCCESS in 36m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3181e17c743e4dee851dd805af6aee4e : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a28be38bd9f46a18b14957dfc11a1d7 : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/395566bb7b27461684dc18ade409a968 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d70c5808c284835b5a5dfd3cd65bcc5 : SUCCESS in 6m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a0907d5c211d46baaa155a84b97e6dc2 : SUCCESS in 12m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0748e9b4b0a54e17a0e049bda6cc450c : SUCCESS in 9m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9bd5b473b5284a4d8ebd36d81e2446e8 : SUCCESS in 10m 59s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a19754925afd4b51a7c498a04e6243e6 : SUCCESS in 8m 58s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/3f4fbb05f8df4f26bf61dccff071b1c0 : SUCCESS in 11m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/33f09332ba084ebfbe181bf8b95048a8 : SUCCESS in 1h 52m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2992295fa6604751b175e5acf059e18e : SUCCESS in 1h 27m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e9d507b7aa894850964e087353d0e25a : SUCCESS in 1h 40m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f2e7f2bbeb2c49479f54e966c1557afb : SUCCESS in 1h 31m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c5622c491c474604a6615be3b39a84e8 : SUCCESS in 1h 41m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c31af18a1e1f4b6aadbd40f303a147d4 : SUCCESS in 1h 05m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/be5f26abd1dc44758f427d367f5db85b : SUCCESS in 1h 33m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ce70947cc97440c0a27d30c9aa0091e3 : SUCCESS in 27m 41s","accounts_in_message":[],"_revision_number":2},{"id":"ba129eb9f3c51dcc18d6c89873dc538a17f103e1","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-02 15:12:29.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"287714cd4ba70ce8d9c94c844f93a49bdaf11828","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-02 16:06:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a9a9c0ec1de71b1d8df960127fd94747c29dff53","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-02 17:42:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"af5860e2cbeae753055370ffa74fd113cb9b0ad2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-02-05 08:55:33.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"468b93ea1d20178444845ed90782f0b5fe456c46","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-02-05 09:06:48.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5eb4a3f08088d1aab60faaff7ab6239d183a784f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 09:08:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4a83571ec6346edb5dfdc60431cce97\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/91fe121b79464736b14dee1ba368821a : SUCCESS in 10m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d0d60ee3d9564340adce53f6fc0f6333 : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1695fd8fd27c4609914321eb7b500874 : SUCCESS in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"71c36f93b061f65edc509ef0cd73c2a300d47647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 11:00:08.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e271418fed44ecd963ee16525b45aa8\n\n- grenade https://zuul.opendev.org/t/openstack/build/657c3f849fce485cb2aca28ecc4bf1ab : SUCCESS in 55m 42s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/38461b47e9b94c0095a26b427e189505 : SUCCESS in 47m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e89b00842fbc4b9f8d23ffd078d9b1e2 : SUCCESS in 1h 20m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/969439e01dd0493b802b0e2dac5f3d34 : SUCCESS in 20m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cded2912fcf44f15b3dcd1ff3ef18a03 : FAILURE in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/442deb9ad12948cea73fc3e5d931074a : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de0cb9c05323474d87ae08eff08becfc : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51915e6ddf124a62b42db3a3586a091f : SUCCESS in 5m 07s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/bcacd83b6e9647009385b7d96de9e088 : SUCCESS in 14m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/674efa469de94cdfa20dc5e033d321ec : SUCCESS in 7m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/18f5cd900e624fc38ea9e6f1ba727176 : SUCCESS in 8m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/98e818fa9d1e4d719ab3ec47b332100c : SUCCESS in 8m 54s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/516dc36e88f84eadb7e7dbb0f93bcb52 : SUCCESS in 11m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/33dd90f37a9f48b38da16b17c197113c : SUCCESS in 1h 55m 56s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/00184f53f56149c282f9ee23191c22c3 : SUCCESS in 1h 42m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/79c4a2bd4ac345c19a1b4e1e7f27c4c7 : SUCCESS in 1h 43m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/729355dc63c144fd954b40996001b12b : SUCCESS in 1h 56m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/08e7bbdbad5c4406b37d4e6f3f1c2064 : SUCCESS in 1h 53m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36cf175446e54223ad5781d563da18c3 : SUCCESS in 1h 11m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/40ab25f0890544f6a0bca0d386a7e28c : SUCCESS in 1h 53m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/84b1b4cf32fc4c578f8db8fffbb518d7 : SUCCESS in 40m 50s","accounts_in_message":[],"_revision_number":3},{"id":"2efbf421b4a9905cf7c90b1389c65b7841dfe96a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-02-05 11:05:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2374f4304d6d4baa8aa00502a020e82adc21135b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-02-05 11:06:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"53a05f2c5907cdd6635818a13afb5a546f289a8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 11:13:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cc463d77cdc4d44865e0c984b311a36\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/99de222f84db4efaa30ce65ad4ceb8e7 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/12f7a98fc054484c954c0d041fd50c5e : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f78c0983cfab4a5ba4c88403fd6c4694 : SUCCESS in 5m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3cda7cabfac627417981c676e9e2bdf0316fb48d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 13:07:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cd0cac937bd495c99358a9637d1e967\n\n- grenade https://zuul.opendev.org/t/openstack/build/74f66246615f42dfb91bc82ad5d9f631 : SUCCESS in 1h 01m 13s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e57e3622410d44a5bdd45c42d1a5a524 : SUCCESS in 54m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c093d7104e8401db3270e2b39ad5dbe : SUCCESS in 1h 28m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfcc88db88d2479dbe0c763e758a1d08 : SUCCESS in 29m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/687be49dd94a4ffb90f1cefb17cf3027 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f06de6f9d5a54d88895ddbb7aa1afc60 : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ad0cfba401774c27b5fac14003bc0480 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4b81b4861c24c018bf78da9e78085ba : SUCCESS in 6m 38s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7f283246debb486c86894db822fa582a : SUCCESS in 15m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e8ed6375236f44f5a7e2cfd29040b017 : SUCCESS in 9m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5e23d25504f0423dbf828caa7e8dc53f : SUCCESS in 8m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0af7849b2e0e4b77b1c037f3f79de7cf : SUCCESS in 8m 43s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6afb8a0a3dfb40078dc87a63ebc2940e : SUCCESS in 11m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b3c1ee4ae0ba4e5a8f6a66d9c0cb675c : SUCCESS in 1h 52m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b58aff29d3424738a21c00b7395be431 : SUCCESS in 1h 41m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fb598a0f547b4bdf9c3cfb28796f5863 : SUCCESS in 1h 26m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1fca9257c5f04018801046386f091f55 : SUCCESS in 1h 41m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f65b7f0d5338439b832dd754e9a7db84 : SUCCESS in 1h 41m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73611b4ee1c44605bc36a74423a924c2 : SUCCESS in 50m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b9171892a35e41069ebd7d77e8e5548d : SUCCESS in 1h 33m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/66d2801a983e496bac14eafb7b43bdfd : SUCCESS in 32m 53s","accounts_in_message":[],"_revision_number":5},{"id":"b57f8dd12bcd01897634afaa4e90062b31e1ff29","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-02-06 09:57:36.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8b0c555a4e2f2bfbe557a6325d20ef951e189f19","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-06 15:41:24.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8e7ac9f999ba5584f99918204d29345104fb8a25","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-15 14:46:45.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b6e36606cf42ba92b59d2d730cd1dcccbc074cb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 14:47:23.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f049883517da406938b9d54215fd38b211ea6367","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 18:00:45.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f57f4b9c6f14e228db376f29f4ea98e\n\n- grenade https://zuul.opendev.org/t/openstack/build/c129c0a7e07f4d5aaa979161fef2e7fb : SUCCESS in 1h 05m 05s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/83994ffff66845cea64a82066410c92f : SUCCESS in 50m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a3fae99d3b3c4ed3881c994e8df389ce : FAILURE in 58m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d199d382596a407a9f735f978a89d34a : SUCCESS in 29m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c663259d9e54cc2a902195dcf11d206 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f9b1ed256bb415fb6fa598024d169bf : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5cf7a7b1eddf4d3c843865abca521b02 : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d603a58404b4a0a8b2c701876571f8f : SUCCESS in 6m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ffe9ef7c31804478a3286123a9119f74 : SUCCESS in 10m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cdf644dac377407996f4f4ebc18c2942 : SUCCESS in 10m 45s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c85db9a848b145c8ba096e28479364de : SUCCESS in 2h 01m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/787bb3f010b3434982707351e133296d : SUCCESS in 1h 42m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7dee38797e3f4f8fa5cb2b45ded60348 : SUCCESS in 1h 24m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be98536325c3485ea07af5d8e9a77e5a : SUCCESS in 50m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a0d4267014b44db48e29377391479410 : SUCCESS in 2h 00m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/00775ffec47a487eb480df8663c1b386 : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":5},{"id":"fbb783cd9601e5b9106fea97ad66ae72018deaa3","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-02-16 08:15:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3918d5bb80a71590b61101ef353ba1be6d6a6d46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 08:24:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22f1237c99404b1abad4d42a36ae5223\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/db5b67e0e517457e83954fac6572196e : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d77af16b3f9546d19cd0697b8af74c82 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b874e0fa606e4ad19344385e01c549db : SUCCESS in 7m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c068b1c7ee23f4a2b4ffdd0d5e1cabe4be33efcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 10:31:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7748835daaf4fb4bc7982eee3d78f77\n\n- grenade https://zuul.opendev.org/t/openstack/build/db5e434af1ec41e0a0839da608345639 : SUCCESS in 49m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8e05ced7286b4c0ead9d409e80af21de : SUCCESS in 47m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/74af1db30c5d4bd0ab4865572bfcd129 : SUCCESS in 1h 25m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b656c51514547fca980f1561f811767 : SUCCESS in 40m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f457e65a6e5a45c29b14508cb7244d31 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/656aeecd0d404fe48f738856cbd462d2 : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/786e6a33391e42b8aa567c12d10609f7 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/976092ac2868413c8ada9b9148e4271e : SUCCESS in 5m 37s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9bb5b4afb7a44ca4ae95758cbfbe7fe7 : SUCCESS in 14m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4d87dfe2e43f457daff767636fd534bc : SUCCESS in 7m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2f0182ce3a354ab7870181832424f9c4 : SUCCESS in 8m 06s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0ce2056d95094c4c9702fdf1f10fcde7 : SUCCESS in 8m 53s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/51783a8441fe48be99924a96fe976588 : SUCCESS in 10m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/299fbf56aa054ed292d9056308a15857 : TIMED_OUT in 2h 04m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e31cd7e7e3ef439f98e77ac9af9a6711 : SUCCESS in 1h 51m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3a19bf18c2f242ba86e369ddaf889ed3 : SUCCESS in 1h 38m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6d14f2896a894d689131656f2c03f2ac : SUCCESS in 1h 37m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5f7973c204554b5c994168918e363ae5 : SUCCESS in 1h 41m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4be90793498c4dd8b59d12105c13ca2c : SUCCESS in 45m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/edc2b09e35db4941a32bdbe748d4a169 : SUCCESS in 1h 57m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a41f0a3064624ea588261781b8354fdc : SUCCESS in 30m 38s","accounts_in_message":[],"_revision_number":5},{"id":"5b8eea8ea808f3ed9ea8c7b9ca994e9c56357097","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 10:32:09.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c90286519fb58a924c2afff105f69ea08fc73a00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 12:40:26.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a80add74d2094e508a620d9022afbd1e\n\n- grenade https://zuul.opendev.org/t/openstack/build/93597edaf43a49dba420bb6eefd43a11 : SUCCESS in 1h 06m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4d1bbc12b02f415897dcf5e1efda705e : SUCCESS in 56m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/302ea050e449436aadce45d08f0e3fe9 : FAILURE in 1h 19m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03f5f722ed144b42a9d207c252b61d82 : SUCCESS in 1h 00m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adec88c4871e4a93af9321d74922c9dc : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a203750ca184056b12c8907aa5712ba : SUCCESS in 8m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2be11a2310964eeaa380eda71f4079ba : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f327c80a848e4e9dbe2de05d4cee2b2f : SUCCESS in 6m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d4d98952849847c38f0addb5e7ed5e89 : SUCCESS in 9m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f72b727904f5446487b1d6ef4a9213f9 : SUCCESS in 9m 25s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9e6b730db3834bc0991fc4b97b4ed276 : SUCCESS in 1h 49m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/eeb1e8cefe4a4ec5be97ca5096b47ae3 : SUCCESS in 1h 41m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/410b1107a1804fd88c0f5fda8c787a48 : SUCCESS in 2h 00m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82e7848e95124eb99500d743d6da7c93 : SUCCESS in 1h 21m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/585eb9e089ff43ea8d073eea8a2d875e : SUCCESS in 2h 02m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2ac45c394ffa4520b969c87871a0f698 : SUCCESS in 36m 43s","accounts_in_message":[],"_revision_number":5},{"id":"e527e729a01fededca1cfe9b0ae71c51755e0a9a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-16 15:06:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9b64f80ad8f7f0ef8c55703b563eaf3b8bb45eb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 15:21:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8441db101d234a1597ffeda7d5fda59e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f1a0ed1d2ce42f1952a48eb4fb40868 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3993e7c3cf1747be87e50ff8c0fd4a81 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f3eeabfb54a74131b1c934bfedbe0247 : SUCCESS in 7m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"617173f0fa1c9d58655bfcdd316fbb341515eacb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 17:37:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb52676fad034527aea825996b5533b1\n\n- grenade https://zuul.opendev.org/t/openstack/build/bad38e5a4eff4467bab92df430fa6941 : SUCCESS in 59m 45s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/64e91e3e3ff445489648c8b96bb876ca : SUCCESS in 53m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4c836873f7964ba090bccc627ef0a0d6 : FAILURE in 1h 14m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70536956dbd04186813d21bbe0657e06 : POST_FAILURE in 1h 06m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cebfb77b2074d2b93c78ca737289714 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f1df9a3425045aab0e415538ca8c917 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bded4f2114dc4e7fa70797aaa7f1af7d : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d523b9bf77754be5935dfe2361c1196e : SUCCESS in 6m 24s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2c69e3b04b8b4ac48ea7634beee706fe : SUCCESS in 14m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0cfe40cf0f2c43028421501e5e8781e0 : SUCCESS in 8m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4ec45d16d70f4544865b5f168125cfc1 : SUCCESS in 9m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/eecbe354ac2f454c8605f2e7fc940624 : SUCCESS in 8m 23s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/87dd2d8113d346588741b28a7256fa45 : SUCCESS in 10m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f26c3ff1f557465ebb3f14c34cfa2837 : TIMED_OUT in 2h 03m 15s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7ed61e1f08614c10a2c27a6971285b3d : SUCCESS in 1h 36m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ea6137a21e8f47b9b00b133e4d221afe : SUCCESS in 1h 44m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2a2fe95992d24cf18ecac9df42773385 : SUCCESS in 1h 45m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c1dbdbad73e548cb8585de0d9cef5387 : SUCCESS in 2h 24m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22f24ae72e0a4a34b3df812e5c067d59 : SUCCESS in 1h 16m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/580913034d094e85b48b239a8e92a9a9 : SUCCESS in 1h 53m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0f72106137474150850a91ca52888740 : SUCCESS in 41m 21s","accounts_in_message":[],"_revision_number":5},{"id":"c6e331b3f7e000b22ddb096404ce3c03fdba343e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-16 17:56:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a7738e64495080be75f966c34d88c1f66055273e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 18:06:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36a48cb6ffbc4e899224ca0622045b15\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/44286119dcf6469fba84e94cb090aa6f : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c2c87d8560ed4ca2a57312a5b59885f7 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cb38e932538b4f309835227e2079bc5e : SUCCESS in 8m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"98e7eadf9f8f72969304ab08941107e338631c2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 19:53:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d928197096e46279e63dfbf16ff1756\n\n- grenade https://zuul.opendev.org/t/openstack/build/d3897758a71f4dce8eda9eb647ec7e21 : SUCCESS in 50m 14s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1c50305108074ad6b6f975e67518cd85 : SUCCESS in 39m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/418c3b9d0f1d4451b1d631205c18fa05 : SUCCESS in 1h 14m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bbc43cc9a9e48a1bb2a09353b260088 : SUCCESS in 1h 00m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07c2dba8ab554b2eb8eca5da646da5a4 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cecf03e333904f0181070470139432ca : SUCCESS in 5m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e4c473418e34eadbc86d51d648aa9f9 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73b511cc22844afb877951bbc2499e6e : SUCCESS in 5m 37s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/86e2e26485b94fe18f58b20d38aaf59a : SUCCESS in 14m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0439d90f9c9f4f96adf8ab05203b4607 : SUCCESS in 10m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/da8ec7edf1564437b94e68af4690952e : SUCCESS in 10m 02s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9bbaf8f4601e4296b3a0cfc71f528995 : SUCCESS in 7m 56s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/9a10f659f6bd49c1b4fa7271271626e5 : SUCCESS in 11m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c13f8ba3b13d43849e3a50a46f972030 : SUCCESS in 1h 49m 52s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d8fc3633938648439f74bf85b6c30198 : SUCCESS in 1h 41m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/753784046e5f4965aa945058ccd2173d : SUCCESS in 1h 29m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/70193478c5c74df28fb1899ed29c9a6f : SUCCESS in 1h 37m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ef40b8d10dea4565ab71bdacf66af88a : SUCCESS in 1h 51m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8dcf88b28274e45b2112fddf42176c7 : SUCCESS in 1h 01m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/75f0cc6318654c5a92937e31d7e132da : SUCCESS in 1h 41m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/80a9afd42768418da4f97efce0b84e23 : SUCCESS in 29m 50s","accounts_in_message":[],"_revision_number":5},{"id":"a513f050baec9645e4d49c82d60e863814c85d6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 19:54:04.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"59717a0410c839a37caab3cd71df0473821e9067","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 21:47:31.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb44aa2882a944c1aea38630ec7f77b5\n\n- grenade https://zuul.opendev.org/t/openstack/build/abb63972729845ec86fcf88cd351b1ab : SUCCESS in 59m 45s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6345a5de79684ddbbea5d4ef4bfcf02b : SUCCESS in 1h 00m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8537a2d8867a4a3aa62762c23087d69a : SUCCESS in 1h 31m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa3fa64df8bc40bdb12134281c9ab8d4 : SUCCESS in 1h 07m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ae3d3ce406c4b828db659a1afe7f61d : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f732526ecb824ce9b2013fde3deeb1d1 : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/230b756114104ef2a538de2162adfb4f : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c2c48e670914f5d955775233283aa09 : SUCCESS in 5m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ef805860714d42f7bc22b2ca902f6ef6 : SUCCESS in 7m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/263bfd8ae1bf4de3bbc584658b5ff6fe : SUCCESS in 9m 04s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/df075d7db3a04db58ad624f358d21ab6 : SUCCESS in 1h 46m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/361f9714e794426d9d9c8cc3b66cfb09 : SUCCESS in 1h 18m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0399105659e24be88c183838df598434 : SUCCESS in 1h 29m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14c89a54d0194bd0bbdf36a1c36d4ba1 : SUCCESS in 55m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6a4578a32acb4fde9bc667158ffb34d1 : SUCCESS in 1h 52m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8b7c9236785d457fad0e4abedb2ac0b7 : SUCCESS in 31m 14s","accounts_in_message":[],"_revision_number":5},{"id":"bc28a26fcc410a0e1518efd1d513ac545f1d6f05","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 21:47:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"c797eb0f9cfbcd4a0b90809f98733a16b5056243","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 21:48:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bb3d1a3971a4d3aa8496fdffd1e5b47\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0030db6ba1f43ef9dd901c7ae6127c0 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5e7e6bfb80ed6ede57f348e688c96dbdfe0e1c2b","revisions":{"7cd5b5ce936193826f78d38554a1b8d0a9e85a7f":{"kind":"REWORK","_number":1,"created":"2024-02-02 11:02:54.000000000","uploader":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"ref":"refs/changes/26/907526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/907526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/907526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/907526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/907526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/907526/1"}}},"commit":{"parents":[{"commit":"03a14a9b9c3fbbe106d1f2cc1e33fb594a5be963","subject":"Merge \"Unset VENV python_interpreter in fips job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03a14a9b9c3fbbe106d1f2cc1e33fb594a5be963"}]}],"author":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2024-02-02 10:48:24.000000000","tz":60},"committer":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2024-02-02 11:02:27.000000000","tz":60},"subject":"Support Stream Optimized VMDKs","message":"Support Stream Optimized VMDKs\n\nStream optimized VMDKs are also monolithic disks images, and start\nwith the same sparse extend header as normal monolithic sparse files,\nso we can parse the virtual disk size in the same manner.\n\nSee \"VMware Virtual Disks Virtual Disk Format 1.1\" p. 17.\n\u003e Header and Footer\n\u003e The header and the footer are both described by the same SparseExtentHeader\n\u003e structure shown in Hosted Sparse Extent Header on page 8.\n\nCloses-Bug: #2052291\nChange-Id: I7d63951ff080dc699b8d11babc0a5998d90621e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7cd5b5ce936193826f78d38554a1b8d0a9e85a7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7cd5b5ce936193826f78d38554a1b8d0a9e85a7f"}]},"branch":"refs/heads/master"},"3db32f608158df98af34634d644c2301bb14b6e7":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2024-02-02 11:13:25.000000000","uploader":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"ref":"refs/changes/26/907526/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/907526/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/907526/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/907526/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/907526/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/907526/2"}}},"commit":{"parents":[{"commit":"b6b9f043ffe664c643456912148648ecc0d6c9b4","subject":"Replace usage of deprecated [DATABASE] sql_connection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b6b9f043ffe664c643456912148648ecc0d6c9b4"}]}],"author":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2024-02-02 10:48:24.000000000","tz":60},"committer":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2024-02-02 11:13:17.000000000","tz":60},"subject":"Support Stream Optimized VMDKs","message":"Support Stream Optimized VMDKs\n\nStream optimized VMDKs are also monolithic disks images, and start\nwith the same sparse extend header as normal monolithic sparse files,\nso we can parse the virtual disk size in the same manner.\n\nSee \"VMware Virtual Disks Virtual Disk Format 1.1\" p. 17.\n\u003e Header and Footer\n\u003e The header and the footer are both described by the same SparseExtentHeader\n\u003e structure shown in Hosted Sparse Extent Header on page 8.\n\nCloses-Bug: #2052291\nChange-Id: I7d63951ff080dc699b8d11babc0a5998d90621e4\nCo-Authored-By: Rajiv Mucheli \u003crajiv.mucheli@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3db32f608158df98af34634d644c2301bb14b6e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3db32f608158df98af34634d644c2301bb14b6e7"}]},"branch":"refs/heads/master"},"080bfe874e035efef6fffc05acf01b0f28caca3f":{"kind":"REWORK","_number":3,"created":"2024-02-05 08:55:33.000000000","uploader":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"ref":"refs/changes/26/907526/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/907526/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/907526/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/907526/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/907526/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/907526/3"}}},"commit":{"parents":[{"commit":"b6b9f043ffe664c643456912148648ecc0d6c9b4","subject":"Replace usage of deprecated [DATABASE] sql_connection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b6b9f043ffe664c643456912148648ecc0d6c9b4"}]}],"author":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2024-02-02 10:48:24.000000000","tz":60},"committer":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2024-02-05 08:55:22.000000000","tz":60},"subject":"Support Stream Optimized VMDKs","message":"Support Stream Optimized VMDKs\n\nStream optimized VMDKs are also monolithic disks images, and start\nwith the same sparse extend header as normal monolithic sparse files,\nso we can parse the virtual disk size in the same manner.\n\nSee \"VMware Virtual Disks Virtual Disk Format 1.1\" p. 17.\n\u003e Header and Footer\n\u003e The header and the footer are both described by the same SparseExtentHeader\n\u003e structure shown in Hosted Sparse Extent Header on page 8.\n\nCloses-Bug: #2052291\nChange-Id: I7d63951ff080dc699b8d11babc0a5998d90621e4\nCo-Authored-By: Rajiv Mucheli \u003crajiv.mucheli@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/080bfe874e035efef6fffc05acf01b0f28caca3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/080bfe874e035efef6fffc05acf01b0f28caca3f"}]},"branch":"refs/heads/master"},"0e5da2f346bfb36a38e632099f98b8fc68588d2f":{"kind":"REWORK","_number":4,"created":"2024-02-05 11:05:21.000000000","uploader":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"ref":"refs/changes/26/907526/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/907526/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/907526/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/907526/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/907526/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/907526/4"}}},"commit":{"parents":[{"commit":"b6b9f043ffe664c643456912148648ecc0d6c9b4","subject":"Replace usage of deprecated [DATABASE] sql_connection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b6b9f043ffe664c643456912148648ecc0d6c9b4"}]}],"author":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2024-02-02 10:48:24.000000000","tz":60},"committer":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2024-02-05 11:05:12.000000000","tz":60},"subject":"Support Stream Optimized VMDKs","message":"Support Stream Optimized VMDKs\n\nStream optimized VMDKs are also monolithic disks images, and start\nwith the same sparse extend header as normal monolithic sparse files,\nso we can parse the virtual disk size in the same manner.\n\nSee \"VMware Virtual Disks Virtual Disk Format 1.1\" p. 17.\n\u003e Header and Footer\n\u003e The header and the footer are both described by the same SparseExtentHeader\n\u003e structure shown in Hosted Sparse Extent Header on page 8.\n\nCloses-Bug: #2052291\nChange-Id: I7d63951ff080dc699b8d11babc0a5998d90621e4\nCo-Authored-By: Rajiv Mucheli \u003crajiv.mucheli@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e5da2f346bfb36a38e632099f98b8fc68588d2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e5da2f346bfb36a38e632099f98b8fc68588d2f"}]},"branch":"refs/heads/master"},"5e7e6bfb80ed6ede57f348e688c96dbdfe0e1c2b":{"kind":"REWORK","_number":5,"created":"2024-02-05 11:06:38.000000000","uploader":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"ref":"refs/changes/26/907526/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/907526/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/907526/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/907526/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/907526/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/907526/5"}}},"commit":{"parents":[{"commit":"b6b9f043ffe664c643456912148648ecc0d6c9b4","subject":"Replace usage of deprecated [DATABASE] sql_connection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b6b9f043ffe664c643456912148648ecc0d6c9b4"}]}],"author":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2024-02-02 10:48:24.000000000","tz":60},"committer":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2024-02-05 11:06:30.000000000","tz":60},"subject":"Support Stream Optimized VMDKs","message":"Support Stream Optimized VMDKs\n\nStream optimized VMDKs are also monolithic disks images, and start\nwith the same sparse extend header as normal monolithic sparse files,\nso we can parse the virtual disk size in the same manner.\n\nSee \"VMware Virtual Disks Virtual Disk Format 1.1\" p. 17.\n\u003e Header and Footer\n\u003e The header and the footer are both described by the same SparseExtentHeader\n\u003e structure shown in Hosted Sparse Extent Header on page 8.\n\nCloses-Bug: #2052291\nChange-Id: I7d63951ff080dc699b8d11babc0a5998d90621e4\nCo-Authored-By: Rajiv Mucheli \u003crajiv.mucheli@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e7e6bfb80ed6ede57f348e688c96dbdfe0e1c2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e7e6bfb80ed6ede57f348e688c96dbdfe0e1c2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
