)]}'
{"id":"openstack%2Fglance~908247","triplet_id":"openstack%2Fglance~master~I4404cd0cb9e7a59ea332b0ad2bac354b53c232d4","project":"openstack/glance","branch":"master","topic":"centralized-cache-db","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-02-28 17:05:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4404cd0cb9e7a59ea332b0ad2bac354b53c232d4","subject":"[docs] Configure centralized_db cache driver","status":"MERGED","created":"2024-02-07 06:05:28.000000000","updated":"2024-02-28 17:07:36.000000000","submitted":"2024-02-28 17:05:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"908247-centralized-cache-db","meta_rev_id":"8f7b4355ec05493672b6c58503913612fa440943","_number":908247,"virtual_id_number":908247,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-28 17:05:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2024-02-28 15:49:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2024-02-28 15:06:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":1,"date":"2024-02-28 15:49:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-07 06:15:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-28 14:29:20.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-02-28 15:49:11.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"4e7226f3842cc9bf88d388e9ebc792942e38bc67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-07 06:05:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"081cbcb34e22ed119a67e2447fafc6d076eed3e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 06:15:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bec93baaa4224af28048e118d7aefead\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f1ca05eb3924ec09f206f4493667031 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b0c16491e4d4dd3bf4fe8cf05397b7e : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"328439e2ca5b92308d739b7327dc674a91bf6ae1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-09 10:05:42.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a36ee47c25a7bbe1061b4f68862a6a8762a1e3a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 10:24:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15286646f31b40699d4ffdf960db5fc4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f89e711b8f6d41808d1f6f1df9564006 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0b4036d841c4c2580823c6f471d3e6f : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":2},{"id":"9f8126978fafc6c65fa476fd4bcb23c4b8600ab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-09 19:54:58.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f2539e78feb6e2ffed618ba1499ca6cae784e025","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 20:22:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfab9217c04b4428b2dc737590650e31\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a43cf9dac144bc5b6bcb620819a4e91 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c799f52b441d4e2a92489ba351e182f7 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":3},{"id":"3dd37a81b9609b87a532e3ce3fd62397726287fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-13 18:28:54.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b54d7b91cbf71e91c90400e8e8e63d232381b3dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 19:11:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16162b50573c4115a4b1716d596c2a57\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e534c19306824219a6a12838090d599c : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e9e0c3aca754f48b232f1c41c8817b6 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":4},{"id":"44dc78951efe9032881bb15bd1bcbfa95301a2a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-14 05:35:50.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"9cfeae68dcbf76d548d631907303dc90aff6e67d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 05:51:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b057a4af5a0e4074a44f0ba83c0cb08c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e0b9eb2f3f24fb18233417e6be7e0e2 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed54719aba6643a79d4bdfbf7f82ee2e : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":5},{"id":"d8c4ceed6b631c640726e8569de2a94b5d827f6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-15 17:46:06.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"6c65738d146ae70ccc28af99c13e4c27d6118481","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 18:00:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/039156f41058480482a218ca208df73c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5980c4145f08420db765e12443669a29 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8077a8ece8eb456784e025c0a4626a6a : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":6},{"id":"8611e1ed4549ae2120c8e6fe44dd20b309cf514b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-16 17:12:14.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"c971502223d459c2f4fe784e8b900726260f2da7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 17:32:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7b35a5c5d0b4358b0652afb6c4a8d60\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc7a4da5b70e4c03a0cddc81feb500c3 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7809581681a946619cf6a0f5cd2b2470 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":7},{"id":"0c46fd98beea3815c0f346780d4767c275bb5da2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-20 20:30:08.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"ea31cff593015fbf8f07068b579d60c844fd7252","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 20:57:42.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cce5060ef8084c4f8b546d5a4c66d46d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ce2516e3c3444cc888040c62b8c10df : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1997df0281d345dc9c4d05ae281fea43 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":8},{"id":"beb031188f66aa714a78c996a6d64c13373657bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-27 18:45:05.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"ba609fc78b491a7e49c196ef97a399ae234ad1d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 19:04:38.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7847b3e9cf3b4739904023b3d6364051\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b2f3e57b7ee43bea1bb313829f11b7d : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aea11980831348ad8063d68e790926ce : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":9},{"id":"54743959f875033e235ee4273c45101a6562b5ea","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-28 14:29:20.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"aa44708d2fb5b1a4e658e571524ce4bb84b6cb8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-28 15:06:06.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"af43e9751b749329786c56828c1a3f2d86d148e0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-28 15:49:11.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(9 comments)","accounts_in_message":[],"_revision_number":10},{"id":"d18df2d2d040ff51c078d78d333788c254765ee5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 16:02:39.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28984232a3cf45748bb910aa89f5fb7b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cda3ff1bef4c48e4851656b56d5e6997 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11224422f08e4da396cb1becfd91b673 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":10},{"id":"d1dc8b583a6272af17503d91824a45623fe8b3be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 16:03:09.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"96da7c875b4423d875b062c84e8cd4e00c43f856","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 17:05:36.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/970dacb1828d4124821b62fd966303cd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f284fbd9dd32414fac25b4e6509c6133 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/051118ebc531410fbd1c9805e65bcd3b : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":10},{"id":"3d73318b1d1bf0d60b587ebc924a777aef5b1a1f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 17:05:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"8f7b4355ec05493672b6c58503913612fa440943","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 17:07:36.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/009f74f7c76e4999830505c02995d4b1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c112e0384a74cd594a7298d448febcd : SUCCESS in 47s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"cd9d521923e3eae1e01d75d6e3c513117bc37847","revisions":{"549d221a62a17c2ea6b8a284a3da886462be1ded":{"kind":"REWORK","_number":1,"created":"2024-02-07 06:05:28.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/47/908247/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/908247/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/908247/1"}}},"commit":{"parents":[{"commit":"8fc297210f1d37909923417f7f652cbdeec06016","subject":"Make `centralized_db` cache driver default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fc297210f1d37909923417f7f652cbdeec06016"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-07 05:44:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-07 05:44:04.000000000","tz":0},"subject":"[docs] Configure centralized_db cache driver","message":"[docs] Configure centralized_db cache driver\n\nDocumented the usage of centralized_db cache driver and\nmigration process from SQLite to centralized_db on upgrade.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I4404cd0cb9e7a59ea332b0ad2bac354b53c232d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/549d221a62a17c2ea6b8a284a3da886462be1ded"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/549d221a62a17c2ea6b8a284a3da886462be1ded"}]},"branch":"refs/heads/master"},"3b3e7bd0f50b5cfbcf6cbf92ff2d7ab600474eab":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-09 10:05:42.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/47/908247/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/908247/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/908247/2"}}},"commit":{"parents":[{"commit":"b812b8eabddd62abc475d6f859f7f9dc0b153b10","subject":"Make `centralized_db` cache driver default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b812b8eabddd62abc475d6f859f7f9dc0b153b10"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-07 05:44:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-09 09:39:48.000000000","tz":0},"subject":"[docs] Configure centralized_db cache driver","message":"[docs] Configure centralized_db cache driver\n\nDocumented the usage of centralized_db cache driver and\nmigration process from SQLite to centralized_db on upgrade.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I4404cd0cb9e7a59ea332b0ad2bac354b53c232d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b3e7bd0f50b5cfbcf6cbf92ff2d7ab600474eab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b3e7bd0f50b5cfbcf6cbf92ff2d7ab600474eab"}]},"branch":"refs/heads/master"},"4ca478ee773d09db4448075b40b1baa5bf2c8db4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-02-09 19:54:58.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/47/908247/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/908247/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/908247/3"}}},"commit":{"parents":[{"commit":"dac28647af962f82ac367e6444605a86a17545bb","subject":"Make `centralized_db` cache driver default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dac28647af962f82ac367e6444605a86a17545bb"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-07 05:44:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-09 19:34:26.000000000","tz":0},"subject":"[docs] Configure centralized_db cache driver","message":"[docs] Configure centralized_db cache driver\n\nDocumented the usage of centralized_db cache driver and\nmigration process from SQLite to centralized_db on upgrade.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I4404cd0cb9e7a59ea332b0ad2bac354b53c232d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4ca478ee773d09db4448075b40b1baa5bf2c8db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4ca478ee773d09db4448075b40b1baa5bf2c8db4"}]},"branch":"refs/heads/master"},"2303cc7eb4aac7ca7849e7a20ef8e2aa14e6eb8d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-02-13 18:28:54.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/47/908247/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/908247/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/908247/4"}}},"commit":{"parents":[{"commit":"e38a117cc3b3dc912f9326364e1531a9f7e22d60","subject":"Make `centralized_db` cache driver default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e38a117cc3b3dc912f9326364e1531a9f7e22d60"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-13 18:08:44.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-13 18:08:44.000000000","tz":0},"subject":"[docs] Configure centralized_db cache driver","message":"[docs] Configure centralized_db cache driver\n\nDocumented the usage of centralized_db cache driver and\nmigration process from SQLite to centralized_db on upgrade.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I4404cd0cb9e7a59ea332b0ad2bac354b53c232d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2303cc7eb4aac7ca7849e7a20ef8e2aa14e6eb8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2303cc7eb4aac7ca7849e7a20ef8e2aa14e6eb8d"}]},"branch":"refs/heads/master"},"baf9b79fbc6d77bb33c27e5c97ff0dd8294ae9d4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-02-14 05:35:50.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/47/908247/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/908247/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/908247/5"}}},"commit":{"parents":[{"commit":"afd1e6439f5fadda983f42ddeedab50c7300a522","subject":"Make `centralized_db` cache driver default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/afd1e6439f5fadda983f42ddeedab50c7300a522"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-13 18:08:44.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-14 05:04:26.000000000","tz":0},"subject":"[docs] Configure centralized_db cache driver","message":"[docs] Configure centralized_db cache driver\n\nDocumented the usage of centralized_db cache driver and\nmigration process from SQLite to centralized_db on upgrade.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I4404cd0cb9e7a59ea332b0ad2bac354b53c232d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/baf9b79fbc6d77bb33c27e5c97ff0dd8294ae9d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/baf9b79fbc6d77bb33c27e5c97ff0dd8294ae9d4"}]},"branch":"refs/heads/master"},"45821d23bf6bdacd2f2ba440b891a34de9ed472c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-02-15 17:46:06.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/47/908247/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/908247/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/908247/6"}}},"commit":{"parents":[{"commit":"4cf46e97002dbe39c9aee4ca4c90bff1f5eadf0b","subject":"Make `centralized_db` cache driver default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4cf46e97002dbe39c9aee4ca4c90bff1f5eadf0b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-13 18:08:44.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-15 17:26:37.000000000","tz":0},"subject":"[docs] Configure centralized_db cache driver","message":"[docs] Configure centralized_db cache driver\n\nDocumented the usage of centralized_db cache driver and\nmigration process from SQLite to centralized_db on upgrade.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I4404cd0cb9e7a59ea332b0ad2bac354b53c232d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/45821d23bf6bdacd2f2ba440b891a34de9ed472c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/45821d23bf6bdacd2f2ba440b891a34de9ed472c"}]},"branch":"refs/heads/master"},"f9d6466c388a09b9acbc1f2a5fec6dbbadd2bced":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-02-16 17:12:14.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/47/908247/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/908247/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/908247/7"}}},"commit":{"parents":[{"commit":"1e9569266f0b0a444390c12b7dc2bc2cfd686688","subject":"Make `centralized_db` cache driver default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1e9569266f0b0a444390c12b7dc2bc2cfd686688"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-13 18:08:44.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-16 16:52:43.000000000","tz":0},"subject":"[docs] Configure centralized_db cache driver","message":"[docs] Configure centralized_db cache driver\n\nDocumented the usage of centralized_db cache driver and\nmigration process from SQLite to centralized_db on upgrade.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I4404cd0cb9e7a59ea332b0ad2bac354b53c232d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f9d6466c388a09b9acbc1f2a5fec6dbbadd2bced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f9d6466c388a09b9acbc1f2a5fec6dbbadd2bced"}]},"branch":"refs/heads/master"},"332bea6ee909021947bcc3e447ef96174ec4dc0f":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-02-20 20:30:08.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/47/908247/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/908247/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/908247/8"}}},"commit":{"parents":[{"commit":"4f973850591dfec379cda9b9cdf190e2e1e2705d","subject":"Make `centralized_db` cache driver default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4f973850591dfec379cda9b9cdf190e2e1e2705d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-13 18:08:44.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-20 20:10:32.000000000","tz":0},"subject":"[docs] Configure centralized_db cache driver","message":"[docs] Configure centralized_db cache driver\n\nDocumented the usage of centralized_db cache driver and\nmigration process from SQLite to centralized_db on upgrade.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I4404cd0cb9e7a59ea332b0ad2bac354b53c232d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/332bea6ee909021947bcc3e447ef96174ec4dc0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/332bea6ee909021947bcc3e447ef96174ec4dc0f"}]},"branch":"refs/heads/master"},"389bba1d5a13988e108134918b0aaaa4909b4533":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-02-27 18:45:05.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/47/908247/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/908247/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/908247/9"}}},"commit":{"parents":[{"commit":"14aa05c1ca0ac512e3146e7ebcbf90adf2bca419","subject":"Make `centralized_db` cache driver default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14aa05c1ca0ac512e3146e7ebcbf90adf2bca419"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-13 18:08:44.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-27 18:25:02.000000000","tz":0},"subject":"[docs] Configure centralized_db cache driver","message":"[docs] Configure centralized_db cache driver\n\nDocumented the usage of centralized_db cache driver and\nmigration process from SQLite to centralized_db on upgrade.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I4404cd0cb9e7a59ea332b0ad2bac354b53c232d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/389bba1d5a13988e108134918b0aaaa4909b4533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/389bba1d5a13988e108134918b0aaaa4909b4533"}]},"branch":"refs/heads/master"},"cd9d521923e3eae1e01d75d6e3c513117bc37847":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-02-28 15:06:06.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/47/908247/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/908247/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/908247/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/908247/10"}}},"commit":{"parents":[{"commit":"781ce285604844643cfc30232406a402fa32c30e","subject":"Migrate from SQLite to Centralized db","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/781ce285604844643cfc30232406a402fa32c30e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-13 18:08:44.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-28 14:43:11.000000000","tz":0},"subject":"[docs] Configure centralized_db cache driver","message":"[docs] Configure centralized_db cache driver\n\nDocumented the usage of centralized_db cache driver and\nmigration process from SQLite to centralized_db on upgrade.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I4404cd0cb9e7a59ea332b0ad2bac354b53c232d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cd9d521923e3eae1e01d75d6e3c513117bc37847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cd9d521923e3eae1e01d75d6e3c513117bc37847"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
