)]}'
{"id":"openstack%2Fglance~908744","triplet_id":"openstack%2Fglance~master~I5d22af388746fb4863691f7f4c5ca91b2d05475b","project":"openstack/glance","branch":"master","topic":"db-migration-squash","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-08-08 01:03:19.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I5d22af388746fb4863691f7f4c5ca91b2d05475b","subject":"Update base database schema from pike to queens","status":"NEW","created":"2024-02-11 16:03:35.000000000","updated":"2026-07-18 18:53:18.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b61dd06b9003992196fecaca047f3844e9907664","_number":908744,"virtual_id_number":908744,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-18 18:53:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-11 16:15:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-11 16:23:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8107fcb6eee03e6131d5a684a56da5abf9f2ea35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-11 16:03:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d02eb50d107094834ccb56ec877d88e5dcdc53ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-11 16:15:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/063b55d0d9a74a44be0d4dcf9c883ca6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e65dd569cba24cc28cb35fc65508b21e : SUCCESS in 10m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0e360f44f5f840ffa7b746d0a7d9a9b4 : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d78d6efa0de94ae39eafbb6c7e7f5087 : SUCCESS in 5m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f9bb38e52b46950c7c1c194ecb7ecc47151c5c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-11 16:23:26.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/908741, https://review.opendev.org/908740 are needed.","accounts_in_message":[],"_revision_number":1},{"id":"1138c92b9dc87c34d730d6cddc48f3da53f6f703","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-11 16:31:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8dcf1e8d69901e24cf884226db9dff4d56a4033a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-11 16:42:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e325db94e4574e58b771493a237433f8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89c50830c14f46d185d5adc7922b6726 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/efe493df375f42c689f83de1b2fb0b7f : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a6b005565ff64a71a39deec717a56351 : SUCCESS in 8m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9d40c2009d84300ae6a1384eaf707f39924bfac9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-11 16:49:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7973f1a1058586fdca9ebec5d12cd6686c497806","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-11 16:50:14.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance for 908740,13","accounts_in_message":[],"_revision_number":3},{"id":"622503e46c7a846bd209c006fdc07e5e47395c96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-11 16:50:15.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance for 908740,13","accounts_in_message":[],"_revision_number":3},{"id":"2bf2319e650c1ee9a5e923fcc2c0b273ce181850","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-11 16:53:27.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4f6d96e80e7c427bb98f1797f7870c6f9bfd4d85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-11 16:59:00.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/908741 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"e637c99a416a2ab7116f28d0c1f770d1834dfac4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-11 16:59:02.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/908741 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"dd927f5252f0c549f300fae7db49d2b6a2696303","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-11 17:32:18.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"d0f222d66a6237d4c304ff4a740a3ccafd2b2fe7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-11 17:34:30.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"537d4add8353f123f7eb4853f27988260862acdb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-11 17:46:17.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2eed3bfdb0f437b92fc2ae7cac512d6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9db94b3010c143c79a66b85e021763a1 : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a6a89911231e4e63a6e95b5f6bbb7be8 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7d2b10c1e7f14275a2948232b3294ff3 : SUCCESS in 8m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"19e11dea9e39c70bde6f1d4dcc8c86796f010e0d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-11 17:55:55.000000000","message":"Topic set to db-miration-squash","accounts_in_message":[],"_revision_number":6},{"id":"82dc080d9b992682752988782f926c63d0db79d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-11 19:44:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e6f1e62ce904054b850794e8c0ab1cd\n\n- grenade https://zuul.opendev.org/t/openstack/build/f6ca597854d54aacb9adefcaa342f525 : SUCCESS in 44m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2e4901af2540420c9de8dcc89582961b : SUCCESS in 40m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7df9dbe9328b497cb0f5335773b4efc4 : SUCCESS in 1h 04m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f63b1c31b87f4e858cd17306b636bd65 : SUCCESS in 26m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f48d84a92de405caadf8b7113909dc5 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8507fbd3be244f549844d86cce2dc42d : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85a052e0da6c4a4a882511eaab3cde86 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01dd437cf14042a5ac2143c3cfc32757 : SUCCESS in 6m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/55d2454a1f664cb08d44c9a1ac1cd081 : SUCCESS in 12m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a4491a779ee14152b75131a26cd1c9a3 : SUCCESS in 8m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ff5119b16770412dafec7d1544c20bd8 : SUCCESS in 8m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c55f4aadeda04e4eadf92f5af415af46 : SUCCESS in 8m 55s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a98a43004e3a467d843046de0e1171a9 : SUCCESS in 10m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/00a66e792ada4755ad6267fae798bb45 : TIMED_OUT in 2h 03m 19s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2f72abf3cda24287a9b2e20c59941ee4 : TIMED_OUT in 2h 03m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/327f6ccac85a4b32add6b20b62cdf8b6 : SUCCESS in 1h 50m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/055d9c53dd634fac8394b8ae9d45e1a0 : SUCCESS in 1h 47m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e2757aafbde14f688a9ca721e2b9cf86 : SUCCESS in 1h 36m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d2b138125604eb39287a532bd4e357d : SUCCESS in 1h 03m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a9d7e418302049ee8660a1ee44c4d921 : SUCCESS in 1h 44m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/138aa5c2929a4cdeb088b111f5e834a1 : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":6},{"id":"7e033fb9f729fbf0f83256d055f15509b607a0d4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-12 05:49:03.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c688ebbaf2587ad6e7d0a60b10db4a95329b8c3f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-12 05:56:36.000000000","message":"Topic db-miration-squash removed","accounts_in_message":[],"_revision_number":6},{"id":"351d03e829782a61044170bceee8b962651d154a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-12 05:56:44.000000000","message":"Topic set to db-migration-squash","accounts_in_message":[],"_revision_number":6},{"id":"5ef0b5b90693e4ff0ce6b55847378fb65be5bc01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 06:00:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95c24d4cba714fd2af1aaa8260203ef0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ea22e752d754eba977a40ecb838ef9f : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/51abc77bf3bc4a1fb0f109eae114953b : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fe497b21b0234f1cb045b1f4d8a4d71b : SUCCESS in 9m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0d674896a252bedd535a243c81615628197d3faf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 08:07:17.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e847e92b7219499ea8d18e1499907c3c\n\n- grenade https://zuul.opendev.org/t/openstack/build/d2cabe3b9c5349e7bfe35988e89d4e9f : SUCCESS in 49m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7ce7f3aced924d43906a12cb3c5e413a : SUCCESS in 46m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c4a25cb85d7b4b808d2fc50967f7c4c1 : SUCCESS in 1h 08m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6338eb577c7e4a599fd73677808d4126 : SUCCESS in 22m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc0d0ed68c094b37a78819ad3b5279b1 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d2f3549d3e847b29f5a6c8f0abaa429 : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e8bef9e3903540a493ac96f5e7eeefdf : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa8c06f4b2c14931af9148b3071e5d08 : SUCCESS in 8m 04s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/469fa47726df4b819c78be823d360ea3 : SUCCESS in 10m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/60fc1edab3a544a797313d18035660e5 : SUCCESS in 7m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d5b71837100242cc8b842b5061eb4b5f : SUCCESS in 7m 25s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a152eafff1c14d08a24fafb1747d0447 : SUCCESS in 8m 02s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0a76a34db4c24d50bf094a5aea52b6e3 : SUCCESS in 11m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7719ff25ff3b4dc3a5f205e189178972 : SUCCESS in 1h 29m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1e98ff50cbc64ee0ad0f7d7b568c4d01 : SUCCESS in 1h 57m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/70119411a7e9481bbe2af844c2f59cf8 : SUCCESS in 2h 03m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/186bc0332fde4450adb06f537ac69186 : SUCCESS in 1h 52m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/75340588c5a8498ba36477b8629fb509 : FAILURE in 2h 07m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1dc9ae0631240b29d4032f4aed6519d : SUCCESS in 41m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/670ac878581642e6acdd36cd1371da2e : SUCCESS in 1h 55m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f7d1a49c84fa48b5aeaa478574f30786 : SUCCESS in 37m 21s","accounts_in_message":[],"_revision_number":6},{"id":"4786c93dd252fae73f35ae65e7341555304e51df","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-12 08:58:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"19e9c31e91be2887bb05fb7f6ba5e900cb45eaf0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 09:10:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a1a30fe5b7d4002a3112bc723072fe8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e587d925ea154868889f40f7b01b25fd : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b8408953ec7c44a1b917c9c737036d27 : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/253660c0d3b647f78cd11748e3cad9bf : SUCCESS in 5m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"276874ea55c93cb42cffcea824f8e7f9488b918d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 11:19:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fc5f3eea35044479754a6318d7f09e6\n\n- grenade https://zuul.opendev.org/t/openstack/build/897f829d8f7f4de982dbecf43917a505 : SUCCESS in 56m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/88db14feb8ec4df0b1f7d0f5504f659c : SUCCESS in 47m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b5b8d56cbef0469ba718944cc877e91c : SUCCESS in 1h 16m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f72b3fb404ba4f17ab77227c9105f235 : SUCCESS in 28m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/571c3a03ebe2479fb1c5673713179985 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ff457456e1e417a96679a5b43c91476 : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c578391b31a741a9b2ff1fea5c902be0 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50c104026fc34cbfaa6d11b7a9fa79ce : SUCCESS in 5m 47s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/542b81a2e40240b2a050253ebaed43de : SUCCESS in 14m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f91856cf045c494e9e8ec30fb0e56354 : SUCCESS in 10m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/958524fca3244852b6ffb7966fdfaf75 : SUCCESS in 7m 51s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/72996312bef24a85be672de328e79fcb : SUCCESS in 8m 20s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ed9de88104a548ee8f713a5074512650 : SUCCESS in 13m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4142cee5336e40b893543fe2d6b78e2c : TIMED_OUT in 2h 04m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b8d98a832a5848e7ba24ab96e87fa910 : SUCCESS in 1h 38m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2d2aed31786d4767891b95f75c4314c9 : SUCCESS in 1h 45m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/97bf9105ff7442f5a25fa163c3b9df3f : SUCCESS in 1h 53m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a41d40c6493547af92d9a22dd7c4b341 : SUCCESS in 2h 15m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31e3a221c3484808a96a5faa1f482621 : SUCCESS in 1h 22m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d3aaf268e07d4e2eae4fdf4a3774840e : SUCCESS in 2h 04m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f45e78ad6552452dbddb4fecfa3818b6 : SUCCESS in 53m 13s","accounts_in_message":[],"_revision_number":6},{"id":"e941bf59202879347ab2131432989f13f5609e4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-07 14:12:34.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"41e0100f40c71e66171cf7faa55650ca0222d380","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 14:31:59.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1af3c29b3ae04763a01dc31e143ccda2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e7f6fc84942e41f1afc7115c184625f8 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e358d0764925427eb677a2e5a7002788 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/097a953cbed643d9927ee99a176233cd : SUCCESS in 5m 25s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a0dbf7b87bc317015ee461f1929c6cf4b0c825c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 16:42:11.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/283197e0702547bda45a5875dfb91d9a\n\n- grenade https://zuul.opendev.org/t/openstack/build/6ffefa22aadf40d3ba95a5327b7184cf : SUCCESS in 1h 21m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/284371f74a3e4c6dbf72b761388e2a4b : SUCCESS in 1h 01m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef16fc84f09641a9a45bdff69722620c : FAILURE in 1h 01m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/615dff15b9f642b3a7967e198319924a : SUCCESS in 1h 02m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f249f41fa29b475badb0a291503e06db : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf6f07ac99d8447f870904df53c06f0d : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8da93bd8b05d48db9cb70099af1bb008 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee204f4cddde438594ce73c7c6b34dc7 : SUCCESS in 7m 43s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9b8627d3929c4b158236ffb26a41b762 : RETRY_LIMIT in 2m 17s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6223007b1c7c49b986ddfc57062cb8b2 : SUCCESS in 8m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1208691cb92a4afd8f9113cd419141ce : SUCCESS in 8m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ea3f19c69ba94a1c8855359bc1e0e92a : SUCCESS in 10m 25s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c4d572537f43466fb711b01a2f3d1d8b : SUCCESS in 12m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3691f47e3c26418fbb8bce7388780181 : SUCCESS in 2h 00m 11s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/94cfac17211948ea8a462a7595735439 : SUCCESS in 1h 41m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b86fc781eb614c04b97e665a7915d296 : SUCCESS in 1h 48m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c93c6878406b49f98caaa70fba63e1f3 : SUCCESS in 1h 38m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7de8853b62de4c74bd38688b71159324 : FAILURE in 2h 20m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c48fc72daba54755b4964859db75ced0 : SUCCESS in 1h 00m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cf4e931f049b463b8454da9eb6fe8531 : SUCCESS in 1h 54m 02s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/316cdb2108754fbe99c91a9235213221 : SUCCESS in 1h 49m 07s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/442285a3d83e4f1b9f10d2deab768cdb : SUCCESS in 52m 26s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9cbb2db6ba3f47c195c8d7cc74203fd9 : FAILURE in 39m 19s","accounts_in_message":[],"_revision_number":7},{"id":"9897cffc451080e55a8909d8d89a190deff0452a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-08 04:03:18.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"93694a28d2415cc87184ac8f184d793a0fe334f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 04:15:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4c2f59e96174daa8c7032e1e634b9a9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6868efa22cb64f38918e79cc2b87b5ac : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0c7545222b71447ab632eff9f22e9f1e : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fae7248ce95241d8a01ea9fd2f547b35 : SUCCESS in 5m 07s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0ccdc62e8329fb8497abd75f48055aeea1c19f6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 06:23:36.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e8c26e5740b47b1a582073df2ed5d1e\n\n- grenade https://zuul.opendev.org/t/openstack/build/80febf0032f44430891d1c0989d6747c : SUCCESS in 56m 41s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/efcbe3a9d42e4fbd9beac3ecfc1bc6b3 : SUCCESS in 56m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3def68671b494f52a985aca14760960c : SUCCESS in 1h 32m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/febea4708c0f4f9b8731baf7859abf13 : SUCCESS in 56m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/385874c3f98b4a6c8781d1e23c4faeca : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e233e93543ca484792dd105fee8a50ea : SUCCESS in 7m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/658125c5cc094fc692a529fc6c3fea24 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a3487ce28fc45b992bf9ed986d445e8 : SUCCESS in 6m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/66e47681752f4667aef4126bdc26db63 : FAILURE in 14m 39s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ae1c905546e34a89879a42fb650299e0 : SUCCESS in 8m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7a7acaf40d35451196e7c32053d52ca8 : SUCCESS in 10m 16s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5dfe694bedb04cdd8ede7bda7570eeed : SUCCESS in 8m 04s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/27e6acf809664f84be59597b699521bf : SUCCESS in 11m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/342e7376ba344dcf8b5bc3f44e03aba5 : TIMED_OUT in 2h 03m 44s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f79d1cbd36ec4f8c8b8ca1c06c11558f : SUCCESS in 1h 48m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a719ecd1dd424d519453936285b2e281 : POST_FAILURE in 1h 31m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/daca923cf8cc4492a6542e394546e931 : SUCCESS in 1h 12m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e641082fb7a84c3ba16e2519ece84b2c : FAILURE in 2h 05m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20154e4f2ff84f9cb534659a9186e678 : SUCCESS in 55m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e1eed822ad2c461b815bc56e0d843c36 : SUCCESS in 1h 59m 01s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/c2375e55efb34cd39a2c1aeeb02e222c : SUCCESS in 1h 43m 09s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/546a33c7aa604a56b21b51ddaac0990f : SUCCESS in 57m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/969a7da2c87841a8ac88d33d0eb2c3da : SUCCESS in 35m 17s","accounts_in_message":[],"_revision_number":7},{"id":"063abcdef154809f6229a402bc06b97733091e0e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-18 00:17:00.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"99443d1a8968e09444f0757f05909ee57a6ad3a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 00:25:42.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5e9b24cabae459d91f9b2b3f75de5f7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff40628675f34b838cc47c65234629e9 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/039f26a9006c4b5d86dc6f9f27348214 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9d57128e88964b8bae7d3696ce70697b : SUCCESS in 6m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"537a6fd9863562a08060e7b70c5071a382f52235","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 02:31:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6dd8e3ff32944e1bc2b5796517b70c2\n\n- grenade https://zuul.opendev.org/t/openstack/build/50ba724d38d04e6190aa63fdd03cf30b : SUCCESS in 55m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8d1c802e0d7d458c9a679a0d1fd49dcb : SUCCESS in 40m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b445193ad80419c963fe587e38fab49 : SUCCESS in 1h 06m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2d2110265d3450189f132edc1d6b9d4 : SUCCESS in 54m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56c027f840cd4861b9b462b98254ced7 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c8a226303504bfe9cebdb1520bffee3 : SUCCESS in 8m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/838bbb530a7a48cb8d2559f94dd123e3 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8b69eddaf6740bb93c4a8e9f08882cc : SUCCESS in 6m 19s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b8c0123b72734a83b0bccc397ee27d8f : FAILURE in 10m 44s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/585b892521924f0fa9b87cba5f2e657e : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cb915818d7f14228a48c1e570eab4157 : SUCCESS in 8m 02s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/846b3dbeee044bf1b1ca9501851682b4 : SUCCESS in 9m 04s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/df2fec96a1494ea9ab7aae88838b95eb : SUCCESS in 10m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ac46a4d614754bd992de5947579ef09e : SUCCESS in 1h 41m 44s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ba7ae14e9368490fa74c7a2937e33794 : SUCCESS in 1h 51m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/af9a5d0a1d734a59a9d55bcc5b98c5b2 : SUCCESS in 2h 00m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/91f80f4207a44cada2190e629345883e : SUCCESS in 2h 02m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b20af1755426469facc51c44c74223e3 : SUCCESS in 1h 32m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3f5eb56d5bb4c3abf6a902de0939a20 : SUCCESS in 40m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e9727e63dcec403eb1e28e2c316080b3 : SUCCESS in 2h 04m 51s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/1b2329d859f44881b95b939d3b2789b1 : SUCCESS in 1h 16m 13s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5443f7a25ff0451591ed55e291502044 : SUCCESS in 1h 06m 51s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f5d2da56766941d5a835b76348daf566 : SUCCESS in 25m 40s","accounts_in_message":[],"_revision_number":7},{"id":"479004bb716eee3cfcb08f5ab6efdc779433f6b6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-03-18 05:36:55.000000000","message":"Topic db-migration-squash removed","accounts_in_message":[],"_revision_number":7},{"id":"272d17dd1bda4dba1af8039783a6ce8b730cf3a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-10 01:16:45.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"5c580c6fd4bd2e116eca9a54b67385d6ee04b22b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 01:31:31.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa10278e4a854ba897039fba98d8bdda\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f002f30cd854bf8aae4a1d5da467861 : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/17ac50cf62a14fb8ab45cd44f61f6837 : SUCCESS in 6m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0aa6297ebe0a089b3f268332b64991240dc52c8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 03:50:19.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1bf0ea068b34ca3bf08ecfac3856f4e\n\n- grenade https://zuul.opendev.org/t/openstack/build/31720ec53b0f40ef9d4892e7c4fb0ff5 : SUCCESS in 1h 00m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3560b08264c74be2826627a6a7e98839 : FAILURE in 1h 29m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8bd7b0a2205e44edb5eb43cd183d3d51 : SUCCESS in 1h 06m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81576b3c14b141efa118aa03aa44313a : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a7a8b46b9164a6bbd1d344603a18460 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3cf13d3e6c3462ab237b41c25cdf3d3 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e5bbc3aa557415db4ec11aab965b778 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a512278466241ee96b5afcda0606153 : SUCCESS in 6m 06s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2f90d9ac579e4df38369699fae0a1453 : RETRY_LIMIT in 2m 35s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/944ed93970d540e9b3cf253828100394 : SUCCESS in 8m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1b413b2938904131b43a93a4b761d10b : SUCCESS in 8m 35s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/48aecd8ee4484474a74c9bb067f12584 : SUCCESS in 10m 24s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/68556614f03748bb8db998bfd4b5399e : SUCCESS in 8m 11s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4f4c4b5999b646e5bb30d009914f3a92 : SUCCESS in 2h 01m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8d01e3b9e6484c1fa6c9c535fb4d27b6 : TIMED_OUT in 2h 04m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9c67cb37f79d42fb877340d339ed07a7 : TIMED_OUT in 2h 04m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7c4cf96f4f8d43b2a0513a20246c9c68 : SUCCESS in 1h 49m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f75ebdf61938458180fc88e3dac45fcb : SUCCESS in 2h 28m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abe7634c977045ce9b5c80703d343d89 : SUCCESS in 56m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5623fe9babb043bd8b5a67490e6c6e35 : SUCCESS in 1h 59m 15s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/441a5a327f9f48bba10788de161c1af2 : SUCCESS in 1h 50m 27s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/badc7e75aec74e34b0e7eeb9c99ddbf2 : SUCCESS in 1h 00m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4ac2dcd5fbdd414990d61d417a3dbee7 : SUCCESS in 35m 01s","accounts_in_message":[],"_revision_number":8},{"id":"79817a91f5e35c2bab137b7b7630e05486bb4c66","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-08 01:03:19.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"73a4fac1ee009b24fef821c17b641e0f68357642","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 01:14:21.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfa87481d4f048f5bc184d04a6a06c82\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41a1e97d177d4321bb29677c9f5d97a9 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4e17b6f8204645c3877ddaccca25201c : SUCCESS in 9m 11s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3fe2c90b027106201e196e07fed4e2e125855440","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 03:05:43.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57d89aaa9393456d91920d042bc99d19\n\n- grenade https://zuul.opendev.org/t/openstack/build/82ca140791b146b98f2d6d47676dcd22 : SUCCESS in 1h 06m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/919310fd2fd944659c057a3cd3796715 : SUCCESS in 1h 52m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1324318d6d914d8c95f92dda08a50cbb : SUCCESS in 1h 02m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71e1fbecfb10408b870a510206407064 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8cd3e2cd8644d5a8682a1f94f1681fd : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b07296fce12d456fa97d398797653d01 : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3a7be724ceb40dfa77e223ac8c2cd68 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d598159289144e0892c2872c91acd254 : SUCCESS in 7m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dc09f561ecc94f8cb93d8f48822971f1 : SUCCESS in 8m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/284e63043914443d8048c5443a8e48ca : SUCCESS in 7m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a707584946d84c33b9cec66da614e37f : SUCCESS in 14m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c3f510ee56d84d0c8b26da80eed44be0 : SUCCESS in 7m 58s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4025d479cff6486da328953e96dd5b21 : SUCCESS in 1h 46m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/df6515d18ed64920952c3efe20144c75 : SUCCESS in 1h 30m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/77d0ec66170c425d9f2c9e7a49c67ced : SUCCESS in 1h 45m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/93e0275f4bc14d62bb646310fdbd7532 : SUCCESS in 1h 38m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6a4694c177b549d99d3cbe1aca694eba : SUCCESS in 1h 46m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d25470ec17ee4ec58589254a42eaa4f5 : SUCCESS in 44m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b6460055da1b4a75bb168ccd477ec4c4 : SUCCESS in 1h 51m 19s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/250b8ece368249df85a5569895efbf01 : SUCCESS in 1h 48m 35s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/339756679bf64873af0e8f89cfc89886 : SUCCESS in 1h 06m 45s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/46dce3341d034a8b9c47c11d19464151 : SUCCESS in 34m 15s","accounts_in_message":[],"_revision_number":8},{"id":"a55f4f3b865363743d69806bd7ca86c522c53ad6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-10 13:39:10.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"870c0a3bbeffc8fb39629bef52a6d62cd90b04b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 14:23:39.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdebcda5b1c54d3eb83f99ac1b9b5119\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5bef798d9d784f499062e36777da3fd3 : SUCCESS in 10m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9e84585df2a145b4b2a3f994c8a03781 : SUCCESS in 11m 55s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8dc509afb0ebc507d029647740a6c5db5289c4d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 17:00:38.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65b7021a3510475a97f2c66203bdcfef\n\n- grenade https://zuul.opendev.org/t/openstack/build/b1abc3ffbbd64d1c802aab4b7d43d596 : SUCCESS in 1h 05m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9dfefc90bfc04628b9f362b0dc40c0fb : SUCCESS in 56m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/26e291c6fd4340aba8ff995543b9447e : SUCCESS in 1h 30m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2938ec3f868640e3b045ad57c46319e3 : SUCCESS in 58m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a078978507144938383854973bc9308 : SUCCESS in 7m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46f5853d99274223a47d01e3dc5caad6 : SUCCESS in 4m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/34b4ea8f360248949bcec490c0566249 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab334a7f48bd4840bcb59f037e7b7a5e : SUCCESS in 5m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/400b608b65db461ab605c866f9dd719e : SUCCESS in 8m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/38e53d0452fe4b54b0220079ef977875 : SUCCESS in 8m 06s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0546855e30d244729145a279a0917ee2 : SUCCESS in 14m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/42d9af982e864f1d97b299088650f893 : SUCCESS in 1h 46m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/25ed7ebe4d3543f6837e98c4d83cfced : SUCCESS in 2h 03m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0557382d98ed4ed9b096ca5b2e8efde1 : SUCCESS in 2h 09m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8f8c6dff744b4990b9984bf6bcfdc7c6 : SUCCESS in 1h 56m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e4e3d5743d7d49ca8ca3601aa3d96a22 : SUCCESS in 1h 27m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e098abb807034eb2b72de9c461acf2df : SUCCESS in 1h 01m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/33229061eaff459893798c5a3a9c06e1 : SUCCESS in 1h 45m 44s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/0dcc0209e82546d2a3dde60c2eaa78c1 : TIMED_OUT in 3h 09m 12s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9ddcbbb3a9f246b6ad28e7515c76ca37 : SUCCESS in 57m 56s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/43b150f5adb94082b49f14d518064720 : SUCCESS in 1h 55m 02s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c8f95b4fb0a74bdc96a7fa85e75cc831 : SUCCESS in 33m 45s","accounts_in_message":[],"_revision_number":9},{"id":"9283ba1f5a043a091bdacb301cd967daeb9f49fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-18 16:52:12.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"932ece0a89e33443b2fee569c555c7173e62d175","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-18 17:02:58.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89a6ef657fa247009adbf4455ba24ecf\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7eebc85aa7ec4c239e982b1a616e44a4 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/39056984064942df980b1f4332a9d70d : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9a69ef1b146e4e1a9dff57e128a7af58 : SUCCESS in 8m 58s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b61dd06b9003992196fecaca047f3844e9907664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-18 18:53:18.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c086acf17b494ea3ac72b8c895c4f2e4\n\n- grenade https://zuul.opendev.org/t/openstack/build/9c7e986c2664498d998c518807ce7224 : SUCCESS in 1h 02m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/24042aa2ffcb4dcdb70bfdfc1270a633 : SUCCESS in 58m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dd2529c2acc24440a7b8a889ecd0d20b : SUCCESS in 1h 03m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf4faf4a902548bcb877365531fe3626 : SUCCESS in 40m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0cb92388d9a41d99a17e4f02c5e25d5 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a69cba68144944618307ce8c9c539867 : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e4154fa1bc74de2bea8df151b6cad47 : SUCCESS in 5m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/32cc4aa56bac4699aeb1f55f2e8bf61b : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8740e5890b54c18a8467d1523c0f1aa : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dc947c7a9024182977daa8e497487bd : SUCCESS in 4m 22s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5ff47bcd88bf4f5090552c1764fe38ba : SUCCESS in 4m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6ee49b5efd64409fbb07751caf3e01db : SUCCESS in 6m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e8961864f4e54dd09a6d70b8b333e0e7 : SUCCESS in 1h 20m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/96825c46fd68488aaaa9d579be710f89 : SUCCESS in 1h 36m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2d712f99e5d04d94a10bfb8531b59dba : SUCCESS in 1h 47m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2f74f4fd55ba4559b1a499b2c39ef438 : SUCCESS in 1h 51m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c935159d8ae04bea96e8bcf379cfdc6c : SUCCESS in 51m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5ae049ec94fa4a1a82a58af04dcbc642 : SUCCESS in 1h 22m 03s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/1ab421ca53914810a00f3a15b635a1ca : SUCCESS in 1h 54m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ebb5cb48dc7e4a70b3d270b760cea289 : SUCCESS in 23m 30s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/91cdd579f0e5492a8f67ec08106eed29 : SUCCESS in 26m 56s (non-voting)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"3492086abec6426a2fb3d00b004cdaaa0b59c49b","revisions":{"a3349be5a10c8bd4222f4dc2072028f466f510c6":{"kind":"REWORK","_number":1,"created":"2024-02-11 16:03:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/908744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/908744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/908744/1"}}},"commit":{"parents":[{"commit":"5d729e3b4a91fb16f6fc24b309cee3937e6e71cb","subject":"Update base database schema from ocata to pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d729e3b4a91fb16f6fc24b309cee3937e6e71cb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-11 16:03:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-11 16:03:00.000000000","tz":540},"subject":"Update base database schema from pike to queens","message":"Update base database schema from pike to queens\n\nThis change is based on the manual squash of schema change in queens\ninto the base pike schema.\n\nChange-Id: I5d22af388746fb4863691f7f4c5ca91b2d05475b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a3349be5a10c8bd4222f4dc2072028f466f510c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a3349be5a10c8bd4222f4dc2072028f466f510c6"}]},"branch":"refs/heads/master"},"539a7bc1765e463992106cfaf7fa7e1e50484f79":{"kind":"REWORK","_number":2,"created":"2024-02-11 16:31:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/908744/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/908744/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/908744/2"}}},"commit":{"parents":[{"commit":"4e2d17abcf67cc57bd8996048e4fe0139d828281","subject":"Update base database schema from ocata to pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4e2d17abcf67cc57bd8996048e4fe0139d828281"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-11 16:03:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-11 16:31:31.000000000","tz":540},"subject":"Update base database schema from pike to queens","message":"Update base database schema from pike to queens\n\nThis change is based on the manual squash of schema change in queens\ninto the base pike schema.\n\nChange-Id: I5d22af388746fb4863691f7f4c5ca91b2d05475b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/539a7bc1765e463992106cfaf7fa7e1e50484f79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/539a7bc1765e463992106cfaf7fa7e1e50484f79"}]},"branch":"refs/heads/master"},"f92beb741d0630f9f090f5136adaed2073a0f711":{"kind":"REWORK","_number":3,"created":"2024-02-11 16:49:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/908744/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/908744/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/908744/3"}}},"commit":{"parents":[{"commit":"76249007dcf45faa8a6405acfb6884503662440f","subject":"Update base database schema from ocata to pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76249007dcf45faa8a6405acfb6884503662440f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-11 16:03:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-11 16:48:51.000000000","tz":540},"subject":"Update base database schema from pike to queens","message":"Update base database schema from pike to queens\n\nThis change is based on the manual squash of schema change in queens\ninto the base pike schema.\n\nChange-Id: I5d22af388746fb4863691f7f4c5ca91b2d05475b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f92beb741d0630f9f090f5136adaed2073a0f711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f92beb741d0630f9f090f5136adaed2073a0f711"}]},"branch":"refs/heads/master"},"2cbfd4be2aeb5100e395c70931b0c608a9b4756e":{"kind":"REWORK","_number":4,"created":"2024-02-11 16:53:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/908744/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/908744/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/908744/4"}}},"commit":{"parents":[{"commit":"8d08f350321e173797a9b99a2d1ef81974b80975","subject":"Update base database schema from ocata to pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8d08f350321e173797a9b99a2d1ef81974b80975"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-11 16:03:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-11 16:53:04.000000000","tz":540},"subject":"Update base database schema from pike to queens","message":"Update base database schema from pike to queens\n\nThis change is based on the manual squash of schema change in queens\ninto the base pike schema.\n\nChange-Id: I5d22af388746fb4863691f7f4c5ca91b2d05475b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2cbfd4be2aeb5100e395c70931b0c608a9b4756e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2cbfd4be2aeb5100e395c70931b0c608a9b4756e"}]},"branch":"refs/heads/master"},"d01a650f81ee56a8a13fdc44dd38f3de156a39c8":{"kind":"REWORK","_number":5,"created":"2024-02-11 17:32:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/908744/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/908744/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/908744/5"}}},"commit":{"parents":[{"commit":"18ff2fbe66daba4269589aebca4ea75ad94ae5cf","subject":"Update base database schema from ocata to pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/18ff2fbe66daba4269589aebca4ea75ad94ae5cf"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-11 16:03:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-11 17:32:02.000000000","tz":540},"subject":"Update base database schema from pike to queens","message":"Update base database schema from pike to queens\n\nThis change is based on the manual squash of schema change in queens\ninto the base pike schema.\n\nChange-Id: I5d22af388746fb4863691f7f4c5ca91b2d05475b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d01a650f81ee56a8a13fdc44dd38f3de156a39c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d01a650f81ee56a8a13fdc44dd38f3de156a39c8"}]},"branch":"refs/heads/master"},"57784f1b29c968791c65670110b112dcb1f3809a":{"kind":"REWORK","_number":6,"created":"2024-02-11 17:34:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/908744/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/908744/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/908744/6"}}},"commit":{"parents":[{"commit":"18ff2fbe66daba4269589aebca4ea75ad94ae5cf","subject":"Update base database schema from ocata to pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/18ff2fbe66daba4269589aebca4ea75ad94ae5cf"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-11 16:03:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-11 17:34:15.000000000","tz":540},"subject":"Update base database schema from pike to queens","message":"Update base database schema from pike to queens\n\nThis change is based on the manual squash of schema change in queens\ninto the base pike schema.\n\nChange-Id: I5d22af388746fb4863691f7f4c5ca91b2d05475b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/57784f1b29c968791c65670110b112dcb1f3809a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/57784f1b29c968791c65670110b112dcb1f3809a"}]},"branch":"refs/heads/master"},"ef10acfb770f9fd4ec1616f8804e104b41b3e23a":{"kind":"REWORK","_number":7,"created":"2024-03-07 14:12:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/908744/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/908744/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/908744/7"}}},"commit":{"parents":[{"commit":"e68a1c0daf0f60af731218c175b91fa84ab3f2ef","subject":"Update base database schema from ocata to pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e68a1c0daf0f60af731218c175b91fa84ab3f2ef"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-11 16:03:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-07 14:12:20.000000000","tz":540},"subject":"Update base database schema from pike to queens","message":"Update base database schema from pike to queens\n\nThis change is based on the manual squash of schema change in queens\ninto the base pike schema.\n\nChange-Id: I5d22af388746fb4863691f7f4c5ca91b2d05475b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ef10acfb770f9fd4ec1616f8804e104b41b3e23a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ef10acfb770f9fd4ec1616f8804e104b41b3e23a"}]},"branch":"refs/heads/master"},"a70ec438a2f77d88daab078cb6097dcce85c4f22":{"kind":"REWORK","_number":8,"created":"2024-06-10 01:16:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/908744/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/908744/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/908744/8"}}},"commit":{"parents":[{"commit":"9da0085bb955aeb251cb96f9904ae4493d6673c0","subject":"Update base database schema from ocata to pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9da0085bb955aeb251cb96f9904ae4493d6673c0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-11 16:03:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-10 01:15:47.000000000","tz":540},"subject":"Update base database schema from pike to queens","message":"Update base database schema from pike to queens\n\nThis change is based on the manual squash of schema change in queens\ninto the base pike schema.\n\nChange-Id: I5d22af388746fb4863691f7f4c5ca91b2d05475b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a70ec438a2f77d88daab078cb6097dcce85c4f22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a70ec438a2f77d88daab078cb6097dcce85c4f22"}]},"branch":"refs/heads/master"},"fd4e9fdded7489c0cd244b800e2d7bdf10c3058a":{"kind":"REWORK","_number":9,"created":"2025-01-10 13:39:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/908744/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/908744/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/908744/9"}}},"commit":{"parents":[{"commit":"b3c13ce7ff60eaefa771af0e28d80b5f80c0a3e5","subject":"Update base database schema from ocata to pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3c13ce7ff60eaefa771af0e28d80b5f80c0a3e5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-11 16:03:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-10 13:38:58.000000000","tz":540},"subject":"Update base database schema from pike to queens","message":"Update base database schema from pike to queens\n\nThis change is based on the manual squash of schema change in queens\ninto the base pike schema.\n\nChange-Id: I5d22af388746fb4863691f7f4c5ca91b2d05475b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fd4e9fdded7489c0cd244b800e2d7bdf10c3058a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fd4e9fdded7489c0cd244b800e2d7bdf10c3058a"}]},"branch":"refs/heads/master"},"3492086abec6426a2fb3d00b004cdaaa0b59c49b":{"kind":"REWORK","_number":10,"created":"2026-07-18 16:52:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/908744/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/908744/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/908744/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/908744/10"}}},"commit":{"parents":[{"commit":"4d853fc4c5ae9f6d3516672eed9b9afc90c7027c","subject":"Update base database schema from ocata to pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4d853fc4c5ae9f6d3516672eed9b9afc90c7027c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-11 16:03:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-18 16:51:59.000000000","tz":540},"subject":"Update base database schema from pike to queens","message":"Update base database schema from pike to queens\n\nThis change is based on the manual squash of schema change in queens\ninto the base pike schema.\n\nNote that all of the migration scripts added during this cycle were\nempty, so no schema change is applied.\n\nChange-Id: I5d22af388746fb4863691f7f4c5ca91b2d05475b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3492086abec6426a2fb3d00b004cdaaa0b59c49b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3492086abec6426a2fb3d00b004cdaaa0b59c49b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
