)]}'
{"id":"openstack%2Fglance~908753","triplet_id":"openstack%2Fglance~master~I5b5fba31421bfedc26f0b2217c6654c76b043487","project":"openstack/glance","branch":"master","topic":"db-migration-squash","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-12 11:17:11.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5b5fba31421bfedc26f0b2217c6654c76b043487","subject":"Update base database schema from rocky to train","status":"NEW","created":"2024-02-12 06:05:54.000000000","updated":"2024-06-10 03:47:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"677a290914fed483e6b5d566fbb9eeacee00c926","_number":908753,"virtual_id_number":908753,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-06-10 03:47:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-12 06:25:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-12 08:15:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"deb128e92e055a9d8a71c4d703a165a941500f84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-12 06:05:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e01127f373b60392ac29c4665f686fd9edb62032","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-12 06:13:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a8a0987c4a55af43e90adc95ca8e0ebdbf773802","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 06:25:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4727ca5da0d34c39bf1dddafec58436c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/21ca7f74f8394d06ae4934b7b2795b79 : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0046d6ec500a45e2b199980594e0c994 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/35a451e4e2794c73abc6d909a5c4ffe8 : SUCCESS in 10m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e16a1df5c38701a3c21f9782d47f24427a184565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 08:15:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d0819a945034fc087bd6225b656c9a6\n\n- grenade https://zuul.opendev.org/t/openstack/build/674c1eeb810c44239c4ad80e78dab2a9 : SUCCESS in 1h 12m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f818ded6227243ce81156048aa4887c6 : SUCCESS in 1h 05m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8bfef1bd062a467481afd96c8ea960de : SUCCESS in 1h 43m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcb8db2f099b4551b8c07cb0149a6c28 : POST_FAILURE in 22m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a318c2ce35f455196c1d460bb8bfcdc : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d18a357b203c4461ae38dc8ccdc94bb8 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a45b36c5b9f84a3f91b0a9a791a86815 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bb6680a1fe642edb11e859863c12c1f : SUCCESS in 5m 36s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/80a12afac62b46a4a993e8a69d91844b : SUCCESS in 14m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/116d4c4e812d4262823eec71a236298a : SUCCESS in 7m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/87a4687a0ae8480283d8add95c755228 : SUCCESS in 7m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e78265ffa1eb44f6ac10b2d295764000 : SUCCESS in 7m 59s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0908d3ff5b4a417a981d65a10ebe704d : SUCCESS in 10m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/59bb303fbb8e4bc5bf0476e9b3214d4e : SUCCESS in 1h 25m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2ced856bd7fc4fa9b77611d2b1abbffc : FAILURE in 1h 50m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/eedfc126cfaf455ea6b261bc43f5b8c8 : SUCCESS in 1h 38m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3c50833a9a3e46119a90c09d92f962e3 : SUCCESS in 1h 20m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9a3ea3ff68fe43918e1058bdf4906077 : SUCCESS in 1h 50m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f942eaefecf460793b5180cc0a871dd : SUCCESS in 1h 04m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9bf84a0b965540d8a95ee480a17a48e5 : SUCCESS in 1h 19m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4e4d698cd3804bee877dd28a3666c6d1 : SUCCESS in 27m 24s","accounts_in_message":[],"_revision_number":2},{"id":"7f28e700f013dffd536e6cb7fcb5a24fc6340664","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-12 08:58:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eed6c1dd8ff3bcdc72a5a3c8d8ffe73aaa04f3d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 09:07:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e47d2f0e050422b8470bff7bd8f0de7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74b155b4c90c4e148de3c8117e8c8e32 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/097616ecd5314d4ea440160d275fd8ac : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/edd75ea486df465c8b7ada46d2c9bc36 : SUCCESS in 6m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0d9be9b96b10f6cf76fc292251635f819c2274c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 11:17:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/811d27b6eb4a46e4a0a2888bdf4ec5c5\n\n- grenade https://zuul.opendev.org/t/openstack/build/06e3b5b1c8a84529a5f5a12686673d42 : SUCCESS in 1h 01m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e41e4ec1c6074c438e74bf3081c09eb9 : SUCCESS in 1h 05m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e9b081f8e66d4124b4266388c064c2a6 : FAILURE in 1h 10m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f57932ff2a94a5d98f7092c9d247777 : SUCCESS in 36m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62deb26138ca42c5a3869308cde7e01d : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bc3552bd73d4bfeaaa98b80d1bb6cd4 : SUCCESS in 5m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4bcd4d2779414767accb5c9bf4561dd1 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d14a907a3e874b40bbb7794c02218e98 : SUCCESS in 6m 30s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8aa3f28e44f1440f833c3816bbee7dc6 : SUCCESS in 16m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6a3e12e822d641b98adda2a6d2c3d7db : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a0f22755f58745709c82ec9b75317283 : SUCCESS in 7m 45s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/402d5f888231440bb938ef53ce20b7ce : SUCCESS in 7m 49s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/3e445ba531e64c459234c3f692122da0 : SUCCESS in 11m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d36c7a3ff8d74a0da33ead239d3229cf : TIMED_OUT in 2h 04m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c5482272ccc745efb9a897c67dfcaf2a : SUCCESS in 1h 48m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0b62f62a8836477f911a77dc22a073c5 : SUCCESS in 1h 54m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b033e970d2944f7eaff1d697e3e03d4d : SUCCESS in 1h 49m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/96280fde098e4135896818832be170ec : FAILURE in 2h 08m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e56144efed1548289ba4029d8c1a14e9 : SUCCESS in 1h 03m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7d9259debd3644c7af6834f0ae8894fb : SUCCESS in 1h 31m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/37a5ddd4bfa1432db705da75a1fa6863 : SUCCESS in 28m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2ea9f4e5870b15581b8ec2e18a51f738d9f4e68e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-08 04:05:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"63a707f1a537d0103eb2c3e369a7e987edbc886c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 04:18:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f753197e7c264aa38ed062327aee91c5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5510ba948f464e03906d578f07701451 : SUCCESS in 11m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/20bd1e615a2349e6a32517077fe48d46 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1d437388208f438e8ea6fcfc4aad9c77 : SUCCESS in 8m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"11daac83e09b531177d625a67ed3ccb8dd98b634","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 06:07:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec09cfdb1b424012881ab6cb142752bd\n\n- grenade https://zuul.opendev.org/t/openstack/build/f6c2309234a94e21832617f7685f9624 : SUCCESS in 53m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b11466c4e4844697ad709709c6a361e0 : SUCCESS in 1h 00m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b0b9850ade6540c18f16c33258db4fcd : SUCCESS in 1h 10m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a0776eddf054f34a8d5a27475965bb6 : SUCCESS in 59m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e5fc964fcc84ba0819a1f672a0cd739 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ed611f02b764705976627a0cb63780e : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e177f3f64754107a1e05f9c39b924c4 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f3033234316463583cd6bef8f05a852 : SUCCESS in 6m 13s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0ec39ad284494c7cb3449b10301dbc22 : FAILURE in 13m 14s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/312157fbc3d34179b7bae88af42ffb48 : SUCCESS in 9m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b47170aee83a4737816930b9acbf59a9 : SUCCESS in 8m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e7f2e3f12da0486487f8eb10c1d419a1 : SUCCESS in 8m 49s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/9cbc1e53a012494d83aa3920342d00e9 : SUCCESS in 10m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/80e98c70fa8841fd818316376d06add6 : SUCCESS in 1h 57m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/57247d423c274d09bc01ad684bf1101e : SUCCESS in 1h 50m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/32bdd1333eba4f08b204f678a39fd0af : SUCCESS in 1h 26m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bb233a560e96406b928a373d267b7b10 : SUCCESS in 1h 47m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/92a32a75ef294b70a0bf0f0dd21d2650 : SUCCESS in 1h 28m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0973855706e4f9aa70ca6d358ea6937 : SUCCESS in 55m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dacc2a9599bf4c39957aff9a7cfe42e7 : SUCCESS in 1h 53m 46s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/29dfbd70e2554627a6695588a7698747 : SUCCESS in 1h 32m 57s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9c65008aeb6943b8963edb6bbccf1cb0 : SUCCESS in 52m 02s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/188fd9c1a61141c8b67d48fbf4df1e26 : SUCCESS in 32m 58s","accounts_in_message":[],"_revision_number":3},{"id":"c552f7df6a0744d545ea5df6eea8545e7ad87416","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-10 01:16:45.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"1052247dd049d59c82da190cc78213cfae401f15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 01:32:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84de7ada25f1417bb91cc089d0317001\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c52697a9a7bf49e2aee3b0810a4700c8 : SUCCESS in 12m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/63184ff9a42f44718a3c40c6ec2dc78c : SUCCESS in 4m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"677a290914fed483e6b5d566fbb9eeacee00c926","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 03:47:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/946b1d2e7f60404db41eea4caa887567\n\n- grenade https://zuul.opendev.org/t/openstack/build/bbbbedd19f76414a8a4be62d5248d713 : SUCCESS in 1h 14m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e5914f837652413e8f530dd7a7745bfb : SUCCESS in 1h 34m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9c9c2c716d24f09bc2d1b59628c2d09 : SUCCESS in 1h 04m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bf086c7efab4778bfe93745847ef6ca : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0dacb34ae56444c893ccea87684bb689 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/22aa6053a26241ef9abf8bd8ac186e56 : SUCCESS in 6m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb1c57db5d074f76967da0cf58b5d287 : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1ebf687e4514bfbb10100d24561ebbd : SUCCESS in 5m 39s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e1453e3296954cf8a5b6fd34c6f6170a : RETRY_LIMIT in 1m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d611a69fd7d945958a4a33c659c45c8a : SUCCESS in 7m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/86c0684c19894bc991fb1681223b7e87 : SUCCESS in 8m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/47dcb6fed0d84d37ab251472b0c8454e : SUCCESS in 11m 02s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/66ee5d8a8c1e4b4b81a6d960a0c65328 : SUCCESS in 8m 24s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/95aa6b55a2ba4728afa9a306c1f06bdd : SUCCESS in 2h 00m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/451552f5e2ac404ca62405c44319e89a : SUCCESS in 2h 00m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0e9b6201c7834206868eb00a01113f6e : SUCCESS in 2h 04m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/810e688c275c49d0bd903f67630af3de : SUCCESS in 2h 01m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e68748ba08864808b1b4c157b1fa1cf1 : SUCCESS in 2h 23m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fe8b7f7533b4d59a1777e76c950d0e2 : SUCCESS in 1h 08m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/64bcfc3b8fb04490acaf3f46926dc81a : SUCCESS in 1h 59m 48s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/951c6bcb825c4e939f7ded7de4092e7b : TIMED_OUT in 2h 04m 11s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/211dc903478c40089766c3c90dabdbfb : SUCCESS in 1h 09m 38s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ebaeea982e2a4529a6b8127bf62177d5 : SUCCESS in 35m 34s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0e92aa9c65514bece89875248826fb0c61e5ee34","revisions":{"ba2db0552db4afc23f9b1a5c5e164b5e3e21a7c9":{"kind":"REWORK","_number":1,"created":"2024-02-12 06:05:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/908753/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/908753/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/908753/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/908753/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/908753/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/908753/1"}}},"commit":{"parents":[{"commit":"e1cb4726114de8a90fb6a25a51d0eb78ec937030","subject":"Update base database schema from queens to rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1cb4726114de8a90fb6a25a51d0eb78ec937030"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-12 06:01:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-12 06:05:29.000000000","tz":540},"subject":"Update base database schema from rocky to train","message":"Update base database schema from rocky to train\n\nThis change is based on the manual squash of schema change in train\ninto the base rocky schema.\n\nChange-Id: I5b5fba31421bfedc26f0b2217c6654c76b043487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba2db0552db4afc23f9b1a5c5e164b5e3e21a7c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba2db0552db4afc23f9b1a5c5e164b5e3e21a7c9"}]},"branch":"refs/heads/master"},"1118ee64a1f181326588e8bae557c012b3fc49d1":{"kind":"REWORK","_number":2,"created":"2024-02-12 06:13:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/908753/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/908753/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/908753/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/908753/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/908753/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/908753/2"}}},"commit":{"parents":[{"commit":"e1cb4726114de8a90fb6a25a51d0eb78ec937030","subject":"Update base database schema from queens to rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1cb4726114de8a90fb6a25a51d0eb78ec937030"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-12 06:01:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-12 06:13:13.000000000","tz":540},"subject":"Update base database schema from rocky to train","message":"Update base database schema from rocky to train\n\nThis change is based on the manual squash of schema change in train\ninto the base rocky schema.\n\nChange-Id: I5b5fba31421bfedc26f0b2217c6654c76b043487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1118ee64a1f181326588e8bae557c012b3fc49d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1118ee64a1f181326588e8bae557c012b3fc49d1"}]},"branch":"refs/heads/master"},"b060bdd62e7493e18a836a15d13f928b740c0673":{"kind":"REWORK","_number":3,"created":"2024-03-08 04:05:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/908753/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/908753/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/908753/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/908753/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/908753/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/908753/3"}}},"commit":{"parents":[{"commit":"1258fbc650b74daa2c0238d114b7d28dfc92774e","subject":"Update base database schema from queens to rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1258fbc650b74daa2c0238d114b7d28dfc92774e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-12 06:01:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-08 04:04:52.000000000","tz":540},"subject":"Update base database schema from rocky to train","message":"Update base database schema from rocky to train\n\nThis change is based on the manual squash of schema change in train\ninto the base rocky schema.\n\nChange-Id: I5b5fba31421bfedc26f0b2217c6654c76b043487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b060bdd62e7493e18a836a15d13f928b740c0673"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b060bdd62e7493e18a836a15d13f928b740c0673"}]},"branch":"refs/heads/master"},"0e92aa9c65514bece89875248826fb0c61e5ee34":{"kind":"REWORK","_number":4,"created":"2024-06-10 01:16:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/908753/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/908753/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/908753/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/908753/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/908753/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/908753/4"}}},"commit":{"parents":[{"commit":"7d56eadf0bd654741322aff4edf762ac63bdd2a2","subject":"Update base database schema from queens to rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7d56eadf0bd654741322aff4edf762ac63bdd2a2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-12 06:01:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-10 01:16:08.000000000","tz":540},"subject":"Update base database schema from rocky to train","message":"Update base database schema from rocky to train\n\nThis change is based on the manual squash of schema change in train\ninto the base rocky schema.\n\nChange-Id: I5b5fba31421bfedc26f0b2217c6654c76b043487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e92aa9c65514bece89875248826fb0c61e5ee34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e92aa9c65514bece89875248826fb0c61e5ee34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
