)]}'
{"id":"openstack%2Fglance~910042","triplet_id":"openstack%2Fglance~master~I82d9e7bf3f1f582030ac7729d729a746784c668f","project":"openstack/glance","branch":"master","topic":"ddd","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-02-28 05:13:38.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-02-27 15:26:21.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I82d9e7bf3f1f582030ac7729d729a746784c668f","subject":"Deprecate glance scrubber","status":"MERGED","created":"2024-02-24 19:01:56.000000000","updated":"2025-05-19 17:14:50.000000000","submitted":"2024-02-28 05:13:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910042-ddd","meta_rev_id":"3ef621785c7ae2cbe37aeac44d2948c5b1841b41","_number":910042,"virtual_id_number":910042,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-28 05:13:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-02-27 10:04:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-02-27 15:26:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-28 01:26:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-02-27 15:26:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-24 19:09:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-24 20:30:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-26 18:07:36.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-02-27 10:04:10.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-02-28 01:02:23.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-02-28 01:26:10.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"2bfe9a9dca4e5c8708553dc5be79950909c61f11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-24 19:01:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d213f3e02fa203165eff01dcee38093b0f75e151","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 19:09:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ebeadc1ee7c408e88222e8fcc38c28c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/64b9ffeb4e6842c9915d1cb544bab4a0 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e8e3cf10fcbf48e38637e008bbe674b1 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/12cc3733e9774d6c820c66e075df2a0c : SUCCESS in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2faf6ab2051c886b1791a9435c8b0d6622057b8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 20:30:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a0f6e1d85324b72b692ea54e3b32ec0\n\n- grenade https://zuul.opendev.org/t/openstack/build/443f672012ab4fabbb382ed5c9e1f3bf : SUCCESS in 1h 01m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3bea3986942e46bcac469b8b79849fce : SUCCESS in 1h 06m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/60731395ad5e4b27920701b9f3a58c44 : SUCCESS in 1h 22m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/caa7ab3be5b649499b4a1c3b1b82fe1d : SUCCESS in 1h 03m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38d1e3ddb0374d94a0d2d7a5f6c3efbe : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ce57ee5d3224248bb39b733fcb1cf40 : SUCCESS in 6m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c4a701f96014c5783cb7d1b0a0ba1b7 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc8f7334f6044d4982bdbf3e222d4b90 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f186876a312241e48578d99211c1e23d : SUCCESS in 6m 04s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/12bec0f45d4342cdb6a253084ee554c6 : FAILURE in 11m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/030d5cd1df2141edbe191283f5ae292d : SUCCESS in 9m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ca911f3de899459285d0fd7ce641f170 : SUCCESS in 9m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7ba07c55cfa54262909d0c975a996add : SUCCESS in 9m 23s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/25394ed78ea945929d950d7880f0cc26 : SUCCESS in 11m 23s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5adecc465c4a4ccabe43a02b42c56e19 : SUCCESS in 1h 23m 08s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/32bc25d2c7d7472f8f0801344d5e6900 : SUCCESS in 1h 15m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1fee388372d34189ad7260e6f0ba065f : SUCCESS in 1h 19m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6ee4925e1d1e40d0ba16f83adb0afb50 : SUCCESS in 1h 19m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1c6051c24ca2472183f52d03e6217bf8 : SUCCESS in 1h 23m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abd2fb9ffecc43c3a4f147f2d0af6282 : SUCCESS in 39m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/71d11afb5fa7406f98e344a96327fb24 : SUCCESS in 1h 24m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6b8cc770c98d4329be2c420f2aabacbb : SUCCESS in 41m 41s","accounts_in_message":[],"_revision_number":1},{"id":"7bf657d6a8b69a486a2d852080b657935f544fa4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-24 21:25:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df66f5f898516e561878dff7e81230c62537d28f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 21:34:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a25755bf07b4d1a8f97d2340aec00aa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/35cae81df1d540ee8d40c00b8e29817e : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/42e1ddb470204ba48bd95faa8f490757 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eb06a652a77b44f5b7a5a556355dcacb : SUCCESS in 5m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5acb4703efa4233f3ba7f4a7d681eca72803c868","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 23:33:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/094f7f718ddb4a2198240e96725bb759\n\n- grenade https://zuul.opendev.org/t/openstack/build/fa138e12e725414bac99e952468aa7d0 : SUCCESS in 51m 48s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/503f2d721a3547699902c2b58d75c71b : SUCCESS in 46m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e7e7cbecdd2d4042ba0ad4f7eac207a9 : SUCCESS in 1h 10m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4140e8e8cd147c1b59048be49544d52 : SUCCESS in 43m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80dbe5aae8414312b5cc3608cf484c3a : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/602df307868b4d949ae5363332bbf123 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9870dbf9348c410a95d3a8026acb6e27 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1caae37d53d54a52890fa5203d59f497 : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e9852655cc84669b8519bee10f3f251 : SUCCESS in 5m 06s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/720d20da55ec459e9a76aab88ebbb6e7 : FAILURE in 13m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7225da8a5f74442ba746d027b3d44c82 : SUCCESS in 8m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e09c83b22d1247dd890d36dd7e1ae2fb : SUCCESS in 8m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6f478e83d4f341d080d94d5099a75f14 : SUCCESS in 8m 07s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c0282ea2b7284e6d8cbb97a3278b5b74 : SUCCESS in 10m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d9fdde527c2e430ca08e770af1694f45 : SUCCESS in 2h 02m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3b6759b6dfe94a3f949b938fb6276efe : SUCCESS in 1h 10m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4c8df7d0300044e192545b5085ae07b4 : SUCCESS in 1h 38m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d37ada6ffb7f419f9394af5b952f39bd : FAILURE in 1h 41m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ebcf4cee8eff40c59e7435490cd08e8c : SUCCESS in 1h 54m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/035e2f3e9c7748d596913f7af844a307 : SUCCESS in 1h 10m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c3e559ca70dd4b288d9ac8186b73d6da : SUCCESS in 1h 52m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9964905f4b544a07a996c45b863a08b9 : SUCCESS in 37m 50s","accounts_in_message":[],"_revision_number":1},{"id":"1a127cd55d2562f89dc2324f090a328f2cda6cfc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-26 13:09:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e8ff7fe86feff26c1b1676c9f5d89dcb159decb0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-26 13:10:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b120e749c1359f8e217658811ab1086feb1405c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-26 15:08:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"100d56f3e595e45537f139c0d97cf2056eba5f18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 15:16:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2407305e096c489b8bf709fb707459d3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b10512dfb77841618b0ea9864b7315f2 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b3547059fcad453fa675d7e9fa976a64 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a71b51b9f7a1454ba487151136a23a12 : SUCCESS in 6m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5c9debfdf99b971026112fbd92e80f442ed3c6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 17:17:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2890ae5732cd466bbf642aa86e877ab6\n\n- grenade https://zuul.opendev.org/t/openstack/build/61108ae9c1a7444db0460b7b59cf0f68 : SUCCESS in 1h 10m 55s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f53e9d5551e94a828c5b7a69385e9903 : SUCCESS in 55m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/40f8ec0b490648c0b0c2cbbb911492cb : SUCCESS in 1h 12m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9aeb17e471041f49bfcea5b2ce3b0fb : SUCCESS in 48m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c76b24fcdbe4305afd19dc8d776ee14 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/183eb434409a40dd9fc0f103ccf50e32 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a51be014efe448a58c0e805e15564c80 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc0a8723d2a14cbcb78f80e155226e3b : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9a074e80bf941cb93c77354d0b59c62 : SUCCESS in 6m 54s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f307f95fcc8543f387bae4e557564247 : FAILURE in 13m 18s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/db70dbf63b6d4841b1bb3c40b37e23fa : SUCCESS in 8m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e1953f7f1dce4078b704aa3273873eb4 : SUCCESS in 8m 40s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b78bfa8574ed466c85120a0bc61b4c90 : SUCCESS in 11m 09s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/1a4464de7ac04f0c9bff0e3a2b8c702b : SUCCESS in 14m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/46db4f9041724849bba7f4ee088a0718 : TIMED_OUT in 2h 03m 55s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/96e336d6d18f4809bd98a44e5c3c8da8 : SUCCESS in 1h 41m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3f8cdae54916425cbf98218e0c5e6a0b : SUCCESS in 1h 51m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f24ae403578b432991f4595c65ad628a : SUCCESS in 1h 49m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/56dc8e15eb82485fbf484267ada214f0 : SUCCESS in 1h 43m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0aeb5112a7e64b2cbad99f712dcfc5f0 : SUCCESS in 1h 04m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3323ed02b5b8487bbc06a9dd4062dc90 : SUCCESS in 2h 01m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b116deee47c64a2fb9ebd4808e5d311e : SUCCESS in 36m 46s","accounts_in_message":[],"_revision_number":1},{"id":"cfd159c2d4b8fbff3c040b4cc1f91dd3e05d25f3","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-26 18:07:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d525a31a0e9812fc5ac919e646d3873710f34ef","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-26 18:08:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f921b3471fc5d74d634a91911a2ce2f0b75c283","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-27 03:03:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84ade56f16fce51f85be7a6b83cca82358f9a8c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-27 03:04:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c79d7607ce1dc649d8c3eb2990d6acd5303a474b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 03:19:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/739d6419d325437fb2e571d616fce32b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/57627c8f05524c939f7cf5a9d73be7d4 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7298c8ac7beb442485dee4fba25e67fe : SUCCESS in 13m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e5f7a2f824be4e0894a94b86da293d85 : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5eef067f0fc182e1c1535a73235062bd72cdb452","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 04:51:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9ff2f3c59ef4312bde73f9e56007b34\n\n- grenade https://zuul.opendev.org/t/openstack/build/034b2a7148e7452c9752274dc57b12a3 : SUCCESS in 58m 44s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/617086521be84ed985d2902f0766caa9 : SUCCESS in 1h 06m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/40becca75e3741dfa50635d62f095482 : SUCCESS in 1h 13m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb4d2d0f089241278cd5c6d797d71573 : SUCCESS in 1h 12m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b29a6f2973d94a52914c5b89fbe2cc73 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4df13a73daee4fbc8e6d7aa2d6bc1093 : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77916de30db948a09e71ed78567c83a5 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ed1c7d6e1b649ddbecc0e4d1e58bb3d : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1bcba501ae54ed6b1fe71575d9fd2d7 : SUCCESS in 7m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5797b1e7cf274668bc1ffbd14703e654 : FAILURE in 14m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a1755c019834417e9de52c19069924cb : SUCCESS in 8m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/541843444e3442f29813eda2ebf97394 : SUCCESS in 9m 03s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/52dd5edc803845578e7a2127610cbff1 : SUCCESS in 8m 34s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a1715ae04977471faed69b40b4c56dd4 : SUCCESS in 11m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/093f0c7dde124862aa724a7349a164ed : SUCCESS in 1h 42m 16s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9adb06ba669a4a4591e1d0e16408943a : SUCCESS in 1h 35m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/488e0fc38d8e4378a394c1bb99ba255b : SUCCESS in 1h 20m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d7aecacadc1d48d2a3b633f8ab74a02f : SUCCESS in 1h 32m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4f65a23c2d7c4cb29ba6db2d273ff0f0 : SUCCESS in 1h 26m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/272348beb390486cba898b2c8910322d : SUCCESS in 52m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e8ac1360704e4bb7967d4b2c6cb76ccf : SUCCESS in 1h 35m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1df03e6ab60a4ff1874b6b9086869975 : SUCCESS in 40m 30s","accounts_in_message":[],"_revision_number":2},{"id":"402cb531a166130477e01a27335855ec3885572f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-27 10:04:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"04e68ec8dacca254bb0e845e041e3e77ebfdceea","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-27 15:26:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d7e847aa4d302bad551b94f960a18ccad56d63b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 15:27:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2688674f60c6f1a717649239d08875c2cf05a7ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 17:44:33.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc0dbf6acba2452592d6ba740c88afcc\n\n- grenade https://zuul.opendev.org/t/openstack/build/bfff706377b747c89b666fcee5e9fc93 : SUCCESS in 47m 34s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1ca349e4922942d09604d2189bb77665 : SUCCESS in 1h 01m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b753cb36b524d4e9755158a522ad9d7 : SUCCESS in 1h 28m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b23f54b38504438b26e28ff11dccba9 : SUCCESS in 1h 01m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc0671bc29e8458b8446306667e97b3e : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c02182b4c9984c3694ec030f9359410d : SUCCESS in 6m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73af76206a3d484c9ea04b4d1d70183d : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4805ee484c664e05afaa141b71084eed : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d1bec10a4e0449d964c6378ccc44e59 : SUCCESS in 7m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ccc11209ef6d4f029241a4db01ceb393 : SUCCESS in 8m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/37ff5791bf7049fdb84aa7ab51b4611b : SUCCESS in 8m 21s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/da39f8de6fb34e88ace72652fc393716 : SUCCESS in 2h 02m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2fbc01a3215f48f88426b3ecb0bf42be : SUCCESS in 2h 00m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d33feb49bd30438b9d880904fac0ca8c : TIMED_OUT in 2h 04m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f90e8ff1a5f247dea7af48f9f48f9ea8 : SUCCESS in 59m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/24f36b09f7e74725a70f023fbbc32a22 : SUCCESS in 1h 41m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/006031bffb2c455aadda8c61320056b0 : SUCCESS in 41m 30s","accounts_in_message":[],"_revision_number":2},{"id":"3b7c033ff425207005607fc69771f566f4010afb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-28 01:02:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4c587d249f803627e61b48ce9baab5b8e192fb02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 01:08:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e42f91fe923e41dea2a1ae95883ebf75\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/27129aec4ada41e1bc0e7d539ce5302d : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/61e1ba3db534425eac78ca9a63992cf5 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bbc6ed4f3c5b4551a959d71ddc625f02 : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b9ba12be5a2cb6c9d013ca9e0e6d9cff7713f116","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2024-02-28 01:26:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"dbed8de2b7bde7130e38190d5183671e65f79495","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 03:10:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/861701ed8bf14db2859423526f78863c\n\n- grenade https://zuul.opendev.org/t/openstack/build/61a3ec83ae1c457d825c7e8f03e4ec45 : SUCCESS in 1h 15m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/45c93ce458914877986e719de681e273 : SUCCESS in 49m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f0ab1d83e1d947c0a8e50b9ca8b8004f : SUCCESS in 1h 36m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5e93debca7349a0a49dcf615083d4ef : SUCCESS in 1h 07m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dec5dc25a9d14ab38beb24251780c395 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0e76f01fee04c469ede6434a2b521db : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aefdd8372be746bebb64a19a0f771a22 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf92ee89ae504a3bae42ffe76a1fe4e7 : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d177a76d72e4a2e876495cebc99f0aa : SUCCESS in 5m 02s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/902821dc967741c5b6a13dfa85a6e835 : SUCCESS in 13m 25s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bd8f57fb7c604d999836d9a83a34c136 : SUCCESS in 8m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c9ec2d3411d94863aa0dec60fc105c5c : SUCCESS in 8m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/48fd29fbc08440ba941d63c0580c7984 : SUCCESS in 8m 03s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/fce0cfd7f36441709a4610b703d384ed : SUCCESS in 9m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cbcedc56eb534451805eff9c6acfadbf : TIMED_OUT in 2h 05m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/04e65ba310b64796a469155dabf14228 : SUCCESS in 1h 43m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4bfe4981b4e64c6c964b792219d91201 : SUCCESS in 1h 43m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/22c9a364660e4662928a4731bdcbdcbb : SUCCESS in 1h 41m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8e14cdaf166f4f31992642b75ffbe46e : SUCCESS in 1h 40m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91ebd8fbca88464eb1470e04c9c2e370 : SUCCESS in 49m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/999eb20d350f4c84b7be0f6d64161eea : SUCCESS in 1h 37m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fe7322bc511740a3aa68219012d1d251 : SUCCESS in 30m 52s","accounts_in_message":[],"_revision_number":2},{"id":"ced3c5573b3401033e63439e42ff7619fd7462af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 03:10:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"80e598abed79a419f1c873dfbbd47bbe85d25d7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 05:13:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/205a11b095154c6a82a6fd16aaff6cbd\n\n- grenade https://zuul.opendev.org/t/openstack/build/1a31eaa8e8ec439fb7a43ff531a3bb23 : SUCCESS in 51m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d322541d97434c2d9ab93cc0d8898cf5 : SUCCESS in 1h 12m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef810b0d87ec4f7d8cceac41dbdc2dcc : SUCCESS in 1h 23m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64536e78bf034f64819d672ee119ea83 : SUCCESS in 59m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d9e476771054714a83851131ead2874 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdd3d154c30d44d6a5beacecc69f01c1 : SUCCESS in 5m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a2469380d5447e2b665fa3fe1c7349c : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/379bf6e5854146f6a64f95a38bffb19f : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4911129d7b8e44a18551a57cebbd6ba4 : SUCCESS in 6m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bc1ddec083434c2eb303d1d5ce66035f : SUCCESS in 8m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/726c293d7fd94e08979f2a73b9890c65 : SUCCESS in 9m 20s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/35964a45af524f3bb012a02e7e97db69 : SUCCESS in 1h 44m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1bc8086c8ad3476bbca815bf15df3493 : SUCCESS in 1h 49m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/80b63a9b9c474b7db30714acd91ff71d : SUCCESS in 1h 47m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/449343853c0444608e11410540333a9e : SUCCESS in 52m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/14e5e33dad0142f796a6e60c4129de81 : SUCCESS in 1h 58m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2385941940b54d9d9fba4e642c36b24f : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":2},{"id":"8cbda0af2609dd920031bce4f02f2eff67867f75","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 05:13:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8a6b592a08db3c7a49ce2b0fb22c2a7f82143722","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 05:14:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/431825d99ff544b8acad07696919b52c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e35e46360a84412eb408b161a0a4038d : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b98c3fe1c9b741dd847a241a6252cb20 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2},{"id":"3ef621785c7ae2cbe37aeac44d2948c5b1841b41","tag":"autogenerated:gerrit:revert","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-19 17:14:50.000000000","message":"Created a revert of this change as Ib96684366c4b5ec4ccd37b06b83bf9962eb03380","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b059afbe2c1fd584bda1a0191be9a792de8e70f8","revisions":{"fcc7ffae81f6ebbf3777935946c3e1f33ab0c461":{"kind":"REWORK","_number":1,"created":"2024-02-24 19:01:56.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/42/910042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/42/910042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/910042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/910042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/910042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/42/910042/1"}}},"commit":{"parents":[{"commit":"02be49c998c578284052e73ec1a6bef7193afc8c","subject":"Merge \"Add new tables for cache operations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02be49c998c578284052e73ec1a6bef7193afc8c"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-02-24 17:27:52.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-02-24 18:57:42.000000000","tz":-300},"subject":"Deprecate glance scrubber","message":"Deprecate glance scrubber\n\nDeprecate the glance scrubber utility and its associated\nconfiguration options for removal during the 2024.2 (Dalmation)\ndevelopment cycle.\n\nRemoval of the scrubber was proposed on the mailing list on\n2024-02-01:\n\nhttps://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/33F37CL4NUE3QI6GYBWG5BL5UNBK4L5U/\n\nNo objections were raised.\n\nChange-Id: I82d9e7bf3f1f582030ac7729d729a746784c668f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fcc7ffae81f6ebbf3777935946c3e1f33ab0c461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fcc7ffae81f6ebbf3777935946c3e1f33ab0c461"}]},"branch":"refs/heads/master"},"b059afbe2c1fd584bda1a0191be9a792de8e70f8":{"kind":"REWORK","_number":2,"created":"2024-02-27 03:04:06.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/42/910042/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/42/910042/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/910042/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/910042/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/910042/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/42/910042/2"}}},"commit":{"parents":[{"commit":"02be49c998c578284052e73ec1a6bef7193afc8c","subject":"Merge \"Add new tables for cache operations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02be49c998c578284052e73ec1a6bef7193afc8c"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-02-24 17:27:52.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-02-27 03:02:24.000000000","tz":-300},"subject":"Deprecate glance scrubber","message":"Deprecate glance scrubber\n\nDeprecate the glance scrubber utility and its associated\nconfiguration options for removal during the 2024.2 (Dalmatian)\ndevelopment cycle.\n\nRemoval of the scrubber was proposed on the mailing list on\n2024-02-01:\n\nhttps://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/33F37CL4NUE3QI6GYBWG5BL5UNBK4L5U/\n\nNo objections were raised.\n\nChange-Id: I82d9e7bf3f1f582030ac7729d729a746784c668f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b059afbe2c1fd584bda1a0191be9a792de8e70f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b059afbe2c1fd584bda1a0191be9a792de8e70f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
