)]}'
{"id":"openstack%2Fglance~910203","triplet_id":"openstack%2Fglance~master~I7490eeaa7352d38fba95c40bbb17bd970887be8b","project":"openstack/glance","branch":"master","topic":"glance-caracal-m3","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-03-07 11:03:30.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-03-07 07:25:34.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-03-07 11:03:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7490eeaa7352d38fba95c40bbb17bd970887be8b","subject":"Refresh Glance example configs for caracal m3","status":"MERGED","created":"2024-02-26 10:33:27.000000000","updated":"2024-03-07 11:04:59.000000000","submitted":"2024-03-07 11:03:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910203-glance-caracal-m3","meta_rev_id":"45956a2e17ffc5a029590172123532257bdf6ed9","_number":910203,"virtual_id_number":910203,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-07 11:03:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-03-07 07:25:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-03-07 07:27:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-26 10:52:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-26 12:42:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-26 14:17:35.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-02-26 14:17:35.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"4fc9b77c7052f2c626aa64bed8699e5a020dc9ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-26 10:33:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b74cb7d1563e151b3e8b3ae50323836c0fb5a4d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 10:52:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eab3c1b1f9ee48529b264e061d6336cc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8eec7a79ddf4a1785d61ec32e882c08 : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e2be8aa47c4547edb7d1da12e1fdc357 : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/83f10b2f57b74b1db874fbfbe5d680c0 : SUCCESS in 11m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd2ba7fbc5a4bc97e71188e144b5c29fc56b54f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 12:42:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1807f86ca27242ff98c78c76dbbece80\n\n- grenade https://zuul.opendev.org/t/openstack/build/b814c5dea27c45c8b424d91e2158f900 : SUCCESS in 1h 05m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ecf3ee9319ab46d2a398ac71e31a4365 : SUCCESS in 1h 02m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7409903cb79e4b9dbcc69d0daebe7b39 : SUCCESS in 1h 31m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e097bc19fc4b440ead534b5bde39c187 : FAILURE in 53m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba01825dedc14c7686a32cb7c0fbd4a0 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f69290c9c2a46e28eec31e5e8bd39a0 : SUCCESS in 6m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/192619d19d6347a29b3e349f0b1a18e5 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8632545f3ec4791b4dbdd6339e52964 : SUCCESS in 6m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ad76caf6f959415ab2d42deca8d55d30 : FAILURE in 13m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6dfd22ccd9dc4f4cb0bfa1dbd6f14076 : SUCCESS in 8m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/582aaa97f9d64787b559471d6313aac6 : SUCCESS in 8m 14s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/06e2b98b22c844ec9e8b0f0070bf4da4 : SUCCESS in 7m 57s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b2c2b2fe14bf4a209fd0e246bf4a902c : SUCCESS in 11m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/828dc473154a4880be1d5475217b6882 : SUCCESS in 1h 33m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e384daa3681b42f48f9e27cf5b9b0b72 : SUCCESS in 1h 26m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cdf030d4f26043bfa19089b9a6f10bac : SUCCESS in 1h 52m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/60c9b2044ff944d4aa11267134338aab : SUCCESS in 1h 46m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6d3e69e8c73f4ee99755e76775b3b144 : FAILURE in 1h 57m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d3f0222468549bea0e05aef1b77f54d : SUCCESS in 56m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f591d6b11cdb4d81ad888140cf3345e1 : SUCCESS in 1h 50m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f9ec62254fe94c5aa191eb01c5af712e : SUCCESS in 28m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a44eacd09801ee4f6009e3eb7dc773cfcdabb006","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-26 14:17:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df1606c1fbab65c1268663497ec871e47c08c2e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 14:24:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f590f5e1e044cebb205f7ea26b8cdee\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/728b603cd9674fa096f7f978c44323e6 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e19fbc93e8a34a13a050fb47c9ef3f2d : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/05493f06ebdf464dbd033cfa9ba37706 : SUCCESS in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da99bf5cb644e5b21c6a369ddf74c00e8bab5722","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 16:26:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0f1f9b0ac024ad49754b5ec9d48e424\n\n- grenade https://zuul.opendev.org/t/openstack/build/6ad3904e119e487b8ad80592f668cc1b : SUCCESS in 1h 00m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8170f7852af042c189b93fe812db25b9 : SUCCESS in 54m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f879fd1c33864af7a3a1f6f6867eadde : SUCCESS in 1h 25m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8aa0451c6c4a4d33a11c116ac21c4874 : SUCCESS in 55m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c1b795860e247bd83f2e9718358dd86 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57db0f9c201346a5ae63efd9c04637ed : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7b5872bb2504fc59c5e2afe2a12d36b : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdcee039c50c4a869043ee81313b6f7a : SUCCESS in 5m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0dfd4887efc44f23abc52680466611d1 : FAILURE in 11m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/017a82340c3a4fb585bd7abb114e269d : SUCCESS in 7m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/21a3d1094f284d47912e7bb24c785485 : SUCCESS in 7m 51s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/425db490005c4b12a2f388851bdd2cfb : SUCCESS in 7m 42s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/33acd6dd58164fdcae2d3552e45762c5 : SUCCESS in 11m 18s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d66a1f6e043d4932b0405ccd09428e44 : SUCCESS in 2h 00m 12s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8cec6ea458fc4f829e163ad9c85e8614 : SUCCESS in 1h 47m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/033ba33dcbb1436ba1d35986dfb9f59a : SUCCESS in 1h 59m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/01d9138b04c347c4a4dc82b7e59bb6db : SUCCESS in 1h 18m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/df5cf321eb1b428c9714da0bb39afbe3 : SUCCESS in 1h 39m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/370c3e51c81f4ca2b14f5c9bbb469d39 : SUCCESS in 43m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b3578a1eaa824803820746b9fbebda60 : FAILURE in 2h 07m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/86b586c2d19c44c198106bd6a49af6a3 : SUCCESS in 27m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7f72f7761df3f11eb60c729662c4c01f57678bd1","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 19:28:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"52ab331b0f3caa680db2b2309a808921904857d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 19:35:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4342b35b64474ab09e9be97b99878f2f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fce809167a314fa68938bd4ecf65a750 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ea7115dd485947a49f03947da671250a : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0e9fedccdd7b43c29382c049af8b697e : SUCCESS in 4m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"15910bdd61ad31cd3cddcc17f5d4ff523e0447fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 21:47:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44ab0e7089e24515b17d3660baf0b54e\n\n- grenade https://zuul.opendev.org/t/openstack/build/e7d62a28da914ba993a5324328a0ae7e : SUCCESS in 1h 21m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a5d5b5bef6a242da87cc58ade1f471cb : SUCCESS in 56m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/09763a94c9ad4fdda48c7eb0b1d4db18 : SUCCESS in 1h 31m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08fe16b2dcfe40a3a19ee8aaed7c7ab3 : SUCCESS in 1h 01m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a18f16cb67440ea95a16f944d9c7163 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/354bbb71f71246dab3c98108890eea81 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c889cf0214ec48a384a6673384ba960a : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca80ffa2324345fea4e60a56fb964198 : SUCCESS in 5m 39s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/daba9e9f5f7b401b99cd9f0f275ce411 : FAILURE in 14m 23s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f40449baa1fa404cb6656a3156ad7d1b : SUCCESS in 10m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e30e0afae45a4c28893687aa0fc54c7b : SUCCESS in 10m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5cb766ddd5c74618a5f1b4cfb2ea3f3b : SUCCESS in 10m 46s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/9a210f4c663d436c918c87e771d7614e : SUCCESS in 11m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ef711693f0a045f89c930ec89c6afa1a : SUCCESS in 1h 58m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2e6ca7cb8e5c453daf649141bae9c56d : SUCCESS in 1h 27m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/df87fa387ed64f3da44b583003045f75 : SUCCESS in 1h 28m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ae6ab35a0f9141d388d2f0dda93b9d7d : SUCCESS in 1h 26m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1d9dc02ecdee432795d25ee590fd0468 : SUCCESS in 2h 12m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e57c5747fae64866a35cb4eb7f1fa032 : SUCCESS in 49m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/612c57d3f03f441483883a39207445d7 : SUCCESS in 2h 09m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4e45bc12307e404e82218899e8f423df : SUCCESS in 30m 24s","accounts_in_message":[],"_revision_number":1},{"id":"873a56be591f1fc1a3f1cccfb9e598f37ce385f4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-29 11:22:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f1cb338484c6baff5c668d3b728dd6dbb0ecab16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-29 11:36:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"93ec4255a7e417a89cc8db2422cfe9ba514a2b5f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-29 11:42:20.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"876e795acb38b9ee0fa4c8ae8af61b7a046567d4","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-29 11:42:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4241e9d751adca39a6b62314fcbacabb566a9053","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 11:46:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff9ef17b90844a3dad656002c1039163\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/71ab0f851fc64849a361b94b90a4ceaa : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e59a731a1d5f4405967c0f87db29e4fe : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/23710bf7d67442c5bf4a5bebf49c7285 : SUCCESS in 5m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1a4e82578aa56b7c66372e8147e0bf7adad7a298","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 13:42:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57b70996f1484b94ba65bee14f471944\n\n- grenade https://zuul.opendev.org/t/openstack/build/63bed66846e1436e977a8be5a75541d0 : SUCCESS in 1h 07m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a3a686552be548d9b7fb79baef1e5f5e : SUCCESS in 1h 01m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c636f34b131a47768c42b54feccd6366 : SUCCESS in 1h 36m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3a8dd9bdd8649b4912fe41550dff461 : SUCCESS in 59m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/468e1da3c9c7452792a80c457e116a64 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94a31dcffee94bca8a3250840b64b38d : SUCCESS in 6m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/24561bd4c5ed432f9dfecd6c2dd525e4 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69b6d29205dd4c68a8332766f93581e4 : SUCCESS in 7m 24s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f3be24c58ad847f8bb25605cdc97cfb8 : FAILURE in 12m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/387d9f8bb62d4ad3b697ad286ae21803 : SUCCESS in 9m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7a1ad7990b5545e48328aaed075bf3e9 : SUCCESS in 10m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5fd74357370b48178d4ce2e63643fc60 : SUCCESS in 10m 23s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f7f96cb64048425e8069f35e8777f53a : SUCCESS in 12m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d351447d48b54e5a8cde611be6dbadbc : SUCCESS in 1h 48m 16s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5377dc8933264ac386d311636bfb4384 : SUCCESS in 1h 27m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c1f67c6ae27e4a12a296d17c36e36ee6 : SUCCESS in 1h 40m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/90de3fa373064f1caecfd132fec28e71 : SUCCESS in 1h 41m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/27dba1bb0bb3422eb5cd0b52522ba479 : SUCCESS in 2h 02m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33e1df10336e44338173aad23fda8bac : SUCCESS in 58m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/af8a0ad02b32419181fd995b0df7918d : SUCCESS in 1h 50m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c61d760a9d2842deb7ef39b046a1e063 : SUCCESS in 28m 01s","accounts_in_message":[],"_revision_number":2},{"id":"28a7a3d72af1d7e5eb5c9911da1a0185f2f94015","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-29 14:53:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e64d8eeac2d3c46f97b90fada12eecad9bd7152d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-03-07 07:03:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"57b74b66d0604b77362c7d4686c3560444bda9b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 07:09:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c36c157b7a14b6198dacc578dc10375\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/621f4a57fece46ed8014218bd55eb71c : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f2d3c363264744bc986e1e591a0f7af7 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f90a9eaf9d71475f91ec8c7650a9051f : SUCCESS in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e89002805d905e7f46eb912da1f34b7d0c0aa5db","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-03-07 07:25:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"dd3750b1b33dc6f93052b9e8163cec14274eb756","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-03-07 07:27:05.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0441c7a8cdc86b552e48915b98a4065482426d1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 09:09:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59763f8610194ac8bbbaee30654432df\n\n- grenade https://zuul.opendev.org/t/openstack/build/e9f90f51c0e546b79597f6b7cb6a0ad9 : SUCCESS in 48m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0838ac82ae334fac9b878cdb53f925c1 : SUCCESS in 46m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/afc4deacd56e400ba73784508dcbf021 : SUCCESS in 1h 21m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59434085ba3b47ef822aa8bbc0e643f3 : SUCCESS in 48m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3e1318960c04d198f727081e0c90cb0 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84bf3b039a3642b78c1ed0fe87eb2ac2 : SUCCESS in 8m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7d6164a5e8a5480b993f28b06917310d : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c05f42dad9634e98a8bd69b151ee38c0 : SUCCESS in 6m 04s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ffcc0eb8655247de84785d8ebca9912f : FAILURE in 13m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/18e8773012d1484889e9f6b038190769 : SUCCESS in 8m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7f69b8a2b67b4b0a954919fb36b21980 : SUCCESS in 8m 25s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/bd9574282c234c2884a6a0e513333897 : SUCCESS in 8m 14s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5f23f2367d454df08c0d4160eb7bb111 : SUCCESS in 11m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cc210d2bf1a24cc6a2cf70a7e7b8effc : SUCCESS in 1h 59m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8f3019b931364e29a9cc023396e411a9 : SUCCESS in 1h 11m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c815c81634854ac186596988d6f25882 : SUCCESS in 1h 27m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c57791cb952d4f11aec8991815a76cce : SUCCESS in 1h 40m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7a057c4b983a4f85b3d2f7c6cfdae254 : SUCCESS in 1h 40m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fe872b531e940cdaf0de84a550ea8f9 : SUCCESS in 52m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ff42553ff35741af960e9368a4f02f95 : SUCCESS in 1h 43m 33s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/6e5e2db927924e04b722c3582798ba75 : FAILURE in 1h 38m 00s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2b0cf4ad760c4fa8a0deb7b333fd5319 : SUCCESS in 56m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3e1f89e6700548b8adfe401abf2a7a43 : SUCCESS in 36m 27s","accounts_in_message":[],"_revision_number":3},{"id":"a72201e634a706ed758c10cd81c6642663447a71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 09:09:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fe5f89a58b7bef7b7eae0bca32e3d76cc876a4e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 11:03:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a90f5bf24cc4d1d869efa59c793b9ab\n\n- grenade https://zuul.opendev.org/t/openstack/build/c3006edb1316483b9cc29d33dc50e8cc : SUCCESS in 55m 01s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d9055b4a7cdf42b08874c9e1ab3b40d6 : SUCCESS in 48m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/32cfb326cb134c6c98c6541a91ce3311 : SUCCESS in 1h 24m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/534bf398736144baa36a8de14062b465 : SUCCESS in 57m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/926a05b0f0f04126b314ed4e43bc0cd8 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b7f3f79a8f943f0b96892d8bd475170 : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3d4664493304d109de5fc69021a34dc : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26d094f69c744980b09a3eb84dd21c66 : SUCCESS in 6m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f66f8d0fe3514069a6e86ef9c307d004 : SUCCESS in 9m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a9d2998fdda149cfb1f9040077419fc2 : SUCCESS in 8m 08s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c8e43a832307420cbd66bdbd91f78fdd : SUCCESS in 1h 35m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/06484552a10d4338bdfe18a25d4ef192 : SUCCESS in 1h 42m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/07a99d7af14e490fa3c92be258419315 : SUCCESS in 1h 45m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/795957e422c346c0b097bd7b961ea2af : SUCCESS in 1h 00m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7e0ce1bd50e943748f9266d551d65992 : SUCCESS in 1h 52m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9630a25e503e48e9bc10c0ec0af370ab : SUCCESS in 32m 02s","accounts_in_message":[],"_revision_number":3},{"id":"e959e6e950ce7935ddade1e280ae0c4d97cb3edf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 11:03:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"45956a2e17ffc5a029590172123532257bdf6ed9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 11:04:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e877451caa0643e68f85e92640d45dc9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db5627a949ab4e67983de570d1c0326f : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"752aab74b1780720bc4162025ecf65f514f45835","revisions":{"67d2828caaf0b3eae8042af4e8e45d407c28eafa":{"kind":"REWORK","_number":1,"created":"2024-02-26 10:33:27.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/03/910203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/910203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/910203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/910203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/910203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/910203/1"}}},"commit":{"parents":[{"commit":"7ce9dd534fdad43d0682b7ebb7decaae394d9a37","subject":"Release notes for Caracal Milestone 3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ce9dd534fdad43d0682b7ebb7decaae394d9a37"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-26 10:29:44.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-26 10:29:44.000000000","tz":0},"subject":"Refresh Glance example configs for caracal m3","message":"Refresh Glance example configs for caracal m3\n\nChange-Id: I7490eeaa7352d38fba95c40bbb17bd970887be8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/67d2828caaf0b3eae8042af4e8e45d407c28eafa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/67d2828caaf0b3eae8042af4e8e45d407c28eafa"}]},"branch":"refs/heads/master"},"627f13d8c5e9de7eecfe1bdaad84070842666526":{"kind":"REWORK","_number":2,"created":"2024-02-29 11:36:06.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/03/910203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/910203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/910203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/910203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/910203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/910203/2"}}},"commit":{"parents":[{"commit":"689e90c238f70fdcc4b9434d21fa7f7f93d29657","subject":"Release notes for Caracal Milestone 3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/689e90c238f70fdcc4b9434d21fa7f7f93d29657"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-26 10:29:44.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-29 11:32:17.000000000","tz":0},"subject":"Refresh Glance example configs for caracal m3","message":"Refresh Glance example configs for caracal m3\n\nChange-Id: I7490eeaa7352d38fba95c40bbb17bd970887be8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/627f13d8c5e9de7eecfe1bdaad84070842666526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/627f13d8c5e9de7eecfe1bdaad84070842666526"}]},"branch":"refs/heads/master"},"752aab74b1780720bc4162025ecf65f514f45835":{"kind":"REWORK","_number":3,"created":"2024-03-07 07:03:50.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/03/910203/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/910203/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/910203/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/910203/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/910203/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/910203/3"}}},"commit":{"parents":[{"commit":"4bfd9322666f5e8d499032f0e4f8df91f470e9f5","subject":"Release notes for Caracal Milestone 3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4bfd9322666f5e8d499032f0e4f8df91f470e9f5"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-26 10:29:44.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-03-07 07:01:26.000000000","tz":0},"subject":"Refresh Glance example configs for caracal m3","message":"Refresh Glance example configs for caracal m3\n\nChange-Id: I7490eeaa7352d38fba95c40bbb17bd970887be8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/752aab74b1780720bc4162025ecf65f514f45835"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/752aab74b1780720bc4162025ecf65f514f45835"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
