)]}'
{"id":"openstack%2Fglance~910377","triplet_id":"openstack%2Fglance~master~I15b19dbbc77e211dff78bb9ba5e7c7210b08a529","project":"openstack/glance","branch":"master","topic":"centralized-cache-db-4.1","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-03-01 15:21:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I15b19dbbc77e211dff78bb9ba5e7c7210b08a529","subject":"[minor] Fix doc string for class","status":"MERGED","created":"2024-02-27 17:49:29.000000000","updated":"2024-03-01 15:22:54.000000000","submitted":"2024-03-01 15:21:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910377-centralized-cache-db-4.1","meta_rev_id":"ee72eb8352d9857c5354f724e7a5d47ce8cd07ef","_number":910377,"virtual_id_number":910377,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-01 15:21:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2024-02-27 17:55:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2024-03-01 13:31:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-27 17:55:28.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-02-27 17:58:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-27 20:17:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0fd5aa3e29b98d960b5a7b6e5181577b5f4d515","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-27 17:49:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cd54d7f414531712d5b637546c7c22dbab87cbf","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-27 17:55:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"974a9edc4786a733ad0ed6bcbf06a9ae45496957","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 17:58:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f99271d6bea04ba988cc5b9f8b14f51c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c539c87326eb4130b2741539d3cbe3c9 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3e5c4b7e9cda4deca4477ad148d5d1d9 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ae2c6c2b542f4affb02a8b744cb80f97 : SUCCESS in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f015b53d11f048e30c0d0b801c2207da295a9553","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 20:17:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8dc9b90ec9bd40e0a6df8813023f0553\n\n- grenade https://zuul.opendev.org/t/openstack/build/82c6c907173644e9b11b2c64a6d06bb4 : FAILURE in 55m 00s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/85461233690e4bb288cad2bd39f25386 : SUCCESS in 54m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/561e259fc33843008590973efdcb7b82 : FAILURE in 2h 00m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82836f9be4ae4293a96861cdb07680b4 : SUCCESS in 48m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aad49b0a1c554251a9bf1bf4ef0aa8b3 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41e52099781a4773bec0e3d35b5d92c0 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e80f7ae886d490cb96d9b13f1e94995 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a7e60fe930b44d98c81b27bd20f7fdb : SUCCESS in 5m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8379f0dd5a2146d5a48e5ef8e71b747b : SUCCESS in 13m 32s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7fe3b279045843c0b226482976b0a4b3 : SUCCESS in 8m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0f807bbc6d204c23b65c3ef02983d079 : SUCCESS in 9m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/04486268cea0441f9fc407fb8d751e0b : SUCCESS in 7m 59s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/cd4c0e586fce4a559790acf577731ddc : SUCCESS in 13m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/12b6e8c7aa8a41c897b860792ee4d12e : SUCCESS in 1h 54m 12s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/71514e6dce5e4573bb4c665cdbaf58d5 : SUCCESS in 1h 54m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4766c17003464b3ebde49bb34431c3dd : SUCCESS in 1h 44m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/22879a8c7bff4ad6b9c6359b49efd60c : TIMED_OUT in 2h 07m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4dec16a4f7554b0a842cfb7843bbac77 : FAILURE in 2h 19m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5626512951544c55a2a786c8f422551e : SUCCESS in 59m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3003afb08d8f46ea80d06c739befa75a : SUCCESS in 1h 33m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e6f29a5eff0e4aaf9b9b86c632d1c8ad : SUCCESS in 38m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6c080954f9ee99c685c8f2ca102f3e892d39c81a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-28 04:23:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"019bd6ae5a0ce279e8c5ff8e205d4918501c39ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 04:33:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e94e4b4ec7d4c088b903db5693eac7c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f4f458297714a2dafc116f330157c76 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/236d36f36baf4849964c4bcb7d3e12a4 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b71ffa4696a74003b6472b44d1115249 : SUCCESS in 8m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"667780adda5bd40606a5f556e5fd8a782d01b39f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 06:38:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9942a70cfc5d4e8c9e9d1cf6bc23eaa6\n\n- grenade https://zuul.opendev.org/t/openstack/build/73adce5ed421429ba4a9730326a962ba : SUCCESS in 49m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/926318711b7a40fe8df823cc11684b8c : SUCCESS in 59m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d67c9da401f945d2a93bae2cb603fff3 : SUCCESS in 1h 23m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aee05cd61e634c13bfc3c32718af785b : SUCCESS in 50m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eddc77a65d4e4df6b00ce7ca6aff2b5c : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54d555f2607d43ba9b97da3e8718772c : SUCCESS in 6m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b14ab7d50604733b51840f926de0ab5 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41adaa931e2249e6bbd47ed8de22fc58 : SUCCESS in 5m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/606760d7c778461eafa2b2a98c230d1e : SUCCESS in 12m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c341e7daa0804f91aca764e72b8a39e6 : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/28b306b25718494da2595257f32b2514 : SUCCESS in 8m 41s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8b6dc7f03beb400e8589b587ea0cfa6e : SUCCESS in 10m 56s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/557455e30de64687b0f5bd00af08eb2f : SUCCESS in 10m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6531d83102b149df89c783909943d3e0 : TIMED_OUT in 2h 03m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9dca6a6121c241a0ba4fccc999370950 : SUCCESS in 1h 49m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/65a19163f5b84273a239929ebd2a656e : SUCCESS in 1h 22m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/37642774bbc946ff93369f4a49fd1cff : SUCCESS in 1h 48m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c182f32fda5347ca9fda21dc50b3013c : SUCCESS in 2h 01m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a11bf9573150426fb498b72475bee42b : SUCCESS in 56m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e40a51e0ee45447fa7683ff46c7e897b : SUCCESS in 1h 39m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4b6dd12694d7487b92cfe37a70c10487 : SUCCESS in 42m 25s","accounts_in_message":[],"_revision_number":1},{"id":"67bf940fd7c766ca20dec7fe4773049b4accd20f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-03-01 13:31:11.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"103e519d345d08f815f50efd6de35022d38c57e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 13:31:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e5b95bad7586ad53e163e1a77f9f1ef66066cb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 15:21:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b76cfbd634464f55b06d0cd973f07bbe\n\n- grenade https://zuul.opendev.org/t/openstack/build/69d39bf9107641c68d0d7de103aa6881 : SUCCESS in 1h 12m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ec42e69cfec947e2b346a567ac7fe502 : SUCCESS in 1h 04m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b07d0e56104c4e35a7d6edea3a94e425 : SUCCESS in 1h 19m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30a3e16b8ae046959f17f522b7784bc6 : SUCCESS in 59m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c1128c8b59e42a3aff3ba221bd5cc68 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76521c490dd1408a8c0e35f6351bfebb : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/714be8b2e30c4a3f903338b263ea046f : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67fd5afe73a04a4980210575b2940fcb : SUCCESS in 5m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dc3ae65a2b7b4ada96725728a88c4563 : SUCCESS in 10m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7eb792b95e6a4ceb9198b8e9e9ad7a9c : SUCCESS in 8m 35s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8468f701384d4fe69b48564e10885207 : SUCCESS in 1h 45m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f4637015827346ff9490dd27eaae3455 : SUCCESS in 1h 30m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4a265815a21b4c35aa64a5f4726b0be4 : SUCCESS in 1h 43m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95864f04bf194982829b3b6c10d6ab39 : SUCCESS in 52m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c3925369699c4d719e3c530d574a67c7 : SUCCESS in 1h 45m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bd41d60df63f4ae991ea18d7a85bfa6d : SUCCESS in 31m 07s","accounts_in_message":[],"_revision_number":1},{"id":"9c046df9bd7fa0a58d6e49d76b177de1cf02cb0b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 15:21:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ee72eb8352d9857c5354f724e7a5d47ce8cd07ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 15:22:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/187f1f13ab734ad880339673e4db6ec6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1509197776254b3eb4601dcb66e4856e : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"591f479442f9c4470bdf19cf8dfd0fb51e564c58","revisions":{"591f479442f9c4470bdf19cf8dfd0fb51e564c58":{"kind":"REWORK","_number":1,"created":"2024-02-27 17:49:29.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/77/910377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/910377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/910377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/910377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/910377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/910377/1"}}},"commit":{"parents":[{"commit":"0bbaec2eeb489b1f362e066749c3f71d05e02602","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bbaec2eeb489b1f362e066749c3f71d05e02602"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-27 17:28:48.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-27 17:28:48.000000000","tz":0},"subject":"[minor] Fix doc string for class","message":"[minor] Fix doc string for class\n\nChange-Id: I15b19dbbc77e211dff78bb9ba5e7c7210b08a529\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/591f479442f9c4470bdf19cf8dfd0fb51e564c58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/591f479442f9c4470bdf19cf8dfd0fb51e564c58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
