)]}'
{"id":"openstack%2Fglance~910640","triplet_id":"openstack%2Fglance~master~I321982013bdf571741a8ef2d0575f65288fff53f","project":"openstack/glance","branch":"master","topic":"fix-flaky-test","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-03-01 00:42:11.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-02-29 21:48:14.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-03-01 00:42:11.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-02-29 21:55:35.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-03-01 00:42:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I321982013bdf571741a8ef2d0575f65288fff53f","subject":"Fix flaky test related to cache migration","status":"MERGED","created":"2024-02-29 15:36:44.000000000","updated":"2024-03-01 00:43:22.000000000","submitted":"2024-03-01 00:42:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910640-fix-flaky-test","meta_rev_id":"cfef500443502f44ea28bbea6a916305b5ccb03d","_number":910640,"virtual_id_number":910640,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-01 00:42:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-02-29 21:48:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-02-29 20:39:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-02-29 21:55:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-29 15:37:29.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2024-02-29 15:37:29.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-02-29 15:37:29.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2024-02-29 15:37:29.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-02-29 15:55:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-29 16:51:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab4d57f25b5db54f3b3395dbab156c383ab46a5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-29 15:36:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e40a7105e5d751f72075a2d076ebb9b552fe170","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-29 15:39:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e5c5949b3128e9a9a1e960af46ee4596934b5d0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-29 15:43:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"198f6ecb8018aea906500fe34979383afcdc0746","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-29 15:43:18.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b3ac7d08440a207f04ef644449da89562b01849a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-29 15:44:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f9104fec938535e8c0ecd0467a94c7f8595e05df","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-29 15:50:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f93f92926e923b7660c569ea6ff5b75ba6de5ffa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 15:55:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/200e1c43decb4a0296ccbae1d925e8a1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/60fcb1cddb1b41159ad41126a86e0e91 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a97f2f64404447f5b4285290a606c1d0 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6ef8cf9733754191b1299b7d8c7a92ce : SUCCESS in 9m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf7ee90bda956cfb13411a1d709e8e7cb61afbea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 16:51:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cfb1912a47a4c088218039eca592254\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0fed91b806f4461c92f750ccb7973628 : SUCCESS in 1h 03m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd83ebea26674998a62eda556f6d7b35 : SUCCESS in 1h 01m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f73168b601c49ed8949d388f6f0345f : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34abcded595e4098be4c50b8b755e234 : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ebe71b71af54dc8b1a5ddab90f5ee6c : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5ede36e8da44d3680c725aa288f2230 : SUCCESS in 6m 53s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9ab3095659cb45cbabaef12c8af2c20b : FAILURE in 12m 15s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9503fac3fd9944078ee6e04c7fefdbef : SUCCESS in 9m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/07b0dfb4411b4cc187911079ef0730ae : SUCCESS in 9m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0649b0bd6efd45fca84148c5a556d855 : SUCCESS in 11m 17s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b53aa442e2fe40fe8b380d5499d85a15 : SUCCESS in 11m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5ff6af7cc7ae4d7ea8c43b76f5b6ce23 : SUCCESS in 32m 19s","accounts_in_message":[],"_revision_number":2},{"id":"b60b41d98972f974c99ac4f06d52cd9f70185f7f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-29 20:21:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"db757adb74e33e4d5f12b2cb57fc85b51811e263","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-29 20:28:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8605cde892617d9700587d424b396e32d8f4b5c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-29 20:35:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"bdebf10dedfa6cbf24197129932ea0f5ccee70c9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-29 20:39:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"691efa9fbc75d00ff7d4ab2bf079693feb96ad58","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-29 20:39:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"18e8a6e468691d66b172edcb760316cc1a8a9a1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 20:41:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1425597f6ddb4db4b950124df2611ade\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/082ac948046a4968b62134fcdd72c24f : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/387681898b7a4591a9968d32749b4f38 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9daec0d6b43541f6a349e3b79534532f : SUCCESS in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"be1b637ff190ab458b3daf97ee3241f399143221","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-29 21:48:14.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8ad54b4edae4d6670a0207586e390eb373387f1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 21:50:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13d668825b044ea995a6a1b160a82c38\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9e05c95273b1400d9379cd7051937017 : SUCCESS in 52m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea999fe11d8c440da6cc46b8607bd75c : SUCCESS in 1h 09m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9225ddbd03484acfa6081becdbee8b38 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0450dc230cb44d189763109fb20b3504 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cfe7731958dd450caa80c05925af6963 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccff4dc395ed4dcebd6592bc8009eed4 : SUCCESS in 5m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a8f8952f205f49109d38dba60cccaa76 : FAILURE in 13m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/187ec4445562440fafaad671c26862e4 : SUCCESS in 8m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3035fa8de3e84200852f07e6ad6b571c : SUCCESS in 7m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1d96a55f0cac4f6893df5b34d8205804 : SUCCESS in 10m 36s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/475b7856a2e94e5ca88a5f0c362f233a : SUCCESS in 10m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/322a8b8721ca4a06b54baf36e6bdc515 : SUCCESS in 29m 10s","accounts_in_message":[],"_revision_number":3},{"id":"92ee71de69c3db09a0b9c4001d43b25eae953c30","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-29 21:55:35.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"029a9c4d5376c033a8b09533f7f2658c3ac1de7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 21:56:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7b6ef03e14025cf2ce552aaaeaa36b876aab5c0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 00:42:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/646e1d96b4404c00b88f2ede1864fec5\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/367b47f9da72463f966c6e40ee5ebbba : SUCCESS in 56m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7477061cd00e48409f063e3156591c66 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37ef8b4b6c3c42ba9ae46c438f4edca9 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e97436176edb41d6bd151683ddfc0b8e : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0ea843bcd3c4f4a93991668e2308c64 : SUCCESS in 5m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9e8ec6370969421aa97a1884f1956eba : SUCCESS in 8m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/33a32b46db5c49989e86357fccd8a524 : SUCCESS in 8m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7ceabe4456f547c0b7179936835bf8d4 : SUCCESS in 23m 29s","accounts_in_message":[],"_revision_number":3},{"id":"17b388aa1121172968d7c99bf68836ec1c8692ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 00:42:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"cfef500443502f44ea28bbea6a916305b5ccb03d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 00:43:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/010b5e2c9b464c0087d3b749ef14980d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4783b59101d9463c806d8bdcabd33fc0 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5bf2f4e7a073ed1c046f541845680484d93d9036","revisions":{"17e2cc70ab629f5aa869eb16a4d0e96f1740bbbf":{"kind":"REWORK","_number":1,"created":"2024-02-29 15:36:44.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/40/910640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/40/910640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/910640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/910640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/910640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/40/910640/1"}}},"commit":{"parents":[{"commit":"703cf5ffddc1dcc9a7086b8dff220cf5627d308b","subject":"Merge \"[docs] Configure centralized_db cache driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/703cf5ffddc1dcc9a7086b8dff220cf5627d308b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-29 14:56:16.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-29 14:56:16.000000000","tz":0},"subject":"Fix flaky test related to cache migration","message":"Fix flaky test related to cache migration\n\nIn cache migration unit tests, I was using same cache.db file name\nin the setup method which was used by all the tests, which was\ncausing a issue of deleting a file by other test and resulted\nin random failure.\n\nUsing random filename to fix this issue permenantly.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I321982013bdf571741a8ef2d0575f65288fff53f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17e2cc70ab629f5aa869eb16a4d0e96f1740bbbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17e2cc70ab629f5aa869eb16a4d0e96f1740bbbf"}]},"branch":"refs/heads/master"},"4c4c2b2913a05ebbe0d83050a105bdeac9f44cc0":{"kind":"REWORK","_number":2,"created":"2024-02-29 15:43:04.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/40/910640/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/40/910640/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/910640/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/910640/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/910640/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/40/910640/2"}}},"commit":{"parents":[{"commit":"703cf5ffddc1dcc9a7086b8dff220cf5627d308b","subject":"Merge \"[docs] Configure centralized_db cache driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/703cf5ffddc1dcc9a7086b8dff220cf5627d308b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-29 14:56:16.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-29 15:22:43.000000000","tz":0},"subject":"Fix flaky test related to cache migration","message":"Fix flaky test related to cache migration\n\nIn cache migration unit tests, I was using same cache.db file name\nin the setup method which was used by all the tests, which was\ncausing a issue of deleting a file by other test and resulted\nin random failure.\n\nUsing random filename to fix this issue permenantly.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I321982013bdf571741a8ef2d0575f65288fff53f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4c4c2b2913a05ebbe0d83050a105bdeac9f44cc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4c4c2b2913a05ebbe0d83050a105bdeac9f44cc0"}]},"branch":"refs/heads/master"},"5bf2f4e7a073ed1c046f541845680484d93d9036":{"kind":"REWORK","_number":3,"created":"2024-02-29 20:35:08.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/40/910640/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/40/910640/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/910640/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/910640/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/910640/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/40/910640/3"}}},"commit":{"parents":[{"commit":"703cf5ffddc1dcc9a7086b8dff220cf5627d308b","subject":"Merge \"[docs] Configure centralized_db cache driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/703cf5ffddc1dcc9a7086b8dff220cf5627d308b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-29 14:56:16.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-02-29 20:35:00.000000000","tz":-480},"subject":"Fix flaky test related to cache migration","message":"Fix flaky test related to cache migration\n\nIn cache migration unit tests, I was using same cache.db file name\nin the setup method which was used by all the tests, which was\ncausing a issue of deleting a file by other test and resulted\nin random failure.\n\nUsing random filename to fix this issue permenantly.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I321982013bdf571741a8ef2d0575f65288fff53f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5bf2f4e7a073ed1c046f541845680484d93d9036"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5bf2f4e7a073ed1c046f541845680484d93d9036"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
