)]}'
{"id":"openstack%2Fglance~910666","triplet_id":"openstack%2Fglance~master~I2878b4162dddaf935002c168bc544eb1eff71efd","project":"openstack/glance","branch":"master","topic":"lock-path","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-03-01 06:50:01.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-03-01 04:47:44.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"I2878b4162dddaf935002c168bc544eb1eff71efd","subject":"Set a lock_path for tests","status":"MERGED","created":"2024-02-29 21:46:55.000000000","updated":"2024-03-01 13:38:14.000000000","submitted":"2024-03-01 06:50:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"910666-lock-path","meta_rev_id":"2f00fe6122cd0b32834715de3077e1b260043b73","_number":910666,"virtual_id_number":910666,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-01 06:50:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-03-01 04:47:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-02-29 21:49:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-03-01 04:47:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-29 21:49:15.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-02-29 21:49:18.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2024-02-29 21:57:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-29 22:35:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-01 01:03:48.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"d0f8f9901a1e91dc822cd5376ef9dd7a112d0718","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-29 21:46:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"409f213dc4103d372d262fef8c892a861757a4d9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-29 21:49:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"574d16ab556a597ed955bf6889e8ee262b542f45","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-29 21:49:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a006b4f8e5a67f7f99cbfc2be0e732e136af634f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 21:57:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85206d521fac42c0b409e8d10c0b09ff\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f52406ddcd945c1aa3d81bb51d2bfb1 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4c263c0bab8b4545a8a149ae54f1d9cd : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/88e7f1b1e5cd474f83ebc1194a1c99e9 : SUCCESS in 7m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6400afe1a15b5180109ee16ded621649c08618f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 22:35:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8fee0057a2548f2a71763beb7ad42aa\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/da356578115f4a0b813b1faab85925cc : SUCCESS in 45m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89233363bc0e4bc1b939c9eaa1a342c3 : SUCCESS in 46m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b21d7e251a6640219fb1e523f5401154 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/112c94188b8f42c4973fb8371cec10d7 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e4edcd08f9a4f7980921a82b41832df : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5785b04f03244798bda383a7613a63c7 : SUCCESS in 5m 04s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5e196d96b9b145f1a9068ef51f0fd9b1 : FAILURE in 12m 39s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0ec506e94766435fa6e2ff4dcb537a31 : SUCCESS in 8m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bc001bff4a5646fda381b76f99d9ee4a : SUCCESS in 8m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fc92f1f693df4041b17b43bf673aa79c : SUCCESS in 9m 04s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/e90d2a7b3c10400589c414c16503c8a5 : SUCCESS in 10m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5efebf2625334305871bbee83d027aaa : SUCCESS in 39m 40s","accounts_in_message":[],"_revision_number":1},{"id":"221579a72e53873270863a1bf45776d4f45082e3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-03-01 01:03:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4303790294f633e8cc9e198eecaf812798ee854","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-03-01 04:47:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8bcfe0e893198b1f1ddf13f20e3e79a5b52aebfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 04:48:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b9ab9e452acc3be68a8fc269d349d3d8fdabb51","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-03-01 05:50:49.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"89cd225116821848659da66228881da7cb85a73e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 06:50:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c86ee8da68004311be31eaced9c3cea7\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e3127be2b4e472ab95c180ab6ff934b : SUCCESS in 54m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d011a39a2aa4b689ee1e4a442d52a2b : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5b33c025f1e447ba5ffdc4fb3c9368a : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8904a290d9454e4ab3146beb088a42c8 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f879aba5c464254ba0cc3bc362dc7c2 : SUCCESS in 8m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1e9a226837cc458f8daee2540eab85b6 : SUCCESS in 8m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4df52405893041999ce67dda424c5cc7 : SUCCESS in 7m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/60c96802ccce4267afccb7e1fa377b5c : SUCCESS in 38m 09s","accounts_in_message":[],"_revision_number":1},{"id":"7ebdfaad9615505484fef110bcd8dcac64ac2ef6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 06:50:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ff776f0637cad363da4f5e86a812470df3169142","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 06:51:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0776ecd81c554ae3b0570aaabfaeb0c2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a5427748dc14e7b9ff7d85ed27fbf77 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1},{"id":"2f00fe6122cd0b32834715de3077e1b260043b73","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-03-01 13:38:14.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9df586a5fe2fd02b4f7497fc9e73b45c4acaf25d","revisions":{"9df586a5fe2fd02b4f7497fc9e73b45c4acaf25d":{"kind":"REWORK","_number":1,"created":"2024-02-29 21:46:55.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/66/910666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/910666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/910666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/910666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/910666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/910666/1"}}},"commit":{"parents":[{"commit":"5bf2f4e7a073ed1c046f541845680484d93d9036","subject":"Fix flaky test related to cache migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5bf2f4e7a073ed1c046f541845680484d93d9036"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-02-29 21:37:17.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-02-29 21:44:40.000000000","tz":-300},"subject":"Set a lock_path for tests","message":"Set a lock_path for tests\n\noslo.concurrency 6.0.0 has relocated the lock_path that is required\nfor external locks from [DEFAULT] to the [oslo_concurrency] config\nsection.  Instead of messing with configuration, we can use the\nlockutils.set_defaults() method to set this for tests.  So add this\nto the BaseTestCase so that it\u0027s available for all tests.\n\nChange-Id: I2878b4162dddaf935002c168bc544eb1eff71efd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9df586a5fe2fd02b4f7497fc9e73b45c4acaf25d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9df586a5fe2fd02b4f7497fc9e73b45c4acaf25d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
