)]}'
{"id":"openstack%2Fglance~912504","triplet_id":"openstack%2Fglance~master~I03776591f225348ffda5204eddbce24b1c465753","project":"openstack/glance","branch":"master","topic":"remove_unreachable_code","attention_set":{},"removed_from_attention_set":{"36609":{"account":{"_account_id":36609,"name":"Youngjun Yoo","display_name":"Youngjun","email":"yj.yoo@okestro.com","username":"Youngjun"},"last_update":"2024-09-04 04:40:41.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-09-04 02:17:33.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-09-04 04:40:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I03776591f225348ffda5204eddbce24b1c465753","subject":"Remove unrecheable code","status":"MERGED","created":"2024-03-12 04:49:57.000000000","updated":"2024-09-04 04:41:50.000000000","submitted":"2024-09-04 04:40:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"912504-remove_unreachable_code","meta_rev_id":"9db431e6c504f35c7fbebd415816d2bf7514615e","_number":912504,"virtual_id_number":912504,"owner":{"_account_id":36609,"name":"Youngjun Yoo","display_name":"Youngjun","email":"yj.yoo@okestro.com","username":"Youngjun"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-04 04:40:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-09-04 02:17:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-09-03 14:13:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-09-04 02:17:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-12 05:00:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-12 07:08:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-30 11:44:31.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2024-04-30 11:48:35.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-09-03 14:13:56.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"1074cd18d955e8713b61d31e5f6f710305e46465","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36609,"name":"Youngjun Yoo","display_name":"Youngjun","email":"yj.yoo@okestro.com","username":"Youngjun"},"date":"2024-03-12 04:49:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4403f23c18536da7d577b7c742e7cefe89be81c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 05:00:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dcf2cb5181d4b1e8b95ae3fd5227815\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e97bd57d981498e86d53e26ab56847f : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c83ca9647b8143159e5032c79a880e7b : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cb5ebfe8eaf146c4b69890a8c395ab08 : SUCCESS in 9m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8290a8c71d4f62040a254de924e9d2b879f921b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 07:08:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fc6a47f908c422d8f2dcfcdebe0c782\n\n- grenade https://zuul.opendev.org/t/openstack/build/e8142209ddeb4cfa988c84fe0c282289 : SUCCESS in 57m 22s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d041481898294f3da117014abe1d9da5 : SUCCESS in 59m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a2f53c41f22f49f09ed803959a0f247a : SUCCESS in 1h 31m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2faa5016d4c5467b8ba8f88873f771f7 : SUCCESS in 1h 06m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed87769d585c45ffb00ea6678f8aba3f : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eeb6caa285524330bc2cf3b6e6fff0cc : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8a4e597451284d5d84159cba747f466c : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98b1638b373a4ba5bc4920970017c52a : SUCCESS in 5m 33s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/fb1cb0881d6d4adaba3dd0f370eda924 : FAILURE in 12m 59s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/306c626f6c6644229c06e797c6a6573b : SUCCESS in 9m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b3cd5c1be7734ae39c40d8e6159020e2 : SUCCESS in 8m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0f6ed34bcd4e482a8a602ccad6f66dd6 : SUCCESS in 10m 40s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/3e3dc8e4f7934747a2666760b960c313 : SUCCESS in 10m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8e341701cc3041898b7956d47660c31f : FAILURE in 2h 00m 58s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0e91196f7d6546eb978cbb2e53423d67 : SUCCESS in 1h 34m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f4a227009b564151bd7e532d5737e160 : SUCCESS in 1h 46m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/405b7e7fb0c444bc96c018247e89524b : SUCCESS in 1h 59m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/22bce7923f364a16a84113f57b5d5d09 : SUCCESS in 2h 00m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/173637da81ae4e1dbba0a3b2a7ec39ce : SUCCESS in 1h 02m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/801ab4282e3c404fa44025250890aaaa : SUCCESS in 1h 44m 12s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/4a584c796fd94f568881343223a5b58c : SUCCESS in 1h 51m 38s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a76350c7857346f8ad324e3a46b65117 : SUCCESS in 1h 08m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/75849ec776074792a7531660f28b5c5f : SUCCESS in 37m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d5dd1d9b63cb8e69b9048234d03f28428e140098","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-04-30 11:44:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f4a664c410c1d250c13a45e16ea19afeb360f80","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-04-30 11:48:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f664c334616644a599643c69e944d77babb2c19d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-09-03 14:10:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a6867096364972e25b3c208fc4878cdbfae6a8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-09-03 14:13:32.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"dbb0dd70195033ea519b1f6f54d2a5333763f0f6","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-09-03 14:13:56.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7723230f913d26cfcf9f35ecdbd62d4fbe65912","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 14:43:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/624b8ff8e0bf456ab15c8b7d2765056b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b7dbbe0b2c894f0bb988c542ceb4dbd8 : SUCCESS in 13m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0a200699f6de4a43ad8cdac37ac62ec1 : SUCCESS in 11m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9d0b013b40f3904e138ff2b69081855885f395c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 17:05:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89beb3cac25f49ada110b76845ebd9a3\n\n- grenade https://zuul.opendev.org/t/openstack/build/b819f6bfe64c4ec1914bd3db102e6d98 : SUCCESS in 1h 10m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e056fdfa4cb5486b9968767858442704 : SUCCESS in 1h 24m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2d047b6816249e491464571bba1f451 : SUCCESS in 57m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6db2488a6df545b4a57dbe877d017f79 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19e6ee4f09af4d199eabd83ef14c0e36 : SUCCESS in 6m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fbec7748c0234ca89519b9096cd00ebc : SUCCESS in 6m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0c1868adbdf4a5dbb3a619d54bd7f00 : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a6a7ded29144bb1bb78185156077f07 : SUCCESS in 7m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7bf16b8ebc4e4360b40ff80d049d6fe5 : SUCCESS in 9m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b7712839593a4c08b6dc7b174efbf308 : SUCCESS in 8m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/69dc035eec2840259c9ced8908995aa7 : SUCCESS in 13m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ebf02b4983cb44e9960f63c47df167a9 : SUCCESS in 8m 17s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2a59d9fff8f4486b8d44c8acf7e17d5d : FAILURE in 1h 44m 00s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9e3dacd719c14bdd9ec929a4a26f7823 : SUCCESS in 2h 17m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f763be59257845fea4f36d0b7689156c : SUCCESS in 2h 01m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bc03ced2ef284692b650e077be3fde66 : SUCCESS in 2h 03m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5762911390a94bc2a799a85bcd0908ca : SUCCESS in 2h 36m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a055852b82174e78800662d0b3730387 : SUCCESS in 53m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cb41764e065c4f698fa13f0a8d505616 : SUCCESS in 1h 34m 43s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/18051d57c59c4861ba5a137634afd893 : SUCCESS in 1h 49m 41s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/18ff325911d54cf28b949215d90a711e : SUCCESS in 1h 01m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2ba80490a68d4716af4e2d48552198c7 : SUCCESS in 45m 39s","accounts_in_message":[],"_revision_number":2},{"id":"22f0c2558c90647ef1f7214691946288ed909deb","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-09-04 02:17:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f6d2f909a80b494be1a17e8f9aab1d57c7f57d6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 02:18:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2be290a6151ef0c914dc191d6ad7dd4ff3579e5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 04:40:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4a2a10bf3944e56b68abe9082bed33e\n\n- grenade https://zuul.opendev.org/t/openstack/build/1fbbd066ef8a48d68c408a9e286ea778 : SUCCESS in 1h 01m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88329863a49e40e18a1fa62c388272ac : SUCCESS in 1h 31m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0ad53d0cb4548f5b3df955e7391fdcf : SUCCESS in 55m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e238a1b8ab945c1814c25718319d4ba : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c965fc4abadf48c0a97b375230c99ae5 : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e537ca90fc442de933e67dc92cdf30d : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0fb968a776d460b877a5084170c529e : SUCCESS in 6m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/55deabe0e76f4261a3537280f62debdc : SUCCESS in 10m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/51bf0eef7e594fa188c8edfa735d18f0 : SUCCESS in 8m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/386b5fdd9ecd44e0a437ab9409694bcf : SUCCESS in 11m 36s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f0226dfaaec84231982df709483e39cd : SUCCESS in 2h 11m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5e513ae6498e40db8cbb7815e1ca2e89 : SUCCESS in 1h 47m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f2e4fac16f3340e3966fd0b3697d7f05 : SUCCESS in 2h 02m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57367c746cd5471ea409edd6bcc6f9db : SUCCESS in 58m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7616d14f4f084fc4a3adffe2783bae9a : SUCCESS in 1h 59m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d545a3384e6e4ba8853d86994893a10f : SUCCESS in 37m 42s","accounts_in_message":[],"_revision_number":2},{"id":"76412280deee03f2729054090c8ecd97c3175fff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 04:40:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9db431e6c504f35c7fbebd415816d2bf7514615e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 04:41:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c869481b32b34ed89c7b25e8b312d620\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff999dced7614e64b84c67898483bb1a : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c4a0a94e60412cca18e7c3eeecd460f4c95afc22","revisions":{"13278bc6893153ffe5de54a5d96f57cd8668606d":{"kind":"REWORK","_number":1,"created":"2024-03-12 04:49:57.000000000","uploader":{"_account_id":36609,"name":"Youngjun Yoo","display_name":"Youngjun","email":"yj.yoo@okestro.com","username":"Youngjun"},"ref":"refs/changes/04/912504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/04/912504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/912504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/912504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/912504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/04/912504/1"}}},"commit":{"parents":[{"commit":"309ca3aec26b6dd49b8d955c4900a5c390d14537","subject":"Fix glance-api if cache is disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/309ca3aec26b6dd49b8d955c4900a5c390d14537"}]}],"author":{"name":"Youngjun","email":"yj.yoo@okestro.com","date":"2024-03-12 04:48:28.000000000","tz":540},"committer":{"name":"Youngjun","email":"yj.yoo@okestro.com","date":"2024-03-12 04:49:54.000000000","tz":540},"subject":"refectory: remove unrecheable code","message":"refectory: remove unrecheable code\n\nChange-Id: I03776591f225348ffda5204eddbce24b1c465753\nSigned-off-by: Youngjun \u003cyj.yoo@okestro.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13278bc6893153ffe5de54a5d96f57cd8668606d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13278bc6893153ffe5de54a5d96f57cd8668606d"}]},"branch":"refs/heads/master"},"c4a0a94e60412cca18e7c3eeecd460f4c95afc22":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-09-03 14:13:32.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/04/912504/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/04/912504/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/912504/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/912504/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/912504/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/04/912504/2"}}},"commit":{"parents":[{"commit":"309ca3aec26b6dd49b8d955c4900a5c390d14537","subject":"Fix glance-api if cache is disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/309ca3aec26b6dd49b8d955c4900a5c390d14537"}]}],"author":{"name":"Youngjun","email":"yj.yoo@okestro.com","date":"2024-03-12 04:48:28.000000000","tz":540},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-09-03 14:12:13.000000000","tz":120},"subject":"Remove unrecheable code","message":"Remove unrecheable code\n\nos.execlp() does not return if successful. From the documentation[1]:\n\n\"These functions all execute a new program, replacing the current\nprocess; they do not return\".\n\n[1] https://docs.python.org/3/library/os.html#os.execlp\n\nChange-Id: I03776591f225348ffda5204eddbce24b1c465753\nSigned-off-by: Youngjun \u003cyj.yoo@okestro.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c4a0a94e60412cca18e7c3eeecd460f4c95afc22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c4a0a94e60412cca18e7c3eeecd460f4c95afc22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
