)]}'
{"id":"openstack%2Fglance~915290","triplet_id":"openstack%2Fglance~master~I24b14812a2a15675beacee720d722c2f3a2b252b","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-04-30 03:07:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I24b14812a2a15675beacee720d722c2f3a2b252b","subject":"Cinder: Remove ambiguous warning","status":"MERGED","created":"2024-04-08 18:30:15.000000000","updated":"2024-04-30 03:09:00.000000000","submitted":"2024-04-30 03:07:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915290","meta_rev_id":"6e1cd88ac7839ee7c89a82e761929198ab3e96a3","_number":915290,"virtual_id_number":915290,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-30 03:07:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2024-04-09 09:04:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2024-04-09 05:06:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-04-30 01:16:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-04-30 01:16:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-08 18:35:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-08 20:32:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-09 05:06:46.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-04-09 09:04:26.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2024-04-30 01:16:01.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"b6d6a7fec573a629145fd40d223392a775000660","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-04-08 18:30:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b13272694560a6b7eb30599e54d85ff1bab708f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 18:35:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8774e075ba6746b99409479058c7a143\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f4c6d4b81b304655ad00ade6a49124c6 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eb871e60d94347bd89c59428674b8c7d : SUCCESS in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"623d709748ee2768ad8c15360d4153bc9eff5a53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 20:32:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da8584cac4cd4005ac6631308b387656\n\n- grenade https://zuul.opendev.org/t/openstack/build/900a078e0acf4d90ac57f33eaa3a6747 : SUCCESS in 56m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/99e4ad1ace7345e48c496e0db2652896 : SUCCESS in 1h 05m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b7fec57c1a149f9a09a00b863d8c060 : SUCCESS in 1h 00m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85e9533da9954f6db7e05e5a898fd36c : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0283d3fa03564b0fb1ea8a21ce2a442f : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8fd444b3cc3a4830b00c853cfcddb174 : SUCCESS in 4m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9825196554ab48f2a712135b78c85d6d : FAILURE in 9m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07a9de47975244d5b054b9f3f6b7afe0 : SUCCESS in 4m 49s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/96fd71ef297240c585781356c3e36d2f : SUCCESS in 10m 08s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e039d1aabb8e4b17aa2603c970e9cf14 : SUCCESS in 7m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c26aebb355ae429a8853e6368340f597 : SUCCESS in 8m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c450c9f344bf4936900f187d5c6f0cc1 : SUCCESS in 7m 49s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/2097a7326adf428ca55d2f2fa3e4cd95 : SUCCESS in 10m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/57bc5472ad9149ac8e6091d254cbaf46 : SUCCESS in 1h 46m 44s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/edce26f3df884562a97410b641643b38 : SUCCESS in 1h 46m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6fa2917b69614cd88736fca4d861e94c : SUCCESS in 1h 11m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9831db2cf4254f278dcb25c5e5a83d81 : SUCCESS in 1h 11m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f393605f278e4dec86e00e51dd5a7c86 : SUCCESS in 1h 54m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/392978680abe43a2b08b6f6f182420c7 : SUCCESS in 40m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8369990cf3c043bdaa9615652f86b0cc : SUCCESS in 1h 19m 10s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b667872110174f1f8bcff32b947e4706 : SUCCESS in 1h 17m 45s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/84f10d3faf054bb7af3a02d2b2e1bcf6 : SUCCESS in 49m 42s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/de4631134a8748c7bb709afbe6dff202 : FAILURE in 26m 15s","accounts_in_message":[],"_revision_number":1},{"id":"bebf4b32531405ab7efd482967b640de658617c9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-04-09 05:06:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69722c099166358a5a18f0395e3ff79dae7ab92f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 05:12:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8288ebc1a2d84ad4921723997c97f76d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dc72ffd68fd542db997269d5d540fd96 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2e7fb2f00e364e3aa40bde8641436988 : SUCCESS in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f98290a461775cb519c58a49d54f4bc463aad03f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 07:25:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcb610700eb04de09f7a1e70dc2e8c3f\n\n- grenade https://zuul.opendev.org/t/openstack/build/89e6d405a4e0413b991b4cf7a38bc172 : SUCCESS in 1h 02m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6396e1c8048c4666b2980b3090167202 : SUCCESS in 57m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05d0d22daa264bd5ae5c43461ffdc954 : SUCCESS in 43m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf64ef7138bc4ddcac9657b0bf7ef2d0 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88732fd99c57465da0bf8a2426ca42f8 : SUCCESS in 6m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e04fde385c63462b8312500dcdac5d1a : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/faae863b56b14a85b10f05cfd2136a24 : FAILURE in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21aabd9dbedf4f28b7e23aef9e708804 : SUCCESS in 5m 17s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/eff1c448cc4641b2a37206b19b3f0854 : SUCCESS in 13m 25s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ffbddd5b26a64102909404507a5c1637 : SUCCESS in 10m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/526149195017431492de275d7a47271e : SUCCESS in 7m 44s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d7e66aec4ff6407f967a19981a9925a2 : SUCCESS in 8m 59s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/78778b6b4c82479089894c1268ee224b : SUCCESS in 9m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9510911ad8aa4edfb1330ad74c0d0d31 : SUCCESS in 1h 53m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/147616588b354279a15ae7cccf67d980 : SUCCESS in 1h 27m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/70d213c64f8547268c6f3658aee28851 : SUCCESS in 1h 24m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/46243e0dc48b48ad96c66c9aaf4a78d6 : SUCCESS in 1h 42m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4856bb66243443b291884bfb3b889c67 : SUCCESS in 1h 57m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd3cc5f336524d46bb11121d8521c521 : SUCCESS in 1h 02m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f41ae21b1492483ca8091199159cb41c : SUCCESS in 1h 42m 03s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/345c5873ec88459f82f090188f50e15e : SUCCESS in 1h 49m 55s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/082a6cd4707547a49a1697fdd17e7c28 : SUCCESS in 1h 06m 18s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eef957af51474272a11b96ee864c17a3 : SUCCESS in 38m 07s","accounts_in_message":[],"_revision_number":1},{"id":"556fa77d3776294ac67f50dba64e052cd6cc2941","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-04-09 09:04:26.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"22464e0a0de11c9480e9ae4a7b4b75694132be44","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-04-30 01:16:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"225b2ce28d4eff0533306a26f80d9a8af4b8aabe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 01:16:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"484b6c22e510ff0c6666cfa450380f86275f6792","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 03:07:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3de3873d2cb44688436ae438af5eef2\n\n- grenade https://zuul.opendev.org/t/openstack/build/83b77d907e3c4ae2b1bf4b5e009edb53 : SUCCESS in 1h 02m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9445a37ac7c940ceba4942fb2e113569 : SUCCESS in 1h 32m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fb91176972547ff83efaa24a045b9ea : SUCCESS in 1h 02m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf24d06aa2bf4d2c9457879f375c5e74 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7cb5a0ab0d5f429f9128ec1c7dcdbfa9 : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/decad84f22a142df8a3a51d645bf61b8 : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9903fd6bef01480d9a97159158f17bef : SUCCESS in 5m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1a6b6470af8046fd82a350afd0b729d1 : SUCCESS in 8m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6412aa655fae4d9d8da638700e44e59d : SUCCESS in 10m 48s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/41f2b958011746e598b251fd98d1cdb1 : SUCCESS in 1h 28m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7f243d6cff484868b68eaf00c7bb2a84 : SUCCESS in 1h 49m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/073250295cb44145a508a9fa8407f820 : SUCCESS in 1h 26m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a59c0557b69453f949090ea6cc37e52 : SUCCESS in 37m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a23b9e21c49f44bf813160498b801b41 : SUCCESS in 1h 31m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f739ca89779349949f82608dda0e16cc : SUCCESS in 23m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b498166565d8cbcf14281d73cdcac07fb718fdb7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 03:07:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6e1cd88ac7839ee7c89a82e761929198ab3e96a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 03:09:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fc2564e0226402eaf67dd3015d96b85\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bd7470600c3418ca2da66dce45f0d53 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3f7b51653a72bb28cfd140cde183bfbdc5f76c21","revisions":{"3f7b51653a72bb28cfd140cde183bfbdc5f76c21":{"kind":"REWORK","_number":1,"created":"2024-04-08 18:30:15.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/90/915290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/915290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/915290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/915290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/915290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/915290/1"}}},"commit":{"parents":[{"commit":"0bcd6cd71c09917c6734421374fd598d73e8d0cc","subject":"Merge \"Docs: Add info about the cinder store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bcd6cd71c09917c6734421374fd598d73e8d0cc"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-04-08 18:11:23.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-04-08 18:14:20.000000000","tz":330},"subject":"Cinder: Remove ambiguous warning","message":"Cinder: Remove ambiguous warning\n\nIn case of Cinder store, when updating location of a legacy image,\nwe call cinder store to perform the update.\nIf the operation fails, we log a warning in glance which states:\n\n\"Not able to update location url \u003curl\u003e of legacy image due to unknown issues.\"\n\nSince glance doesn\u0027t have knowledge about the actual issue occurred\nduring the location update, we log this ambiguous warning which is not\nuseful for operators to debug the actual issue.\n\nThe responsibility of logging the warning should be delegated to\nglance store (like done here[1]) since it can point to the right\nissue as it knows which operation failed exactly.\n\nThis patch removes the warning and further warnings should be added\nto glance store only.\n\n[1] https://review.opendev.org/c/openstack/glance_store/+/911005\n\nChange-Id: I24b14812a2a15675beacee720d722c2f3a2b252b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f7b51653a72bb28cfd140cde183bfbdc5f76c21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f7b51653a72bb28cfd140cde183bfbdc5f76c21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
