)]}'
{"id":"openstack%2Fglance~915799","triplet_id":"openstack%2Fglance~master~I9050a3cc75147febc1862c6daa929c0d8d25179a","project":"openstack/glance","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I9050a3cc75147febc1862c6daa929c0d8d25179a","subject":"Imported Translations from Zanata","status":"MERGED","created":"2024-04-15 05:06:49.000000000","updated":"2024-05-27 08:03:40.000000000","submitted":"2024-05-27 08:02:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915799-zanata/translations","meta_rev_id":"0cbc151f22e98532e82ccf35ce9e15e9aecb286f","_number":915799,"virtual_id_number":915799,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-27 08:02:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-05-27 07:37:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-05-27 07:37:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-15 05:13:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-15 05:31:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-27 07:37:54.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"6dd1c2936155917f86fd8772d0b4af0a29bff5b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-04-15 05:06:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6253132270c47d49375ccbf7433abf7699836c1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 05:13:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a2b546554c04f4eb2a5be90bcf5575d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/768e7d1c7aff4d89861f7c5f4e779633 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6f1309cd49414bd6ba4a4ab20dfa447e : SUCCESS in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"615a47826acb99ae63fd23dcdcbbfd440b0c6867","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 05:31:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea27297befc34617a2ba3f680186c67b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c88e3f32f9043c295112441239a0443 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/659a706ac8e14363a1151d441503a840 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d260905bbc6c4b4aafb5241df77b071a : SUCCESS in 5m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a7cf59e61f424043bb57f8fb97dd065a : FAILURE in 8m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfa86d42315a480c92e9e983649181a4 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8ab3dfa6de84b01a840561288a7d7b6 : SUCCESS in 6m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f59fbd799853402e8c89930ee2054855 : SUCCESS in 13m 48s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6e1997e33e4d425d977e9e4a94b9192f : SUCCESS in 9m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2d6727d9ba2c4533bdecac5163f7671c : SUCCESS in 8m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b748ffa2f3254fec9823547fea73c6ad : SUCCESS in 9m 40s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/48dce5a9e3b14a5fba11204f21c40d62 : SUCCESS in 10m 17s","accounts_in_message":[],"_revision_number":1},{"id":"926e7991a3d9d85bf002c7bb13d21fd84f24aee5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-04-18 05:12:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"579066d737ea307b6d6acf25d5b3aca1f8610ba6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 05:19:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ab54aa5a96742ddb56784643ca0ce7c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c18df3257094498bb0463dba1ccc33c7 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9e9159dda9dd46f0b045c1440e86126c : SUCCESS in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ba769068417e4583d9331fa90fde4a94d423a161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 06:04:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8aad81658414b578bb31428635f0e05\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f926c1bdbda14cc792d797a1b494828b : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f7062e99ae0406b9ff66ad40156ec80 : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d897e6f4549f42b38aa8b034d451af29 : SUCCESS in 5m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dc8d024e60ff45fc983ea488abf65655 : FAILURE in 9m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1616a1ca6f6a46b2a34db71f37f7eb01 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59b0de5c82b14b4382a14248057f6a23 : SUCCESS in 5m 59s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f83d5f8b1551480eb3a1fd4bbe097358 : SUCCESS in 10m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ddaeca6e6f7144a99b21d00c0ce0af2a : SUCCESS in 8m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d76ce02fa40a4eaea06e2869a7dd620b : SUCCESS in 7m 02s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1633d38ea8f8407d87df96628a9204a0 : SUCCESS in 9m 05s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/4c59185e046841d3a5502518116ac62f : SUCCESS in 9m 25s","accounts_in_message":[],"_revision_number":2},{"id":"86ec7457baae19d57abff2abe530a93c528bc2ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-04-20 04:59:25.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"559e50b3d1971a53c0f5f16881382f262ca75de7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-20 05:07:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03cdb933ffd440768f998cbe4a2e92d0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/61d9a370b1854fc6a1e767727eba231b : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0e11c8b3e4324b4e94b17f494b7134fb : SUCCESS in 5m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1dbcac110e8ba23301639673271571824c90fb85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-20 05:28:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1356b3a8037404ab078e52e44b717f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c628dda27e9c4df7ac4224693989e0fe : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d38de763f4f4efeb9a3a8ccb1b72a38 : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c66ef5ab26a440cb8c4478ae60b6a7a7 : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6733daf94cc44177a34b8cae6c04dafc : FAILURE in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97253eb70d3f437889769bd0e8f356a5 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/459bec58259d4400b153a72f13627332 : SUCCESS in 5m 32s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8f73296dfc1d4a6eb860451fe9f88e89 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dba526b136e24e1f943ac6791bae9414 : SUCCESS in 9m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c261fa8698ce4122b4ca53588d3e529a : SUCCESS in 6m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d32a9316f3614665ab8adbb48b2c8f5b : SUCCESS in 8m 30s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/480e675b2fda4343a39144a439100929 : SUCCESS in 10m 46s","accounts_in_message":[],"_revision_number":3},{"id":"b690f9f768aa63963fd1abb5010a886ab7013861","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-04-24 04:43:27.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"bed7c08209f1170ecae43c128f9bf6bc62755c68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 04:56:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b2beeca060147a285124dc9239cc023\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/17a205e9b9514062893d93cede69e762 : SUCCESS in 10m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0565171b6ad148d48409107dddf813a8 : SUCCESS in 10m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"af05407c81554f71ef2451d6cebd3550bb6c3550","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 05:05:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18820eb20ece443ca55ce2e9d8a87cdf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4c94159e97a42e8b670ea82480308ad : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26325db564f140bb8ed445ced6c7693b : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6acde97df9b6448395e8c123387531f9 : SUCCESS in 6m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/242db73313e9487ebe698d168e37cbff : FAILURE in 8m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9baf8d83fdbf49bd9939c0806a1ab6b0 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f68d1d83f6a44b2bc2ad5e0df27b707 : SUCCESS in 4m 40s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0421a2d9cacd4ff5a0f75ba5e2d5f3ae : SUCCESS in 12m 18s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/61ca633a314647bbb51228fef1c1d6f5 : SUCCESS in 8m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/40a7b1453f4543c48723c42837058b88 : SUCCESS in 7m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/375b45cdaf3547fc83ece6649a857720 : SUCCESS in 7m 42s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/1ddc33c704974b598fc1ef39de3e0745 : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":4},{"id":"d78cd7da91d29d2d5c9938631dd9324367066a1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-05-01 03:32:32.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"380e5bc3f7c69e5c05984d892e76ceccadfe0138","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 03:38:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9218e47fa7f143ffb429166f05bef92a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1368f85762424814bd11bd6365999b14 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/92b514dcb4a5498caad0f38d3e235923 : SUCCESS in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c1b7a0d271629c4b9500539d323fd82ec8b5583b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 04:18:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a64ed9750b0549128ab11e1648bd884b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de4285ed02674fac9b9eff032d9241c8 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d098626b94e5433cbfe5cadfe02fc596 : SUCCESS in 6m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b1ada0ac5fe48cda8a4e1a1192d0897 : SUCCESS in 6m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9df6e5d947140d49d20f44150116ea2 : FAILURE in 11m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf52574323a14a20a0ae36cbc313756a : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c574caa3d86e4621981ca9ece18bd0d9 : SUCCESS in 6m 16s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e2ecd3c9470a47fc8f865c457b5f6818 : SUCCESS in 15m 10s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bb7fc5870a224664bc2be18f14993970 : SUCCESS in 9m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ac1b96f7ec3949c9a4746b07a7d3828b : SUCCESS in 7m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/72cec3aeaf194b62bc805574a7ed60e4 : SUCCESS in 9m 02s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ae3cecaa7c08498e855207375bf7bee0 : SUCCESS in 12m 52s","accounts_in_message":[],"_revision_number":5},{"id":"ac2a9144c0c23c5803e233f79231333382275aef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-05-03 03:33:50.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"5c5b7c3cabf4de587ce828bf48e4dfc2736af039","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 03:39:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbe575dae4ad4ee89ea5ff00805f10b7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b39e72ebe264c599b5cb78d31515d57 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3e7d01e4fcfc40b5baebbfd86c0a1330 : SUCCESS in 4m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e1c8f51174a011051c2db1c048373a68116ef468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 03:52:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99a4b6dc412b499eb2223f0adce797b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbf91d5236d344078eca3cec38b7b5a8 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99e7754b837441059ee82c53f45d9cf8 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1041df6995904b078d346ca2aa0bda25 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ddaf0ab21e7847e99e8d24b7da861a5d : FAILURE in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3890d037252641aeb01e7ea42fdeeb8f : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36e7594891004024a4a8158c9beae789 : SUCCESS in 5m 32s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e82c6aad4cd54e179f36789b127285fc : SUCCESS in 12m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dc4ed2f79cbd4abbaa2d0d0f5240b330 : SUCCESS in 7m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f15d93036b8e4619812aabc06037f832 : SUCCESS in 9m 16s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/380966b8687941dba5cff96256a95202 : SUCCESS in 8m 17s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/796073d06e6e42888eca7ef7ea5bf3ea : SUCCESS in 12m 25s","accounts_in_message":[],"_revision_number":6},{"id":"04f86af3f6e5f7470dfb9274d6e8257ca33c0007","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-05-09 04:11:39.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"6965374100beea95099432836b99c8aaa44331f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 04:17:30.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/923568ce90034374a3ac46d6bd0b63c0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/730f13a8e78e443aa8be5f9404180498 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0c8a88a7f89349ea98040deedffa8ff8 : SUCCESS in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3b50e48e93e06f79bbdb8ba3d60ab98eb6e8e4b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 04:36:22.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3bbe444735342ebb115d2eac46953ed\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c22d0ec92d9d4e24a6c151450f394a23 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e062cfe889bb4aa7969f2a09edb0f0ea : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/734757b4bd43486db29f548bb63c0a94 : SUCCESS in 5m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a7464acfd8d44ddb18b29d89c0abceb : FAILURE in 9m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb16fd3f76e5422884fc46a2fd255ecf : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/969748f270e14d74a6da1fa35fab024c : SUCCESS in 6m 19s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b53bcdbaafe749c0a9a21c2266623750 : SUCCESS in 13m 03s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0d3f606cd36d4dc38fde5e58ee97ef51 : SUCCESS in 9m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/309e478146264a908078e8b4b4a4442e : SUCCESS in 8m 22s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/36f16317878c4fd19f1ca6734cd39075 : SUCCESS in 9m 09s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0db66426a3bf4e9abd7936f6311b30ed : SUCCESS in 10m 37s","accounts_in_message":[],"_revision_number":7},{"id":"15870ed2803e2807b7853e34b45fc3c549effcd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-05-15 04:16:30.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"99ce973ece703a81595e4b42177d37449b6941f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 04:22:00.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb8ce9c1b87f45ccb2578b4a89207783\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7658db99aaca478ca1bd0cd3cceb8aca : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2b9de8eaafce41cd86bbf261abdcfded : SUCCESS in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"50ecbf430b007e65c685842b9d58d2ca63616647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 04:37:01.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b761e86ca6a34881a1c28d4b37a31e66\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47f65940948c4cc4832119bc7353fe4d : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ea247e022924e2d90e4ac4eb3ad2d0c : SUCCESS in 6m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c541863cf014667bef7512e8c1c342e : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/141c4d059e6344b29b9cedb40a73e71a : FAILURE in 8m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cd5a35dc3774b7bb6daad128e4c6a42 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4da7b74c6fc04f709ca5bf4f3df2c4ac : SUCCESS in 6m 08s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/020b6798befe4aaca5281055526467bf : SUCCESS in 13m 38s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/953c839ca36f4114a6cfb642fc3e3c9c : SUCCESS in 7m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1ed16a638e094ba598ab98c5f7ef6868 : SUCCESS in 7m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f235a0cf3fa3481badd76ca2c6bddb0e : SUCCESS in 7m 41s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bd4bada6fb3c3dc07672d46399643327861a92d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-05-17 03:55:31.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"7f7d509a756a843bb50b1eba2f4c13fff1d2b2bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 04:07:29.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/140b0b8608e54c3da4dfa707d42913fb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aab19576ad6a45a783eb26991e3df19c : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bbc4a2fc86af4d4a80a1ba100dfa6f2c : SUCCESS in 10m 12s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a36575923ed82a57877d0ad0f54f79a9f528a616","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 04:14:34.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4286f90933434c6eb7b9204c9474bffd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1756eef9706e486eba69170e3702c635 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18a1fe9acfe34804a9562db3de055f28 : SUCCESS in 6m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69e8ab4a718444febe30a418c0b04573 : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae13ef95dde8481d826a234df10ab7fb : FAILURE in 9m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1f83f5a564b496699ac18b3009a7fc1 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40e7ee05788f47f8a0628aa5f102740c : SUCCESS in 6m 42s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f057d79b345c412bb802fcc0c320037c : SUCCESS in 13m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/001bdac4be6a4c88b7601867eed3d73c : SUCCESS in 8m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5890bf1e98b047498a884cc3340aa46c : SUCCESS in 8m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a31adc8051994566a59a1eba3a92ad9e : SUCCESS in 8m 11s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e77654e1c6552cf723da6ddaa9ff5111db95dae9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-05-18 04:07:33.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"dadb99c3b2a8a5f4f6e2fa3351104d05fcf4ecce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 04:17:58.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74999a231984476f8f4f76152f6f78d0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5617304777eb42e8bd5c28b8d1d2d21d : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2564ad7bc9634458bec5eddc02710f4b : SUCCESS in 8m 23s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"2d55e54412f3effde21eaf13e817d163e1bde762","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 04:27:22.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7c3e6b284af4063954655caf705f10d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c51bb9e5a9c143b2993765647326272f : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7dd8872b092478d9d99325c8ab69a6b : SUCCESS in 7m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/031c6afffa904017b6cd2dce3bd5bffa : SUCCESS in 6m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a672fdd69794e149dcbcaba0b16f6a0 : FAILURE in 9m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bca044d7f6f417793a67f0b4779ea65 : SUCCESS in 7m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6531f9f06ee41b48e2c0606f8959126 : SUCCESS in 6m 48s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f9ae9139cb2e446095ce0ab3b4d2bde6 : SUCCESS in 12m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d6981b2d8b5f4e0ea1ee550cf7db002d : SUCCESS in 9m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7fc3bb61193b471f84b6ec9d8968ab8c : SUCCESS in 8m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7c299b700c9940fd82976c10dd63cffb : SUCCESS in 9m 05s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0a2f9bdbb0817c85fce4abc42fdf5da185cdc37d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-05-23 04:23:18.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"231f91845ee2690d4f7dd52a68aa1c15baff3959","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 04:29:40.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39f5728f9df74931957b77aaa8089bdc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/238dc9890f3541f9a7c9be7a08c83154 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c0c9b155120a47e6ba86d5f9dacf6282 : SUCCESS in 5m 29s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"8da5181f5d7535bb631232d17ca02bcc24374e5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 04:40:09.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9badaa749ce4ce19038fc41cabfc8cc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b13f8f139804d97958d4f6a78dcab87 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be00641739ca4393a35f404f62ef6ef7 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/121cde45d29544da9363d11bca41d3b9 : SUCCESS in 5m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ab9ae5e18c944089a91e6c37ecaca3ad : SUCCESS in 9m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/597f69f5f5ca4456be9be80935b0a585 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aaa3341eceb44f9d9d42810fe9de8569 : SUCCESS in 6m 17s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8f6c569318ef49dca5805e1f12b351b9 : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/070a73c0258e43b7a1b8e360b1c644f2 : SUCCESS in 7m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a6b7c1edcbf0421fa7051645cde69463 : SUCCESS in 8m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0813581dd32948988d99f512ca1d72ae : SUCCESS in 12m 09s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d20c600007ba4310b88451b05788a013 : SUCCESS in 8m 36s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"991bfa1fd43e5b655414d64d590f0fbf8dbad3fc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-27 07:37:54.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"cf2588bab5f86859494daa1fe360c37b3bff9085","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 07:38:13.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"671b6fddc2a77106660e262999568dcae2d87c06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 08:02:30.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e085f82b74794c4f8cbd563b5d6606e1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d613a97001f848b5ac04b7d44a1df972 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fbda3b9d42d4c1c87b48191e973ab70 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/184089a398474b10894801d1cf73d798 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ba5dc90081a48b8ac91c057349ec59a : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8f370b5d8054bab90f04c675e383bc2 : SUCCESS in 6m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5674dc6d98fb46d1a704fa8f02f2228a : SUCCESS in 7m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ddd7f1baf80b4e31a6aedb6f3f612a75 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":11},{"id":"a8c43b5c5dafcc0098318e3015939be9b47f2655","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 08:02:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"0cbc151f22e98532e82ccf35ce9e15e9aecb286f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 08:03:40.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/871551b7cedc46c091c81ceec40264b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ee36347569144a29a4babd6c62c9dd5 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee73466165dc4435991403d4acad4e98 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"3f49730e627206c464163a3e1f00580b09cc5ebf","revisions":{"53c1f35afe4bf13f7920e22cb27d6c012a170921":{"kind":"REWORK","_number":1,"created":"2024-04-15 05:06:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/915799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/915799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/915799/1"}}},"commit":{"parents":[{"commit":"0bcd6cd71c09917c6734421374fd598d73e8d0cc","subject":"Merge \"Docs: Add info about the cinder store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bcd6cd71c09917c6734421374fd598d73e8d0cc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-15 05:06:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-15 05:06:48.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I9050a3cc75147febc1862c6daa929c0d8d25179a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/53c1f35afe4bf13f7920e22cb27d6c012a170921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/53c1f35afe4bf13f7920e22cb27d6c012a170921"}]},"branch":"refs/heads/master"},"ac197551a56cca7cd1a4f10661ade8f21e845e27":{"kind":"REWORK","_number":2,"created":"2024-04-18 05:12:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/915799/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/915799/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/915799/2"}}},"commit":{"parents":[{"commit":"835e7f36a52d8ea22052bb59be84ae792499d62b","subject":"Remove unused run_sql_cmd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/835e7f36a52d8ea22052bb59be84ae792499d62b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-18 05:12:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-18 05:12:33.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I9050a3cc75147febc1862c6daa929c0d8d25179a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac197551a56cca7cd1a4f10661ade8f21e845e27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac197551a56cca7cd1a4f10661ade8f21e845e27"}]},"branch":"refs/heads/master"},"8341a3a1df88081030da0248fb693cfd186a5209":{"kind":"REWORK","_number":3,"created":"2024-04-20 04:59:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/915799/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/915799/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/915799/3"}}},"commit":{"parents":[{"commit":"b8375913bf7645d13ba0b8360453527e151c4bdf","subject":"Merge \"Replace remaining usage of [DEFAULT] sql_connection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8375913bf7645d13ba0b8360453527e151c4bdf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-20 04:59:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-20 04:59:24.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I9050a3cc75147febc1862c6daa929c0d8d25179a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8341a3a1df88081030da0248fb693cfd186a5209"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8341a3a1df88081030da0248fb693cfd186a5209"}]},"branch":"refs/heads/master"},"9df6756db733a0f1226c4a3438909d59e8d87025":{"kind":"REWORK","_number":4,"created":"2024-04-24 04:43:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/915799/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/915799/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/915799/4"}}},"commit":{"parents":[{"commit":"7b6e89155968c531f63e41e05af89a75fceadd06","subject":"Merge \"doc: remove trailing dot from hypervisor_type list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7b6e89155968c531f63e41e05af89a75fceadd06"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-24 04:43:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-04-24 04:43:26.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I9050a3cc75147febc1862c6daa929c0d8d25179a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9df6756db733a0f1226c4a3438909d59e8d87025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9df6756db733a0f1226c4a3438909d59e8d87025"}]},"branch":"refs/heads/master"},"a93676a8822ae5a1daa80175a276d25922e06b08":{"kind":"REWORK","_number":5,"created":"2024-05-01 03:32:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/915799/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/915799/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/915799/5"}}},"commit":{"parents":[{"commit":"02fbc47733ea50b73fbd2afe716b63f7c5d1402a","subject":"Merge \"Change DB migration constant to 2024_2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02fbc47733ea50b73fbd2afe716b63f7c5d1402a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-01 03:32:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-01 03:32:30.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I9050a3cc75147febc1862c6daa929c0d8d25179a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a93676a8822ae5a1daa80175a276d25922e06b08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a93676a8822ae5a1daa80175a276d25922e06b08"}]},"branch":"refs/heads/master"},"2c0ae003f829368baf738632d764e1f58d35d4ce":{"kind":"REWORK","_number":6,"created":"2024-05-03 03:33:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/915799/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/915799/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/915799/6"}}},"commit":{"parents":[{"commit":"95770a25e33cd531216104c788ac5dacda4e36cf","subject":"Merge \"refectory: fix comments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/95770a25e33cd531216104c788ac5dacda4e36cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-03 03:33:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-03 03:33:45.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I9050a3cc75147febc1862c6daa929c0d8d25179a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2c0ae003f829368baf738632d764e1f58d35d4ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2c0ae003f829368baf738632d764e1f58d35d4ce"}]},"branch":"refs/heads/master"},"fbaadb0c2bcec1644e6d4ba156235fdd7fb03656":{"kind":"REWORK","_number":7,"created":"2024-05-09 04:11:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/915799/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/915799/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/915799/7"}}},"commit":{"parents":[{"commit":"32ee57ccdeea90a86bce1e5e33d10fab115ed2d3","subject":"Fix broken glance-cache-prefetcher utility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/32ee57ccdeea90a86bce1e5e33d10fab115ed2d3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-09 04:11:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-09 04:11:34.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I9050a3cc75147febc1862c6daa929c0d8d25179a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fbaadb0c2bcec1644e6d4ba156235fdd7fb03656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fbaadb0c2bcec1644e6d4ba156235fdd7fb03656"}]},"branch":"refs/heads/master"},"3172ba006eb6644d50ef5a38ca0b577486a5aec7":{"kind":"REWORK","_number":8,"created":"2024-05-15 04:16:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/915799/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/915799/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/915799/8"}}},"commit":{"parents":[{"commit":"462ac84445c4c446dbbb11f6495c1a94bc05b369","subject":"Merge \"Remove SQLAlchemy tips jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/462ac84445c4c446dbbb11f6495c1a94bc05b369"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-15 04:16:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-15 04:16:29.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I9050a3cc75147febc1862c6daa929c0d8d25179a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3172ba006eb6644d50ef5a38ca0b577486a5aec7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3172ba006eb6644d50ef5a38ca0b577486a5aec7"}]},"branch":"refs/heads/master"},"e03b31d18f89c2e5e3893cd0d7dcbce9f3e2d1e1":{"kind":"REWORK","_number":9,"created":"2024-05-17 03:55:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/915799/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/915799/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/915799/9"}}},"commit":{"parents":[{"commit":"705b145ab83defb3cb2f8d3243e1ce900fe337b9","subject":"Merge \"Refresh Glance example configs for dalmatian m1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/705b145ab83defb3cb2f8d3243e1ce900fe337b9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-17 03:55:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-17 03:55:29.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I9050a3cc75147febc1862c6daa929c0d8d25179a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e03b31d18f89c2e5e3893cd0d7dcbce9f3e2d1e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e03b31d18f89c2e5e3893cd0d7dcbce9f3e2d1e1"}]},"branch":"refs/heads/master"},"77d6b3414872bb3c10bfbc49d5d500c45052e224":{"kind":"REWORK","_number":10,"created":"2024-05-18 04:07:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/915799/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/915799/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/915799/10"}}},"commit":{"parents":[{"commit":"da1b9792f269cb87b97142005927833cbeab3f32","subject":"Merge \"Docs: be more explicit regarding oslo_limit.endpoint_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/da1b9792f269cb87b97142005927833cbeab3f32"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-18 04:07:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-18 04:07:32.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I9050a3cc75147febc1862c6daa929c0d8d25179a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77d6b3414872bb3c10bfbc49d5d500c45052e224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77d6b3414872bb3c10bfbc49d5d500c45052e224"}]},"branch":"refs/heads/master"},"3f49730e627206c464163a3e1f00580b09cc5ebf":{"kind":"REWORK","_number":11,"created":"2024-05-23 04:23:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/915799/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/915799/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/915799/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/915799/11"}}},"commit":{"parents":[{"commit":"9f4754cd8060efafb67ca73efe71708c556a5fc3","subject":"Add non-voting functional-py312 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f4754cd8060efafb67ca73efe71708c556a5fc3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-23 04:23:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-23 04:23:16.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I9050a3cc75147febc1862c6daa929c0d8d25179a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f49730e627206c464163a3e1f00580b09cc5ebf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f49730e627206c464163a3e1f00580b09cc5ebf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
