)]}'
{"id":"openstack%2Fglance~916162","triplet_id":"openstack%2Fglance~master~I8b1dcb76d16ebd6a2c90b75b6ad3de34be3cba99","project":"openstack/glance","branch":"master","topic":"db-migration-version","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-04-30 13:55:17.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-04-19 04:35:07.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-04-30 11:42:25.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I8b1dcb76d16ebd6a2c90b75b6ad3de34be3cba99","subject":"Change DB migration constant to 2024_2","status":"MERGED","created":"2024-04-18 05:49:20.000000000","updated":"2024-04-30 13:59:17.000000000","submitted":"2024-04-30 13:55:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916162-db-migration-version","meta_rev_id":"bf64d28bba4705617d55cf2619bcc237107d5be5","_number":916162,"virtual_id_number":916162,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-30 13:55:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-04-30 11:03:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-04-30 11:42:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-04-30 11:42:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-18 05:51:00.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-04-18 05:51:00.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-04-18 05:55:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-18 08:08:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f78f0f4e324aa6769171cd509fdcdf2335eab435","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-04-18 05:49:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44b7c3829694f25716becb80597b973ec29bb60d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 05:55:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b23b3c4371c4eefa3c0c5fdba37174a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/49c57110fb544835b8c3e0b6917777f8 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/35d11e5be93349a882ec938432023cd6 : SUCCESS in 4m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87fcc65e5b88b998f521900a593f2b261a38e6b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 08:08:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/583a191c89d54d0c8510979743339005\n\n- grenade https://zuul.opendev.org/t/openstack/build/2b7ab8e81aa94760a23e4f5254b625c0 : SUCCESS in 1h 01m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/072b13cf78214198bfacff1fb1abb69b : FAILURE in 1h 34m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a0f969133cd4285a68ce31ebb1612d3 : SUCCESS in 42m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d6904f62a3143729e3861bd72ed3ac7 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b825051f7fd4045b075883ca689d2c5 : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff21c3a6f2234f67a87c956598caea6f : SUCCESS in 5m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c8937ab140645aebe8c3b3447746aa9 : FAILURE in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/237c250e015f4b3db58829635f091bb9 : SUCCESS in 6m 14s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5fc0ebef834d4c8e8b88081247a464f6 : SUCCESS in 13m 56s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/520294764e634cc0b4a08119a541216c : SUCCESS in 9m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b911e000d9bd459d9fad791fb61054a6 : SUCCESS in 9m 02s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a62d3c89317b48cda729d08815a9d01c : SUCCESS in 7m 50s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/7178c557140d4d5ea0b2aaf6a4f64061 : SUCCESS in 9m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/efc2b84b4bef4cea9a557f5c86181805 : TIMED_OUT in 2h 03m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4f92ca970d6b435d87d0e8f866c73398 : SUCCESS in 1h 47m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ce37019cd65a445e8a27cf7d009a6396 : SUCCESS in 1h 15m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/678a8323010f4881a498f76bec8ab374 : SUCCESS in 1h 24m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f62aee68f3524f4696f11afb79312326 : SUCCESS in 2h 11m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b2895f951b94103b6b6695837e563fc : SUCCESS in 1h 05m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/31612a35de5d4ed882acf6b3a2506272 : SUCCESS in 1h 27m 38s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/6877b851736848e19406255f952d3ae4 : SUCCESS in 1h 44m 56s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/41925ab274fd4510939d7d47db872415 : SUCCESS in 1h 06m 04s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/32603171277143209ed3171aa7929681 : SUCCESS in 39m 48s","accounts_in_message":[],"_revision_number":1},{"id":"57fc8534be6a04d5f12bc576f217e86a43499f8d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-04-19 04:35:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"778d186b726323c9e699dd1a8b1ddb99e0f024c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 04:44:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17887c09efa5472e852a52e681761f9e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e93de0fec6fc40229fc5bf71bf39d9c9 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/339b79e8862a47908ec1cc808ecebbb9 : SUCCESS in 6m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"020b4c5bdc68f3ce6690fe982f7251450aefd70e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 06:46:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ab81b60ef3a46f092cc69d176f6b71f\n\n- grenade https://zuul.opendev.org/t/openstack/build/6a7c8680253e4938ba0c6335c4213ad4 : SUCCESS in 46m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0c0d5fb1516c4fd8932dc85fb8d27808 : SUCCESS in 1h 12m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4311b6be20fd4bd9b09077dc5ac36dc7 : SUCCESS in 46m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19768cb8a13c456baaef0f8557971a8c : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a753b02047714db583c66546a7dc83d4 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa28b848b6684387aaf3cbe20064cab9 : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c4917b1e05b454d87aa1de8327fa2d0 : FAILURE in 9m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6487164c58b467babde3e71d04892a4 : SUCCESS in 5m 50s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/cbac47f8071c41b7bab237d18f64ba55 : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ee34fd2275054205b4d003c7fc4196ee : SUCCESS in 8m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e2f3c88bffe245dbb751b64386457ed3 : SUCCESS in 8m 04s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/dd83645a704c44e3ba7688b2c1626791 : SUCCESS in 8m 36s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ac647e688e5e49fc86d93e2223a42d6c : SUCCESS in 10m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/00f1bb390afc4ee5b83e8be7fc836087 : SUCCESS in 1h 45m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5c4dde8af3564202a6ea45aa3ee3ddf0 : SUCCESS in 1h 27m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4cbecfa9ead6475e9eb8023720b4d90d : SUCCESS in 1h 46m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9f06fd24b26c43469c4830cb79637f79 : SUCCESS in 1h 37m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/29ae1209a07e4d6d80790250e405ab6e : SUCCESS in 1h 29m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c97232e991048f2b23496b967dace73 : SUCCESS in 1h 04m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c3b0dff06fbd4c8ca179960a3fea558b : SUCCESS in 2h 01m 01s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/0b08181a7a9442e3bf27d724e1263030 : SUCCESS in 1h 41m 20s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/428743cbe5b944918a7a042372f5e401 : SUCCESS in 51m 52s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a8807e6dc8d2444aa65c6cdaacad5385 : SUCCESS in 38m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ed10f39c8f089f8dd203c81e36866463f109e45d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-04-30 11:03:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2d39ba034afb0ce9cf020fc43b58a95bd1aa13e8","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-04-30 11:42:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c3cdffc1cccc0bae6efbc85b944ecfdb06e00001","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 11:43:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b1267d8500e22a535a4d39c8e19a4daf8f77d7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 13:55:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23ff9ef5a8354a45812bd5959e2e86f8\n\n- grenade https://zuul.opendev.org/t/openstack/build/3aa7f6852eac4b49851b85fee1220e85 : SUCCESS in 1h 02m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/72922a52e8a84699a5aa996a33ddd5c2 : SUCCESS in 1h 37m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b7f21ef83154569996922f690a6e22c : SUCCESS in 1h 23m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf2cb9fac843474b973b1d9e24cf5678 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d9c7e86360c48beb856e52690a6f48e : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa598512d7d041b2964cd4763d6774bd : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d76e36149440b0aca87153b10b06c3 : SUCCESS in 6m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d99a3cd0d83b4a65ad74ffbcea91554c : SUCCESS in 9m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/28b2f824c6b8444ab22e2493a1283c8a : SUCCESS in 9m 19s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2e801abe29c34495a2e1eb08c6c15939 : SUCCESS in 1h 43m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2fde2f9c746a42d09f3b84f27f5078a8 : SUCCESS in 1h 47m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b6e66f7d2b844c4fbf1a367d324a8c63 : SUCCESS in 1h 44m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36fa37078cf64d6aa485ba9b3a0dccbb : SUCCESS in 1h 07m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/22c87e3a4d4d47959e470243e0589c7c : SUCCESS in 1h 40m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/04cff40fce1d4632b6dfa4c9401a6d7f : SUCCESS in 40m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3579769b33ffb3b8737ade949884491afe305e82","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 13:55:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bf64d28bba4705617d55cf2619bcc237107d5be5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 13:59:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de10a89a2c6d43dcb0a11c2b3e66c0d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/913a29afc71f43e187dd21fcbdfbf25c : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b4e36f160869d74a9e63563096739dc2a52e3d3","revisions":{"4b4e36f160869d74a9e63563096739dc2a52e3d3":{"kind":"REWORK","_number":1,"created":"2024-04-18 05:49:20.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/62/916162/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/916162/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/916162/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/916162/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/916162/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/916162/1"}}},"commit":{"parents":[{"commit":"835e7f36a52d8ea22052bb59be84ae792499d62b","subject":"Remove unused run_sql_cmd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/835e7f36a52d8ea22052bb59be84ae792499d62b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-04-18 05:42:40.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-04-18 05:42:40.000000000","tz":0},"subject":"Change DB migration constant to 2024_2","message":"Change DB migration constant to 2024_2\n\nChange-Id: I8b1dcb76d16ebd6a2c90b75b6ad3de34be3cba99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4b4e36f160869d74a9e63563096739dc2a52e3d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4b4e36f160869d74a9e63563096739dc2a52e3d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
