)]}'
{"id":"openstack%2Fglance~916301","triplet_id":"openstack%2Fglance~master~Ie682a09d238c9de1dd120dc58e9f6a6206be6eb1","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-04-18 15:08:36.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ie682a09d238c9de1dd120dc58e9f6a6206be6eb1","subject":"DNM: Run py312 unit tests with modified taskflow","status":"ABANDONED","created":"2024-04-18 13:56:44.000000000","updated":"2024-06-10 01:14:05.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c33dfe9059492af1598df0e8d04ecd85696267d1","_number":916301,"virtual_id_number":916301,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-04-18 17:24:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-18 14:03:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-18 14:57:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7143c2c5ee44f3602fe73ff23fbd3bad12c79d0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-18 13:56:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f7912b27568be70f203700ea6f61e4f9e5d02c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 14:03:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65110c209c9a41998366302282efcffa\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/805c8ace574748fa99eb924ef52a92f7 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2af90519e4464c6ca343bda06c929aa5 : SUCCESS in 5m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31c62a90e589ada7a04b05ddc2ba540144ba27e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-18 14:03:40.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"02fdba0965dfae0b903894bfd8123c0b0814372b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 14:09:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/718feb92c6eb44cbaa58c0bea0eb2f5e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb314fb656cb4482b77295af4d8113e0 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b8c107409d5f462bacbc1ebaf0e4da75 : SUCCESS in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8d2b79a78073d22dc56af37005faadb38432ba79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 14:57:59.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/916147 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"c7fd234109cd0268bd6a8745fb8c7b0c4c313549","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-18 15:08:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"95a2390a86e29ac879f98ed367e3c05794685e70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 15:15:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7edc671056de4b3384fec476b78f8a9f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da0c5cf573024bf6bca090e747b21e11 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e67c090a105c4980bfdf04104554a637 : SUCCESS in 5m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c7d3d16d98e717fd422c405a35b296c688037944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 17:24:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dafde61dfd54ba69241dfc53cafe738\n\n- grenade https://zuul.opendev.org/t/openstack/build/f6b149ac563d423ca6e2463740aceb12 : SUCCESS in 1h 16m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e2b9bce5a6fb40e38c2aacb7b4ca8f6f : SUCCESS in 1h 40m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/522f2fa3f55f4af6922e73b28e527fac : SUCCESS in 1h 04m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d073e5eb17e4599be48457e2495c761 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2de98dc06b84ae9ab62f2d8c075bb86 : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7d478f0492b840f1b98a2830fb96eb11 : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e94647a05694e37b647844cc306eae2 : FAILURE in 9m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f91ab26d8484153bfd4425241be0032 : SUCCESS in 5m 57s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d1f30ed637c345fcb25ca28565990060 : SUCCESS in 13m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/433ec134ab9e46999ff1505831ab31f7 : SUCCESS in 7m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d8a253ac17864ee7b9c05265da96b4ab : SUCCESS in 7m 38s\n- glance-tox-py312-tips https://zuul.opendev.org/t/openstack/build/33f68005999b41f2a37fb0ecb203c966 : SUCCESS in 7m 55s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f22807bf452647a49c4e88c6343c28b6 : SUCCESS in 8m 52s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/dc0dec6da30a41e0bb1a43e035f7b13a : SUCCESS in 12m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c58cb2fd6d3042798c182005b3c3b5b7 : TIMED_OUT in 2h 04m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/25b2908667fa4421b7fa27e46d84ca05 : SUCCESS in 1h 31m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b5e9a554a95b4abca31330a195283f7c : SUCCESS in 1h 50m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bcbd8d6a49504d40842ca84ac12724d8 : SUCCESS in 1h 48m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d41bdf44d2384828b9e24b8392055a71 : SUCCESS in 2h 13m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5417520e02fc4d38888f096032e4e58b : SUCCESS in 1h 04m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/64e3a18b41f14575a0f4ef2ca5eda334 : SUCCESS in 1h 49m 17s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/498b8ca83a4c4318822eefaf42878873 : SUCCESS in 1h 41m 34s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/458da78d90184e05914e62b511377a34 : SUCCESS in 1h 08m 37s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bb90d6af27e445f9bb4d60dee9aeafde : SUCCESS in 34m 39s","accounts_in_message":[],"_revision_number":2},{"id":"c33dfe9059492af1598df0e8d04ecd85696267d1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-10 01:14:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eb37e3e72ff4ce2ea920121d9c0805bf8e0eb214","revisions":{"7ea1683b9be2bb577eff1ad1e2438c7856293c95":{"kind":"REWORK","_number":1,"created":"2024-04-18 13:56:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/916301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/916301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/916301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/916301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/916301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/916301/1"}}},"commit":{"parents":[{"commit":"0bcd6cd71c09917c6734421374fd598d73e8d0cc","subject":"Merge \"Docs: Add info about the cinder store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bcd6cd71c09917c6734421374fd598d73e8d0cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-18 13:55:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-18 13:56:30.000000000","tz":540},"subject":"DNM: Run py312 unit tests with modified taskflow","message":"DNM: Run py312 unit tests with modified taskflow\n\nDepeneds-on: https://review.opendev.org/c/openstack/taskflow/+/916147\nChange-Id: Ie682a09d238c9de1dd120dc58e9f6a6206be6eb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ea1683b9be2bb577eff1ad1e2438c7856293c95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ea1683b9be2bb577eff1ad1e2438c7856293c95"}]},"branch":"refs/heads/master"},"eb37e3e72ff4ce2ea920121d9c0805bf8e0eb214":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-04-18 14:03:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/916301/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/916301/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/916301/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/916301/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/916301/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/916301/2"}}},"commit":{"parents":[{"commit":"0bcd6cd71c09917c6734421374fd598d73e8d0cc","subject":"Merge \"Docs: Add info about the cinder store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bcd6cd71c09917c6734421374fd598d73e8d0cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-18 13:55:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-18 14:03:40.000000000","tz":0},"subject":"DNM: Run py312 unit tests with modified taskflow","message":"DNM: Run py312 unit tests with modified taskflow\n\nDepends-on: https://review.opendev.org/c/openstack/taskflow/+/916147\nChange-Id: Ie682a09d238c9de1dd120dc58e9f6a6206be6eb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb37e3e72ff4ce2ea920121d9c0805bf8e0eb214"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb37e3e72ff4ce2ea920121d9c0805bf8e0eb214"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
