)]}'
{"id":"openstack%2Fglance~916422","triplet_id":"openstack%2Fglance~master~Ia4784f59d16660e8d40c0e409f092ac4e46870b4","project":"openstack/glance","branch":"master","topic":"remove-pysendfile","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-05-16 04:02:32.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-05-14 05:11:37.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-15 23:50:30.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-05-16 04:02:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia4784f59d16660e8d40c0e409f092ac4e46870b4","subject":"Remove sendfile support","status":"MERGED","created":"2024-04-19 12:27:44.000000000","updated":"2024-05-16 04:04:15.000000000","submitted":"2024-05-16 04:02:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916422-remove-pysendfile","meta_rev_id":"bfdcedbef1847388f17b073f8865cd1455f541a7","_number":916422,"virtual_id_number":916422,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-16 04:02:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-05-09 08:31:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2024-05-14 05:11:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2024-05-14 05:11:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-19 12:38:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-19 14:43:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-30 11:26:59.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-04-30 11:34:17.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-04-30 11:34:32.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-05-09 08:31:18.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"}],"messages":[{"id":"4c85798ccac54b947e2e4ed44ca6fc159bb4c268","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 12:27:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c77f79d39dacdd0ab5faeab1030040d9ce994bbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 12:38:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ce5373c0f494d1eb9edd50e98e043ff\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a96d69b0bdd4974ad45cf97d43f7049 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ea3b81e5acfd491b8b0bd24e57f1f065 : SUCCESS in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d796dabfde5ee9f828b9a14a6d73a60e1e6f227","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 12:41:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8e1d526782da0cbae87d8a04be64faba952b22e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 12:56:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efcd12d76fd14d73bb24abb14bbf0c0c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4daa5055dfaf487b8fef359e2ff0975c : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/04d47a8a31154e4b8d8da47cbabf1986 : SUCCESS in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"93af721d6d87ef7a840d2e40a667ed5584c7e90d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 14:43:40.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e8f8a1dbc3a43e29a119328c8cfa3f9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f868ce72601e4c5bb2c0f407b0047725 : SUCCESS in 4m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/431c243ee16840a8b26613e8bf556119 : SUCCESS in 2m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/bc8407e934a04dd0b1ee3af4623ca043 : SUCCESS in 1h 15m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8003a8f93afb434babbe75327bf0ae69 : SUCCESS in 1h 19m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c54f17d3fd34e74998827ee8417629a : SUCCESS in 1h 02m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/260e69878e0c4722b402c3ac8c6428ae : FAILURE in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acd932c197474f1384742416744bff96 : SUCCESS in 6m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2db20fb2086d47d893e446a5c5dd8aa1 : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f0906b24ab5d4edb8133c28ecad26e5d : FAILURE in 9m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a3d0f9d15b747e59a9f453d6e338e8c : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5b7c1875a1a4aa4b3aaed51c6b16311 : SUCCESS in 7m 24s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2a96f89178884f20bb91ac5f7ba713e5 : SUCCESS in 12m 45s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fa3d3656526e4fc6bbcc510b47ee0c57 : SUCCESS in 8m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/41acc0889c3248a79c877566096da58e : SUCCESS in 9m 30s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0b27040c184d4c65ab3c2b97fbcc886c : SUCCESS in 9m 11s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0c30538d55774bc790528301d6af3c65 : SUCCESS in 10m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a3d882a258914723b1f3caaad4315d75 : SUCCESS in 1h 42m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5872b878b4e94750b5661f97da604e49 : SUCCESS in 2h 00m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ac71a7743c604ee88c3d6ad6339c4045 : SUCCESS in 1h 49m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/06559491bf494e678cbe61538c0739dd : SUCCESS in 1h 42m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/366e6b3db14b4aaeb752ef76370bb699 : SUCCESS in 1h 51m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7663a0ab2da47cd98d073de19171985 : SUCCESS in 1h 04m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a4a3d908d960426ca6933ed7f7ba3e63 : SUCCESS in 1h 47m 07s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/314428c2a66a4d628faf0dda5a6781b4 : SUCCESS in 1h 25m 25s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d9a841f239a34d63a2fdb641cde52ff1 : SUCCESS in 58m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1743d3a99c8a4fffb3d8ee796f1629e1 : SUCCESS in 27m 27s","accounts_in_message":[],"_revision_number":2},{"id":"1a0d5cb6efb191060f0353f2a2c25d0b6f040329","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 15:08:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8ba110e2ecba5c2b87f6f93d0f58051a3c1b9a54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 15:15:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"672f02c0796b3dd80c6666efa94575619e0b2f50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 15:15:33.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"a7da4397da55bb9f5ce6c091081c8a1e0bac7654","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 15:19:26.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f4f09c273d9a2966be6a8b87a1f64973ea758423","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 15:26:46.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"3617e029fa1fae2a0b7cecc85dc2dd0c8ef11eb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 15:28:10.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"a875a07544528d3b844b1c35c02fd91c90d54173","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 15:43:47.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/642ba36ff1af450e9e81b0688310e60a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b48cdb4899f4905ae115a019fdc7294 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/99c65f0c30824114bf7d0c1a02e0d981 : SUCCESS in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"43facbddc157e5c82262c1f5927a4072331c4a33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 18:04:38.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8792444ce46145c1bc618c3f1c742950\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d263ea58af7d4350b066fe83be839382 : SUCCESS in 4m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/123779b39828431ba4c7361a9e993c41 : SUCCESS in 2m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/3e2e7abfb8724677a731f0c549e840af : SUCCESS in 1h 16m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ed8a18553ef040d39b73bd5eea40643b : SUCCESS in 1h 30m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01d65b8d6547471d84527d7d6e049e02 : SUCCESS in 1h 00m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/930f1ded4b294699a79470c13e64d0f3 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a45326e032614bffa20a22c6e4ed5c40 : SUCCESS in 6m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c3684b2a9fdc49b4b74058e045051274 : SUCCESS in 6m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c50827606d1d406784237efa37ae6837 : FAILURE in 8m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f39c342278cc477684aee4e4652fa4da : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc6314a241ee4f70ba00fc43d5c21289 : SUCCESS in 5m 57s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7faf588dcb844f6385a14d949f5ed1d9 : SUCCESS in 12m 21s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6e1ac7ebd26f4c5f99c606e41c8f7e0c : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/83fb397f6c704727be121d445787f800 : SUCCESS in 7m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/65466abfb4da45c686b9be831356c819 : SUCCESS in 11m 00s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a09c0e71b1bd4581901853589f0f8e63 : SUCCESS in 12m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6485981206914ff9aea4c339c2ce6d94 : SUCCESS in 1h 55m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/85d36fdb6c6e4c4b8839e602963cffaa : SUCCESS in 1h 54m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/afe190072bd44b77b7ad04994619fdfa : SUCCESS in 1h 35m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0da1d8d142a242798ffc174925817ec9 : SUCCESS in 1h 47m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b4b5f19b56b54ec19529f14d698bcb3e : FAILURE in 2h 22m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/099e99021e474aca9ceca93fb65a2385 : SUCCESS in 1h 06m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/70da3d47fcad4799aaf1c4200edc482b : SUCCESS in 2h 09m 30s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/4e16a35e6b4c44b9b94ebdbf9c56ae00 : SUCCESS in 2h 00m 46s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/6c41d173cd164e009732dafcd3f20687 : SUCCESS in 1h 10m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/74b9d709ba8a4f9b836e49d33b5782f5 : SUCCESS in 42m 22s","accounts_in_message":[],"_revision_number":8},{"id":"74dbec05f7f6f6492c39474720e48a2fc8eed8bb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-20 14:50:27.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"14b9a16c0ed973f8aed508f82d9b6355b41c3930","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-20 15:02:59.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24e7c62a1490419dbdc92934502a0a55\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b60289dbf4f4737bd61a8bd7d8e3c00 : SUCCESS in 10m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b44e10d3b1a04051a44735e690195291 : SUCCESS in 7m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"022fe95468bc4f0674bcc345c7dfb11036ad8ce8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-20 17:04:06.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d3a05fa6a1f4f088f76d462b2c92de9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ce996b27aa3443d9bed2951ab79d5f92 : SUCCESS in 6m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6618243fae61412f917fc67df498ba61 : SUCCESS in 2m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/4529ce33e2a142a08cdad391591fe0b6 : SUCCESS in 39m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e647e2043e854f5dbef4b277500baa16 : FAILURE in 1h 02m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d94f5653b0d94ede95c32ccd391e498a : SUCCESS in 1h 04m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afc3addb45bb432cbcfa8ccb8aa3f2e8 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4848786e41a40c5b81aa2e8b99dda51 : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e8e785fa30342bab4f5a6cb86bd6662 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e0e914f9b264d569a749778ee404f32 : FAILURE in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85d5e050252c41549f040261adfa5d74 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18d5951fb1b74242b4faed2a6639a192 : SUCCESS in 6m 27s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8f95cb5b4f0e4992847c78357fb16d45 : SUCCESS in 13m 24s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/51c7ae5f75b24a569e990611e91c81a3 : SUCCESS in 8m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a3e3536beb784abca4cc1c8fd001eaa3 : SUCCESS in 9m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7892020c8c2f4199873ee24aec6d13be : SUCCESS in 8m 15s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/37f44412628d418b85922f544e868fa8 : SUCCESS in 11m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/81e31e4b51df49788ee29152188992a5 : SUCCESS in 1h 29m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fc207474586d42b8bc7204de221a573d : SUCCESS in 1h 34m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/503e303a017343439b0b34a62a1a7df1 : SUCCESS in 1h 41m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/60e8d17c40de4962a5d2b58905e6d5ee : SUCCESS in 1h 31m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2b870a6712184a7a8bb2c3bc822512f6 : SUCCESS in 1h 42m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca9b6abd79584f37922b959e42b6e2a6 : SUCCESS in 58m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6865fcb312a44c56aec18544a5c183be : SUCCESS in 1h 45m 15s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/17790a94f9a44416bdde3260e2c427c5 : SUCCESS in 1h 28m 51s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8b606b03b2904de49e8195b649cde560 : SUCCESS in 1h 00m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fdebeaaaa10c4fe682778d649d6f1607 : SUCCESS in 28m 40s","accounts_in_message":[],"_revision_number":8},{"id":"e1bd02682af982071b38e2930403fc418736ddec","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-21 15:37:04.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e025de6a63cba4dcd231afe66c6eb384cd8aab78","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-21 15:44:27.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75f3ff2d328d4fbd911b016cdbaca93d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/624f1b467a234f2a881cb22bf540f3db : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/915ecb15ffd14c5eb18dfbc24177fd3f : SUCCESS in 6m 38s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"caca85e0ccb0521bf44db31f22a1c4096d18ae7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-21 17:38:39.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da790f9bb5ed42beb00b0def7a925900\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b368adcb4d1943b1b0a5955c8e074f07 : SUCCESS in 4m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/92113f88554047f78ee648303428c028 : SUCCESS in 3m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/7b9636eae2774879bc48b374a77b29de : SUCCESS in 42m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/725211a61f064a9ea1df4ed9ba964f0c : SUCCESS in 1h 04m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b75a05794c64d1e8cd066923325044e : SUCCESS in 40m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c4b34f74df74dd0bdd57b1618108e48 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b1ea5a3c3694dd48f7efbae4a7f001b : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a57cc37abae649beb3ba7fd22da9faaf : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1a7cade36ac549b9869f705136b807c8 : FAILURE in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f279dfd92bd48a6adf89c326da2c9d3 : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34f4e945a9a9459f8a01f52649343941 : SUCCESS in 8m 16s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/cd0730303aac4714a7d751fea1e5f336 : SUCCESS in 13m 44s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dfd0fc81d3f54ab9a9c23e35340e23ef : SUCCESS in 10m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/93af4bbe5aca4309b8af8953ee437cc5 : SUCCESS in 7m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3713544031b64f43bdfcb1b1771553b7 : SUCCESS in 11m 15s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/44fd5b9b7c1746c9bdfdbf101de62d8a : SUCCESS in 9m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7affadff57c246c594bb45856b22a007 : SUCCESS in 1h 52m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/35587a10362145e09777a8dd9c40b4c3 : SUCCESS in 1h 53m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3a8901323f214da9a364e1686dad20bc : SUCCESS in 1h 21m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/97715e182fbf4d32abfc0393249c2b20 : SUCCESS in 1h 42m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fed904c6076b4981a7ea0b36fedb2eea : SUCCESS in 1h 40m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/176360c7fa9f4f24aaebc4010bd16a25 : SUCCESS in 1h 02m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/beab9b80a9a9473b82bf6ec011ed6e59 : SUCCESS in 1h 43m 51s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/4fe43213380540838b0082fc4353fdea : SUCCESS in 1h 39m 48s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a023dfa3acc04b15aed14c79384ba053 : SUCCESS in 48m 04s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6b7a274006474f949edf78148bdc827f : SUCCESS in 40m 35s","accounts_in_message":[],"_revision_number":8},{"id":"b3e3e99c645e0db0b36a49ce2d887c014e04ed7c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-04-30 11:26:59.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"aa8805edc683a0d8e5a05e7e6f1acc5f9e9f470b","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-04-30 11:34:17.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"42eb4bcfa9a8177a146a084bd660c512620dd1e2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-04-30 11:34:32.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4f1e6cdd7f5c8fcc2681aab675738f52c14aad30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 11:34:58.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"becb7e4feb3e24304d4e19644d5763d6beeff17f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 13:43:13.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fae37854651491fbb7433866235e157\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/726bcf9e1b024b9aa96e72db91222602 : SUCCESS in 6m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6cd4b61fe8b6456ebf6712b95d927a41 : FAILURE in 4m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/99b16ea335e94cd59205d1b214c53932 : SUCCESS in 1h 04m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6629ca30d56e4c088e1a6e6abd644006 : SUCCESS in 1h 28m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d106c1d35e6a4624bfe06f09b5f090ba : SUCCESS in 54m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/598e7cc0c98d4c3bbcc3bbc36105d6f1 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d47525e4a23b439f95a7aab8acb6f4ac : SUCCESS in 5m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c3de6be28afa42ceb40c600feca4d498 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/920bee213779462d92dd314877a04652 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe38e9cbcab9486ea2aaf4e2fe177e45 : SUCCESS in 7m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cb09cf85e2fd4b29bcdf68da7b322fbd : SUCCESS in 9m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0b58a81187e64d21afc0185f447a036a : SUCCESS in 8m 56s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/67dfcbb875394a8e90bfd075b6c2d90b : SUCCESS in 1h 57m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3bf6aed0a8204560bf297e1e40153f2d : SUCCESS in 2h 03m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/22384789b1f048e88ce69419dd4c00d7 : SUCCESS in 1h 45m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee33272bd6d44b5d999a4cc7d3d6ad9f : SUCCESS in 1h 09m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bb7f5f8ff6c1490784d7ce9eb6c17df5 : SUCCESS in 1h 57m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bdea2a03ce094480ab361b908f578c1a : SUCCESS in 35m 44s","accounts_in_message":[],"_revision_number":8},{"id":"ede1abf3473ccddd80da9ffce89a0727aeb19bb7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-02 01:13:43.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3b3ec1de2fbcb81eae6296f753722b91482312cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 01:25:29.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4651bf045f049a3880f08e9449dac1e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3beb2207e74545a59863f6d3a6188663 : SUCCESS in 10m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6e0fa0c616c04d73a446119900315b7e : SUCCESS in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"08a521317f579c3a29bddaa85d4cee7c682ccacd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 03:29:09.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a954d53b3044c2ab16fb8a0bc80e56d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d1d7018808414faa9c160b1aea0c39d9 : SUCCESS in 4m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1031d8d95c384e2f97b6dbe2cb92834c : SUCCESS in 2m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/c98b9d6ac7e841a3b28c76d93d5d804b : SUCCESS in 1h 01m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e8b89339b991489080476b9452ca6e81 : SUCCESS in 1h 34m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cfe3ca2143f4bd18118c3884bc8292d : SUCCESS in 1h 07m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7972e657b0a241e98ad11959c69aa13b : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b4f80d34c924ae7bd9e616b609d0510 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84f87748240c474b8f246d4113e8efda : SUCCESS in 5m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55e9bf159ee1432fb2fbf351f01f5ad0 : FAILURE in 8m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e48772c9062b403da021e130bf9d98c5 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6df808279da4b30a5cc4b2057f87878 : SUCCESS in 5m 11s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2013936c338d42d58ea41bcf5bf37ae5 : SUCCESS in 11m 51s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/06d781a1f12b4cc491f9c6a140f90a54 : SUCCESS in 9m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/00b16e8cd80d4f9fb2d939729ecb91b9 : SUCCESS in 8m 59s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/595603a8f93141eebb7c967439058448 : SUCCESS in 11m 33s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ada05205d0c441fca27c6ca23fd85032 : SUCCESS in 10m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8e3a4c1ba2fd42ff9ebd698ec086cc5a : SUCCESS in 1h 52m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e8dfc5f07370432481ad468505e55980 : SUCCESS in 1h 44m 31s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7f36645926e84cda89137b2b84c2ed01 : SUCCESS in 1h 31m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d6f095ac4bd1468abdaffc965d64b1e6 : SUCCESS in 1h 44m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2c1dcea030cd4fb682e3f376d08b0551 : SUCCESS in 1h 54m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13fe0c269f544f70b18885cb5dbb6317 : SUCCESS in 52m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c797853790b54331b82a89eeb719a625 : SUCCESS in 2h 13m 19s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/a7909bb81fd3424a93d943eb6a98c72b : SUCCESS in 1h 44m 59s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e88e3fdbc4a24293b0ff77f0b5a78d04 : SUCCESS in 1h 10m 52s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8974ae3598b946328dc04eb37201a745 : SUCCESS in 33m 43s","accounts_in_message":[],"_revision_number":8},{"id":"ad5388f6ea4f8371b66593988863f12bc80ed10e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 03:29:49.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"0c8146bd7ccc65e6ec8a6009c454bcefa966d1c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 05:20:04.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2503bfaf6bd648e9a0dcb75c4abca008\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bc64d1fbf3ad47df8e637e453111d74b : SUCCESS in 4m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a4bb491f18024fc0952b5d6998b2b9e2 : SUCCESS in 2m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/98c6e8147f0e482293c68ed9350e9bf0 : SUCCESS in 1h 02m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/30809c7c6a9c49c380e76a27dfc16536 : SUCCESS in 1h 13m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b616f9356ee64a7c914ab9c0b29cc7ad : SUCCESS in 55m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28935759ab834bcba310396dd3329fed : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/061c764cf5514bbaac26a131950e4062 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27f6f80b6390419e92f2ad1c78238411 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58209a2fbd0d4335aa57371541743fa3 : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e17cff3171794244ab19efa855e3dec8 : SUCCESS in 5m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ef605781f9a34e7e9c0d779f57949eb7 : SUCCESS in 9m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8dbf99d763d14d229134b084d85f7c9b : SUCCESS in 9m 47s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/89d85587a19e4646827eb2e4854ec5e7 : SUCCESS in 1h 30m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/89a193e40cbf450fbf777d740fe37105 : SUCCESS in 1h 37m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5e1c45cccf664c39b16afe178f9527a5 : FAILURE in 1h 39m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd427279be2d449ea71640031855cc5d : SUCCESS in 59m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9fb73d4eaece44f6ac88cd43a5b0e5ca : SUCCESS in 1h 25m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9ce986a3b9724772aca8f4375d8ea879 : SUCCESS in 22m 49s","accounts_in_message":[],"_revision_number":8},{"id":"0e67d950dfea60af4751fbcba319f8211839c13f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-02 06:12:51.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d9e7bcba253fcfb07f707d4f42a45a77d1267818","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 06:19:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23b8a603cf9b48e79df6fb7d1079a193\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46147808bfbd4a13b1da1e61c868dfeb : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c80624c7beeb4b3db6a5d5b1600f644b : SUCCESS in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ffa343844d3f3031faa209c89a497e6ed2da8107","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-02 07:16:13.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"18541f90cdc845b9bb965ebdfaec460662ad1dd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-02 07:17:15.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":9},{"id":"d1f7bbf5a1762cd3a976665a594c8dcff1513e1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 07:23:30.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87447d852eb8498d836a8228fc6352ed\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2e5ac705c96f47c996a4b02a3fdd0d4e : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3384ee1a16ea4def9d33c2058d9bbad3 : SUCCESS in 5m 29s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"aa0c4c5b2b2bcdf26446e14b89b06d990dd519e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 09:20:18.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62d64c14273a4c8ca9d5aa1ca65f195b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f05c216e33e14488a9548a82b1569161 : SUCCESS in 3m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1b3d525308b942cfaabb930b8e355040 : SUCCESS in 2m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/a9361a01f3d34707bef7f361b2e75ef0 : SUCCESS in 1h 02m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/24a123cf14b74ac8b3b9bf05b0fe3b65 : SUCCESS in 1h 11m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b73e79964f049d5a32d7923219c4c7c : SUCCESS in 59m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a01268c8074e4fcc9ba12f329d4523ec : FAILURE in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a0a94c366a748168e13d4c5bb4bd9ec : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cfe7c9989ce4495aaba25a424921175a : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09212f8dc8744726bb3aceb42e3efa1a : FAILURE in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e9213b9b86c422fa439f6cc878909ab : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0d0a3075d094c49b90cdb7ca7e1d2ec : SUCCESS in 6m 14s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9363b6c4c63846f0bef0f0aae1e76fac : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f06e69e75866497fb4f7d6c884494d73 : SUCCESS in 7m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e328f84977dd4670947153b58324ed09 : SUCCESS in 8m 12s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/73bb3a31e875405db8f6f88e94ff01fc : SUCCESS in 8m 31s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/00d8b473585d477cbf85b0d9b5f7112b : SUCCESS in 14m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fd97346f6a0b474f869df9f4ab7e8d89 : SUCCESS in 2h 00m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1ec051789e144d90840f421b4df4afe4 : SUCCESS in 1h 44m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/509f7f57b9ec4b2d98ba221ab7ff8970 : SUCCESS in 1h 44m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/76c0ffdf72024ddb91fae2fc6a4febec : SUCCESS in 1h 45m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dd52e6e9764f4450a132bc9aee3b2265 : FAILURE in 1h 37m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3feb1e5e7131440d97cdd5953a276ff2 : SUCCESS in 39m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/61435deb44e84e7d91df54e5bfc98852 : SUCCESS in 1h 58m 05s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/197c470c28fd499b9f3c0177d25fbd64 : SUCCESS in 1h 15m 00s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/642e9bf336f74523b41446cefb09f6aa : SUCCESS in 1h 00m 52s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4e5c080958f14da8b05a2f1346e91338 : SUCCESS in 23m 54s","accounts_in_message":[],"_revision_number":9},{"id":"61f279eaf6da8263ca5320bb9d88350e6359ec63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-02 09:21:00.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"14a5cd6eab515541ded4fe813a30714925b5e3a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 09:33:35.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d12fccbc952e48d39dfa6f99fcd353b0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bce9ff4b54944699aca7d7e9f59143c8 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/29eab57065e245a1988b60e1314d894b : SUCCESS in 7m 10s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ef59c89accdefe10ee438d122c5d4a45d0dee90a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 11:43:04.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1656b3144c054c68af13cb966ef47afe\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1ef64caf4416429299731c5bb0576e22 : SUCCESS in 4m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/acaf2166aeb746a68d17bcd0e9e9494d : SUCCESS in 3m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/a3afdcb73de84025af9142f86720c799 : SUCCESS in 1h 03m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a5b7e9faa2c490b9cf42ebea3d8a35d : SUCCESS in 1h 21m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22985abc0f374b7c93308ccc2f5a96d0 : SUCCESS in 59m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43759fd992f940689f2c73b7a41ebdef : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d7ad583fca04ebaab75c61f38c3afb4 : SUCCESS in 7m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e101659f8e949dca57bf7e1fcb1da6c : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b862362e9778462192a2ac2bf4a67620 : FAILURE in 10m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36b11bec80d14435b5d3226a79197b30 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ba1b44e7ca141408720952ce753ce54 : SUCCESS in 7m 46s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/31489b5b3fb54023a80cd00a46158fcc : SUCCESS in 13m 28s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/39d2afdaeda746d9a5557431b6ac4aea : SUCCESS in 10m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/682b4325fd294342a70fd3108d4484f2 : SUCCESS in 10m 23s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/35f8529d09b84e9cb8ca1033fca7c778 : SUCCESS in 9m 17s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/95953ed5432245c4b7f48b810c63cc67 : SUCCESS in 13m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7eabf263fe4146838d3a6f197e3944a5 : SUCCESS in 1h 44m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/259b74e65e604f67bb6ffcd96c1c39a6 : SUCCESS in 1h 20m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5d678f81fbb646c782f826a62c11edff : SUCCESS in 2h 00m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ec586f75348346f7a9cc514f0c627dc3 : SUCCESS in 1h 41m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ca2edee3152a4a74b67ed7a9dae510e1 : SUCCESS in 2h 01m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56c838b074204a779e97de60dd5fc74b : SUCCESS in 51m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/748567ca865648d88bf7041ab5444ae3 : SUCCESS in 1h 48m 49s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/0f53db29feb94681beef3e06de1f2530 : SUCCESS in 1h 51m 46s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8bf929db987c4b0ba9f162bce29cc0c5 : SUCCESS in 1h 14m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/735d6d99154e4ae0ab65552675fd87b5 : SUCCESS in 37m 30s","accounts_in_message":[],"_revision_number":10},{"id":"602e2a8803f2ad14afe12aee853eba6e72a21a31","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-05-09 08:31:18.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"31181e5b50c5ce421060eafe62690430cf8a35e9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-14 05:11:37.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"b0bba38de15ad041d89f3c7f351afd5b2d2f25c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 05:12:11.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"7cf0e658c00ba14c84b3fa5ccf5e8086b98b005a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 07:36:46.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0355a85f577542a6bbc5c0cff22f2e31\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/61654632b4ea44048f3e1a7c5f7e00be : SUCCESS in 3m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0726e60c5e194faca26c8d4a007fc081 : SUCCESS in 2m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/63f964c4592147c789120ffe23273349 : SUCCESS in 42m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/52051721e1414365a31bd5d9474fd66a : SUCCESS in 1h 47m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca4723f6902d4f4897ecf86b4e7f5da5 : SUCCESS in 1h 02m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/566383cdf3724d0586997959f0eb1029 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5235e4a6c2b4c4ea5f3e5a931feaae6 : SUCCESS in 6m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3fea7ca9145b4197afc5ce60163a0f79 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c51acea737e74e7ba836eccf49a57000 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02fdf20a396c40afb303447f813965fb : SUCCESS in 5m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/767d4ad4417f45f9810f8f320d89d7ec : SUCCESS in 7m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6e5b6e3cdde042f79069059b7633dd04 : SUCCESS in 7m 20s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d9186462a41446608dd6a25c6c2ae259 : SUCCESS in 1h 21m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/80a9021aa6ca4818a15d24e8886cb767 : SUCCESS in 1h 19m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/751fde3748f246b48557516385189586 : TIMED_OUT in 2h 04m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0eae7f07a294476a792a974ee0372a8 : SUCCESS in 41m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/47337db42cdd40aa820b144ac88c3b5b : SUCCESS in 1h 47m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/70388e88634745c7b79f029415c63cf0 : SUCCESS in 23m 51s","accounts_in_message":[],"_revision_number":10},{"id":"8ffe695be862b040ed349e0ddf9cc6b50e1069c0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-14 07:40:24.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4aac6d104ad255a3a354cdf8144bb75ec4c44227","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 07:52:17.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e79e82fcdf81433d8cd0e2177b58e3ed\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/09e76803eb88415d982b044d31ff6fa4 : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e5dfb66d06a14ac0836a76a52c20837f : SUCCESS in 5m 44s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6c64e1042dc40a17e81aa5d747803b3355e02107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 09:38:28.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4fdab09376145cc9e63ca70ccfea340\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cde669d49e1847eab1bf098585bd8ddf : SUCCESS in 4m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/da2cbd98f2ef482b96b623d65e44c682 : SUCCESS in 2m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/c295806ab15f4a15afc891d57db3e753 : SUCCESS in 1h 09m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/152967d411fb45cf9038da01f71ef5b1 : SUCCESS in 1h 25m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53a44b8f9d8e4ed0b561025b632a17d7 : SUCCESS in 1h 01m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4d4ca87b74640bea94edc4964f7be22 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fe7326a051e47f7a8adf548d7fdc6a4 : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5dc689ef05a04165b1054c58ec6e8816 : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8b053c70b6f64d61bfeef4b39f704fb7 : FAILURE in 10m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9839acb896b4913b3a9ddee8343d55f : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cf56f2e08e54774bf402167658e1ea1 : SUCCESS in 5m 16s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/6b6384bc2b6b4c528033ce9115b12df0 : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/419ed7c0e8af4e0aa319f056fce470c2 : SUCCESS in 7m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/042b54c11c6a424389b55f99b4bad1e8 : SUCCESS in 8m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ea38ccca8638490d9bd153d08305a571 : SUCCESS in 9m 04s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5c3e63710a0348feb34f175e857a7af4 : SUCCESS in 10m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/473040cc97ce47a2a1a9d48f6fc0b400 : SUCCESS in 1h 52m 47s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/90c7c47206b04399a8aec90a2f4820d0 : SUCCESS in 1h 33m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2c6ee3a979df497b9fedd75a96485156 : SUCCESS in 1h 19m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2c866b1547e543b4a7c71d26d0e4803c : SUCCESS in 1h 43m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/63cd8135f43f4b5a8c6e1e94293c6aed : SUCCESS in 1h 48m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5301bf771e8e431a8873bf83e110858c : SUCCESS in 57m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/20d22b5a37fd4315ad0447a6ebd5ace1 : SUCCESS in 1h 51m 40s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/9493250b81ff4d7fbc0da697b66054e3 : SUCCESS in 1h 44m 35s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1734a9bf7b884f61b7a81ebe7bb77a3a : SUCCESS in 1h 09m 28s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9a89b139ff574b1e8bacaa1fbfb47021 : SUCCESS in 33m 59s","accounts_in_message":[],"_revision_number":10},{"id":"cdbe23ddb6eed7c6b9ba7498ca560bd681e2db4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 09:39:12.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"d0e4bb6b144eb05c75e9bf2d3a2eadc8d83f0d78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 11:46:10.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38780e1955a549d69c50319812ff1475\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0377849a2e7a4ca3b01ee70f9ed8e5b5 : SUCCESS in 4m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5e1a86cb2ca44efab6be3a76761c0450 : SUCCESS in 2m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/c874759e71de41b2bb52e1719ca3356c : SUCCESS in 1h 02m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e4fcc60e5dfd46f0a88c7839b7a8398c : FAILURE in 1h 20m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d86a081ed72b4d52ba22be0003df72a7 : SUCCESS in 1h 03m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a7c5c4b89dc4a0ab29fc81700007a87 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/591e4820e6f14c6797914fa2a95304e4 : SUCCESS in 5m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8371461b03ba47da9d6a7bf882d83570 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a3c0c919ae425fb53077a98b5302c2 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b98ddcc126d1497a8fe4c02c6b3d5d16 : SUCCESS in 5m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b6cdc173cc014574b10c6625c1d1e7ee : SUCCESS in 9m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/061627d6d956477e9bb0d05af5cfcb44 : SUCCESS in 8m 08s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/89b016b0b95040a78d6b59e63229a1b0 : SUCCESS in 1h 51m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/891e3906a55a4336b0808a4fbd951ab8 : SUCCESS in 1h 31m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/59d4aae039244f64afb69fff149fffe0 : SUCCESS in 1h 23m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb7d05af7f824eb185b67186ac6d6d1b : SUCCESS in 45m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f8f93abd1973457ca60f1d3dd23d1c60 : SUCCESS in 1h 45m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0d2993fb608a40f9836e2ed5038fbdc8 : SUCCESS in 26m 55s","accounts_in_message":[],"_revision_number":10},{"id":"9930fae070611cd0c25bdbc00ffe51edadd53eda","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-14 14:08:16.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"5a5929802f8d506d55d2afab37d7922af1afbb4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 14:19:51.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/723d5329eae24ba0807004b36d3eea1e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e7820ff5d55e4f78b31554dbbc870526 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/664670cb93c74c4ba0c7731f33e0ee74 : SUCCESS in 8m 31s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"66e456d46073aea4c153df13fdc17c2d8f979bd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 16:18:27.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e398d71703c4de2b333a4a9ab1673d3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ae88e8ae47e24221a0a44d4a51ab84ba : SUCCESS in 4m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0a4b4f165e834f20adc827489f8bb49d : SUCCESS in 2m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/7b40dd9bf0aa41ce8807184daedfde98 : SUCCESS in 1h 10m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5e8a2b1527404dfd8d96ead8e01b53a2 : SUCCESS in 1h 29m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cd27567dcf54fb1bbc3d3ca8c49876e : SUCCESS in 49m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae85514125fb4058a670564dc821196b : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/725f282e43574e56b807dc8ff54024c0 : SUCCESS in 7m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/da8ef8abb1fa410eb91ff1b44f4db5a9 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/39cbefec95bf43cba5828cc624f509c0 : FAILURE in 9m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a704544880b84c6abf510bbee6fb50b8 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3318ad7834e04505a5f3778bd98bf128 : SUCCESS in 6m 28s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7cb9206cbcf7468aa8a952ec9329c8b0 : SUCCESS in 13m 59s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2fcfa1490765417ba453a2c41626ef57 : SUCCESS in 9m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/70706d8d4c0643fda8f6b1c5c712d281 : SUCCESS in 9m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8c541e0a08ef4ec9ac2468d6c3fa335e : SUCCESS in 9m 26s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ed37ecfadcab460d8b708feeda2f389a : SUCCESS in 1h 50m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8b5670656ab64faeb90d6a3dec93cb1f : SUCCESS in 1h 54m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/da0c653711e246e7a6d1480a3f4c0776 : SUCCESS in 1h 36m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b85087c5d12e4b78a5c3b43e1cb5412e : SUCCESS in 1h 40m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/69775b319ef1465cb65e2b1bb2b1f691 : SUCCESS in 1h 42m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4bacc8477a914f2b9b29955240782407 : SUCCESS in 48m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e1f01ab62c504693842e600206a66dc7 : SUCCESS in 2h 00m 30s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/752683775e1e4fb1b01f58c3132dca6c : TIMED_OUT in 2h 04m 32s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/57746f49b90f43e19525a3ab91c61d1a : SUCCESS in 56m 59s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5b3588951eb74eb4ad25ec9e9c9bfe63 : SUCCESS in 42m 13s","accounts_in_message":[],"_revision_number":10},{"id":"ce47a0b833b927aa047affac9d773ed83e3e92fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 16:19:35.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"2242017e9c2690dd67efeea29049f49aea88a313","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 19:11:04.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee9ffd35610c482d95480ccd77452cb4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ed0dd727641c47519e0e5d64277ef726 : SUCCESS in 4m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3c53bf951dd5414caca2bff671e9c295 : SUCCESS in 2m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/6d1551584bb540d488bc119a120ba16e : SUCCESS in 1h 05m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5ba0d80547fd4cf7a172b1b9bd19dbe2 : SUCCESS in 1h 34m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d25dd0d3a3a647348aacb0cc7f5cc3f5 : SUCCESS in 1h 20m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba8a668e5aac438ab8b9c18974c8744a : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/668034147a344458bbbe9742a44391bd : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b81cbc95532482c8e56aba65c06040f : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c5c705ced1b4ace81f04c94c8acb3a4 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c96df35f6984e9497c985293e67475b : SUCCESS in 6m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/defd50aa40af46849b44b62ecb5b78c0 : SUCCESS in 8m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9f9b500ba38f42cdb78779a3a614ce99 : SUCCESS in 8m 41s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5f6fa22e14f249799f7b301d8a837f46 : SUCCESS in 1h 39m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c588ddce15e6481383d4e3315ab3789d : SUCCESS in 1h 17m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ce937c8a76ea419ba79112512ba064cb : FAILURE in 1h 47m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e3b447242a149b78b718d31489c25ad : SUCCESS in 57m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f26b3c757f37474ab04aec48862d5ae5 : FAILURE in 2h 13m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d4e5902084e4402589c5bece3c46c9e7 : SUCCESS in 33m 26s","accounts_in_message":[],"_revision_number":10},{"id":"49f7836e83e4c32d91800ead6f2a5b8e3c71d57f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-15 05:25:15.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"763f5010528e5bc9e19dac3251feddece976a8fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 05:38:34.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d900f789e324faaa46eedea70344b45\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/494e2e02bf4b44c3ac92fb5e84b8aa43 : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b1ee1c68ed14494aa13d235a0fa6563d : SUCCESS in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7c6556ea7f390492b33ae64c804656274a588d99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 07:30:02.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98c9d4efaada4923b0e7c291933d8471\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6e713ce97078445999625006f8a56dc4 : SUCCESS in 4m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cae599dfa9b7456dac790db745458fcd : SUCCESS in 2m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/917b721a85134000bed248b8db2e2d49 : SUCCESS in 48m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/be5680facd7e494db19c5b6120e8b611 : SUCCESS in 1h 36m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04f523d7533845eaaa06bdccd9fd2c06 : SUCCESS in 50m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43ce3553a47e40658b1c6a12179a3585 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55e035738eff4127b00f568b1533b72c : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a8f448fc4dc49c2ba71edd004fecead : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d594db25257470ea45b633f714adbea : FAILURE in 8m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d19a87d30194cf09631c51bd49d8f8c : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f09f5abe8124443adc3171bb0623fd4 : SUCCESS in 6m 50s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/65dadef649a044b1bfeb2f3c4ffd71ae : SUCCESS in 13m 08s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/203d6197198c48a7b43ce1845c24473e : SUCCESS in 9m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/252d91032fcc48ec8785d3622870e826 : SUCCESS in 8m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9702df4f6a644b8ab42d78f019e90893 : SUCCESS in 9m 00s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4f734c6bcf99473aa08dd47d22350252 : SUCCESS in 1h 56m 27s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e1d22dc6b4c5421eb8598c37148139be : SUCCESS in 1h 18m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f64e630e222541e8931e89915b710744 : SUCCESS in 1h 25m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7000dbfa99504e828350858d5c55302e : SUCCESS in 1h 17m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/22eca5acfbd8470ca59932127788049e : SUCCESS in 1h 32m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bfc5230bd6144a6936fed5c7e3f644a : SUCCESS in 1h 01m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5c1c904905614abdbc7c57090c26f8c3 : SUCCESS in 1h 52m 41s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/7d7991752f0544e784e4b058c43f441c : SUCCESS in 1h 41m 23s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/55033fc20f40480c836fd2e6a06bc19e : SUCCESS in 1h 07m 56s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/066544da64c04fcba08f339fa9f63ede : SUCCESS in 42m 30s","accounts_in_message":[],"_revision_number":10},{"id":"0918a84c6ee7fc35ed3d65951b05d8cb48825ca9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 07:30:37.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"db6755e52e2c6066476864164185ffc6bfb9d8b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 09:33:11.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf9ee640d7b34070b35625ef44bb4ec4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/41c6a1a33ddc43acbcf35180fe24983e : SUCCESS in 5m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8d781183fd6b4f45b74dba5ed5635cac : SUCCESS in 2m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/5bd44117c6634c4ca9059e90e17e7f79 : SUCCESS in 1h 04m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2c461ec40c874f4ca10ede922ff655f4 : FAILURE in 1h 08m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f86d3b9728384ed586b8e1ff9aa430d0 : SUCCESS in 1h 03m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aafb34ea0acb4845a8e92cdf563188f5 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28ce92896d8c48d882cac284046902fe : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9256a91a3965498cac2f3555fc027b71 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7f388aeed904e46a7f5d30b026e7e29 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b0ef743ca04436193d78326870627b6 : SUCCESS in 6m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6ab5957142ee49bea941481c348ac500 : SUCCESS in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/95c22639e7f94ffd8a528f3ac69f5d78 : SUCCESS in 8m 49s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7d0efb93c6fc4415874bc9879e619c59 : SUCCESS in 1h 57m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/63a7aacad9374e7fb5fe5c61a6404fd9 : SUCCESS in 1h 34m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b9762d4d8bb74c7ea720188d3d1ebd8d : SUCCESS in 1h 52m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd91a88714374cf59e54228ae66376bf : SUCCESS in 50m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ed3860ba2d484426938ecd404df5979b : SUCCESS in 1h 50m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4d9012a137cb4763b6138eb3c87b7889 : SUCCESS in 26m 33s","accounts_in_message":[],"_revision_number":10},{"id":"936ab2afaabcf2722fbeb78bf69a176b6443a05b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-15 14:12:24.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"9ec1bd70dd4183a7e4db154727e754c98d7d5f13","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-15 14:13:20.000000000","message":"Topic set to remove-pysendfile","accounts_in_message":[],"_revision_number":10},{"id":"ec14c1cfd44cde6be91b5f67d539c0649e5e7c02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 14:18:42.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d48e8f5e42cb428eaf0a5de4b0740af2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5a10d77d639497ea80a257765d79fd4 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1ce8649eae15472bab9054e1e7ea5b68 : SUCCESS in 5m 04s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c5ee98d6bf6d1996e7f7b0663d3c08dd264f0bea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 16:27:13.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d43681eed3cf4cc085316ac50f491078\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/70cf1514256c4feab379dad99050f56d : SUCCESS in 3m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6855b9823f454692832abca407bfb46b : SUCCESS in 3m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/ce253f69f081427daeb5d732ca06cc80 : SUCCESS in 1h 04m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c5601bc70b124d36a4846cf7c6e3cd0e : SUCCESS in 1h 34m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8991b1784e3d4e1ba83abe88fcae67c6 : SUCCESS in 1h 02m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72694dc8209d4887978fb4a0a1640149 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db790a7892f64e24a6d89d91a99de816 : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e96138452c58434085eff8bd14c5cdf7 : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/05680e36b0a34f388482b9d8f240bfeb : FAILURE in 11m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98dadc780b634b488560cc3e49e42e18 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a04703ba66a431e87d28e05a05a0e53 : SUCCESS in 6m 51s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2e9707450670482998efef03b61df8ef : SUCCESS in 13m 02s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/876d47de6bf44d79b5eb939dbb683f26 : SUCCESS in 8m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bd0ff190e30041819d7c94e648a8b730 : SUCCESS in 8m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/70b6cc7c917b4d3098d06dea6a736b24 : SUCCESS in 9m 42s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d38a27d4afcc4aff8e9b9a6d7c2ce073 : TIMED_OUT in 2h 03m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b777ef4ee39f45fcbbb65d184e8d0462 : SUCCESS in 1h 46m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d99076d2e846400294257192b989a3d0 : SUCCESS in 1h 54m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9779820b7c804728afd5de76b4f75c66 : SUCCESS in 1h 49m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8d77519c5046422eab017df3b9f20c0f : SUCCESS in 2h 13m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0e958eed7ae444484eb7a5ae9fd4823 : SUCCESS in 1h 03m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cfa0e6bbf8a648baadf407423d04f9d0 : SUCCESS in 2h 08m 28s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3676bcd498194d72856f33ce2ea42c47 : SUCCESS in 1h 58m 40s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/0f4a50e19e224d959c254cfdf41c30bb : SUCCESS in 1h 05m 06s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/590516f927914e63ace1ca38e27263a2 : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":10},{"id":"4a1ccfd67f2629a3e5da139093a1ebf9df6b49e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 16:29:23.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"1867c86fa218ad53c6825eeae74324f152a4a857","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 18:24:15.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c77bc31b6a504771a490de758024e955\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d45a56190bc740ec883158c17b192516 : SUCCESS in 4m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4beec7e9e79b4003921a613a236d272f : SUCCESS in 3m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/85a92f3aeb5340c4b712c6d62e55d347 : SUCCESS in 46m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/581258b88475472db6a5a3df9eefa311 : FAILURE in 1h 08m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e356c3b6c3343228febb7aa9108d353 : SUCCESS in 54m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72e0eab12e9c4660908b4b57158d5e89 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fde25de02c64c41840076742c866243 : SUCCESS in 5m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a190a9301aeb41d4b539016d80e56fa1 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e94ac39db2ed4a4ba135f28684b8be2a : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f91c2c1d77d045da9363006558dddddb : SUCCESS in 6m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/082019fef8df4eb49625bf0cdec50c5a : SUCCESS in 8m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1a018d9faa8640ed8ed9d7733038e74a : SUCCESS in 8m 07s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/34d6381b362f4d5d9abb3502de673bf6 : SUCCESS in 1h 50m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0e97f031d85941d4bcd206b4131775c9 : SUCCESS in 1h 42m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c0cf6d26f4d74d41940e36ed62be2caa : SUCCESS in 1h 19m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9db987044f7a4926b8d7d748ba37f57c : SUCCESS in 39m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bb02cb6c97ef49739ad5d1f68ff1f2d0 : SUCCESS in 1h 49m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d1dd0ae4205b403499237f429b64302c : SUCCESS in 25m 23s","accounts_in_message":[],"_revision_number":10},{"id":"2bd9eab80a44ef9ab95df2d520be26c8e3918084","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-15 23:50:30.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"49a1eb9baa1e8268158df1c969ad17429bc85f22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 23:56:29.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/511d889b1bda42e39c94f4550a0e8b68\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32bd1faf71e5476f8c3f8ff38fdc84dc : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/37ccf438b552494eb9c392bc328ab711 : SUCCESS in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ebbd4ea704d96d61fe477f9d5909e2cc1ac81788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 01:57:10.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4169b2b15ead4e2c9646e711c81b7106\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/025fd663ced4462cad57bd6e09092609 : SUCCESS in 3m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c69e0d66d7184a9bb50ea58d174ef62e : SUCCESS in 2m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/23fe7acfbe3f436ebbad642912acb98a : SUCCESS in 45m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fd80309f90744b988f0461fa59130927 : SUCCESS in 1h 06m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbdfc1ac274b431590a308a55cbc1543 : SUCCESS in 43m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b745ef193ec476c89ec5ad107bcb2de : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e647c3f000441e29f72bf1229450f96 : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e46d90b332e44b9190c503104a7d5d14 : SUCCESS in 6m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4749c902f3b84f26b4cd5e31ea02bb70 : FAILURE in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60fd0344f5a543d98e7e08a0dcaf87a3 : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbea71646d0f447d8dc7e7f90fe95630 : SUCCESS in 6m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/6160d9cf19c24bd7bb0ca53ad1bc8ff7 : SUCCESS in 14m 21s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0c2b285f262b4097bc085682bafa65be : SUCCESS in 7m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/14ea4dc2a89a4a51b9ea086e6b943668 : SUCCESS in 7m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/43655a7aef374317ace2ad52cf7e8be7 : SUCCESS in 8m 20s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5f26f5ac07b0462eae2cec7a08bb749f : SUCCESS in 1h 53m 48s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/72bfdec8928047e79957fff8d501037f : SUCCESS in 1h 55m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ab3b0f1c8e0f46f886bb992d366e91fd : SUCCESS in 1h 53m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ede118405d524e46b5ac6996d7608070 : SUCCESS in 1h 40m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/809a6b4185a84b98aa343afdb2a509df : SUCCESS in 1h 53m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da9436d0726c41289ae7cbc777724025 : SUCCESS in 1h 02m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1ea4b613c8cd4909a65f2df2cbbfe25d : SUCCESS in 1h 41m 43s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/c80270d2dc2e4630883089d31547629e : SUCCESS in 1h 49m 19s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/83817632b2bc40f7a6d5cda3bd10d096 : SUCCESS in 1h 00m 25s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/84e2983ada354b32a4b1d2ef31ae4ba2 : SUCCESS in 30m 40s","accounts_in_message":[],"_revision_number":10},{"id":"fe10e28e4b5e3c027a15c5350c4707347f6ee7d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 01:57:46.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"7944d4ac4c304b7b106719a06b813df7fda38637","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 04:02:31.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25763ebc85b2415e86c171a8d0d0a38f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/665065cefbb340fea2eb08560680dacf : SUCCESS in 4m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6ae6bb7ec42d4ae6b2b5e7f427275563 : SUCCESS in 2m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/83276d4a67e34389993a216ef3544119 : SUCCESS in 1h 11m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f019479f4ccf46c39e6878d61a3529c1 : SUCCESS in 1h 32m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e37d2f6f17514e6798ed4c6bdefa4f04 : SUCCESS in 1h 03m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/758435f9033447f9b70fc09ac53f0887 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bf5498e988c4f5989fc331f66c2736b : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6c30641982048f79b4be3369521086c : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a66dd784e2984792acb8c8b6e458a5d4 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/055018886d3b4c87a1ecf7cac97b9b5f : SUCCESS in 6m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/25cfa3e59bc84a9cbffedbef702734de : SUCCESS in 9m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/716797d8288e4eef87f23b7465b47c62 : SUCCESS in 8m 25s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5d8387bb64394140a0ad28955b8f6e2a : SUCCESS in 1h 27m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7032ff44ef9448e695ba9bf5fae70f69 : SUCCESS in 1h 49m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f140db7c388042cf8a545b4685968057 : SUCCESS in 1h 55m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8762073290794ed49455ed100a511a75 : SUCCESS in 1h 00m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2dc0ed6b61124416bf149b51ed0c75ed : SUCCESS in 1h 57m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/76bf3e104028451a8ec5df43aa09738a : SUCCESS in 29m 57s","accounts_in_message":[],"_revision_number":10},{"id":"ec55887b337eaee45bbda04682727b3af9926ae9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 04:02:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"bfdcedbef1847388f17b073f8865cd1455f541a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 04:04:15.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/100b14c80b63493685d7e4b2fc762742\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/608284be98e547f3ad029897c8326a1f : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da2811c679794cff88a48763ffbad93a : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2925598b53214251bed889cb45ae22a6 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"205ca3336a7646ad3fbf92d8cf2509116a10bfcf","revisions":{"5d111385e320540e59262e73a86fabf3d485554f":{"kind":"REWORK","_number":1,"created":"2024-04-19 12:27:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/916422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/916422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/916422/1"}}},"commit":{"parents":[{"commit":"0bcd6cd71c09917c6734421374fd598d73e8d0cc","subject":"Merge \"Docs: Add info about the cinder store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bcd6cd71c09917c6734421374fd598d73e8d0cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 12:26:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 12:27:21.000000000","tz":540},"subject":"Remove sendfile support","message":"Remove sendfile support\n\npysendfile[1] was added as an optional dependency but the library\nhasn\u0027t been maintained since 2014.\n\nRemove the implementation dependent on the unmaintained library.\n\n[1] https://pypi.org/project/pysendfile/\n\nChange-Id: Ia4784f59d16660e8d40c0e409f092ac4e46870b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d111385e320540e59262e73a86fabf3d485554f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d111385e320540e59262e73a86fabf3d485554f"}]},"branch":"refs/heads/master"},"bf0d134b978ec60cacead1f15d5c50461761e58c":{"kind":"REWORK","_number":2,"created":"2024-04-19 12:41:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/916422/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/916422/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/916422/2"}}},"commit":{"parents":[{"commit":"0bcd6cd71c09917c6734421374fd598d73e8d0cc","subject":"Merge \"Docs: Add info about the cinder store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bcd6cd71c09917c6734421374fd598d73e8d0cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 12:26:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 12:41:08.000000000","tz":540},"subject":"Remove sendfile support","message":"Remove sendfile support\n\npysendfile[1] was added as an optional dependency but the library\nhasn\u0027t been maintained since 2014.\n\nRemove the implementation dependent on the unmaintained library.\n\n[1] https://pypi.org/project/pysendfile/\n\nChange-Id: Ia4784f59d16660e8d40c0e409f092ac4e46870b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf0d134b978ec60cacead1f15d5c50461761e58c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf0d134b978ec60cacead1f15d5c50461761e58c"}]},"branch":"refs/heads/master"},"d8da4e0ce252fdfaaaea9a184ef815ecefe26ad9":{"kind":"REWORK","_number":3,"created":"2024-04-19 15:08:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/916422/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/916422/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/916422/3"}}},"commit":{"parents":[{"commit":"0bcd6cd71c09917c6734421374fd598d73e8d0cc","subject":"Merge \"Docs: Add info about the cinder store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bcd6cd71c09917c6734421374fd598d73e8d0cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 12:26:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 15:08:29.000000000","tz":540},"subject":"Remove sendfile support","message":"Remove sendfile support\n\npysendfile[1] was added as an optional dependency but the library\nhasn\u0027t been maintained since 2014.\n\nRemove the implementation dependent on the unmaintained library.\n\n[1] https://pypi.org/project/pysendfile/\n\nChange-Id: Ia4784f59d16660e8d40c0e409f092ac4e46870b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d8da4e0ce252fdfaaaea9a184ef815ecefe26ad9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d8da4e0ce252fdfaaaea9a184ef815ecefe26ad9"}]},"branch":"refs/heads/master"},"628f8d5bd07a6a90a10baaa2f46d1454a0f3a5c0":{"kind":"REWORK","_number":4,"created":"2024-04-19 15:15:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/916422/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/916422/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/916422/4"}}},"commit":{"parents":[{"commit":"0bcd6cd71c09917c6734421374fd598d73e8d0cc","subject":"Merge \"Docs: Add info about the cinder store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bcd6cd71c09917c6734421374fd598d73e8d0cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 12:26:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 15:14:50.000000000","tz":540},"subject":"Remove sendfile support","message":"Remove sendfile support\n\npysendfile[1] was added as an optional dependency but the library\nhasn\u0027t been maintained since 2014.\n\nRemove the implementation dependent on the unmaintained library.\n\n[1] https://pypi.org/project/pysendfile/\n\nCloses-Bug: #2062573\nChange-Id: Ia4784f59d16660e8d40c0e409f092ac4e46870b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/628f8d5bd07a6a90a10baaa2f46d1454a0f3a5c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/628f8d5bd07a6a90a10baaa2f46d1454a0f3a5c0"}]},"branch":"refs/heads/master"},"8cc23d6d136d4390bb9df6aed71e14e63f47de9e":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-04-19 15:15:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/916422/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/916422/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/916422/5"}}},"commit":{"parents":[{"commit":"0bcd6cd71c09917c6734421374fd598d73e8d0cc","subject":"Merge \"Docs: Add info about the cinder store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bcd6cd71c09917c6734421374fd598d73e8d0cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 12:26:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 15:15:16.000000000","tz":540},"subject":"Remove sendfile support","message":"Remove sendfile support\n\npysendfile[1] was added as an optional dependency but the library\nhasn\u0027t been maintained and has got no release since 2014.\n\nRemove the implementation dependent on the unmaintained library.\n\n[1] https://pypi.org/project/pysendfile/\n\nCloses-Bug: #2062573\nChange-Id: Ia4784f59d16660e8d40c0e409f092ac4e46870b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8cc23d6d136d4390bb9df6aed71e14e63f47de9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8cc23d6d136d4390bb9df6aed71e14e63f47de9e"}]},"branch":"refs/heads/master"},"2740e132b744f30d787c810965162e35a77e2391":{"kind":"REWORK","_number":6,"created":"2024-04-19 15:19:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/916422/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/916422/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/916422/6"}}},"commit":{"parents":[{"commit":"0bcd6cd71c09917c6734421374fd598d73e8d0cc","subject":"Merge \"Docs: Add info about the cinder store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bcd6cd71c09917c6734421374fd598d73e8d0cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 12:26:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 15:18:02.000000000","tz":540},"subject":"Remove sendfile support","message":"Remove sendfile support\n\npysendfile[1] was added as an optional dependency but the library\nhasn\u0027t been maintained and has got no release since 2014.\n\nRemove the implementation dependent on the unmaintained library.\n\n[1] https://pypi.org/project/pysendfile/\n\nCloses-Bug: #2062573\nChange-Id: Ia4784f59d16660e8d40c0e409f092ac4e46870b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2740e132b744f30d787c810965162e35a77e2391"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2740e132b744f30d787c810965162e35a77e2391"}]},"branch":"refs/heads/master"},"a24d65afad2d914db601481b64ac8413fbb3c088":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2024-04-19 15:26:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/916422/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/916422/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/916422/7"}}},"commit":{"parents":[{"commit":"0bcd6cd71c09917c6734421374fd598d73e8d0cc","subject":"Merge \"Docs: Add info about the cinder store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bcd6cd71c09917c6734421374fd598d73e8d0cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 12:26:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 15:26:01.000000000","tz":540},"subject":"Remove sendfile support","message":"Remove sendfile support\n\npysendfile[1] was added as an optional dependency but the library\nhasn\u0027t been maintained and has got no release since 2014.\n\nWhat is worse, the sendfile implementation is not actually working\nsince The SendFileIterator class was removed[2]. (Follow-up[3] removed\nthe remaining reference to the class).\n\nRemove the implementation dependent on the unmaintained library.\n\n[1] https://pypi.org/project/pysendfile/\n[2] 76c3620c7effb15be577450c276db1e394c2de3e\n[3] 0c151d7d7ea00627fad182898958db55b1fef3fd\n\nCloses-Bug: #2062573\nChange-Id: Ia4784f59d16660e8d40c0e409f092ac4e46870b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a24d65afad2d914db601481b64ac8413fbb3c088"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a24d65afad2d914db601481b64ac8413fbb3c088"}]},"branch":"refs/heads/master"},"5cf02829747fd2ab82c0b1fcbe0c467a4877ca6c":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2024-04-19 15:28:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/916422/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/916422/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/916422/8"}}},"commit":{"parents":[{"commit":"0bcd6cd71c09917c6734421374fd598d73e8d0cc","subject":"Merge \"Docs: Add info about the cinder store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bcd6cd71c09917c6734421374fd598d73e8d0cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 12:26:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 15:27:10.000000000","tz":540},"subject":"Remove sendfile support","message":"Remove sendfile support\n\npysendfile[1] was added as an optional dependency but the library\nhasn\u0027t been maintained and has got no release since 2014.\n\nWhat is worse, the sendfile implementation is not actually working\nsince The SendFileIterator class was removed[2]. (Follow-up[3] removed\nthe remaining reference to the class).\n\nRemove the incomplete implementation to get rid of the dependency on\nthe unmaintained library.\n\n[1] https://pypi.org/project/pysendfile/\n[2] 76c3620c7effb15be577450c276db1e394c2de3e\n[3] 0c151d7d7ea00627fad182898958db55b1fef3fd\n\nCloses-Bug: #2062573\nChange-Id: Ia4784f59d16660e8d40c0e409f092ac4e46870b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5cf02829747fd2ab82c0b1fcbe0c467a4877ca6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5cf02829747fd2ab82c0b1fcbe0c467a4877ca6c"}]},"branch":"refs/heads/master"},"e119070a2c3f1c49a77a25f38aafa6d183933373":{"kind":"REWORK","_number":9,"created":"2024-05-02 07:17:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/916422/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/916422/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/916422/9"}}},"commit":{"parents":[{"commit":"0bcd6cd71c09917c6734421374fd598d73e8d0cc","subject":"Merge \"Docs: Add info about the cinder store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bcd6cd71c09917c6734421374fd598d73e8d0cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 12:26:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-02 07:16:33.000000000","tz":540},"subject":"Remove sendfile support","message":"Remove sendfile support\n\npysendfile[1] was added as an optional dependency but the library\nhasn\u0027t been maintained and has got no release since 2014.\n\nWhat is worse, the sendfile implementation is not actually working\nsince The SendFileIterator class was removed[2]. (Follow-up[3] removed\nthe remaining reference to the class). The broken implementation has\nnot been detected because the client is not currently used to upload\ncontents.\n\nRemove the incomplete implementation to get rid of the dependency on\nthe unmaintained library.\n\n[1] https://pypi.org/project/pysendfile/\n[2] 76c3620c7effb15be577450c276db1e394c2de3e\n[3] 0c151d7d7ea00627fad182898958db55b1fef3fd\n\nCloses-Bug: #2062573\nChange-Id: Ia4784f59d16660e8d40c0e409f092ac4e46870b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e119070a2c3f1c49a77a25f38aafa6d183933373"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e119070a2c3f1c49a77a25f38aafa6d183933373"}]},"branch":"refs/heads/master"},"205ca3336a7646ad3fbf92d8cf2509116a10bfcf":{"kind":"REWORK","_number":10,"created":"2024-05-02 09:21:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/916422/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/22/916422/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/22/916422/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/22/916422/10"}}},"commit":{"parents":[{"commit":"0bcd6cd71c09917c6734421374fd598d73e8d0cc","subject":"Merge \"Docs: Add info about the cinder store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bcd6cd71c09917c6734421374fd598d73e8d0cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-19 12:26:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-02 09:20:59.000000000","tz":0},"subject":"Remove sendfile support","message":"Remove sendfile support\n\npysendfile[1] was added as an optional dependency but the library\nhasn\u0027t been maintained and has got no release since 2014.\n\nWhat is worse, the sendfile implementation is not actually working\nsince The SendFileIterator class was removed[2]. (Follow-up[3] removed\nthe remaining reference to the class). The broken implementation has\nnot been detected because the client is not currently used to upload\ncontents.\n\nRemove the incomplete implementation to get rid of the dependency on\nthe unmaintained library.\n\n[1] https://pypi.org/project/pysendfile/\n[2] 76c3620c7effb15be577450c276db1e394c2de3e\n[3] 0c151d7d7ea00627fad182898958db55b1fef3fd\n\nCloses-Bug: #2062573\nChange-Id: Ia4784f59d16660e8d40c0e409f092ac4e46870b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/205ca3336a7646ad3fbf92d8cf2509116a10bfcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/205ca3336a7646ad3fbf92d8cf2509116a10bfcf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
