)]}'
{"id":"openstack%2Fglance~918384","triplet_id":"openstack%2Fglance~master~Ie56882b1a8f0b80faa370fb1e5041ddca95fc606","project":"openstack/glance","branch":"master","topic":"New-Location-Apis","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-06-25 06:16:40.000000000","reason":"Change was abandoned"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-06-20 05:24:28.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-06-25 06:16:40.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie56882b1a8f0b80faa370fb1e5041ddca95fc606","subject":"glance-manage: Add image command","status":"ABANDONED","created":"2024-05-07 06:22:18.000000000","updated":"2024-06-25 06:16:40.000000000","total_comment_count":23,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"c2ab41acaf5e176146326e835834b3a339f7387d","_number":918384,"virtual_id_number":918384,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-08 11:19:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":-1,"date":"2024-05-17 06:41:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-07 06:33:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-07 06:43:03.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-05-07 06:43:03.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2024-05-07 08:33:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-29 00:56:52.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"}],"messages":[{"id":"43d0a273171f9b484b6fc226cc4f56d543a71099","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-05-07 06:22:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc6fd3df1c5dbac41efa91e48a0a18738d64e103","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 06:33:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3f3a36a041543d08db1581e97d9efd1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/145807e232b54cdcb95eaed1c2b91964 : SUCCESS in 9m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aa113d5e46ad4e25885dea3c14286150 : SUCCESS in 5m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4eb8fe164823ab7f1079771342a70e034251a20d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-07 06:50:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3570a6711d2ab2c75b5e80288fa0c05d05c3bba2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 08:33:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2475acf4ef54d12b290c72f086d2d40\n\n- grenade https://zuul.opendev.org/t/openstack/build/8398848ec15f4070a4d06a5f2a4b281f : SUCCESS in 42m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0383dcff03b141fba2c7ee5b95abc51d : FAILURE in 1h 24m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fc4466c95fa4c8bb05a0f7750eb85cd : SUCCESS in 59m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ace608c594ec4c0bac43242038ba14c0 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9201ffac09aa490a86404a9dccf4cbe8 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ce60e107809429fb8c58a39792f95cb : SUCCESS in 5m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ddcd6236e2a4a319fb5f5611a8e24fc : FAILURE in 9m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10390e746f9d4ae0860df12d21361e38 : SUCCESS in 5m 44s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ff762da9f564461aa7630a5a48f61bdb : SUCCESS in 14m 09s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/336b92da708346a3864a3827c4c37e97 : SUCCESS in 7m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e054e38eaa744cf89f4264569bf585df : SUCCESS in 8m 03s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4ed6ca1ae9fe406285502faf3473260e : SUCCESS in 8m 14s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/16045279df9b4fafb33874b59291f75f : SUCCESS in 12m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4bf80741a06245859fe910039269feed : SUCCESS in 1h 51m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fd792334e0074b89a1547969f690aaa8 : SUCCESS in 1h 36m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0006fbdc0e7b4e4c979cd63255178334 : SUCCESS in 1h 55m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c0bee190a2654a529aad42196ec1ee97 : SUCCESS in 2h 03m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/35640cf94e7f4db6837520caec4abc9e : FAILURE in 2h 00m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/679ab7f6d3804bd3b52689aac11a7334 : SUCCESS in 40m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/62d6a7c557fb4700bd546e9e1e01f765 : SUCCESS in 1h 57m 22s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/a4cce9126f3447dcabae3dd394e2d8b4 : SUCCESS in 1h 42m 24s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/dccc10ddd8c745e998d56e9f5105644b : SUCCESS in 1h 09m 21s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8d91e9bda14a4fb79c5933441cc835f2 : SUCCESS in 43m 49s","accounts_in_message":[],"_revision_number":1},{"id":"95fbe86697d1211db9b92f68dd4e787226ca5e7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-05-07 13:08:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"78d3ada6e47a24fe5704cf7b8fc7a068e853c1f0","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-05-07 13:09:53.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d85285cb86a9495690efc615d6b1f2c32a4aef09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 13:16:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8552c65dfb5c4388a7a803aa846fbd58\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f0d6d9cfa67749f5a11e22387c3c2719 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3c9f689979c54f29a84d947c541a581e : SUCCESS in 5m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e046c1dc3fb93ab86ca222ec74cef75af3bc7f7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-07 13:27:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b576364ba3ffdc6cdc5264e02e90624a252b499a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 15:26:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b2b3fd985c24755ab1e1c9be2b5c331\n\n- grenade https://zuul.opendev.org/t/openstack/build/f2e5ec8b5ae744a191efe389a5e97006 : SUCCESS in 52m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7f6faacb41404e32a132d25820f46e04 : SUCCESS in 1h 15m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/300e2ee2e164495c8cd325eaba2b60c7 : SUCCESS in 1h 05m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04fc41f8e03d45a4af2881020ee1252b : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52753f714b924c74830122025439cb1a : SUCCESS in 5m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a24a2f4c48c24d89b2557d23546085bf : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6b8bf1bbff924c909b0d80d6d2c3fb5f : FAILURE in 11m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54a9d7782fb348209d6acd4a3df677b9 : SUCCESS in 6m 15s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7c4460113b9e4f2fac43db8a74f4ee6f : SUCCESS in 13m 15s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0db6d148d4a740efbdabbd8ca93e4039 : SUCCESS in 9m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/544cbcbef1aa4162b577779929811a01 : SUCCESS in 8m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/54eb0b0bcfdd45cd9810b879005f8d3b : SUCCESS in 9m 04s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5ab15b4a98704a66b762439184ae9589 : SUCCESS in 10m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a05f21bc7ed342b991070fd218b09b80 : SUCCESS in 1h 57m 08s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e3496430a25043ce99be0432ae3be7b3 : SUCCESS in 1h 37m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f2d8cac4503e4df583ab423c1a6c5e77 : SUCCESS in 1h 39m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f2eeba69729b4a98a0976c95d93f2cb0 : SUCCESS in 1h 31m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d27982afd7e640baa2159d6069a36b95 : SUCCESS in 2h 06m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f8f1d3060fa403cb5dfa1142e29f8ef : SUCCESS in 1h 02m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/07c3713d45b4459fa7497515d59d189b : SUCCESS in 1h 38m 56s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/90c2f2d08ec6426391b65832739e6c6a : SUCCESS in 1h 20m 34s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8889a26b9e2f44a2b0fe49431476c541 : SUCCESS in 59m 49s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d8e0769122eb4955b14307204805e2c6 : SUCCESS in 28m 32s","accounts_in_message":[],"_revision_number":2},{"id":"86c0b4a40f7c0e09a870fcd9d6797705b59ab96c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-05-08 09:01:11.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"15ff5c28b5cf6eb3435c8d1342a48ae67c92fe44","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-05-08 09:02:35.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b9dd3690aa7700862acafaef5381c6307224e27b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 09:12:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9240403d1ff44534a2ddffda8d08d224\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15435d31fb8247f68777bafb575d17cb : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/67490c4c2a854e4d855740d648619184 : SUCCESS in 8m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"79cb130bf7a6e857641d19906150b35de426b98b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 11:19:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c76f0360c0214fb292deaf1736c8fa20\n\n- grenade https://zuul.opendev.org/t/openstack/build/8cb55af0f3434e588ef2570b89fdb20b : SUCCESS in 1h 01m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d3903ffd02ef4bbb9649647b9083502f : SUCCESS in 1h 23m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d439a921637447108267c3d9bd2b2f21 : SUCCESS in 59m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8603a69a46849e099d42b7d5e8acbbe : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cddcdf73407e48f5b13c84391fc0e257 : SUCCESS in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7349c22db8a42a79dee6569b2534e02 : SUCCESS in 4m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/639c005e495648309a276368353f80aa : FAILURE in 8m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bdb3aee18894d8192b54f995b82aab0 : SUCCESS in 5m 23s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/079883ae4fa94feba3e251669847feef : SUCCESS in 13m 42s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e59f98d0749142a89df4833f7f88afd3 : SUCCESS in 9m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0c349a9d8fbe4fb3afc43c13e967e409 : SUCCESS in 7m 34s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/09a1a2dfe40844c691e76ab1ac70b23d : SUCCESS in 8m 13s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/3b6d0447b238480786cf1ce4eefb1300 : SUCCESS in 10m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/923111cdc453426eae4b747bef459d07 : TIMED_OUT in 2h 03m 37s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ec7429454f6f406483d093acc65b357a : SUCCESS in 1h 38m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/64997cf65d324437b01bf3a8e500a2a7 : SUCCESS in 1h 23m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5daff22121ee4354acc739320fca48f4 : TIMED_OUT in 2h 05m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a6b1b62094fb45b49b21bc0686b0402a : SUCCESS in 1h 53m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c69aaa6e1a24d3b9d81280928e026f2 : SUCCESS in 1h 02m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f67d1cf69017485fad751577f30c7ad1 : SUCCESS in 2h 06m 52s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/8b1733bb8a7f4a34b55d07aba161e490 : TIMED_OUT in 2h 05m 17s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/173d54bc680b4a449b3d68c1b6a10e8b : SUCCESS in 58m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/96678227774d41c8b79798e7035a3b79 : SUCCESS in 35m 37s","accounts_in_message":[],"_revision_number":3},{"id":"aa9b1c41d4a78664b8139bc4284e2690bf182f01","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-17 06:41:10.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"963090661985f697785a2b387a5d3e974cc70eb4","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-05-29 00:56:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5b4c1a57cd9780a7676269cfd80279384ec1d2a5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-06-20 05:24:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c2ab41acaf5e176146326e835834b3a339f7387d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-25 06:16:40.000000000","message":"Abandoned\n\nSince we decided to have a separate api for hash calculation, this will be no longer needed.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"55631d4bd744c5ad4c3550672286cf9be21fac5f","revisions":{"3fff62de9420899212dd349633f8c5dd14bf704a":{"kind":"REWORK","_number":1,"created":"2024-05-07 06:22:18.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/84/918384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/918384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/918384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/918384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/918384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/918384/1"}}},"commit":{"parents":[{"commit":"65ec7c3dac511ecd2c51f32442bf0f874f060c2b","subject":"Make location URL compatible with cinder backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65ec7c3dac511ecd2c51f32442bf0f874f060c2b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-06 11:30:36.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-06 11:57:23.000000000","tz":0},"subject":"glance-manage: Add image-hash command","message":"glance-manage: Add image-hash command\n\nThis adds image-hash to calculate the hash and checksum\nfor image if image is active without os_hash_value and\nchecksum.\n\nChange-Id: Ie56882b1a8f0b80faa370fb1e5041ddca95fc606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3fff62de9420899212dd349633f8c5dd14bf704a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3fff62de9420899212dd349633f8c5dd14bf704a"}]},"branch":"refs/heads/master"},"02a3fb7d08160caf7f062af4a5c151dd27c97ad9":{"kind":"REWORK","_number":2,"created":"2024-05-07 13:08:53.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/84/918384/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/918384/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/918384/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/918384/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/918384/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/918384/2"}}},"commit":{"parents":[{"commit":"65ec7c3dac511ecd2c51f32442bf0f874f060c2b","subject":"Make location URL compatible with cinder backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65ec7c3dac511ecd2c51f32442bf0f874f060c2b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-06 11:30:36.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-07 12:56:22.000000000","tz":0},"subject":"glance-manage: Add image command","message":"glance-manage: Add image command\n\nThis adds image command with calculate_hash action to\ncalculate the hash and checksum for active images\nwhich are without os_hash_value and checksum.\n\nRelated blueprint new-location-apis\n\nChange-Id: Ie56882b1a8f0b80faa370fb1e5041ddca95fc606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02a3fb7d08160caf7f062af4a5c151dd27c97ad9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02a3fb7d08160caf7f062af4a5c151dd27c97ad9"}]},"branch":"refs/heads/master"},"55631d4bd744c5ad4c3550672286cf9be21fac5f":{"kind":"REWORK","_number":3,"created":"2024-05-08 09:01:11.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/84/918384/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/918384/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/918384/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/918384/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/918384/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/918384/3"}}},"commit":{"parents":[{"commit":"65ec7c3dac511ecd2c51f32442bf0f874f060c2b","subject":"Make location URL compatible with cinder backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65ec7c3dac511ecd2c51f32442bf0f874f060c2b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-06 11:30:36.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-08 08:48:09.000000000","tz":0},"subject":"glance-manage: Add image command","message":"glance-manage: Add image command\n\nThis adds image command with calculate_hash action to\ncalculate the hash and checksum for active images\nwhich are without os_hash_value and checksum.\n\nRelated blueprint new-location-apis\n\nChange-Id: Ie56882b1a8f0b80faa370fb1e5041ddca95fc606\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/55631d4bd744c5ad4c3550672286cf9be21fac5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/55631d4bd744c5ad4c3550672286cf9be21fac5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
