)]}'
{"id":"openstack%2Fglance~918472","triplet_id":"openstack%2Fglance~master~Ic51de5ba8179708c946e42b9f2e504750c57710e","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-09-05 03:03:57.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ic51de5ba8179708c946e42b9f2e504750c57710e","subject":"Stop using deprecated defusedxml.cElementTree","status":"MERGED","created":"2024-05-07 18:12:16.000000000","updated":"2024-09-05 07:51:04.000000000","submitted":"2024-09-05 07:49:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918472","meta_rev_id":"34862c7319f399546eb0301c5da9b9e8b2eb57f4","_number":918472,"virtual_id_number":918472,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-05 07:49:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2024-05-09 08:24:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2024-05-07 18:18:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-05-29 00:57:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-05-29 00:57:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-07 18:18:36.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-05-07 18:20:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-07 20:16:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-09 08:24:31.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2024-05-29 00:57:58.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"fbd1f49babd7eaf0cf7eab15e6fd668cbd585963","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-07 18:12:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e57d16648d2e5d6eb05492f78f198aa8a2cf3f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-07 18:14:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fb7a927fdb4871dd33a8a370f66e0784e79b86a2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-07 18:18:36.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"53f5e106ec17fd88dd4614c9341b6b4b8f33bcff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 18:20:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be6f26d2bf4c4f22928689472efcbfd5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d609750a75124351a1fd376e8822f8ed : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bfac70e11e6a4abea5d05fcb1605ce60 : SUCCESS in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"17bf3c6778c007acfbef9f96c14b30013b3476ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 20:16:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33f4cce3abe94b53a4c8bf84e20fdde5\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0e2d531ea04d4c28852d2a49f0b68f1f : SUCCESS in 2m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/ee107bcdd8394cca93914345c75691ce : SUCCESS in 48m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aecf387623234f338b86031faf9c4c87 : SUCCESS in 1h 21m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc1ef26b387d4e8ca5e35d09e5ca2e7c : SUCCESS in 46m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e053d71c85e48e2a29147488f1eb214 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6dcc8ee49a84973b113a3f0c1716d20 : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70af4bd0d4ef416b85a56c5f7798afcc : SUCCESS in 6m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e4fe809e21e54b55a2e9999dcc700de2 : FAILURE in 9m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95cbbbccf6d146c2a81298456a6acdc1 : SUCCESS in 5m 35s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/200efd83de334acc886a406d8e7ff3c4 : SUCCESS in 13m 56s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/92fabe4754894de68dafd4c27e471824 : SUCCESS in 10m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dd7ddc72b14a42b7abd9c86165a625cc : SUCCESS in 10m 29s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f012ce9085e34ee5b4213081eb4c669b : SUCCESS in 8m 31s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/de86a67ba935472b8ac2865a447dd181 : SUCCESS in 10m 58s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4252723900b74a33b7f4303573433070 : SUCCESS in 1h 55m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/de45d8c71902473e87432bcccc6b9812 : SUCCESS in 1h 45m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/092e0228664b403584a53413edbf95a2 : SUCCESS in 1h 53m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/86a043a6a6664c47be04dcf5cee63324 : SUCCESS in 1h 53m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b1494ea5b800435ea24b97b7dd6abdcb : SUCCESS in 1h 56m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b49f69280d7425b9f71d8e4e5ecac16 : SUCCESS in 1h 11m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d964ccfb8f5e45d781258c2885b85288 : SUCCESS in 1h 56m 49s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b4f682c6844f470eb7f687a24e3b18e8 : SUCCESS in 1h 44m 25s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7fb87731caaa4d88bc925b6c7d4c840f : SUCCESS in 1h 08m 23s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7ceb583be3f04996afe771a4ef4d8830 : SUCCESS in 41m 12s","accounts_in_message":[],"_revision_number":2},{"id":"6b14a313e1fcad93c838d9e9aeefd6552f62a2c1","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-05-09 08:24:31.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"49fc5afd2d2a69768e761e4f6f5b2d2adf994563","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-05-29 00:57:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7b7256bbe28ad0d1ea027c8238ced837c67488b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 00:58:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9fdb7d21e1a551ef4091391c29778dba80457311","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 03:09:21.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/12ecc9ca50ee4f72abf70eae409ff50c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3f7f19a2d1384a40b4d28a0614bdd784 : SUCCESS in 3m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/19ac8d8afd104f55b52c9f45dd029421 : SUCCESS in 1h 00m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/53498fa8510142a7ad1b45e057e21a48 : SUCCESS in 1h 11m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7b71a1c4a8d4426897be49d60b8429f : SUCCESS in 50m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a79f78a4adad4d78902943b9f3b2737a : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b98d287d032c480184d283517abfa62f : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e5b2b4402c5c4225b3bc9dbbd0169dd9 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf3bd702072943eb93fab29042671a85 : SUCCESS in 4m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bc4e49098e444deab61d584ed9f85c9e : SUCCESS in 8m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f32b23fd374b42ecbd0425a77866b670 : SUCCESS in 8m 34s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ac122319eb41495e9ccdccd09fbbad65 : SUCCESS in 2h 00m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6408576d84824d4eba78af8fdc7cd2b4 : TIMED_OUT in 2h 05m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/537058d42b26458392f385b852bdab9a : SUCCESS in 1h 56m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea0296d72205415cad17b0c1215f7fbe : SUCCESS in 1h 09m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/741495af3f8f46a996c6ea2db652eb7a : SUCCESS in 2h 00m 31s","accounts_in_message":[],"_revision_number":2},{"id":"94b1eb3e92663c9cdcbe3199bd0c519055edf8bc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-29 03:20:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aa02b91d5d4f9e390abbb7f40c94e5ba3a19150d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 03:25:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f3243f5553c49c18683109f125e0baf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a2c0e6e2c1984bc09cec79ab1651d4ae : RETRY_LIMIT in 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4d7b8511937241cc8b6850e3b7e29e9a : SUCCESS in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"200d178c0abe662e9caed9416f59ab5e8b0f3347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 05:42:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1b70cab79ef40d6b9cde158c486324f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/07accf072fea4dbcb4196a15aefd9ed4 : SUCCESS in 2m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/258b38d2983a4f8eb3bd562415d2754f : SUCCESS in 1h 06m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1027a80d90a04d96a38e7923c587d21f : SUCCESS in 1h 35m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42fd3ff0df1c4751808a74c6e252f8bc : FAILURE in 1h 04m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45700fe9e7154c03b728861740de5cd0 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8f6ef7435e9455fa094218991d1e2a7 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e5cb82b44fd943b6a278a7edf20e917a : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1040e18b8e7640e183dc4a2631aeab65 : SUCCESS in 9m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be20608d291f4b2fbaee9f72e57508a6 : SUCCESS in 8m 42s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/545091a5da624ec99d0c3e85964d11e7 : SUCCESS in 13m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b4af446d3caa471d8ed31768ef4941df : SUCCESS in 8m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b82b971896414e7b8ce548423d1768c5 : SUCCESS in 11m 21s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a6943822645441a98728c16125999917 : SUCCESS in 11m 33s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ba34d579c82a48a59d0b943102859a85 : SUCCESS in 9m 39s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1de3c9dcf81447e49b53896d0d1fba59 : SUCCESS in 1h 58m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2cd3dcf16a9342c7ad87667d2ff37d07 : SUCCESS in 1h 38m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2bbca30c43464f6b92cd718d076cf9e8 : SUCCESS in 1h 50m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1b3da0a5f0df423196073efe5dc32db5 : SUCCESS in 1h 36m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f4b11ade2eb04538873abc9d5e25f4b5 : SUCCESS in 2h 16m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5bf5dece02b496eb33807d0bdd3f6b4 : SUCCESS in 45m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9e85338db6374be2ac55a821a040aeb5 : SUCCESS in 1h 41m 29s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/d53555fb687140b6b48fc8dab1cf79df : SUCCESS in 1h 39m 43s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/66995bba51f24107834aa4af593781cc : SUCCESS in 48m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"91631a646d65a1b859bff273ac899f6010db90ae","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-29 07:15:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bf1926e0794d7b2040a48845332038d3d1f1b5cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 07:21:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/524fb4bf486a4abe8593b9bc49bf88f0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a2b90bd11fc24bf4b5b185e5d5c4fe97 : RETRY_LIMIT in 1m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d0cf361c0f534b3d81bcdb2b4f3f358d : SUCCESS in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7e6d65c6fb14ba74938c263e929768c2117e9eef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 09:41:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6275fcbf7ad5468291e62d356a7ba518\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1c6bb4baece3470c98e93a41901f4611 : SUCCESS in 2m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/f1a0e2eb62a04d9b93f0f3361cbf79b3 : SUCCESS in 50m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2a8be19dc9b64870af4378219093d815 : SUCCESS in 1h 23m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a0355c9ab634b09a97195191d629478 : SUCCESS in 53m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56cf188e34bc4d8ea3ac9e01cbb3240f : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/999ce96a18204636a8328efe5a5ae911 : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c12cd0eb57544e187999962cc9ef951 : SUCCESS in 6m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64b06e7bed694b0fa72a5f7628549081 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7064c1a04a624dd1a36d4824a4c7e467 : SUCCESS in 5m 33s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9ae7361b2731423f9f73b0069d52feac : SUCCESS in 12m 48s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/88b76bf983c640c890113a8e4795aefa : SUCCESS in 8m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0ecce58c22444f0bbb9de5b7aba02bc1 : SUCCESS in 7m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/315533f0632b43ffafd75a301ce5872e : SUCCESS in 12m 11s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/462aed00b9e14f018907aae8707cac03 : SUCCESS in 8m 29s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/baa199dee87d4cfaa828ad02951aa09c : TIMED_OUT in 2h 03m 32s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3169be37db1c4f008450bb9941375929 : TIMED_OUT in 2h 04m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5fb69de70f56469a94600cf4ff32885b : TIMED_OUT in 2h 03m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/be323b57283244b3bf61b5eef02d57ab : TIMED_OUT in 2h 04m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2d24cc525eaa45de91ffcee84c7416ab : SUCCESS in 2h 10m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e37d79a7d64c450098d895865c021d66 : SUCCESS in 1h 02m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/62fe1a0c04584092bad754482ec4997c : SUCCESS in 2h 19m 37s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e8a20d6d03bb42f5b5351339f0034b8d : TIMED_OUT in 2h 05m 07s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a86e3cab14db4971b9ae82cb34fa9592 : SUCCESS in 1h 02m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9d0e8313844039ecadd2c045e5820fffc40b8f83","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-31 08:29:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e9d9b4238435c3e71d404d535b73d8a98e5099f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 08:44:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c364c943d87f4e2696ac838b4387aa9d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28cdb3e0020b42109574a18ffcd07b4d : RETRY_LIMIT in 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c2f2b31efc384f87b19772a02c78d3bc : SUCCESS in 12m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fd27f74c4216d41130ba51fecde1cbdbdf7245ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 10:49:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/14f33bfcbccb42ccb802330bcb2a79aa\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ab35be185a9841a8946e915c9729b035 : SUCCESS in 2m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/0c72da2b3a9846149d5a29eab6c4862a : SUCCESS in 55m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1d81238f0f534b45a5f4d3b4f1e64ca0 : FAILURE in 1h 30m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eeffafb64f68463fa32169e772181a01 : SUCCESS in 1h 01m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6a9af6d325b4b89b25f93b353c7edbc : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b916c25e26344a9ab8e49d99b8c5f45d : SUCCESS in 6m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1a79e3d3a44479495d9c89d218393a2 : SUCCESS in 5m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9c4cfad931246749255243523df5c55 : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f444ffe2e09f4429a5142ccf00198df7 : SUCCESS in 6m 40s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0a8a9ea842934b95ae214841becddab4 : SUCCESS in 13m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/520e661d26d347b8a0dd853cd13e7780 : SUCCESS in 8m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9ba0595eeb174153a81052d33e5db88a : SUCCESS in 8m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0055672d450245fea7caead4897fe90e : SUCCESS in 11m 00s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5b16bef2c2d54a89830d0f4693569e79 : SUCCESS in 8m 28s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d691cb9e91e443c890bcf9e0d13d2c9e : TIMED_OUT in 2h 03m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cdb3275a67f242858b81c296fcf5af0d : TIMED_OUT in 2h 03m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/eff728820ca54a7eb34818b03fafc4a2 : TIMED_OUT in 2h 04m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/604ecc8273b24cb0b90a57b6a83f8d42 : SUCCESS in 1h 46m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d79a1806b54d46e989dd5f79a4dcbbac : SUCCESS in 2h 07m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/142d18660475407bbb006b5243958974 : SUCCESS in 55m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f54a5954c8b64b1aa6ef11aedcefe874 : SUCCESS in 2h 08m 47s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/1668318240244e3c887b80d2c4fd88ec : SUCCESS in 1h 53m 34s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/38bf1bbbbdf44f76a1ba79df0a562c48 : SUCCESS in 1h 06m 38s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6fc819fe8fea4af4b3bbadf99215d780 : SUCCESS in 29m 59s","accounts_in_message":[],"_revision_number":2},{"id":"2e17f7c4054cfc3da651ffd4fb4ed61a67cddbf7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-01 01:08:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a323c8889137872b08e1ff220143e9e33b094b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-01 01:15:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dad4fad73584f9f9e73a70b61af63da\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/98143526968342ce9906abbcdad7993c : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b699363047594f2290a99c4d4d0a080d : SUCCESS in 5m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0f40e6ae980ca385ab7616b2dc3fb457d6542bc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-01 03:26:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd356b84134b4a31a257ed1244989228\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7638a653d0034b32aeab2a53149c785d : SUCCESS in 5m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/8d7ab8f8cf61417ca60909e860ff6b3b : SUCCESS in 55m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef9a14749a4d4f4399eda97469db7e33 : SUCCESS in 1h 22m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3d59097ee694db381ec6fd8dc7929a4 : SUCCESS in 1h 04m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/828bf6d4f3444d4bb0780371172d89fd : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2dafefc7bf6d4687bd8c4b04ee3abd15 : SUCCESS in 8m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/13d4841388504ad7bd299e4d4e48c7f5 : SUCCESS in 5m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/34b108e02ac34d54bba4ddd4ff07f382 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23f61cdb8d0b41bcb48c3f49ec76356c : SUCCESS in 5m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/89349bc8b47a4ac3b8bf300d96ab31e8 : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/83e4c9b0a59d413c9b276e948362ccea : SUCCESS in 8m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a0d2542823a048639cc7b4ceb7eb9fa6 : SUCCESS in 7m 13s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/795169b71400444b98a262e7ee00efb6 : SUCCESS in 9m 06s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8cddfba18fd145fdbc980795f3ae21b6 : SUCCESS in 7m 49s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/47d3dcc3af3f4e358cad464452cf1a47 : TIMED_OUT in 2h 03m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0cdc7901213a437482bf622b9509e8ea : TIMED_OUT in 2h 04m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bfe09ec31e8d401abb9119179476f75e : SUCCESS in 1h 28m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/855ec01608014f2981fb0e2c628da166 : TIMED_OUT in 2h 04m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fc6d0280bf8143299b027cbfdd7da7da : SUCCESS in 2h 04m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2dadafd5fed4220b98ba2603a261872 : SUCCESS in 49m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/733a848ffe8c4d49950488b7ad6ee5f0 : FAILURE in 2h 10m 51s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/ffa7573c235f40d39c51a771ebff97c7 : SUCCESS in 1h 51m 30s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/94df124399154f7298987eab5582d3dc : SUCCESS in 1h 08m 04s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8ccadf87fa764026a385e370ced936bb : SUCCESS in 32m 22s","accounts_in_message":[],"_revision_number":2},{"id":"8b71c161a99bccbe11e6ba896511da92bc9b3437","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-02 07:13:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c3ac1dac6a898307bdcf7b0317e64ffc36be53fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-02 07:23:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a7f29977f6848ddb92cc27a3e944601\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8242629951e84d8fb2c32adb2e605134 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/514f506394734a128ffca41e5ad89b6f : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7c9b441549294cb64678663cfa662245754c41c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-02 09:27:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d71e10812c7749a7be3bad75de9d5d23\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c3cc3946760f418c928c5878d92c6942 : SUCCESS in 2m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/571217dace4646bda06c076d1b4a2c56 : SUCCESS in 58m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/294704d929d24644aae09efa9bf588c9 : SUCCESS in 1h 10m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61ccb1c7e64c4546a1090decfcf9ba63 : FAILURE in 1h 06m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/802954434730401d95a1e921fbbe8212 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec11900eebc74646be30b028b90ae9a1 : SUCCESS in 8m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b427990e8d8c423a95231900f2a53ff4 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ab68ae00ec1c47f1a4f9ce3315d2bd81 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/531f1b64746148259c93ab9995f34681 : SUCCESS in 5m 38s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/410630a81eb54953b34eb08ea7e5df1e : SUCCESS in 13m 30s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d080a7d21a1448c4a66cd12bb06ede03 : SUCCESS in 7m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b32bc63751284230861bb31ac0a36308 : SUCCESS in 8m 18s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/08e7f7ca72a7490ca8ccda620f12547d : SUCCESS in 9m 10s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4b90d9c1a2384775b13f19f87c210abb : SUCCESS in 9m 04s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/eef5e21dba6e48f8bcc72b5ee185195a : SUCCESS in 2h 03m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/eadbced4b20143b28bae5f22e3ee17dc : SUCCESS in 1h 57m 31s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/25613cc28ccc4e0e924487ada34c5f20 : SUCCESS in 1h 31m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5911199e55da4000bdc2d4b3dbdb280d : SUCCESS in 1h 49m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b15d636f8eb74dd587a4bb8af46558c8 : SUCCESS in 1h 40m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b162ba36eb3d42259c2f8da83c73e6ca : SUCCESS in 1h 03m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ec19012c00f9430c9916c620714373df : SUCCESS in 1h 35m 41s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e512644548114950a162c658dfa16559 : TIMED_OUT in 2h 04m 01s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/24e9c84773534a4f9400c127909d3057 : SUCCESS in 59m 58s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8032722b5d734f219b3d47eded85f02e : SUCCESS in 21m 45s","accounts_in_message":[],"_revision_number":2},{"id":"faf08ba86f004153f2f70d2e06c3bb9c35858fb3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-02 15:14:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c306c4e46c625e88fe5ba161f08f3003aa84060e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-02 15:20:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29819966b4d142c397ef696affe12520\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/910cac93ca044c0a93b7f64736e6bc74 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6054d04d40b94e5aacbe63ee4246aba7 : SUCCESS in 5m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f818770fb8a541d30151c420b89bb2361f838b2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-02 17:36:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e0dd830e03e41e386b514e641eb97b6\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/86fa4f23bb8746a4804f6ccc93c1b21e : SUCCESS in 2m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/272f6fe0199e41c295eb3765075fff58 : SUCCESS in 41m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f5865baf4200484c980b749205959b34 : FAILURE in 1h 24m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6de44736c59d47dcb00298959709a269 : SUCCESS in 1h 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7f8b4cf63d44aacb5ef050ba8e45e6f : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19c478ba8d07408196580cc7dc2f08ac : SUCCESS in 8m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a531006d8d7401c9a56b5baaccdf831 : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2cfb4146f1d449dca77f3ecb16db1991 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47b2ca0996e947ed80882283cc067dc0 : SUCCESS in 5m 45s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7bab4248b97b49bda6acbff718fa5129 : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d747cea4ddd544a498ebcffa6d561b22 : SUCCESS in 9m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fe2a04d44dd34de7a20a60d1c71753c8 : SUCCESS in 7m 55s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4f91f0c0831147b9a27585ecf2284d40 : SUCCESS in 12m 00s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7693e24fb8964913b6f4a928e2304cbb : SUCCESS in 8m 22s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/651d9fc76e15419b8801fe0de99193c6 : SUCCESS in 1h 55m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c3eab6089572461ca86b4ec6dc1c96ec : SUCCESS in 1h 25m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d718d021792e44c38a7ca3b7b2ca9fb2 : SUCCESS in 1h 53m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/633c1ab4538c4e57b5df685d59f44894 : TIMED_OUT in 2h 05m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8c96f6fa731b44538fc0f2ae90a4fc35 : SUCCESS in 2h 16m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1fec38ecfd148829c88c3056b13a695 : SUCCESS in 41m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/86df4c03b31947a8be8ec50e25783cb6 : SUCCESS in 1h 29m 02s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/ca4c0d0556d24718b3adc795e7595530 : SUCCESS in 1h 43m 43s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c1f055cb12664089be69c823e962ed26 : SUCCESS in 1h 06m 14s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2ff0cdeca5374ed9a3ebe69c8c183bb6 : SUCCESS in 21m 13s","accounts_in_message":[],"_revision_number":2},{"id":"2214cf0b3fb8118d66fdf8fc1c0add4e74391351","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-03 03:24:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8f8a74668e9509447164c911f22ce7aa2c91a575","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 03:30:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f8755892aea43edb754c3e020262508\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0d979317e5e4746a8585e2a3f2d5708 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8e6f633e49224e7c9436c95a96a8b29f : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e29ac391970420a49882b838d30579dc7cdca0c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 05:36:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb484e82ed7d40f2a631caf9397d5a90\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9d4f27810501495d829db61c0193e4c5 : SUCCESS in 2m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/9973e0ed4f204b6d83ca8ed6edd90807 : SUCCESS in 1h 02m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d1ad80d7a7744617ba74d7ab4d33feab : SUCCESS in 1h 20m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7174955bd504844ad430968af654176 : SUCCESS in 47m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5335a988f1e456c9554fe54b51db434 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb2e7bae7f8945e4af34a1d9d9ee2452 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3dd1dcbbbe548bbad21e0679a0331fe : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/12323e9f06eb476991401842ca264ee6 : SUCCESS in 9m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97ea1bd967344edcb7796c49a2760c75 : SUCCESS in 5m 40s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9428dc10d06b46f2b118b41e61a9d798 : SUCCESS in 13m 15s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a280c26a54c84e3ea090831e4f26a60b : SUCCESS in 9m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/560cbad1581b401f8b6332449c22c91c : SUCCESS in 8m 03s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5bb4edde50c34568a6dd176fe7a23621 : SUCCESS in 12m 14s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/55194dc154a0450ea3f5a9386be0ad7f : SUCCESS in 7m 35s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ddb513fe92ac4db4a3d9caa0b49dc60b : SUCCESS in 1h 34m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/deb1edca88d24bc5a829c051a6d54959 : TIMED_OUT in 2h 04m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c64e4f94065146b2addb853c584ed928 : SUCCESS in 1h 49m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8503aa86d0794e6a89ea8590f3425dc5 : TIMED_OUT in 2h 04m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d7aa74e4261a492cbabb6ba0077184ea : SUCCESS in 1h 54m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4450bf718c94689aedd00f3b9237f0b : SUCCESS in 44m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/423f7a285c0b4d80917ce7b1ea4b1937 : SUCCESS in 1h 43m 52s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/90d154cb968e48938fa697f1781ba3b0 : SUCCESS in 1h 26m 26s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1ffcd8c241064db6859473ebd7af946c : SUCCESS in 1h 07m 17s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c323e6604a4a4cdfaa168277c3094c1e : SUCCESS in 36m 05s","accounts_in_message":[],"_revision_number":2},{"id":"8df39f6265baf53ff38b5858593d08a32e95ef99","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-09-04 13:32:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d934307eb92acd7b35aec15752eac961447e0582","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 14:29:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b186525cc7d1400ca4388d3a661786db\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/53a108ecf08947158493c268f3b463b3 : SUCCESS in 11m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0c27268665774e029ac42c7e6f55e28e : SUCCESS in 11m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0a74fb3f8887da9230ee906b6c2de03823af541c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 16:05:55.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ef0d56e5969469f900197548a730a77\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6acd6742efd64ffa86429eb8fb0df151 : SUCCESS in 2m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/7fdb28dee7d74cdba0b499d8c703f162 : SUCCESS in 1h 25m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3e5ff0cf84b54ee4aee9f63e0425c48e : SUCCESS in 1h 55m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b45582e57d047a49ecc139e675ea15a : SUCCESS in 1h 01m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71986735e432454fbafd2301c62f91ac : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0306dfdd86245598c7b104afb49626b : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1b82a780864425a8d7ed78087f138a9 : SUCCESS in 5m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/94c3f9ee54bd4dacaf62411084933117 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f33d5461f0734667a6859fb38e074f97 : SUCCESS in 7m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/82393b5fcd0043ad99adac7185c4b780 : SUCCESS in 11m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e1c92f7d686d4c23a53408b7eb4fe471 : SUCCESS in 9m 21s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7cb0baffc9344281aafd80752754fb86 : SUCCESS in 12m 51s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4ea63721c88d4749a2ba18fea8378f38 : SUCCESS in 9m 33s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fcb4dcb0fad24f8d8acf4c5bc6caecbf : FAILURE in 1h 48m 11s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b416ae2bb3d14d98be04571fbcba3147 : SUCCESS in 2h 11m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/438d0add60094c9ea9d05772f63c9620 : SUCCESS in 1h 54m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3129be0628684a2e9daf8d40e84ea886 : FAILURE in 1h 44m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2fbe1d15228d408c80dd3b3a53dd7fcc : SUCCESS in 2h 30m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be5954642d3b47b5ae7a840a8a260eb0 : SUCCESS in 45m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/831ffe3406d54733ae060abe314ae948 : SUCCESS in 1h 35m 36s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/7111a99c43a24894973ced036f6d8458 : SUCCESS in 1h 44m 56s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f0a6508ef05947babb1f72d4bc6f380c : SUCCESS in 1h 20m 17s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/689abdfbe5fc49719cfef15e06137fe7 : FAILURE in 28m 42s","accounts_in_message":[],"_revision_number":2},{"id":"136818056f869a544e7a8d117ef8958cca137aa5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-05 00:28:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d54c80335404cddec0065fc99cbdbfd202505cbc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 00:45:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/526f8e483b1c4c26b13089caaaebed1c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f3d2ea67c2948faa28feb983b5736b1 : SUCCESS in 15m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/254da8802d5b4725b0067753c59bf1ff : SUCCESS in 11m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"49dae59a99b4e9a57275f700422352081413a886","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 02:32:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/30aac5a7474f4a37be6c145ef13fc64d\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c5ed095895ea4577bfa8fa9a50d63814 : SUCCESS in 2m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/0ef968d7ab56404d9bffbefbd4a9b40f : SUCCESS in 1h 03m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ba55d3826b3047639a7fdc3c625138bd : SUCCESS in 1h 24m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fda91227964f4e65b2c1c44837a6fc3c : SUCCESS in 1h 05m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d665859ad44542979e1deb426f374c98 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a6d644a639b41f6b5ec317caa5e99bc : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa1ab29600b94d0db783213ec0a64bbb : SUCCESS in 5m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8eef607cb3c2449c932742cfddc7818b : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d9276fa524b424aa1c8ea7ab01b3850 : SUCCESS in 6m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/242c74357ca44bcaa7063c7f4308704d : SUCCESS in 8m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b5a5bf445d4e4bf88cac59ab6c8a3549 : SUCCESS in 8m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0f43accf7b9f4441a1b7bf0f3c09964d : FAILURE in 12m 03s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e78dc8d7d2324230a8dcacd3baeef254 : SUCCESS in 11m 15s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7adf6e90476d43de957033997599395e : FAILURE in 1h 27m 31s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f2e3e4c96de54340b4d71889d760bbaa : SUCCESS in 1h 31m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f533485101744572b704e73ba19e18f2 : SUCCESS in 1h 58m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/23e7f031eddf4513b0db4102d7fbbe01 : SUCCESS in 1h 55m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/414278040ce64f9b8d71ba7e1d7a38c3 : SUCCESS in 1h 59m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8cfb83b015f64552b1781864c55aa603 : SUCCESS in 45m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5e9de34addf44cb7bfccb3c929755f42 : SUCCESS in 1h 38m 59s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/4aa1eaa1f6a74ebaa6a5ae24eef887bc : SUCCESS in 1h 25m 55s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/85ce66328fa546538db16cd46545c835 : SUCCESS in 1h 10m 08s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/74004884090443eca5bb4ab4bcdf9b5d : SUCCESS in 32m 44s","accounts_in_message":[],"_revision_number":2},{"id":"c6be5b431f71864fc59cc8bf73553dfe6a9b3823","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-05 03:03:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a69b375ed8bee31b6c98af81fc2370deeeae5c3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 03:17:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a6c27ceeb6a4b848635c22080dc09cc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/592d461a96a24de2ad55c1d594c87548 : SUCCESS in 11m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cd2598b7a9414426be9f954e2607ef21 : SUCCESS in 9m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a8dbc7b63c8a9c0d4ec03f16c0650d144d55ef69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 05:31:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a73e45de61f46e4a1822c342a95c3cb\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d80caa110f4a4525b4ec175c22d27a7d : SUCCESS in 2m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/48ed6bd8bc8240609c695bcb307bf54f : SUCCESS in 1h 05m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b0b935e3c8704effb3d238b91c8ee322 : SUCCESS in 1h 32m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bcfc654c96dc47278c935132253761ea : SUCCESS in 58m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/869b5b5f74f4429582d668feebd00589 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d6fc32168774cb9b7e90a43adf7aade : SUCCESS in 6m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/98b3c9d74c6444bba4f07b27b92b178d : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/479e954d45994959ac9ad9cbefeeb98f : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65c85d778ecb4e7e986fe7f658ca1922 : SUCCESS in 6m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7afa218f3e3945708bf155e6201e3d3b : SUCCESS in 8m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0caf1d56101f475d993015119d30eb67 : SUCCESS in 8m 37s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/db5adf41e2464ea69f6c0ce44cf8bf21 : SUCCESS in 11m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a344cd167001439891a0181d288da8ba : FAILURE in 1h 51m 11s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7e75ae65ba2143deab51b5949e24a8a4 : SUCCESS in 1h 24m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4ca4822f26ed42f6a3e07808823ed115 : SUCCESS in 2h 03m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ad91dfee9dd047d5a2e3d37d76a226c9 : SUCCESS in 2h 17m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/030b4e0364154ffa916becdece32adc0 : SUCCESS in 1h 48m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4703b7ddf4304fcdb016b6022b0f6678 : SUCCESS in 53m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/07f5a96f5ebd42efbce3bee4a0e2a785 : SUCCESS in 1h 49m 57s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e8246857abf24d5fb02f2ed1eb23a1d2 : SUCCESS in 1h 33m 22s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/fc26d8e219684e7e90aea59008d53bac : SUCCESS in 57m 09s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cfa65793bb4c45fb98b63a0d808b4e95 : SUCCESS in 36m 45s","accounts_in_message":[],"_revision_number":2},{"id":"e0a68fe5100986649b11e11b32c798195edea195","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 05:32:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"33e9aca17719297926c40e8841b153d1cb0ee17c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 07:49:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9947ae8502f4e20ac2bf8c75e39e998\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c1e76ada6f054867aea4d11eeccf34da : SUCCESS in 2m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/9b91ca6079d34e548d2f387d01b0ed88 : SUCCESS in 47m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d540bc7379c84e1e93977f09627f2aa5 : SUCCESS in 1h 14m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cacd5656b0824bb984c0c5938d77fbfa : SUCCESS in 45m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d58114da49bf429289cc17752a70c2a6 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94367dd7ee684e649a8cbaed5b3fab3d : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30e48cbb49b64da18d6aa9ef7e86d4a0 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97f4ed4bc7934a54841367808456fcbd : SUCCESS in 4m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/752df401b02e45d19fd478ac337184a9 : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9f27d6a65a38460dafda1f3e502a2277 : SUCCESS in 7m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0ae43c23c8264d83960c32f6c7691991 : SUCCESS in 11m 45s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d4d497dd17504aa3bbe7e53442d5c8e8 : SUCCESS in 1h 35m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3f2a3a4d962147a284498af513c05db5 : SUCCESS in 2h 12m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/77ea82e2e62e4892b9a20843d1d96232 : SUCCESS in 1h 53m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/807d60c55da747b6993a4ae3e828c36a : SUCCESS in 1h 03m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/168a2c9f192c456c8a63872f51b0e594 : SUCCESS in 1h 59m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/259838c5b90647b1844e88ce1ce6a5c8 : SUCCESS in 33m 11s","accounts_in_message":[],"_revision_number":2},{"id":"db52276acdbd737a9eaea8cabb8006e5bcbae393","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 07:49:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"34862c7319f399546eb0301c5da9b9e8b2eb57f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 07:51:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/952b8929de284c5fb9547bd7b31492c1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7bff6dca83b453ab3856a20111b34c7 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ef58d77be0d8964435e9d3350617fae545d8d334","revisions":{"5678cc5ccfac8bd78c4072b6c15ab64dc4cc285a":{"kind":"REWORK","_number":1,"created":"2024-05-07 18:12:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/918472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/918472/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/918472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/918472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/918472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/918472/1"}}},"commit":{"parents":[{"commit":"95770a25e33cd531216104c788ac5dacda4e36cf","subject":"Merge \"refectory: fix comments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/95770a25e33cd531216104c788ac5dacda4e36cf"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-07 18:11:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-07 18:11:39.000000000","tz":540},"subject":"Stop using deprecated defusedxml.cElementTree","message":"Stop using deprecated defusedxml.cElementTree\n\ndefusedxml.cElementTree was deprecated in defusedxml 0.7.0[1] .\n\n[1] https://pypi.org/project/defusedxml/#defusedxml-0-7-0rc2\n\nChange-Id: Ic51de5ba8179708c946e42b9f2e504750c57710e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5678cc5ccfac8bd78c4072b6c15ab64dc4cc285a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5678cc5ccfac8bd78c4072b6c15ab64dc4cc285a"}]},"branch":"refs/heads/master"},"ef58d77be0d8964435e9d3350617fae545d8d334":{"kind":"REWORK","_number":2,"created":"2024-05-07 18:14:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/918472/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/918472/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/918472/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/918472/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/918472/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/918472/2"}}},"commit":{"parents":[{"commit":"95770a25e33cd531216104c788ac5dacda4e36cf","subject":"Merge \"refectory: fix comments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/95770a25e33cd531216104c788ac5dacda4e36cf"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-07 18:11:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-07 18:13:48.000000000","tz":540},"subject":"Stop using deprecated defusedxml.cElementTree","message":"Stop using deprecated defusedxml.cElementTree\n\ndefusedxml.cElementTree was deprecated in defusedxml 0.7.0[1] .\n\n[1] https://pypi.org/project/defusedxml/#defusedxml-0-7-0rc2\n\nChange-Id: Ic51de5ba8179708c946e42b9f2e504750c57710e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ef58d77be0d8964435e9d3350617fae545d8d334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ef58d77be0d8964435e9d3350617fae545d8d334"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
