)]}'
{"id":"openstack%2Fglance~918636","triplet_id":"openstack%2Fglance~stable%2F2023.2~I0c09dd7c0d81178ebf2aa5efd19c54866fc8791b","project":"openstack/glance","branch":"stable/2023.2","topic":"fix-prefetcher","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-05-09 09:21:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0c09dd7c0d81178ebf2aa5efd19c54866fc8791b","subject":"Fix broken glance-cache-prefetcher utility","status":"MERGED","created":"2024-05-08 13:34:14.000000000","updated":"2024-05-09 09:22:23.000000000","submitted":"2024-05-09 09:21:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918636-fix-prefetcher","meta_rev_id":"222b107a7c2877d03888aa00e50ec8d64177127b","_number":918636,"virtual_id_number":918636,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-09 09:21:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-05-08 14:29:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-05-09 02:24:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-08 13:34:14.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2024-05-08 13:49:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-08 14:29:28.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-05-08 15:44:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf4e3fe2c77b666950f2f103645b02caee7cc9ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-08 13:34:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"121ba39377f66516a0eff998ed4ba7f5a544e2ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 13:49:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1964379c150c4dc1b336c8ecbf24b003\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ea36e8568814d8aa7f62c3af8194ff4 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a4f3c23d7414681980fbb1c2b12ba62 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/73142cca660e4f9a8ba0a0a4995ae124 : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/34e45c6664a44ac587a132988e015120 : SUCCESS in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8c91c06846e8d55ea460b88b8f3e036e9d8e669","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-08 14:29:28.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c1f4b6a4fe9c12f7e76da2c513bea8cc40a32a14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 15:44:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d708e09600034c9994311a37825a80f6\n\n- grenade https://zuul.opendev.org/t/openstack/build/7a9674b8dfb74c00bf61ae97331e54c7 : SUCCESS in 1h 00m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e29d169ddc60454397b9e4131508bc74 : SUCCESS in 1h 34m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c83eb5b9d0e64a10bd029e3f9ee5d851 : SUCCESS in 1h 06m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b99dd3a9d7f541a383af0627aed9676a : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/752b79b7267245b2b5747162e421b36a : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/442fca5eeb7c4c9dbf646cc926032d52 : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d6653e6cc8a4feb9ff3c47d3f1b23a3 : SUCCESS in 6m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6335f0a25e7a48249919560af345bd2e : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/938bfd79dae14562803c84e3c3c32765 : SUCCESS in 6m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/58babad5fb5647c3a53db3f09af05703 : SUCCESS in 12m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3859dc58937444e693bc95fa15bcce69 : SUCCESS in 9m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4be9289780ef47e5b573f6fd84936192 : SUCCESS in 8m 52s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1dd2ae93281e4596a3c7c191d2616893 : SUCCESS in 8m 34s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8955ba26274e4e3cbc22b4aeb0f882b4 : SUCCESS in 1h 30m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f04a65ef120846aab1bb81670ec2ed10 : SUCCESS in 1h 54m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/93fe6619628d42b5971957bd0f67f681 : SUCCESS in 1h 27m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/aacfe909c71e4737bc1588bc18dba532 : SUCCESS in 1h 40m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1f374bb44645409582504eeef4020c42 : SUCCESS in 2h 00m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fadd3bfb297d47c8a621ac706e489412 : SUCCESS in 44m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/18c2a530707940f69854223651493f1c : SUCCESS in 1h 50m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a2c5ec2ca4884a30abbac3e1d227e4cc : SUCCESS in 36m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8c324b6e616dcad901891a5e8029274938b92385","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-09 02:24:38.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"87b9bf4157cc828b68151ca398fceb65bdbe83ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 02:25:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fcdab49c6928b41b8e31955e1ea5f5fb54704cfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 05:13:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d286ed233ceb4ac5915164c206150925\n\n- grenade https://zuul.opendev.org/t/openstack/build/6829a849edef4f7fa353d0a2f4313945 : SUCCESS in 1h 03m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/098753baa379458284952780dde4d1f1 : SUCCESS in 1h 13m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/274fca6f81254dcaa3f38c16e03cb161 : SUCCESS in 48m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87b40e20498448ad9bccb6bb7ea32a3b : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ba5acb7b0ed4e1da3e5f1bc62d29871 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e7b15cde47447ee820067662b66edb6 : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5cfc21dc4064897b6a04bfc90ec2883 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c334c01986ad43338f75f37cd2f27232 : SUCCESS in 6m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fee1013edb81469d81d9f1ebe45c2246 : SUCCESS in 8m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0858e0be037e4847934b6b1f427e0542 : SUCCESS in 8m 19s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bbdef99f12eb4cdbac95c5aeeb3f5812 : SUCCESS in 1h 34m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/45e1a19253d1400e861f7ebadba97069 : FAILURE in 1h 31m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e85ac4bdf57041be9c19ab0817f1ef2f : SUCCESS in 1h 34m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f17e704e857943768c45b2aefa469ab5 : SUCCESS in 1h 00m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1afa1e6566464aaca38ddf34b77c3272 : SUCCESS in 1h 52m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a607cefbc67d4d7c97946b6865ecbb8e : SUCCESS in 36m 00s","accounts_in_message":[],"_revision_number":1},{"id":"fe9c52ae212d98754d94943b4221ffa911dc39b5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-09 05:18:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"585ef21e8cb13d829d673fea341a0c57427fdf6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 05:31:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faf332ced2ad422aba266bc72e5eb62a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7220985d16d24bd0a9b40e085745dfc3 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca80c7fa422141e39dc1c30f1bd70e20 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b39844cc4e9e4d8b9449f095fcfef606 : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d1682c889e684fb48665fbfa3cbd8dac : SUCCESS in 5m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0de687febf8d5c02e2d67ed2fbe8e13bba39b179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 07:23:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0388def989a419ba5f4714451dbef0d\n\n- grenade https://zuul.opendev.org/t/openstack/build/51516327896f4e9cb08f5d54e6e5e8e2 : SUCCESS in 1h 07m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8ea5b9bc3b384ca0ac18a64e514eae8a : SUCCESS in 1h 24m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1157aef3e3e341eb9bf6a87da4494b5e : SUCCESS in 59m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d45acf27215b4e08b7cb0fb4fc774147 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49f56feac16a425385705c8d34d67c98 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14281a1682f046ff9375d6d08a8966b0 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c812b3cba864498794add76d43dd03d4 : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c2731f86a0c485ea40d42ef73948314 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd8809fb3cc140c0a549e457fb1dc5e2 : SUCCESS in 5m 30s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/96697ba18fed40baa9b95858a3e99110 : SUCCESS in 10m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1086e3e1601b40f58227e53ea7a0ebc2 : SUCCESS in 8m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2cd48900bc8b4f07885264fad5270e88 : SUCCESS in 8m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e9efbe3038c14fe1a872dc48231d65cd : SUCCESS in 7m 43s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a2f39f8b0188447c804e5d12c8aa17e1 : SUCCESS in 1h 23m 44s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a51b886ad4db4dabbaab816768ff481c : SUCCESS in 1h 19m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/18c642e2f01b4c24b10b10bc64ff3534 : SUCCESS in 1h 15m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/daec61d4c6dd44c297f894ab0489752c : SUCCESS in 1h 20m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/326c5415e12f49baba10c8c25a9ec9cd : SUCCESS in 1h 59m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/235fbbd13b3d4fb58acbb0965924d933 : SUCCESS in 1h 00m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/86d8f08cca7a4f96a211f57c290c842e : SUCCESS in 1h 53m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/36e0a8f2ba2742feaab1031cfc47da79 : SUCCESS in 35m 44s","accounts_in_message":[],"_revision_number":1},{"id":"86980912b544d3a7943c815713e4c7650b7818f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 07:23:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24543a977ef372e189b88a48339d1ea834aea127","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 09:21:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39add0826e7647bcb8d03c8e6d9e3822\n\n- grenade https://zuul.opendev.org/t/openstack/build/36ee78b50a044012a254eba329f9b1e1 : SUCCESS in 53m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e534cd89af7a4b109ddf144b64c9b53a : SUCCESS in 1h 13m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a0dd366222b4230807eeb175017f5a1 : SUCCESS in 49m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fe7b5750f5a4a4586d8b297a0f48d53 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05ecf2d667d94b81bf19073a566f785b : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae8d0fd69f3b4f6baa399b54c922aa3a : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1b44e24ca254eefbb49de6108dccd90 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4c2c77e651841cf82b6fd2af806635c : SUCCESS in 5m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0a9e76e5861247d39fef990c540c1c81 : SUCCESS in 8m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/73840e59347e4034b91a10f92e12e274 : SUCCESS in 8m 11s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2b05918147424903b75966f8cef2bd8c : SUCCESS in 1h 38m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/630cb16ce7ce4b17a6e6b9fe5e9203a8 : SUCCESS in 1h 32m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a3eb389e02c04ca393ed374675109e05 : SUCCESS in 1h 38m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f49983a6ad24403e84f050420ff3c7da : SUCCESS in 58m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/19609750570945698a4c2cd8df9ecc0b : SUCCESS in 1h 16m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/26de778c1f334c8e8e597215539bfc2f : SUCCESS in 30m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f36e9663b19eb52570e689dd67666b2ce5001280","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 09:21:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"222b107a7c2877d03888aa00e50ec8d64177127b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 09:22:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37488ed29f9c498f96865e2788413c10\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbfef0a5fed84fddbcd0f5b9e62a563b : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cec35e92bb36f2bfcdaf2cc1f05d7b5a0f3a07cc","revisions":{"cec35e92bb36f2bfcdaf2cc1f05d7b5a0f3a07cc":{"kind":"REWORK","_number":1,"created":"2024-05-08 13:34:14.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/36/918636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/918636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/918636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/918636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/918636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/918636/1"}}},"commit":{"parents":[{"commit":"9a4a3067b5c7c7f8ee9363bd939a3d86b260d660","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9a4a3067b5c7c7f8ee9363bd939a3d86b260d660"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-05-07 18:05:24.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-05-08 13:32:36.000000000","tz":-240},"subject":"Fix broken glance-cache-prefetcher utility","message":"Fix broken glance-cache-prefetcher utility\n\nglance-cache-preftecher utility was failing because it was not\nsetting global threadpool model while initilizing the tool.\n\nCloses-Bug: #2065087\nChange-Id: I0c09dd7c0d81178ebf2aa5efd19c54866fc8791b\n(cherry picked from commit 32ee57ccdeea90a86bce1e5e33d10fab115ed2d3)\n(cherry picked from commit 99540a990fceca89d0e4e73493f4cc05dbd5441e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cec35e92bb36f2bfcdaf2cc1f05d7b5a0f3a07cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cec35e92bb36f2bfcdaf2cc1f05d7b5a0f3a07cc"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
