)]}'
{"id":"openstack%2Fglance~918644","triplet_id":"openstack%2Fglance~stable%2F2023.1~I70b4a970d7e16174f715c54bdae9467cc321c13e","project":"openstack/glance","branch":"stable/2023.1","topic":"fix-prefetcher","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-05-09 13:13:13.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-05-09 16:23:34.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I70b4a970d7e16174f715c54bdae9467cc321c13e","subject":"Handle case for multiple cirros images","status":"ABANDONED","created":"2024-05-08 16:09:26.000000000","updated":"2024-05-09 16:23:34.000000000","total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"cherry_pick_of_change":886071,"cherry_pick_of_patch_set":1,"meta_rev_id":"b629e7bbf98c5425db9fea3f721df09fc865e48c","_number":918644,"virtual_id_number":918644,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-09 16:00:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-05-09 14:03:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-05-09 14:03:46.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-08 16:09:26.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2024-05-08 16:15:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-08 18:01:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-09 05:04:29.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"05fb3891e34b53c585da0b6c9bcd15274e542e9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-08 16:09:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c501cce504bb77c64e1f123b5c69a4e87f02f34c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-08 16:10:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a0427e7aa680fe9fdb920b69fdc0c6e7d645d192","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-08 16:12:43.000000000","message":"Patch Set 1: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30545daee255e377bd281c9996b52490ceac133f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 16:15:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdc8a76c288242999f81268c6aeb66c1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b236fc240334735b5120ae1e8f60942 : SUCCESS in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"75c12ebfb514e6fb3a165226639fd58da1219483","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 18:01:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47c598ca10404cebb5ab00f65fbffc69\n\n- grenade https://zuul.opendev.org/t/openstack/build/a3db4b96cd90480da7e1906609ef9e80 : SUCCESS in 1h 06m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0cb6d1f47d4c407b9a6234461b3c07ac : SUCCESS in 1h 03m 06s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c86b1b88b6754716a092c2e9d7f7b5dc : SUCCESS in 1h 26m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/541b3a2f57b9497cac352ff316e58873 : SUCCESS in 1h 13m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fb629c521ba482498b2c7c0fdcb69ba : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edcdeaa1248c4f179f763aa89a6d7122 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa63300f33474250ad60279dcd8c40a9 : SUCCESS in 6m 55s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/82d350cc3d914a21b296ef5282897b93 : SUCCESS in 13m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cf347a8a87e74538b1d7f90f75a7cef0 : SUCCESS in 9m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4f78ce9980b647a88f2eeac44b38bd24 : SUCCESS in 8m 52s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/20eacd7b2b944a6bb5f4583e10eb58c9 : SUCCESS in 1h 32m 16s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/979034c91e5744739a3a30c0b85ea1d5 : SUCCESS in 1h 38m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9fa0ceb4cbb54414b7b4b1225e602557 : SUCCESS in 1h 24m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/61051ed81e2046f08287d9f82d567883 : SUCCESS in 1h 16m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/27445e9ec973456b978718deb3d10600 : SUCCESS in 1h 21m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/159254cfff3047b9bf992f11e4bf0cd6 : SUCCESS in 59m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ddb0abf027c74b43b0e638a039a2a3d3 : SUCCESS in 1h 46m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a88cbf2bb8b74fcda981b66395536044 : SUCCESS in 38m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f4eb52b2f6936c2049b6a566f1109adde7effd33","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-08 18:10:16.000000000","message":"Topic set to fix-prefetcher","accounts_in_message":[],"_revision_number":1},{"id":"f9123d4d6e50c16d9b4a07a880e91282b242b344","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-09 05:04:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a571e6e5e13945654713b3106e44dd02b4385925","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 05:05:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00d50aaecdcd1b6762174d6eea95b0c4cb6a3b26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 07:00:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/96cf37274d7d4db2988aa641a469f4b8\n\n- grenade https://zuul.opendev.org/t/openstack/build/8ff45cf659ee4c26a65521d8ab5eaafa : SUCCESS in 39m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/69687fb8f6544cb08ff4820e9ee08d3b : SUCCESS in 59m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7641354091d441ea9dc6f20464d7747e : SUCCESS in 52m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/124fc416eb794b06898c27b79a931c7d : SUCCESS in 55m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fd6ae07ce654e9d92176e73779f205a : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d64a17431274d6ba7be125f3deabcf5 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81b3659b9375450e848e7636f17846a4 : SUCCESS in 5m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b6fe9d17f91442108806ecb3f0f6d2cb : SUCCESS in 8m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/96c59cb938cf44fd98640ee97ca23afc : SUCCESS in 8m 42s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/218261dcba2c40998213c2f9ce170bed : FAILURE in 1h 14m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7b1c68554d1546d88dfd6706e8dfb27b : SUCCESS in 1h 20m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/df62c42ec4ee4dc5aba0860dd67885f0 : SUCCESS in 1h 24m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e4380c24a2e4cb88836833f65e1df25 : SUCCESS in 1h 09m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/113c1fd0fc224fb4b91a47dd520d01be : SUCCESS in 1h 49m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/82b61bcdfff34a2c96dd0272314c63e8 : SUCCESS in 38m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8bb875ad9c002e6604445f9f7602b9c6dcc288f6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-09 07:08:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f1f3bdc6846b2c9cc0eb18155be2345b9b78c66e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 07:13:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50bf3f5eb9be4667b54bcf4f9ff0322d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/daa8d742e7d94e52a68af6bbf5f7a87e : SUCCESS in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9969b875dcde9de74faab8b13a8c112f8f3765ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 09:06:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/154fe49af23f4a2f9ee181b25d19e3d7\n\n- grenade https://zuul.opendev.org/t/openstack/build/7f44b59257224bc88b20fbf70111e864 : SUCCESS in 59m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e312bb2ab1124baabf9913957cb88eb7 : SUCCESS in 59m 21s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5cb0687cf83f4e758acc6a7cb98700ab : SUCCESS in 1h 20m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1538548fe3a24becabc2b67bc91be3bd : SUCCESS in 53m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b13cf76d8b4b4698838be6941179dd48 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c17e526b1db41068bf162332bf5ae60 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a0f6185c5b04e45a1f62c9aa54b9be9 : SUCCESS in 5m 52s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2ac43623efe34438af8e68e5c06e592b : SUCCESS in 11m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3bb94ca193af4b87a71098a25bd90b25 : SUCCESS in 8m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/886b4f5832394fd99b31097e54432e0c : SUCCESS in 8m 46s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c91f68fc68f1494bad720355070b809a : SUCCESS in 1h 36m 11s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ef4983c809364a78abc5e3c3f2bc87d2 : SUCCESS in 1h 18m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/18e48789c7b849de9cf0bd86a3e39cf3 : SUCCESS in 1h 23m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/11750817b5f74ee88221d50e14f5b110 : SUCCESS in 1h 23m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fbcce96b52c440abb17a498f237cc18e : FAILURE in 1h 09m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a47c3ff56c94ee29f1af66fa75ccae8 : SUCCESS in 54m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e95066817fe04ebea05e1edf58b692a0 : SUCCESS in 1h 48m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3323bdc643234089b3e8e6a9c1d9bc90 : SUCCESS in 37m 50s","accounts_in_message":[],"_revision_number":1},{"id":"c16dba93485a824ef1edd82877def92fd5bf4fff","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-09 09:07:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08c9a95c0283202b90711d6c868a38e4ffd25dd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 09:12:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff11f14062d24cf8a388574e23e2c7b9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5edcda23c84940e186b4537b29a2cde5 : SUCCESS in 3m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"679ce5b981450fa83ef6ec1ee5db3193fed5e285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 11:37:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/747d8d99062349df81266f1aca9aa5ba\n\n- grenade https://zuul.opendev.org/t/openstack/build/7524d1554cf549ec85124990f77d64e5 : SUCCESS in 46m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2d7775ae79a04476a52aa494337b064c : SUCCESS in 41m 58s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de335da0ad374f25b042f65aeeaaa707 : SUCCESS in 57m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ba65d75418f4194b2ed6d883aa476fe : SUCCESS in 40m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f3f3c29b6334b8fae2238d3415de21f : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/888b173ff2cd4b04a6f5796f1bd3cbd0 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7691337549c641afb5ec15259ebbca1c : SUCCESS in 5m 52s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a3cd94923453471c8759d4654102a95d : SUCCESS in 10m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3411eedd9cd049dca0dca602801c0c83 : SUCCESS in 8m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/850a881f15cf47fb8d36670f4a1d9583 : SUCCESS in 8m 34s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fcf80e99bf6c46ad9be3d3ae04000936 : SUCCESS in 1h 41m 48s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6f7f5f439e7c4b5db461462581e3e395 : SUCCESS in 1h 15m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/65478b7d208f4fcca2cad9749998414b : FAILURE in 50m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e2c9d382964242a7825460a64f0bc7c4 : SUCCESS in 1h 21m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0b6c22c3c5314620bd3c568b49667c21 : FAILURE in 1h 31m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a229640d6a844b598122dfed9bcb964 : SUCCESS in 35m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b266202c0bb34130809f18fb48becbf4 : SUCCESS in 1h 30m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5f96751a53644fb79cf97a437d1dd82b : SUCCESS in 23m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f466062e11b2311fa234f184087bcb8bc05c7567","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-09 13:11:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"303fd00113b50484ee16a927dccad84c2c297bf4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-09 13:13:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5daacfe81295172ac65922b4ed4948f78d248461","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 13:18:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32c3a443632c452ea658c2b4edd57921\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9949657bd16e4261af75d28cdf006ed1 : SUCCESS in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9765bfd5e49b01dc33be7ca99c0998bbd67687f2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-09 13:18:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84a810dc515bedb77074d8dba09b92e25b299296","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-09 13:38:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fd1736637d2aa6e5beee602c18649c3e28a38b1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-09 14:03:46.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9e40d6d4481a865148093b67bad3e29d50fe4e82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 14:09:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6bffb6d814a4656844faf0a56dd66cd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/206a268b604c407da6bc1a193c4ace7a : SUCCESS in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"01e2c2d602fdc531111dc609e0544376aef5ead9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 16:00:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9adbde65686642b89409bf204b1aa554\n\n- grenade https://zuul.opendev.org/t/openstack/build/d77a5e6b5d824225b1021f2fbdc2effc : SUCCESS in 1h 03m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/569be7dd03d04767a78564ebb82b3421 : SUCCESS in 1h 01m 34s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/863172a3c8f04273ad7ebbbd2ceb389e : SUCCESS in 1h 16m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b362f4803f66455da32e9955fe823663 : SUCCESS in 1h 02m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5610c21d4004df397bc05dd13e3bc09 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b5b444bf6784c9cbc351b7f77d4b0d8 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee906e046de744b59619d869a4a8d02e : SUCCESS in 5m 57s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c2ca182043944366bc4795480cf4cba6 : SUCCESS in 13m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dfe640d9832e4891badb53bd984b06c2 : SUCCESS in 7m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c99a93b2bc234148aa74836346c200a0 : SUCCESS in 8m 37s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b5f3dd556c2548b69c316abac0e47383 : SUCCESS in 1h 30m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1777baf2457f4ad7b1fa46a4130044ab : SUCCESS in 1h 08m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3f82500870fe4b8fa2b3b96a27d1c566 : SUCCESS in 1h 27m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6b2664a08eed4f9695b3a7373e7caf46 : SUCCESS in 1h 27m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/142360d352384f089e564ba3bdd60578 : SUCCESS in 1h 29m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db48b9a31d994b59b9cc404267bf0c07 : SUCCESS in 43m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c040d18fda774760ae6b6735af067534 : SUCCESS in 1h 19m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c7976440540248f9b07c26927c62ad9c : SUCCESS in 38m 38s","accounts_in_message":[],"_revision_number":2},{"id":"b629e7bbf98c5425db9fea3f721df09fc865e48c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-09 16:23:34.000000000","message":"Abandoned\n\nThis has been squashed into https://review.opendev.org/c/openstack/glance/+/918734","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"98657968a3359fe93e2f484140a6a73b218961cd","revisions":{"f333a86485039c98d5f620a1360a3eec38256849":{"kind":"REWORK","_number":1,"created":"2024-05-08 16:09:26.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/44/918644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/918644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/918644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/918644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/918644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/918644/1"}}},"commit":{"parents":[{"commit":"285021bdb0ea344f995508fabb8b4e2485f2cf3c","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/285021bdb0ea344f995508fabb8b4e2485f2cf3c"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2023-06-14 09:24:03.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-05-08 16:09:26.000000000","tz":0},"subject":"Handle case for multiple cirros images","message":"Handle case for multiple cirros images\n\nThe change[1] configures multiple images for the devstack-tempest\njob and all the jobs deriving from it. This causes problem with\nthe current post-check-metadata-injection.yaml playbook since\nwe only expect one cirros image and perform steps accordingly.\nThis change modifies the playbook to handle multiple cirros\nimages.\nFailure in the glance-multistore-cinder-import job can be\nseen here[2].\n\n[1] https://review.opendev.org/c/openstack/tempest/+/831018\n[2] https://zuul.opendev.org/t/openstack/build/68a4d3ec6ce04c87b21d73a333f5b5cd/log/job-output.txt#23161\n\nChange-Id: I70b4a970d7e16174f715c54bdae9467cc321c13e\n(cherry picked from commit d47803613d501b18f9fa4f6aa3c899bd16f9ddde)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f333a86485039c98d5f620a1360a3eec38256849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f333a86485039c98d5f620a1360a3eec38256849"}]},"branch":"refs/heads/stable/2023.1"},"98657968a3359fe93e2f484140a6a73b218961cd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-09 14:03:46.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/44/918644/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/918644/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/918644/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/918644/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/918644/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/918644/2"}}},"commit":{"parents":[{"commit":"714f2ab5fbe40bac79cb78bcecca7a68e1a4f1ab","subject":"Adjust settings for integrated-storage-import jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/714f2ab5fbe40bac79cb78bcecca7a68e1a4f1ab"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2023-06-14 09:24:03.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-05-09 14:03:46.000000000","tz":0},"subject":"Handle case for multiple cirros images","message":"Handle case for multiple cirros images\n\nThe change[1] configures multiple images for the devstack-tempest\njob and all the jobs deriving from it. This causes problem with\nthe current post-check-metadata-injection.yaml playbook since\nwe only expect one cirros image and perform steps accordingly.\nThis change modifies the playbook to handle multiple cirros\nimages.\nFailure in the glance-multistore-cinder-import job can be\nseen here[2].\n\n[1] https://review.opendev.org/c/openstack/tempest/+/831018\n[2] https://zuul.opendev.org/t/openstack/build/68a4d3ec6ce04c87b21d73a333f5b5cd/log/job-output.txt#23161\n\nChange-Id: I70b4a970d7e16174f715c54bdae9467cc321c13e\n(cherry picked from commit d47803613d501b18f9fa4f6aa3c899bd16f9ddde)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/98657968a3359fe93e2f484140a6a73b218961cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/98657968a3359fe93e2f484140a6a73b218961cd"}]},"branch":"refs/heads/stable/2023.1","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
