)]}'
{"id":"openstack%2Fglance~918734","triplet_id":"openstack%2Fglance~stable%2F2023.1~I6cd5b8bbf39f0cfae466f0800faea5a74e960e69","project":"openstack/glance","branch":"stable/2023.1","topic":"fix-prefetcher","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-05-10 02:51:12.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-05-09 16:26:07.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"I6cd5b8bbf39f0cfae466f0800faea5a74e960e69","subject":"Adjust integrated-storage-import jobs","status":"MERGED","created":"2024-05-09 13:34:34.000000000","updated":"2024-05-10 02:53:52.000000000","submitted":"2024-05-10 02:51:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918734-fix-prefetcher","meta_rev_id":"7dce8bb43d5321c43730b9e50a0e8068a0f17adf","_number":918734,"virtual_id_number":918734,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-10 02:51:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-05-09 16:26:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-05-09 16:26:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-09 13:47:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-09 13:50:23.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-05-09 15:22:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85c2d9382c73585941cff75fa1f63a325a820ca5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-09 13:34:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f19703f0373f4830c4b9d29cd4fbb61f809c1a1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 13:47:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c3f51a3e81b42acb9876f085224f1ff\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a831fc8c37642cbbb1087a47306f49f : SUCCESS in 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"33727813afd7b15050ea1102b6d9ffb52b8cae98","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-09 13:50:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8d1de5fd52c795568d3a072067b702f7f976a77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 15:22:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/027076c584894810878e43e3f8a7a79f\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/58ad842b4f05431494049fa27e498612 : SUCCESS in 51m 18s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42ce669d8f764b6cb8c1d880a3a5165b : SUCCESS in 57m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4f7f4b6d2f84f2288596f40499085e2 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d912e50271d5454d86a09f819131d2ab : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bb86fe49d774c4b8e812bb1edc8624d : SUCCESS in 6m 49s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8c261bac7b0c47ff968e8e700fda59d7 : SUCCESS in 11m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/249b008955d742a4bb7be1f351f56000 : SUCCESS in 9m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ecfb2e64c7bd4c31aeb967fd7491d4a5 : SUCCESS in 9m 16s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a3e667fc4a1b4ac29a1b10c2bd95c86e : POST_FAILURE in 1h 29m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9908484320be4db691cae11c66d6f21f : POST_FAILURE in 1h 21m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4e0025a247164ee3a26bb1046f905d39 : POST_FAILURE in 1h 47m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f52bc83a2903494485cada07f97088fc : SUCCESS in 37m 20s","accounts_in_message":[],"_revision_number":1},{"id":"bacbafc76a447ae3ae56a5d54826716c1579119a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-09 16:00:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e206dbbe757948c102f097948d8801348c0a0820","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-09 16:22:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f0fa7be2402939bed70d81f3a011fe965a6d8c8d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-09 16:26:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f274fe5c06a5a84fb28462487a92d0c778216afb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 16:33:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fe97aae1ff14053aa8bbb166ac7f9e9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4810bb860f5e482c9dede0fcea8cf0a8 : SUCCESS in 8m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b79677247827d6ef593377c8f062ef3ef3af7d09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 18:26:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90d6548bf05042fd8c70587bbfdc7ca9\n\n- grenade https://zuul.opendev.org/t/openstack/build/98067f65eaae4c98883e04cf8a0f05ae : SUCCESS in 1h 01m 02s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b9302fc138a14100bff6cc945e1beebe : SUCCESS in 57m 51s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/68e2191ad6d143adb0a9772b2e95339a : SUCCESS in 59m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85e8effcead848efbd5641ff6b14e214 : SUCCESS in 48m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9f71e766b3f402fbdba6c3db6b46da3 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b849cc2d511439bb7fa62011981ebd2 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf5b925696c74162b0ba9b23ebe6697b : SUCCESS in 6m 13s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/10c10131ed13439b8141b74844115de7 : SUCCESS in 16m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/575a243755e6448b9fbc02cd6bed1e15 : SUCCESS in 10m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/96af8f7bcebc4c53b149300e14c4ce7b : SUCCESS in 9m 07s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f054a8fea60349858dc5fadb439ecbf1 : SUCCESS in 1h 41m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f7c975a38e7e433a8eeba0d4f656e18f : SUCCESS in 1h 11m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ba2a9a4464d64898acfc4fb752a4d714 : SUCCESS in 1h 28m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/13c7e3fa5ea74f798407ead78bed393c : SUCCESS in 1h 41m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5c2b907bf4474ea0a3f4efd6ea38a34a : SUCCESS in 1h 55m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16419297176c48bf94a981c0c76e0b72 : SUCCESS in 50m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/99dce9bc100e4086b5bc3545ed530800 : SUCCESS in 1h 49m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/de7391c5012e49e5aaaa05e22812bd76 : SUCCESS in 35m 27s","accounts_in_message":[],"_revision_number":2},{"id":"5bd8cfe5d55bb95f7e2f165e67256d045165f383","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 18:26:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"119acc535c716db2114864ebec2c8cce2fa6d6c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 20:37:56.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/735771c164d84581a774d9cf703bdee9\n\n- grenade https://zuul.opendev.org/t/openstack/build/6ae1c931483a48a58862def156376541 : SUCCESS in 1h 01m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/40d4f56c6de545c08dd95cf96c452197 : SUCCESS in 1h 00m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e37225899b964ec0b866ebef3ec2d34b : SUCCESS in 1h 22m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7108982bfc4147b997de089596f0ec06 : SUCCESS in 43m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8750862009194d8c991cb24ad9a3a096 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8585d2e4801445cb829edc1fe937e22 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c634d8b8bf9940a9b3fe1c503589fc81 : SUCCESS in 6m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/24592cf08fc24e4f99d6f0e60f1e31f0 : SUCCESS in 9m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/02b218d33499487b98af2f95272d2433 : SUCCESS in 9m 11s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6e589ac951ed401e96246caa5bde8ace : SUCCESS in 1h 21m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/401d89382b8645fa8d2b25e0329c9346 : SUCCESS in 1h 30m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0646d63043024ef1b0aafe3df8ac571e : TIMED_OUT in 2h 04m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a820fe0de2346d7be142cf00df1b6b0 : SUCCESS in 55m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1997deb89d6f46818ea78a981b4441fb : SUCCESS in 1h 58m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c71cbf9346a54a85aabfccdffda5ef75 : SUCCESS in 37m 30s","accounts_in_message":[],"_revision_number":2},{"id":"21ccec6b7411e386d5b023c44d24bd9b20488157","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-09 22:47:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"847cabc4f7823ed6e5dc2e74e74bd99ffb5cbcb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 22:52:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0a0cd8de3994d898a0e8c7e3e3f9685\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/41f64719d88e4565bb396444b693ffd0 : SUCCESS in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e0013bf4a29fd5b7270e93ea0d656d60e5fecb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 00:38:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18afc50b027749fbaa98961adaec1b40\n\n- grenade https://zuul.opendev.org/t/openstack/build/9ca99d51338546d184da14fec6c865a1 : SUCCESS in 1h 09m 00s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/15ae45660eee4476b2bcc03fc0212132 : SUCCESS in 1h 00m 57s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/19ad514b267244b889a8e2ab94b64460 : SUCCESS in 1h 22m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a43f730e751d41ce974a3b2d30a56805 : SUCCESS in 1h 01m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b50c9b4a3bfa470a942af56fc63cc07b : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27d6034ee5f149de9777598b976c3cb7 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bb1029856ac44db983c87a3daa4c1e9 : SUCCESS in 5m 25s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/50ec14aa637949caadf5b3bc59a21925 : SUCCESS in 12m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6d2c24efccf34e0b9a8640dbf3af0965 : SUCCESS in 9m 25s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e7b76fd7e0bb49788b34bbc02bffee25 : SUCCESS in 7m 58s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8552e9f3dd66425a9f203a92a551ad6b : SUCCESS in 1h 45m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0c0bb2e596234465b6fac756c6f8e51e : SUCCESS in 58m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8b12054f6ae54a2aba695b321b5c1445 : SUCCESS in 1h 31m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/494f0612455546e5be45abd79d9d6cdd : SUCCESS in 1h 14m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9f93823023bd449fa85f8c0f667e43e1 : SUCCESS in 1h 14m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d293ad3cedf642c6b35fb690fb54f2fe : SUCCESS in 35m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/00e96bf82e5840ab82b2ef8844f45cdc : SUCCESS in 1h 15m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/535b37ad2fde4f629c8f59d9bf7db4fe : SUCCESS in 22m 01s","accounts_in_message":[],"_revision_number":2},{"id":"3922867c8beb2e8bf5bcedd614d7046c30f10eb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 00:38:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"63a128ab0737760af1cc2e77c7fcad173ae510f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 02:51:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfdaf089069e440d86c0887b3de9207f\n\n- grenade https://zuul.opendev.org/t/openstack/build/dc14d49bc6b043fb8642f72bb9a8a045 : SUCCESS in 44m 05s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6ff958318333428e936342f95b49f905 : SUCCESS in 39m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/87636939679c485c890ae795ef6eb2b9 : SUCCESS in 57m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f412edbea37c4eb09bc9768004d58b2d : SUCCESS in 57m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d15f78827bf64f66b67b6f21825ea9a0 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce2129e7b8b1453ab280699513b01143 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c11518da3aa546f28772bea739aa4476 : SUCCESS in 5m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bb4d575f79b94295a0f8200c5629f4d5 : SUCCESS in 9m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cad711aaea5d460c99507b82579b48cf : SUCCESS in 7m 31s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/58a6633273024ad598a9c689cec5607c : SUCCESS in 55m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6079faa75d52446ea4d078d6e7847a97 : SUCCESS in 1h 30m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3428f12356b44d318d0e942067e676cc : SUCCESS in 1h 34m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/076537e9860f487fa1b1ab243252c316 : SUCCESS in 35m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6720ba5a1475442790cca8b4cf18403e : SUCCESS in 2h 06m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2e6131dedddc48b8821168a735aca4e8 : SUCCESS in 39m 08s","accounts_in_message":[],"_revision_number":2},{"id":"e6edc63b64829e036cc2f628e875e34b12ae982e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 02:51:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7dce8bb43d5321c43730b9e50a0e8068a0f17adf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 02:53:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c0a6cb4acce419ab099304c284a014b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6342eaa2afdc4d82b7d369fd40f6f838 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"065e2d7e9e0c605cd4e60c9b150337b6ac6a9fde","revisions":{"714f2ab5fbe40bac79cb78bcecca7a68e1a4f1ab":{"kind":"REWORK","_number":1,"created":"2024-05-09 13:34:34.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/34/918734/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/918734/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/918734/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/918734/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/918734/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/918734/1"}}},"commit":{"parents":[{"commit":"285021bdb0ea344f995508fabb8b4e2485f2cf3c","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/285021bdb0ea344f995508fabb8b4e2485f2cf3c"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-05-09 13:30:57.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-05-09 13:30:57.000000000","tz":-240},"subject":"Adjust settings for integrated-storage-import jobs","message":"Adjust settings for integrated-storage-import jobs\n\nWe\u0027re seeing OOM killing mysql, and timeouts.  These were addressed\nin Bobcat and later by increasing swap (I073216d1bbddbd3bdecd2) and\ndecreasing tempest concurrency (Ieb96eb6ceb6f2a9673dfee).  Not doing\na direct squash of these because the former included a change\nrequired by a modification to Bobcat and later devstack, so not needed\nhere in Antelope.\n\nChange-Id: I6cd5b8bbf39f0cfae466f0800faea5a74e960e69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/714f2ab5fbe40bac79cb78bcecca7a68e1a4f1ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/714f2ab5fbe40bac79cb78bcecca7a68e1a4f1ab"}]},"branch":"refs/heads/stable/2023.1"},"065e2d7e9e0c605cd4e60c9b150337b6ac6a9fde":{"kind":"REWORK","_number":2,"created":"2024-05-09 16:22:56.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/34/918734/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/918734/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/918734/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/918734/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/918734/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/918734/2"}}},"commit":{"parents":[{"commit":"285021bdb0ea344f995508fabb8b4e2485f2cf3c","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/285021bdb0ea344f995508fabb8b4e2485f2cf3c"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-05-09 13:30:57.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-05-09 16:06:10.000000000","tz":-240},"subject":"Adjust integrated-storage-import jobs","message":"Adjust integrated-storage-import jobs\n\nWe have to address two issues simultaneously:\n\n1. handle multiple cirros images\n\n   The change[1] configures multiple images for the devstack-tempest\n   job and all the jobs deriving from it. This causes problem with\n   the current post-check-metadata-injection.yaml playbook since\n   we only expect one cirros image and perform steps accordingly.\n   This change modifies the playbook to handle multiple cirros\n   images.\n   Failure in the glance-multistore-cinder-import job can be\n   seen here[2].\n\n   [1] https://review.opendev.org/c/openstack/tempest/+/831018\n   [2] https://zuul.opendev.org/t/openstack/build/68a4d3ec6ce04c87b21d73a333f5b5cd/log/job-output.txt#23161\n\n   This was change I70b4a970d7e1.\n\n2. change the the job config to address resource constraints\n   * partial backport of I073216d1bbdd to change swap size\n     (we don\u0027t include the change from that patch that was needed\n     to address changes in Bobcat-era devstack)\n   * backporting Ieb96eb6ceb6f to change tempest concurrency\n\nChange-Id: I6cd5b8bbf39f0cfae466f0800faea5a74e960e69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/065e2d7e9e0c605cd4e60c9b150337b6ac6a9fde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/065e2d7e9e0c605cd4e60c9b150337b6ac6a9fde"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
