)]}'
{"id":"openstack%2Fglance~918875","triplet_id":"openstack%2Fglance~master~I5c58cb11179014e69b6456f643a596e4f1e48ded","project":"openstack/glance","branch":"master","topic":"bump_sphinx","hashtags":[],"change_id":"I5c58cb11179014e69b6456f643a596e4f1e48ded","subject":"[DNM] Testing docs bump with new Sphinx","status":"ABANDONED","created":"2024-05-09 19:26:52.000000000","updated":"2024-05-09 23:08:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"100036489f36125d13a2eabbcda9bb6079049e37","_number":918875,"virtual_id_number":918875,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-09 21:55:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-05-09 19:38:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-09 21:55:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cfbe67d8b682dcc5667249a04f714bfa82af8d21","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-09 19:26:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8db8cc22a1708f642e55f6ee51103fdef49c31ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 19:38:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0eae227e352421282c93cce01b7326a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0375ad34bb04a0d9a1ed96c12c53c03 : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/01f3382476ca46868c7814109e095c06 : SUCCESS in 8m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c05bd4da2cd271cafaba45b21037fd30fd56bae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 21:55:24.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/664695ad6af94381ba0f15b37cf7a339\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/493a321d11874781bbb73b28a2f2fc60 : SUCCESS in 1m 57s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b63d91586d284730b1a0c748e2806d52 : SUCCESS in 6m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4cc597aedd744808872184210942ff83 : SUCCESS in 2m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/ba8339754ffc40d294bbe3e5a9147841 : SUCCESS in 44m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d401519500a8451183104e550c386822 : SUCCESS in 1h 21m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a4e880aa5ef47d2afda2d4ce92d8b63 : SUCCESS in 50m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0817b821cb142f9b21f117bf31747cc : FAILURE in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/722764f725964ae098e81d657205af16 : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/02fa91b69a6445cebc335ad4c2283bec : SUCCESS in 5m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec95d59c0f3043c29850c421296523dd : FAILURE in 9m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42c568c49fce48d8b076772b4816d689 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4f9d4e7285142328244c054951ef712 : SUCCESS in 7m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/72dbb6ecc1764a05b5e0468ccb4fdbcc : SUCCESS in 14m 42s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/76cb58ca18ad4501828beccd921fbee7 : SUCCESS in 8m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6e0733aaf830416daa1d852d6dc7f56a : SUCCESS in 9m 04s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f331704393ad42e080b448f211b10625 : SUCCESS in 9m 06s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/bec35cd0844340db878b19685ae09d2a : SUCCESS in 10m 36s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8a732d18282743868fa2accc9bb3d35e : SUCCESS in 1h 51m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6eb57cf5239d410c8885547d835f5462 : SUCCESS in 1h 38m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7b98587a958c4d1e86d71efab0bf5e44 : SUCCESS in 1h 21m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6410e4e4ac7a4646b0b101ae45c1fa52 : SUCCESS in 1h 46m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/72495917980443e186d1f36dd2692062 : SUCCESS in 1h 59m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcde3a257d41487ebfb33a2077a6278a : SUCCESS in 1h 01m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5d2e0391ec294c1d9cf3bcaed9cdeb3d : SUCCESS in 1h 50m 13s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/4fcac89e19c4450b8441fe904d5c9fee : SUCCESS in 1h 52m 39s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/bb837bc868b2425598e550d9231a13d1 : SUCCESS in 1h 03m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0fabf9d9870b425586938ed8946ad90f : SUCCESS in 34m 45s","accounts_in_message":[],"_revision_number":1},{"id":"100036489f36125d13a2eabbcda9bb6079049e37","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-09 23:08:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b3efe683fa6d42e2e5a6a1ddd6442203c27e444","revisions":{"0b3efe683fa6d42e2e5a6a1ddd6442203c27e444":{"kind":"REWORK","_number":1,"created":"2024-05-09 19:26:52.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/75/918875/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/918875/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/918875/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/918875/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/918875/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/918875/1"}}},"commit":{"parents":[{"commit":"32ee57ccdeea90a86bce1e5e33d10fab115ed2d3","subject":"Fix broken glance-cache-prefetcher utility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/32ee57ccdeea90a86bce1e5e33d10fab115ed2d3"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-05-09 19:26:46.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-05-09 19:26:46.000000000","tz":-300},"subject":"[DNM] Testing docs bump with new Sphinx","message":"[DNM] Testing docs bump with new Sphinx\n\nSphinx updates can be scary.  This change leverages a constraints bump to see what breaks.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/916565\nChange-Id: I5c58cb11179014e69b6456f643a596e4f1e48ded\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0b3efe683fa6d42e2e5a6a1ddd6442203c27e444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0b3efe683fa6d42e2e5a6a1ddd6442203c27e444"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
