)]}'
{"id":"openstack%2Fglance~919643","triplet_id":"openstack%2Fglance~stable%2F2024.1~I96ae5ba24a9e32994f4b7b111a8b56e56db33cbe","project":"openstack/glance","branch":"stable/2024.1","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-20 08:38:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I96ae5ba24a9e32994f4b7b111a8b56e56db33cbe","subject":"Remove SQLAlchemy tips jobs","status":"MERGED","created":"2024-05-15 03:54:36.000000000","updated":"2024-05-20 08:42:19.000000000","submitted":"2024-05-20 08:38:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919643-sqlalchemy-20","cherry_pick_of_change":915561,"cherry_pick_of_patch_set":1,"meta_rev_id":"7cea8e4918e19dc157d3ee5e161a367fd03e96d4","_number":919643,"virtual_id_number":919643,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-20 08:38:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-05-20 06:24:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-05-20 06:24:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-15 03:54:36.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2024-05-15 04:05:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-15 06:05:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-20 06:24:33.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"486ffed11bdf5a29a37798ce7f90ab51492a29db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-15 03:54:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2c85dee689556da84c7f8295f22d4c80b8168e03","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-15 03:54:58.000000000","message":"Topic sqlalchemy-20-stable/2024.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"2c6bd4d784bb0a4f15eaa406cbc3b0975ec91b44","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-15 03:55:01.000000000","message":"Topic set to sqlalchemy-20","accounts_in_message":[],"_revision_number":1},{"id":"23bec5143482402678261894ce701c55b2d61c3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 04:05:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aff12a21a360417d93e27189457e9c20\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ff291f0e3654390befb6a0bb6b770b5 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f24b1c32c93f42ef94371f740e39c1a1 : SUCCESS in 8m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ce05c9a4f2e435706f529f1df243913be97c0297","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 06:05:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbc86c58a50240e59bd8e1d347ab1ca3\n\n- grenade https://zuul.opendev.org/t/openstack/build/62971fcdead94fa4be28150e8784965e : SUCCESS in 1h 04m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/60c9167e5eee4d3d806b5ff149b8cb02 : SUCCESS in 54m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cdfd002726a64101b1e196ac48af0804 : SUCCESS in 1h 21m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7b9b44c7af94f8ebc29b10264f7e362 : SUCCESS in 57m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/321f8b163c8a4768ad11031b6d7f0981 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c02e54f0aa6e4a9cbd705f2db26b61f9 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb17f4b03572410292ad3c8db3998cee : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94130573a89b4126b9902f1039893c82 : SUCCESS in 6m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/97b4c3c1194148849c9ed1e91a1e8e86 : SUCCESS in 13m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/97f609fef4da4702a1edbaec304bdd36 : SUCCESS in 8m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e0f6323c7f6044a98eb0ecc783a6c879 : SUCCESS in 8m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9ea5657d11db460192b61bdb8f542f78 : SUCCESS in 7m 28s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5a9f7859bc424e628f45cb86cddd3f7d : SUCCESS in 2h 02m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8bd478659606445f89b803da583892e9 : SUCCESS in 1h 24m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a9d766a8bf1340069e5064bf67bb755d : SUCCESS in 1h 16m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/70492c5cd6ad4373b4c324fb6d1a7999 : SUCCESS in 1h 52m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0fa52726c3174b348183ea919cf9b6dd : SUCCESS in 1h 27m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72b31270c0584f39af0d6ef6446468ae : SUCCESS in 45m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0181a0cab5e4497f9224ccaae33e3b53 : SUCCESS in 1h 58m 25s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/84ce0d8f258c46feb490580ab7165cb4 : SUCCESS in 1h 47m 32s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d2290980c6574e0d8a325abd56376967 : SUCCESS in 1h 10m 09s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9a198dd0fa5f477eb0ef7f617d522bea : SUCCESS in 36m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6db6a27a366da607de9de9b468ca927c86f0a6ae","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-20 06:24:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0815158507f293543493ac346ab5a759655e99b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 06:25:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"767d2ebd579c652d073c0e1cb7c632ac951e6281","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 08:38:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a26a70fe069b4bf0af230b7457decfd0\n\n- grenade https://zuul.opendev.org/t/openstack/build/dac5a03c0ccd476895f985b6326acf1e : SUCCESS in 1h 03m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5acee7c5e9764ac49784cf32d4ed7437 : SUCCESS in 1h 00m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/224d2795669449f5af8c3e46d8655680 : SUCCESS in 1h 25m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff6bede184024a4e924a63de11ed2dff : SUCCESS in 1h 01m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5a3cf7dc5ad453cb97953b1bed68292 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/377bbf8d11a548f7b67cda7f8be9a7e1 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f9965e135dba4087b8d82e75e4adbb08 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3d2d31ab6d14cce8a3f1cf93ba1a7b7 : SUCCESS in 5m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bb9d92d3b6e74d0c842325697c0b6c7f : SUCCESS in 8m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/da37cda8284b4c73b285513673a405ad : SUCCESS in 8m 41s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/166ab5b91d9f4744bd41a83d7dc7b99e : SUCCESS in 1h 37m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2eb4b7fe96334ce1b97422117917e640 : SUCCESS in 2h 00m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b575dc9c740e4ecfa5636b01af09959c : SUCCESS in 1h 31m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31f5cb915b104c5a910996d44d469616 : SUCCESS in 1h 00m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d68ce977a84e4638bdff955d5ceccd84 : SUCCESS in 1h 30m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8152eed6efdc4686b4a78c75a80a4d01 : SUCCESS in 32m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ab0698c0fdd826dd3cbfa5d51386c4071d48532b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 08:38:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7cea8e4918e19dc157d3ee5e161a367fd03e96d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 08:42:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/194aec95726646d19168a29441690d13\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70493d6af471481d8a0fdb5142da78da : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb3eb4ea269f608a7b534bb205f917309420cd51","revisions":{"eb3eb4ea269f608a7b534bb205f917309420cd51":{"kind":"REWORK","_number":1,"created":"2024-05-15 03:54:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/919643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/43/919643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/919643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/919643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/919643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/43/919643/1"}}},"commit":{"parents":[{"commit":"99540a990fceca89d0e4e73493f4cc05dbd5441e","subject":"Fix broken glance-cache-prefetcher utility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99540a990fceca89d0e4e73493f4cc05dbd5441e"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-04-12 10:39:09.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-15 03:54:36.000000000","tz":0},"subject":"Remove SQLAlchemy tips jobs","message":"Remove SQLAlchemy tips jobs\n\nThe most recent SQLAlchemy and Alembic versions are now in\nupper-constraints. As a result, this job has served its purpose and can\nbe removed. For more information, see [1].\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/RBHXHTO3GUOOXVSZXD4C2O3TKDOH2QSC/\n\nChange-Id: I96ae5ba24a9e32994f4b7b111a8b56e56db33cbe\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 513311f9752166a333e00f94653c70c49a4a4914)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb3eb4ea269f608a7b534bb205f917309420cd51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb3eb4ea269f608a7b534bb205f917309420cd51"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
