)]}'
{"id":"openstack%2Fglance~920170","triplet_id":"openstack%2Fglance~master~I6cbd11fd0ee5758169a2e0e9961660edb844e65f","project":"openstack/glance","branch":"master","topic":"bug/2061947","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-07-11 22:55:27.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-06-12 08:53:01.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"I6cbd11fd0ee5758169a2e0e9961660edb844e65f","subject":"Fix 500 if multi-tenant swift is used","status":"MERGED","created":"2024-05-22 10:09:16.000000000","updated":"2024-07-11 22:56:42.000000000","submitted":"2024-07-11 22:55:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":36,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"920170-bug/2061947","meta_rev_id":"e3e7ebcd03cd935cbd124b81931d2d8fc49f1fe0","_number":920170,"virtual_id_number":920170,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-11 22:55:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"all":[{"value":1,"date":"2024-06-27 08:44:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":2,"date":"2024-06-12 08:53:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-07-11 14:22:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-07-11 15:40:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-22 10:19:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-22 12:36:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-27 07:35:54.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-06-27 08:44:10.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"},{"updated":"2024-07-11 14:22:24.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"84e8a9ded68b60e6f8477f4f8d5dcb02df61ec9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-05-22 10:09:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5ca363ab75b84ea412abca3c9e17709b8054aa5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 10:19:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92b1b39fbf4d4c35a5875ebef3440662\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9669ff8fd48945d0961a59b55ac143c9 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/416e11d943ab4abf9c8a28af4abead87 : SUCCESS in 8m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40fe06953003204e6f8914d2fdb38c7ae4b72210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 12:36:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84474cdf3225459bb5fe76f47e5caad6\n\n- grenade https://zuul.opendev.org/t/openstack/build/5f386871d0a64e4aa5003cc54e881707 : SUCCESS in 55m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1c038146752543d5b0d347a76e4fed95 : SUCCESS in 1h 14m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cf10e11893749b69846edb4be2feba5 : SUCCESS in 1h 05m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f79e1930f1b74b8dbe5a9b97f8b47912 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e85fd6bf59da4b3fa17d256aa3883b4c : SUCCESS in 7m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1884c53e6f8b4a15a1d3444d281609c6 : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5176b6655d7346009d10f7a25da09c8d : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa8ec2c0bcad4baab97730f473173d89 : SUCCESS in 6m 31s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2d743a8b13f5409cae58f0fbeede5f6f : SUCCESS in 14m 59s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/55d9f43df992453eaec396d655d9992f : SUCCESS in 8m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/504e76243ad84d5ba1397b96c70cc4e7 : SUCCESS in 8m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/94bbefcc1803418c880317917928cc50 : SUCCESS in 8m 18s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6b81557d8b734ea7beaed6b2d9cbcd84 : TIMED_OUT in 2h 03m 22s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a37dc86a68d84ce580d1a7ef268682bd : SUCCESS in 1h 33m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/be33d0afd22e4dbc9ac8e9775b382271 : SUCCESS in 2h 02m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/98e74a7b5a5e4afc834dc7f1d5e93317 : SUCCESS in 1h 46m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/96cdf25af65647368ea5d55038f01969 : SUCCESS in 2h 11m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6585764cdcb435598cd3cdab18478b3 : SUCCESS in 1h 06m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5aa3f6c5af7c468dbb9c4a28f5228fcf : SUCCESS in 2h 12m 06s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/ffbde8677e6c4b65ac99ff1806ff74cd : SUCCESS in 1h 47m 17s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/cdade81514094ccd8697f26b191b089d : SUCCESS in 1h 00m 40s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c371311a4f32462b9bc43245546a9d69 : SUCCESS in 24m 49s","accounts_in_message":[],"_revision_number":1},{"id":"11b0e61128cc429330ce2b60674e7446ec4675df","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-27 07:35:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"14c9caa53b4a966e27c8e819bed12b084bc3ab2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-11 07:45:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c5c61d3b1e855de0ba74cc06694fa18e1622678b","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-11 07:46:34.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7544cdf51f26b151e2a05981fde33adedadd4e8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 07:51:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e46ac27a89a4d42b603e3c2b418c699\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5c149da2ef3434f9b414be0d605c4ed : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b7026edf0cf84dec892b7411b2cd0b1a : SUCCESS in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4e1c66e8f1416d608c9a09301d857f92b780a992","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-06-11 07:52:25.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dc97f836522c82d88471f129b171f1d5dae51e4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 10:07:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84f7e465bb3e4ac6876179f9f3786e82\n\n- grenade https://zuul.opendev.org/t/openstack/build/9f2e925924244e69809cb35200148536 : SUCCESS in 1h 01m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/32a8bf5613774f5cab631a78e270b574 : SUCCESS in 1h 41m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ba64a6593134bff9abc1b104a5aebbc : SUCCESS in 1h 07m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edbc04bbe56b4b2da1be007a35264fb4 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/940c84f42ce445e18d90ee7181e3ab19 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c9ee018b017842c1b4af4a8ce93c466f : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac84d3bdca464025afe465112236fd3a : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ce5e4e74ae44ba7ad5a1bb65d96fa56 : SUCCESS in 6m 15s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1fe5b1e5eaf840cc9bdfc51824c67aed : RETRY_LIMIT in 1m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a54baca23b5344c99653e16f5fbbb476 : SUCCESS in 8m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/207105ea65084008a56077dd217e5c10 : SUCCESS in 8m 03s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f510145bd53a4b2a980c73ba0f5a6fbf : SUCCESS in 11m 40s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8c2a21cacd06478bb5c6754c5249e383 : SUCCESS in 8m 41s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/05f87239e1d34255a19054c031f880e3 : SUCCESS in 1h 56m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c45072222bf64191a377f0709ff95090 : SUCCESS in 1h 47m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/57ffa5e026034704af91f0c4063037a7 : SUCCESS in 1h 55m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/60cab28649584472b6e13f16e6bf2b60 : SUCCESS in 1h 49m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bed0b054a9fb4796b2ad6003c1fbedf7 : SUCCESS in 2h 13m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d2d73450b3043a6b2eee8d3751f2cc2 : SUCCESS in 1h 00m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3b4047a0d14c4042b18f727426806af1 : SUCCESS in 2h 16m 29s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/956692097fc841e5b8c86246021aba07 : SUCCESS in 1h 54m 57s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d1957cbc71aa4073bae5c840d53479c8 : SUCCESS in 1h 10m 47s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e583c86aa3b24171a21ce6a951a14942 : SUCCESS in 40m 26s","accounts_in_message":[],"_revision_number":2},{"id":"769c90eafc5921253ab098dee0a45dbbd12c0779","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-11 13:32:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a7d9d9a3f5b720b19e5988d144bc75d66ac3b12c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-11 13:32:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"98ccc5c06566410b43cb54bc326f30ec1e255831","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-06-11 13:40:49.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"24ff226669c6cdf2953b950ac6d0120d9541855a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 13:45:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b248660d36c741998dfbcb0d82ab6698\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ecc5e34c70ea456a84b622e1e352dbfc : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0f680abe1ce24291a31ab996c736976c : SUCCESS in 11m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a7fe839915a42bd649a3a6f664c38f100391187a","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-11 14:02:13.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1939c1a32932af7513fb6afc58cc78d5b0593a1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-11 14:02:54.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"3147b8d11a826a87164fa712fa0745e31f820412","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-11 14:03:35.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cd4bc694fed16831274b12a1607f06e2ea724405","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-06-11 14:09:56.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"20832efc8d46c641d2f8e69a986113f94d3cdd3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 14:10:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc3c575f26604f5483c518ca41d69f1d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/06d930769c8c430f92854dae8868e8cc : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6d35de292b304f4bbb895f4b6598ba15 : SUCCESS in 6m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9c43ed0ec3242a9f065469df5a38f549fa748d04","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-06-11 14:11:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d06198570a0239ae465a878df9453ff18e235207","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 16:27:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2be964f5567b490c84acb717c7b810f0\n\n- grenade https://zuul.opendev.org/t/openstack/build/02f16cb3c46b4e449539e2773a5929f7 : SUCCESS in 1h 17m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/54e87906b4e6448fba580dbb686fbd94 : SUCCESS in 1h 34m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/161fdc08779847649e62a79b59f7f12b : SUCCESS in 1h 09m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47b1397bf06f4f928b2ae714af1f7176 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd43a68937df48f5954db710bf7ddde2 : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/acb40096f9654d48935e04b835d550a1 : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/21609b9094f4455ab994a850e27ea25f : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d24d2d564c544ae8879fc5d76f89286 : SUCCESS in 6m 08s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d4f65a42d76e4b00853bde88d402fcf3 : RETRY_LIMIT in 1m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/948c15579ea0415398ac559feb2a66d3 : SUCCESS in 8m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/95ec33bfb1784a40800ffe5245cc424f : SUCCESS in 7m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/66001c3261db4bd79161534bcdda03e5 : SUCCESS in 13m 46s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/32abccf1fc3742e293b19f4eb831abd1 : SUCCESS in 8m 08s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4f764576b5694243bfc977eabfe0d89e : TIMED_OUT in 2h 03m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9702cc5b2d004c4da9a375d6db4b7e96 : SUCCESS in 2h 01m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/de616d1183b64baab389b28eefd548a0 : SUCCESS in 1h 57m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9e86aee9d04342b4a945b90de5cbf508 : SUCCESS in 2h 03m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/350ccf1887034564910dfb2dde05afec : SUCCESS in 2h 17m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37a1249a745f44dca8c3ded489b3a55b : SUCCESS in 49m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c86367f08d7b48b0bf214cceaf3b6afd : SUCCESS in 2h 08m 00s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/8fd43fd2c3f34df994c763c030a8d732 : SUCCESS in 1h 55m 15s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9d1d5c17194e45ac9f41504c07cbd189 : SUCCESS in 1h 04m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6e8aae8b9c6244f8b01b849ef776f314 : SUCCESS in 38m 24s","accounts_in_message":[],"_revision_number":4},{"id":"78aa6443d8dfb7d135c723516c3d84ff5d9069f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-12 03:42:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"34c89f9f3e2c86beaeb451104ced507d98fafd1d","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-12 03:43:25.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"882d9aec9a02634d1106164f76e8a4fa2fc54b06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 03:49:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23d23392b97448f2a37b86a9c6021425\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b3fae04411954a0888275b8fb9157708 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/83687763228540c5b064fc27584c34e7 : SUCCESS in 5m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b93068f624dccb76ff970320b480823d79298d64","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-06-12 04:52:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3940b6e66c735be5d8289680f1623cd119313fe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 06:38:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a934bafc7db49378e95f4b767e2fe6c\n\n- grenade https://zuul.opendev.org/t/openstack/build/c53de43904b54917b6208e92a2e8c552 : SUCCESS in 1h 03m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/987d7ef91fea491da34044ca34d6679b : SUCCESS in 1h 35m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6aa0ff654fb4a35bcfb7282fc6b9d17 : SUCCESS in 1h 14m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d81ba157e2040e1aa9ab30c2ac901e4 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a1e55a8a9d041028f3ae29799be588c : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5f13852f9c248f08bfa912612aaf082 : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/77d65d6bf818471780b595974c3a8607 : SUCCESS in 10m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/886329b7580047b588c7a888f88961b8 : SUCCESS in 5m 12s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2f487ae5fd3c4600920e64e095216807 : RETRY_LIMIT in 1m 36s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d5a34fdb3e4b424193870bd5161ce5f7 : SUCCESS in 8m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a721a57f04ea4a45a9d6c9fa95b0e9e8 : SUCCESS in 9m 03s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/069e12e1e3584f369a2d9cf5067ccf30 : SUCCESS in 12m 05s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7115b89bfd8a448d83eeb791270bd4b7 : SUCCESS in 9m 38s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d6d669a8e1ef4fada84ff59d7b68e432 : SUCCESS in 1h 53m 11s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/836a71f6fcad40a2887b9f0ad562a50c : SUCCESS in 1h 42m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b0790aa146514ed78f68adce77192b63 : TIMED_OUT in 2h 05m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6d74de98f07d41f49b0d9a98be6bcc5d : SUCCESS in 1h 49m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9846dc9b4a804b98b6d7c3daaa12f031 : SUCCESS in 2h 49m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/413a4a3a8f414a1cb80246828b93e715 : SUCCESS in 55m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/53ee92a1ca7a486db96e0eecf7585926 : SUCCESS in 1h 55m 45s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/093ac3063c3b4070890fb252b73cdf6d : SUCCESS in 1h 49m 54s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/925529b237f243d6a52d23c145f32de0 : SUCCESS in 1h 04m 35s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e2e4e879e1a247a5895d4ef882a8b58c : SUCCESS in 36m 39s","accounts_in_message":[],"_revision_number":5},{"id":"0f7f7da00a68410f04f382fe414b847f5c978dfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-12 06:59:32.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"49b240a23557bed50c99fdc178addaae16543da4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 07:06:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64671970b48e434587699a5eea3b86db\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ae2ca64e88054a878f48e181397c6aa0 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/010dd682cfaf4c3180717301296c42c9 : SUCCESS in 5m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4fff4dbb1c46cd2aef56d33a612e7bb41bed4262","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-06-12 07:06:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"86c173f85489bf9c3aed9f73a3b918e2ef21d0dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-12 08:00:04.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"410ef395fd85b5cf15558d331913125962055c45","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-12 08:00:36.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a1e75a2d90942a3c801daec4c23c92868be8776f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 08:05:43.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0bcd573fd1c4aa8bfe6d93e67607162\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a466eb2464e4521bffd1c7bf05cf644 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a3c59437b06b4b60a32ea6a581f048e4 : SUCCESS in 4m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cb2f539bcac2ddc2b7ab735f98247bc31c493b78","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-06-12 08:53:01.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"3b8913c416f00d343de63793c2e2d578c7fe8d78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 10:24:58.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/311173ad97f548fc88ea6f8dcc6f69ac\n\n- grenade https://zuul.opendev.org/t/openstack/build/78c6e7bf339e4e08ba22ec3d56566e62 : SUCCESS in 1h 05m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bbd7344560cd466680309485c36a06b3 : FAILURE in 1h 25m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de478aa2d6b84c5ab90c441e42383f0b : SUCCESS in 1h 07m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e10483405d42407fb07cb41e23cc0bc0 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16836c80f9b44e66833091f657739b34 : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c587f4b3d754e699e1aee8dd232bb83 : SUCCESS in 5m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7fd1655a68db446a911721f2746c761e : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/434a2a18db1741c680bd63925408d107 : SUCCESS in 6m 43s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/896188602c91446fbef05dd1920086c4 : RETRY_LIMIT in 1m 47s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/229d0eaca0504576be4d42cbf97baaf6 : SUCCESS in 8m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a3ae8c53fd1f415490f4b45a1528e45f : SUCCESS in 8m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6f9e9589b5c342a8bd735ec35bba3577 : SUCCESS in 12m 48s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/362c86deeceb45b49142987172aef76f : SUCCESS in 8m 39s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/14e152eda7e444dc95d3ebdf67942ad9 : TIMED_OUT in 2h 03m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/958bd8effc124bfeb63f143c871d2784 : TIMED_OUT in 2h 03m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/acc892a76ba0464c87aad798cbd1318d : SUCCESS in 1h 54m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2a80edb9752c47b986fd60c5e5a3b43f : SUCCESS in 2h 04m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9c342d218ccd442c8fd2338115fee818 : SUCCESS in 2h 19m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12436218218a45be8f67cda42f7499aa : SUCCESS in 57m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c56775738e7a4c5b84c7bbbe86ca0a9a : SUCCESS in 1h 57m 12s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e016dafb6add4c14b0066511caab748d : TIMED_OUT in 2h 04m 30s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/27379fd92b694594b251f5a371d32239 : SUCCESS in 1h 13m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/df82b84a67944c7ebc29d0eed4ab8e97 : SUCCESS in 38m 53s","accounts_in_message":[],"_revision_number":7},{"id":"86c661dcec8a83083efa488bd8643ae8761fa72b","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-20 13:09:34.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4da94d99a4f318cbe28e3fd39b44e17c18f8971e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 13:15:49.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f41f5674b0340959c52554346446b9d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/221a0ba8f9bf4688930688b5566fb4df : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eaea0d106f1c473d93a5db63b7d4bd5b : SUCCESS in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"763f8429a91b68dd2ddf16f363cb1289d92c17dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 15:26:17.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/67e3d31b30db4a0daea48cacddf7c4e3\n\n- grenade https://zuul.opendev.org/t/openstack/build/ff0fcf91a76e416da877607fb9ec7a47 : SUCCESS in 56m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c74593e58bf74a12a099d2fffcea9cb1 : SUCCESS in 1h 26m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24fb98a3a39c4f3c8acc3ed78507909d : SUCCESS in 59m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/264e5a4518f740f8ae0f60e0676d4be1 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a1a9232d0ce4c5dac52e4932c19a5d9 : SUCCESS in 6m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52287e7de30a4c9fa127021a99992a55 : SUCCESS in 5m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc742f52e3cf4293a7803e498603e123 : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0090d594817548c6bbc782d3abc98380 : SUCCESS in 5m 33s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4c1e0a6f53ba443ea1cb77d1a14c2ec9 : RETRY_LIMIT in 2m 08s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a6f18a06575a417a82fdd06329dbc078 : SUCCESS in 8m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3b000e31e7fd47a782a78c0f05ff1803 : SUCCESS in 8m 48s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0496525931724be7808c0097c91172fb : SUCCESS in 13m 46s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/379d13d97a02424bae1f01e92b8642e5 : SUCCESS in 8m 40s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c09203cdf6274be2ad61bda2ac402be1 : TIMED_OUT in 2h 03m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1be48ffc032f46bf981999ab017f64e1 : SUCCESS in 1h 55m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7a4982df68744b17aa0e365876f47602 : TIMED_OUT in 2h 04m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c853710155604a3e9cb16f6aaedc60f9 : TIMED_OUT in 2h 04m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/23312958c3fd43eb878ba161a916227d : SUCCESS in 2h 10m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7e36a32fa5a4a4f8be9a9a6f4750778 : SUCCESS in 1h 02m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/26d7aa0cf8e440f0b48a80ccacb89a2b : SUCCESS in 2h 08m 19s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/92adea11c92149358a599dc047a89c4c : SUCCESS in 1h 47m 55s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/dc1ee28b99bd446fa0de28a16ec29bc5 : SUCCESS in 56m 11s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f640d69a399d45a59f62674e6f0b53b3 : SUCCESS in 28m 17s","accounts_in_message":[],"_revision_number":7},{"id":"729a09de6920fa521e4a71c90f80ce0ee793eb01","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-25 09:46:01.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ef1faa71518b1a275855e1eeb6e41656e1ba8f4a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 09:51:28.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfaaf6a4dbc644319d81c6768781d875\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/94cdbf9d8eef406186df7c7e073d9358 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/39b1121a9c9040889af11a6b45c35d0a : SUCCESS in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a3a7da1cdbeb7fbf408c70f1b1360886cd226964","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 12:03:39.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/218baae29b71456fa7baa41bd549db08\n\n- grenade https://zuul.opendev.org/t/openstack/build/16b88c4de60840d1bc1156eece1d5128 : SUCCESS in 1h 04m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4310bfc98ec94340afcb886278d9c05c : FAILURE in 1h 10m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc9c27830991451fb2cbd9237f6fc0a7 : SUCCESS in 54m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc5013572a024739a41f22dd0a89feba : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f9204f150424ccbb2f8fcd4b11e3ee1 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7227983222cf466e9327b39803d76add : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cef85d80c74c4602a8001ff30d08573d : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/085224a7fde04614aee4e85464b89e6e : SUCCESS in 5m 39s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d26647407e18474fa90c728061ddd6d6 : RETRY_LIMIT in 1m 31s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0306265b6c0149c8b13b9e1fa6ff9397 : SUCCESS in 8m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5c15c8f452e242548352f0852e1afab6 : SUCCESS in 8m 21s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9327a9a82b964cfb89c8ee7152bea20d : SUCCESS in 12m 11s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6e4e19d72259471eb4d6087713945dee : SUCCESS in 8m 41s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e9bf5be5f6054afebd96bfed00292b77 : SUCCESS in 1h 44m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/97bac900120243c4b4321891f4f55a57 : SUCCESS in 1h 46m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6a8e5f60a7bb4dbe8a23e0102a15c267 : TIMED_OUT in 2h 04m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/113f6961534f43f9ae394069a4242309 : TIMED_OUT in 2h 03m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/03f841d9a7ef42e1a577dbf8986b5ca6 : SUCCESS in 2h 16m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e83cc88e453486f9a8f368acc90bc6f : SUCCESS in 44m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a68c7466840e4ccda692354dc3c0f531 : SUCCESS in 1h 49m 14s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3fae51a5c54648f9ba5f273eadd04c57 : TIMED_OUT in 2h 03m 33s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8dafe69c0dda465ab9cc6e182dfefabc : SUCCESS in 1h 19m 26s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/086f5aaf7b1c4381955a887122e4500e : SUCCESS in 32m 40s","accounts_in_message":[],"_revision_number":7},{"id":"a50b29be902b36c0a8c430b386ee5211054097d7","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-06-27 08:44:10.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ad970221213a3cfacce33baef9cf360f5635393d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-11 14:22:24.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"f1126dd25eae2dc132b0e5d91c55e298184dc3a2","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-11 14:32:26.000000000","message":"Patch Set 7: -Workflow","accounts_in_message":[],"_revision_number":7},{"id":"be76d026cdc2b9b173d24bbcd9aacdaf3d3b3a42","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-11 14:33:06.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3d8eef7c217225ec7a57e62fc74416375a3d1a0b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-11 15:15:04.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d3f65734a60d760a6b4365538350764a409cb8ed","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-11 15:40:58.000000000","message":"Patch Set 7: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7d50865f1ec3ee19f6f933c705ba34ad550bf967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 16:35:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/452a455e5f4d4fa9854adc79ce6f3167\n\n- grenade https://zuul.opendev.org/t/openstack/build/cd948c18d01f440fb0777bfaf802107a : SUCCESS in 1h 01m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b7cf42579b8242e18e40abd7d678d943 : SUCCESS in 1h 29m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d90643ff89fe4d6ca406dfc13fb77ec3 : SUCCESS in 1h 01m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/359bc4797a00488ab6a0b84eea783203 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6813dc7ae4a94929ac2b50c79af3827a : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df8d0934c1e748598b03ac254bf56e70 : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2ab736385c374d31a7161de98f9ec926 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6b7af66f9814a139f67587d33eb18f3 : SUCCESS in 7m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea82e56e8099473da2abcc0b39712e93 : SUCCESS in 8m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/41d4afa40c2d4696b864990c603e5bc6 : SUCCESS in 8m 20s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3db5bfba9aa9471abcee746f1359076b : SUCCESS in 12m 24s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0856ab4b359b4ec9925b29c712b2380b : SUCCESS in 8m 28s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/05ebed2469bd461ca7e12cc0092cc1db : SUCCESS in 1h 55m 31s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2394076f42934713bbf610c6c725ddeb : SUCCESS in 2h 06m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9a37603279d64c8689341524889d172d : SUCCESS in 1h 47m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d3c5d8a1756845909f350667b751c678 : SUCCESS in 1h 41m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d8c9907834ba4333b0ef13d2ef446927 : SUCCESS in 1h 54m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/954a0d0ddc9744e79c42542bf97b1f33 : SUCCESS in 48m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1eeab11362a94c83b096906197f0d5f0 : SUCCESS in 1h 29m 22s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/92ca5c6a6c52449d82dbf440d9a19166 : SUCCESS in 1h 47m 11s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f5d65ae459d24bba91884ac5e40cdfd6 : SUCCESS in 1h 01m 48s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2d897ccd914c4ac09dafd1b313649391 : SUCCESS in 24m 11s","accounts_in_message":[],"_revision_number":7},{"id":"ace207b6b2e0b18885793a732e3e1c57c810b7fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 16:36:33.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"bd85ca7a18acc34a2c8db2accf52549b29f59e69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 22:55:26.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f67fa3fa98f470e8610fbda0e649714\n\n- grenade https://zuul.opendev.org/t/openstack/build/a34bd5e9268744828b58775507b7ad3e : SUCCESS in 1h 15m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9166b965281640e3ae8a0eb1d9585a3e : SUCCESS in 1h 31m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/854a150812784fc4aa5c4882a3ea1c6b : SUCCESS in 1h 03m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e65d95a90df4b10a9c2dbd19b3eb4c8 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d19ef1d865d409595784df40b1d795e : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6672945e49b43faa153f1347cb5297c : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/608ade37ab834cdbba05c7ff05ab1d8a : SUCCESS in 6m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/09c17c294ce541298700a01f3cbf1d43 : SUCCESS in 9m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c072dc864b3e44ce92dc117f97ecb250 : SUCCESS in 7m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b668647b193d47d89a13086af4d13d1a : SUCCESS in 14m 25s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f5a11d0d0e274701870703237ae2323c : SUCCESS in 1h 49m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a6cb499f82034c98b859c1fed8d5fd12 : SUCCESS in 1h 47m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a283a6d8f78a4405886ead479bdb071a : SUCCESS in 1h 43m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/985ffb16729b4d8f8a84fa2fd88bc229 : SUCCESS in 1h 03m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cd366626c0c64da3b1f537dab9563619 : SUCCESS in 1h 39m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a9c84381e400417c82b123bf2eb9581a : SUCCESS in 25m 44s","accounts_in_message":[],"_revision_number":7},{"id":"eaf28a88054345f543910f83f7af136e264a7817","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 22:55:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"e3e7ebcd03cd935cbd124b81931d2d8fc49f1fe0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 22:56:42.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e84405b61834231bd1bc65152b33bf2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5e421416d0d4ee5ab3573c764ddf385 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"06c727cfb32db38d9d705b28236d699e1c2c2083","revisions":{"e3ee4a3557b4301b26f392f7d93e017ed0ffbcb4":{"kind":"REWORK","_number":1,"created":"2024-05-22 10:09:16.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/70/920170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/920170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/920170/1"}}},"commit":{"parents":[{"commit":"da1b9792f269cb87b97142005927833cbeab3f32","subject":"Merge \"Docs: be more explicit regarding oslo_limit.endpoint_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/da1b9792f269cb87b97142005927833cbeab3f32"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-22 09:51:59.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-22 09:51:59.000000000","tz":0},"subject":"Fix 500 if multi-tenant swift is used","message":"Fix 500 if multi-tenant swift is used\n\n1. store-info --details and image-create apis fail with 500 if\n   swift configuration file is set in g-api.conf when muti-tenant\n   swift store is used. Fixed this by raising HTTPBadRequest with\n   appropriate error message.\n2. Also, store-info --details fails with AttributeError, when\n   multitenant is enabled for swift store since it doesn not set\n   the container value during store configuration.\n   Fixed this by returing `None` incase of multitenant swift store.\n\nCloses-Bug: #2061947\nChange-Id: I6cbd11fd0ee5758169a2e0e9961660edb844e65f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3ee4a3557b4301b26f392f7d93e017ed0ffbcb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3ee4a3557b4301b26f392f7d93e017ed0ffbcb4"}]},"branch":"refs/heads/master"},"e0c5805fb410419415ab131c298625328560f58f":{"kind":"REWORK","_number":2,"created":"2024-06-11 07:45:55.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/70/920170/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/920170/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/920170/2"}}},"commit":{"parents":[{"commit":"dfbfdad77fdc91899f64d2ff186097cfb2065ab7","subject":"Move all periodic tips jobs from py38/39 to py311/312","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dfbfdad77fdc91899f64d2ff186097cfb2065ab7"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-22 09:51:59.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-11 07:44:03.000000000","tz":0},"subject":"Fix 500 if multi-tenant swift is used","message":"Fix 500 if multi-tenant swift is used\n\n1. store-info --details and image-create apis fail with 500 if\n   swift configuration file is set in g-api.conf when muti-tenant\n   swift store is used. Fixed this by excluding the unregistered\n   conf parameter and added log warning.\n2. Also, store-info --details fails with AttributeError, when\n   multitenant is enabled for swift store since it doesn not set\n   the container value during store configuration.\n   Fixed this by returing `None` incase of multitenant swift store.\n\nCloses-Bug: #2061947\nChange-Id: I6cbd11fd0ee5758169a2e0e9961660edb844e65f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e0c5805fb410419415ab131c298625328560f58f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e0c5805fb410419415ab131c298625328560f58f"}]},"branch":"refs/heads/master"},"695a80a530930d7ee4d0ed2b190dbb22518fe928":{"kind":"REWORK","_number":3,"created":"2024-06-11 13:32:27.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/70/920170/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/920170/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/920170/3"}}},"commit":{"parents":[{"commit":"dfbfdad77fdc91899f64d2ff186097cfb2065ab7","subject":"Move all periodic tips jobs from py38/39 to py311/312","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dfbfdad77fdc91899f64d2ff186097cfb2065ab7"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-22 09:51:59.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-11 13:27:55.000000000","tz":0},"subject":"Fix 500 if multi-tenant swift is used","message":"Fix 500 if multi-tenant swift is used\n\n1. store-info --details and image-create apis fail with 500 if\n   swift configuration file is set in g-api.conf when muti-tenant\n   swift store is used. Fixed this by excluding the unregistered\n   conf parameter and added log warning.\n2. Also, store-info --details fails with AttributeError, when\n   multitenant is enabled for swift store since it doesn not set\n   the container value during store configuration.\n   Fixed this by returing `None` incase of multitenant swift store.\n\nCloses-Bug: #2061947\nChange-Id: I6cbd11fd0ee5758169a2e0e9961660edb844e65f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/695a80a530930d7ee4d0ed2b190dbb22518fe928"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/695a80a530930d7ee4d0ed2b190dbb22518fe928"}]},"branch":"refs/heads/master"},"04dd124131c39a97a4d8828552eb9cbac3dbf51c":{"kind":"REWORK","_number":4,"created":"2024-06-11 14:02:54.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/70/920170/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/920170/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/920170/4"}}},"commit":{"parents":[{"commit":"dfbfdad77fdc91899f64d2ff186097cfb2065ab7","subject":"Move all periodic tips jobs from py38/39 to py311/312","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dfbfdad77fdc91899f64d2ff186097cfb2065ab7"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-22 09:51:59.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-11 13:58:04.000000000","tz":0},"subject":"Fix 500 if multi-tenant swift is used","message":"Fix 500 if multi-tenant swift is used\n\n1. store-info --details and image-create apis fail with 500 if\n   swift configuration file is set in g-api.conf when muti-tenant\n   swift store is used. Fixed this by excluding the unregistered\n   conf parameter and added log warning.\n2. Also, store-info --details fails with AttributeError, when\n   multitenant is enabled for swift store since it doesn not set\n   the container value during store configuration.\n   Fixed this by returing `None` incase of multitenant swift store.\n\nCloses-Bug: #2061947\nChange-Id: I6cbd11fd0ee5758169a2e0e9961660edb844e65f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/04dd124131c39a97a4d8828552eb9cbac3dbf51c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/04dd124131c39a97a4d8828552eb9cbac3dbf51c"}]},"branch":"refs/heads/master"},"cb67bb13edc62c8d2c551b53cb710fc42cf72b02":{"kind":"REWORK","_number":5,"created":"2024-06-12 03:42:14.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/70/920170/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/920170/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/920170/5"}}},"commit":{"parents":[{"commit":"dfbfdad77fdc91899f64d2ff186097cfb2065ab7","subject":"Move all periodic tips jobs from py38/39 to py311/312","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dfbfdad77fdc91899f64d2ff186097cfb2065ab7"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-22 09:51:59.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-12 03:36:48.000000000","tz":0},"subject":"Fix 500 if multi-tenant swift is used","message":"Fix 500 if multi-tenant swift is used\n\n1. store-info --details and image-create apis fail with 500 if\n   swift configuration file is set in g-api.conf when muti-tenant\n   swift store is used. Fixed this by excluding the unregistered\n   conf parameter and added log warning.\n2. Also, store-info --details fails with AttributeError, when\n   multitenant is enabled for swift store since it doesn not set\n   the container value during store configuration.\n   Fixed this by returing `None` incase of multitenant swift store.\n\nCloses-Bug: #2061947\nChange-Id: I6cbd11fd0ee5758169a2e0e9961660edb844e65f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb67bb13edc62c8d2c551b53cb710fc42cf72b02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb67bb13edc62c8d2c551b53cb710fc42cf72b02"}]},"branch":"refs/heads/master"},"059176a007b293d57b0f7cc8b5f1a9a25a4f1361":{"kind":"REWORK","_number":6,"created":"2024-06-12 06:59:32.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/70/920170/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/920170/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/920170/6"}}},"commit":{"parents":[{"commit":"dfbfdad77fdc91899f64d2ff186097cfb2065ab7","subject":"Move all periodic tips jobs from py38/39 to py311/312","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dfbfdad77fdc91899f64d2ff186097cfb2065ab7"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-22 09:51:59.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-12 06:57:18.000000000","tz":0},"subject":"Fix 500 if multi-tenant swift is used","message":"Fix 500 if multi-tenant swift is used\n\n1. store-info --details and image-create apis fail with 500 if\n   swift configuration file is set in g-api.conf when muti-tenant\n   swift store is used. Fixed this by excluding the unregistered\n   conf parameter and added log warning.\n2. Also, store-info --details fails with AttributeError, when\n   multitenant is enabled for swift store since it doesn not set\n   the container value during store configuration.\n   Fixed this by returing `None` incase of multitenant swift store.\n\nCloses-Bug: #2061947\nChange-Id: I6cbd11fd0ee5758169a2e0e9961660edb844e65f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/059176a007b293d57b0f7cc8b5f1a9a25a4f1361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/059176a007b293d57b0f7cc8b5f1a9a25a4f1361"}]},"branch":"refs/heads/master"},"06c727cfb32db38d9d705b28236d699e1c2c2083":{"kind":"REWORK","_number":7,"created":"2024-06-12 08:00:04.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/70/920170/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/920170/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/920170/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/920170/7"}}},"commit":{"parents":[{"commit":"dfbfdad77fdc91899f64d2ff186097cfb2065ab7","subject":"Move all periodic tips jobs from py38/39 to py311/312","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dfbfdad77fdc91899f64d2ff186097cfb2065ab7"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-22 09:51:59.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-12 07:58:06.000000000","tz":0},"subject":"Fix 500 if multi-tenant swift is used","message":"Fix 500 if multi-tenant swift is used\n\n1. store-info --details and image-create apis fail with 500 if\n   swift configuration file is set in g-api.conf when muti-tenant\n   swift store is used. Fixed this by excluding the unregistered\n   conf parameter and added log warning.\n2. Also, store-info --details fails with AttributeError, when\n   multitenant is enabled for swift store since it doesn not set\n   the container value during store configuration.\n   Fixed this by returing `None` incase of multitenant swift store.\n\nCloses-Bug: #2061947\nChange-Id: I6cbd11fd0ee5758169a2e0e9961660edb844e65f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06c727cfb32db38d9d705b28236d699e1c2c2083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06c727cfb32db38d9d705b28236d699e1c2c2083"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
