)]}'
{"id":"openstack%2Fglance~920679","triplet_id":"openstack%2Fglance~master~If9238dc4243b1bd23ee0a61a1531ff784ad19655","project":"openstack/glance","branch":"master","topic":"voting-functional-py312","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-06-20 20:35:34.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"If9238dc4243b1bd23ee0a61a1531ff784ad19655","subject":"Make functional-py312 voting and add it to the gate","status":"MERGED","created":"2024-05-28 15:05:52.000000000","updated":"2024-06-21 01:25:12.000000000","submitted":"2024-06-21 01:21:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920679-voting-functional-py312","meta_rev_id":"cef81760b6e697ea5a763efe26a182adc6d07cd1","_number":920679,"virtual_id_number":920679,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-21 01:21:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-06-18 23:09:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-06-20 05:56:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-06-20 05:56:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-28 15:06:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-29 06:22:16.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2024-06-12 04:29:05.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"},{"updated":"2024-06-18 23:09:43.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-06-20 05:56:47.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"be2d8507b2441bd95dfef93145dd8ddeb7196886","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-05-28 15:05:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e49c06def60a96c2c731d933f788af9ffe0fbe26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 15:06:16.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/glance on branch master.  The\nproblem was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][3][\u0027openstack-tox-functional-py312\u0027]\n\nThe problem appears in the the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - integrated-gate-storage\n        - openstack-python3-jobs\n        - openstack-python3-jobs-arm64\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n  ...\n\n  in \"openstack/glance/.zuul.yaml@master\", line 361","accounts_in_message":[],"_revision_number":1},{"id":"246d1e44232b71fd21c07e92fa19bc1cfb21b51d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 15:06:23.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/glance on branch master.  The\nproblem was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][3][\u0027openstack-tox-functional-py312\u0027]\n\nThe problem appears in the the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - integrated-gate-storage\n        - openstack-python3-jobs\n        - openstack-python3-jobs-arm64\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n  ...\n\n  in \"openstack/glance/.zuul.yaml@master\", line 361","accounts_in_message":[],"_revision_number":1},{"id":"b8176340a77a1db936e8c75ddf4e71e8a11d26bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-05-28 15:10:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9f5470169bd9cad695a0b8967c753839fd8c21d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 15:16:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50fc58dc81f54f0597129cd6477b67ee\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91d8a48eefc14b66b495e8d403c5b828 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7202ae5deb1b440da7ee5c266a276ab1 : SUCCESS in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c4fe25ba5b51fb4d1de09ff1312f5ac69bb88f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 17:50:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/760f7f126ee64c13b6e0a716d6377a12\n\n- grenade https://zuul.opendev.org/t/openstack/build/d7aad995fd2d46ba97dbfefa2cf28986 : SUCCESS in 52m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/54a0740a2a084c918d7b05edb6723f14 : SUCCESS in 1h 16m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b74e95d3323848d389c29ef66d7c78a0 : SUCCESS in 1h 08m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0b4fb62640c4c04bb127649c0a2d8ac : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ef009e3c7314ff7baff3005924ec8ac : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6c8950f7f9341c5a60ca067b0f4b051 : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6137c8e0940433ab976d440b95a673e : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bebac9d32fa1423382fb0636216546af : SUCCESS in 5m 33s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a1fdb54c73a1416691bc0b3aeb3aea6a : RETRY_LIMIT in 2m 36s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6b39679dd9264c52965b0cd43da8e3a5 : SUCCESS in 9m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/785359211fc54d2a85eb6041bf6619aa : SUCCESS in 9m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d443aa75c48e42adb9b978daa99501bf : SUCCESS in 11m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/518b48c851ca4e158a64085f04758219 : SUCCESS in 7m 50s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a423cfc4342745ef9f9bbddc70d8106a : TIMED_OUT in 2h 03m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/258ff288e65946b0a884ad544f0501ea : TIMED_OUT in 2h 04m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c3413feeec7b4154a724b675cdf00e32 : TIMED_OUT in 2h 03m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/757f24be1fe043539611f13131a37ab2 : TIMED_OUT in 2h 04m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/db1f13d17fdc40c6810364a3a554e78b : FAILURE in 2h 31m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a57c54929a3436aa9c67877e3d6275a : SUCCESS in 1h 05m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7c11a813224f41bdbb85dc3122317f09 : SUCCESS in 2h 10m 23s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/6650099d2803479ba6d085121aca988e : SUCCESS in 1h 44m 43s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ea0d55c96f114a118cd9bce51294c088 : SUCCESS in 1h 13m 04s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f86a6b12674c44dc8e36d21d6532c595 : SUCCESS in 37m 11s","accounts_in_message":[],"_revision_number":2},{"id":"566cb6aba869452ea16d0f873a42fca77592ebc3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-29 06:22:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b6674773190c26ccc6c65034978792f8af37865f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 06:22:36.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/glance on branch master.  The\nproblem was:\n\n  Job glance-secure-rbac-protection-functional not defined\n\nThe problem appears in the the \"opendev.org/openstack/glance\" project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - integrated-gate-storage\n        - openstack-python3-jobs\n        - openstack-python3-jobs-arm64\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n  ...\n\n  in \"openstack/glance/.zuul.yaml@master\", line 361","accounts_in_message":[],"_revision_number":2},{"id":"ca2c2c27df51f4489f2552f97d6fe21f6f4cd8ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 06:22:44.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/glance on branch master.  The\nproblem was:\n\n  Job glance-secure-rbac-protection-functional not defined\n\nThe problem appears in the the \"opendev.org/openstack/glance\" project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - integrated-gate-storage\n        - openstack-python3-jobs\n        - openstack-python3-jobs-arm64\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n  ...\n\n  in \"openstack/glance/.zuul.yaml@master\", line 361","accounts_in_message":[],"_revision_number":2},{"id":"4be79ba1c09fb37c6e029201fd787196a8396229","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-06-05 11:03:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4924bd7f21cddf85905fa5764d1f4f44c6d65b7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 11:17:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a66b6506891949f99ed9d7cc4cb25fd0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/705df65527c847aba7d2babd97ff3188 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b2d2c61c319147548cf0f0cbf2242704 : SUCCESS in 12m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e34537b903925af402f9b16117aa22fe491a76e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 13:32:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a68256c5deb46df99e19d09417b6635\n\n- grenade https://zuul.opendev.org/t/openstack/build/8a30a75870c243e08f5330ea50764456 : SUCCESS in 57m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/688f14f8a070496db4bd699bd998466b : SUCCESS in 1h 41m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a14a36d4881c4582aca5fca4a4735f63 : SUCCESS in 1h 01m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3319d60cbee14509ab49b6c2e28e5e05 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b0b1c1772e74ec1aeb7f56ea9f89816 : SUCCESS in 6m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/224f2183b46048f2ad1e3e2553dcb50e : SUCCESS in 6m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9099290099b4d5184c4c4c943d9b022 : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ca69e904010458cb01b1a6ce64ddf31 : SUCCESS in 6m 36s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f2f912946d834f3191636cc3457db4d2 : RETRY_LIMIT in 1m 45s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6bfdf8c164a141589d06cea8b0dbd0db : SUCCESS in 8m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ce7d70c7fadc41d7a42e9e75f2e69cb9 : SUCCESS in 9m 05s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0a875dc989f44329a509e8e0d502d40c : SUCCESS in 14m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ec2578666e394c68be32bcfb894d6989 : SUCCESS in 8m 52s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bc2b92bbb07a46e5ba842280671a18c1 : TIMED_OUT in 2h 04m 02s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/81c1644cae6043508e38582e0e17c652 : SUCCESS in 1h 47m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ff57c72b38d74b9d8a506ebe9afb5c1a : TIMED_OUT in 2h 05m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b9ca4c3d04894623a11d6517c216f962 : TIMED_OUT in 2h 04m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0013f740365e4ae18820d59c01694897 : SUCCESS in 2h 22m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/913d218f38a34a74979d5dcb72b85c4b : SUCCESS in 48m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2db11ee4036d4fef829c59d5ea6f267d : SUCCESS in 1h 59m 50s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/d33beb788de14c66a4d12bb55d62323a : TIMED_OUT in 2h 05m 07s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1041a919abb04b91a34aabf66e9edb4c : SUCCESS in 1h 10m 49s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cf73e990e75e4bad91166c2ee4bea586 : SUCCESS in 30m 02s","accounts_in_message":[],"_revision_number":2},{"id":"9634ba53abd55410b85298ba421a0190edbd281e","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-12 04:29:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e3ac7d2c318dfb587575b2d8b0997e592108a064","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 04:35:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f231228207064ff28c6ecf3ca96b12de\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e358b78e753c472cab95a95e4b2366db : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d3b6d7cb9bd14bc58bdc082408515c9b : SUCCESS in 5m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"23609079c4c1abf80dec3358433d685a93d85b30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 06:59:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0039dfdc14145eaa72e774a04336935\n\n- grenade https://zuul.opendev.org/t/openstack/build/2861ccd978094f908d1986cc4f148e6a : SUCCESS in 42m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3957154b6f4d48489ddb02f50c75d817 : SUCCESS in 1h 14m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acd726940f974318b5126189b54f48a6 : SUCCESS in 48m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45f2516c68bd484dbd3bcfbb0630aee5 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6b8dcbcb0434909abdcde42a5b72167 : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa36415ecfbd4467a841cdd8185284d3 : SUCCESS in 5m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8b661de0b10146268c211c8e31d6ae34 : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ade9b30de9f413fb3ceb1714e6fae31 : SUCCESS in 5m 41s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7ec0c6e5ce144a489c3b3e59d1606f80 : RETRY_LIMIT in 1m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a3fc0130ef83486e92fea0c6bbbb7ac5 : SUCCESS in 8m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5223b5955a6c48b298e2cc9c2dadb787 : SUCCESS in 8m 35s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3263e9748ac4444f9cf91436846914b3 : SUCCESS in 12m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/dddd0b1eadd646c8957841ea04669296 : SUCCESS in 9m 07s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fc4d6bd484634fa19de78917a5a6dbd8 : TIMED_OUT in 2h 03m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4c933a977a3741fd961bfb125ce98f0e : SUCCESS in 1h 34m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0001064f280d47e5877c9ab9550647c6 : SUCCESS in 1h 38m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e1f8e0fede7f43d594edf347d6fd2da0 : FAILURE in 1h 50m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/421d4eb9d88143cfb2ddae2351e06f53 : SUCCESS in 2h 05m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/272a944ae3a042609896e08d8ea6d582 : SUCCESS in 39m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7a7bd1390e684c74b344f60aa7e80727 : SUCCESS in 2h 19m 13s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/2928ab680ac4425384e32aaf0c8d9c9c : TIMED_OUT in 2h 04m 24s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1b41a8ecc58b4a578ea1d2fc862828f6 : SUCCESS in 1h 06m 49s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4ac157b914d641b38f045308b0982dcb : SUCCESS in 38m 22s","accounts_in_message":[],"_revision_number":2},{"id":"de6c7ebe91ba7ba731688b1d00173d8ba43641b6","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-06-17 21:33:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5557d9892275c7d13f73bfb4e3319aa5c735a4a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 21:40:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1c3c4949f7045c5ba95b6137af72aee\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a67d1bfab9cc4a479851e924d663d905 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/177a74a91d054d4e8ed939db9d617358 : SUCCESS in 5m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b48dcd287b631d8489d37e6616be9fbb9db91b12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 23:58:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f95d1a62f88e42de97346c593b4e29ec\n\n- grenade https://zuul.opendev.org/t/openstack/build/f2be22465a58481994e417b1bd528424 : SUCCESS in 51m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/25076d2da0f94adf8c9604c476b6b43d : SUCCESS in 1h 17m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eac4f4b37cac42098fa5d615aeb7f2c1 : SUCCESS in 51m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19074b22f7484f859f0d889b0a11141d : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/543f572e292b41dbb532367c508c6079 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e8300e3cd3d04640aad40600c0ec40c6 : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7c7d7280640c4f01906569d7ac9f79d5 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/582a9c33224342459af5c26d0346d860 : SUCCESS in 5m 33s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3d056225d3b94cacbb01bf672678e3f6 : RETRY_LIMIT in 2m 32s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/853979c0b4aa4476a6cf1a670ed3f137 : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/35c5b9765f734b83abc2bdc08e8e2613 : SUCCESS in 9m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e242e80406344a0282c6a8297be241f1 : SUCCESS in 12m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ce1dd18e782447d7bdf87b8e419ddcc9 : SUCCESS in 11m 22s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7d2d4f26bb7740a09146dbafcb005b37 : TIMED_OUT in 2h 03m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/28160c9e6741407f9b137212e923f784 : SUCCESS in 1h 56m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/39c27b15e0124b59bdb01c7aa863c57c : SUCCESS in 1h 58m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/369e179286ef4d9096f6602f34929035 : SUCCESS in 1h 54m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fe5d92fc620b489da40ee67983115162 : SUCCESS in 2h 05m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7fe38c1c79647929e8752a4560a0183 : SUCCESS in 1h 03m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3659f44884b3416db730e1666185e892 : SUCCESS in 2h 18m 54s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3be18ec6ce20439988e68a6e33179d09 : TIMED_OUT in 2h 04m 11s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/de1654ffbacc452f9aae5f885d01cd9b : SUCCESS in 1h 12m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3e2bc3cbf3704966bbf9742c09e3f168 : SUCCESS in 37m 11s","accounts_in_message":[],"_revision_number":2},{"id":"8400238300b3d9dd8f2446443a3937efe85b7c08","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-06-18 23:09:43.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"696b20f3edd01547f7d5b538bf9021889fe45832","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-06-20 05:56:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a266f11d4462ac88b2b7f5ddbe820c78c5303430","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 05:57:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c4febf71979fce16cb9ebddafaefbe31f45f5b45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 08:15:08.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5092235290d644f789f20f8546347da6\n\n- grenade https://zuul.opendev.org/t/openstack/build/83a58745889549f4a971e6af20323c85 : SUCCESS in 1h 03m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/917d1ed0ab004d2eae4c32918c17ac63 : SUCCESS in 1h 41m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07a00d99501e439e9ad2dde44a7b789e : SUCCESS in 1h 02m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b20e437fe3564947b134ad5c12ac80e8 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb30d34982fc4a5f8f10b86b7115b2a8 : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee34d81cf496405388178b051e7bc695 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47f1d3375b464b10bf5d5ae80fcfc75d : SUCCESS in 5m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1897899daf8442ada9b653e4a2762cf6 : SUCCESS in 8m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cb7a7eaf10e44fde9a2dfd005ec33728 : SUCCESS in 8m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a5d7c48ab131493dae5b95ab35477df9 : SUCCESS in 14m 36s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9f89419c7f9a4b289811af21daff020e : SUCCESS in 1h 23m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/10dfca41b8ba4c71a997ea52d163b0ff : SUCCESS in 1h 48m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6c4dec54f3f0444dbc02ad72c2107da2 : SUCCESS in 1h 59m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d588db91b20b4ef6aebed07ce5435812 : SUCCESS in 42m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/337924ae8d66477fb76ac1bb84fd2b16 : SUCCESS in 2h 12m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c73c8624e9a34cb4abed0c372acee8b9 : POST_FAILURE in 46m 34s","accounts_in_message":[],"_revision_number":2},{"id":"c63ff3722255c99d2a3635ecd27421ba84fd02a3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-06-20 08:22:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4481a6c8d09e58a2a1d6d27f820e2d8505e7296","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 08:29:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10a142bd586f4e22b2a59a81d7d45959\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed6b717de74744b8993b33bb0d74b323 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/898031ceb9524539968a047b3fdb4569 : SUCCESS in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"30f91f11132d9b5656b7bf92b6dbc4100d84a8bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 10:55:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b41fb18371a5410abb0826264cf48ee3\n\n- grenade https://zuul.opendev.org/t/openstack/build/248af5e7e3594e7f8bcf9c7c147e3681 : SUCCESS in 1h 04m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4fbca7a6d9fe4d60adf8ecf47a85f569 : SUCCESS in 1h 46m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/842b5a67f4b346b4a18e63c7267a25aa : SUCCESS in 1h 03m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1be1650cba2e4b11b489788f93128e3d : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b7ed8567f1944b487ae3ad89304a7b1 : SUCCESS in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f999540f3e8443269e43e005524ebd85 : SUCCESS in 5m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97ffc76ef1034422be1e72f59a585f4f : SUCCESS in 9m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a44bb39e6fc94d67817b237dc4911825 : SUCCESS in 6m 53s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/814bbe9be0784ce7a42c3a04bf64a66d : RETRY_LIMIT in 1m 35s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0f91f75b288d4781bb9ff486d3e4abb5 : SUCCESS in 7m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f7f19b1fb2334814a822e6b96deaae59 : SUCCESS in 10m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b9cda85e2a664c5eb42f6f9fdd0761fa : SUCCESS in 15m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b3c5f4d062dd4d0a998a10c32da9a8c5 : SUCCESS in 7m 35s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e457968eb057429b872d914852084b40 : TIMED_OUT in 2h 04m 32s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/277a6085127349808c70ce0d3e1a7f8f : TIMED_OUT in 2h 03m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/20c452007d2d45aaba112eab4c742a12 : SUCCESS in 1h 51m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fa3b24b2666f4ec3be3995873db18285 : SUCCESS in 1h 57m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/09c6b984b7ee4d8a8a4f1d624422874b : SUCCESS in 2h 25m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afb0aa8c20134e6196ae7486ad886226 : SUCCESS in 1h 09m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/87e94f5918654eae92123a675b42f6e5 : SUCCESS in 2h 21m 23s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/c2639eb7191e4ed19165d9d037f54a17 : SUCCESS in 2h 02m 28s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a170583bb41a461599b6cd683b5f5d0a : SUCCESS in 1h 12m 56s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7df5dba2c00d4d97b4c9baa5b21579a7 : SUCCESS in 47m 57s","accounts_in_message":[],"_revision_number":2},{"id":"7848f1152100f42ba532132bdb974d223fe73b58","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-06-20 20:35:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"03ef94daac2a2834ed17e3eeb8145e72b6e6d83a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 20:43:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7763575117ba4571bdf1d57e9383e3c8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b1c449e8b694063a18c78b6aa887d8d : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1060db4b88cb4c78abff7d1271d0a1e5 : SUCCESS in 6m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1025cff600d1c94bd0e1778da83b2d0c17d6791f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 23:03:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/075ae2f2e8604ac3a6c6bfbc6dfd076a\n\n- grenade https://zuul.opendev.org/t/openstack/build/bb606ae49bbf4cb482c9c377d7f3bcca : SUCCESS in 1h 03m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2bb16d400b70467e8c5397c86f629ebf : SUCCESS in 1h 36m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d75837a7c55c4d5694c705cc8e723e38 : SUCCESS in 1h 07m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f80259720a6a4760a2886e9237e39c63 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2cda467cfa140ffb5c4bbf7507b6b02 : SUCCESS in 6m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/595c4ecd01404feab68ce4f8fc592c4c : SUCCESS in 5m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11c9b0f0782f4270929a36b497cafb7a : SUCCESS in 10m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5c17301d7624c66938b3a14e9a61473 : SUCCESS in 5m 37s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/53863d1a9f684c598bcf9ecbb4aafc0c : RETRY_LIMIT in 1m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c889b2c739a84f45920ca63b8e9beb61 : SUCCESS in 8m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1f3f646498aa4942bb0923e7934dda73 : SUCCESS in 8m 46s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/01d6c96fe02646219b04fa2680e2b19f : SUCCESS in 12m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b2f080c3b67649f595d399d310a70409 : SUCCESS in 8m 08s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f079d176ae424ce5baf042d6a07391ff : SUCCESS in 2h 03m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f43b96439060453c9115cf84062caf50 : SUCCESS in 1h 59m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a040d56ad1e5474cad2f6e2a2ea0e8ab : SUCCESS in 1h 55m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e0d2886e5fcb4e32bbd8807737ddaa54 : SUCCESS in 2h 02m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/487f9b6342ab4978b1ffc7147ca0da5c : SUCCESS in 2h 07m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afc932a910a24ad5b67210a74ffb1c7c : SUCCESS in 1h 02m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ff73adc2eef34222b1fc078d49f78009 : SUCCESS in 1h 45m 51s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/6bf137303111483c80b13ff705c39525 : SUCCESS in 1h 57m 56s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a1f9c84af68d4ebebbc4a8d09bcbb8c2 : SUCCESS in 1h 08m 24s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3e8dc56623024ce0b80f7d18e4ea0589 : SUCCESS in 35m 55s","accounts_in_message":[],"_revision_number":2},{"id":"7efef8067f9acc58af34e4ce55f3b8316c9b6912","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 23:04:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"094bbe76cc9b63a32f27212dfaae205cdd16d88c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 01:21:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40dc2237c0ec4304b862d081deca6382\n\n- grenade https://zuul.opendev.org/t/openstack/build/d6d67542c3144eeb9a5c445bc29eaba6 : SUCCESS in 47m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/83ddb9a7528a402d947da2240211d202 : SUCCESS in 1h 27m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/631cea53898d4013bfa271beb83bfc8b : SUCCESS in 58m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/023aec84e18d4d608a3bb0b5bdee43be : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23e0287430204a77851dbe0bf64f1065 : SUCCESS in 9m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ec9fdc75172419992bc3ea7c87f123f : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b34817f01a0e47a1bb32d04e72f1c38a : SUCCESS in 8m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6aaa25b10d6c403f934451469b837fad : SUCCESS in 7m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/abbe3dada04b4529840b3a55b5b6be16 : SUCCESS in 9m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e267602b60e74aeb86a7d90b8b592b1b : SUCCESS in 10m 32s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/961b4403401946bfba600919d79e22ec : SUCCESS in 1h 49m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d6918e23e5b24e52beb55fb13ad6321f : SUCCESS in 1h 38m 24s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7c1392aea93f4030a9d433dbde9bd30b : SUCCESS in 1h 39m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ca07f2ef551422983fb864e7c9b4fe5 : SUCCESS in 59m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ba23c110c29f422f9761ca373dd43523 : SUCCESS in 2h 12m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/df0da646bc7a4bdbb4bbf0518b17b8cd : SUCCESS in 24m 55s","accounts_in_message":[],"_revision_number":2},{"id":"159abb6a89060fc83ab026a383b2eeac0b71ccfe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 01:21:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cef81760b6e697ea5a763efe26a182adc6d07cd1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 01:25:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e6481c0b80f4b2880e4161529981f2c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dfb0110ce75486498a7056bf7e7072a : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"95abad115b2e1e1227f8060580853a60bd768f0f","revisions":{"5caf9fd0b8bf19ad3d9e29f69319d0930821dcfe":{"kind":"REWORK","_number":1,"created":"2024-05-28 15:05:52.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/79/920679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/920679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/920679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/920679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/920679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/920679/1"}}},"commit":{"parents":[{"commit":"3f49730e627206c464163a3e1f00580b09cc5ebf","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f49730e627206c464163a3e1f00580b09cc5ebf"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-05-28 15:03:51.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-05-28 15:03:51.000000000","tz":120},"subject":"Make functional-py312 voting and add it to the gate","message":"Make functional-py312 voting and add it to the gate\n\nChange-Id: If9238dc4243b1bd23ee0a61a1531ff784ad19655\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5caf9fd0b8bf19ad3d9e29f69319d0930821dcfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5caf9fd0b8bf19ad3d9e29f69319d0930821dcfe"}]},"branch":"refs/heads/master"},"95abad115b2e1e1227f8060580853a60bd768f0f":{"kind":"REWORK","_number":2,"created":"2024-05-28 15:10:06.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/79/920679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/920679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/920679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/920679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/920679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/920679/2"}}},"commit":{"parents":[{"commit":"3f49730e627206c464163a3e1f00580b09cc5ebf","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f49730e627206c464163a3e1f00580b09cc5ebf"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-05-28 15:03:51.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-05-28 15:09:49.000000000","tz":120},"subject":"Make functional-py312 voting and add it to the gate","message":"Make functional-py312 voting and add it to the gate\n\nChange-Id: If9238dc4243b1bd23ee0a61a1531ff784ad19655\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/95abad115b2e1e1227f8060580853a60bd768f0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/95abad115b2e1e1227f8060580853a60bd768f0f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
