)]}'
{"id":"openstack%2Fglance~921300","triplet_id":"openstack%2Fglance~stable%2F2023.1~I7d63951ff080dc699b8d11babc0a5998d90621e4","project":"openstack/glance","branch":"stable/2023.1","topic":"backport-stream-vmdk-origin/stable/2023.1","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-06-05 18:56:56.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-06-05 21:03:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7d63951ff080dc699b8d11babc0a5998d90621e4","subject":"Support Stream Optimized VMDKs","status":"MERGED","created":"2024-06-04 15:43:05.000000000","updated":"2024-06-05 21:05:48.000000000","submitted":"2024-06-05 21:03:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921300-backport-stream-vmdk-origin/stable/2023.1","meta_rev_id":"18d15b7d6dc6c7eeaa0e23e3f993a462e5b1497b","_number":921300,"virtual_id_number":921300,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-05 21:03:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-06-05 18:56:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-06-04 19:18:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-06-05 18:56:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-04 15:43:05.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"state":"CC"},{"updated":"2024-06-04 15:48:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-04 16:12:31.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-06-04 16:13:48.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-06-04 17:55:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc8a8260c0d2af6518289400c61fe1e86adaa1e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-06-04 15:43:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2efb72f5cd95fe6df149a63efab112b4d72433b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 15:48:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b172b40f5ea6494a825fbc42a0c6711b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed5d4a4c030e4bfaa55743f0df2d6d8c : SUCCESS in 4m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc10e481df13b37d9002204f1cfad922c4658bb6","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-06-04 16:12:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8cf052e0f6a8448342a33cad3f0dd8e507758377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 17:55:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50ef60ca87894a6a8a4ce4a255dfe806\n\n- grenade https://zuul.opendev.org/t/openstack/build/6f50e54c744b4ef288bc7cc90734f24c : SUCCESS in 59m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6b0d090d81464ccfb7fd70ad1fd357d5 : SUCCESS in 51m 15s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bebcb19f39d44cd6838545f827f039f6 : SUCCESS in 1h 30m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb6b70607e7147c5ac60eb9ab5520933 : SUCCESS in 54m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d9e1920a9134ca9b419777e662eea43 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb01463f316b4781bb7fa55c1b8565a1 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54ac56b9f07c4cfb8445353f3249f96c : SUCCESS in 6m 01s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/998d462e08c346f0b86a7dbfc004114b : RETRY_LIMIT in 1m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a7de72446c5148bda9c43d9a9a91e4cd : SUCCESS in 8m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0d0f5033b3a84a32a416680efea5502b : SUCCESS in 8m 32s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ae7742e827164b8ba01ddc91292b1541 : SUCCESS in 1h 26m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/13644e61e4cf45079c144f869c0771cf : SUCCESS in 1h 14m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bc030edd2d2644f18e2c8758e9fa4204 : SUCCESS in 1h 22m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/eb3c5a38048e47c38a91d91e2d3cc013 : SUCCESS in 1h 22m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6b289a2332264be2b9b6ff99ee518838 : SUCCESS in 1h 46m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1c67e74e9cd419daa973c803584ab98 : SUCCESS in 57m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6b7a195cd0d6423083bc2da063253e47 : SUCCESS in 2h 05m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/69a75b5f6adf42a8a09bb979d0670098 : SUCCESS in 39m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e7f582840f860c551bb22711459697f28eddd35a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-06-04 19:18:49.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"237d128342cf0577510ca2a878eddc8702c4647c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 19:29:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9c2d98bc1d8433c942888f8d99fee86\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c84edc0351c8463892ab179da1c6879e : SUCCESS in 8m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a5f609d1666050215ca4a41ed3f8bd1b10c9f5cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 21:39:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30e778bec9a54fe8b5882ca93d12acac\n\n- grenade https://zuul.opendev.org/t/openstack/build/471e4c498d04439c9bd61d95d6fdb69b : SUCCESS in 50m 05s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cad097cd7346435a8ab8d204cbb59b13 : SUCCESS in 49m 42s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6dad59f64fa145a6b4a08aae9910647e : SUCCESS in 1h 03m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e47065d66804379b2613253ae40b20d : SUCCESS in 58m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8a070ba4b4947518c720ddf126514b7 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee9d4cdf331e4b6cb2bc86f64b5a2bf2 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7fff138c4aa45cda86fd1f47cdffbb1 : SUCCESS in 7m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a10b86e206da44378d1209db59400c72 : SUCCESS in 8m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/91b92a9adb9e4224b690dfb0597b0546 : SUCCESS in 8m 44s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/68f43fb332524aaca395649c497caabd : SUCCESS in 1h 52m 00s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/60a0875b75a242ca96fd1aba9b6a49c8 : SUCCESS in 1h 14m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/90670cb6383a428ab5bac0d4a59e20f6 : SUCCESS in 1h 33m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c657acdec3334802b2e2f470389697f6 : SUCCESS in 1h 21m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8d9f36351e6a49268946eda09350271c : SUCCESS in 1h 38m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01f12534e81346dc9e16b9f7236ae3b7 : SUCCESS in 41m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5b329e7649b44f5caeff3792806d0f8e : SUCCESS in 1h 53m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5432a8c394ee4d82b40a0b0fa0a5852c : SUCCESS in 26m 10s","accounts_in_message":[],"_revision_number":2},{"id":"e17726cf60de409a322a71e112207113859fdffe","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-06-05 18:56:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e12ac6c77218e1ff76b1ddf663b6131c4cac6eb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 18:57:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4ee5470dce0bbcb2bc9cf213883692da000870aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 21:03:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8aa57ae21ee431ea541fba46b9fec87\n\n- grenade https://zuul.opendev.org/t/openstack/build/0185385c209d4727bb0b08fbb8d12f7d : SUCCESS in 58m 27s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e75d5d7dca4e4c5dbac267dab8315ecd : SUCCESS in 1h 01m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/571839cacd0541dc82c4389682cb3945 : SUCCESS in 1h 22m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0794f5a3c9c248f784cff73ad6492bb9 : SUCCESS in 58m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4336b64d9b43498e9026381c722eb15e : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4115724407ca44e5b43ebdb0091f4433 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b12d251a34a4cd68ffdf0c07bbd333d : SUCCESS in 7m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e2b7b23ffbf5417e8717a9e0b843a99f : SUCCESS in 8m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/82630a411413495d911a4facb84ff84b : SUCCESS in 8m 39s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a3af7d42d03d4be38236fb3fd7813998 : SUCCESS in 1h 23m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b54eac6a29254cc49b63608f1e3e11e0 : SUCCESS in 1h 42m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0205039d51b945b88931e2306955dc0e : SUCCESS in 1h 17m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6da773c4d9124f0aab932f676d47c848 : SUCCESS in 53m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d04d8f28893e424889fa1c138eb741a0 : SUCCESS in 2h 01m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4cc9beb95ba642aabcd68e9507ad67d8 : SUCCESS in 28m 15s","accounts_in_message":[],"_revision_number":2},{"id":"4cc725f290985b59947790cd06905f82883e550c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 21:03:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"18d15b7d6dc6c7eeaa0e23e3f993a462e5b1497b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 21:05:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f7d48bd03f7412897aeaf037597700d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17e0ffdb075b4c409fdd599428557286 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ef22a7425a5fe4f7006f9434a33dd092e2594830","revisions":{"0b08c530fd9b0b00c0bcbb3ad5f61d6a1148b30f":{"kind":"REWORK","_number":1,"created":"2024-06-04 15:43:05.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/00/921300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/00/921300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/921300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/921300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/921300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/00/921300/1"}}},"commit":{"parents":[{"commit":"70170d2a99f9d30bddcc55a9183133a16ed14c06","subject":"Fix broken glance-cache-prefetcher utility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/70170d2a99f9d30bddcc55a9183133a16ed14c06"}]}],"author":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2024-02-02 10:48:24.000000000","tz":60},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-06-04 15:42:06.000000000","tz":-420},"subject":"Support Stream Optimized VMDKs","message":"Support Stream Optimized VMDKs\n\nStream optimized VMDKs are also monolithic disks images, and start\nwith the same sparse extend header as normal monolithic sparse files,\nso we can parse the virtual disk size in the same manner.\n\nSee \"VMware Virtual Disks Virtual Disk Format 1.1\" p. 17.\n\u003e Header and Footer\n\u003e The header and the footer are both described by the same SparseExtentHeader\n\u003e structure shown in Hosted Sparse Extent Header on page 8.\n\nCloses-Bug: #2052291\nChange-Id: I7d63951ff080dc699b8d11babc0a5998d90621e4\nCo-Authored-By: Rajiv Mucheli \u003crajiv.mucheli@gmail.com\u003e\n(cherry picked from commit 5e7e6bfb80ed6ede57f348e688c96dbdfe0e1c2b)\n(cherry picked from commit 9d45e8d4b87e992be23974a831811bff563ce721)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0b08c530fd9b0b00c0bcbb3ad5f61d6a1148b30f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0b08c530fd9b0b00c0bcbb3ad5f61d6a1148b30f"}]},"branch":"refs/heads/stable/2023.1"},"ef22a7425a5fe4f7006f9434a33dd092e2594830":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-06-04 19:18:49.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/00/921300/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/00/921300/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/921300/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/921300/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/921300/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/00/921300/2"}}},"commit":{"parents":[{"commit":"a45667b7140ec49a74bd14ba786624646967bb91","subject":"Remove py38-fips jobs because stream8 is dead","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a45667b7140ec49a74bd14ba786624646967bb91"}]}],"author":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2024-02-02 10:48:24.000000000","tz":60},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-06-04 19:18:43.000000000","tz":-420},"subject":"Support Stream Optimized VMDKs","message":"Support Stream Optimized VMDKs\n\nStream optimized VMDKs are also monolithic disks images, and start\nwith the same sparse extend header as normal monolithic sparse files,\nso we can parse the virtual disk size in the same manner.\n\nSee \"VMware Virtual Disks Virtual Disk Format 1.1\" p. 17.\n\u003e Header and Footer\n\u003e The header and the footer are both described by the same SparseExtentHeader\n\u003e structure shown in Hosted Sparse Extent Header on page 8.\n\nCloses-Bug: #2052291\nChange-Id: I7d63951ff080dc699b8d11babc0a5998d90621e4\nCo-Authored-By: Rajiv Mucheli \u003crajiv.mucheli@gmail.com\u003e\n(cherry picked from commit 5e7e6bfb80ed6ede57f348e688c96dbdfe0e1c2b)\n(cherry picked from commit 9d45e8d4b87e992be23974a831811bff563ce721)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ef22a7425a5fe4f7006f9434a33dd092e2594830"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ef22a7425a5fe4f7006f9434a33dd092e2594830"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
