)]}'
{"id":"openstack%2Fglance~921327","triplet_id":"openstack%2Fglance~stable%2F2023.2~I7b86f9036914dec59db01347ac8a2163e6e9843f","project":"openstack/glance","branch":"stable/2023.2","topic":"backport-stream-vmdk-origin/stable/2023.2","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-06-05 17:25:09.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-06-05 12:09:05.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-06-05 07:05:21.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-06-05 13:39:39.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-06-05 17:25:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7b86f9036914dec59db01347ac8a2163e6e9843f","subject":"Remove py38-fips jobs because stream8 is dead","status":"MERGED","created":"2024-06-04 19:17:51.000000000","updated":"2024-06-05 17:28:39.000000000","submitted":"2024-06-05 17:25:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921327-backport-stream-vmdk-origin/stable/2023.2","meta_rev_id":"731da98daf634de0a794a94ed76b61c5ba5cfbc1","_number":921327,"virtual_id_number":921327,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-05 17:25:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-06-05 07:05:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-06-05 07:05:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-04 19:28:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-04 21:05:07.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2024-06-04 21:05:07.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-06-04 21:05:07.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-06-04 21:05:07.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-06-04 21:21:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d406e1d3bfc3bbf0d7eb25fa4f565af3cc722581","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-06-04 19:17:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8baf1e1b62cea982f7802cc97038661987ee2c93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 19:28:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4908049ed5874c178149841244265fed\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c84d3ef0cb744ffb53883d285a7ce6b : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/772564c487bc43acaf45a9071a825247 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4a6ecedea9b9437eb267dfa6e251ff4f : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/810a435525b44f0f9bc672800ecd87a1 : SUCCESS in 8m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0396d624bea423f5e518a36af0a93c7743ac6cc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 21:21:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ed5f2e4721a4f30b71248cce6482a31\n\n- grenade https://zuul.opendev.org/t/openstack/build/ea49059ced1a4e67bf3618348051c9a3 : SUCCESS in 1h 05m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c8a81e79a189400ba16727aeceb774f3 : SUCCESS in 1h 09m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ba2b5896374490684f4b47f3d2ccebe : SUCCESS in 58m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1216986c2cde413ab98baee4f6499276 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16fa4949b95545de9aeea91afeb8b567 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ab16b8a499349eda0a746f6b1e7f726 : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e40d6d124614ee8bcd4a28f727e7b2a : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea6e7eff20a34385af90ce2a25f5ae33 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c2e69de7dcc4b7eb464b48ab8fd284e : SUCCESS in 6m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f09e73f840f144d2865bdfa20e291059 : SUCCESS in 8m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ac68b4268ca6412381a59fbd769c6694 : SUCCESS in 9m 33s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/83618258faea4f6eba5bdac0101431a1 : SUCCESS in 8m 14s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/99555677d95e48f7ba49e67ff4f904e6 : SUCCESS in 1h 40m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/12da285924864cdfab603059e171a5c4 : SUCCESS in 1h 51m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1afcf73de62d4804ad0c21cfa02a4502 : SUCCESS in 1h 47m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b5ec623d8f1b480d99ee274d015ba412 : SUCCESS in 1h 24m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e8b3c645561d4fd78d68097a4edb4343 : SUCCESS in 1h 47m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d611b35efa2426aa67817b9f246fd87 : SUCCESS in 48m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dd839a2a89d542348a872658ae4ad869 : SUCCESS in 1h 45m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ebb20b450362425b97cff10b62137d17 : SUCCESS in 38m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b5bdf42ea6734e26a37ca7a9b38f42f1af89ceba","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-06-05 07:05:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"26391bdc6ee5a3abd29cf72000c6b9fb09d6d197","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 07:06:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5490b70dfca5d1dd1059406a45de646a2ac44c07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 08:53:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2262e468885843d79706f729480f9a86\n\n- grenade https://zuul.opendev.org/t/openstack/build/7bff8d700c7b41d7afeee778df90ed69 : SUCCESS in 54m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88ff85caff3147aebc2f2f44d39e42b1 : FAILURE in 1h 08m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39ce00d1e7b7451caab963ad81316671 : SUCCESS in 58m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b61f9ac338640c8a4c9004b3583dffc : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df5c78ee84da4f5d8f6ebc86d28375f5 : SUCCESS in 7m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/680e5cd50afb451db9f3132b7952bdd5 : SUCCESS in 6m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b474ea5d59874b9b8f5837a0b1b5b5d7 : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b73ed1eb8af4e51ae03814d627d1b34 : SUCCESS in 6m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e07055f9a2a8418a947e52d965a79a7b : SUCCESS in 8m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/86300f11d3434d32b47c2176e619ce86 : SUCCESS in 7m 15s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2b35e01461664ac9b14eaa0dae686b2c : SUCCESS in 1h 39m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b155c104fb30437896461674f0f4e5fd : SUCCESS in 1h 31m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/95f86383fc27410ca125326b03a4c874 : SUCCESS in 1h 25m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32ff477d8e214912bc0b8f23e2467756 : SUCCESS in 59m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ecda5b5c512d441ebc8fc29d5ecd50d2 : SUCCESS in 1h 42m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/69493ea7f0d5414aa8c80087fa9204a4 : SUCCESS in 36m 58s","accounts_in_message":[],"_revision_number":1},{"id":"7a69ffd15da0d1c9044d987632d2bc648941d1d9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-06-05 09:35:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c6bf7a0fe3daccffb927b037db5827092f18391","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 09:43:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efd339cca7f940f0be271f3323a71b3d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c4244932619403f961da2695b854981 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f4395233c69433f93e078a623f98f3c : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c3e6d052d3ef4e899b7e83f844f0ace5 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/63bccf519ffb4f4bb6e5f20eefb3118a : SUCCESS in 6m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"791c34bc16e34cce27d07afed7f00d64dbdb7775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 11:44:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/acf0ffc789194b3db1d72da78e18ebf0\n\n- grenade https://zuul.opendev.org/t/openstack/build/3a0efc389ce84c798ee520009953f2cb : SUCCESS in 1h 01m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4d7e60f40fe24bf68c07d44ed280666d : FAILURE in 1h 28m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f3d93dcd70f46f9901836d31d6ef9e0 : SUCCESS in 1h 00m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c293dec325db43b08186060d967a8f80 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/029e6ec10337407598b140c09c529a1b : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d1ab42241634299bc35c9797e380279 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a9cfbafb19c4b07884919984792ba84 : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9777bc3e0b6f4fafad4740dccfd0c3fd : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45783beeb01645aa9550877353f19d5f : SUCCESS in 6m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f3651dc666e24ce193809b16c40e3419 : SUCCESS in 8m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/209d0c49dd3344589e7c2a8cfa342414 : SUCCESS in 7m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f131abd28eb84a3b8ebfe214d848d1d1 : SUCCESS in 10m 01s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f0b3f5ee88894a329f36287af9f66df8 : SUCCESS in 1h 41m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bb7371df94e44d03be57818fa41af9f9 : SUCCESS in 1h 40m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e2f3e513137244988fa1b1b9ae6b2ed7 : SUCCESS in 1h 36m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f14d1778682c497e97466c51cc54c429 : SUCCESS in 1h 46m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c6cc8f1076454b8b82773c455fc75c2c : SUCCESS in 2h 02m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65bde9de579a4c399ea54dbfe0985b38 : SUCCESS in 57m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6c20c7ca3b8f47489e7809cacafbe2f4 : SUCCESS in 1h 48m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ee2d51e76b27412a80a4c678a113c9cb : SUCCESS in 35m 37s","accounts_in_message":[],"_revision_number":1},{"id":"792b1f299e036dadcb0c4b63d3fdf466ede1032c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-06-05 12:09:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"476149e6a9fd6a81666f0d292daccba78d24c2f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 12:16:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ede44b2dafaa4d769ae52f780e0f504f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f793a9260b494f0b910e52585bf26856 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5bde8f2d49844bae98d023206e143d1a : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6e96f39602ad41bd90ec3a5e1b342d65 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/569bc0864907438686b1bc4a575d005b : SUCCESS in 6m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aa742a672712ba2ee7581e00ed53174e886ca8c3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-06-05 13:39:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c388052717b8422fa85ccb316177e4d3c2d745bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 13:51:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/301bdf1a063945f983adbad863e56f6f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c05dd39781b2431db5d385e80cc02689 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46665359706e42e9a4ea3dc6a741ed24 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/23ead13498b24aedb35a4879cff10f88 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d68246dd44f24ff6ace2faad5223c898 : SUCCESS in 9m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2456e7375d162afe08fc895986048a6acb7c0ae0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 14:01:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06c240408798413d94dd7f697b119d0e\n\n- grenade https://zuul.opendev.org/t/openstack/build/c4928fc7af904144a80b4069a0cf96d0 : SUCCESS in 49m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e82b4be3bcb14cef9e889723152009e1 : SUCCESS in 1h 21m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c73f7104cd4543ca856f2ac171f3a3aa : SUCCESS in 55m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/886a328ec24249a68819f6f3811b6f49 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/600d4148367b4a748b0acba0e24f5710 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93c4feb3fa754fe4817c4d547eb321ca : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f50601ed3ce94557963e2715bc1aad22 : SUCCESS in 6m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47b20fb0eaeb4e438a024f06a1979568 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/451ca74c070d48f2986104de865b6518 : SUCCESS in 6m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c47b45c86046481ca3e6ea3efd283a36 : SUCCESS in 7m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/714f7a72af644bd4a44e695c6442205b : SUCCESS in 8m 41s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/82e7a030731d45ffa1ba9bb7805de812 : SUCCESS in 8m 26s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c0e534badc1d44fe9c4969941def1aed : SUCCESS in 1h 40m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/77aaa28b0d204cf1a9164272a73a6464 : SUCCESS in 1h 30m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/19268d5430734b0a9ea8d1f40ae5b204 : SUCCESS in 1h 41m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/df623484c9114027ae9f738a5b2c7a3c : SUCCESS in 1h 39m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bd06089971d049bcbf246504edd1810e : SUCCESS in 1h 46m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f6512dbc1a3405392203aef0f2fe4fd : SUCCESS in 1h 12m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/19f16079b9324059b8f59890d0a30f8a : SUCCESS in 1h 39m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ab53180efec1442c914c98d3fea8ad86 : SUCCESS in 31m 26s","accounts_in_message":[],"_revision_number":1},{"id":"92a6af2eaba5f99239c8f8674a5646f04d1e4359","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 14:02:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6dae49752cff6d55a4af3aebcca48c6741c3570","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 17:25:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d45e0d1fdb9340a2a4a8fc17431d2dbe\n\n- grenade https://zuul.opendev.org/t/openstack/build/7a5b9c11d2ab4863ba51378ece5c6ed4 : SUCCESS in 59m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b276b98eafe54ba199720847b4234312 : SUCCESS in 1h 39m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81d1b55cc015454ba9c2aa0b04e931c7 : SUCCESS in 56m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bea687512004eb4bba46877791caad6 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5fc344966e3456e9ab7f0fc8ef71e14 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74048ebf692f4429bd5576ef67de0984 : SUCCESS in 6m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dfdfe35ba93f4500ad080387d36e94a2 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e75556a9a0f4ed2871f9cd6102eb3fd : SUCCESS in 5m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2cbe2a4a0b154319ba688c63c68340a4 : SUCCESS in 10m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/deb371c8c5e64ea29bf3aa7910dab05c : SUCCESS in 8m 06s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/49d58390750c438a9e82755b2bc54a1f : SUCCESS in 1h 26m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3a3e529d73f34f1787c98db97f3ed549 : SUCCESS in 1h 51m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5cf31c2acbfe4f72b8b1d9435e700486 : SUCCESS in 1h 50m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3566674d04bc4ab39a24d76b38e30e79 : SUCCESS in 47m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/437a9448434f413680a64a686187bcb2 : SUCCESS in 1h 40m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f292c933413f4ac98547eeeac4fb7827 : SUCCESS in 38m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e0f91b5ffacd8a5f40e2a4f3c630ca59c8010df4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 17:25:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"731da98daf634de0a794a94ed76b61c5ba5cfbc1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 17:28:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/616679d7022a43f28b00b0b62706ae93\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48e9df7d02f144c785ad8fa1dc727d16 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56da508eaf9dc23deb4a9adcc564162943f1954b","revisions":{"56da508eaf9dc23deb4a9adcc564162943f1954b":{"kind":"REWORK","_number":1,"created":"2024-06-04 19:17:51.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/921327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/921327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/921327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/921327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/921327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/921327/1"}}},"commit":{"parents":[{"commit":"cec35e92bb36f2bfcdaf2cc1f05d7b5a0f3a07cc","subject":"Fix broken glance-cache-prefetcher utility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cec35e92bb36f2bfcdaf2cc1f05d7b5a0f3a07cc"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-06-04 19:16:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-06-04 19:17:45.000000000","tz":-420},"subject":"Remove py38-fips jobs because stream8 is dead","message":"Remove py38-fips jobs because stream8 is dead\n\nThe centos project has yanked their entire mirror of stream8:\n\nhttps://mirror.dfw.rax.opendev.org/centos/8-stream/readme\n\nAs such, these jobs can\u0027t run at all and thus will block any patches\nfrom landing.\n\nChange-Id: I7b86f9036914dec59db01347ac8a2163e6e9843f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56da508eaf9dc23deb4a9adcc564162943f1954b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56da508eaf9dc23deb4a9adcc564162943f1954b"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
