)]}'
{"id":"openstack%2Fglance~922114","triplet_id":"openstack%2Fglance~master~I0cfb06895d8eaf88fd2b623d09993260c27ed471","project":"openstack/glance","branch":"master","topic":"bp/image-import-plugins-api","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-05-21 13:53:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0cfb06895d8eaf88fd2b623d09993260c27ed471","subject":"Add image plugins status","status":"ABANDONED","created":"2024-06-17 14:18:29.000000000","updated":"2026-05-21 13:53:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c2f740cba9a7f9ba41c7accb0e349f7f95158a81","_number":922114,"virtual_id_number":922114,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-08-21 14:27:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-17 14:28:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-17 16:36:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9dc11bbc57713fcbe37c6fee92b4fd649d844476","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-06-17 14:18:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70692e406e796886b0519c06692fe79bd8e2de69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 14:28:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e89c958862924c46811d48e63d33502e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91d53ea1f69f41b9aaf88ab1d578de87 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/accb7030ae784977ab33be2776fbb9c5 : SUCCESS in 7m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e263a26631e49625b81702c038bbb31430c72f40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 16:36:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e682743e143940febf3ffa363cd23f05\n\n- grenade https://zuul.opendev.org/t/openstack/build/c6a2f3b9e60b4c608f8306f90d6b0321 : SUCCESS in 56m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6b353dc6f80842a5b89638a5b3e6bce7 : SUCCESS in 1h 36m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ceee4ab39894b0db73ce9c5e1d8588a : SUCCESS in 1h 20m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1017c60672cc40109ae3cd83bb57741b : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3948ab5b314745f88cd94d26d6252dd7 : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/14dc9fa7832143f29b4f13e63a92373d : SUCCESS in 5m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a1d15f32539943738c502efe26c13a82 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41598ae31f084ae08650088c6add612d : SUCCESS in 6m 24s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/fa3cc54b06df4a9b94da005f9cdcfa06 : RETRY_LIMIT in 2m 36s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/346d576491cc4e7581444da59861927d : SUCCESS in 9m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/31bd4f67a21340df85b2401c51a72ac6 : SUCCESS in 8m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2d54741109c949be9ef6e7adb6ecf096 : SUCCESS in 12m 13s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/87ade0ad086f4791bcffc52af075b1f0 : SUCCESS in 9m 02s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b127131397ab406883b02d485024c00c : SUCCESS in 1h 54m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/13b7d9b925ae416faf169ab1a8b464d2 : SUCCESS in 1h 51m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/62301d92d0364e879f981a47c26bcd5f : SUCCESS in 1h 50m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8c6dd039b98d4cf38854879c67c2d64b : SUCCESS in 1h 54m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a85945f8fa0f419b91d9642d4b92583a : SUCCESS in 1h 51m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ae85385c5564a0a9d799be208d39867 : SUCCESS in 1h 04m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5ce462d246f44a12879d7a1685d0e760 : SUCCESS in 1h 48m 51s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3c7399358b0247dcbbc93c7ff5321f92 : SUCCESS in 1h 54m 33s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f7dcb863dfa14e8da496d1425a4da8f2 : SUCCESS in 54m 37s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/74b59b8f1c5e4e6daece485d4f15b70c : SUCCESS in 38m 35s","accounts_in_message":[],"_revision_number":1},{"id":"fec4f496ba7b914acdd3e9dc28e3fb859ea7e268","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-06-18 10:23:55.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d01560b4cbb72668eb6ba0a5a27854307c4af338","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 10:34:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63f0fe30e8e24fc0adb71d7746a2f9c3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d9436b18b1148ec8816e606bd17bed3 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a42a25965fbf4eff86f3dd8dbd793cda : SUCCESS in 4m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9597110c7f5fcb898502d138d55d9640aafbcb48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 12:55:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8de70d12db51452ea477e414f4c932c5\n\n- grenade https://zuul.opendev.org/t/openstack/build/0c6f77104926415184cabc0e2fcc9c96 : SUCCESS in 42m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9b595ba1202e468bbb9452927498ed83 : SUCCESS in 1h 54m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbfb0c9bca9e43ffa6e1fb0cffc59abb : SUCCESS in 1h 10m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dd812107eea4f1283a07c9c31a38350 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d2ebb7550c4410f85022fed00016a93 : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8a75a77338140ce90d8e2decd499a5e : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e35879a825c42ccb8e0981b16a5f4af : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b97b912c7e4474c815f17ac1ada486b : SUCCESS in 6m 38s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8ac46d06ba964cb6a0a3ab706a74535f : RETRY_LIMIT in 1m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/03e6c6433ffd4819958a1dc7a13af4ad : SUCCESS in 9m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b7bb5e7dae444439a6941e7d61f96e99 : SUCCESS in 8m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bda7eb5e7dff47a1a368ce3d0b8eee85 : SUCCESS in 12m 01s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/291683ce542a424498d47f8571989053 : SUCCESS in 8m 35s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ab0658c907914ebcb20d2d12628d855a : SUCCESS in 1h 52m 58s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/101ae4fe5a2d4099b506c53dd0a32c70 : SUCCESS in 1h 56m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/35d2b9a2cd6e4480ba724fc5308a8552 : SUCCESS in 1h 34m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8da38a12ff834a018af01ec1a7b0b2f6 : SUCCESS in 1h 23m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ad1254dba5164f89b941289984e08110 : SUCCESS in 2h 07m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c71d966b4694ab3bb6616c4e0379c89 : SUCCESS in 1h 02m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7ebd4b78d9aa4ac0b1b2be0cf7b4d3fd : SUCCESS in 2h 09m 20s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/98d2a1543c314312ba91f4cc7fe8d0c7 : SUCCESS in 2h 01m 56s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9a01e9c4b7ff4b8cba8a86e57caefa83 : SUCCESS in 1h 03m 28s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f914fb1cb1944198a1bfc46569e8aa20 : SUCCESS in 26m 59s","accounts_in_message":[],"_revision_number":2},{"id":"6c80b604ceaef34562d9e1c6c84e7c58ea954175","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-08-21 11:29:07.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"95e33c2bc91f1841f465e6c40a8de36c46c6c1b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 14:09:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae6d23ecdd6741da964ba4d7f43b4634\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fed70397973c436ebcde5db1180e1a37 : SUCCESS in 10m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b12f9d1c5c644967ac57aeac2c5dfd00 : SUCCESS in 9m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4b1c3e21f983a5c215323cd6fc0d7282070b50c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 14:27:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9427c17cbd854980900b3fca94a10550\n\n- grenade https://zuul.opendev.org/t/openstack/build/2f2a4d29923f492c89973ea4da1b735e : SUCCESS in 1h 08m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7f7f3d8150f948628334c82d1bc0e94a : SUCCESS in 1h 22m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63ce64c0df78497dacd7e84bf9b8c911 : SUCCESS in 1h 02m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ab44171de7b460caf778db3f2356eca : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a9b297488844c73bdd324968d60630a : SUCCESS in 5m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7285b7a03d5b4f819b07402043f3b075 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2cb36896861a42a8b434304fe8c7490f : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4919dba009a4b988f8167142c681beb : SUCCESS in 5m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7be286f8c5a441eeae592cde277f2602 : SUCCESS in 10m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f6adb45b90d44695871d34fc6a8a1401 : SUCCESS in 10m 30s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/19ae9676094d45738be5e0bff58d2515 : SUCCESS in 14m 22s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fa2c30d330af426dbfd074e55233e9c2 : FAILURE in 10m 46s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/35d61e20b22245c6a50b2f1bbf5c3d5b : FAILURE in 1h 35m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/420dc24ba1ea4992841313bbb0fa7eb8 : SUCCESS in 2h 10m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/767bb9fe4b7c4db2860e59346c617c08 : FAILURE in 2h 12m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/43af8eb17bdc4af99ee0c8e0d38f4547 : FAILURE in 2h 02m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a5aa67f5da714c428f4bf767f8a97c32 : FAILURE in 2h 54m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bfa202d365d547099a5698de9e63109a : SUCCESS in 1h 05m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4682e55b5aad4aea921caa8581a5d92b : SUCCESS in 1h 50m 20s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/511516aa7b484ad5abb9721592c13038 : FAILURE in 2h 11m 24s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/be5938b2ff9c4ceb821295c14c9c48ca : SUCCESS in 1h 06m 57s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6b7d996f860145df9a74e95146924b14 : SUCCESS in 35m 28s","accounts_in_message":[],"_revision_number":3},{"id":"c2f740cba9a7f9ba41c7accb0e349f7f95158a81","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-21 13:53:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1a04e7637eb17d88ab8493a3b3d47eb93a93e364","revisions":{"8ecf365441f51da9cb46143649ee2db228df1a2b":{"kind":"REWORK","_number":1,"created":"2024-06-17 14:18:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/922114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/922114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/922114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/922114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/922114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/922114/1"}}},"commit":{"parents":[{"commit":"e3c62035764c4b7c060bf67a416bfbc389fce19a","subject":"Use class, not function, for image import plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3c62035764c4b7c060bf67a416bfbc389fce19a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-06-17 11:04:31.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-06-17 14:15:06.000000000","tz":60},"subject":"Add image plugins status","message":"Add image plugins status\n\nAdd a new \u0027status\u0027 class method to the plugins that, as the name\nsuggests, allow us to retrieve a status for the plugin. It is expected\nthat this method should return \u0027None\u0027 if we don\u0027t want to report a\nstatus (because the plugin is not user-facing and/or reporting the\npresence of the plugin could be an information leak). All other plugins\nshould return a dictionary with useful information for the user.\n\nWith this in mind, only two of the in-tree plugins provide a status:\nimage_conversion and image_decompression. Of these, the former\u0027s status\nis an empty object and purely exists so users know that conversion is\nhappening (which is useful information due to the fact that this being\nenabled will result in a different hash value), while the latter is a\nuseful demonstration of this attribute and exposes the support\ndecompression formats.\n\nChange-Id: I0cfb06895d8eaf88fd2b623d09993260c27ed471\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nPartially-implements: bp/image-import-plugins-api\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8ecf365441f51da9cb46143649ee2db228df1a2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8ecf365441f51da9cb46143649ee2db228df1a2b"}]},"branch":"refs/heads/master"},"67dc506a2975b9fb32e1584099dee92de11b4a0e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-06-18 10:23:55.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/922114/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/922114/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/922114/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/922114/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/922114/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/922114/2"}}},"commit":{"parents":[{"commit":"dd30d37b4304d629fba1033bbefae871a295d638","subject":"Use class, not function, for image import plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dd30d37b4304d629fba1033bbefae871a295d638"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-06-17 11:04:31.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-06-18 10:23:42.000000000","tz":60},"subject":"Add image plugins status","message":"Add image plugins status\n\nAdd a new \u0027status\u0027 class method to the plugins that, as the name\nsuggests, allow us to retrieve a status for the plugin. It is expected\nthat this method should return \u0027None\u0027 if we don\u0027t want to report a\nstatus (because the plugin is not user-facing and/or reporting the\npresence of the plugin could be an information leak). All other plugins\nshould return a dictionary with useful information for the user.\n\nWith this in mind, only two of the in-tree plugins provide a status:\nimage_conversion and image_decompression. Of these, the former\u0027s status\nis an empty object and purely exists so users know that conversion is\nhappening (which is useful information due to the fact that this being\nenabled will result in a different hash value), while the latter is a\nuseful demonstration of this attribute and exposes the support\ndecompression formats.\n\nChange-Id: I0cfb06895d8eaf88fd2b623d09993260c27ed471\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nPartially-implements: bp/image-import-plugins-api\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/67dc506a2975b9fb32e1584099dee92de11b4a0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/67dc506a2975b9fb32e1584099dee92de11b4a0e"}]},"branch":"refs/heads/master"},"1a04e7637eb17d88ab8493a3b3d47eb93a93e364":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-08-21 11:29:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/922114/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/922114/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/922114/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/922114/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/922114/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/922114/3"}}},"commit":{"parents":[{"commit":"367562768adf560f07e613895d937031ca736afc","subject":"Use class, not function, for image import plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/367562768adf560f07e613895d937031ca736afc"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-06-17 11:04:31.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-08-21 11:29:02.000000000","tz":60},"subject":"Add image plugins status","message":"Add image plugins status\n\nAdd a new \u0027status\u0027 class method to the plugins that, as the name\nsuggests, allow us to retrieve a status for the plugin. It is expected\nthat this method should return \u0027None\u0027 if we don\u0027t want to report a\nstatus (because the plugin is not user-facing and/or reporting the\npresence of the plugin could be an information leak). All other plugins\nshould return a dictionary with useful information for the user.\n\nWith this in mind, only two of the in-tree plugins provide a status:\nimage_conversion and image_decompression. Of these, the former\u0027s status\nis an empty object and purely exists so users know that conversion is\nhappening (which is useful information due to the fact that this being\nenabled will result in a different hash value), while the latter is a\nuseful demonstration of this attribute and exposes the support\ndecompression formats.\n\nChange-Id: I0cfb06895d8eaf88fd2b623d09993260c27ed471\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nPartially-implements: bp/image-import-plugins-api\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a04e7637eb17d88ab8493a3b3d47eb93a93e364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a04e7637eb17d88ab8493a3b3d47eb93a93e364"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
