)]}'
{"id":"openstack%2Fglance~922128","triplet_id":"openstack%2Fglance~master~I43ac62029637c015d6befe9071f1b594e9eec4c2","project":"openstack/glance","branch":"master","topic":"bp/configurable-hash-algorithms","attention_set":{},"removed_from_attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2026-05-21 13:53:13.000000000","reason":"Change was abandoned"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-05-21 13:53:13.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-21 13:53:13.000000000","reason":"Change was abandoned"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-05-21 13:53:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I43ac62029637c015d6befe9071f1b594e9eec4c2","subject":"conf: Add \u0027[DEFAULT] allowed_hashing_algorithms\u0027","status":"ABANDONED","created":"2024-06-17 15:45:22.000000000","updated":"2026-05-21 13:53:13.000000000","total_comment_count":13,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f4f36ba1db80f71ed1bb17eb6eb4c7a090eea391","_number":922128,"virtual_id_number":922128,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-10-01 11:51:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-17 15:52:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-17 18:07:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-21 16:00:41.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2024-09-03 07:46:36.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"CC"},{"updated":"2024-09-03 07:47:22.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"},{"updated":"2024-09-09 09:13:33.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-10-01 07:03:47.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"}],"messages":[{"id":"697e625ca1d380d51c251215f4f89a29167309ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-06-17 15:45:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41ebeac59b8b1ce2eb72b41d63d686bef86c8a18","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-06-17 15:46:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1154014e3d6c1afeeebb34283b2d7c45426174da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 15:52:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c241543bf10c48ea943f2e9b0a5bfb41\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1720fcf1b27e4a758aab92bdd4f66444 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e81493354f5e4f5887ce009a85fc59ef : SUCCESS in 6m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c32d0bdb5fb1061cdc3b0c036abd43dffb96f1d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 18:07:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3302d59f87a446dabd9ff2d2e068624d\n\n- grenade https://zuul.opendev.org/t/openstack/build/51f466ef64a7469e8be75b6bbb1e7305 : SUCCESS in 1h 03m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/718b3d7c199947299d8b1be456b4e73a : SUCCESS in 1h 45m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f714fe96db840a9bc3ca9797a26da17 : SUCCESS in 52m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b23702d61d9c4bedb23b7884d98fcc07 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac475d8635214ccea31f1876c6469c8d : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec9e7e144fe74040aa2622a5de840bc0 : SUCCESS in 5m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d1ba2097ac2a493eae8ad7f6069eae2a : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28db692ac7dd48c892fed7a1e4b0c68f : SUCCESS in 6m 24s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e62ac806bfff43179e62d9bcabe54483 : RETRY_LIMIT in 1m 44s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8ad16a86fedf468db45587689fa4535d : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1b1f091b5bc34d8ba07718e838a602ff : SUCCESS in 9m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fdc7864fa46e452197b1d4c1879dc21c : SUCCESS in 13m 23s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fbd38dd5212847a3b5ebde3b5e864634 : SUCCESS in 7m 59s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/56bc252af0a6439bbbdec8f191816b18 : TIMED_OUT in 2h 04m 37s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5f59cf4aa62c48b0891469188811f54e : SUCCESS in 1h 50m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e818a98ddf634a45aece51fd7467e81e : FAILURE in 1h 54m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/513b3ca41a0e4cd8b634e07f15a12fe3 : TIMED_OUT in 2h 03m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d148c340b1d446c291d5a25395d6979b : SUCCESS in 2h 04m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd87adf6187e49cd96642dcaaa708385 : SUCCESS in 59m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9eb09b105934452e93246490a575d86b : SUCCESS in 2h 16m 53s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b3caeedc35d744dba888177353381b5d : TIMED_OUT in 2h 04m 22s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f03927cd9ec44a7db849ada054906cc2 : SUCCESS in 1h 03m 51s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d7461102ab7c4066a13d70f3f58d9f06 : SUCCESS in 28m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9e463fee8d0342ccd36c5ce70ce02648e515c009","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-06-21 16:00:41.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"35cae370450e5a3db8b7aa3284878a62593a22ae","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-08-20 13:46:10.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"58852a75b7bde4df4212e0e95b588ee3dba1c637","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-08-20 13:46:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6334f31970ceedca91627ff0e75f4394a5aaebd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 14:19:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1349ffe3a4e24cf9a8ef31f06ac23f3f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/58b359cd5b694ac7bb16ebc9981560f2 : SUCCESS in 16m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ffd59ab30095483f9bd4edf6066f5f01 : SUCCESS in 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3b4b451432854164707804ff293bf15207f5b6b8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-08-20 15:29:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4ad743a7d248a0afda8c3164ba1f4d0587be2de2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 16:36:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/284def94abf44b37b9666b755cbabd93\n\n- grenade https://zuul.opendev.org/t/openstack/build/ce257be64f2b46328c62cd4061bc6f37 : SUCCESS in 57m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/19e9f258be704f96bab91ab87002a325 : SUCCESS in 1h 24m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf95ccdfd81a4fe892177274e908b67f : SUCCESS in 56m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19d5a6487b6240889729a09161144c45 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00f69ce2abc2440e868dc52fbca146f5 : SUCCESS in 6m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/82d72d9ddd6041d3b5870d6a0e1ee710 : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6515050a0a38494eb296c4efd1c21ba4 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccb906b697a84946aec9bfa64ed3e7a0 : SUCCESS in 7m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fca1d5cae45f4fd5a232f35946495555 : FAILURE in 8m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/420b06ca2af54cdb96916e56d9c92fca : FAILURE in 8m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/12b6c4e068b34037956879b2f99750ad : FAILURE in 12m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/da9bb43dfd7447d19e2af6dadee994c2 : FAILURE in 9m 44s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/24277e5505fa4989bf76a12e8bc088e6 : FAILURE in 1h 50m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/125c21801f7f4e57bef5faef63fba5b8 : SUCCESS in 1h 49m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8b0a4e76615d49dfbf6e70ec8af4b197 : SUCCESS in 2h 01m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/78c816c3337d44949a59a3682614c033 : SUCCESS in 1h 48m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3e37392dced04f4b990ff828c36c7c14 : SUCCESS in 2h 43m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da8c830804c143f08c2c5e0a55e666c0 : SUCCESS in 53m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8b2e1c267959478f8fe1c74dda0a813e : SUCCESS in 1h 46m 05s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/cb98dff875aa414e911faed512645e10 : SUCCESS in 2h 29m 25s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b91c6f86c8eb42c2a758d5ab96fcba93 : SUCCESS in 1h 07m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/193772f06a2343dd87a0d2a3f79ef4a2 : SUCCESS in 30m 20s","accounts_in_message":[],"_revision_number":2},{"id":"ebd256a1e2054987748d8b570095a1b7b6504b93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-08-27 14:51:31.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"09152c435dfa5ac312388848355f87a832e96c31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 15:35:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e9e20f7491b41ccbe77881bbb52e6b5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ad83a2156454db781ffbc2f92d75b84 : SUCCESS in 13m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/285beb4d7d234bcc90aa863653983dec : SUCCESS in 10m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7ccf74e84a1d2c86f564e2adf674db8d866be35c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 18:02:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d80782b4742453193d86f36b1ac3fee\n\n- grenade https://zuul.opendev.org/t/openstack/build/deab733427a04315b39620aca3195010 : SUCCESS in 1h 13m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1499be5305eb487aa8bfab72105eb15a : SUCCESS in 1h 42m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6086c74000524e238bdc6b83a4e508e2 : SUCCESS in 52m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/678292d50c1b4126bd93bee67ac30aeb : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1bff721ccc24d96920c31c70d2480f1 : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40b1b7f02c694ce39e282e77631b4b76 : SUCCESS in 5m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6cfe9b557158426e938b1a8df954bbea : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39f008d1de794cd4a36b3414a5fa5dbb : SUCCESS in 7m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a3b976fcdfec446bb17dd8a68cc0480b : FAILURE in 11m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0558e4af03f54001840391ee3acb25a1 : FAILURE in 10m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/98b91861a7ec468fb1956b7c5f87784b : FAILURE in 14m 59s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/eaaa127cfaa44872848548f80a2778c9 : FAILURE in 11m 01s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/86b4a3b5410847caaec9e2d3d9bde331 : FAILURE in 1h 48m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8cf1a7788008401ba6608ca98977fb3e : SUCCESS in 1h 59m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b78ea3bbb1a7497a8831d357e1a5d8bb : SUCCESS in 1h 50m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c342463132f8404ab53fbea7f925ba0c : FAILURE in 1h 49m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/84457821c3274474af1d8fcea401bc1f : SUCCESS in 2h 34m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c733a589b38140f3b6ed12f95bd03fac : SUCCESS in 57m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/af01484014ec4e11b9ea0555c5475954 : SUCCESS in 1h 40m 09s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/8f9cb38c45724c32a3f377304d79bb7a : FAILURE in 1h 32m 00s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7845226fbe684b12a5986436e1457e1a : SUCCESS in 1h 10m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/22eecd6ca86744b399154676dcdb443a : SUCCESS in 40m 06s","accounts_in_message":[],"_revision_number":3},{"id":"73d23ac2e93dda04d94db044b5c1a85838d5eabd","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-09-03 07:46:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1271446164c6afc71141d25b67e036f9b224d052","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-09-03 07:47:22.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d35c82387ee21215e0bcb227563f76d21585ee9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 08:13:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd4d15e72ca247aab9fcaef4cee4bfa6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb368ec558cd4b14b031c232e3edfe1e : SUCCESS in 13m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1b685e7f6e9842a2a3fdfb6739cefa72 : SUCCESS in 9m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"41ea0e14fac0d1dd3be31f4f546e28aa3d1cb4a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 10:11:24.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1147b577ecf6485193e9dc5e93548c1a\n\n- grenade https://zuul.opendev.org/t/openstack/build/0753bc39ed024883ae76392579c93f22 : SUCCESS in 1h 23m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/77f2c9f6b1a9453ea047da8b6fa14e6a : SUCCESS in 1h 16m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5c9b4539d6c4ef5a10126fd26c59c04 : SUCCESS in 51m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8afb6853d8b44245829e36f85b7c0b6f : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e4c43729813449daa23044924dfa9e6 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27707a7556cb4b3383e6b82232787def : SUCCESS in 6m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e04abd91bc6b4203b31f37e03f0d3350 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2e1248176494c7a966f5c856504e106 : SUCCESS in 5m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/69d9ed45b23e45c2af68914a81e7a5bc : FAILURE in 8m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b4025cd2e6a94d788ab479ae9683cf14 : FAILURE in 8m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/829504e8c1b340b9b2c485338da6d52e : FAILURE in 13m 32s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c3713a2c5e1b46ddaa400272fe374695 : FAILURE in 9m 04s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e17d41b472f24c7b92ead0b08602e47b : FAILURE in 1h 59m 37s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a59c3611948349a089d76399f17aabd6 : SUCCESS in 1h 50m 31s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/97c799711d2c48708247ed05be0f2fcd : SUCCESS in 2h 12m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9ad067f044634e0884f27015ff990b66 : SUCCESS in 2h 05m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4be8f0866af44fb3bf49294419ed8ed0 : SUCCESS in 2h 18m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7fa1f75176243618329bb5b0b39dbb8 : SUCCESS in 1h 13m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c5e747dff74a4c32a4a9a952ab5db7be : SUCCESS in 1h 52m 09s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/7123fbca861144339653ae65c43b8579 : SUCCESS in 2h 16m 41s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b0ac7e33485042cc8bd26778190b6f22 : SUCCESS in 1h 18m 40s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3aaca2a62ab3440286b9a3b33fb53add : SUCCESS in 47m 52s","accounts_in_message":[],"_revision_number":3},{"id":"420ba23555e0614d21b059113f528026c5aef53c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-09 09:13:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a1298909ed118d1decf29058e5d8840ce969df85","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-17 11:20:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ae8a7fc037a19f39815120578358b44971881ba9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-17 11:21:53.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9de7d2e9aa35eff7688013051f9f08db354c1364","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 12:27:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30e464dbc0ec4d5993cc8c8cf746eeec\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a24a2d68ad5f4b8ab06481f558869656 : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b7d75d9fb5b244a89e27cfa57a3f2c60 : SUCCESS in 7m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"62047d5b8ddc72408a645ede188c48a61323245b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 14:08:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/323fb821b6dc4e5e8ce14259a97eb61a\n\n- grenade https://zuul.opendev.org/t/openstack/build/506172e539ea4527a4d81ee88379f1c3 : SUCCESS in 48m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a76b11143d344c579c72e6334ae0119c : SUCCESS in 1h 29m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31112e3f6e7845d49dfd4e05b91cdc0c : SUCCESS in 28m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3bf6592102e40e79fad88e76eff9b23 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/931caa687c804e769675cda1eddce26a : SUCCESS in 6m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36887febfba4410d8bff2c217aba5fa9 : SUCCESS in 6m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a0f94fcfb8b4a57b86784e52d02920c : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30859ff38dbe435997d08324dad8975c : SUCCESS in 6m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f239f45864264897acb390e3fc7c3858 : SUCCESS in 9m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cf07598669ed4dd483ea18c8ef039765 : SUCCESS in 8m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4903844a0fb6470e90d72455452a465e : SUCCESS in 10m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4488333e6ce04154a768e41ebcab3ba8 : SUCCESS in 1h 51m 28s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bb81ad9171c044e683721de1a80ca57a : SUCCESS in 1h 47m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c325966d93324690bea28ba989c4de27 : SUCCESS in 2h 00m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0afa327eaccf42d08acf851afd542a4a : SUCCESS in 1h 14m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5f76372ada524759b27e5a51382fb8f3 : FAILURE in 2h 39m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf5b8d4cf1c74726b4e94375b5f39b61 : SUCCESS in 52m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e24cfe0c36a348d8b61d6f5d4bfe6f69 : SUCCESS in 1h 47m 14s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/0a78adddcedc4aa89297d26e3c5991dd : SUCCESS in 1h 50m 00s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b842fe2c88c9459d95e34fcedc43ebc3 : SUCCESS in 55m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4e481726a54f4f3cae700e3605ce870f : SUCCESS in 19m 51s","accounts_in_message":[],"_revision_number":4},{"id":"b9402c4ffbc7a139c374158067f28cb85439f15f","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-10-01 07:03:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ece7c10b71cc2ace07d04c02010a1f5f0f0930b5","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-10-01 07:06:34.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"748ebb3b6df7734a280b19ec7a203cce5be41ec1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 07:17:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4366a1f1cb43400f9fdf6c593596055a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d30de6d1f5714994a2d6351e72d284b7 : SUCCESS in 9m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ea893c8123e84001b3510fdc19c57107 : SUCCESS in 11m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7a70576df7126272d45a9d8a775b706b37c13b35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 09:10:30.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dfbac6f770024d45ac354e00371f91e4\n\n- grenade https://zuul.opendev.org/t/openstack/build/fda7a9a97dd746eba0c117216d40e02c : SUCCESS in 59m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/550f8c46b41c47858a85a9e2ecfc4e1f : SUCCESS in 1h 03m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f8fb14486a15481f9a745afb36e6481c : SUCCESS in 1h 37m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7cba306bf7f8470bb2f1dcbc397cf9b2 : SUCCESS in 30m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/444cccd4748b4b94bb46ae5f8cf94732 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09fff4a6b7a64368bf047645b3129a08 : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/488ea85e595346a6bff36fb1d367912a : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39a2c8aee342459d91dcb852d418df78 : SUCCESS in 6m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9f5399ff28504c028bcbd6accf5cbeac : SUCCESS in 8m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/04aa8e7f19d54d2dbdf62a78c54a3ca7 : SUCCESS in 7m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/da6b20d8c4754b8286b189ad00659139 : SUCCESS in 12m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/79c09219f0d04223bc3c0fac5b621dcb : SUCCESS in 1h 29m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/45240957ebbd431eb060805f666ae2b6 : SUCCESS in 1h 55m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b1cb338770e94b6d8523f59c5cfbd082 : SUCCESS in 1h 51m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a6ead7727bda41e6a9ebedd254ebc609 : SUCCESS in 1h 45m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/649e474b2a5142abb0c2ce298edea715 : FAILURE in 2h 02m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d43e38b0da0a4d9dbfdb96d65ac820bc : SUCCESS in 1h 08m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/efe2092a38a34bdfa9ffde0e8f5f732a : SUCCESS in 1h 41m 08s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/8936aa31edb24906a6d7d90b3548cc87 : SUCCESS in 1h 59m 24s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9d6c5e1a9a9345ab891ade28f8bba41f : SUCCESS in 1h 01m 04s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5c7faaeb6a6d4d39b8115369dbad416d : SUCCESS in 34m 26s","accounts_in_message":[],"_revision_number":4},{"id":"d5162c113de2a7f244c3c4e33c0aadb9811456aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-10-01 09:27:15.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"c28ffadc80cd968ecebaf4be4c8862de6e2796c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 09:47:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52abc077cfbf44eab75d6e8772f6a021\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd1674ad10014b8d8eca63b314496798 : SUCCESS in 13m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b85216d9492d4bac9e3c7462838cff89 : SUCCESS in 18m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a46ad4bd00472c5e1fe6ccfe8c42080257d5aedf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 11:51:05.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/10200fb5e3f947b2be7f0b4b3f28d42b\n\n- grenade https://zuul.opendev.org/t/openstack/build/ee77a703e39d47be8397764637fe34f1 : FAILURE in 53m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/76aff67d7e904acc8b55159efa6dae8d : SUCCESS in 1h 01m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f01f240fcf8f4480b9ff0f1a2229b1c1 : SUCCESS in 1h 26m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7eac041d155f4fbe98b21369412ea52e : SUCCESS in 25m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c9d7ba8345c4a7dbf330d15dbbb2536 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c92c932207994461ba16c45560d18879 : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08b70e06e7354c33b90318decda708a2 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77a6823fb95449659b52f1fd15d54a85 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e86e5e5dca0848378b1666d28bec7733 : SUCCESS in 5m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/76d46c6124284e93865bf90f6d6fbdc3 : SUCCESS in 8m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/613716492e184b78a4444e18389c1145 : SUCCESS in 6m 48s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e207b397c9514c6a8ef27ba4a1d52ab0 : SUCCESS in 12m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/74c5d7c92eed4511b2b5f6e1d978e165 : SUCCESS in 1h 41m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e51908765d134a5ca2f87eda46389465 : SUCCESS in 1h 47m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7af71d755bc949ceaadb206167d28489 : SUCCESS in 1h 12m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/14878a96b19942aa9e0ea06f62b2552c : SUCCESS in 1h 48m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e75997a35b704c9c8795fdf2d31fd8bf : SUCCESS in 2h 19m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25ac17f31d884a378ae368aceb501bd8 : SUCCESS in 58m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c949486a797a466aa69be524c547954e : SUCCESS in 1h 44m 17s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/f190bbdc8fb74861851d5df6925a123f : SUCCESS in 2h 10m 25s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1123c80371a54e4a8a7af31b08f29305 : SUCCESS in 1h 04m 23s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/faefad96637b4d5987bc3ba37d8828ae : SUCCESS in 36m 39s","accounts_in_message":[],"_revision_number":5},{"id":"f4f36ba1db80f71ed1bb17eb6eb4c7a090eea391","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-21 13:53:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3e9116bf0dc583b00706f5adc92b4edee267d9b8","revisions":{"f4d6afc7c2e35a051c777bf0c92e1c8ddbc99e6e":{"kind":"REWORK","_number":1,"created":"2024-06-17 15:45:22.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/28/922128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/28/922128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/922128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/922128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/922128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/28/922128/1"}}},"commit":{"parents":[{"commit":"dfbfdad77fdc91899f64d2ff186097cfb2065ab7","subject":"Move all periodic tips jobs from py38/39 to py311/312","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dfbfdad77fdc91899f64d2ff186097cfb2065ab7"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-06-17 15:11:55.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-06-17 15:15:41.000000000","tz":60},"subject":"Add \u0027[DEFAULT] allowed_hashing_algorithms\u0027","message":"Add \u0027[DEFAULT] allowed_hashing_algorithms\u0027\n\nNothing uses this yet but that will come soon. We also rename \u0027[DEFAULT]\nhashing_algorithm\u0027 to \u0027[DEFAULT] default_hashing_algorithm\u0027.\n\nChange-Id: I43ac62029637c015d6befe9071f1b594e9eec4c2\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f4d6afc7c2e35a051c777bf0c92e1c8ddbc99e6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f4d6afc7c2e35a051c777bf0c92e1c8ddbc99e6e"}]},"branch":"refs/heads/master"},"8a4252c34db56c792e8faf9add0f64786a827748":{"kind":"REWORK","_number":2,"created":"2024-08-20 13:46:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/28/922128/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/28/922128/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/922128/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/922128/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/922128/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/28/922128/2"}}},"commit":{"parents":[{"commit":"35477aff73398cd5fd742c3d1e32f30f6d59f59e","subject":"tests: Simplify version test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35477aff73398cd5fd742c3d1e32f30f6d59f59e"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-06-17 15:11:55.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-08-20 13:44:27.000000000","tz":60},"subject":"conf: Add \u0027[DEFAULT] allowed_hashing_algorithms\u0027","message":"conf: Add \u0027[DEFAULT] allowed_hashing_algorithms\u0027\n\nNothing uses this yet but that will come soon. We also rename \u0027[DEFAULT]\nhashing_algorithm\u0027 to \u0027[DEFAULT] default_hashing_algorithm\u0027.\n\nChange-Id: I43ac62029637c015d6befe9071f1b594e9eec4c2\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a4252c34db56c792e8faf9add0f64786a827748"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a4252c34db56c792e8faf9add0f64786a827748"}]},"branch":"refs/heads/master"},"3a0eddce89f5949ea581909fc2dc271fce842f98":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-08-27 14:51:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/28/922128/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/28/922128/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/922128/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/922128/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/922128/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/28/922128/3"}}},"commit":{"parents":[{"commit":"ea5ae117f11f5960b694b411825cef5238455126","subject":"tests: Simplify version test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ea5ae117f11f5960b694b411825cef5238455126"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-06-17 15:11:55.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-08-27 14:51:25.000000000","tz":60},"subject":"conf: Add \u0027[DEFAULT] allowed_hashing_algorithms\u0027","message":"conf: Add \u0027[DEFAULT] allowed_hashing_algorithms\u0027\n\nNothing uses this yet but that will come soon. We also rename \u0027[DEFAULT]\nhashing_algorithm\u0027 to \u0027[DEFAULT] default_hashing_algorithm\u0027.\n\nChange-Id: I43ac62029637c015d6befe9071f1b594e9eec4c2\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a0eddce89f5949ea581909fc2dc271fce842f98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a0eddce89f5949ea581909fc2dc271fce842f98"}]},"branch":"refs/heads/master"},"3a38d00fba5605091c9a48656f5e7a272dd738e9":{"kind":"REWORK","_number":4,"created":"2024-09-17 11:21:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/28/922128/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/28/922128/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/922128/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/922128/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/922128/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/28/922128/4"}}},"commit":{"parents":[{"commit":"c3af590baa9ded93dacff7f8934e6c8f1b85bb68","subject":"tests: Simplify version test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c3af590baa9ded93dacff7f8934e6c8f1b85bb68"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-06-17 15:11:55.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-09-17 11:20:01.000000000","tz":60},"subject":"conf: Add \u0027[DEFAULT] allowed_hashing_algorithms\u0027","message":"conf: Add \u0027[DEFAULT] allowed_hashing_algorithms\u0027\n\nNothing uses this yet but that will come soon. We also rename \u0027[DEFAULT]\nhashing_algorithm\u0027 to \u0027[DEFAULT] default_hashing_algorithm\u0027.\n\nChange-Id: I43ac62029637c015d6befe9071f1b594e9eec4c2\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a38d00fba5605091c9a48656f5e7a272dd738e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a38d00fba5605091c9a48656f5e7a272dd738e9"}]},"branch":"refs/heads/master"},"3e9116bf0dc583b00706f5adc92b4edee267d9b8":{"kind":"REWORK","_number":5,"created":"2024-10-01 09:27:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/28/922128/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/28/922128/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/922128/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/922128/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/922128/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/28/922128/5"}}},"commit":{"parents":[{"commit":"5d4973717f50184c1e7d8d694b09e86186f3f844","subject":"tests: Simplify version test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d4973717f50184c1e7d8d694b09e86186f3f844"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-06-17 15:11:55.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-10-01 09:26:30.000000000","tz":60},"subject":"conf: Add \u0027[DEFAULT] allowed_hashing_algorithms\u0027","message":"conf: Add \u0027[DEFAULT] allowed_hashing_algorithms\u0027\n\nNothing uses this yet but that will come soon. We also rename \u0027[DEFAULT]\nhashing_algorithm\u0027 to \u0027[DEFAULT] default_hashing_algorithm\u0027.\n\nChange-Id: I43ac62029637c015d6befe9071f1b594e9eec4c2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nImplements: bp/configurable-hash-algorithms\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e9116bf0dc583b00706f5adc92b4edee267d9b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e9116bf0dc583b00706f5adc92b4edee267d9b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
