)]}'
{"id":"openstack%2Fglance~922198","triplet_id":"openstack%2Fglance~master~Ia83a27ec897ebc2f0f477b1b0acd9484dd7beac0","project":"openstack/glance","branch":"master","topic":"bp/image-import-plugins-api","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-05-06 17:03:29.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia83a27ec897ebc2f0f477b1b0acd9484dd7beac0","subject":"DNM: Revert \"Use class, not function, for image import plugins\"","status":"ABANDONED","created":"2024-06-18 10:25:35.000000000","updated":"2026-05-06 17:03:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"49a12f685201c364db60818c680046e88a66d680","_number":922198,"virtual_id_number":922198,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-06-18 12:50:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-18 10:32:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-18 12:50:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"15f7f3d1c650d7c55fdc09a0cbdb7693c6f8c694","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-06-18 10:25:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a31a4ab162b5e059d05bda2b7ee0b74601c68c42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 10:32:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bb8984733a14a189370171005cdc1b8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ac6118c55b94c14ba0238e41d655769 : FAILURE in 4m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2b1fede9cf024139800c7a7cef41d2af : FAILURE in 6m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a0be2bb73e49d028bf4181eb96ab49eb4f0834e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 12:50:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/93bfd2b3fea24f438b5730d5cef1b97a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/86b11a8ca57742a6ac60655a7c69605e : SUCCESS in 1m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/78128a5a4abe41328f4d44631ddc554f : SUCCESS in 1h 06m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/98b07c8eef954be6bc1a2e6ed0cdc502 : SUCCESS in 1h 24m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfb600497b8f45d483a6e4a63fc166bd : SUCCESS in 1h 08m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f645c533b7b4ee78b55558a166bff2f : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f01ab12ef1d47c1915cc0af958c2042 : FAILURE in 5m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23e323714eb4449287f296d739996d8f : FAILURE in 7m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65552da6acfe4351b01d71ecf19a2840 : FAILURE in 8m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcfe9850c28840c381bc1d7e5902bc82 : SUCCESS in 5m 38s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1a4dc016fad24042bfb46332e55b4d9f : RETRY_LIMIT in 1m 51s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/215f3218fa0c465fbe74879de34019d9 : SUCCESS in 8m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/90ad66534a1146a08581bbf8e6ab6171 : SUCCESS in 8m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/33bcb2ee8c974935b55f3b0fe7809c3b : SUCCESS in 12m 25s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a8e0b9fc01de460190e0b5e32ab012fd : SUCCESS in 8m 21s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/de09627ec7f942d0ba53a4c879e17a98 : TIMED_OUT in 2h 03m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d207175ac1644091b134be30edecaacd : SUCCESS in 1h 42m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c74126e74c6b45d48a612be70d87c78d : TIMED_OUT in 2h 04m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5ca6cc25c12d4a86b8554386d11ac927 : SUCCESS in 1h 56m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/53f592c3293043a4b6d2efbc7217b1d0 : SUCCESS in 1h 53m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3e7c7d8ad5443a1a7a4dfee6347ff4d : SUCCESS in 1h 03m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8198f46fd8704f33a5625b254c866fd5 : SUCCESS in 2h 10m 15s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/57b67008a6064870a53614bdd946d65f : SUCCESS in 1h 56m 38s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5e3a2d1ffb9040d9afaf8937ba0aa859 : SUCCESS in 58m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1a3f98b537dc4b33b5567eeb57f455cc : SUCCESS in 32m 48s","accounts_in_message":[],"_revision_number":1},{"id":"49a12f685201c364db60818c680046e88a66d680","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-06 17:03:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d78480bd8e28c5c1135040cd99077a15162eba33","revisions":{"d78480bd8e28c5c1135040cd99077a15162eba33":{"kind":"REWORK","_number":1,"created":"2024-06-18 10:25:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/98/922198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/922198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/922198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/922198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/922198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/922198/1"}}},"commit":{"parents":[{"commit":"dd30d37b4304d629fba1033bbefae871a295d638","subject":"Use class, not function, for image import plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dd30d37b4304d629fba1033bbefae871a295d638"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-06-18 10:23:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-06-18 10:23:00.000000000","tz":60},"subject":"DNM: Revert \"Use class, not function, for image import plugins\"","message":"DNM: Revert \"Use class, not function, for image import plugins\"\n\nThis reverts commit dd30d37b4304d629fba1033bbefae871a295d638. We\u0027re\ndoing this to test that old plugins still work as expected, in lieu of\nunit tests.\n\nChange-Id: Ia83a27ec897ebc2f0f477b1b0acd9484dd7beac0\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d78480bd8e28c5c1135040cd99077a15162eba33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d78480bd8e28c5c1135040cd99077a15162eba33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
