)]}'
{"id":"openstack%2Fglance~922316","triplet_id":"openstack%2Fglance~master~Idd1cb8982b40b85a17821596f76dfa10207f6381","project":"openstack/glance","branch":"master","topic":"New-Location-Apis","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-07-05 08:07:56.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-07-05 08:07:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idd1cb8982b40b85a17821596f76dfa10207f6381","subject":"Fix: optimized upload volume in Cinder store","status":"MERGED","created":"2024-06-19 13:05:41.000000000","updated":"2024-07-05 08:09:13.000000000","submitted":"2024-07-05 08:07:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922316-New-Location-Apis","meta_rev_id":"bc3932d7e87df25780f75282035b1326ee7b8f8e","_number":922316,"virtual_id_number":922316,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-05 08:07:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2024-07-04 17:39:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2024-07-05 04:08:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-07-04 17:39:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2024-07-05 04:08:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-19 13:12:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-19 15:31:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-20 05:23:32.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-06-25 06:14:09.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2024-06-25 14:40:14.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"52161cd8c84462fb14866609662307514a9e2683","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-06-19 13:05:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01bb5a363c0e0ed5f43061754d5f1d879c07b9c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 13:12:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d99eafaf1ad41cf9789942b398c2e9a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05b40b70f67540ae970f312dab255447 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0fc5841a244a406598e8f5400d6cc8da : SUCCESS in 6m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d95456ac68ebcfc7768ca0486f5ca90cebfcc76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 15:31:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/15940607c3bd4146a09ebfc178581ccf\n\n- grenade https://zuul.opendev.org/t/openstack/build/12120314abe547969601a8bd19e64953 : FAILURE in 56m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a2377a8fc84649f7b289bc6131bb006c : SUCCESS in 1h 38m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fbb4369f02947109203d9bceeed3f90 : FAILURE in 36m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/335e007b60d74b05b8aa41fc72a9e6eb : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/310ab3187142444c99ea622be675c01d : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2fa9dcd205f04727b452bac58651ce6e : SUCCESS in 5m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c19201bb27a64a88a32c2ac06a4272b1 : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e31b839ab4cf4f19a62917efb402c1fb : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bef11322a0ba4cd587989bee6c7b1775 : SUCCESS in 7m 35s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c01b80d90bcb4ac8a2e5a3e5d9702ff0 : RETRY_LIMIT in 2m 01s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9034ecfed8ff44ca9b84011b65b84a22 : SUCCESS in 9m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3f177b6954044bed89fc646faf2dd3e4 : SUCCESS in 7m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d7347a8f7f344229ae9368ef179c3ebb : SUCCESS in 15m 26s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e05b256dfd904d87b12fd7ca2fdf7672 : SUCCESS in 8m 53s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0ba47167a88240d98f6007f46050ce80 : SUCCESS in 2h 03m 44s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3e773a99956a4f9eb6e0bd5cf296cb33 : SUCCESS in 1h 52m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e9fc8bb352b5476da4e63dad25d6bc6a : SUCCESS in 2h 02m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e33054d1f74545a39a647e9adcd958df : SUCCESS in 1h 50m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4730f41a50ac4ea3a39bfb229330cc8a : SUCCESS in 2h 06m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10ff6c52647641578de133ab6255d325 : SUCCESS in 1h 05m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1a39c496f81444f9888b714d260ebccf : SUCCESS in 2h 18m 06s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e7fc4a6539a84be489e2c4d52ecbce70 : TIMED_OUT in 2h 05m 20s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/708c6a4b5d30476a919ecb2de451e87d : SUCCESS in 1h 11m 06s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a215311da6b7470cac3c29ab606e8544 : SUCCESS in 29m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a0373d98647aa009c0e73847ea9ef6ad07edbb6d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-06-20 04:05:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0507d6614ae47068f45f0eb3c62e8cf185d2b18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 04:10:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64547456b29e41e9910560c9cb0e52cb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a91690d489904b91a01cc4c59590c57d : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7f402fe13b9b47a48cae06c295807b0c : SUCCESS in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8ccc8165585a11b8036b68d572a549f26f29c6b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-06-20 05:23:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d1d9a207b5ecf9c0fd721367be382bb5297418b8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-06-20 06:21:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76f8eaa1b50465758f295a29f0bf40f73b82b5a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 06:33:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e7db86fc6c045109480fa94eb057d6b\n\n- grenade https://zuul.opendev.org/t/openstack/build/fe21cdf37ea94d7f9083dc4179c7f258 : SUCCESS in 56m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fa345108910e4454bd8d2d79dd258178 : SUCCESS in 1h 26m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/885c29d7c0994733be98a165a2b1513b : SUCCESS in 1h 00m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac47bdb5cf5c4e14bc6b7d206a7564ee : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/948769c8844e4476a22c540976fea665 : SUCCESS in 7m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1673d9598d443b0ba2bbafade225cbb : SUCCESS in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ea43e8efff64125b282b5d0d9c1820c : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08bc4c4de5364abf9735a6a9b455e31e : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83d17342a2d64af888455cd961d2f9fe : SUCCESS in 6m 40s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/360739fd07a74e2596a647c5aa457b0a : RETRY_LIMIT in 1m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/106c6e5e248d4a0494bcfac4482ddc96 : SUCCESS in 8m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/edd05691cf844d698189f09d86de8004 : SUCCESS in 7m 55s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f45d7919b8494a13ae92abda12055a93 : SUCCESS in 12m 26s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/117f641fa783447ab3813205a37ab7cd : SUCCESS in 8m 11s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d928cd8da75d43859068b21f3cf7400a : SUCCESS in 1h 42m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9b5ca66961bc47948fb9026f8f3d3561 : TIMED_OUT in 2h 03m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/12c22fdd1287426baaf980bdfdca3b08 : TIMED_OUT in 2h 03m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/219216e20bef43a2bb287621323469dc : SUCCESS in 1h 52m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9b8c16d56016489ba1a8f3dffad93c79 : SUCCESS in 2h 21m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bf50caf74f548cda1d181e274d36fb4 : SUCCESS in 1h 01m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/247577b61f9e46cab25a7184ad3be788 : SUCCESS in 1h 57m 43s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/9925b7c8e0174eb8a3b09d2462593e01 : SUCCESS in 1h 58m 51s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5a257635578f4949a6f1bd3f7ef2ba1e : SUCCESS in 1h 12m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f16aac1ffa97417785f011f69ea5898a : SUCCESS in 38m 01s","accounts_in_message":[],"_revision_number":1},{"id":"6bb2846a011cda62a7edbb2081c530b564c49951","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-06-23 16:03:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9b8f84807ddf52456fe6cde8a3935b82472f9d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-06-24 05:06:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"48d6528a5d74e25a157ad4ab0611f7c8941b6289","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 05:13:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4ebdcb495b644de80b7b4510d43fe8d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e11b20c2818748739c4adab01651c9b4 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/82e1e342d180439dbd01c824e2a2c984 : SUCCESS in 5m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"746de3ca90879e6cb4da02cde590d9c1f74babcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 07:40:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/205f32799ad640318b2bd29499b94ec3\n\n- grenade https://zuul.opendev.org/t/openstack/build/2ddd8fafbbb94fad98800793b16c20a3 : SUCCESS in 1h 01m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5c9dbbdb9b5b4dbc89cf02e1b354111e : FAILURE in 1h 30m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ceb5243150c449baca92dc9807f2849 : SUCCESS in 1h 01m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eeddbe4688a4db399a475b92577e9f2 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e163701974334e04b7f667874b810fd6 : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b6383c421a84a35a9f435a4f9d24a67 : SUCCESS in 7m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8da12175f9074ef5ad2dfd4773a8bfe8 : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ab927503bdd430b829fc9a378902444 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81923c3039844e92ac82009a894bbf7b : SUCCESS in 6m 34s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5c4c949e2ad94d8ca7f75fb10cc063b9 : RETRY_LIMIT in 1m 35s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/09f1d86053124cca8e3806ab92c1e034 : SUCCESS in 11m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e3aa46c7e2e54d478bc6255f26f5125c : SUCCESS in 7m 46s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/59f1f376d7bd47b09b89dcaade37e2bd : SUCCESS in 12m 22s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f07fa49222c14ea2bedab9fc82e5a107 : SUCCESS in 8m 02s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/52e7205365604e648f85ddae491b897d : TIMED_OUT in 2h 02m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0d4792a964474a41863df1a5b329919b : SUCCESS in 2h 03m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/90eae4fa1bdc41f68b6f506115952749 : SUCCESS in 1h 59m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d8c463ab2af04885b863931a082e78a8 : TIMED_OUT in 2h 03m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f57bbc5297914ac390492fcb2d78b273 : SUCCESS in 2h 12m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8043320188c5400f914d6bae773dbcb5 : SUCCESS in 54m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fea93d8f64e04f168d04dadf442db368 : SUCCESS in 1h 47m 18s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/787787aebdda4e4e87da2aa6e6a7cfaa : SUCCESS in 1h 58m 34s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/65808e7eacb74011851660664e21cba4 : SUCCESS in 1h 08m 52s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6ef0b39cb8f54889aa4d39db359022d9 : SUCCESS in 29m 41s","accounts_in_message":[],"_revision_number":2},{"id":"a7b4ca31bf41b820885a55098580ce5893c68f70","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-06-24 14:45:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c6d4993ff40f80d075b3fadc341ddd84d659a382","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 14:51:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db563b9f4f794f15a74cb4c9c39f2f46\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/719b861eaccd4cdfa08eee44c5e988aa : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dcd20d10103e462f9c3cd438003b1e93 : SUCCESS in 5m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"46ab3f7f42f99e6defc780c671a44b2248895bb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 16:56:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/428e48303ffa45438aba5732c8d6d06b\n\n- grenade https://zuul.opendev.org/t/openstack/build/43992c819f414e15aef3c157feba42a2 : SUCCESS in 1h 00m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c65fdfbbefbc45a5854be0045e6c2190 : SUCCESS in 1h 47m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3bcf470e4f274293b51f3f3aff06931c : SUCCESS in 1h 06m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cff1b165114645858e8cf7a8de985567 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a88279d24844815ae163e1de4c4ab50 : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b42a72d552e14e0295409ae4c61a7b21 : SUCCESS in 6m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/177ec20d25624117941686435721cfc8 : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f24b3d23a89418b8a08dba9be004ad8 : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2709ad75e588459da01009d9f1a00c09 : SUCCESS in 7m 32s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9484edb7bd0f48a8b6e8fce51f0cdb00 : RETRY_LIMIT in 1m 52s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4479c397c5fe44b09fd124ade34de701 : SUCCESS in 9m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/518f3b8ec7a249e0913b3baa0efc0995 : SUCCESS in 7m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/09e464262a7f46be845b0ee18ccebeb3 : SUCCESS in 11m 42s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/27e4cf66f5b64681ae355eed7f938c0c : SUCCESS in 8m 24s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/853eb26e1ee845c886b9ec6737b828b3 : SUCCESS in 1h 41m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/14bdb9479e9e49e6ac32825ad66e3e5c : SUCCESS in 1h 47m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/87fb4a1ccad046b88623752316e80da3 : SUCCESS in 1h 53m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/571f17804c214fa18e2ed66cde4c86ff : SUCCESS in 2h 03m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7c344c48086749dfb8384abc86064db0 : SUCCESS in 1h 54m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42902f18fc144c96aa94b70379b693b5 : SUCCESS in 1h 06m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9f6ccd3d82a24cb8b36309207a4e970d : SUCCESS in 2h 10m 01s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/f04823e6d605456aa0f8bf23ed98bab8 : TIMED_OUT in 2h 03m 48s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3f96cbc0f20243c495a85fbe3e9f52d5 : SUCCESS in 1h 04m 54s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f5998c51d956456596f74f10c73e2ea0 : SUCCESS in 32m 58s","accounts_in_message":[],"_revision_number":2},{"id":"e9b525852639603804ed4f76e2115f7be3a38329","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-25 06:14:09.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3dbda00d212aef4a50458b0713f69069811277fe","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-06-25 14:40:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"baedfef89337aeebc8f856750ddc340679080cd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 14:41:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"90a3aa305e14ad5ff89d028bca61ecc5c1158916","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 15:29:50.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/906359 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"4d1f84783c382d42e98798ffcd1deeab15c6084f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-06-25 19:20:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"680c447750afa6a20fbb03768e34e2b38b3381f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 19:32:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97a413c55ef041e1af4efc5eb20bd52e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/44bd7db5b2ff49dc9bff0e4189c19563 : SUCCESS in 10m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ddb83778d99d4a798b500888d3a47f7b : SUCCESS in 10m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a0462d0598d333c975fd187e7b00f4bad2cc9260","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 21:27:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee60a44b7a7f499ba114bbb2444b83f6\n\n- grenade https://zuul.opendev.org/t/openstack/build/61aa2a20a4344cb5a63b551fa58fa6cc : SUCCESS in 1h 03m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/67e4d65457884ef9b70487c24c528322 : SUCCESS in 1h 25m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/244858965fa1419a850a52fac1b8aab1 : SUCCESS in 59m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6beae02c9cb14e468f48fc4994d2db50 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5444c060edb40799e8c374e04d1e65a : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d06629d43d084502a3c0275395ffa235 : SUCCESS in 5m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/26d149ff0f094553b75c490269c7411e : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35176f36074c44b8a3cd7f26d6f71c17 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6a9d6d5286c438da17e21fe02825d83 : SUCCESS in 6m 40s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/55b1f2a8f5844bed86b6c9c0e695ef92 : RETRY_LIMIT in 1m 55s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a05fa7c5a4d441369c2214e47cc19d48 : SUCCESS in 11m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/df7a6f91bba94cfbbb02b8e8b600ec93 : SUCCESS in 8m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/74b69f1426624757859a0536d95a83a7 : SUCCESS in 14m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/467f04bccc9841ed9ff5cc241d8df37c : SUCCESS in 8m 45s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b6db37b7e0d249698c1a97e99be45e8f : SUCCESS in 1h 47m 00s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/64470b116dce4dfb803ab9bbd5999e96 : SUCCESS in 1h 48m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0453282002aa4deea02ab4ea398f7921 : SUCCESS in 1h 48m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8d37708bc3cc4ca4a6e1614b602ead10 : SUCCESS in 1h 39m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/54808b576e924e34b5fbf908375ff3b8 : SUCCESS in 2h 01m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4a662135aa641209a16380ba76f2a26 : SUCCESS in 1h 01m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ec4db659aec64986b1a36bd5bdda709e : SUCCESS in 1h 41m 50s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e346682a44c84fccb33477fab345fad1 : SUCCESS in 1h 55m 00s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/0c0f5809c8294663acd0e6bfe9e11192 : SUCCESS in 50m 58s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/45441fa822b14037a9a3a602fdb7f643 : SUCCESS in 27m 56s","accounts_in_message":[],"_revision_number":2},{"id":"1e198d8e479e7e3a05021a30341fbfe77b9bfe24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 21:38:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"43848bc906b39ccab311d684fc3bbf1f8bfd44d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 23:49:25.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/906359 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"5c881ccd83abdb62c641bcac957e77c1911d6b5d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-06-26 06:15:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3e917c5840684cd160ddf251c2fa5eb8bd4588bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 06:21:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e1e8b1fbaa0401aafec2ed4f6956332\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/59c70f97810e417ca53d007b0cfae1d2 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1af18919ab3040b481544b58f4062889 : SUCCESS in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5a2894cba6a14c0acd22cc76a3d1b2c731bfb1ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 08:38:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a9f4f37f66d40a99f3f03fd3085b6cc\n\n- grenade https://zuul.opendev.org/t/openstack/build/3c834e81e9e34b80864093d91f6772bf : SUCCESS in 54m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f8a5716d3c4d4712b8804db7ef8163d9 : SUCCESS in 1h 26m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/865da6c2fc964b0bb02d9de7c02dcb59 : SUCCESS in 52m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b64dee310a7d436983e0dca3f9f95f53 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ef5a3dd019b44579cb662f8933e6c92 : SUCCESS in 5m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/78c181fca7b64e13a303e63c21f3f918 : SUCCESS in 7m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a46fe010b4f24a85b39c4a268d6a3f9f : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2ef27e4669b4a77b61de34488f09199 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c368a0fc37e4ad4a8f5cd7869c85a10 : SUCCESS in 6m 15s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d36bbfbadf8541deaf14408ef3ee8b80 : RETRY_LIMIT in 1m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c2fd6633650d4e46a8857b15a3c5a7e2 : SUCCESS in 8m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9489e6a8ade54a37a9e6da8957a154f1 : SUCCESS in 8m 30s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/492f97e346a2457bb09f228be1b351a2 : SUCCESS in 15m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b7397f9d2f2049478e1f5cbfaa6c5203 : SUCCESS in 8m 47s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/53346178473d4fed9b0c9a2ecfd6e6be : SUCCESS in 2h 01m 48s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/47bbfcbfa4f44b9c857e6c6380f9e821 : TIMED_OUT in 2h 04m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/23d81288b77541de8c5741dfd03792e6 : TIMED_OUT in 2h 05m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e3f6f61ceb834457a0b6950830c09046 : SUCCESS in 2h 03m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fd0c4d06296e47c08b813e4929f3a355 : SUCCESS in 2h 16m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e3f7ec44fc84a69969f892db4a2c2a5 : SUCCESS in 56m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6347ca6263684882802a08a7b8548ace : SUCCESS in 1h 52m 35s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e01dd951e8004264aa2c940fb3efe093 : FAILURE in 1h 50m 10s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/cefe3a571782457db040cd639c86d9d8 : SUCCESS in 1h 12m 17s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/891b0622f0a74fbbaf0b26e2b7883124 : SUCCESS in 32m 57s","accounts_in_message":[],"_revision_number":2},{"id":"15c2ff42195d5744d17652f378602cc8b27253da","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-06-27 06:35:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6496f84ac3a3712b9759ab11853dce7268f503d6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-06-27 07:16:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d5183cc4bcf1c32b214a954a9ac44503cef43eed","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 09:14:21.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"37e18f0e3c54938c05b0e997896be9ab67a8689e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-03 14:05:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"14b904e1258c3c1d977b928bca9afcba5ab229ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 14:12:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/797694df9d4447cfb63b6862f3d3feb2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cdea394a6bac47e3a8d7866cdaa4ce75 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4d767d12136e40e28cb52bfafb95802c : SUCCESS in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4de043edcad6a19e10bdff9e2e606bfc05415631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 16:45:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/10cb9d9af1c648339bd3aae3d6cf5c15\n\n- grenade https://zuul.opendev.org/t/openstack/build/595b5908ebc043049b980dbb6edf87c4 : SUCCESS in 1h 26m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0cb7b9bd0068460f9a0c75b3eab24a0f : SUCCESS in 1h 23m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b190b5230d0d47e19f450aaddc6fe564 : SUCCESS in 57m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05d92d767816403cbd8a8ff4bc224c21 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00a78bd44b0244f7b393de4f40c01d1c : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eaebcc19866b484bb5a0378cf9f8ee96 : SUCCESS in 5m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c05f55bf452d4ed19aa1b96c3fe4940c : SUCCESS in 11m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aea4a454a5841f7bcd9849a88edd9f5 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/533fc9246b4e43c4bff1f63a194bc056 : SUCCESS in 6m 32s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e35ca081c75b4f8dbd6b12d261b5aaf7 : RETRY_LIMIT in 2m 18s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/135b56753c57472aa99fbd76eceb9005 : SUCCESS in 9m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eaf4955b0e844f26a6064fe5537adfb2 : SUCCESS in 9m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3083b7c76e7b4b20be25113c3138910d : SUCCESS in 12m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/00c5572aad9c4bfaa27e9e8416e7d572 : SUCCESS in 8m 15s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f8affacb0cee4734880b86c62810c99a : SUCCESS in 1h 53m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8a58e279cfe0484f950835d51f01ce0c : SUCCESS in 2h 01m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bfb8cfb12049408eadb5cfca01b4f253 : TIMED_OUT in 2h 06m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2ed9327026fa4f25945d39280631a473 : TIMED_OUT in 2h 04m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9f410f5d39e4461089b50980a8caa8bd : SUCCESS in 2h 10m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1e6a8b2c6b947caa148ef9b22c23ed0 : SUCCESS in 53m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d24690a5d55744d8ae4366ab20b98a6f : SUCCESS in 1h 44m 40s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/0fe9e92ce697433bb21a8cf30afed593 : TIMED_OUT in 2h 05m 55s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2622fb78bed04c51b331416f393198c0 : SUCCESS in 1h 13m 05s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d70dac4a82e74d43a00cef70d4cf32eb : SUCCESS in 44m 11s","accounts_in_message":[],"_revision_number":2},{"id":"0f8daee444fb8f4d540b126376690f59cbcbfb36","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-04 07:28:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"01ceb1e1f8f34465fc4197b2cb4795576bebfee0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 07:39:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5812f865f60d40e39e4b2af0e7ab29b0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b3fa7d193954e67a4a12cafd00d582b : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0f32317408884a9986ba9487a7965135 : SUCCESS in 9m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"085d40694c0e0c94eb9ff8d32dee3aee7bea026c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 09:52:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f75759ab3da44f3cbd4de27a2b5e3f8e\n\n- grenade https://zuul.opendev.org/t/openstack/build/b590be668b204333b20b6c2785d55930 : SUCCESS in 1h 00m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7fe875dbc5a24bd8b0b7ab657c5eceb8 : SUCCESS in 1h 37m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0da1568b71704de19453b0ea0b9a33dd : SUCCESS in 1h 06m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e138ffae9f5d430db35051195e51d66f : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59ecb215106c468a80ae6225fade83dc : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d67efaeb1bd453db9e1a3129c0f8967 : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/865cbafb70554cdc95f2ce0141788d78 : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eb2779d39d943bea053b9fbb402b5f7 : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/951e969431b146aba2968f3c6529f3e3 : SUCCESS in 7m 00s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3909be99ab8b498fab62bb8aaa589263 : RETRY_LIMIT in 1m 39s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c2ebeb3dc3174e3283054aa0f3f4407b : SUCCESS in 8m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bdf60597566f4693b040e53f63de604a : SUCCESS in 9m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bbd11bab53c14a4490af4b3781c3449b : SUCCESS in 12m 46s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/509a40a1e03140dfb096ba1c0d0538c7 : SUCCESS in 9m 26s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/40d4678584ed4826bf99f813bec26d51 : SUCCESS in 1h 52m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5929febd2fd4418c8b5a11b14b3e3702 : SUCCESS in 2h 00m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/988bc73729c34d8392533e2d9f04e7a2 : SUCCESS in 2h 16m 09s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/76b8692d5ab74b2393d5e12819affdc7 : SUCCESS in 1h 53m 55s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5c08ce512a88423097ceb7abadc82c4f : FAILURE in 2h 09m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/892c858db834430ebc340c076de219b7 : SUCCESS in 1h 02m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f27c035736cf48a0b999505de3e1bd8c : SUCCESS in 2h 09m 43s (non-voting)\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/8151fa9a3ebd4281a7a4375917b36eb3 : SUCCESS in 1h 59m 16s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1b79ad3f55814f5cb1511c1d6ab67e6c : SUCCESS in 1h 09m 10s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8f75d6754b924c07a1f8692ee970dca5 : SUCCESS in 32m 21s","accounts_in_message":[],"_revision_number":2},{"id":"37e05ca0f5e65c44adb5817fad567b534830575f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 16:59:57.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0392dbf76f6d46affb76a6b5018c862778ce01ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 17:00:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c1a73446b95f71ebdc96764f3ffd5e3a7a935108","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-04 17:36:50.000000000","message":"Abandoned\n\ngate is behaving strange, abandon/restoring to restore gate jobs to sanity","accounts_in_message":[],"_revision_number":2},{"id":"cf10ab1cdcff88040772688ffc5326e514cdeb6d","tag":"autogenerated:gerrit:restore","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-04 17:36:59.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"91bc36c7de7afc01f424146efa955de44289a77b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-04 17:39:28.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0c4fbb97e0886082b3060b5959f561d2b0cfcf99","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-04 17:39:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b3660992fadd644bafbe9863b31a931588366ca7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 17:52:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d11da4f9f134d6d8661c9a3ec56d26b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d04def711fa640dfbca5d9be5ebd99c4 : SUCCESS in 11m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9553c164013f4646ad7afee9676f8c10 : SUCCESS in 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"311034b1f6c7c821a27519eabcb62e989225944e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 19:56:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f356fb8e873b4d45832d6fd6ae028edc\n\n- grenade https://zuul.opendev.org/t/openstack/build/9a01f88c43cf4c88b53ddf49cf309ae8 : SUCCESS in 52m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/53f597b2f9e34d529546f102cd809de0 : SUCCESS in 1h 40m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/245dccbc7b6e48af81507f6944f17934 : SUCCESS in 59m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/009086d78c6e47cdb5df899035964590 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53233189bfba4d8083d8185a851b9843 : SUCCESS in 6m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5f198427385f4025890f53a181e1ed36 : SUCCESS in 5m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f637b53f37540358e97cc225fb2c7ee : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a1c18d01ea643689f4b6a208ea3f823 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9921268ce3d44968e99a90c0133f909 : SUCCESS in 7m 04s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9e012ebc12f243e2bc203745a1f1de98 : RETRY_LIMIT in 1m 35s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0bda794bc4524a06aacdd8f08a0904b5 : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ece86130fc6b4df1a68ecdb45a81ae5b : SUCCESS in 8m 05s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/038ad2cc7b4f4ca6b338accd73a2525e : SUCCESS in 11m 25s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a55293db01cc42b8abc53606f43fce91 : SUCCESS in 8m 33s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c96f6589c8e74a42ae80f526b11b19e6 : TIMED_OUT in 2h 03m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a5ab2493951c4b65a586b7dbb5fb5e66 : SUCCESS in 2h 06m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6cad16fc4ec3417a853e29ef4f0e9711 : SUCCESS in 1h 47m 29s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e083c64c7f294c8baa3b02dc7f8d5baa : SUCCESS in 1h 43m 30s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/395916bf02594d9d8a0b5d29909d4f91 : SUCCESS in 2h 01m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a99d4ad5c9b147e7bbfe59747ad18c30 : SUCCESS in 59m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/294ff41dd5464e63bd690f1c97d6b5ff : SUCCESS in 1h 56m 17s (non-voting)\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/f0de8d06ab764130a4ef2c08ac597d96 : SUCCESS in 1h 51m 23s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/bb6000ce744f4ed7921ab2bf50689e27 : SUCCESS in 1h 10m 35s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a22dc54100d74c3aa2fd78a178f5c15d : SUCCESS in 25m 12s","accounts_in_message":[],"_revision_number":3},{"id":"5a68a1f31f12fc604a3ae6bc95a0d6fdd6531625","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-05 04:08:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9fb546f6eab230462451377d2267c7fb4bd8cf78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 04:09:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3c6890278145d0b6ca52fd8c8e2b2d71c1f37998","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 08:07:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec76594f6f84423b8ab18a04078ffbff\n\n- grenade https://zuul.opendev.org/t/openstack/build/c829be6e92a54ec38c59eb6692337aba : SUCCESS in 56m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b13f306137b84076b5b1ad08378dcf02 : SUCCESS in 1h 49m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6267373f1ec34a728e469c814e0fe7c8 : SUCCESS in 55m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f52f7ab52d244331b3fcde98de89318c : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/221d3113bdee4d2f91c8d7bfe8c78d21 : SUCCESS in 6m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5e6c59ca76f4e7a95e5b4e341d7d218 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d200431830a44e9ca4efc162377d189d : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb4aac37a30c4b0f843309bab5bd20f5 : SUCCESS in 7m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c9058d7ddfe641099456e49587ea5b92 : SUCCESS in 9m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ced36a809ed0474793c283f398456c9f : SUCCESS in 8m 06s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cbfb1c65e25d407aaf4c0b715b8cd8a6 : SUCCESS in 12m 20s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/79f61951a46a414e8d707d436c12b822 : SUCCESS in 1h 54m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b2d1b6116937420e9cc41806985c3061 : SUCCESS in 1h 45m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b2d089d603604941a7570048d62aff2e : SUCCESS in 2h 06m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6aed85067ea4a5997757b9450605999 : SUCCESS in 54m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/64cab6c2bbf7406dbbb6da03d105ab57 : SUCCESS in 1h 43m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2cd65ab14b3c408a90ed620404b745e2 : SUCCESS in 41m 45s","accounts_in_message":[],"_revision_number":3},{"id":"418b8692e69f15e807b673e5e32e09935c32d8de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 08:07:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"bc3932d7e87df25780f75282035b1326ee7b8f8e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 08:09:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cd4c34625d54249ab53ecc74a787271\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df717cb53e0844caa2a8e388e45d211c : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0bdf62c6ac345bf9e0a213e1fb27da7 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8318da1d5fad2b47fc5747c049b90fe9255c28ad","revisions":{"c25f4997cc79934de10b361cda9295f253daea23":{"kind":"REWORK","_number":1,"created":"2024-06-19 13:05:41.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/16/922316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/16/922316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/922316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/922316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/922316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/16/922316/1"}}},"commit":{"parents":[{"commit":"c0e96abf3fb195f666431123f1ef37e12b950834","subject":"Make location URL compatible with cinder backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c0e96abf3fb195f666431123f1ef37e12b950834"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-06-19 12:25:52.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-06-19 13:05:27.000000000","tz":330},"subject":"Fix: optimized upload volume in Cinder store","message":"Fix: optimized upload volume in Cinder store\n\nWhen Glance is configured to use Cinder store and we upload\nvolume to Glance in the optimized path, it fails with\nInvalidLocation error.\nThis happens because Cinder is not aware about the store\nin which we will create the image and supplies the old\nformat URL i.e. cinder://\u003cvol-id\u003e whereas Glance expects\nnew location format i.e. cinder://\u003cstore-id\u003e/\u003cvol-id\u003e.\n\nGlance has code to update the format from old location format\nto new location format but it isn\u0027t triggered in case of\nold location APIs.\n\nThis patch adds the context to the update store ID request\nwhich calls the Cinder store to provide the updated location,\nhence fixing the optimized path for upload volume to image.\n\nCloses-Bug: #2054575\nChange-Id: Idd1cb8982b40b85a17821596f76dfa10207f6381\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c25f4997cc79934de10b361cda9295f253daea23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c25f4997cc79934de10b361cda9295f253daea23"}]},"branch":"refs/heads/master"},"749189d979b9dc81547d77386656b89b0814b87f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-06-24 05:06:58.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/16/922316/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/16/922316/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/922316/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/922316/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/922316/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/16/922316/2"}}},"commit":{"parents":[{"commit":"61a63881111ec0d0f7f0d2ccfa68df153878abac","subject":"Make location URL compatible with cinder backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61a63881111ec0d0f7f0d2ccfa68df153878abac"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-06-19 12:25:52.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-06-24 05:06:58.000000000","tz":0},"subject":"Fix: optimized upload volume in Cinder store","message":"Fix: optimized upload volume in Cinder store\n\nWhen Glance is configured to use Cinder store and we upload\nvolume to Glance in the optimized path, it fails with\nInvalidLocation error.\nThis happens because Cinder is not aware about the store\nin which we will create the image and supplies the old\nformat URL i.e. cinder://\u003cvol-id\u003e whereas Glance expects\nnew location format i.e. cinder://\u003cstore-id\u003e/\u003cvol-id\u003e.\n\nGlance has code to update the format from old location format\nto new location format but it isn\u0027t triggered in case of\nold location APIs.\n\nThis patch adds the context to the update store ID request\nwhich calls the Cinder store to provide the updated location,\nhence fixing the optimized path for upload volume to image.\n\nCloses-Bug: #2054575\nChange-Id: Idd1cb8982b40b85a17821596f76dfa10207f6381\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/749189d979b9dc81547d77386656b89b0814b87f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/749189d979b9dc81547d77386656b89b0814b87f"}]},"branch":"refs/heads/master","description":"Rebase"},"8318da1d5fad2b47fc5747c049b90fe9255c28ad":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-07-04 17:39:28.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/16/922316/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/16/922316/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/922316/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/922316/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/922316/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/16/922316/3"}}},"commit":{"parents":[{"commit":"0d8e79b713bc31a78f0f4eac14ee594ca8520999","subject":"Merge \"Add file format detection to format_inspector\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0d8e79b713bc31a78f0f4eac14ee594ca8520999"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-06-19 12:25:52.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-07-04 17:39:28.000000000","tz":0},"subject":"Fix: optimized upload volume in Cinder store","message":"Fix: optimized upload volume in Cinder store\n\nWhen Glance is configured to use Cinder store and we upload\nvolume to Glance in the optimized path, it fails with\nInvalidLocation error.\nThis happens because Cinder is not aware about the store\nin which we will create the image and supplies the old\nformat URL i.e. cinder://\u003cvol-id\u003e whereas Glance expects\nnew location format i.e. cinder://\u003cstore-id\u003e/\u003cvol-id\u003e.\n\nGlance has code to update the format from old location format\nto new location format but it isn\u0027t triggered in case of\nold location APIs.\n\nThis patch adds the context to the update store ID request\nwhich calls the Cinder store to provide the updated location,\nhence fixing the optimized path for upload volume to image.\n\nCloses-Bug: #2054575\nChange-Id: Idd1cb8982b40b85a17821596f76dfa10207f6381\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8318da1d5fad2b47fc5747c049b90fe9255c28ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8318da1d5fad2b47fc5747c049b90fe9255c28ad"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
