)]}'
{"id":"openstack%2Fglance~922962","triplet_id":"openstack%2Fglance~master~Iaef62e4d00d92c65b3d2466863a5ff59d0f2bf5e","project":"openstack/glance","branch":"master","topic":"wrap_socket-py312","attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-06-27 18:29:05.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Iaef62e4d00d92c65b3d2466863a5ff59d0f2bf5e","subject":"Python3.12: do not use ssl.wrap_socket","status":"NEW","created":"2024-06-27 15:57:56.000000000","updated":"2024-08-30 08:40:06.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"9f28af70830c7ab96ec4faba40e10feaba76ac55","_number":922962,"virtual_id_number":922962,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-11 16:28:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-08-07 05:46:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-08-30 08:40:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-27 16:02:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-27 18:29:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-09 17:25:22.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2024-08-07 05:46:17.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-08-30 08:40:06.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"8458c18fba0d60307f9a46049bfc1c8ab8cc066d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-06-27 15:57:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22fefb2df9b3faeb5e6a2e0937e295d5e10b5a0d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-06-27 16:00:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"733fbcc65c06e7e65b17c4e074b5a6610900e5a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 16:02:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f8e4fa5302845d39cdbad8a18c8cbb9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f60110aae380419aaacba8e7c0dbf27a : FAILURE in 3m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7c71cf407060451fa1f3992d8f8847fc : FAILURE in 3m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"94cd6d2497595d3c692c7d2f2b56478cf397b58c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 18:29:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/267c28e774bd47819e44fba1d1fb7aa6\n\n- grenade https://zuul.opendev.org/t/openstack/build/c488de77bb5f4c569e45e103ae71b231 : SUCCESS in 55m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d9cf7deddd1c41539d541bc292af755a : SUCCESS in 1h 18m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15dae4c098c645ae97a4f8742f2ef8d1 : SUCCESS in 1h 07m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdecaf6d559d4f6ca0046d0131247880 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28545068085a4c6cb6f20561ce9a2e32 : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c07044f816a4411cbcf3c4e994c79c05 : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40844d01c0c94c33bf2df7d168776585 : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5293751155144decabbcb5a788a9d709 : SUCCESS in 5m 43s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b4d33032e1d54a90aa87b8ad35d68485 : RETRY_LIMIT in 1m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6fc9b26dafaa4666859894992acdac05 : SUCCESS in 7m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9687fbd0f02f4d18b379ab072f725fb4 : SUCCESS in 10m 07s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8d11e77710804dd09ff01253a81787e7 : SUCCESS in 12m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a141f236cd9f40a0b1fce06913c3c804 : SUCCESS in 12m 29s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d9a41656a53e457db0a26e03d10686ca : SUCCESS in 1h 44m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9d9b0fe171df4f1996bbeb8a90fd77f7 : FAILURE in 1h 44m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0155e4c9df264399a1b1a8c696c22943 : SUCCESS in 1h 49m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2f1ae03143a24798bf1c917047dd669c : TIMED_OUT in 2h 04m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b675cb6d3b1e4c99ab5ce40e60dda1fd : SUCCESS in 2h 13m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e99fe41f7af40979689122822009aea : SUCCESS in 52m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/afe5391cd230454ca4b627189337089f : SUCCESS in 1h 40m 47s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/5918072500cf48878d16a73914bc1085 : SUCCESS in 1h 46m 39s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5607d8a3de554eab9aa912904e2ab92d : SUCCESS in 1h 03m 01s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c2fd9e298ac242baa55f594493cea8ab : SUCCESS in 34m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ffef73a44eba3ffea74fe068bb811c64fbd5e65b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-09 17:25:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84ed321719b8b359af91ba845d8faae078b4920e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-09 17:35:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30a047c72a7d4c372911e3096ab7c786fa2ec288","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 17:39:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79553d6ca4be437597c9fc27692595be\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/377133c4ab2c4a95a26a0f54bf065efb : SUCCESS in 12m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/91598cf979c54804a7fa357fe8378bb3 : SUCCESS in 9m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"de1263ecfa3b0053a7b62e492106185a2964ca17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 19:50:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/648589319d1e4a58a38478a8996801b6\n\n- grenade https://zuul.opendev.org/t/openstack/build/076cacac40f44889aa559dbd79912842 : SUCCESS in 46m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/86ebd33750ef41b1bad4889002f2f941 : SUCCESS in 1h 21m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f5d34b8512e471ca8494c2f03523b99 : SUCCESS in 48m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a1018675da3458facbd09e2ade89d33 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8f6077872b34c429fd24e1967f595b9 : SUCCESS in 6m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/596a9cc77b9145419aa83e7993901e11 : SUCCESS in 6m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d6c5cdaec8f1431a91bddf0a43bd11e9 : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b904def4c44142c5a4eefed9f12938a3 : SUCCESS in 5m 16s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0ad5db8de12747ae809c545ebb4bf766 : RETRY_LIMIT in 1m 39s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cbb857d53798431598523350531b06a1 : SUCCESS in 8m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d2b01dfd8a29407a884251cc783dd036 : SUCCESS in 8m 13s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f908c0fcdb8e45f9848d023c931de4ef : SUCCESS in 11m 24s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6e13b1ea77b940b6a327028d34a81ef2 : SUCCESS in 9m 16s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d43df57d42af44d3b85c0f7e20efca76 : SUCCESS in 1h 38m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d821a352c6ea4530a1f66e7ea40cbca1 : SUCCESS in 1h 40m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4b4bb937849741afbc1899acbc25d15e : SUCCESS in 1h 52m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2fbf75bbeaa9400abfdd63f183cdb3f1 : SUCCESS in 1h 52m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bb3612cc126b4308969a8f7415367839 : SUCCESS in 2h 02m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7077d2a0e8b14d61a2cd61aa290e61ad : SUCCESS in 58m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c228394bc6c64ec091a147754f81c426 : SUCCESS in 1h 57m 15s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/a6d5ad4e0b784caa81e876de910c24bb : SUCCESS in 1h 57m 29s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f6fb68a9558b4784b96eeb5f5338b02f : SUCCESS in 1h 05m 16s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0fce13f7337e401480b53b59fd5f2638 : SUCCESS in 28m 02s","accounts_in_message":[],"_revision_number":1},{"id":"bb9b4dd57cd2b2b7ae9827c5f17dde1a4d2f5219","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-11 14:10:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5c5665014bc22711b42537daf75f04b28a72eef2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 14:15:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23008166e6fe4cc7b3aa9f0c0e274464\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7b8463b7fd3c44d2861b6180aff2123c : FAILURE in 3m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/946e0a95a1d043a7b53152e6055f5bbe : SUCCESS in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1fabaa7e9a07da615335364c1ac5bb2350cfa92e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 16:28:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b587f446950f4f2baeb0b9e61931abe9\n\n- grenade https://zuul.opendev.org/t/openstack/build/27a16299e48e4fc4910d99fc2422e15f : SUCCESS in 1h 04m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a4e7ae30dbb14580868c0bab8008ee32 : SUCCESS in 1h 09m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b0e17b91c0c40d49f227521847b6f01 : SUCCESS in 49m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbb08354ff0645f2b302daa32ad25ee3 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80a8e602d09e4e89b53ba1700d9c8e2e : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ad42d60f7a3545fca04d921a4df67413 : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/796fc3f6cd5545398acf3a5aed4faf24 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eda94276f164831bb72d285ced053b4 : SUCCESS in 7m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8392acc9452741fe9fb36cf925c088bb : SUCCESS in 8m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7086d1ea29d7429a900ee0ecd7367ec9 : SUCCESS in 8m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5ef3f16cbe4c4654af02fc74698f640e : SUCCESS in 14m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ccd4ff24564543d096f073dc7116b116 : SUCCESS in 7m 28s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ca575af977e442debd26f896f5cd1e4f : SUCCESS in 1h 55m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0fea326c1af9472ab6f7038d0eb2cf9f : SUCCESS in 1h 30m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3333176a5be544bcb2280a560be46dc7 : SUCCESS in 1h 31m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/47449fad373f4c2a9d00452a4441c276 : SUCCESS in 1h 58m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d56eceed890944e19c3bdad22ce0bb4c : SUCCESS in 2h 12m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0f1ab94a17549eba652f109abb8d296 : SUCCESS in 42m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4d69e75d6ee2440b9a04b8e4aae827c6 : SUCCESS in 1h 41m 12s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e357c96ede8342ebb3a2ddc0c035137b : SUCCESS in 1h 33m 03s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ed073cbe43e34303af4af11bed181bab : SUCCESS in 1h 06m 21s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a2e2796a2b894ab08917693f8f3550e9 : SUCCESS in 29m 18s","accounts_in_message":[],"_revision_number":2},{"id":"d42d2fba3997614eea68723c188adc0921105778","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-07 05:46:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9f28af70830c7ab96ec4faba40e10feaba76ac55","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-08-30 08:40:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"de86377b3874fac929deabf3e803acab68674c29","revisions":{"1c5974f44ad66e6eeda1ef421cf328867a284a0a":{"kind":"REWORK","_number":1,"created":"2024-06-27 15:57:56.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/62/922962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/922962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/922962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/922962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/922962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/922962/1"}}},"commit":{"parents":[{"commit":"2e6afec7a315586a1c3e1ca274c1911eae6c7e28","subject":"Merge \"Drop sqlparse\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e6afec7a315586a1c3e1ca274c1911eae6c7e28"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-06-27 15:56:19.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-06-27 15:56:19.000000000","tz":120},"subject":"Python3.12: do not use ssl.wrap_socket","message":"Python3.12: do not use ssl.wrap_socket\n\nThe ssl.wrap_socket method has been removed in 3.12.\nSSLContext.wrap_socket should now be used.\n\nChange-Id: Iaef62e4d00d92c65b3d2466863a5ff59d0f2bf5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1c5974f44ad66e6eeda1ef421cf328867a284a0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1c5974f44ad66e6eeda1ef421cf328867a284a0a"}]},"branch":"refs/heads/master"},"de86377b3874fac929deabf3e803acab68674c29":{"kind":"REWORK","_number":2,"created":"2024-07-11 14:10:32.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/62/922962/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/922962/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/922962/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/922962/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/922962/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/922962/2"}}},"commit":{"parents":[{"commit":"7cfa378b2f75b31c30456d6cdd64d3a26687150d","subject":"Merge \"Release notes for Dalmatian Milestone 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7cfa378b2f75b31c30456d6cdd64d3a26687150d"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-06-27 15:56:19.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-07-11 14:10:07.000000000","tz":120},"subject":"Python3.12: do not use ssl.wrap_socket","message":"Python3.12: do not use ssl.wrap_socket\n\nThe ssl.wrap_socket method has been removed in 3.12.\nSSLContext.wrap_socket should now be used.\n\nChange-Id: Iaef62e4d00d92c65b3d2466863a5ff59d0f2bf5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de86377b3874fac929deabf3e803acab68674c29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de86377b3874fac929deabf3e803acab68674c29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
