)]}'
{"id":"openstack%2Fglance~923108","triplet_id":"openstack%2Fglance~master~Ib06df932a9c2f0d9e98b9e923c4e1424e67fb994","project":"openstack/glance","branch":"master","topic":"Calculate-Hash-Api","attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-06-30 15:32:30.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib06df932a9c2f0d9e98b9e923c4e1424e67fb994","subject":"Add new calculate hash api","status":"NEW","created":"2024-06-30 13:24:49.000000000","updated":"2026-02-04 14:38:43.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":12,"unresolved_comment_count":12,"has_review_started":true,"meta_rev_id":"471f25560f2ed2dabae78d5fafe4f0cbd3c0ff48","_number":923108,"virtual_id_number":923108,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-04 09:17:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":-1,"date":"2026-02-04 14:38:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-30 13:31:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-30 15:32:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-04 14:38:43.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"6b6f652484dd5cce4bb1e1b63e31ff683c7b53f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-30 13:24:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"058b1a106a73b8c82398456501f92a33f6a1016c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-30 13:26:13.000000000","message":"Topic New-Location-Apis removed","accounts_in_message":[],"_revision_number":1},{"id":"a5f22105b5359da0b3214fd48f5bd7d6fe95a44a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-30 13:26:28.000000000","message":"Topic set to Calculate-Hash-Api","accounts_in_message":[],"_revision_number":1},{"id":"a1622e0ce160e42d8b193c52b3e0e4327d1115e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-30 13:31:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb1b3cf9303443a083717a2ab6508ad9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9a65ac337b334d07bc55b2a570c63d24 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/46557c79a2394672981feffebf03e2dd : SUCCESS in 5m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ffd95877d43163ac321e6114d333ac81e70ffd5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-30 15:32:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec953238ef19437282be73a547cc8bec\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/78639535c51f41b78bb1755d2d1f93c7 : SUCCESS in 2m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/14544382f4394c0f81f01496fbaa8dd6 : SUCCESS in 44m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/46d59d2716f043c4a2c90b0542ef010a : SUCCESS in 1h 30m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00371af549fd4dbdb5d936071cfc0515 : SUCCESS in 1h 04m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/677637ab03404d8bbd2b38932be0afed : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2250f9ed80bf4ff99c3a3bedb32fa66d : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d94f21b11fe5400abf48c6def4c0f53d : SUCCESS in 5m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d3883b77fb7f4591a8336a9694054de6 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/353e7e657b8c4072a8f2c9f0a52548d8 : FAILURE in 4m 59s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/bfb73ae71cab40d2bb26fbcad04f6bbe : RETRY_LIMIT in 1m 55s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/36ccdc5e26264d78969558bda23ab104 : SUCCESS in 8m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/71cf0cc4636c42cd96fa118d6b125e8a : SUCCESS in 8m 37s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/548e5ab5801145449f86ac7b0e726fa2 : SUCCESS in 12m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3792aa8ce98c4ee8b7b0b64a39ff2661 : SUCCESS in 8m 49s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4af4d7b0972b45c79d4ff84582c4fc04 : SUCCESS in 1h 47m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7417f46adc664e93a40e9ea96a7b7537 : SUCCESS in 1h 46m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/081cf5039d7f437082b612183ade6e46 : SUCCESS in 2h 03m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5abba9061ffd4815bba876ed6710157e : SUCCESS in 1h 30m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/98d4453defed43b9b4f50e7c68e63ff1 : SUCCESS in 1h 35m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5bbdbc822974d5e84f26003301880c5 : SUCCESS in 42m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d6a31e0c17de48ea89570dc0d50e3efe : SUCCESS in 1h 18m 52s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/36443afe560f4c09b95cfe1b4acc2b6e : SUCCESS in 1h 24m 14s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ffab0dc5abaa4e78946064d58d834533 : SUCCESS in 52m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2403259435f745ea8a3aa0823394e396 : SUCCESS in 20m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ed9a7859207a11c1975afdb53d47951fb6fd6b9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2025-01-21 10:07:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e1ab63a46553dede6ed74497d2640fd3e139fef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 12:55:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75f0252f0bb54a0886180b8531830127\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4cab923cafe7493088c1dfccd5022e85 : SUCCESS in 1m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/113a7bc6f678406fa8ace3ca94d3663e : SUCCESS in 1h 04m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4052fdb1af4b4405a73fd7bc620c002c : SUCCESS in 58m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7fb3735f2edb4d9abe4b496c28e317c6 : SUCCESS in 1h 37m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58b5289b222244fbbe3c6811b25babe8 : SUCCESS in 1h 05m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bda4e2b63a54791a37125ef09787707 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6366c2b41a0545a2955b774faccc6ccf : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f7608d07c70d44c5a9852d8b6373c465 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c97c50ebd9824831bad9e8e5d8ed42c8 : SUCCESS in 6m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6378977430c2493fbcdb2fe55d0423e8 : SUCCESS in 8m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f0f0b322a7c646bd842638ef391a01cd : SUCCESS in 8m 18s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5b4f6e6b349c47c3b013ec37ae8ffb1d : SUCCESS in 10m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/022c1d69774d44de90e1a06b524c8166 : SUCCESS in 1h 35m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ce262b78a578478cb992f8e167593907 : SUCCESS in 2h 35m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c67eeced0a5246ae88ac795683f878f3 : SUCCESS in 2h 38m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bf773dd30fdc4e868ed2f55ed21c6eae : SUCCESS in 2h 03m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b3e6411e501144399a2e78909a0069b8 : SUCCESS in 2h 23m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6e81e91b80d48d8afe426bfebc6655c : SUCCESS in 1h 08m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1cee07d898a149839dbeddcfeb49b806 : SUCCESS in 1h 44m 45s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/a44c47d8c0144e778ac9932c3addf852 : FAILURE in 2h 05m 01s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/da80560eb66b432ebdb072fa3deeeda7 : SUCCESS in 1h 01m 29s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/8f78eb04df694410a63487924e5aa698 : SUCCESS in 2h 12m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/909d7e75fd784c79a28efb2e91d69526 : SUCCESS in 40m 49s","accounts_in_message":[],"_revision_number":2},{"id":"f1832ea51075a9f6e9200077f5742fb0a0d3f168","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 13:11:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1356a21d3d1e4ee092803fa57e980477\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8d7fdfc4bca480bb1869b69866e154c : SUCCESS in 17m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e9b45c9fe09c4ebb93c26b0cb6ccf727 : SUCCESS in 19m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0accbe3115e7564b7b0edf31533cb6fb0d8b0e9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-02-04 07:02:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4af5b1a462a9d51f9020a1ce8c297f40603b1265","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 07:17:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d29ade6d7f9b432bbe3908ddf8a5461e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/185e893b5651456faa457f91dc252102 : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c2789743e0404574884cd2d27cd729ab : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7b4207f82e1347bab7fec3039ac67613 : RETRY_LIMIT in 3m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c2ba8b8744a8b23204c90ee953126fa72a820bc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 09:17:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c5d7895869a465ea59f286b108f1399\n\n- grenade https://zuul.opendev.org/t/openstack/build/dd92efbaafac49d68986747dd2b88661 : SUCCESS in 31m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1314144965ae42098318d30cc919c0c8 : SUCCESS in 41m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2fc55ca497b74cf4b9299508ec5b4e16 : SUCCESS in 1h 27m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2786d498f3ed40fb969b8ad261fdd0ad : SUCCESS in 43m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8c9f8e95c394977b69bb0a9b8cf2609 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2e7a660e5924498a6a19832ffc785b0 : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/00539be7e674414b890c8c545dfe6491 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3dee19c995541e69bacc3ad2b178dd7 : SUCCESS in 3m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5a8d34ebfab14188a24ea7e12c141625 : FAILURE in 4m 42s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/37ae8df0c62e4030aaa0c5db506b1877 : FAILURE in 7m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/19c594d8dbbe411f8982faaefcb5b4ec : SUCCESS in 1h 52m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/95b827e394624d79b0b944315eb03ae0 : SUCCESS in 1h 56m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/55a7aa884e6b43718339701474d959be : SUCCESS in 1h 49m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5ae2e305b5094ee3b07d78aba3570f7b : SUCCESS in 2h 08m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43731aceb5bb4ca69e9f67f77c85d594 : SUCCESS in 56m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3d05b3d960db4c39b4d605adcdc3951a : SUCCESS in 1h 24m 54s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c6e6534266c040a0a6fdcdc06ac92f53 : FAILURE in 15m 21s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/98f82e31bce04e0bb415199915070a5f : SUCCESS in 2h 03m 14s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3c5e3fc5aa4b4ddd99430e82d9d6a676 : SUCCESS in 36m 13s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/776ca5c5f6944f408bf630553391278f : SUCCESS in 32m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"471f25560f2ed2dabae78d5fafe4f0cbd3c0ff48","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-02-04 14:38:43.000000000","message":"Patch Set 3: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6e14206a9e88e2d443fe351e7ce3976a11a23916","revisions":{"4755098e98bbc213a78f9b4ebb609ab2d693d372":{"kind":"REWORK","_number":1,"created":"2024-06-30 13:24:49.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/08/923108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/923108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/923108/1"}}},"commit":{"parents":[{"commit":"5cc1c826fea1252fb3076db241e4785a3ee90132","subject":"Bump Images API version to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5cc1c826fea1252fb3076db241e4785a3ee90132"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-30 13:09:49.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-30 13:09:49.000000000","tz":0},"subject":"[WIP] Add new caclculate hash api","message":"[WIP] Add new caclculate hash api\n\nThis change adds a new calculate-hash api POST\n/v2/images/{image_id}/hash to calculate the checksum and\nos_hash_value for the image which is in `active` and with\nno checksum and hash_value set.\nThis operation will be admin only, and validated by the\nnew policy rule `caclculate_hash`.\n\nImplements: blueprint new-calculate-hash-api\nChange-Id: Ib06df932a9c2f0d9e98b9e923c4e1424e67fb994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4755098e98bbc213a78f9b4ebb609ab2d693d372"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4755098e98bbc213a78f9b4ebb609ab2d693d372"}]},"branch":"refs/heads/master"},"22b5ccdd577079169a0b052670be537bcc3f10c1":{"kind":"REWORK","_number":2,"created":"2025-01-21 10:07:28.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/08/923108/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/923108/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923108/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923108/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923108/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/923108/2"}}},"commit":{"parents":[{"commit":"afaa1900283fb7f09e401b2faff16289aeeb687c","subject":"Merge \"Bump minimum version of glance_store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/afaa1900283fb7f09e401b2faff16289aeeb687c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-30 13:09:49.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2025-01-21 10:04:57.000000000","tz":0},"subject":"[WIP] Add new calculate hash api","message":"[WIP] Add new calculate hash api\n\nThis change adds a new calculate-hash api POST\n/v2/images/{image_id}/hash to calculate the checksum and\nos_hash_value for the image which is in `active` and with\nno checksum and hash_value set.\nThis operation will be admin only, and validated by the\nnew policy rule `calculate_hash`.\n\nNOTE: Unit and functional tests are pending.\n\nImplements: blueprint new-calculate-hash-api\nChange-Id: Ib06df932a9c2f0d9e98b9e923c4e1424e67fb994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/22b5ccdd577079169a0b052670be537bcc3f10c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/22b5ccdd577079169a0b052670be537bcc3f10c1"}]},"branch":"refs/heads/master"},"6e14206a9e88e2d443fe351e7ce3976a11a23916":{"kind":"REWORK","_number":3,"created":"2026-02-04 07:02:04.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/08/923108/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/923108/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923108/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923108/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923108/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/923108/3"}}},"commit":{"parents":[{"commit":"ff1a8dd978334389915f472631a42415e9059d2e","subject":"Merge \"Raised NotImplementedError instead of returning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ff1a8dd978334389915f472631a42415e9059d2e"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-30 13:09:49.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2026-02-04 07:01:57.000000000","tz":0},"subject":"Add new calculate hash api","message":"Add new calculate hash api\n\nThis change adds a new calculate-hash api POST\n/v2/images/{image_id}/hash to calculate the checksum and\nos_hash_value for the image which is in `active` and with\nno checksum and hash_value set.\nThis operation will be admin only, and validated by the\nnew policy rule `calculate_hash`.\n\nImplements: blueprint new-calculate-hash-api\nChange-Id: Ib06df932a9c2f0d9e98b9e923c4e1424e67fb994\nSigned-off-by: PranaliD \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6e14206a9e88e2d443fe351e7ce3976a11a23916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6e14206a9e88e2d443fe351e7ce3976a11a23916"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
