)]}'
{"id":"openstack%2Fglance~923197","triplet_id":"openstack%2Fglance~master~I15d388793f12ad8305095b0993a4e70d1bea6851","project":"openstack/glance","branch":"master","topic":"metadata-encryption-key","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-08-01 17:42:46.000000000","reason":"Change was submitted"},"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2024-08-01 17:42:46.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-02 09:56:41.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-07-09 15:19:59.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I15d388793f12ad8305095b0993a4e70d1bea6851","subject":"Deprecate the \"metadata_encryption_key\" option","status":"MERGED","created":"2024-07-02 08:19:18.000000000","updated":"2024-08-01 17:44:02.000000000","submitted":"2024-08-01 17:42:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923197-metadata-encryption-key","meta_rev_id":"cca913e429f0a1ca3afd4f400bac49ae1379ce97","_number":923197,"virtual_id_number":923197,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-01 17:42:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-02 09:56:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-08-01 14:50:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"date":"2024-07-02 18:16:09.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-08-01 14:50:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 08:25:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 08:46:36.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"},{"updated":"2024-07-02 08:46:36.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-07-02 08:46:36.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-07-02 12:01:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b73f60ae3d20ed62906ca76a41d12ddf72f88eac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-02 08:19:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7728f3905d67d6bf1946d106f071f084f702891e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 08:25:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c31b327ce5247a88749298b0d540258\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/faf681e5745c49148c1b65131e40b02f : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4a05bab5113548349096f2be80b0d82f : SUCCESS in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"213679c6cbb90a67931e7b4e869bff69f1d4c9d4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 09:04:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"95059e848415d7342451d0befe5b880e5ccf594d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-02 09:50:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"32d0670c91c7b631e86c809aa87062ea67c16635","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-02 09:51:45.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3b92f7b9f43d929da453a0a0c03f7a1601209e33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 09:56:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/674b673b653349698696a068444b6c51\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b1f2476a0b948a1993c45febd22810d : FAILURE in 3m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4a1b3e2738624845b86d47838423c8c6 : SUCCESS in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"793778e271850e930254b8e542e2c6139a46af56","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 09:56:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f949ec54d27e8e1166a564f86acded791f42142d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 12:01:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09500aa10b934d928fc78dfe84ad0118\n\n- grenade https://zuul.opendev.org/t/openstack/build/d19d854678e8413a894202e8a5375d89 : SUCCESS in 57m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4dd86a0297584096bd58f97567e7e0b1 : SUCCESS in 1h 27m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3cdadb56ea64c1f8df58bf6ee72f755 : SUCCESS in 58m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b8fd45ad25e44e0a2ad6aad92e66d5a : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b827d2eb4d145a686e0cbe2da20e2ba : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30bf85dd64e74a4fbb032e084230ca3c : SUCCESS in 6m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/173bd0d01f68459480dd53bd74ad18a2 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba9bca1560db4dc48c628589ddafd334 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e7305be02294ed696ae66c365413203 : SUCCESS in 6m 30s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/75a403b25d914c4489d922ef3a923786 : RETRY_LIMIT in 1m 32s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0ccbe9743b664e2385e8714a4862d3e5 : SUCCESS in 8m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c25a5ec53a154b708e92d4ce8063ac46 : SUCCESS in 8m 33s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f728601d3714416ab5ff438338454e28 : SUCCESS in 11m 38s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/04ba155fffcb4a1585b4d2a4841cf88b : SUCCESS in 7m 40s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dd88189f99574c02a303a8604a32946e : SUCCESS in 1h 46m 05s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c08534112c7e4c0993e2b962355fe3d7 : TIMED_OUT in 2h 04m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0e62dd00e1e341a49f1566907332ffd5 : TIMED_OUT in 2h 04m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5c9bb9b6894a44b081e160de8f2a8f89 : SUCCESS in 1h 52m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/89f821a6643d42758545253fb101d9ce : SUCCESS in 2h 03m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/148303349d98406a946d661796017f34 : SUCCESS in 50m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ef193446a1d340719fcfb8885e6f25c8 : SUCCESS in 1h 46m 37s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/bdbfe931d25e46d58c07d6460bec8b59 : TIMED_OUT in 2h 04m 16s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/233d205ce532489ba45557e7f99bc6d8 : SUCCESS in 1h 09m 00s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0a7ab7aa59cf48d89336a862660a63ce : SUCCESS in 31m 51s","accounts_in_message":[],"_revision_number":2},{"id":"f20befda0c781074d9198aa4711229c5f2023ace","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-02 13:33:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0942948baf2a3f5c6bc97212df4f2958348ecb7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 13:40:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb2f8e19a70a46209535e5c484066d4c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/90d74cc8ff8d454d8318c8cfd137832e : FAILURE in 2m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f8b7a54b9ab34616a611eb90f41b3303 : FAILURE in 3m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3b31036f9c65e0cb0181ed7bf00cd1372d0ba5a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 15:55:53.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc2375ef801d4c63a349cb3c523ddcd4\n\n- grenade https://zuul.opendev.org/t/openstack/build/fd4c17b95ced49909c796745351a764d : SUCCESS in 1h 03m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c132f570a190429594f5cacda1332797 : SUCCESS in 1h 32m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf29c914971d413b9a0364bb8a1a5ff8 : SUCCESS in 57m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb0fb39c93794b86b48f58c47c7ab4c2 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a3ee89830a0492bba4d6fa9a8ee31cf : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b1d9637f9ed84ad091383ad48fbc1313 : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9a11a1c0fb74c879429df4d8bbf7b92 : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2fd8caf95f6444d82b1c05d19f100e5 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40944af1e4d447fd8a67f64198c8aa1e : SUCCESS in 6m 45s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/19591df267e4409e9e568e593d8cf255 : RETRY_LIMIT in 1m 36s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/35c76e3109bb4cbe91b726fc1f28b73e : SUCCESS in 10m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c9861025fef94650bb5c18dc1d5ff37f : SUCCESS in 10m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2936d51744bc4506a51f6b2ec4a5c92b : SUCCESS in 15m 38s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a6f95d483b554f12b46a3acd5e223c44 : SUCCESS in 9m 19s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/06cbeafa961440379f0d37647ded4898 : SUCCESS in 1h 55m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/93632619dbf845f88562967559a71128 : TIMED_OUT in 2h 04m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ff72de974ab9471ea30079071cc71055 : SUCCESS in 2h 01m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/149228d042c6478ebef8e5b45199d23a : TIMED_OUT in 2h 04m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/28d16ac3b7514eb5a8bb67b8dae7b1f3 : SUCCESS in 2h 15m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c378eb74c2124e359dd0248227f46dd4 : SUCCESS in 55m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0a8d28e5a80e4703989713b32d188682 : SUCCESS in 1h 53m 37s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/74591cecc52f4a4dada4d4e741958455 : SUCCESS in 1h 55m 22s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/6214610c929c4f5f8aa6600a376d19dc : SUCCESS in 1h 15m 31s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/033def84d9ea44b6b2f3039ee358b296 : SUCCESS in 32m 40s","accounts_in_message":[],"_revision_number":2},{"id":"d4f41b5687f9688b03f0a5f98d38711d88a04b4c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 16:55:18.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c66e980e2eee5efa1b3f3baa86c70f33920fff68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 16:59:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29c036a4a26247f585b93735dbb19eb6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f56ae75c202e4c409bc7cc45e5ce797b : FAILURE in 2m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0cb0eae6c3384a5c941ef2ea5bef6cec : FAILURE in 3m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"06846ecf89929917894c78b91ed0792bf6cc7349","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 18:16:09.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"00bc76ff61e0725fbed824bbbbdc92834345fb43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 19:53:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c2d05f9ecdc42eeb7191bc0dbe15088\n\n- grenade https://zuul.opendev.org/t/openstack/build/3f82b517965f4095a5f0a20036c942e6 : SUCCESS in 59m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/38541d6932f74bc38983f43520413cad : SUCCESS in 1h 33m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4db47d2ae4814f55ae22cd964a5e2f29 : SUCCESS in 1h 04m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d2a8f9c7c7a469795824007e756736d : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aab785cc338543e4a6d52b17149c76e5 : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1203f19457ab4ae3a0cdc11491991eb7 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6bba9551967e4440a5aec8d32a755b7e : SUCCESS in 12m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5c113ae4480447995f07f064a7f1108 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a52982c9f824d7e861684372ae175b0 : SUCCESS in 7m 12s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d8da02df56164ab98305bf3a6abbabac : RETRY_LIMIT in 2m 14s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8a38f1cb3bf14e22a15deec438bb826b : SUCCESS in 8m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/27533a2e39df4dbea9d1d4ba5d0f2291 : SUCCESS in 7m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ab45b98d676d41208a9a3b432f72cdce : SUCCESS in 12m 35s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3e32cdb8ce3f46eab03d21fe37d1c8f7 : SUCCESS in 8m 43s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d77f2ea1ca5440158bd723ffd88c7bcc : TIMED_OUT in 2h 03m 55s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9d5b3267049446c8aa233976f186ca1a : SUCCESS in 2h 02m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ff1fd8e97e4145f4ba1dcb26f090e7ed : TIMED_OUT in 2h 04m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/82366bc36b204390b43957c92a36a3b9 : TIMED_OUT in 2h 04m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ce0be859bf2d461292888ea9f2fb87a7 : SUCCESS in 2h 05m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c4e2cfa6a39437284cda7d09d0e7c70 : SUCCESS in 50m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/20b70f4dc3bf4824adb75b72a686fc5e : SUCCESS in 1h 45m 36s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/aa8d03000acf4418827e089e57b31be1 : SUCCESS in 1h 40m 24s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8f225bf82c434fffb10955c8812ac7af : SUCCESS in 1h 01m 05s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2e6c90fe6b3c4ce58776430ee51c9739 : SUCCESS in 24m 48s","accounts_in_message":[],"_revision_number":2},{"id":"3e1525fc5662d4c3725dff211b9e0dd876567ff6","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-09 15:19:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e314aeff7b98659dc6dee281f832e799160d64bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 15:24:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b467b98280a5459fb2f0238bd20b8842\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4486fb04a24f499bb18d586c2afebec4 : FAILURE in 3m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f4ce43faba164aa08e26e50ec9e95a62 : FAILURE in 2m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e3ead72c62645e69baa3b0f7b806b9bc5ea08b38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 17:28:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eb7be697eff4e69aa7463a6941909ba\n\n- grenade https://zuul.opendev.org/t/openstack/build/92b230c1ac1d40d796d9675bf2b64bcf : SUCCESS in 1h 02m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3822237197f44f2ba2299b99bb0e4c92 : SUCCESS in 1h 11m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b66db4b2a46a498aad8a26eb4abb6df8 : SUCCESS in 50m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de510ca7dc3849c6a9441f622cf34fb0 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fac6d3b73a04d16a4abd5d741926523 : SUCCESS in 5m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/58f684bd412f4faa9f996560feb20870 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/90ec9feb243d407bb0a1e3281d196f0d : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15da9a919287478c90b1a041cae71cea : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1da25fddd3ed47b38a134d840521a493 : SUCCESS in 6m 06s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d3d2451daaf040da819e45627f7373f8 : RETRY_LIMIT in 1m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6481c54cef5e4c3fac541fe8bccdc02c : SUCCESS in 7m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2654ef6d8f354340b5b33b56948199da : SUCCESS in 10m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/28557b7256cb4d9eaec0d4745df9689c : SUCCESS in 14m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d2cb9f16dd874c35a4242344e2326ae5 : SUCCESS in 11m 39s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0a8554f582bc49789530f8c149ec80ec : SUCCESS in 1h 42m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f61928c42a1b49e48b97abfcae67335a : SUCCESS in 1h 59m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3a32665dea004071a729a5585668d378 : SUCCESS in 2h 00m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3363667c2e284c59a29f3f45ac80817b : SUCCESS in 2h 02m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/79698e8e88c44557ab51ab5a5e7f1bdd : SUCCESS in 2h 06m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81b9b30610d849feb676397e292d2068 : SUCCESS in 55m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c9d22e9b160a4ecda05dc0efcbccf34f : SUCCESS in 1h 41m 11s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/99a93dc995224ee88460ec921cc41472 : SUCCESS in 2h 02m 17s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9f0b6d20d899433dace59a36fd4f8951 : SUCCESS in 1h 05m 17s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/029d1607f4474cfdb5f78045d3ad33ee : SUCCESS in 33m 29s","accounts_in_message":[],"_revision_number":2},{"id":"fdf0f3050758124353790e6951fb90647e847732","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-08-01 14:50:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3800f211092d55e636ea494e838fa7b198e67861","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 14:51:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2b0ed4a7b09361c2bf0ae8168cbda720b172ba84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 17:42:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/121884d37bf94393b98846857c426124\n\n- grenade https://zuul.opendev.org/t/openstack/build/6b0a2962ab2249d78436069eb9a76444 : SUCCESS in 1h 04m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8cacb30ac6af4203ab43df90fe5da1cf : SUCCESS in 1h 25m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9eba0744067494e998c70d123157e19 : SUCCESS in 54m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/447b1ab485484268b7c9299b5580e7b9 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7d9cdef17dc42ec88a785df0cc2686e : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c1e8f6ea5094ca38308799335ce655b : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4df9bd4268b94a84a91648f711ac794f : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82fd80d947f54b649883f9fbc726693a : SUCCESS in 6m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9e7f54c8d9ae44349d33809088323f58 : SUCCESS in 8m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8dd7fc6762a84fbc9677678544b7d73e : SUCCESS in 7m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/002617dd6cae496a96818e6a319fb229 : SUCCESS in 14m 27s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5efa9c1a375041e5a4fd09830fae39c0 : SUCCESS in 1h 43m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/36dc58f8d24340c7be1452b39998c40b : SUCCESS in 1h 45m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6c7e3c53d9554d4995135c90d4db17d3 : SUCCESS in 1h 58m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/716035e34bce411ca187fe814fa0609e : SUCCESS in 57m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e490728fdb074744b1e83d9e6b2ea9a8 : SUCCESS in 1h 39m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/72869c36c42b4afb95f2a0ff37003f76 : SUCCESS in 39m 29s","accounts_in_message":[],"_revision_number":2},{"id":"ed379b4007e353daa115ec0ac0003e9760326b1a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 17:42:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cca913e429f0a1ca3afd4f400bac49ae1379ce97","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 17:44:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dba94678e42f42a994d135e21e431eca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef2f627c1bf3488191aac88f46d2a343 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f524561437cb4312b8eba4c850c8cff8 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"006ef41c0b0f8d23ff1a65863c83c06072ebfa2f","revisions":{"a365669ad9c62fc7e1c0883f4cfc7257047a11c8":{"kind":"REWORK","_number":1,"created":"2024-07-02 08:19:18.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/97/923197/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/923197/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/923197/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/923197/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/923197/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/923197/1"}}},"commit":{"parents":[{"commit":"4dec531b0aea15e8d970fa775a6e90068ddd5b03","subject":"Merge \"Use constants for common rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4dec531b0aea15e8d970fa775a6e90068ddd5b03"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-02 08:16:16.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-02 08:16:16.000000000","tz":0},"subject":"Deprecate the \"metadata_encryption_key\" option","message":"Deprecate the \"metadata_encryption_key\" option\n\nChange-Id: I15d388793f12ad8305095b0993a4e70d1bea6851\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a365669ad9c62fc7e1c0883f4cfc7257047a11c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a365669ad9c62fc7e1c0883f4cfc7257047a11c8"}]},"branch":"refs/heads/master"},"006ef41c0b0f8d23ff1a65863c83c06072ebfa2f":{"kind":"REWORK","_number":2,"created":"2024-07-02 09:50:56.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/97/923197/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/923197/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/923197/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/923197/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/923197/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/923197/2"}}},"commit":{"parents":[{"commit":"4dec531b0aea15e8d970fa775a6e90068ddd5b03","subject":"Merge \"Use constants for common rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4dec531b0aea15e8d970fa775a6e90068ddd5b03"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-02 08:16:16.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-02 09:46:13.000000000","tz":0},"subject":"Deprecate the \"metadata_encryption_key\" option","message":"Deprecate the \"metadata_encryption_key\" option\n\nChange-Id: I15d388793f12ad8305095b0993a4e70d1bea6851\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/006ef41c0b0f8d23ff1a65863c83c06072ebfa2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/006ef41c0b0f8d23ff1a65863c83c06072ebfa2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
