)]}'
{"id":"openstack%2Fglance~923248","triplet_id":"openstack%2Fglance~master~I6326a3e85c1ba4cb1da944a4323769f2399ed2c1","project":"openstack/glance","branch":"master","topic":"bug/2059809","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-04 09:21:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6326a3e85c1ba4cb1da944a4323769f2399ed2c1","subject":"Reject qcow files with data-file attributes","status":"MERGED","created":"2024-07-02 14:09:02.000000000","updated":"2024-07-04 09:22:23.000000000","submitted":"2024-07-04 09:21:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"923248-bug/2059809","meta_rev_id":"426484781c195013ef3d80ce35084186b9d7d477","_number":923248,"virtual_id_number":923248,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-04 09:21:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-03 05:58:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":-1,"date":"2024-07-03 05:55:53.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-03 05:58:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 14:09:02.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 14:21:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 16:06:37.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-07-02 19:45:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47e1b2baede3308897f323086f42c31d6b6f7423","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 14:09:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a416de14d8da1dcdd9b1dcf4de509e130578223","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 14:21:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3b4a341b33b43c595916fd0e78aef62\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/34668d202a9b4c7ebbe6a38e5c8149ee : SUCCESS in 10m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/743f34a1cef4433c8d4040f813fe05ff : SUCCESS in 9m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c05127cb6a1c36342e507da7041cfb53570cac6","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-02 16:06:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d4f95d6346880a58d44b997a7f83c90ff0c5e868","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 16:19:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c33945d22b8fc67c832fd4636b84982c6237fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 19:45:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/82aef21cd63c47958f27c190f03bdf9c\n\n- grenade https://zuul.opendev.org/t/openstack/build/b74d12a7a5784cd9a2abbd9138f1723e : SUCCESS in 1h 03m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ebab7969c4b84c36a977946c890eef20 : SUCCESS in 1h 25m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7703323a125642f3b1793c8344aa632f : SUCCESS in 1h 04m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beca6f41a16c41f4b4d8ce89d522c873 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/875416192c9048daaabee7fa2676a8e2 : SUCCESS in 7m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40299b51ba3d4a4b9faba8ac0c4ba467 : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/121453db367e46aebac6473439e71df3 : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d19e5d945854e4fbb7dc99c1f7c2202 : SUCCESS in 5m 42s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/369e7aa09fdf408e87d4736918e6d6dd : RETRY_LIMIT in 1m 55s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3bcd27c309a14faa8c7c5d736f4d1c47 : SUCCESS in 10m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/19e8f916adbf42f084f3da95c3757232 : SUCCESS in 7m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b68a2e50a58942c993daec969ca2c479 : SUCCESS in 13m 11s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2385569cc5404d8b820ef023522492ad : SUCCESS in 8m 32s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/95b5ed05ef1b4da4a6f585ea809222a2 : SUCCESS in 1h 49m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/230ca4dfa50c4c82ac0e14cfc35f385a : TIMED_OUT in 2h 05m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/33b5fbb4f351486ca650662712961727 : SUCCESS in 2h 02m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/51a678c799e7417b834e69c19d2b57d0 : SUCCESS in 1h 57m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cdc892fb1f2949e7903d0a20703934be : SUCCESS in 2h 15m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c232e81cda849d2820393c17ec80b9b : SUCCESS in 1h 04m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/25c00b515d8240aa9ca2c6d0eb71d9c3 : SUCCESS in 1h 40m 39s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/c3a9f5d98308426a820bdeb896d30880 : SUCCESS in 1h 55m 04s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/720c2e57d0d74cb18bedf6175366c2b0 : SUCCESS in 59m 42s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/de75d14ef3c348f2aa83b74937641706 : SUCCESS in 29m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c88d08b412d27af6753dfdb8623216192a431c2f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 01:42:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0255e116aaa4a1376f0453bc3f9e561c9505380","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 01:55:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cadc72f010054fc0b78206e60abb953a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23782cad9e054084825d33c0db4f5e3a : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d0c50c0432414ca4b6d37a133ec39398 : SUCCESS in 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7de931a12f2ac3487ea54bf9c292fe34cf6416a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 04:42:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de31fa8974424215a4f21cf26705a38d\n\n- grenade https://zuul.opendev.org/t/openstack/build/0c9ff831054f4c0f90cb465428184744 : SUCCESS in 1h 13m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3ae4b14d825d4d8d92ed9bcb2fd4c11d : SUCCESS in 1h 34m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c60812aa884249139681ee96bf2df833 : FAILURE in 59m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8a66dbfe741492aa21258484dbf8b9c : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/350b65f7e525406ba129fb52a8964b30 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc241fdbef494d5d898f2c564053a40c : SUCCESS in 5m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d5246bf2f674c9d94022ad58726d784 : SUCCESS in 9m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74b322acdaa448e28f2158bb3af87875 : SUCCESS in 5m 28s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/304f012633454bbdad01e29e1ffc6ee7 : RETRY_LIMIT in 1m 37s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/583e948a543e407fb1be86b7a51e73cc : SUCCESS in 8m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1cf7fc414b634eef819a9eeacd4615b4 : SUCCESS in 9m 09s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/170a34046f9b4f82bfb6c230ab1af849 : SUCCESS in 13m 27s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d34a936d77d84849b14bfd48a72340d6 : SUCCESS in 8m 10s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5ffc166963fa4546aab2c719fd6ff9dc : TIMED_OUT in 2h 03m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e94a0ecd8fee49fd8f811671298ea308 : SUCCESS in 1h 49m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2e05027bf04f45d794c9ddbb42d0d42b : SUCCESS in 1h 52m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1a8365a671484b14a8641d60c8366d96 : SUCCESS in 1h 52m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2dc67a73d0c94de2ba0f3bd3b3f94de2 : SUCCESS in 2h 29m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26ec8b5dbee4410684baf76b97f4cd62 : SUCCESS in 55m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8f771e2fc2814905969ba4af8ae8f36b : SUCCESS in 1h 56m 46s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/559bdae5a5a34ee78dda3458bf48c54b : SUCCESS in 1h 49m 38s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4744d984ed504c599c1e8e34c60e35cd : SUCCESS in 1h 21m 11s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/710e24258f3e46a8b806276477545792 : SUCCESS in 25m 49s","accounts_in_message":[],"_revision_number":1},{"id":"7037319b133f01705639efe70444238411720eab","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 05:39:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b1e777ad67ad70c059df3f6ce880244df1b6473b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 05:46:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd4bf37de9f449f39ae258a8be7e1913\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4605f7155652453da5b20662a1238ff8 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3f7882f3c42146dabdd8516f56b4fb5d : SUCCESS in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0750716a9fa228fa6ed4144a0731c53baf83645d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 05:55:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"719ab14edaae75cd2380399fdc6d7ee91e50fdd8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 05:58:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"74a3c6616c2dc91afee9ead2bd198ca3d6562ddd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 06:02:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21bb7c77ea2e41479dc7929653776cc7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a80eda28b054b4d90f2aba4e539ebec : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e5a77cceb2f640179138d3cccb34eafc : SUCCESS in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"37448c7546289d002f08d9c7eb703fefd1bce609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 08:15:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cdd40dd108384d9f82c2314377a82f61\n\n- grenade https://zuul.opendev.org/t/openstack/build/3eead54c9e0249138f382bf09c3a4dd7 : SUCCESS in 57m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b2cfb369b7ed48b7bdbd9be4a0996e6d : SUCCESS in 1h 27m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3d8e31ca98b4531b3b6ee6cd904cc1e : SUCCESS in 1h 01m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e96f0a7490f443886e06658026c0912 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56ccb1ec29b5427695252998140e00fc : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/426ccad3d8c24cba804133d7ef2db34d : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/115f6dea01d947cda28cf57e7c574dae : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/382a4972be884dae9ea239d5e98f02d8 : SUCCESS in 5m 29s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/298dd849cd4948109ea743298a1fe9e5 : RETRY_LIMIT in 1m 49s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/88cc6c26b44c469e8b5342544bb17927 : SUCCESS in 8m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ea5cfe83b68e4092aee33cdfe30d4dc7 : SUCCESS in 7m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8dc8a9c225b54dbca37740bcf40a7d42 : SUCCESS in 12m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/451df326fae04851ba92b30907c578cb : SUCCESS in 10m 42s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3112ca2c06334d91b78d44c64e47844b : SUCCESS in 1h 47m 16s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/570ebe1467ec4febb8a6f41232a95e60 : SUCCESS in 1h 50m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6fa6b5c4d26b4a5ca88e0ffde712acb4 : SUCCESS in 2h 00m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/edbf3e11ecec4019a1df79bc68433df5 : TIMED_OUT in 2h 04m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/baca35774ae747e9852ad2fdec3b7c05 : SUCCESS in 2h 12m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7bb8f99c21247a29868a477a3715deb : SUCCESS in 46m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e3597dcd526d4d41be439ea0a1b88164 : SUCCESS in 1h 55m 07s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e956a4da7b694d7e87bb555a284e5082 : SUCCESS in 1h 53m 52s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ee56d1e592f841e29c2fe2443e2b3e2f : SUCCESS in 58m 21s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/77db47347f93476fb09de67c0917c9c0 : SUCCESS in 24m 46s","accounts_in_message":[],"_revision_number":2},{"id":"a4cf071bc2f5eaa8da015cbd927a5ebac4c53edd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 08:16:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e237bff9d9c1b89ee4ab34f24d75e02006fd86ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 08:31:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86fa9e634e874153944fc3471c0994a3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dbc9793532734b109508f2a4065dde33 : SUCCESS in 13m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/78db7343867d4bf9a8e84c47eefc1992 : SUCCESS in 13m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c3d81a0532cb3f7431a9f4fab73f8a9a8f4a73a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 10:34:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c865ae36fb1413ea972f343f9c1cdaa\n\n- grenade https://zuul.opendev.org/t/openstack/build/07bb73515ec54d05a1df26f41dff48d4 : SUCCESS in 58m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6f4fc8058b8c491baf787044275fe8a6 : SUCCESS in 1h 39m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd3cc3d4980c4b8db53f18a603f9d34f : SUCCESS in 57m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d84233e2aebe40249b2847144700855c : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1ac0be269304e1bbf0b268363bbb1d0 : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e9174fcd98dc43fa8c7fccfaf7e2fd08 : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0f79d947ee89407596a30e248e92f506 : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e533c6c6e1c4f5a9c326a07f9c894f3 : SUCCESS in 5m 15s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e98c9e9c63e0471492ac0ff5779ef889 : RETRY_LIMIT in 1m 56s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/37cc5e1000764f64b0445ed2ff056a36 : SUCCESS in 8m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a1c7ca677d8444cdb5ffd2e58c09ad46 : SUCCESS in 8m 02s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4c7e0f3a77ed434991b58b9824cd9d1c : SUCCESS in 14m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2ec864fbaf894853931d6c03d6e65b9f : SUCCESS in 8m 08s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/775a5a5977ba4bc7868079c94dc7fd74 : SUCCESS in 1h 45m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7f651efc7269403d9cbca010fd1d0d59 : SUCCESS in 1h 57m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/10b2c290beb540e4ab95d966bcc16e13 : SUCCESS in 1h 54m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b83e849781cf480d8328afcd0c891e9d : SUCCESS in 1h 47m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3e1dfa3eaf8241358962d3a49eca1713 : SUCCESS in 2h 09m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd6617730d0c4b5a8d7d85757b7c57a6 : SUCCESS in 56m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/34bcfaa5dc4c4f0bbef525952ae6f805 : SUCCESS in 1h 49m 34s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/5bcbe586320543558759ed01c66841d2 : SUCCESS in 1h 57m 35s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d056f16d33fb4efd945fa4f7b3aa2072 : SUCCESS in 1h 25m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8dfbe3466cff453a9e2f3114ccca2312 : SUCCESS in 32m 47s","accounts_in_message":[],"_revision_number":2},{"id":"da8132bbf052084feebd6138e5f440ba7bbc84b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 10:35:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a8e29b5fcc8c59930fbf98e66f4d59c08f196bdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 12:56:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e24188a6e88a41388bb7a9932271a8ff\n\n- grenade https://zuul.opendev.org/t/openstack/build/4c889e2df27043f797fce9b815aa3e24 : FAILURE in 59m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b5767662725e4e52ae4ea184391727d4 : SUCCESS in 1h 22m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1404494e4574481b3fed78ad06aacad : FAILURE in 59m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c6cce2746714377aed44302257ff903 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66aaa4945a234dff8770480ea8265fdb : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5abd8b6944334d0cb64b3526cc1e9cfb : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6cf054a57eb4e55bfd90bde3706bae5 : SUCCESS in 7m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/047a746e58e542c59871761ffd9f59be : SUCCESS in 9m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/832f6125606d43bd99d2406419906aad : SUCCESS in 8m 43s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a2066a87883847e7bff8aa4096248e70 : SUCCESS in 13m 02s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/56c421f501e24e89afa8d321eadf456d : SUCCESS in 1h 52m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fb7b3621cc8e4ef69f5189c8cd5c0ae0 : SUCCESS in 1h 35m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/18112c30aa7e4b3fabec03b26c303fea : SUCCESS in 1h 56m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1bb1a88d6cd4a95ba7330df4fe6ca28 : SUCCESS in 45m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ee4e9b92e6d04a71a771a3cf36129868 : SUCCESS in 2h 07m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6086fe3b6c4c4499b66eec8bdc559dee : SUCCESS in 35m 20s","accounts_in_message":[],"_revision_number":2},{"id":"d047f2d365d6a84ec580e4d643e252bf4e205986","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 14:10:14.000000000","message":"Topic dnm/tempest-fix-test removed","accounts_in_message":[],"_revision_number":2},{"id":"6b5fe8b909b84587c8dd1d4efa62752abd4e7764","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 14:10:19.000000000","message":"Topic set to bug/2059809","accounts_in_message":[],"_revision_number":2},{"id":"95e4f7428667f94b0c426c165b1d1cb4a506d7b6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 14:10:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3f6707e263aa40d736267048afff1911f20a5617","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 14:17:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0c714fac9f74693a582c7df6e9dfa09\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bdfdf864178744ec9b1710f3b98f2d86 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5a8bd3edc59148d2b7dc3b8898db04d3 : SUCCESS in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1dca520748411b80e71bebb8e48ad17572eb610b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 16:43:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/15f30f8aa53e40ff8494aef8f97b2ed8\n\n- grenade https://zuul.opendev.org/t/openstack/build/10f93825ab2241edb66dca259fbec9c4 : SUCCESS in 56m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fa36645115564512810f75cd5f3ace38 : SUCCESS in 1h 25m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d8df3b33012489ebddb87007bdf0e4d : SUCCESS in 55m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43d66735222243c0a0367c64539a381e : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9b5827698da4a5a821b82371d72fbc0 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/277dce433fad46fd8dd96dab4a55de3b : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/80225b8c28b34ddf86504e678b7967f1 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87e5125965f64959a7d82631b64a2572 : SUCCESS in 5m 14s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/80a8909d21844bfe8fe99ca76e198e1c : RETRY_LIMIT in 1m 56s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/68fd8fc8a4b14ea8838b3db76916e790 : SUCCESS in 7m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ffaa5e5832c64bd69c99d3e17840597e : SUCCESS in 7m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5d5627a747de42f1afc8d01aa24b88ca : SUCCESS in 16m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fa87880b5e1e455ca86b63f0104b9039 : SUCCESS in 9m 02s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/236a1cd56ef34d2cb16d086e8de466ac : FAILURE in 1h 56m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1047f24c1eb94c58948377c34c4193e1 : SUCCESS in 1h 59m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2e59870286ea4f0dbcfcbdc1919bdd62 : SUCCESS in 1h 55m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9de739a4338944459b9aa227bd0cb0d7 : TIMED_OUT in 2h 05m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e4774cf815534df99495e5c1014479ca : FAILURE in 2h 03m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1bbeb76a60eb4de3897b64fbe3dab9b9 : SUCCESS in 1h 03m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/20e6e8f15b0c4d2d927d68b2420c32a6 : SUCCESS in 1h 46m 54s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/79c566acc4f347a4ad013694f52e7fbc : SUCCESS in 1h 57m 48s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/42e105e2cddf4bcabc7d3c0501f92014 : SUCCESS in 1h 10m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0842ec232040480aa66b1a527161e3d4 : SUCCESS in 40m 16s","accounts_in_message":[],"_revision_number":2},{"id":"0d7ffdddfec34b7c967d7d559a2eecb886c93c0e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 16:49:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eabb574e7b55f7849e7eeb1b49b122aeebde92be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 17:00:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6463a285f3846baac0911ee00abfb0e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e0705729713e45de9dda8a79184080be : SUCCESS in 10m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5767590ad0e24e09aa91b31829528914 : SUCCESS in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4b0924d79204f8df3ab816b0376719d7349b10f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 19:06:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13ec094acb6d429ba72ca3582bb9a5dd\n\n- grenade https://zuul.opendev.org/t/openstack/build/973110eb76604693be630f8200f4829e : SUCCESS in 1h 13m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6c29a926ad77473d9b34d88ab0ee9718 : SUCCESS in 1h 33m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee5a0155e46d49d3b1ea86e9e6dfd800 : SUCCESS in 1h 04m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f75fe4cceb414b6b9fc676c1e63966b2 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d9579dd96b34eabab2978362b5ab05c : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df532d62153e48e480bb5fb88a8b5345 : SUCCESS in 5m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7cfb36695e86451d9c67ca7c1c690fec : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e23a61dd85df43b8b61123a064beddaf : SUCCESS in 7m 13s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/93146fdef41343a0acf62ced03020d01 : RETRY_LIMIT in 2m 06s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d778e23955d54ac18869e360caefc809 : SUCCESS in 9m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d208677e7ea84dfbb8e376ea03834523 : SUCCESS in 9m 30s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/76f9d409cd664432995306758609e5a6 : SUCCESS in 17m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/97bf55bca14642b39b2289123f1ed3fa : SUCCESS in 9m 33s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/68bc99043c2e49da84221c84427905ef : SUCCESS in 1h 58m 02s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/46555e20a90a4bae86f4d63d36224633 : SUCCESS in 2h 00m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4e8d14e897ae46feadbd4f3c0a40624a : SUCCESS in 2h 03m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c826c553520a46b5a37e3c04e424eb21 : SUCCESS in 1h 49m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3904cd90448a42e882fd297f8f6aa2b8 : SUCCESS in 2h 00m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80c6a32fc1cb4230857a8b2e60b88837 : SUCCESS in 56m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/68ac99fa94bd4539bc9616eadd5596d7 : SUCCESS in 1h 50m 44s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/122b697afbfe48de828e32954cd9a60a : SUCCESS in 1h 50m 58s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/73a56bb22eaf4cfdb0b618098a24f426 : SUCCESS in 55m 59s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/65ed382ab93749ab9cd683ab116421b6 : SUCCESS in 34m 02s","accounts_in_message":[],"_revision_number":2},{"id":"de4fb937b54fe2ac9094ec2b9c7ac761313e5367","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 19:56:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d697f036d0a24cc5563907825b32c0f0c9e851a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 00:45:42.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b3c8fb745f640db8babc18ac5de83b6\n\n- grenade https://zuul.opendev.org/t/openstack/build/5af9b6bfe24149318fb346f3fc8b69b1 : SUCCESS in 57m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2c125f60da6d47369d5f756eb01d3c6e : SUCCESS in 1h 20m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e07b01f17510471f94fe5cd8aee92af7 : SUCCESS in 54m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b4c9655ea9c42e394207344b9d22731 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50558be852204d008fbd4f6bc6c89727 : SUCCESS in 7m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d4ddee4a4a0436b911f3e7576f68965 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/046849b39a6542aa90664fde0c88a9ee : SUCCESS in 5m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2f013882e42d4b8e8d2903269e2e06c8 : SUCCESS in 7m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0238c8fdc7c2480994bd40669e7ea128 : SUCCESS in 8m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/981379a6494b484583be34201f862816 : SUCCESS in 12m 30s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3b3a9a303b9249099b9a658e3bb64d39 : SUCCESS in 2h 01m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/930c56073156421fab503b89c2722bc8 : TIMED_OUT in 2h 03m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/21459157172b4381b8388a72febc9729 : TIMED_OUT in 2h 03m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a06628f0c8c441eb1c53c1e52636d35 : SUCCESS in 1h 01m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0cd1dbcffa8a469ea30be4e5ce6720c5 : SUCCESS in 1h 58m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0a6b5dd42d854351b581d14aaf405cc7 : SUCCESS in 35m 05s","accounts_in_message":[],"_revision_number":2},{"id":"6f9bac6e6488cd08aed7afc54d78781cac25eb43","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-07-04 01:16:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a8fe9273105d8a3fbd0956249249e9ba641d7e65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 01:22:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eae82503330f448eb114640a9d35e7d0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca5116853b764d768b9b820d1dc0d120 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3f87fd7ca11246ee9686fbd91aba4177 : SUCCESS in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7235f6c038b6ba57353d0987b0ea65fd1ec35a01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 03:31:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71e0ef02d72c4347b522573099e90e79\n\n- grenade https://zuul.opendev.org/t/openstack/build/1659a8924e6741a394bfaa2400f17b50 : SUCCESS in 1h 10m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e1bd85f93199432bb87c49618117bcb7 : SUCCESS in 1h 48m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acf8394060224c6fba424c006a44417d : SUCCESS in 59m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e71f927cf97b4afb89b2bc24a95bbab2 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd7bbe3b83aa4001849048be2d893774 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0561f3d5c797425aab147d311318fb3b : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d38834ef6ab4d148571c7015039b42d : SUCCESS in 13m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/890fe2355cb24f2d8e68e44357e9ad3a : SUCCESS in 5m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f17deb83dcf64186b4b9d3007a94e442 : RETRY_LIMIT in 1m 30s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/32e268a793b14fbd87f003322a6b4a0a : SUCCESS in 7m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2ddfe1c328b742f98952df560115665a : SUCCESS in 8m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cb14c40eea82450ba492895333b83e85 : SUCCESS in 15m 16s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/056f13072e5848e88caf3183d42e6501 : SUCCESS in 8m 34s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fd274997694741b6a523cf48fc5695c0 : SUCCESS in 1h 47m 12s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1929eccfb8c441168f32f6ee9d857a84 : TIMED_OUT in 2h 04m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3dce7ccb92b542f3a35d9c694c2cfc17 : SUCCESS in 1h 56m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f05c84e858514721a90246dad288df75 : TIMED_OUT in 2h 04m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0bec99ec511f432c930f65f41b7527cc : SUCCESS in 2h 10m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cda60b55cee84d32b79c833cbce9cb15 : SUCCESS in 59m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8039557e563141f5ba349c516efb2d34 : SUCCESS in 1h 55m 19s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/5873352888d445efa2e06482f058e3bf : TIMED_OUT in 2h 06m 21s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9b11a29a764f4c7e81f6e50b7ed71f4a : SUCCESS in 1h 04m 39s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3989afcfdede49f697216c1ebb9cdc95 : SUCCESS in 37m 24s","accounts_in_message":[],"_revision_number":2},{"id":"4a005df3750960b7d0f62720c683a40a1fd45ec2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-04 05:10:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c28640ae40f9bfe05a88f13206f8cc10f8f41b22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 05:25:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61ed8318c40840ccb477ca8223c09463\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aff4b3647d234c54b87b1593761ca799 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/077d61fdd71b4d52a6916f360c5afda9 : SUCCESS in 12m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fd4ae11fa03b7fd69b66b3c07abf0eb2c4f68679","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 07:32:52.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/73a0e64979274cc5ad4ee676ed257a59\n\n- grenade https://zuul.opendev.org/t/openstack/build/f2bdddf183b442db8674362ccd04389e : SUCCESS in 56m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4b21906431284424aaa8791e5b2cf796 : SUCCESS in 1h 23m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5360b9b414e42a39bc2b7adacc08065 : SUCCESS in 1h 08m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4a7b31e6a8e49c88094860d981a59a9 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abafcd74214346a6a9a92f14eda2ccef : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8893f7fa9aa24690bf83d757288ed6a1 : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0de77dfa51f40a49f60155ee4fde09d : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/131dfa25dbb64149ac90f4c5a670430f : SUCCESS in 5m 26s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/86a60a05a7634474a0ad118a8a3dca6c : RETRY_LIMIT in 1m 44s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3174119301d745acbce99c99fa94a01f : SUCCESS in 7m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/144f5051aa314d4483a648ad48c6f267 : SUCCESS in 8m 30s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c44dfb719ea543dd8f7b410ffdbc9228 : SUCCESS in 13m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f9be07945456429292120eb9d89cea69 : SUCCESS in 8m 41s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f43aa17b9baf4884bb2121458f1ef31f : SUCCESS in 1h 50m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/309c93bf437c422e958e7fb05dd98928 : SUCCESS in 1h 42m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/01e116e688b844af9423b4cee7e594af : SUCCESS in 1h 55m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b66ba8a2bcb94677beed0cb0a6f21c52 : SUCCESS in 1h 52m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4c0584f1ea8146b4bfa11f55f1c5688c : SUCCESS in 1h 54m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a96a8445076437498fd12f8396c34ee : SUCCESS in 58m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/00b885dbea494650b5712bddef24322e : FAILURE in 1h 49m 12s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/10ffee8de99240ecbe8cf673613b84bb : SUCCESS in 1h 43m 01s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/135113951fc84f1981f7efad979d64eb : SUCCESS in 1h 05m 26s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/39c78b30a9fb482b96087d50498a9ff5 : SUCCESS in 36m 47s","accounts_in_message":[],"_revision_number":2},{"id":"387ca6cd0142de2cd69c54df44874b02c7febf84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 09:21:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c5b4579da9d4e1782f94da884b310e8\n\n- grenade https://zuul.opendev.org/t/openstack/build/17e1fb5daaa9405aaf2f4ca0d2716583 : SUCCESS in 1h 02m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f4788dded3284277846bf4046a24204b : SUCCESS in 1h 40m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3528d577ee4647a8884e513f2502cb0e : SUCCESS in 57m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/466ead6a524444d39ea0f7d96a046542 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23afb6aaf1ae4514a45da9bbc20cf9dd : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2df86aa422ef4b2b9b3dacac70be603b : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3402f16dbaba414098c981dc12515b62 : SUCCESS in 7m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/384d3d085b2444a08bcd26dfda97ebd1 : SUCCESS in 10m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/472c706783fe4c0ebefc07603b05f021 : SUCCESS in 9m 21s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/92c4bab7d43f45ae995aed972489dc63 : SUCCESS in 16m 44s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f9691048b70b4ff7a112c2e6cc4db2b0 : SUCCESS in 1h 55m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/10029b3dc33e4f339ba69673a5cf8e77 : SUCCESS in 1h 48m 09s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ac9a6aa4f4424032beeb69f13582ecb2 : SUCCESS in 1h 59m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3898b2612b2401892ebe2e309e8920a : SUCCESS in 1h 06m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c9f09e5888634d53b9e89b95a4c48da5 : SUCCESS in 1h 38m 00s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9a27c5776a02439c88472b5ecdbe47fb : SUCCESS in 35m 21s","accounts_in_message":[],"_revision_number":2},{"id":"6b95ed089fc3c07878294314a2bef4e69d09ed44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 09:21:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"426484781c195013ef3d80ce35084186b9d7d477","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 09:22:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a53957897b094910982e7567f4d8cdee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c88df169fc5e412597f878ef4ce41afa : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"937f7d3e2959632813f5e0040946c6c2504f2fa0","revisions":{"ec15a65136e31cb9b0a72c24a569ec9e1ed636df":{"kind":"REWORK","_number":1,"created":"2024-07-02 14:09:02.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/48/923248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/923248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/923248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/923248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/923248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/923248/1"}}},"commit":{"parents":[{"commit":"4dec531b0aea15e8d970fa775a6e90068ddd5b03","subject":"Merge \"Use constants for common rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4dec531b0aea15e8d970fa775a6e90068ddd5b03"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-01 15:06:31.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-01 13:24:10.000000000","tz":0},"subject":"Reject qcow files with data-file attributes","message":"Reject qcow files with data-file attributes\n\nChange-Id: I6326a3e85c1ba4cb1da944a4323769f2399ed2c1\nCloses-Bug: #2059809\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ec15a65136e31cb9b0a72c24a569ec9e1ed636df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ec15a65136e31cb9b0a72c24a569ec9e1ed636df"}]},"branch":"refs/heads/master"},"937f7d3e2959632813f5e0040946c6c2504f2fa0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-03 05:55:53.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/48/923248/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/923248/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/923248/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/923248/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/923248/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/923248/2"}}},"commit":{"parents":[{"commit":"244ccb68dd9c1be8b8c6223c50998464a394b688","subject":"Fix import job to provide valid disk-formats list to tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/244ccb68dd9c1be8b8c6223c50998464a394b688"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-01 15:06:31.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-03 05:48:58.000000000","tz":0},"subject":"Reject qcow files with data-file attributes","message":"Reject qcow files with data-file attributes\n\nChange-Id: I6326a3e85c1ba4cb1da944a4323769f2399ed2c1\nCloses-Bug: #2059809\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/937f7d3e2959632813f5e0040946c6c2504f2fa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/937f7d3e2959632813f5e0040946c6c2504f2fa0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
