)]}'
{"id":"openstack%2Fglance~923251","triplet_id":"openstack%2Fglance~master~I0554706368e573e11f649c09569f7c21cbc8634b","project":"openstack/glance","branch":"master","topic":"bug/2059809","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-04 09:46:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0554706368e573e11f649c09569f7c21cbc8634b","subject":"Reject unsafe qcow and vmdk files","status":"MERGED","created":"2024-07-02 14:09:02.000000000","updated":"2024-07-04 09:46:19.000000000","submitted":"2024-07-04 09:46:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923251-bug/2059809","meta_rev_id":"bbb3735b6ee443d6d7f1b134e737cdf61af79d73","_number":923251,"virtual_id_number":923251,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-04 09:46:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-03 05:58:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-03 05:58:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 14:09:02.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 14:16:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 19:21:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac2085ba4a9df00d67b125be2ff040176d5327ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 14:09:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2736a9a6d01fc1e731bb3f262664f5804b8d7ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 14:16:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c7406431e244f65aa8a6c8bf6cf6355\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b2fcc8c1d8045cb946e889c98d51847 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d93e41e728a046e6b669271a4303286b : FAILURE in 3m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d728edf65977fd35df475a853776aa413778a276","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 16:32:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f7edb7e9169cfd86395d78dd22ebccbdd7dd9811","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 19:21:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a964fef8033c44faa57a38a7d9d7a73b\n\n- grenade https://zuul.opendev.org/t/openstack/build/b1e7ba8817d844f6bd636af3ef9b6dcc : SUCCESS in 1h 07m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c78db6659bb54b829296c57aa9a13360 : SUCCESS in 1h 28m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bcbd5888a43345b58dae578255289a6d : SUCCESS in 1h 06m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f609d675b1114d2581e8f455db9503d0 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63b914aa4ebd467c8719c8fc8cd8e0b9 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/db271f14573742108d60caf58fd18850 : SUCCESS in 6m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/169a941308e646518be18414cb076bad : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37cf593e1ec8481a9767abb8b34a39ff : SUCCESS in 5m 35s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/acc3d6dd30824a529e0548d58fa9f3f4 : RETRY_LIMIT in 2m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6fa5fe5fb9ba41419431a1090278392a : SUCCESS in 8m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/78bc27831d7248c692d92e34534d8008 : SUCCESS in 8m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0da14a68acdc4828be13581ac8f8c06d : SUCCESS in 13m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fc8a209271ed420bb930dac095450fb7 : SUCCESS in 8m 58s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/030e272bb2ea40899c725b81164f4c37 : SUCCESS in 2h 02m 55s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/23676e0963ef4a0499bcb7e346db1f13 : TIMED_OUT in 2h 05m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/38379becb37145ecb327d583c6b901d7 : TIMED_OUT in 2h 04m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6b905e144d9f45338a930ad8e72f5898 : FAILURE in 1h 51m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c1fa11123c5b41d59459cf41aa114220 : FAILURE in 2h 07m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55b2286dae7f471f9bd672bfcdbe52b4 : SUCCESS in 53m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0b1b99a960e84eb59654bb8f04d71fb8 : FAILURE in 1h 57m 08s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/d24e43ea7e5b483dbf25cadc2dcbe4f8 : FAILURE in 1h 44m 15s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/fd4ba47df3934b5dbd515a388d5121c5 : SUCCESS in 1h 08m 14s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/23df809cf98944acbbd27b0d43c5d4b4 : SUCCESS in 31m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f29d1db336779d46f052482b9bb69d67ba4f5924","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 01:44:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"516a14d37400d34e4e3d9283462c3cbc189467d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 02:00:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23c2adc697e54fd583345af9ca608de4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/edf007a8562f43a7a421ac68a39a7aaf : SUCCESS in 14m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bd53038c11f945959c5c90ad422a7d3f : SUCCESS in 5m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29d3ae6c2d56b638ce2ba6ef5ab84905ea763a33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 04:03:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4623e19852cd4dfd992df689b1aed806\n\n- grenade https://zuul.opendev.org/t/openstack/build/bc0259cbd1644759b4bf29adb4ae6c92 : SUCCESS in 1h 12m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ffa3f889d70a4334916e12e3acf23fa1 : SUCCESS in 1h 35m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35cff1c414074f658b5077d9220d6971 : SUCCESS in 59m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3a843a5bc0344529b0a9ab66ade1e13 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e29d59a6e44c4b6894501d68dfb4316d : SUCCESS in 6m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e640c51c2e247c69dcac57d1a0341a6 : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a34f96ca9064a3ea81773ecb32e84b7 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/414dfdc0acf14d2fbf7f0a25069af218 : SUCCESS in 5m 29s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c5d6c1b24f304097837915b5d828a2b7 : RETRY_LIMIT in 1m 59s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d1f12c062b9245b19b1d8c93b0dc7ad4 : SUCCESS in 8m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/82edadc35a3f4a96a31af055c7e6898f : SUCCESS in 8m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/39856ba823ce498cba493f3e4d3e65ed : SUCCESS in 13m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9686d9562d064e27832269a87196f7e9 : SUCCESS in 10m 28s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/65ff43bd47da4d8fb823f4f9c2952f1a : TIMED_OUT in 2h 04m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7280d53fc7114598a388a70ec5240bb3 : TIMED_OUT in 2h 04m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9c8de064da86403883667d75488c404a : TIMED_OUT in 2h 03m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/09148b76f44243299293251d93eed96e : FAILURE in 2h 01m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/791aa5cf61094ef6bd859a07252bad96 : FAILURE in 2h 12m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/115be0dbc9974e328830c9a3610df199 : SUCCESS in 42m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7ed33e6287a44a0fa4c2ea196fe808ce : FAILURE in 2h 13m 11s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/5cf67a4bc57f4f04806c35091bf3f903 : FAILURE in 1h 53m 28s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9974a7c46907419d8127d5a8a3435058 : SUCCESS in 1h 09m 27s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/29b326aa24854bd6a11fa144de73c6fb : SUCCESS in 27m 28s","accounts_in_message":[],"_revision_number":1},{"id":"377c0f626308a5828abf9224f44dbd8ddc9770dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 05:55:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d5352c1b8b0440efde4071b85545aa85fe1331b3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 05:58:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5d0209982ad86c56f4c38ce9989525451cf4a77f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 06:03:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25f0d0adcd1e4e47910211b860643cd4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/47fc5aa0e4e34568a74653e03a5219ef : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/abe1320de20f475aa463af5bf01bf0aa : SUCCESS in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b819fa8a989b20f015aebaa21cb2804a88e4211","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 08:49:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cafaa22c03a04e5096e45a8ddbb800d3\n\n- grenade https://zuul.opendev.org/t/openstack/build/b09ae2d29fd2438f9a41c0ab28d31e60 : SUCCESS in 59m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/51e8b951c93842cd880bba9468b9f878 : SUCCESS in 1h 36m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2354845406fb4a0dad3ac76f11da6e39 : SUCCESS in 1h 04m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b234242ec4f4e5eb49df06d901fe69c : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b86a21ff3cee4b12a31a4f6a3506afbf : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dba4c38c4b084c23bfb238d5b87a8bb3 : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2e1dba21b6f346c98a49ea2caf8bb33f : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a901ff29bcfb4e35a1ecea0551340f38 : SUCCESS in 5m 39s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/50c03125cfd344e2abd671d0c8dfe098 : RETRY_LIMIT in 1m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/895d79aa63d746db80915c803815c0dd : SUCCESS in 8m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9bf2e7ecd05640fb9627dbcbdd47b739 : SUCCESS in 8m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/11798f864cc0470d807f82922251305b : SUCCESS in 13m 44s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d66b00f01faf48b9b6dd0df546150ade : SUCCESS in 9m 32s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3d76e3c1cfa049eea577971077d9e4b6 : TIMED_OUT in 2h 04m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0c1882177a624172a370cf9cd33c3e45 : TIMED_OUT in 2h 04m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/075eb01d028a4df9a7b5c033d7b78b26 : FAILURE in 2h 03m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4658532fabe54fd9b9ef8f138bd9727b : TIMED_OUT in 2h 04m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7ddc17d446444125bf298394a03fdc13 : FAILURE in 2h 14m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91b9a2c2ad9e492a8d65e85568fdc871 : SUCCESS in 1h 04m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f48ad725760b46d5ba1dbbc24e54e579 : FAILURE in 2h 11m 08s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/f998f9bf5f8c45748e1eca496edbd9ff : FAILURE in 2h 03m 30s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5a1e1c57e9d441989e0b23e845bd6108 : SUCCESS in 1h 05m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/00a4191dd2054519911245ea55baa828 : SUCCESS in 29m 31s","accounts_in_message":[],"_revision_number":2},{"id":"57045ed95cd027ab1c5f4a64a0345873bd4d96ae","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 14:11:58.000000000","message":"Topic dnm/tempest-fix-test removed","accounts_in_message":[],"_revision_number":2},{"id":"42e70cdf5e406fa6b20009e68d1139aa1e1ce57d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 14:12:01.000000000","message":"Topic set to bug/2059809","accounts_in_message":[],"_revision_number":2},{"id":"ad92a395330437877f8f4ebf10bfc455a6116ce7","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-07-03 20:54:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8e0bdee79369235b8a4c1d516bebfc48351776d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 21:07:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/111f3797adc34931869e5a14fefd8461\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7174986776b34b73aa7403e2aa081a6a : SUCCESS in 11m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/98d70d5eac814bde87984b142e7cb8a2 : FAILURE in 3m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"25d23abc0a115d5f655abfd3e99ac69d7994948c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 23:29:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2f9fbba7b0842e7b4eb08fb2b6cee7e\n\n- grenade https://zuul.opendev.org/t/openstack/build/f41793d92aaf4b2aabfa521cabf23086 : SUCCESS in 59m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fcb9f6cfa96b4a5abcc687914bba1ff8 : SUCCESS in 1h 25m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8de2f31b50c94dc7ac70022d5452222b : SUCCESS in 1h 01m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4459506bab574774b4ef4a0d6355388e : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67cdd34d47044c2e8a8f05362401e88c : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ff396d00ce948b495c58d9c83d39a4c : SUCCESS in 5m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2ee12ff5bac3488292e260306f6d8768 : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79a0d5901ecc49f29b4a5cf9af6f4b5c : SUCCESS in 5m 27s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/fca3796db80a432abf80c2b8405c5b93 : RETRY_LIMIT in 2m 42s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/92e21880537548e285d40d96a01e2087 : SUCCESS in 7m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5a3284546a1f423389ca5a05536adca2 : SUCCESS in 8m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/070d993a855d4dd480940c3422b3c885 : SUCCESS in 12m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/aa6cceea76004467960f4a736219ad19 : SUCCESS in 8m 18s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/45b1bdea359242bd95bf7f0f88aceab0 : SUCCESS in 2h 02m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/72a9fe438c744ec19897cdbbd07d3904 : SUCCESS in 1h 55m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/984718b1679547b0b8db5b9bef16fcd3 : SUCCESS in 1h 56m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6465f0e570a24f359895b2dbe14aea42 : SUCCESS in 1h 55m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/27067a4603654d04ad8c4d46a03ef883 : SUCCESS in 2h 13m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aec0c2397d8c41c69ef0910aedf581dc : SUCCESS in 54m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/42755e717ab7416eb8d10bdc3ccb95ea : SUCCESS in 2h 26m 55s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/ed143f05d9ff43f9a3ff0db92cbeae65 : TIMED_OUT in 2h 04m 12s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/07f29f8705eb4491a4f8e909187c2cbf : SUCCESS in 1h 00m 54s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/aea4a539310c474fad91a5db2bd9e610 : SUCCESS in 20m 27s","accounts_in_message":[],"_revision_number":2},{"id":"27a9baac7ecebf8ebd9caa5be74a0ddbebacb746","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 09:46:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c0780e040f54a9e9646f1a1ff89f8f9\n\n- grenade https://zuul.opendev.org/t/openstack/build/462af15721e24d1dbe0b96f2e89dcada : SUCCESS in 1h 02m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9df3821b182748f5a4cc4d70fa941d3a : SUCCESS in 1h 14m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26b14484dfdc48a2b228db7e35ee013d : SUCCESS in 46m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c937b1ea8e1f43d48c037b3c9fc00797 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bbf1a32a9df4a84bf5ac69aa4fd03a3 : SUCCESS in 6m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e7166e896f74ace96522f4b9a3d7445 : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d84245625d40c396db178d045d88d7 : SUCCESS in 8m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea49e4dc562a4cdb88917d5b8ffacf84 : SUCCESS in 8m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3c271aca230a4ed7a9e111e90f07aeae : SUCCESS in 8m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/df8e50cd3aba4570b1f5a9471844cd06 : SUCCESS in 11m 23s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d8c703bdf27c4b0dbac40eee41be05d3 : SUCCESS in 1h 37m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b803de41e0684489ae81510b2a7d104f : SUCCESS in 1h 54m 45s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/43617d7f4032421896d401969cc87e70 : SUCCESS in 2h 00m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9eba2b7a380f4d058a1ef1507e022dab : SUCCESS in 50m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f663806c7cca43509e19818465c9d304 : SUCCESS in 1h 48m 28s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e028fd6ced524e01a9af8da2cc8fabe8 : SUCCESS in 31m 30s","accounts_in_message":[],"_revision_number":2},{"id":"bbb3735b6ee443d6d7f1b134e737cdf61af79d73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 09:46:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"797e30e602769a6084643b22a18f1834602c432c","revisions":{"0f17ac471d2d6015c9aa0a645387935265d519c4":{"kind":"REWORK","_number":1,"created":"2024-07-02 14:09:02.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/51/923251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/51/923251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/923251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/923251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/923251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/51/923251/1"}}},"commit":{"parents":[{"commit":"0dcc983e569ba88276a8fbf4c00c71b80d14cc7e","subject":"Add VMDK safety check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0dcc983e569ba88276a8fbf4c00c71b80d14cc7e"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 18:20:48.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-01 13:24:10.000000000","tz":0},"subject":"Reject unsafe qcow and vmdk files","message":"Reject unsafe qcow and vmdk files\n\nThis causes us to use the format inspector to pre-examine qcow and\nvmdk files for safe configurations before even using qemu-img\non them.\n\nChange-Id: I0554706368e573e11f649c09569f7c21cbc8634b\nCloses-Bug: #2059809\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f17ac471d2d6015c9aa0a645387935265d519c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f17ac471d2d6015c9aa0a645387935265d519c4"}]},"branch":"refs/heads/master"},"797e30e602769a6084643b22a18f1834602c432c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-03 05:55:53.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/51/923251/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/51/923251/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/923251/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/923251/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/923251/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/51/923251/2"}}},"commit":{"parents":[{"commit":"e18ebdd11839b0a20b16899305e25c5098893f51","subject":"Add VMDK safety check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e18ebdd11839b0a20b16899305e25c5098893f51"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 18:20:48.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-03 05:48:58.000000000","tz":0},"subject":"Reject unsafe qcow and vmdk files","message":"Reject unsafe qcow and vmdk files\n\nThis causes us to use the format inspector to pre-examine qcow and\nvmdk files for safe configurations before even using qemu-img\non them.\n\nChange-Id: I0554706368e573e11f649c09569f7c21cbc8634b\nCloses-Bug: #2059809\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/797e30e602769a6084643b22a18f1834602c432c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/797e30e602769a6084643b22a18f1834602c432c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
