)]}'
{"id":"openstack%2Fglance~923252","triplet_id":"openstack%2Fglance~master~Ieea7b7eb0f380571bd4937cded920776e05f7ec4","project":"openstack/glance","branch":"master","topic":"bug/2059809","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-04 09:46:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieea7b7eb0f380571bd4937cded920776e05f7ec4","subject":"Add QED format detection to format_inspector","status":"MERGED","created":"2024-07-02 14:09:02.000000000","updated":"2024-07-04 09:46:22.000000000","submitted":"2024-07-04 09:46:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923252-bug/2059809","meta_rev_id":"e0939e04e440a5f154ff39668d11f6929bff9484","_number":923252,"virtual_id_number":923252,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-04 09:46:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-03 05:58:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-03 05:58:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 14:09:02.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 14:15:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 20:12:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff31c32f84a38d3edf9ec5afe3ada64980038130","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 14:09:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a6b628be878a84c942366bafe5f2f6c1494e08e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 14:15:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ae8db4e1ef94a98b713d733dcddbfb8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/78acdad0a0b74f9980d016d217c08b24 : FAILURE in 4m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c9701aef235a4c8a89d70e106cdc6fa2 : FAILURE in 2m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5346ca449eecd24333eb7bd8f4b4ac2c7ec56bca","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 16:33:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa8cb9dd6f1c85532d479a991133ad7edda1c2f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 20:12:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3329d669abbe4ecebc04dd6de7cef105\n\n- grenade https://zuul.opendev.org/t/openstack/build/a8e18411e72a4e8cad8a8df5d3b14639 : SUCCESS in 57m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b65fa6cbf4ca4b24ac6d8018b268f37c : SUCCESS in 1h 35m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b526e31555c845a2816da9a18f124b47 : SUCCESS in 57m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/869c1f132d5d40a1b70be336cf2700c1 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85307ea335824c0fa7419022feee87f0 : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2bb4ed5209c84701bfbd78bac266e3ec : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3bf731488afc41c8a0d78534390822a6 : SUCCESS in 10m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b659827de7a24f91b79be65c98a20f8e : SUCCESS in 5m 48s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/34b6e670e90e47288841158d34bc61b4 : RETRY_LIMIT in 2m 01s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b160aae00e644dcabc1f9cb441142464 : SUCCESS in 8m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3778c2bcb0614c3aaa9e906b1a848120 : SUCCESS in 7m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/327115438af549d594fbe988dbb8010e : SUCCESS in 11m 17s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1aad20cb3bbe4eabb242d63915450479 : SUCCESS in 8m 46s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/97040d4f74dc4ec8b23edfa55352334d : SUCCESS in 1h 58m 08s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/389a58eb0c79498b8a2ad0559281962c : TIMED_OUT in 2h 05m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/407b4ad45af44b3899b7d98ec0e724d6 : FAILURE in 1h 53m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a978691237644bd4aa248f1801d62d71 : FAILURE in 1h 52m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/759fa1b8e2e446fc8153146fc6bbbdb2 : FAILURE in 2h 27m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32be397cdebd41149234723bf6d06338 : SUCCESS in 1h 06m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b598fa22152542fda172ea7c1d05a022 : FAILURE in 1h 45m 49s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/85c8c84ca656435682d82dc39edb5ca5 : FAILURE in 1h 44m 20s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/16fd75d52dc04c88bed652ea14a21f00 : SUCCESS in 1h 08m 45s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2b0aa46e5a114a35b70c086f651b01f9 : SUCCESS in 25m 28s","accounts_in_message":[],"_revision_number":1},{"id":"74f52e6dda2dab5b2b0e2b24b5b4c3862c5acee5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 01:44:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08859ecac5f7437550d4aec1706b715cea23e30e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 01:57:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/739e16f9c994423c92730138f3b5c042\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6828d4f4882d40f8b386ff24cbd98ba3 : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eaedef1855d849b99344841b0e53d031 : SUCCESS in 10m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c806aaef1e5ea44e248400d3648dd8fdc6662ef9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 05:04:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ecdd7f0c4e14701bba8e413adfa9a6e\n\n- grenade https://zuul.opendev.org/t/openstack/build/537f72843e5640d7a721d4c24961ea47 : SUCCESS in 56m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/37ccf3b965a9402db42fcae8d7a606f9 : SUCCESS in 1h 31m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6320cff43cfe4f0a8aefca3296d958e5 : SUCCESS in 1h 06m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3de735bdd1324a54ba273b69f9d28ae8 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf361aaf156741efba68f2fc94c2366b : SUCCESS in 6m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee2da34c770b44349f80b35b58b1abff : SUCCESS in 5m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/13cc25cf173b4fc4849c45567b5e7466 : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbcd9899c14346d7be7ab2080036a372 : SUCCESS in 5m 32s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d76ee4f4038449198f41ca0952a3ff21 : RETRY_LIMIT in 2m 04s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3be3bbaee727441fab429dec52ed14d0 : SUCCESS in 10m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/23553aa11b0146ee91471de6a317f8a3 : SUCCESS in 7m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fb0a2d7296624e6daefc3184a95add39 : SUCCESS in 13m 04s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/585bbf83edbe434a9bcb6b6210606db3 : SUCCESS in 8m 12s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d7e261a6d0a549c489086daf265d007f : SUCCESS in 1h 53m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a93bc8de748c4e3081109bbdf5300eb9 : TIMED_OUT in 2h 04m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3cab6e63c7c3404ca21e763890b08e22 : TIMED_OUT in 2h 04m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/45870bd45ae54bcf90b6086e9bf8748b : FAILURE in 1h 50m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/43adc82ffd094504b9c45bef79859bb6 : FAILURE in 2h 32m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fbc5b9e7dfc842e8aee729d16274288d : SUCCESS in 45m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/73511327044f43a984853714b8553e9c : FAILURE in 1h 57m 14s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/85e8cd4fd5544fac8644ee43303bda65 : FAILURE in 1h 48m 01s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4466b7466b444af09e4c1c9a9929984b : SUCCESS in 1h 00m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1ad853e3861f4b8ca776b4f4e45ebc12 : SUCCESS in 29m 38s","accounts_in_message":[],"_revision_number":1},{"id":"2c7cbf31b393dbd981fe1a8755349300c0b4261f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 05:55:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"dc70f33de3c7cc636e2af2e3340c882900f841cb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 05:58:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"694b117c0436fd29440644f562a13597e3679161","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 06:16:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9834241e97046c4b7ed9542ed4b9c8d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/320a4876c3984188be356ac3e6dbcd82 : SUCCESS in 18m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d622b7cb80a84227992b1f840f098106 : SUCCESS in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a03ff746688cc5a57abbf5e083f5034da9daf9df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 08:08:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/758226763c1243558fba80e50a2cba4a\n\n- grenade https://zuul.opendev.org/t/openstack/build/c5d36d850b8349cc98fa199c17daeb47 : SUCCESS in 1h 08m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0cdbf9488e134db99d3c6d66315a122f : SUCCESS in 1h 51m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/588ee35fabf843bb8c2f8419f1855cd8 : SUCCESS in 1h 03m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3795c4e19393424b87a3b8a6b668509d : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da615cc2ceb1499db07bdc36bbbcaf5b : SUCCESS in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/759e1ae8b5ee43fca714364a6b542741 : SUCCESS in 6m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af2c0199569e4cc581d5a6b8c843a451 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da2f399882f54a91bf2f21573bb359b6 : SUCCESS in 5m 12s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f92cd45a73164c24a67986905e013649 : RETRY_LIMIT in 1m 39s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ac877587f1c54b15ad2ac944c3566f56 : SUCCESS in 9m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c4d3ff47aa8848b09ae0ac76aa469b18 : SUCCESS in 8m 23s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/49bfc25681ff42e08a5b5908b8317f79 : SUCCESS in 13m 46s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e753a86c4da04dccae99ea43a330d383 : SUCCESS in 8m 39s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/befcac5ab628481dbc9df062f144dc67 : SUCCESS in 1h 41m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9def1bccf3e245e6b5531fc34d7643d3 : SUCCESS in 1h 54m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6d91a56d44134675a690f527118020bb : TIMED_OUT in 2h 03m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fb4f6d5dd1c74661a1f7dee6bec57ded : FAILURE in 1h 59m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5ae449cc59734e21ad72718be7dcc17a : FAILURE in 1h 59m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1719ea7cf07b445e9516b648b9734e5c : SUCCESS in 45m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4f15842453904adbbec347f446e59991 : FAILURE in 1h 44m 30s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/d01fae006f9c45f7acd77f724a461500 : FAILURE in 1h 34m 23s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4a7f798c523f40e781ba4fc1ba91365b : SUCCESS in 56m 39s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b6ee29899cf240ae93dab8bde2492cc5 : SUCCESS in 32m 46s","accounts_in_message":[],"_revision_number":2},{"id":"6e37af15aa368a312a30ab060297b9f71b0d26b1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 14:12:05.000000000","message":"Topic dnm/tempest-fix-test removed","accounts_in_message":[],"_revision_number":2},{"id":"7a439e76e383623ef03447d56cba1ff12601495e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 14:12:09.000000000","message":"Topic set to bug/2059809","accounts_in_message":[],"_revision_number":2},{"id":"51d9b0661559924995a507159cd023ac53f6d414","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-07-03 20:54:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"58caea3663283611edc31e8315b1fb0f2761307d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 21:07:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1927809e63b7468da201228a1ebd8646\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0facf3e776d24f04a21a3219a99239e0 : SUCCESS in 10m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/46e4276c479e4efd8e47daabab2f28e3 : FAILURE in 3m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e8a4083085a84e25e67c0efc648eed8463789e03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 23:24:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/033b04eb20f448be92ea92eda5eae29c\n\n- grenade https://zuul.opendev.org/t/openstack/build/43a00dab777b4a65bbd56c5e761eb66e : SUCCESS in 1h 07m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d36c34f949d64446947bf8f779d130ff : SUCCESS in 1h 39m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/abd77b1d241a418f86f143060cc825db : SUCCESS in 59m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8529c7846c94131b07ea6dae6f09975 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d27036988fcc40fbb3e8b223867c567a : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94b2c03ef73c4b47ae1269029e68b5bb : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/644d103af6924e74a43a37c8d6a3daaa : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed8f350b8cf6449e86f51c8c39ce6c82 : SUCCESS in 5m 37s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c0068f8c4a1048f68f6b78f5fee11763 : RETRY_LIMIT in 1m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/09cf090f3750430a8381764de292b4b8 : SUCCESS in 9m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1e75933eae854ce98c91d6b844da5b39 : SUCCESS in 7m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3e005596eb9c40fc9bea6d24f41ef26d : SUCCESS in 11m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3562f89c70034cdcbde5666a9a1a8922 : SUCCESS in 9m 30s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/da5d1613231d4cc29012b56a44bf97c8 : SUCCESS in 1h 59m 52s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d9fc06010d514fe09fe6799dbf95d91a : FAILURE in 1h 58m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3c00687544554e0d8590e78ad29207aa : SUCCESS in 1h 45m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/705310e731094abea73af8cbfb0780d2 : TIMED_OUT in 2h 04m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c0893bb8a4d14422a7098ba5c839db75 : SUCCESS in 1h 59m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d14f40b69f0c4de4914c46261abcfba2 : SUCCESS in 47m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3fcd349c7ee04247ac073cb72f50aa28 : SUCCESS in 1h 36m 23s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/aaec363f4f1f4838b7d911465442393c : SUCCESS in 2h 01m 53s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/19cf4b0590fc40f6a495fe296d2c4110 : SUCCESS in 1h 07m 40s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e0c1f5b0c986462aa02dab7bf3f62533 : SUCCESS in 32m 08s","accounts_in_message":[],"_revision_number":2},{"id":"789e258b217cc14288a635eaeb18af713668ad5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 09:46:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44cc1f30618b463eb4c0f121211a7d05\n\n- grenade https://zuul.opendev.org/t/openstack/build/d1e33b142d1d4099a636eadbe8db5ae4 : SUCCESS in 1h 07m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/03f6914ff02f43fea73a6a83a7af4768 : SUCCESS in 1h 30m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca7b9f9aa3cb4e74b4aef5d4efacf4bc : SUCCESS in 1h 02m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd6ce0dee9624e12b3c69136a73ab1d7 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35b905f053864c5b84114c6bbc2a8d3d : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c85c1882dd7b4699b828a8f0d6083cac : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcd061cfbf9745be833b8f85c1ad130b : SUCCESS in 7m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c34e5c80a44846b2a642ef2928efc287 : SUCCESS in 9m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/98a1ce068cd54c60baa2c34270bea892 : SUCCESS in 7m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b8d4ef581e9a40b38154f0e38d770bf7 : SUCCESS in 11m 56s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/835857fe3b3e401698cb8509f66d192e : SUCCESS in 1h 55m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/52c88f10ad9c421bbbd30167e3023dd8 : SUCCESS in 2h 01m 39s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b20d3c032e914974803defffb7994dc7 : SUCCESS in 1h 58m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c66dbd02ef1419ea96c607d1cba509a : SUCCESS in 1h 01m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9f72a536b1bd4ce5b5987fcfba16c38f : SUCCESS in 1h 33m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/798e7541379947d4a4a9182a1facd229 : SUCCESS in 23m 59s","accounts_in_message":[],"_revision_number":2},{"id":"e0939e04e440a5f154ff39668d11f6929bff9484","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 09:46:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f596965b606bdec109b764aa598ccfa6a16bfda3","revisions":{"07e9e8fbf437e11ee57cc2a13a3dc9c655892684":{"kind":"REWORK","_number":1,"created":"2024-07-02 14:09:02.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/52/923252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/923252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/923252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/923252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/923252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/923252/1"}}},"commit":{"parents":[{"commit":"0f17ac471d2d6015c9aa0a645387935265d519c4","subject":"Reject unsafe qcow and vmdk files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f17ac471d2d6015c9aa0a645387935265d519c4"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-06-27 16:33:55.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-01 13:24:10.000000000","tz":0},"subject":"Add QED format detection to format_inspector","message":"Add QED format detection to format_inspector\n\nThis merely recognizes this format and always marks it as unsafe\nbecause no service supports it. This prevents someone from uploading\none that we will ask qemu-img to inspect.\n\nChange-Id: Ieea7b7eb0f380571bd4937cded920776e05f7ec4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/07e9e8fbf437e11ee57cc2a13a3dc9c655892684"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/07e9e8fbf437e11ee57cc2a13a3dc9c655892684"}]},"branch":"refs/heads/master"},"f596965b606bdec109b764aa598ccfa6a16bfda3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-03 05:55:53.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/52/923252/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/923252/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/923252/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/923252/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/923252/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/923252/2"}}},"commit":{"parents":[{"commit":"797e30e602769a6084643b22a18f1834602c432c","subject":"Reject unsafe qcow and vmdk files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/797e30e602769a6084643b22a18f1834602c432c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-06-27 16:33:55.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-03 05:48:58.000000000","tz":0},"subject":"Add QED format detection to format_inspector","message":"Add QED format detection to format_inspector\n\nThis merely recognizes this format and always marks it as unsafe\nbecause no service supports it. This prevents someone from uploading\none that we will ask qemu-img to inspect.\n\nChange-Id: Ieea7b7eb0f380571bd4937cded920776e05f7ec4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f596965b606bdec109b764aa598ccfa6a16bfda3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f596965b606bdec109b764aa598ccfa6a16bfda3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
