)]}'
{"id":"openstack%2Fglance~923253","triplet_id":"openstack%2Fglance~master~If0a4251465507be035ffaf9d855299611637cfa9","project":"openstack/glance","branch":"master","topic":"bug/2059809","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-04 09:46:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If0a4251465507be035ffaf9d855299611637cfa9","subject":"Add file format detection to format_inspector","status":"MERGED","created":"2024-07-02 14:09:02.000000000","updated":"2024-07-04 09:48:30.000000000","submitted":"2024-07-04 09:46:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923253-bug/2059809","meta_rev_id":"960efa13dbe87c23acf23946445d9982e7485577","_number":923253,"virtual_id_number":923253,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-04 09:46:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-03 05:58:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-03 05:58:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 14:09:02.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 14:14:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 19:10:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6cf979f493488651066c510121ad416aea0a9824","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 14:09:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5ba7da0bb8004acbcdcceb4175f1b2d5047ebfe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 14:14:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/316f2a011508466fb8f392ca9ec85f66\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e976a300bdc43b6993a663bb61b1df1 : FAILURE in 3m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b9678960eb2b4783954537cee70a2814 : FAILURE in 2m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62c09c05ed439ff42a80154121d178ec0e6b96c0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 16:35:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f4d8e0b769da8279488a3abf27bf41bccb97ad37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 19:10:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5952a5b87807492c98bd723a86640709\n\n- grenade https://zuul.opendev.org/t/openstack/build/0b81abbf975d4b559820e51000942993 : SUCCESS in 58m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/27a54445e67a4ae9bb28d98dafdedcc3 : SUCCESS in 1h 31m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec17fe004f2a4774b7e895c1393b20ee : SUCCESS in 1h 01m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31a0e095a1434777a04595c8d7fea489 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94339e98e3a24b378610443e7244cd86 : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a3658d4c14b4c68bf7367989fa180cc : SUCCESS in 5m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac5f8c826ca1476eaa9cf51bd6fc8a18 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afbf6d4f949d4221b846c6642f053b52 : SUCCESS in 5m 44s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/678a7b9e2dd140828948c8443b921cbf : RETRY_LIMIT in 2m 43s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/228b57bb347d40d8b9a61314f33f2860 : SUCCESS in 8m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/45d107abf1214f08bcd8499dad4514dc : SUCCESS in 7m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/386fcb66101d4c8e9cd7c2d01aedc47f : SUCCESS in 13m 32s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e2b0d42da1ed41dcb0ee36d5b600d281 : SUCCESS in 11m 44s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/660475d3d71541acb66640f8f5691a0b : SUCCESS in 2h 01m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9f3f2779129b4de2aff1a55ad08673e7 : FAILURE in 2h 02m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fe327d1812ee407d93505250c5d6315e : FAILURE in 2h 02m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1168e68f2d5d4a50bfb4e3215b8d0e7c : FAILURE in 1h 54m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e6320851414e405eb9d9715caca5ef41 : FAILURE in 1h 49m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8b84d50a1b64766b29e8a902c960fc8 : SUCCESS in 53m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d6908199f73a4e64a7727615c385e6b6 : FAILURE in 2h 01m 02s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/ac4b88713a12476ea540efcc41cc69bd : FAILURE in 1h 58m 42s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e9af151838904b12b52b5cdca77dff63 : SUCCESS in 1h 11m 59s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c61075f8dd0c46df87c2ffe677a358b0 : SUCCESS in 36m 23s","accounts_in_message":[],"_revision_number":1},{"id":"84bc21b8bcf683416c57b29f1340c592159ff2e5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 01:44:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"259476444c2dc4046a67b2808439bd72260a0ef5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 01:58:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d8b2fc51c3444dc9785d833f9e55723\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3bdee4b696e24c4fa888ffeae9cc87be : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c601b0ea5a0f492096e129475023524f : SUCCESS in 11m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b37de012a9c9976e07ac8ef6b0287782cd8b7e92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 05:48:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87943ceb28c4425e9a1aa44281b4ef49\n\n- grenade https://zuul.opendev.org/t/openstack/build/f280a475216740c49915c32768f0fbf6 : SUCCESS in 1h 03m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9a653c4853234467876a62f7a757f67e : SUCCESS in 1h 32m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3aa47885cef0400885bd5d14589daae9 : SUCCESS in 1h 02m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d597392e698347428db560b7717c4811 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/926d7ebdcd4148139cb801a396dd994a : SUCCESS in 6m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f1401771316f40fdaec1e6dbdc44d90e : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1260c5d6e4124eff83b01ac401756b98 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee2de1bfd6f44cb0b286a6e319c25df2 : SUCCESS in 5m 27s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/24339edd6cfd4f65a2374ad5c98d188b : RETRY_LIMIT in 2m 05s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1e52201b84d04ff9ab03db3cf3aa9ecf : SUCCESS in 8m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/874fc386a33c41d4bd946402436bf7cf : SUCCESS in 9m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2ffb3af439594fc984643d9ba702990e : SUCCESS in 11m 44s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/703bf0056b784c4cb40501803f7c3186 : SUCCESS in 8m 38s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1437e93d679b4bc4a8fd0c17b2a532c8 : FAILURE in 1h 54m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ffbbd308c14249a08d8f830b1683c9c9 : SUCCESS in 1h 59m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2e43d10706bb4985a6914bf1987f28c1 : FAILURE in 1h 52m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/92a3c86877df4120b2ad22d4654bf0be : TIMED_OUT in 2h 05m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c88660d309a5482188e1b87f9bd19c61 : FAILURE in 2h 19m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a78bcac1f5b4577bf7a0da17b2307d1 : SUCCESS in 56m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b55cce4a74824a879f7fc06f9ac780ab : FAILURE in 2h 08m 55s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/63fe0819779c4683affe6a31a631c251 : FAILURE in 1h 57m 01s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a6b7b08c4e154f7e875381be338eee2c : SUCCESS in 1h 10m 54s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/383e33c9f80a47d6899d3b567fe9c3e1 : SUCCESS in 34m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9f4088c1a68c0acb75c06df2c1ccc10e5d3cfefc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 05:55:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a33c18e481d22ca6df5620adc053056b21685e38","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 05:58:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cb1c559bbee8ac663c1c9beae4c0ab57dd6c1460","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 06:16:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a8f5d82065f417287ec5ad480da2239\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b62a76aa93546f6b038240223d422de : SUCCESS in 18m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9683424e2bca4c55915244a670314cb5 : SUCCESS in 13m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"90cd38f27eaeb46fbdd63ae23342645d75669731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 08:14:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f94101b459f401eb21712db842dc2d3\n\n- grenade https://zuul.opendev.org/t/openstack/build/96643bb544f3409394f38aadc5914ab1 : SUCCESS in 1h 05m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d456e898feff4610a57615268280f4d8 : FAILURE in 1h 27m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46f47ffb50f8453bae80f59e4249cef9 : SUCCESS in 58m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac707b3403c446eda3a5c30a4ff0d66b : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7072553e81fe4450a27cea784481bd7b : SUCCESS in 5m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/625377e882c146629184466c4391d763 : SUCCESS in 5m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b423d8f369f44e6cafdac614c6e14b37 : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c60d4ce7217e4e5cae6388496aba4308 : SUCCESS in 5m 55s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8721277612eb4d74a706bb9461e9c3a4 : RETRY_LIMIT in 2m 49s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aaea87ea24e84ccc9382affaf07cb20e : SUCCESS in 7m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3a8f8b4f52104b1bb6b9e342ec94cbeb : SUCCESS in 9m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9514926ccbaf43b7a5fd64111c34d197 : SUCCESS in 12m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/220a8ae3fd2f40488680bef72622a222 : SUCCESS in 8m 06s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7c89543e2642440eb16fe7c6927b67a1 : SUCCESS in 1h 47m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5ef45a51cad0499b8bb9db112963d13a : SUCCESS in 1h 59m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3c07b427dad54839947344e9650c1a57 : FAILURE in 1h 47m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/af8e247c7bca4429a80536e4792d90bd : FAILURE in 1h 48m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/61522986570f4cc281ac83d948cb4077 : FAILURE in 2h 14m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/798d184a0acd4ee7aa953f04641a97e3 : SUCCESS in 1h 00m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/45191c8329114d65b63117165b8d252c : FAILURE in 2h 01m 57s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/c8d69856f07442d6b899a47be507fb52 : FAILURE in 1h 46m 40s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9b6eba00c33d43118635241ed644450b : SUCCESS in 1h 01m 11s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4eb43609592648bcb20365b6e1779934 : SUCCESS in 38m 02s","accounts_in_message":[],"_revision_number":2},{"id":"96c296b306c03b5baf9ee7b9c929c78cfac69b63","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 14:12:14.000000000","message":"Topic dnm/tempest-fix-test removed","accounts_in_message":[],"_revision_number":2},{"id":"062487a7f9281670189ef3f5a6c6c1fc803e5e84","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 14:12:17.000000000","message":"Topic set to bug/2059809","accounts_in_message":[],"_revision_number":2},{"id":"e63132a7dc28da1bb1846329aed349c545b1aaed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 09:46:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/196b3ab752de4ad3967aaf2652626a09\n\n- grenade https://zuul.opendev.org/t/openstack/build/7189fc6217954ded888cdc67835881e5 : SUCCESS in 1h 00m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/834589e309ab4a8fbf45ff50c422c78b : SUCCESS in 1h 24m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79d667fdc0394966a4e8f8b2a55b6aba : SUCCESS in 56m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36fd1952ab7441978c0b1841f1c2e03f : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9aa18938f5674212b797f00dda492006 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc72438be97d4dc7b698798b3660f259 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1abba0254674eeb8eb3fbe4e85cd33e : SUCCESS in 7m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5a5b7bf978c8405898ed44b5d467eba3 : SUCCESS in 7m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/85804d7363f042dca6b8bb2f3e92e1aa : SUCCESS in 7m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6308af2da0af4cd1a88703d1965d69ec : SUCCESS in 13m 15s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/321a3d930524421ea4efc36084372f4e : SUCCESS in 1h 44m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e9410422e253498e8a1f0fabd1a4876a : SUCCESS in 1h 37m 50s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b6f9f54c34204264a57a7a3f406e6812 : SUCCESS in 1h 47m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4958f589ba747c699fea252c1103d4f : SUCCESS in 52m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/80398f5c617c48cc9e576cefaf985ee8 : SUCCESS in 1h 43m 18s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3f113119d98c44f185bb7880acb888ab : SUCCESS in 30m 24s","accounts_in_message":[],"_revision_number":2},{"id":"756a24c105ccee320b4d938f30d573112aaa8a51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 09:46:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"960efa13dbe87c23acf23946445d9982e7485577","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 09:48:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd1bba272007465da151c91ed88243b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db4b82d0f0664494956d540286a56c6a : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"22d59d93b900a6128530b35412e856f8d43ef444","revisions":{"0d25293ef9c1d3fef979ae02feff1ae12372323a":{"kind":"REWORK","_number":1,"created":"2024-07-02 14:09:02.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/923253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/923253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/923253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/923253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/923253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/923253/1"}}},"commit":{"parents":[{"commit":"07e9e8fbf437e11ee57cc2a13a3dc9c655892684","subject":"Add QED format detection to format_inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/07e9e8fbf437e11ee57cc2a13a3dc9c655892684"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-18 15:25:24.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-01 13:24:10.000000000","tz":0},"subject":"Add file format detection to format_inspector","message":"Add file format detection to format_inspector\n\nChange-Id: If0a4251465507be035ffaf9d855299611637cfa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0d25293ef9c1d3fef979ae02feff1ae12372323a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0d25293ef9c1d3fef979ae02feff1ae12372323a"}]},"branch":"refs/heads/master"},"22d59d93b900a6128530b35412e856f8d43ef444":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-03 05:55:53.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/53/923253/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/923253/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/923253/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/923253/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/923253/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/923253/2"}}},"commit":{"parents":[{"commit":"f596965b606bdec109b764aa598ccfa6a16bfda3","subject":"Add QED format detection to format_inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f596965b606bdec109b764aa598ccfa6a16bfda3"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-18 15:25:24.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-03 05:48:58.000000000","tz":0},"subject":"Add file format detection to format_inspector","message":"Add file format detection to format_inspector\n\nChange-Id: If0a4251465507be035ffaf9d855299611637cfa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/22d59d93b900a6128530b35412e856f8d43ef444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/22d59d93b900a6128530b35412e856f8d43ef444"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
